)]}'
{"id":"openstack%2Fswift~915763","triplet_id":"openstack%2Fswift~master~Icc65dc20a177e485ad7fe9a479fe93c1fa9ac093","project":"openstack/swift","branch":"master","topic":"utilsreorg","attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-04-13 15:13:10.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Icc65dc20a177e485ad7fe9a479fe93c1fa9ac093","subject":"wip: HASH_PATH_* refactor","status":"NEW","created":"2024-04-13 14:00:17.000000000","updated":"2024-04-13 15:13:10.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"98e24ab93dcf17c64ff88f4cc527a6831e6a0eee","_number":915763,"virtual_id_number":915763,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-04-13 15:13:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-13 15:09:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-13 15:13:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"643fa342d15ab611469f6021f7da4b0499762c25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-04-13 14:00:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c13403a2de654add0acb94fe66ae831d67183fdd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-13 15:09:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/615d35d902e6470d8f0d58eca3051937\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d27a13c9d84b40d2993832aa9ea12c12 : SUCCESS in 19m 44s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/37e4bfc1a0f943ec9eb3188e9ca4da82 : SUCCESS in 18m 43s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/dce9d6e838614edf989d84b23b9e6d95 : SUCCESS in 1h 07m 09s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f5d45f86bb4f45c4b93b3ff0e9647acf : SUCCESS in 19m 28s","accounts_in_message":[],"_revision_number":1},{"id":"98e24ab93dcf17c64ff88f4cc527a6831e6a0eee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-13 15:13:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/84f9c3869a2b4aadb073e1aff3e000da\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4826909c034a432f9af86ec9444c5a89 : SUCCESS in 9m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/a3bfd017d87e445d8606e66cabe226e4 : SUCCESS in 46m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/93e8248b9d6245dea2061c7a00dd8b6e : SUCCESS in 44m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/764ce0445303433998def7a9660a205e : SUCCESS in 1h 04m 27s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3abe2c1351f947a08e404fdf396b883a : SUCCESS in 18m 02s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ca5dbc40ba084b81bdf0d9824a0edbce : SUCCESS in 18m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/23e036e362a94e61bf2a845122840acf : SUCCESS in 19m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2b67da84c9414602a227905f95b4d021 : SUCCESS in 10m 57s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/17d85c9f5f4848d2813b7b2d39e2c03b : SUCCESS in 5m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ea9c7d85396f4916ad6eaff2a744baff : SUCCESS in 19m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/65d4bb4819ff4c209600030630fa0aa4 : SUCCESS in 19m 15s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/502d60c4b072447898b78bed44e949ec : SUCCESS in 17m 03s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/62961f0815ab44fabacd0e9631163942 : SUCCESS in 16m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ee1a2c7be4bf4b9d958ad9dfd59c83b5 : SUCCESS in 20m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e6fce3c8f1c648b19094406b7ef13514 : SUCCESS in 19m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fe3ebbda6ccf4ce496b975e23e53d7da : SUCCESS in 18m 29s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/492da26733d947f0ac71f882b92a685c : SUCCESS in 16m 18s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ff3a5e90fcd34fafb10901372baac74f : SUCCESS in 16m 03s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/24be3940d7274b0d9c682f048ae8ab6c : SUCCESS in 14m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/05bac244f2e94b1a915d9974306e6d5a : SUCCESS in 5m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6474a6fdf849435fb34cfe09d2d556ce : SUCCESS in 18m 23s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a0775e9af56a4118bd0c0f3d41506cbc : SUCCESS in 7m 05s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d5c6359a55a644ac94d71edd3e5f06ed : FAILURE in 52m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/09ad2c8529104d82b3f9b028f5301d05 : SUCCESS in 33m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/15debc4d78894596a22464019f85d477 : SUCCESS in 47m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/98f54d41c5634854acd54c49406c092a : SUCCESS in 18m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03eb351876e742c9830ac995049dcb70 : SUCCESS in 3m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/194ed77199c440cea603838ec0fbabe9 : SUCCESS in 17m 42s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6f4d15246a934294ab075e34ebc89566 : SUCCESS in 36m 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"48dc319212ac77e6a80ab0cc36168d627f3736dc","revisions":{"48dc319212ac77e6a80ab0cc36168d627f3736dc":{"kind":"REWORK","_number":1,"created":"2024-04-13 14:00:17.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/63/915763/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/63/915763/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/915763/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/915763/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/915763/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/63/915763/1"}}},"commit":{"parents":[{"commit":"86a979827796ce6b54f378000f5f55640478180c","subject":"wip: sq? utils refactor","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/86a979827796ce6b54f378000f5f55640478180c"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-04-13 03:13:19.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-04-13 13:44:05.000000000","tz":-300},"subject":"wip: HASH_PATH_* refactor","message":"wip: HASH_PATH_* refactor\n\nChange-Id: Icc65dc20a177e485ad7fe9a479fe93c1fa9ac093\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/48dc319212ac77e6a80ab0cc36168d627f3736dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/48dc319212ac77e6a80ab0cc36168d627f3736dc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
