)]}'
{"id":"openstack%2Fswift~916026","triplet_id":"openstack%2Fswift~master~Ia8aa11ad13c9b7f2c4b470799d8a9624f217b43e","project":"openstack/swift","branch":"master","topic":"random-expirer","attention_set":{},"removed_from_attention_set":{"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2024-05-31 20:53:07.000000000","reason":"Change was abandoned"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-05-31 20:50:23.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-05-31 20:53:07.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ia8aa11ad13c9b7f2c4b470799d8a9624f217b43e","subject":"distributed parallel task container iteration","status":"ABANDONED","created":"2024-04-17 00:18:17.000000000","updated":"2024-05-31 20:53:07.000000000","total_comment_count":20,"unresolved_comment_count":11,"has_review_started":true,"meta_rev_id":"517e216fe28e24edcf3cf3f3873c3bf74e6d1bea","_number":916026,"virtual_id_number":916026,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-04-23 20:50:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-17 01:33:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-23 20:46:44.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2024-05-29 05:26:24.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"CC"}],"messages":[{"id":"be84f3437c98d8b6f033aa551028283a44756863","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-04-17 00:18:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bd77994314ac02cd1fae8d78607c607a1d49847d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 01:33:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/740d207579b742e1875aac136ff56074\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4611374e298246478f66e7aa10c883fd : SUCCESS in 7m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/95c6057eab714470be829ef8f4d6a96c : SUCCESS in 1h 08m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8e5e6c1eef3546b687bdac789045aea7 : SUCCESS in 53m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/79591421b0614992b985adeb040381b7 : SUCCESS in 41m 26s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4692ec2c949e4b7799ebad9bae756bde : SUCCESS in 18m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bdd415120370486ba62f38eb3ac34ec1 : SUCCESS in 19m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/80921314412941efbc3a376ce050d405 : SUCCESS in 21m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4dc58bd85de441f1a8458712972b47fd : SUCCESS in 8m 27s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8c5fa23f89774963a4418883bdc218e8 : SUCCESS in 5m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/df01c4b9b24c4356a4d41ed20d7192d8 : FAILURE in 22m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1ad9bdd9c2674e949432b19a0b0db97a : SUCCESS in 18m 19s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/bf2ff42eb646490ebab12fa29837b76d : SUCCESS in 17m 07s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/caf95b1e88fa45be8aabd311a0ab0ac8 : SUCCESS in 12m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/91b3aac382f34c3dab244fe4155b215e : SUCCESS in 17m 53s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8b673d5c7278483d85216223de04a8ce : SUCCESS in 20m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3387832b11034fc580eac89265c3c97a : SUCCESS in 17m 46s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2af83c743c8f4212a3eccb74c6b56428 : SUCCESS in 12m 23s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7d4cf8298abd4e2ea57e1d4cf7a9011b : SUCCESS in 12m 26s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/394847c20c8a46ec9714875c01d96f5f : SUCCESS in 18m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/260b00fdf28a429e9b3537822bc6f227 : SUCCESS in 13m 31s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/99ce399dcf8a4ce28a30ba0e69b5468c : SUCCESS in 17m 48s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/96c30a939f4d45d7a28681800fc5eb82 : SUCCESS in 6m 36s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ad7c4d2e56f94c7ba30b0e3254960fda : SUCCESS in 56m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5efd7fbdf4c64ab38da813580d61bae8 : SUCCESS in 52m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c2e3d2f1284a4cb2aaa9c327be379063 : SUCCESS in 36m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bd608b400b1147ee896f72725acad47e : FAILURE in 13m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bc067a8ee64480e98bfd1ebbcb9e916 : SUCCESS in 4m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7ab5c92af34f4af1a08b356bf88205a2 : SUCCESS in 20m 51s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4efde70a10d842c1aaba29ff11ce899f : SUCCESS in 1h 03m 40s","accounts_in_message":[],"_revision_number":1},{"id":"5f2d90fba59a1c744bc8c8565f265e89be24bff3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 01:39:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5cdd667e33845c58adeaa044144bb4e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bf8daaf170e447aea5c012a5d62e2e8c : SUCCESS in 1h 13m 18s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6d924a4d2c8c4dc0a8e22dce6eed07fa : SUCCESS in 1h 18m 41s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/877ccf12e2a64ba1985b2ee6ea6b4e1d : SUCCESS in 24m 17s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/79522ba8ab27482c845ce271596378de : SUCCESS in 1h 14m 10s","accounts_in_message":[],"_revision_number":1},{"id":"11999047918dd069a7796ad5c4179fd3c225c586","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-04-17 15:13:28.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"2db264c3e397cbc94814fd2f94d5d400cae451f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-04-17 15:17:44.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"98f12757682fe8c43ad4770afcc509d67ac96751","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 16:05:10.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a0cb94f21454a5298d354e336c838aa\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3e28be6b125741b7a9211f091312f6ce : SUCCESS in 19m 56s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2c1706d3716d403682dd11fd569ffc4e : SUCCESS in 40m 45s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8ed716e68cd7497ab9561706c2df3c88 : SUCCESS in 24m 17s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b2801a42956b494cb436c24ec784a770 : SUCCESS in 45m 34s","accounts_in_message":[],"_revision_number":3},{"id":"04cc41732884bd00e918e77b29c24d0be5c574e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 16:25:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55f091b6cbfb4e7cb7cbabb7c77183e7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7869fc048764466bebfecf8c666cf72 : SUCCESS in 9m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/79223648f8f048bcbc286e2797ce6a70 : SUCCESS in 1h 02m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7a82fd5572d64102bb8056741f930240 : SUCCESS in 36m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/728abc90298a4a608266faf3d9a6cd4e : SUCCESS in 52m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d4843f3bf1dd45c2afd5278bdc4ee0a0 : SUCCESS in 19m 47s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/05dcbc36b0aa4b769705b3f519946998 : SUCCESS in 19m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/741fa9ea282f49068d84d901bf123e57 : SUCCESS in 24m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b3f54e2893174b55bf6d4e3b01199bb0 : SUCCESS in 9m 46s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1fd2d186153747acbbcbb56f8fa534dd : SUCCESS in 5m 07s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/15c28207254d4e8ab68191f7b5723922 : SUCCESS in 21m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5605657d39da4b0eb9b6929449781b13 : SUCCESS in 20m 56s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/63892bc6d6064db490dab6119a790b69 : SUCCESS in 18m 21s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/84369f4243b543ecb137a83750e719e9 : SUCCESS in 17m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/445b12b912544502b768d728d24c77b7 : SUCCESS in 18m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/af65b3bdb3fd488f8ef61cc43379a39a : SUCCESS in 22m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/42ad6dabe189422fafeb452da0e010b6 : SUCCESS in 24m 57s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d1aaea65c9184741b2f3dc889cc2ae5e : SUCCESS in 15m 21s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1cece9d31ea24abab6b695e71586f7f5 : SUCCESS in 14m 14s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/06391a2fb5f34578988f6d4b882f5cc7 : SUCCESS in 16m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b94889faf29c4b18980bc1d3fb26f72a : SUCCESS in 6m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/490e15c6e0344af19826082cd2f2ba40 : SUCCESS in 14m 40s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d168b7aefd774699ab28a032358dba5f : SUCCESS in 9m 31s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2fb5ae82980a4834b685982356d489cf : SUCCESS in 59m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/50fb27ba35e64c069f0e0c0c9a13ecc6 : SUCCESS in 48m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/acaebb1082474f18be211bd6e418216d : SUCCESS in 49m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/055cc30e3c8b4693b8e9b05d10443575 : SUCCESS in 22m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a35caa99f8947a7acca731dab2e74c9 : SUCCESS in 3m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4941f301d9e84909bd0327ec26cdd4b9 : SUCCESS in 18m 30s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f7a3b1f6719845ceb2fd7db18d609c8b : SUCCESS in 53m 58s","accounts_in_message":[],"_revision_number":3},{"id":"15cb06f00f795914936f791cc2b6648c9fca20c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-04-17 19:24:37.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"69625effa519b024f229bea201de751689ebc0aa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 20:18:34.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2a25f61bde74106afe3db5506341f2d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/74953ce0cb71481e9af2a125ff1f35a9 : SUCCESS in 14m 20s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/245d37dfaca24406a1979be0145c0b8b : SUCCESS in 43m 02s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a8b8c43c91fc4b67a544a7940008d967 : SUCCESS in 51m 59s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/aec0abd74d2242fb986290a03108189f : SUCCESS in 20m 58s","accounts_in_message":[],"_revision_number":4},{"id":"52a67f6dd716ca7c8a50e3bd7468a12a2f38dd07","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-17 20:18:49.000000000","message":"Patch Set 4: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChanges https://review.opendev.org/914713, https://review.opendev.org/874806 are needed.","accounts_in_message":[],"_revision_number":4},{"id":"f6e5797ba4778830986d6132fe8fa776e6205b8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-04-18 16:09:02.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"2d690b787ca8a0cda2101e99b3530bc1668fa1ec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-18 17:00:46.000000000","message":"Patch Set 5:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/914713 is needed.","accounts_in_message":[],"_revision_number":5},{"id":"ed42b38593dbfd8cecf48ebd540c3dee1d379246","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-18 17:00:48.000000000","message":"Patch Set 5: Verified-1\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/914713 is needed.","accounts_in_message":[],"_revision_number":5},{"id":"52dae6bc97353ecbe9e90697e83bd4b72137f606","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-04-23 19:33:29.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"58ee032ab191c345d1669e022f0ec4390fe5e8f2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-04-23 20:46:44.000000000","message":"Patch Set 6:\n\n(7 comments)","accounts_in_message":[],"_revision_number":6},{"id":"67900cfe59730bcfdde1b898f12563c9a30346dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 20:50:19.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2432597955014363b2b6e3223e4729b9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/962fab4d94d542a2be016c0c564f7300 : SUCCESS in 8m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/baea2c8b00274dec9f0cb63320f1c897 : SUCCESS in 1h 05m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2d064da4f9d046308b2f3ba73cb7316f : SUCCESS in 59m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6d7a3995a7024a31ba5b90b6450a9417 : SUCCESS in 1h 01m 42s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dbfcab08ebd74c76a44feb6813da8dcf : SUCCESS in 20m 44s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e0a8884ecb3943948c503a19f3603a89 : SUCCESS in 20m 53s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6dc1d6783f6746c4ba73369801e00977 : SUCCESS in 21m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ee4f2f6c015f41ed8bb00cc9f4b8e212 : SUCCESS in 11m 56s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/bb6312fece9b4a2a8d9aad4affa5a2ab : SUCCESS in 7m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f04b487518cc40e4b87c7e9e70ac660f : SUCCESS in 21m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5457bc2b340c4a4e834741ee22721799 : SUCCESS in 15m 56s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/bc46b2f5d0674a31b6c453e9dec8d1e4 : SUCCESS in 23m 30s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/52fabc3d8f4940cfa993732089cb6cd2 : SUCCESS in 17m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fb6f743baa9c41ccb76e91b0a1009fcd : SUCCESS in 17m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c773b0004bac4046b35e976a5ac75768 : SUCCESS in 17m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ab9684bee25e4616856f159bc3d41821 : SUCCESS in 18m 44s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/14653f8d5494459b8778644bbe577ff0 : SUCCESS in 17m 13s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/88398fde4b5f4c99b1780e61482375d1 : SUCCESS in 15m 57s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/92f7c96e706d4165a16afa15812a9125 : SUCCESS in 17m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/06fe519f1f404dc38e5d61a816178af0 : SUCCESS in 6m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c83121fda59e43c2a49173e3363f8be8 : SUCCESS in 16m 51s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ee731f27d5d048b49608135c98282245 : SUCCESS in 8m 14s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e4f2f4062c234760a1ef3ba0d171a553 : SUCCESS in 1h 06m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/02eaf1900043458c865903a11d7ac4b9 : SUCCESS in 54m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f6fe5a60f0334f8bbcd3ec2a5a877cc8 : SUCCESS in 53m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/34aa2547b8d84dd7bec74c2b257f3d81 : SUCCESS in 19m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06491824671449499f106d71b8367914 : SUCCESS in 4m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/eaedcb43440e4721b36ac237354b7882 : SUCCESS in 26m 59s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/20bdfe13adca487e8508c24c72cbe6d8 : SUCCESS in 1h 10m 02s","accounts_in_message":[],"_revision_number":6},{"id":"609b61f3d74b51408e6538494ad67965905fd2b9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 20:51:34.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31e77d6bb2a94faa83401b267d76bbfd\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4514ff630b8d45a592311f3cb96e00d3 : SUCCESS in 1h 16m 00s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e5d7ec45c52a4b71a80f11b6cf084298 : SUCCESS in 42m 57s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/013a60017d654f95964f5b57ca28a3cc : SUCCESS in 19m 54s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/52bcc3c1082d4b14ae9085934c143aff : SUCCESS in 1h 03m 35s","accounts_in_message":[],"_revision_number":6},{"id":"987de7223fe356c135aa80f16149ca757625e2b2","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-05-29 05:26:24.000000000","message":"Patch Set 6:\n\n(5 comments)","accounts_in_message":[],"_revision_number":6},{"id":"f1c2afa91590074a40323d3e83e6e9a5866f72ee","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-05-31 20:50:23.000000000","message":"Patch Set 6:\n\n(8 comments)","accounts_in_message":[],"_revision_number":6},{"id":"517e216fe28e24edcf3cf3f3873c3bf74e6d1bea","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-05-31 20:53:07.000000000","message":"Abandoned\n\nthis isn\u0027t the right patch anymore, the work is going to happen in 918366: Parallel distirbuted task container iteration | https://review.opendev.org/c/openstack/swift/+/918366","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"8d8c539250237530a53ad5794a7f9273290bb69e","revisions":{"57269a19697f15fde20cf11193837aec8a4ca5e7":{"kind":"REWORK","_number":1,"created":"2024-04-17 00:18:17.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/26/916026/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/26/916026/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/916026/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/916026/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/916026/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/26/916026/1"}}},"commit":{"parents":[{"commit":"baaf3bd30e487d15ab6936d20e13fa6e60f10e95","subject":"expirer: new options to control task iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/baaf3bd30e487d15ab6936d20e13fa6e60f10e95"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-04-17 00:11:29.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-04-17 00:16:11.000000000","tz":-300},"subject":"wip: distributed parallel task container iteration","message":"wip: distributed parallel task container iteration\n\nChange-Id: Ia8aa11ad13c9b7f2c4b470799d8a9624f217b43e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/57269a19697f15fde20cf11193837aec8a4ca5e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/57269a19697f15fde20cf11193837aec8a4ca5e7"}]},"branch":"refs/heads/master"},"e11533b8e2543013ced963d14e4f5c6ba206ede7":{"kind":"REWORK","_number":2,"created":"2024-04-17 15:13:28.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/26/916026/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/26/916026/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/916026/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/916026/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/916026/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/26/916026/2"}}},"commit":{"parents":[{"commit":"ba64ac6c882c3ccb3709af218f0766813532b6a7","subject":"expirer: new options to control task iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ba64ac6c882c3ccb3709af218f0766813532b6a7"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-04-17 00:11:29.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-04-17 15:13:18.000000000","tz":-300},"subject":"wip: distributed parallel task container iteration","message":"wip: distributed parallel task container iteration\n\nThis change reimagines the \"cycle\" of an expirer configured with\n\"processes\" such that it will only list a 1/processes worth of\ntask_containers (at least one) each cycle instead of listing ALL tasks\nin the queue each cycle (most of which it will not handle).\n\nInstead, each process is allocated a smaller subset of discovered\ntask_containers (often only one task_container if there\u0027s 100s of\nnodes), and each process is much more likely to handle the majority of\nready tasks that it encounters when listing.  Together all nodes will\nlist all containers and each process their portion of ready tasks w/i\nthose containers.\n\nThe approach in this patch also introduces improved reliability in that a\nsingle process rotates is hash_mod/process_index based on it\u0027s\ncycle_count, so that over time - if it runs long enough - it can action\non every task object in the queue even if other nodes are offline.\n\nChange-Id: Ia8aa11ad13c9b7f2c4b470799d8a9624f217b43e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e11533b8e2543013ced963d14e4f5c6ba206ede7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e11533b8e2543013ced963d14e4f5c6ba206ede7"}]},"branch":"refs/heads/master"},"aeb6be2d3f8531b9af1c720eb8485713c4f11b1e":{"kind":"REWORK","_number":3,"created":"2024-04-17 15:17:44.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/26/916026/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/26/916026/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/916026/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/916026/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/916026/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/26/916026/3"}}},"commit":{"parents":[{"commit":"ba64ac6c882c3ccb3709af218f0766813532b6a7","subject":"expirer: new options to control task iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ba64ac6c882c3ccb3709af218f0766813532b6a7"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-04-17 00:11:29.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-04-17 15:17:27.000000000","tz":-300},"subject":"wip: distributed parallel task container iteration","message":"wip: distributed parallel task container iteration\n\nThis change reimagines the \"cycle\" of an expirer configured with\n\"processes\" such that it will only list a 1/processes worth of\ntask_containers (at least one) each cycle instead of listing ALL tasks\nin the queue each cycle (most of which it will not handle).\n\nInstead, each process is allocated a smaller subset of discovered\ntask_containers (often only one task_container if there\u0027s 100s of\nnodes), and each process is much more likely to handle the majority of\nready tasks that it encounters when listing.  Together all nodes will\nlist all containers and each process their portion of ready tasks w/i\nthose containers.\n\nThe approach in this patch also introduces improved reliability in that a\nsingle process rotates is hash_mod/process_index based on it\u0027s\ncycle_count, so that over time - if it runs long enough - it can action\non every task object in the queue even if other nodes are offline.\n\nChange-Id: Ia8aa11ad13c9b7f2c4b470799d8a9624f217b43e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aeb6be2d3f8531b9af1c720eb8485713c4f11b1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aeb6be2d3f8531b9af1c720eb8485713c4f11b1e"}]},"branch":"refs/heads/master"},"a0c9de18ba78e19bd97e7f789693d5b00eea8899":{"kind":"REWORK","_number":4,"created":"2024-04-17 19:24:37.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/26/916026/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/26/916026/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/916026/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/916026/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/916026/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/26/916026/4"}}},"commit":{"parents":[{"commit":"ba64ac6c882c3ccb3709af218f0766813532b6a7","subject":"expirer: new options to control task iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ba64ac6c882c3ccb3709af218f0766813532b6a7"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-04-17 00:11:29.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-04-17 19:21:09.000000000","tz":-300},"subject":"distributed parallel task container iteration","message":"distributed parallel task container iteration\n\nThis change reimagines the \"cycle\" of an expirer configured with\n\"processes\" such that it will only list a 1/processes worth of\ntask_containers (at least one) each cycle instead of listing ALL tasks\nin the queue each cycle (most of which it will not handle).\n\nInstead, each process is allocated a smaller subset of discovered\ntask_containers (often only one task_container if there\u0027s 100s of\nnodes), and each process is much more likely to handle the majority of\nready tasks that it encounters when listing.  Together all nodes will\nlist all containers and each process their portion of ready tasks w/i\nthose containers.\n\nThe approach in this patch also introduces improved reliability in that a\nsingle process rotates is hash_mod/process_index based on it\u0027s\ncycle_count, so that over time - if it runs long enough - it can action\non every task object in the queue even if other nodes are offline.\n\nChange-Id: Ia8aa11ad13c9b7f2c4b470799d8a9624f217b43e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a0c9de18ba78e19bd97e7f789693d5b00eea8899"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a0c9de18ba78e19bd97e7f789693d5b00eea8899"}]},"branch":"refs/heads/master"},"054d2495173d0da9de42d0e5f50348699f959331":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-04-18 16:09:02.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/26/916026/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/26/916026/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/916026/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/916026/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/916026/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/26/916026/5"}}},"commit":{"parents":[{"commit":"5d47c76b0ee5c49776f483995b726aaa96270508","subject":"expirer: new options to control task iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5d47c76b0ee5c49776f483995b726aaa96270508"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-04-17 00:11:29.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-04-18 16:09:02.000000000","tz":0},"subject":"distributed parallel task container iteration","message":"distributed parallel task container iteration\n\nThis change reimagines the \"cycle\" of an expirer configured with\n\"processes\" such that it will only list a 1/processes worth of\ntask_containers (at least one) each cycle instead of listing ALL tasks\nin the queue each cycle (most of which it will not handle).\n\nInstead, each process is allocated a smaller subset of discovered\ntask_containers (often only one task_container if there\u0027s 100s of\nnodes), and each process is much more likely to handle the majority of\nready tasks that it encounters when listing.  Together all nodes will\nlist all containers and each process their portion of ready tasks w/i\nthose containers.\n\nThe approach in this patch also introduces improved reliability in that a\nsingle process rotates is hash_mod/process_index based on it\u0027s\ncycle_count, so that over time - if it runs long enough - it can action\non every task object in the queue even if other nodes are offline.\n\nChange-Id: Ia8aa11ad13c9b7f2c4b470799d8a9624f217b43e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/054d2495173d0da9de42d0e5f50348699f959331"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/054d2495173d0da9de42d0e5f50348699f959331"}]},"branch":"refs/heads/master","description":"Rebase"},"8d8c539250237530a53ad5794a7f9273290bb69e":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-04-23 19:33:29.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/26/916026/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/26/916026/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/916026/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/916026/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/26/916026/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/26/916026/6"}}},"commit":{"parents":[{"commit":"0ace9d59ebb766fba1911eb515a45063e40780d2","subject":"expirer: new options to control task iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0ace9d59ebb766fba1911eb515a45063e40780d2"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-04-17 00:11:29.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-04-23 19:33:29.000000000","tz":0},"subject":"distributed parallel task container iteration","message":"distributed parallel task container iteration\n\nThis change reimagines the \"cycle\" of an expirer configured with\n\"processes\" such that it will only list a 1/processes worth of\ntask_containers (at least one) each cycle instead of listing ALL tasks\nin the queue each cycle (most of which it will not handle).\n\nInstead, each process is allocated a smaller subset of discovered\ntask_containers (often only one task_container if there\u0027s 100s of\nnodes), and each process is much more likely to handle the majority of\nready tasks that it encounters when listing.  Together all nodes will\nlist all containers and each process their portion of ready tasks w/i\nthose containers.\n\nThe approach in this patch also introduces improved reliability in that a\nsingle process rotates is hash_mod/process_index based on it\u0027s\ncycle_count, so that over time - if it runs long enough - it can action\non every task object in the queue even if other nodes are offline.\n\nChange-Id: Ia8aa11ad13c9b7f2c4b470799d8a9624f217b43e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8d8c539250237530a53ad5794a7f9273290bb69e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8d8c539250237530a53ad5794a7f9273290bb69e"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
