)]}'
{"id":"openstack%2Fswift~916547","triplet_id":"openstack%2Fswift~master~If0382d433f73cc0333bb4d0319fe1487b7783e4c","project":"openstack/swift","branch":"master","topic":"open-expire","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2024-10-18 21:20:43.000000000","reason":"Change was submitted"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-10-18 21:20:43.000000000","reason":"Change was submitted"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-10-18 21:20:43.000000000","reason":"Change was submitted"},"36763":{"account":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"last_update":"2024-10-18 21:20:43.000000000","reason":"Change was submitted"},"34892":{"account":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"last_update":"2024-10-18 21:20:43.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-10-18 20:10:17.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"If0382d433f73cc0333bb4d0319fe1487b7783e4c","subject":"fix x-open-expired 404 on HEAD?part-number reqs","status":"MERGED","created":"2024-04-19 22:14:04.000000000","updated":"2024-10-18 21:21:44.000000000","submitted":"2024-10-18 21:20:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":168,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"916547-open-expire","meta_rev_id":"994f872cc5dbcfff6fb0774759383dc6b28963df","_number":916547,"virtual_id_number":916547,"owner":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-18 21:20:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"value":2,"date":"2024-10-18 20:10:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"value":1,"date":"2024-10-18 20:10:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-19 23:42:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-19 23:44:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-22 17:54:30.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2024-04-25 21:32:28.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2024-05-02 22:46:08.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"REVIEWER"},{"updated":"2024-05-16 00:10:51.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2024-06-25 06:06:33.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"f9e47a9d1d57aa02cfdc98cb04b04460010e6deb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"date":"2024-04-19 22:14:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9a4bc37e09dae017b9b359f3909cae4c345aa69a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 23:42:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96921eb6d56a492b98df06c1cd9ba400\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a1744c26993f4db79fdf9659270848ac : SUCCESS in 1h 12m 37s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/047f8d9a376a432e870c6fe470ab49eb : SUCCESS in 53m 03s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3220dcb8a71341baa7b0830001fc904b : SUCCESS in 1h 26m 43s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4f1f9e0454e64849a0a4f152f771acd0 : SUCCESS in 18m 52s","accounts_in_message":[],"_revision_number":1},{"id":"19e1f676cd296ecabfb8ddc5f6c55cae08167784","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 23:44:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f3d36815b1e4d7e9e5f47464d02ec00\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cd9cb0039aa46d89fe5107d0bdac3c3 : SUCCESS in 9m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/11fbd38f9d214730aa395b7e6c72e843 : SUCCESS in 1h 22m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/849f70e9f9b04c38b37a449cccaf2986 : SUCCESS in 35m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/079e6cbe42924782abe6b5c7b1af809f : SUCCESS in 59m 34s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1c4a451c188d4fa7a0ded4621d84149d : SUCCESS in 20m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/07e85ccbf9ce42809bce13a9054188a9 : SUCCESS in 19m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/644de68bf869469f9d46f005c3487785 : SUCCESS in 19m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/deae4add1fbe48339065f32692169a0e : SUCCESS in 10m 01s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d78373fc75014128bee2a2d2c16cac3c : SUCCESS in 4m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ca310391d51041e380abed7167d93979 : SUCCESS in 16m 35s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f3b2aab7829c4e94a08d049913d9a549 : SUCCESS in 18m 31s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d4f51b3117a740d4b21c0aa01ec3906c : SUCCESS in 16m 42s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/8b905fd5820642fca56e6d288fc5d33f : SUCCESS in 16m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/de2c33f5ce284c20ac4269d31db0ea48 : SUCCESS in 17m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7141af19338d474e8bd4fc8d1821bc25 : SUCCESS in 19m 40s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/dd7822d37c4446c89d255237271c82b6 : SUCCESS in 18m 47s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d88d74fc9cf2406e9b05d10c24d4a538 : SUCCESS in 14m 11s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b74047b6d74246818d3d7d3fc0bb008e : SUCCESS in 17m 31s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ef0eef539bff45bea2ca20c567d50ea0 : SUCCESS in 19m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f9a769f5855541a08356fbf534602264 : SUCCESS in 6m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/13664bf8ea80440b90fbfa4606a1ef3d : SUCCESS in 15m 10s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c90e06f086674bdc829e6846eaa8f12b : SUCCESS in 8m 46s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/79b5041e527540af9e87fe7daa5c3881 : SUCCESS in 50m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9bd3ab193a114e56aaffe444191fefc9 : SUCCESS in 36m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4f53fcbec88a4a6ab93e4c011af38eea : SUCCESS in 51m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a5d97084e294b23adbd4f2b3c6b2bfe : SUCCESS in 21m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce4887b30e9b40c5bfa67dec4423de8b : SUCCESS in 5m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5aab9e09e60b4e5aa3817474acd47aa3 : SUCCESS in 20m 37s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1f8f4e7ebe774cd08bd5c1db2a351e4e : SUCCESS in 58m 42s","accounts_in_message":[],"_revision_number":1},{"id":"0432a95a994caa17e6826cc2dcd08de149d03ac4","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-04-22 17:54:30.000000000","message":"Patch Set 1: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":1},{"id":"7901d97a0d48a338540c399e1f88650f8a672f01","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"date":"2024-04-25 16:39:51.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"448d4d604a028ab51fdad63206acdaf17969f79f","author":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"date":"2024-04-25 16:40:06.000000000","message":"Patch Set 1:\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ff18b25da08029c014bf58ee77ddebc1cc3b9e90","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-04-25 21:32:28.000000000","message":"Patch Set 2: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"b32fe84826524d774a478c0dc6e7e02890a8ea07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-05-01 15:56:50.000000000","message":"Patch Set 3: Patch Set 2 was rebased on behalf of \u003cGERRIT_ACCOUNT_36763\u003e\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"}],"_revision_number":3},{"id":"7a16aacd0177c8c93eb036eadf3bee922d93df25","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-01 16:06:12.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f38569bc6aa41d8937a31100b510af9\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/092fd4b8ddc54e2b90d65a083daaf6d7 : FAILURE in 4m 32s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2d5c3cc655ac4ab79643eab520f7e9b7 : FAILURE in 7m 44s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/76467b612a774843b422af0114bed1dd : FAILURE in 3m 36s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/93313f42f2894cd1adec024c27cf3d86 : FAILURE in 6m 20s","accounts_in_message":[],"_revision_number":3},{"id":"767e0d6fce25f0cd9156e41cbd82d661e1a8d62b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-01 18:12:29.000000000","message":"Patch Set 3: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8a7c001dcdc249fabc4a42bfab5e97b9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6a7b74910244b0eae7f0224522d90bd : FAILURE in 4m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/7456f9a26a724fafaf2eec8a6239af8b : FAILURE in 43m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3ff5b5c9297b46ff80c3c0110c7ea843 : FAILURE in 19m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f01d017b70ba4a3a904f03b165111695 : POST_FAILURE in 19m 17s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/08816ba7c89a4dde9477f1a3734b0c10 : FAILURE in 5m 45s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b9576b2adf97442881a8a7c6e2152d15 : FAILURE in 6m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8ac3a9a46b3e40b1bbcf5911bb75f2c7 : FAILURE in 6m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3e786f013261431fb393977f20feda80 : SUCCESS in 11m 21s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/426720784046444b9220e586bb3e57ca : SUCCESS in 4m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8f6bf762d3fd40e6863e9ea51d10c3a4 : FAILURE in 4m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cd47ea00fce7456ebd49b249c4d5c789 : FAILURE in 4m 18s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/716d7f82cdbe4ef68fbad4c7cbe1b60e : FAILURE in 4m 44s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/a1fa490fa948441f9062d8a4b2143504 : FAILURE in 4m 39s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1145e325c07141668b87d2c20a6aa71c : FAILURE in 5m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f5fe5f21c81b49b1830f55f7fb3443db : FAILURE in 4m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c4bfad43f29b4442b3733496a9e07b8c : FAILURE in 5m 36s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e33eb7ff39c44bbeba82d03b21f84aa3 : FAILURE in 4m 03s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7d5f778366b141e3a1be0070b1b6fa29 : FAILURE in 3m 51s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/50c087914f934b449e54fb4168c68796 : FAILURE in 3m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fe70e014ed624276bf4509e8bf19d283 : FAILURE in 8m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ffcfd30908f54349b969569540865d8e : RETRY_LIMIT in 5m 05s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/269d2d5747f44b718104ed7feb55adb5 : RETRY_LIMIT in 5m 59s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2ae6625fd15342e287f632ecc15e3795 : TIMED_OUT in 2h 00m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a35a78b758bb4c4b9c6798e8b05eab4b : RETRY_LIMIT in 24m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9792058b39f84c9ab391809bde5fbabc : RETRY_LIMIT in 21m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ee883319fc6642159fc04fa0e20b2415 : FAILURE in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/385a635e754a403a85d20d24baa8b92a : FAILURE in 4m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/da4312c90bf14193afad2b9533d92a5b : SUCCESS in 28m 06s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c7babcc7f1394a118f70280c34b88d5c : FAILURE in 18m 20s","accounts_in_message":[],"_revision_number":3},{"id":"9463fcea038850766c8ea29cb9a333ed7c5879fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-05-02 22:45:34.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"b17f01f3459e5f19752532a8d852195eba34db97","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-05-02 22:46:08.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2fb0938bb9003f1fe713ae231c1c39244f624bde","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-05-02 22:46:22.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"4d9c61b15a155dc03cceb6f41dce44f39c84ab9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 23:55:40.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/73deaafeb2874192aa1ca35dad49eb7c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd34ccedf10245109caefe6175084f20 : SUCCESS in 8m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/0a0bb2f8c11a433aa41a04d951d3739e : SUCCESS in 52m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3705d91e3cc942f488026dcfb1edc7fa : FAILURE in 39m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f02a9b405e4d423681f71091aaad5427 : SUCCESS in 1h 04m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1730cdc98cdd49e3a9adb1e1477bdb5d : FAILURE in 28m 19s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9e6599a3b02e4493b83a8f1134c00b01 : FAILURE in 24m 26s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5340a4fbba904c7f855d6574611a70e9 : FAILURE in 24m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9ab25bd2b0204faba2728627b843bca8 : SUCCESS in 11m 34s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/db427486f7f44c6cb635572bbd51a513 : SUCCESS in 5m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/02e8a2b0acb0458d91b2cfd6230a34e3 : FAILURE in 25m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ae33405732e4422d84910d676dc0aa47 : FAILURE in 31m 23s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8f749b9cec0d4ef286e795d5e5df91e2 : FAILURE in 24m 03s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/57fd97bcc2f04438827ac9d1517de841 : FAILURE in 23m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7358c35b5227429a81ef5552cbd4ac06 : FAILURE in 27m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b8cf79fba69d4c22a1b296dd90dcef6f : FAILURE in 24m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c8c494c051354315a8b9d67e38f34f41 : FAILURE in 33m 18s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0cd61f494518466390da71f1f3ccd418 : FAILURE in 21m 06s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/18ac0828fec145f4b569ebcb582b9d5d : FAILURE in 21m 43s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4584fb27ecf542feb7c2e17c92b4028c : FAILURE in 20m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/33165500aa784889b0c5c101334f3c70 : FAILURE in 6m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f97aa69c06494ffe89ced10dc7b0157f : FAILURE in 17m 34s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/15cd69effcdf40aeb457832749924507 : FAILURE in 7m 15s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/289cc83d6eea4c0d819220e1cebb4efc : FAILURE in 1h 01m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3f52707515704f968b2bd404027283d2 : FAILURE in 26m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/02bbf68157b645c894fd4afb00e14986 : FAILURE in 39m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1c392f8b4c2c4980a5ecb0b64be669ba : FAILURE in 25m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e509219dc3b24eef9e8e0255f5985822 : FAILURE in 3m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9cd3ebce4a4647058aad422280e3d760 : SUCCESS in 18m 54s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/76a8b31f07e8434791ae829c4e21cce0 : SUCCESS in 50m 27s","accounts_in_message":[],"_revision_number":4},{"id":"dc5ca111ce7be9ef68fc16b2cb2475ad5225f573","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 23:57:43.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec83c0c34799493f8e64e1d0ffcf6d33\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d0a9e894a2db49739abcf90a8fab7553 : FAILURE in 18m 36s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3afd3dec5dc04bb399a35559aa628820 : FAILURE in 17m 40s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/6df19be7bf9f4fe3b810047f79c0c7da : FAILURE in 1h 10m 01s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ed4aebb5a0a1471cb3064182a804183a : FAILURE in 29m 23s","accounts_in_message":[],"_revision_number":4},{"id":"b2cdceb419c961da01a1713fa505dfc4509b3d8f","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-05-06 18:02:31.000000000","message":"Patch Set 4: Code-Review-1 Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"23812cca714b4c62703ea9da5f30f96bbf8f7735","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-05-14 18:06:46.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":5},{"id":"a45df6edb1ab17b0de01fa444190a3e66291d467","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 19:11:09.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7bad1da799df48e1ad5c5d4125906e80\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b8b95f68d1b446da392c1176a14138e : SUCCESS in 8m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/d7afac63ff0745deb5ccad0bca6bd3ce : SUCCESS in 54m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e25abe5ffac74e098ccc8ffd191ee823 : FAILURE in 49m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1ca375fda2964c8caeb51a72f2215e15 : SUCCESS in 1h 03m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6c84c201f210434a80896ebf03426334 : FAILURE in 23m 53s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/008d57966eae4c52ba933881f87999ff : FAILURE in 20m 41s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e9c59c99ba7543af8829a5de3b8d5a32 : FAILURE in 26m 50s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3db46407daae446181461cb0eebd5b9a : SUCCESS in 10m 11s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/86b97eb048d040dc807d8f8f21f0e363 : SUCCESS in 4m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/02b67451c7204f1fb178a83229ec015e : FAILURE in 24m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d5b4633a49ab4e5fa3a8b4948aa44f50 : FAILURE in 26m 29s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ee67446a730548c185e6e6874a1e24e1 : FAILURE in 17m 46s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/0d3ce49faf2d4048b5b89fc855fe36a3 : FAILURE in 20m 04s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0ac7ba33ace749c8b704194f38c63975 : FAILURE in 30m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9b093894be4542ef99974aabe741d3af : FAILURE in 28m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/28d442de764e47bfa65ff108e1ce360c : FAILURE in 23m 46s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a6043eac0dc44219aabb0e23ad8d8b13 : FAILURE in 28m 37s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9fef484dde164f4f8d0a02edab7f49f0 : FAILURE in 21m 38s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3242c4cafaca4433a5156f17316fccc2 : FAILURE in 21m 50s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0b40cd740b15467eacac92907131fd2e : FAILURE in 22m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6514040977cc46c0a1c3cc6088e0c73b : FAILURE in 8m 18s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/545470a17ad6478bbeebc5896095fdca : FAILURE in 19m 25s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0729d99fea834461977b218c2d67652c : FAILURE in 8m 19s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2f92548e00d9427d8c9aa0c13420d461 : FAILURE in 56m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9ab8a3b992944a0887a7f3a57affc9ca : FAILURE in 37m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a036830a846640ae8b8ddc4699e6b665 : FAILURE in 40m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/401bcc05a2f84779aca0dac5a2e64933 : FAILURE in 28m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6811c3e47d7451db211a8f498ccf8ea : FAILURE in 3m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5326014fb2e44e77b82c7d25cd3bc29d : SUCCESS in 19m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6e220cc961be4e0fa475d77aebff405e : SUCCESS in 48m 02s","accounts_in_message":[],"_revision_number":5},{"id":"01f99f1c420f08c7551463dc5475cfe7e8b3b0b2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 19:17:33.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f56043c889a44008ae1d2d24e00ea666\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1a89693192774e59a839ca95c4dd07c1 : FAILURE in 1h 09m 07s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c8631458f45d40778ece41aa7c6bcb50 : FAILURE in 58m 51s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d3ce94b8a2f64743aec8ddbe651aac65 : FAILURE in 1h 06m 39s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4dfcb552e60045da8277f628aba72af5 : FAILURE in 23m 53s","accounts_in_message":[],"_revision_number":5},{"id":"2c12b378a940c8cebc3dd4433d088104c32a4a5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-05-15 03:03:51.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"0e560690109d834fbfe448e69b520d24ad42f90c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-05-15 03:05:00.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"8f4f882db2455afa9934a2597517142a1fd9df96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-05-15 03:05:08.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"da5d62d78ace3cb1d1bea2cda3bc129cb27143c8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 04:06:15.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2cd24ca6f1245b39211a5c96be9cbe5\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6d97a84788244dde99c50a57b55d2796 : SUCCESS in 16m 43s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c69c4483ce92416ca7fcfe4ac5cc6326 : SUCCESS in 16m 15s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/693580cac22a4c63ae9cdb05616c69e1 : SUCCESS in 55m 16s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/27a19d6a524046798878dc11cae93a28 : SUCCESS in 24m 53s","accounts_in_message":[],"_revision_number":8},{"id":"99c09f67ccbca8f81da3f56566da9577403c378f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 04:20:35.000000000","message":"Patch Set 8: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3f7e3eb87e12451ab08b46373813ee90\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5fc405b96e040789faabe479709ccea : SUCCESS in 7m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/86703191c4cb4d35bba3900c3367edb9 : SUCCESS in 1h 06m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/170aa584cf1f48f3b59ecbc1c7a8af94 : SUCCESS in 58m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/98c7579ced3144018845b8822c066662 : SUCCESS in 52m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c2a3f2c1f50c4c7399d93c338567c692 : SUCCESS in 17m 34s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/281f1bae4ee449978559587496be61d5 : SUCCESS in 17m 02s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/edb46b2ee44a40988ab70d49d88cb004 : SUCCESS in 18m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3e149b99e10b48f2bb3e57cba17cad10 : SUCCESS in 10m 37s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/25c9d15cbeaa4ccebf5de964b8ff9607 : SUCCESS in 5m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9abf66bdc9f34621aad98b8b52982ecf : SUCCESS in 19m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/50ff5bd05cd84a02b8af7c1bffb017d2 : SUCCESS in 17m 34s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/7f312350117e4e0c8d921cce12b6cd07 : SUCCESS in 18m 17s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/abd5455e922f4471b43ace9c5bd1f598 : SUCCESS in 18m 03s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6757fb5253b64b438441847e6cdee01d : SUCCESS in 23m 21s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/766b05a62a564481ba48316abb35c26b : SUCCESS in 20m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0b634dd5864941f592de9339e298373d : SUCCESS in 22m 41s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a5c59546f07f413fb59ee63c1d393ef6 : SUCCESS in 20m 14s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/99b6cbcc01df4143be6340355daea663 : SUCCESS in 15m 50s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b9712c2743634478844a30acd2a82d2a : SUCCESS in 15m 37s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2c43fd6c7d8b466cac1a19aea05509da : SUCCESS in 22m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/23c23b278e2c43c499fe52008dc9b810 : SUCCESS in 7m 45s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/61f06274098f4f8f9c902a08a09bc52a : SUCCESS in 17m 40s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ec14bb32a70c46b088942f24727c1b25 : SUCCESS in 6m 48s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/dcedf4fd129c4a92b5087bde3c24eb95 : FAILURE in 1h 11m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/40b5da0dbc5740d7afa0785d203fc811 : SUCCESS in 1h 00m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9750a25838f4492b93827198f3531c4e : SUCCESS in 53m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0b4d2702f12249b48f6ac518c72fa2c2 : SUCCESS in 22m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e022605266cc41ea9ff1f89b780cad7a : FAILURE in 4m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/104cf2356b67480db4c5bc4ddd39231b : SUCCESS in 22m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8ddd1fbfb5e64480a8a85ba2e5fa1bb3 : SUCCESS in 44m 14s","accounts_in_message":[],"_revision_number":8},{"id":"2be62326a634ed6597eae62955587123667c4157","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-05-15 19:39:16.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"07cc81538374a0ba3ef246d89e180303a1a61343","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 20:29:42.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f2ea23d11124fcbbae3565e5539aaa7\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2a9adc57c32c4739995a0f79c4e795e8 : SUCCESS in 16m 27s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/133e785324b34a6abfa643008e44dd10 : SUCCESS in 17m 32s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/383c984ff2134ab4bfa7c9c606b07ff6 : SUCCESS in 48m 50s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/5bf0e8b6d7bf4e919cfeea6032fbc0b2 : SUCCESS in 25m 27s","accounts_in_message":[],"_revision_number":9},{"id":"287f392af61d97f41b93a9db2e686bb050a328de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 20:47:19.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3784b79e67de4a96b19b27b20ec68c9d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3c715d73cd34b9b9f4d29d25e2ba574 : SUCCESS in 10m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/d88627918b5441f3b43c14f84e4c7404 : SUCCESS in 1h 00m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/302ccb3948d84dce85f8bb7632d7963e : SUCCESS in 54m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6c1f14a367c44199a604aa91cc6e77bb : SUCCESS in 48m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e36964f3b5874d239fe866b460533fca : SUCCESS in 20m 23s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7cf00363126240d5944f717cc4f25023 : SUCCESS in 18m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/93b1b22c7bac431daecde15a17aae76c : SUCCESS in 17m 52s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b5f354f5b1d445de99396b9a1702412c : SUCCESS in 10m 34s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b7ea729f0e234d3696746668bad83beb : SUCCESS in 5m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/483326180f2c4b35a5d561ac6e12c73a : SUCCESS in 17m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3c6ab53a53784ff1bcfb56848d699aec : SUCCESS in 18m 52s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/7a1a7ea9bf2a417f8d268ebc02260613 : SUCCESS in 16m 11s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/cccecdd948a9428f87fe3b5756f5ccd3 : SUCCESS in 19m 35s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/fae6e32c1f9b492bb0a49e66f976a7a0 : SUCCESS in 21m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7c5f627da89e4ba7bedf1972cf3c632d : SUCCESS in 20m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4e99781f89ec4752960471a521eac4ee : SUCCESS in 22m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8811d5a0a0824621b4871c0f36958f06 : SUCCESS in 19m 10s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/72a637e7791445c4b39da8e750390624 : SUCCESS in 19m 43s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/26c34bf846e34cd9aa68dc2907413f25 : SUCCESS in 18m 04s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/37c0cb5f328e4ef7bba657d6285ac73e : SUCCESS in 16m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e3f91ea4e8974038a4fb0fcc7a0d8bd7 : SUCCESS in 5m 36s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/714ee7b6d2c9449698c3a55445606437 : SUCCESS in 17m 24s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/03b00b384f4b4508b8560c27d92c82a4 : SUCCESS in 7m 07s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c5ded9e66c134483abf404a27f5a8f03 : FAILURE in 53m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5d0403c92c1a45b5a42a68905e504ca4 : SUCCESS in 54m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8266a4bdf49846bba66f9de3ecb53f88 : SUCCESS in 1h 01m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f5df4a80f2514e2796c93f4c15999c5e : SUCCESS in 18m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fb91b704e3e4a4f9bf95b199a3997ae : SUCCESS in 4m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c4d3e2279773476686844cec1ba7c512 : SUCCESS in 19m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4ef484d284ae46eb9370dbdec34cc69e : SUCCESS in 53m 42s","accounts_in_message":[],"_revision_number":9},{"id":"0c155ef910e3bee14ad900a4def4dfa41aa19c7d","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-05-15 21:37:41.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"ba9d19ea6ec2d77076e9188419854505597c24dc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 22:20:55.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f48d620c4344115a7452bbeae2d5e1b\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1b4f81256b284e42acd4040089196e43 : SUCCESS in 35m 22s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2ab75cc7876743df96cf36066393d571 : SUCCESS in 40m 08s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f040ee84ff3a4f8fb458f668e9b2d99d : SUCCESS in 41m 39s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1941297b87d64914a7241efab4c27ef1 : SUCCESS in 23m 03s","accounts_in_message":[],"_revision_number":9},{"id":"52f39911e7739dec2854f83ce47f9f07707630ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 22:58:08.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/70f2d85db1554ce79c903dd8fe1e9eb1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea04d125fa244e8c8799f574cf30fe87 : SUCCESS in 8m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/7a31b6f0c26042089f49176d7f57f590 : SUCCESS in 47m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/708aab27364a4e2ea14619e42d640735 : SUCCESS in 38m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/72079165c8fc430aad6c4c90cfbfe030 : SUCCESS in 47m 23s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c88818f71b8e4db58bfbd101b3e53a4b : SUCCESS in 19m 53s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3ccc4a09208b42a9b6883378071f4cc2 : SUCCESS in 19m 42s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1b4f7eaa13a041afb637f4f294a9a354 : SUCCESS in 19m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d82807723fbf4ed9a48b277e7cb04927 : SUCCESS in 9m 35s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ca6f4ab7457146f682a53629b00d7708 : SUCCESS in 5m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/68cc44afe6f9400ebaceb0c05583fa5d : SUCCESS in 16m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f948e96bf1524031be5441aa10ec1bcf : SUCCESS in 20m 51s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/21d3d9560cdd44c4aea619833636a2f0 : SUCCESS in 15m 26s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/a17d848ea875467dbd16a7be6b5dbe3f : SUCCESS in 15m 56s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2cb6a539c2e74aadb7ab3d1d01c81782 : SUCCESS in 23m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8e6771ee45544e74a0c78f81879364e9 : SUCCESS in 22m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8e67f0ff878748a6a8f67c1622edaffa : SUCCESS in 18m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c31d4d2548084af28405c531a30c9273 : SUCCESS in 22m 51s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6dac8132e04b459eaf9565e0b74ff338 : SUCCESS in 16m 31s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/77b7c922ae5542bab9c6970601bee7b6 : SUCCESS in 16m 21s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/38d6b372f8cb41599c667940dcb6352f : SUCCESS in 16m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d9a32198bec04faa8d340aee480761a6 : SUCCESS in 5m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e470c3a5291143a8b61fb2aa145a8afa : SUCCESS in 14m 23s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d827a5e208084e13a810755c98484bdc : SUCCESS in 7m 07s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/bf5d7f20e03642369f79c4c0916146af : FAILURE in 1h 00m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ba044bd2324b4bc5a46712b2b5cb7cbc : SUCCESS in 53m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2bbf7373ac504355a46a972f304c0f91 : SUCCESS in 53m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9fd8baebed99462a991712d62c4db1ab : SUCCESS in 21m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03381cf9ed5b407c99519f8a12f23389 : SUCCESS in 4m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b388c761827b4e8f8cdbd7e35942bc8f : SUCCESS in 18m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b465d46715bb43739b9ea0238ce9f791 : SUCCESS in 37m 20s","accounts_in_message":[],"_revision_number":9},{"id":"126de9bcc09a42f6daa51914a10bc97fc810acb3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-05-16 00:01:19.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"fc35116ded0a0376c04a3c12395eafde86e67728","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-05-16 00:10:43.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"83011c1ac82ef33af6a0959757ef4f9a96982b6e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 00:23:37.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05349737a79d474388e67e02954eb1f9\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9d109561f11344a3a44b86b82f54ddd0 : SUCCESS in 16m 01s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dccb2b16486d46edb96c7c9889b7402a : SUCCESS in 15m 23s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/cab08a073ae849ffb32a68d41d5d6fed : SUCCESS in 21m 15s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/fffe3f62e0574a60bf9c603d51dc8d5b : SUCCESS in 17m 03s","accounts_in_message":[],"_revision_number":10},{"id":"a668b09b37ccafcecd3dbd21e8340472722e1bfe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 01:29:32.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b85e7ebea43d405aafb819fbd0d1419b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f262cc59576e4430b4d9409d40476de6 : SUCCESS in 6m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/99a17498a63a425baf1a34cb488fff5c : SUCCESS in 58m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/978c9af15eaa4277a1fc54634e3c32ac : SUCCESS in 34m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/60487e6269a64ff5bf9d90dfc3d04bd9 : SUCCESS in 1h 03m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fdc416fd430d4cfaba40dec45a43d2a9 : SUCCESS in 22m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c32e27529cd446d0830d46306dd4decb : SUCCESS in 22m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/aa20b2869774469683fec88f3ced6d29 : SUCCESS in 22m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ca54562afa1f47c4891de25cae9fdae6 : SUCCESS in 9m 34s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/332806566783459c96a8d33a6d80dc01 : SUCCESS in 4m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bea68d763f024684a41ec9e74aca4478 : SUCCESS in 23m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bffa31233fe74b1892d30ed8100e0413 : SUCCESS in 18m 23s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6a1483a76fd74f7dbe2197ce4fc49e73 : SUCCESS in 18m 49s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/aa4c507652374f239c0a87f0201f7a59 : SUCCESS in 19m 31s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/24b6661fd5b34a61a5c04a63b38dce69 : SUCCESS in 23m 34s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e16e0a4ace5c410895450e2aed1f0220 : SUCCESS in 17m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/aed2c1402a854c3db5b162c0564e06c9 : SUCCESS in 18m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8565389c881c47b68ec367b943b75226 : SUCCESS in 19m 12s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e520055d28c1482b99618e3b17f689fa : SUCCESS in 14m 44s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c7cdff8c23e4438db4d0c46d0e376674 : SUCCESS in 15m 52s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0c07efbad5834a04a5d54d9e315eaa72 : SUCCESS in 16m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2cb563189ca1465abd2fa08a75f9fbcc : SUCCESS in 6m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d2bbc2cf0ca34fa79324378a768ebae5 : SUCCESS in 12m 28s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5679101f8a5849bf9d1821351ce8d294 : SUCCESS in 6m 55s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e708bd3c95bb4f9d9e81a151db6c0ec0 : FAILURE in 46m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9693c4b31bb941809e6c6c7a634a5c73 : SUCCESS in 54m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d34de40bff7e4897b1bccad425ce7f31 : SUCCESS in 48m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4c8c6ee6b0a849e59c74955e5faf1a66 : SUCCESS in 23m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b55f44bec0049e4aa8006e038ad6158 : SUCCESS in 3m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/601b65b74c594de1974e7f90aed8c83b : SUCCESS in 17m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c54b0020413545108f7857d368ba4bef : SUCCESS in 42m 37s","accounts_in_message":[],"_revision_number":10},{"id":"9416a3b7a8d7ad5d64aa4a74ac0ddffe63f768ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-05-16 20:30:21.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"a71661fc36b0768bf696f326bc2a565f3d0b16af","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-05-16 20:31:23.000000000","message":"Patch Set 11: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"34a221e24d51a98a08f68d9537e12a144d672440","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-05-16 20:45:44.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":12},{"id":"287ff9af399d1d40d91b253049b3cdc5c3c94b45","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-16 21:30:57.000000000","message":"Patch Set 12: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":12},{"id":"0be773d991c298e47fd5c678f1f3a45ea65a4a32","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 21:36:28.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6d048fd73964299bee47ec4e4c99eee\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dea7419628654a439dea250f3dbdbb0a : SUCCESS in 15m 24s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0c9aee9f4104411fb0a0269404e871f9 : SUCCESS in 15m 37s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/064213c3b0bf41a49d2de4f46cce3807 : SUCCESS in 49m 00s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a2d98db80e6741399da0beb63f4b6c29 : SUCCESS in 24m 50s","accounts_in_message":[],"_revision_number":12},{"id":"6b382cef4f0f33500e88bb8d77a1012bf2eea196","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 21:52:06.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8262abba35f4ecbbb1f6697dafd712b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee8c9af8edcc4e349cfc751e5dfad456 : SUCCESS in 8m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/70c9290e9acf4cb7a71e43f2c42dd886 : SUCCESS in 54m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bcc3ee2ada5841b49e35e4985f30f27e : SUCCESS in 42m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/256c314b454943c09a48e033257843d7 : SUCCESS in 52m 11s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/03bab5e6fd46476b8b2990fc7fa01616 : SUCCESS in 21m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7b06a7cf95a34dd6961ade1434bf5cad : SUCCESS in 19m 50s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1a2afdbc58ed4b6ab23815dcba3383ad : SUCCESS in 19m 10s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/827e116353cc4c799e2acafaab3095ab : SUCCESS in 8m 39s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/174b70539718402f99d983b0692a83ef : SUCCESS in 5m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6b8e657bd64d4dbaa0aa6ea7f8d9fdb9 : SUCCESS in 13m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b7bec75879a548539b3f7775c73ad11d : SUCCESS in 19m 21s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/76d8cc5143a941dd8c9519c2d16071a2 : SUCCESS in 15m 26s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/516a9563745d4acd94e235a23d148fc3 : SUCCESS in 11m 48s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b8c5aad9496f4f669400ad64aa9a85dc : SUCCESS in 25m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/81406bba7b414ffeae3c3836fdb79ae0 : SUCCESS in 16m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/37855fe45ecf4b3f86d25475cc79741f : SUCCESS in 17m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f3359ad8561f42beb522f4058522bde1 : SUCCESS in 18m 02s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7529849ddb03477f9951c751a7911d0b : SUCCESS in 15m 00s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/fe23793bfc8e4f238e34114f2ba1a29c : SUCCESS in 16m 19s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f4f1e5264a1249eea65f486aed699c0a : SUCCESS in 13m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/93291c43234f41c6b14d1751355a8585 : SUCCESS in 5m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/75086972d2684af682b5605dee5288b9 : SUCCESS in 15m 36s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9fe3c5a0e5184ab6986a060947d3b4f2 : SUCCESS in 10m 00s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7f1af58a493d472aabb9ec68da9eb6e1 : SUCCESS in 1h 01m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c05f7f29ef68452e8f8b5bb8935d6842 : SUCCESS in 34m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b49273339cec4570a552d6f370feae3b : SUCCESS in 50m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/52a16ecfde0c47ccbbde8d281e1eeffa : SUCCESS in 19m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ad9c2a3552f4c1085bfddc5b8e1faa7 : SUCCESS in 3m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fdf77cd7fad14407abc89c26ee8e69ca : SUCCESS in 18m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ed9725c10fa348c09dd605b57eaf5a9e : SUCCESS in 47m 43s","accounts_in_message":[],"_revision_number":12},{"id":"e476fce699fb50600e6d3a91c4ca886f00d66754","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-05-20 21:28:58.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"4d4a51436e0deaadf7c5d09057ce55bbda9cb662","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-05-20 21:30:10.000000000","message":"Patch Set 13: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":13},{"id":"0fa6ab4a190463f07825ca6d9f29196ba5e334e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-05-20 21:30:18.000000000","message":"Patch Set 14: Patch Set 13 was rebased\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":14},{"id":"d32490d7d153bf30dfb1f53467c2a13263c02d96","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-05-20 21:30:45.000000000","message":"Patch Set 14: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"6628772ad244809e77fb034dfa52418cb9eac5f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 22:32:36.000000000","message":"Patch Set 14: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cbce1023e916441eae2c112bce904673\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/358412ad7eec40acb5a0b3b655f9d375 : SUCCESS in 8m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/c1371a686e29463ba921557f3007176e : SUCCESS in 43m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3b1aa03958664ae493c37e85dbece24f : SUCCESS in 37m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4f0958c83a594a9b8f9535344d951263 : SUCCESS in 46m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0348341d6fab4e88adb4b7fa25cad3ed : SUCCESS in 15m 56s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3da04ea751254fb3ae5cf835b758bb74 : SUCCESS in 19m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7a0a21150e9b4fc98d1acb2e88e110e9 : SUCCESS in 17m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6314c0b365ee4ec7acde7435517c913f : SUCCESS in 9m 19s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8e70b274635b4d0a9022a7b9c65fa9d8 : SUCCESS in 8m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5e4850d422a64ad883b37e6563a29620 : SUCCESS in 20m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a9ba39bb5c2b458ba157c5909e2db463 : SUCCESS in 17m 46s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f80b38253490418aa14aa7cfb76c626e : SUCCESS in 17m 41s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/01ea2f82511d41fba0c3c150ad7ad766 : SUCCESS in 16m 16s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7cbf9fb44cc34d3597ef265c251cf9e8 : SUCCESS in 20m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ff7a94ca1bb04d08916480f550a32ee7 : SUCCESS in 18m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8dc6cb61fc82498f94016fc62ecb8850 : SUCCESS in 19m 08s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e5c88696ad9b4697925e4c984159f911 : SUCCESS in 18m 15s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/80dc178d992f450cad2963a233fff3f2 : SUCCESS in 17m 00s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a10dc705e36745d9abf2beb562f64348 : SUCCESS in 16m 35s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e397fb0c47e140bcb1a7b90a72dc6840 : SUCCESS in 19m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/58ab30bbe6564cf58cab055797778b1d : SUCCESS in 6m 00s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/28a0b6f1ac0b4389987edd57e42c6442 : SUCCESS in 17m 19s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/086aed6615d1435fb29eff8d98d6b1bc : SUCCESS in 7m 40s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0b42a8b25160443dacb53a35f399844d : FAILURE in 53m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/eaaa56dc6ef24adfb11e79ebdbc5037c : SUCCESS in 38m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9a54ca6a1cd7459da6bf33d6fbcd6f14 : SUCCESS in 42m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3921fc6b72d14fa4b0a12b8ad86317bc : SUCCESS in 19m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/009367d849534bbb93d0bf990efc8b0a : FAILURE in 4m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6e3fe2d26b784164b981549b8d57e602 : SUCCESS in 21m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/64af8ec15dba482dbc1820b1a6eb5588 : SUCCESS in 45m 46s","accounts_in_message":[],"_revision_number":14},{"id":"3b9a53994c4a90822f6f800f9a6c62b5a268e093","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 23:02:45.000000000","message":"Patch Set 14:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f270bc0c5ee4148bab202b68fcbfe2e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4295a1dc24514121ae01cc0a09501be0 : SUCCESS in 48m 53s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fbff8bf524654094b04cc859ea0754fc : SUCCESS in 16m 37s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/067aae361ff34d79a3453d31dd9a19c2 : TIMED_OUT in 1h 30m 58s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/9906251546984b4cbf6ad57bfd11bd9a : SUCCESS in 20m 39s","accounts_in_message":[],"_revision_number":14},{"id":"1acb9a1f5774b229ab34b62182ef78ffb8ae571f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-05-23 01:18:36.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"44c20efeaee1b667abccb706c07cd3755845f9ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-05-23 01:18:50.000000000","message":"Patch Set 16: Patch Set 15 was rebased","accounts_in_message":[],"_revision_number":16},{"id":"3ad2a71069a120ad2fd30cf78d5c755039e6fcca","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-05-23 01:19:03.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"fcd9db88777130cc861f53edd54645bff5a9489f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 02:20:05.000000000","message":"Patch Set 16:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/699da623fb5946f09a3cc90ca94588a8\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/17fb561ad984463092fea2f296bc5944 : SUCCESS in 52m 57s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/457d0d3628a34e1692d341a35ccf66a3 : SUCCESS in 14m 30s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/9345c6d81c6d469c874e44e421bbd20e : FAILURE in 59m 32s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6b59c2ff2bd64517b45e8909d9364a58 : FAILURE in 21m 23s","accounts_in_message":[],"_revision_number":16},{"id":"c883f08bc08d70d96ac8ba877e22b7f9f287c381","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 02:26:48.000000000","message":"Patch Set 16: Verified-1\n\n(10 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5624c7cbaa284095abea73036de4d1e7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de76d4e373da44ecaa8d431dc58d4bf8 : SUCCESS in 8m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/f7aa4bb958774c3d9966a7148f6827a0 : SUCCESS in 1h 02m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4b44662e26714e4eb3b001b5f87fb5fc : SUCCESS in 49m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/203bf4c4292f4d04bdc59772c2fb9370 : SUCCESS in 49m 14s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4c56c1fa0cee4233a28d49fd805a2afd : FAILURE in 17m 22s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1e57699c5d1546f4b3a2c95c4fa9321d : FAILURE in 18m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1798a0bb47c840fe88318544fedafe5f : FAILURE in 21m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/69fb1d2cdbec41869c0fab0a0b4fec43 : SUCCESS in 11m 33s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2b463b0358bb47cc94d4031daf9fc1b8 : SUCCESS in 5m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/42ec06a8a4f44844a47300d26856911f : SUCCESS in 21m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f97bd9a2e91c491bbc64c0236d76f767 : SUCCESS in 20m 46s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9d058ce16d1646a5b1a6a04bcdb567d3 : SUCCESS in 17m 03s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/2266fb297c674c1d93b4c8d2c572a2fd : SUCCESS in 18m 34s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/653eab22efa1420cb46e31eda4fc2fac : SUCCESS in 21m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0671999aaf4d46278e399a0e4e9679e8 : SUCCESS in 17m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0f18a0ba21e14411b40959ae36125fe6 : SUCCESS in 17m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e912ce81f17b40468ee074cfa3602428 : SUCCESS in 22m 12s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/04563d7c0e184c19b03a0ff9a23bbecf : SUCCESS in 21m 57s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/da7324593d8f4d0290300e0309c45450 : SUCCESS in 14m 45s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/dd55680168f64637ad70cb9d69bbc569 : SUCCESS in 16m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/35e8128ddf97483e93361e407620d240 : SUCCESS in 5m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0d666bb28bd34fdfa3d88011375d449f : SUCCESS in 13m 48s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b0ce9d4469ce4eafb01d9357a41f11ae : SUCCESS in 7m 24s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2c054a3e1fa54a078f145fd1ecadb866 : SUCCESS in 56m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/920015e4a5a44f169ce2d37057f34f90 : SUCCESS in 35m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/121f7077df6a4ca1b2cb1ced99e2bd0e : SUCCESS in 36m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f29e2f992f4d464c9c9ce120251122dc : SUCCESS in 13m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5a52ad606e94c77bf1a8cefafb38a9b : FAILURE in 4m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f53d702e3e2d4334bba20eb13f42cf97 : SUCCESS in 18m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f4d4de300b884b9fb2da722304997eb7 : SUCCESS in 40m 16s","accounts_in_message":[],"_revision_number":16},{"id":"d637115c3a63acc2818a2ac9d53f3d1f88aa1f7d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-05-23 03:37:13.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"57ce33e7d16bbdda75616135d1923878dfa72209","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 04:37:55.000000000","message":"Patch Set 17: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/28f85bf02431406788dc175ee46bf697\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/492786435b144a07bb4032d59204a973 : SUCCESS in 10m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/89052442a5c3442aa1d5618b8436f4da : SUCCESS in 55m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bad70e81fff1492293c22db5423297e3 : SUCCESS in 46m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5d0fb12b1d47439bb601133dfd038b9c : SUCCESS in 58m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/51e110b1cbfd4d81a916baa60bdd0ae6 : FAILURE in 21m 42s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/be6f046dbdb742c3b43a671514d107aa : FAILURE in 21m 47s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/65ff0e9849e94ff19a6711e986fb5c3c : FAILURE in 20m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8bff346fd0314d11bf97841ade0d7a52 : SUCCESS in 9m 25s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/12bfd3ddb8904ef2a0e8a691fca8e091 : SUCCESS in 6m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a3a500d9966141d78ed285101145222b : SUCCESS in 19m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7976ea54a95740ebb8799a4d5a9094a6 : SUCCESS in 17m 37s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/24e9d3e935484fec9cf4f06f47a2cd72 : SUCCESS in 18m 45s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/8f0158a9445d408d8cb175142ec00e56 : SUCCESS in 20m 47s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/16bf75d20aae4f678a806d6501938325 : SUCCESS in 19m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ed54165daf9640d48dc4633d1ce1a78f : SUCCESS in 19m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ba32d814ef304a018b526a0ef1f0aba5 : SUCCESS in 18m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0f62016dd4e44b1c89302f4a5aa10db9 : SUCCESS in 18m 29s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4a663a15e87a44d3a026f676e673137f : SUCCESS in 16m 24s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1c8e5cb30ff045279f6b9064abbb4194 : SUCCESS in 15m 41s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9f466e137bcc4f8fb29d7fece01fbc22 : SUCCESS in 19m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/69d857f36b2249bba103cdbbbb39e659 : SUCCESS in 7m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b55a81dc07024122a97d40d0d5e223ca : SUCCESS in 19m 18s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f0d481edae544e94a8229d99215a4017 : SUCCESS in 8m 51s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/301d2ce972a944b88bf6ddd3f4f0126a : SUCCESS in 52m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/acaf4e0b106949968804ccc8b80f12f7 : SUCCESS in 41m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d5aeede9465341859c9c1268fda24c97 : SUCCESS in 37m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8cae0cfd085840dfa5906eacf89021d3 : SUCCESS in 18m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea23715607ea4694b8ef86e8add982c0 : FAILURE in 3m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/994dae0084a14ccfafe9963efa2dff9f : SUCCESS in 21m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ae5fe489cb194f74ae5339299da14a7e : SUCCESS in 42m 53s","accounts_in_message":[],"_revision_number":17},{"id":"15c506f19e7d435e6ccb33e1953e93b24c763f38","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 05:09:59.000000000","message":"Patch Set 17:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3df42f1905044102a40c166ef3d6a66b\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b00e1634483b46528244c1d715058798 : SUCCESS in 18m 29s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fc3a9b10cdb841edac9f17a961e75dd8 : SUCCESS in 45m 24s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a6477a23a1294eaba5968b6fdafea732 : FAILURE in 21m 21s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8473a803318640348fb24e82f1ed3c9c : TIMED_OUT in 1h 31m 01s","accounts_in_message":[],"_revision_number":17},{"id":"8a1962e3bd8acd80ca8408eb95ea155f515a1250","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-05-23 16:08:39.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":18},{"id":"556d52868610940b3efa1322b7ce549d5dbf5592","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 17:05:20.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66647f3efea7414980423a581467ada9\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2b3f9be083f244869cf8c0f6d0408695 : SUCCESS in 22m 00s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/66ec2efbd2c2440e95308647ff1a4753 : SUCCESS in 23m 00s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a9e7044a01dc4df9a223145ae9b6db2d : SUCCESS in 55m 05s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8d0f96da18a54893981ae57528473bfb : SUCCESS in 26m 54s","accounts_in_message":[],"_revision_number":18},{"id":"64f92e358e03bb309c782ed880d8a41ef88b2817","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 17:16:32.000000000","message":"Patch Set 18: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/01b44ca59ab7406e9384befb967b27c0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e954099c34ec4096bf2a6948b3ad2e32 : SUCCESS in 8m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/fe613340d52f4b5bbcec62ecd7f41d6c : SUCCESS in 1h 02m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b78f65a0e1e842528867a8fe6426686f : SUCCESS in 49m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c112280a70364e8db83d6a4ad6c0b2b6 : SUCCESS in 47m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5f47aca7396845849739d7b1a8791a70 : FAILURE in 24m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a6e2a8d6744741e98b9072d03cd76862 : FAILURE in 21m 13s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2028678fd5d74e36b1d915d11d669bca : FAILURE in 22m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ac23b53e80ab4ebfa714f78cb267b009 : SUCCESS in 10m 52s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/72c9452a3139407abb9146fc76b2b6d1 : SUCCESS in 5m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ae6ef64ab10845e1a97335b13d2b99f3 : SUCCESS in 21m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ef2af4d42adf4e1a80ae7e4eb241ddd2 : SUCCESS in 21m 13s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3631d4e2913a4382a2877f52df65e1b1 : SUCCESS in 18m 51s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/a030d18f16914e9abe7c074b90d8c5e5 : SUCCESS in 18m 12s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7d1ac2c9bd4145f6b05dabc9ceb50edf : SUCCESS in 21m 20s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e7c3f564d52d411bb9a3ac03d37539a2 : SUCCESS in 19m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/494dd2ff408445bd84f6b5b36f654357 : SUCCESS in 20m 26s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/06941e142bcb476db2c26d2b17d41025 : SUCCESS in 21m 15s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/bea60314d39c458fa0b0bd22eae547fe : SUCCESS in 14m 41s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f290c87603404941a9762c9c2a9c6755 : SUCCESS in 23m 36s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9182daef256c461e88750f04c74a536b : SUCCESS in 18m 15s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cacfdc3c2ac3475c9d4ec37fab0517ca : SUCCESS in 6m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6ba4915fccc84c1c870a645e41ae18d8 : SUCCESS in 16m 28s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3b74ec0950764395adaa9735804a86f8 : SUCCESS in 7m 51s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/42df40f244ef43ae8ec5e0614754550a : SUCCESS in 59m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a2e64ffe762d4827986364b1e4110008 : SUCCESS in 1h 01m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8fd45d9cd1b04811b71e34383083845e : SUCCESS in 56m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a25fe0beea37413592df1559d14b568a : SUCCESS in 20m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96b176e62e55473eb2c03885819b181a : FAILURE in 4m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c4fc452a8c914b779082357fbd451b3d : SUCCESS in 22m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0b14a1aa2c8b41c0b424dd9d46d8694b : SUCCESS in 53m 44s","accounts_in_message":[],"_revision_number":18},{"id":"52fde4bfb6e6be669e8d7c84db11ace9aa701d0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-05-23 19:15:18.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":19},{"id":"a9182815efb02c02108637059f5f1af5d96f8d43","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-05-23 19:23:19.000000000","message":"Patch Set 19: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"56998cd87c57dec877790f305137094d3cf6ef14","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 20:23:07.000000000","message":"Patch Set 19:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e7b7c5acb974961835b503123874adb\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/15eaf9549bae41b5ae51b59b0a753457 : SUCCESS in 1h 06m 05s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b521e122eff442e5b5d3649b0dce7416 : SUCCESS in 15m 58s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/becedb3b0161449083f82492003b8ae7 : FAILURE in 45m 45s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d188d038601b4b1ea0ef7f689b59fcd1 : SUCCESS in 22m 37s","accounts_in_message":[],"_revision_number":19},{"id":"8809a19cf98dc33f23a0c636424ffaf0cde69341","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-23 20:39:26.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fccbc9a12d864591b5726a2e755143bd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0e80f7f807f4cfb92c25a23dda21425 : SUCCESS in 8m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/9cd419741ce84694bf4b66402d000cef : SUCCESS in 59m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ada4d2d1c99d466185f47e4ebd6f5425 : SUCCESS in 50m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bb149657e0d34bcb9d56dd3d1f0de10d : SUCCESS in 1h 02m 58s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/307b089928f4404994f6243c48484bc7 : FAILURE in 17m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/668be260b2104c5680691642f9fe7717 : FAILURE in 19m 26s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0b1a2c3f0bf84898ac643a8e3854cbaf : FAILURE in 22m 10s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/290d03d8bc014f3087ebbef74974d097 : SUCCESS in 10m 34s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0c40e4f1098a46d281e38d151f048835 : SUCCESS in 5m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f26311270f9640cfae282e8513ba1579 : SUCCESS in 18m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/dc21f4e92be440e6866855749fab01bf : SUCCESS in 21m 04s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9986eb08e3f64b7582a01f1a5f5c7688 : SUCCESS in 18m 33s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/4fde1cfec45546ba93cdb3f9cd90aacc : SUCCESS in 13m 22s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/68bce5523cb144418eb8aa609601e4ea : SUCCESS in 19m 27s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/dac34e5caf784526b494927d94a89aad : SUCCESS in 18m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/371597ccc6774e5c85e241a9fd95e099 : FAILURE in 19m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6c8222c1a751455aac54afc4558f67bd : FAILURE in 18m 24s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/feff91379e5f490ba74be4906c7162f6 : SUCCESS in 16m 39s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/46562147954e40cd98dabf38f5026976 : SUCCESS in 16m 37s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8d55d05c9dd4479babf7fcab18a57e11 : SUCCESS in 17m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3c098a38a90f4677a1bbd4da40d70813 : SUCCESS in 6m 52s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/194a8ead94a844949ceb68d00fc3c258 : SUCCESS in 15m 57s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/75c463bf939e4774ab81b1b5320a52f3 : SUCCESS in 6m 32s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1d6f7400a45149ce93c21359108d75f2 : SUCCESS in 58m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5208496e9c65489e8d758281397a1f82 : SUCCESS in 38m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/88601b2b6a8749f18012e112281e3614 : SUCCESS in 1h 02m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4af39a9a858a40189e690fca9910c183 : SUCCESS in 21m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64e2b758a95d42a0a99f93dd198464e0 : SUCCESS in 4m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b1ff980789d64811a0861edae024b1ad : SUCCESS in 19m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/df2a6c4dcf8f45b5813ef70ea1a13ddb : SUCCESS in 38m 52s","accounts_in_message":[],"_revision_number":19},{"id":"1aebcdc694394c4f24efbfb5a6a93a2da2f1b70e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-03 16:57:34.000000000","message":"Patch Set 19: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":19},{"id":"e57f63651a5381145caf051778239dd11d82860b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-06-04 20:25:47.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":20},{"id":"86844bd3e5a68d09026a64058b14561d991c3f3d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 20:48:19.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17ff4d358d0440a6938ccfa2741593be\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fd9f9e73e26b4cd3b8244b7ffd4ec5a1 : SUCCESS in 15m 20s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7159958e7e734babbe3cc0ebc881028d : SUCCESS in 17m 36s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f27f6c08e2be459688265889d3fc115d : SUCCESS in 21m 26s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/150c69d6b55945db91d8152958e90be7 : SUCCESS in 17m 58s","accounts_in_message":[],"_revision_number":20},{"id":"f439873a147478321ee45a83736a432a5090e70b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 21:46:58.000000000","message":"Patch Set 20: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/164262d3bb6a46f0aff7a279e6fe98cb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/104ccfde51cb42e189ee561aedf402fa : SUCCESS in 8m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/5e8eff353bb84c5b869b288855c9745c : SUCCESS in 1h 14m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6118fb1cc0e8485495f6de2994d76d0f : SUCCESS in 54m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cec646c3680241f791e63a60f7c716b8 : FAILURE in 1h 03m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0b67d8d2ac08427eb77eeaa145a98059 : FAILURE in 19m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e605fa788ece4d55b5e743d0484645aa : FAILURE in 22m 06s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/17ef6fb3337a43028f7612d9755c9a9d : FAILURE in 24m 03s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/20c839a3105e4d00b759b5863b87f615 : SUCCESS in 11m 37s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/91cf3c11b05e4d888ef3c8ef0830c5bc : SUCCESS in 5m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/38e10b9938494fd2966a872855402f3b : SUCCESS in 21m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/13e9ac024bfe4ee89d7ac8132bb6a4fd : SUCCESS in 17m 46s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/bace497438284bdcabbdc980a8286a5c : SUCCESS in 18m 05s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/7d8add93dc45423eb39b1e031c8fc6d9 : SUCCESS in 19m 58s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1aaf4af839704827837218f1d0fc61c8 : SUCCESS in 26m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5e6959721f0e40c3adcfc88ae0d23e3e : SUCCESS in 20m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/54858f1883b5467691bc042e3ec77c5c : SUCCESS in 22m 09s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0df15a3049c04adb859dc7efcb075308 : SUCCESS in 18m 53s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/53de0c3638104ad6abc4dedaed6478c9 : SUCCESS in 14m 45s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b1b5a13923f74bc9b49b6f5064a28aa0 : SUCCESS in 18m 58s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/08da981dae4246e3845d43ae9b3a7ef1 : SUCCESS in 21m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c6d6d884872646e59021d80d4a393a77 : SUCCESS in 6m 15s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b0919eccaa15422a94f9c7fa1224abfa : RETRY_LIMIT in 2m 01s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/675a8c07a20641a49a20621db5485854 : RETRY_LIMIT in 1m 58s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f8a2910daf1a4b5b8f043e50db43a85f : SUCCESS in 1h 01m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ad924f41c65940b3a2174b9963ca99c3 : SUCCESS in 54m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/52322a0d28ab42ec95c421c1e5eb18b7 : SUCCESS in 57m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a9d4a44c0991489d994bb7c54537368d : SUCCESS in 20m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/490b52289fc544a09432114ba11c34b9 : FAILURE in 4m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c7268f5f2f2546a29273ccec93d0b48a : RETRY_LIMIT in 2m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6912b00ca1894652b4e7879e4abba0b0 : SUCCESS in 44m 20s","accounts_in_message":[],"_revision_number":20},{"id":"bb9d1bba787527bc1130ebe16271adada54e71d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-06-04 22:28:57.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":21},{"id":"9bb10b59a836b2abd382cfea53a2cf8f712cd7c9","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-06-04 22:29:20.000000000","message":"Patch Set 21:\n\n(4 comments)","accounts_in_message":[],"_revision_number":21},{"id":"7dff8a4975d353cd8f30110c0344f2d88b8e1237","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 23:13:47.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76c9287c60304455be1afd50d85c1110\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c5cdcc1be2b34d06b2ce8e04947da2b2 : SUCCESS in 16m 22s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/738322a94fb9436a94914cfb5bf92bfb : SUCCESS in 14m 50s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d5d15bc76f584c4e9380c67343ffda19 : SUCCESS in 43m 05s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/369e1257c448489b8a311da8d59080d6 : SUCCESS in 24m 30s","accounts_in_message":[],"_revision_number":21},{"id":"c2a3cf35b1fa91c392970da1255f5c28887954a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 23:35:49.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e96d930118884dc4b5af8f35ac90850e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9a1b9be6a344f698386518aaaf4bc5b : SUCCESS in 9m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/9ac808a2fe354cd68177ebff2fd08070 : SUCCESS in 1h 01m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d3c08b0a13294e1b913a627187a610bd : SUCCESS in 55m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f21206a55b034221a2a79fde0aeef6c0 : SUCCESS in 48m 50s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/483ac35fc1bb4ca9bfbd2304c2a6f791 : FAILURE in 17m 34s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4d5f280e42444ac4b11dd91ee4aa6950 : FAILURE in 18m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ecd82ce7f0d94c718945455165078504 : FAILURE in 19m 43s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/240b56c74f25467084e3abb26bc2ea59 : SUCCESS in 13m 24s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/70c3492cbd984c6b89cc23251da6692a : SUCCESS in 5m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ec4182214cc54e98a257fb0b68ff49b3 : SUCCESS in 20m 34s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/91e763c5fdbc466c8b9e8d6a0b88eb77 : SUCCESS in 16m 50s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0659738be3d24d5c82698a5c6b9dd9f6 : SUCCESS in 19m 21s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/ad8bebf40e934324ac562f1c99b37b69 : SUCCESS in 18m 19s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/51ab9d4d8c984d2b9ef6e170f82f94a6 : SUCCESS in 25m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/34ce10dfa4334fe2aee647a4325c79af : SUCCESS in 18m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d1b6aa16a679466490430afc0b9f7f0f : SUCCESS in 18m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8c8f25cd78094a77be12a92cd3d7807c : SUCCESS in 17m 21s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b8dc1ea9f0c142d0a0c0ff14cee98c2f : SUCCESS in 15m 06s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/891a0ac61c9b4246bf1960463e0826d5 : SUCCESS in 15m 32s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/07ab0f7551f947d49b64957a7afed11b : SUCCESS in 14m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cb69d4f5c3b84ee9a9e139c69a273cea : SUCCESS in 6m 17s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/20cb90ada69d45379d9bec40a81844f3 : FAILURE in 15m 20s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b2b18f58329c4adea04506f000b9788d : SUCCESS in 5m 57s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a37eb3283fca42249127130583374a2a : FAILURE in 52m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/14a74e47968c49a3a890c8ef56788fa4 : SUCCESS in 38m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4843431be5294cab9a2446893df7ece8 : SUCCESS in 51m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/12b7eb1338244466b6ed25326eeb0fa2 : SUCCESS in 19m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53387de4413a44888e08aa7dfcd01f29 : SUCCESS in 3m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/049b86cfbefc4fee89dcfc075ac030a2 : SUCCESS in 16m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1aa910d02ffb4428bea914121daa2eba : SUCCESS in 56m 31s","accounts_in_message":[],"_revision_number":21},{"id":"eee2baaed24fc847cbcb212494189503c4b309d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-06-05 00:04:45.000000000","message":"Uploaded patch set 22.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":22},{"id":"3e767776bfe1beddbef7a78f1afcc83e6ad3d0e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 01:22:34.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8fa8799665fc488fa78e2b3151f65562\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ab0a6c71e5e47b9ad59b07d550a49de : SUCCESS in 9m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/b54015d6025d4b2b96a76ca50ca4daf4 : SUCCESS in 1h 04m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a596b934ae2947d2bf3cfd5c7a795f3e : SUCCESS in 56m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b9b0b2789f994202b1d9ae2963802824 : FAILURE in 1h 05m 19s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0d13909377964837b07b90c62a43785b : FAILURE in 19m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5566b4a0bdf0426ea2b73a33a6472e0a : FAILURE in 19m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7316b064620a4c53a0108de209b43fbb : FAILURE in 17m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/11d2d41bb84e429aa32bf99073ea7701 : SUCCESS in 10m 52s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/98b1ff917ca7482fa776dc6656f98178 : SUCCESS in 5m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c6978ef15fd340ada288b97643d27255 : SUCCESS in 20m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c07bd13ea99049ef88d62cc234bc962d : SUCCESS in 18m 09s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6f2b545f2cc54d569098da71ecb27b40 : SUCCESS in 16m 43s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/bd231726f6864384b974d388c504c8cb : SUCCESS in 27m 32s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8a2b013492244cd0a3b4f189150f9388 : SUCCESS in 21m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/955b64a346c14fdf95bd468aace68a61 : SUCCESS in 17m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4bd9507ff8c3480d817bc102a3092523 : SUCCESS in 17m 08s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/83b276cde7b84afca74cf4c222707ee3 : SUCCESS in 17m 07s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8c9a14802eba48d0a037d1786fbf573f : FAILURE in 13m 57s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/720e74edd1a744b89300f9a0748e20c0 : FAILURE in 18m 36s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/bef1cfbbf64142dda529898e50da2b94 : FAILURE in 16m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6b8ae8f1143c45918bbdfe8074fd1501 : SUCCESS in 7m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e1c948f9104b4862a46717a1d14e3465 : FAILURE in 14m 07s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2850264e0f8b4f76b44502278c4400a5 : SUCCESS in 6m 12s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3627cf37313b4fceb1aa2820d15e2415 : SUCCESS in 53m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b2f388a1868d46c291c1469bc4a063a1 : SUCCESS in 47m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/56c3ba3580304b9e8887d7afaecbaba6 : SUCCESS in 39m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6e5784a85eb943eabca44f80f0b92f82 : SUCCESS in 19m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3247124ceca4d2993ce2ca734297f98 : SUCCESS in 4m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/54bde2ffe2a74dc6be5cea76a89985e2 : SUCCESS in 18m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9db5b2f7620f4bc4bd41b3526437c279 : SUCCESS in 56m 47s","accounts_in_message":[],"_revision_number":22},{"id":"03ebccec87d6046e92d4e1de17f6fa9328348562","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 01:23:50.000000000","message":"Patch Set 22:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ba1a521b4c04b3387e4e2fdf310f6de\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e1981eae10dc44ca904e16fc46d9a8ec : SUCCESS in 14m 44s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/29b34660df204dfdb0092b9315dd1cc9 : SUCCESS in 17m 09s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4b0adad6aed84c36b46edbf73b45f28b : FAILURE in 23m 52s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d72dfc43c0cd40edb6b3c1e28819475c : FAILURE in 1h 17m 07s","accounts_in_message":[],"_revision_number":22},{"id":"fce00972e45f09731399894532e9ab583493f58d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-06-05 15:53:42.000000000","message":"Uploaded patch set 23.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":23},{"id":"a6e33d0452e9be100d28fbd295554e31f0742b18","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 16:16:49.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f94b77c418584001a53c164510b01fe1\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c7753c2a1aa44cb9b605dfad1ffcc9d5 : SUCCESS in 17m 23s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f8b6e94ef99440229d655bdf14f5214b : SUCCESS in 15m 26s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3d5df64bc38e4c24ae6c64bce46c1423 : SUCCESS in 19m 41s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/0eaa317596984836ad5bdb568c57b8cb : SUCCESS in 21m 52s","accounts_in_message":[],"_revision_number":23},{"id":"a5befa507c8a94b05a564560aa93397f40bc7c70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-05 17:13:57.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0015a92d643b4b0cbbcf6243b000063f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84c187df8c894ca289511eb948b53093 : SUCCESS in 8m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/e93697e0c2884b5584bcce859c07119f : SUCCESS in 42m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f59c5babe3a34b509e4f0b1eeec6d451 : SUCCESS in 49m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cf483a8d3ecb4a28aa04b778e89ec972 : SUCCESS in 1h 07m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d4dd671b10ce4c1ea639f7b4d2990a20 : SUCCESS in 19m 14s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5b5b63928cd644f1a391e987569844b0 : SUCCESS in 16m 50s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0348bf4fe5fc4a949da11da4d862736e : SUCCESS in 21m 04s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/83fb6ef34dc34fc19b1facb066375495 : SUCCESS in 11m 52s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5e6eb68c1bf34bc5b7e534d1901c605f : SUCCESS in 5m 15s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/649c0f3df76d4e59994e66cc8e921cf9 : SUCCESS in 17m 40s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/43ee548003bd4f97bbcb2b010bf101f8 : SUCCESS in 20m 01s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5f47ef3a1d2b4d9491aa54035d9c96dd : SUCCESS in 21m 54s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/360f77d0b7e145c18566a5c74621aea3 : SUCCESS in 18m 23s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5fa2a1afe8164b41ac146595fa422aad : SUCCESS in 19m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/44107b77bb474ac8886fb4dd5c627b75 : SUCCESS in 16m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f1bffc2ff94a41f4b8bb46cde6500170 : SUCCESS in 19m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9b155b45e87a4d8ba4de0129d10513cf : SUCCESS in 19m 35s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/be73b3e976b24609a072f33df22f8d68 : SUCCESS in 15m 55s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b2b4822853e44a81afe43bceaa4e5c94 : SUCCESS in 14m 53s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7542590d873e4c6095eba182572d8245 : SUCCESS in 17m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d4d88c4f7e82420599f690973171e934 : SUCCESS in 6m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6167c9a7477c4a1184a5f304b29be900 : FAILURE in 15m 33s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/04e82082dbcf448897097d234e937fff : SUCCESS in 7m 08s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ea14df298b7f462d8aca40a13e3bfb8b : SUCCESS in 1h 00m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/249830ce5e6e4ba48a37ce94d7e2bdfc : SUCCESS in 47m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a45f366d2bfd477a873389ba19a42ff9 : SUCCESS in 45m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bb8ba3741894451a9e8c2ae08aa3e846 : SUCCESS in 16m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94ef13d2f5354f1f97e35e806f0e2cf6 : SUCCESS in 4m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ba8936594d1642118ad002c479100a93 : SUCCESS in 18m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5fac66b0c710407b9a4591589eae93d7 : SUCCESS in 1h 01m 12s","accounts_in_message":[],"_revision_number":23},{"id":"70c1b4a9d6662ac7a0a6f11b2c620403545f1f0d","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-06-05 21:31:53.000000000","message":"Patch Set 23: Code-Review+1","accounts_in_message":[],"_revision_number":23},{"id":"31655b71174f641aa7ac1336faf540b719e8622f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-06-18 15:12:46.000000000","message":"Patch Set 24: Patch Set 23 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":24},{"id":"47266e3f167bdddd7989a079b17fb459c72644e4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-18 16:10:05.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b3a5126f9294a8b952c815a3a8e338c\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/949ceee3affb4a66bc98fa673d9ad110 : SUCCESS in 16m 58s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f2eba0a5ba304e068be26a9115a9ef87 : SUCCESS in 15m 42s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3a282b9b921843329bf1eac37edc3c89 : SUCCESS in 20m 10s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d778c258722f45088178c03e828220b8 : SUCCESS in 55m 27s","accounts_in_message":[],"_revision_number":24},{"id":"3bb9256143711f7849bf12ef17048029518ee31f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-18 16:34:04.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39f36e4fb3e84cc698cc253452d02f73\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26c03409143a4fe382b38cc76c8fc096 : SUCCESS in 9m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/5b8c8fdb6c624786b1d855cbd2501f8c : SUCCESS in 1h 09m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0e8fc3d0a1f7451fb631e50a84348534 : SUCCESS in 59m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d3b8e716ce5f4e84941408e5af170566 : SUCCESS in 1h 06m 55s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c8322f1ba9e843f88623bad2cbe2cbf4 : SUCCESS in 18m 57s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0a18274c0703497bbd81dae023507763 : SUCCESS in 21m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0be986a1221a4859a17c2224ae452a42 : SUCCESS in 19m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b171b7ba5ab34739bcc83a43dd8f8479 : SUCCESS in 11m 56s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3711618b81154129b87b1f675a57c756 : SUCCESS in 5m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9984f0afd0ec43b39e0a998572e8c63f : SUCCESS in 20m 21s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/da7ae2c23e9b4de682cf08d449b6ffb9 : SUCCESS in 19m 12s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5af9e6492f524406b2cf1fb75a42ea30 : SUCCESS in 18m 34s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/de0f5e57a91e474bacb99d32902ecab2 : SUCCESS in 18m 50s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/630df31c893644c7aa2923b1e5ca4103 : SUCCESS in 22m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/368ae2341dc2496db93e5388eec56880 : SUCCESS in 16m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8bc0df1eee614cc7bc6880f81433c4c9 : SUCCESS in 19m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f511970f3a914fa2938c24f3a686f216 : SUCCESS in 20m 27s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d6caf18d890048b785b12cc38bc34ebb : SUCCESS in 16m 16s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/598d6b0a93754cfea032f53a30270c24 : SUCCESS in 16m 43s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/64fc375bccf848f4ae4f3515ee167bb7 : SUCCESS in 19m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/62cbaadfd51d40d6ad1b3efc4f54751a : SUCCESS in 5m 38s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/591f8a054b1c49539ad7cc998d6d45e4 : FAILURE in 15m 13s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8cfc4298fc584020b99320bbe5007717 : SUCCESS in 7m 05s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4b6f30d24e9547bc86b93fe1a24f3f89 : SUCCESS in 1h 01m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c95fb8ed46a845eb8d748c694eda6778 : SUCCESS in 48m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7574f79d05a24347a35c6f4fde0b5749 : SUCCESS in 58m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2c2a61ffa3f04916958c6771c72dac04 : SUCCESS in 19m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eac9e853b26b4d96989d58e6cb48aebb : SUCCESS in 4m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/37627076aba54856904d6568ce74202a : SUCCESS in 24m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e2baccbda4e94dc4b274015d6cdd0424 : SUCCESS in 1h 02m 55s","accounts_in_message":[],"_revision_number":24},{"id":"1512471ea205b47c380d903439dd3c0705763c57","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-06-24 16:34:00.000000000","message":"Patch Set 25: Patch Set 24 was rebased\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":25},{"id":"830486eb34d2dfb86af5c48ee612a6415fb5eb30","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-24 17:34:53.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/505cdcd1917749c1b157665f196fef95\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f8971ef7569c4d80b5cdb2a9de4c2741 : SUCCESS in 17m 27s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/104cef2107c74961bd7c862d57afca46 : SUCCESS in 21m 14s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3e1f48a0244840b48668580ba2329e62 : SUCCESS in 23m 42s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/30d9c8e6c1234e72bbddc9c3678dfdc8 : SUCCESS in 59m 02s","accounts_in_message":[],"_revision_number":25},{"id":"d1b7fdbcba4dd7913d19adca81b5c2c4f88599ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-24 17:44:12.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5d033c2b9444ede8ee13929502278ca\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c20e68d83e3748e7a1a9c4e7177fd196 : SUCCESS in 7m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/c6b642efb0d841e8a5c9a4b6cc8933f1 : SUCCESS in 55m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e06c6617dcfc4cdead45ac255da6da3a : SUCCESS in 44m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/377ec72a6cdb414cb9e505de2088323d : SUCCESS in 1h 07m 01s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d817b99ca62f4e719754c3df4e189c65 : SUCCESS in 16m 53s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a420d5c3f5b245cfbd03e0557e1274ad : SUCCESS in 15m 29s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d90ee2b2adb0456494ccfd7ad6b57ef5 : SUCCESS in 20m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3c233c2dfadb4ca8b75a1f603ae173e8 : SUCCESS in 8m 52s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7c8ba61ba2894ac283ad17abe876cdd1 : SUCCESS in 4m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9ae7a3f01d064a4fbf640d845a9fc2ae : SUCCESS in 16m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/902f973134ec421abef8fb9eae3fcb0f : SUCCESS in 13m 00s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0ac79096e4174ca1b2d66f6278431238 : SUCCESS in 18m 41s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/640279c3277843c3b195cd9ec4590e78 : SUCCESS in 12m 32s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3f68f832bbcf405f9954e0c5a7b3a02a : SUCCESS in 22m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/99cde8d5f3ea425ead06d6700bc4111b : SUCCESS in 16m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d8fda13dec304f27a966c21718cc37ce : SUCCESS in 17m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d93a7237a7c44677bf852322cc120e4c : SUCCESS in 19m 07s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/912b10e6b9924fac9f87621b5f1979f4 : SUCCESS in 15m 32s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9572022bfcc64cf4a3ea0dacda5f8169 : SUCCESS in 16m 03s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/08eb7362208e49b181ec3b60e4abdfc6 : SUCCESS in 16m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f6902b795a72472a8d17a7fa1f0a8d35 : SUCCESS in 6m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cf7ff3c7d38843d8bccdf3d3f7985c4a : FAILURE in 4m 50s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/244833902bbd4fe38cad984ba440bbce : SUCCESS in 6m 45s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/426a25b11f694714bf1870f448a02566 : SUCCESS in 58m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/480d9279373e4ab68b345e9cef6b9b58 : SUCCESS in 50m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/109942e9c6d34d7995c9e429004e64b8 : SUCCESS in 52m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bdf4bf1558b849078824621a4921da1c : SUCCESS in 19m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9785b224aea24a238a10945cd539cf4b : SUCCESS in 4m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1efcb4eed8d74bf9993e3fa011b790b9 : SUCCESS in 15m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d13bd0629f964e8da859c6d2cd476b81 : SUCCESS in 1h 04m 03s","accounts_in_message":[],"_revision_number":25},{"id":"45a7c142505e9c8b5521e0a7ed93c768662892c5","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-06-25 06:06:33.000000000","message":"Patch Set 25: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":25},{"id":"7fbe433b0d7163d7908a93b9b7a81c82a202e0ae","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-06-25 16:46:41.000000000","message":"Patch Set 25: Code-Review-1\n\n(11 comments)","accounts_in_message":[],"_revision_number":25},{"id":"1d6ec4164e23a859d39c81acab7b645d787095c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-07-09 01:47:31.000000000","message":"Patch Set 26: Patch Set 25 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":26},{"id":"e33584c1440387a933016c96b81b2e1a9167e853","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 02:09:06.000000000","message":"Patch Set 26:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d61527a51618400487b6cce235ca2970\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9b3dc8731a914163b75d0b6944289a25 : FAILURE in 2m 45s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/74c959e9bb8345a798199ca5b87fd082 : SUCCESS in 20m 26s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8dab9a4767a04f50990a1df6ee2f8daf : FAILURE in 2m 52s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/2135b49c82954697a926a85d880cd618 : FAILURE in 3m 10s","accounts_in_message":[],"_revision_number":26},{"id":"85e5e43160687ed0ed67aa35717478368d0c617d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 03:23:08.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c87029016a942d8a59a75606543ff32\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6e4d2a3cb23462baee4f3b77190b32f : SUCCESS in 9m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/d020dd7786f945b486836b36df8aa71c : SUCCESS in 1h 27m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4d914c01bbf244708045354197e84bc8 : SUCCESS in 53m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/47053ddf98734261ab76994cba41e37b : SUCCESS in 1h 00m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/71e6b5cf89c3484e9c59033fa5533472 : SUCCESS in 20m 02s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c7f0a0c0095149c191dbd8abd1059b36 : SUCCESS in 20m 38s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a430aa84a64d4f1d83e608568d0a8e0f : SUCCESS in 20m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bd12fabdd56845669dba7c4e5b433692 : SUCCESS in 4m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1ea0956b90084c75a2413594a2cee2ec : SUCCESS in 28m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d60abe06b20a4bc9a4c30a27ad884a11 : SUCCESS in 18m 24s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/7871e44efd3649aead2670b60a07b0f7 : SUCCESS in 20m 47s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/f26a4c70fd4e4050b7a8c4cdbe577787 : SUCCESS in 21m 37s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ee7e764ce10b4dee983965c69711588c : SUCCESS in 34m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b891e1a9897f492f920d5124e94b63e4 : SUCCESS in 18m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/40dc8b970421458bb02b4b76bb15a7eb : SUCCESS in 19m 08s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a4651c958bb64649a4fd129197ab13b9 : SUCCESS in 25m 33s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/dfeed33bb64749d1b666a6c9416cf6fa : SUCCESS in 18m 56s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ceb8771d55aa42e6b96e3eaf8709bd8e : SUCCESS in 17m 10s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/09e6172c1a7e4c6d8e50a7fa585c4cb4 : SUCCESS in 19m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8ad82bb58f224e7b83eef1b471d3ec88 : SUCCESS in 6m 57s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d24b473fc1fe4325b27a7114993d59ce : SUCCESS in 14m 49s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2da747b7fe39487ea8f1947d99fcd866 : SUCCESS in 6m 14s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d8f6e3074782465a9628efbbe97c8772 : SUCCESS in 55m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6d281e5d23344b8989b13bf62d59ebfc : SUCCESS in 41m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/746e938db67d46ce82b3e55832d7a531 : SUCCESS in 40m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e90e12c10be344d0a423fae22667cccd : SUCCESS in 17m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d663b350306484c9477e85b37cfd962 : SUCCESS in 4m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c823a7076a574211b7428915a5a05d10 : SUCCESS in 19m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6786f7165c8c4bb59a9e500839be58ee : SUCCESS in 59m 15s","accounts_in_message":[],"_revision_number":26},{"id":"5034bab8b3cc93862402d037c721d516965c719e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-07-31 16:45:47.000000000","message":"Uploaded patch set 27.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":27},{"id":"4c9de853873eff15329303eef5dd735dd25bda4b","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-07-31 16:46:00.000000000","message":"Patch Set 27:\n\n(9 comments)","accounts_in_message":[],"_revision_number":27},{"id":"6d484c105c7ad2b9041a5d55f48b118cddae5429","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 17:53:37.000000000","message":"Patch Set 27: Verified-1\n\n(8 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a14a6b1a80764fbb9d44c7021f0f1fb2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b65cd2afdc440149ba0fe868e63b648 : SUCCESS in 8m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/25d0c0c1d2a541b89d88de792f0a7d69 : SUCCESS in 56m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/da8930e74a2f4a9eaf17daca8f2cfaa1 : SUCCESS in 47m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/557228f1276a4bd7af8e4ce1d60e5fc2 : SUCCESS in 1h 02m 20s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2e37745b7394427297844180a2fc2acf : FAILURE in 20m 18s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/210b9c632c56451393fa94f5d3151593 : FAILURE in 18m 53s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4c4c9c2f449f4adb8c072ca77131614d : FAILURE in 20m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6bd180500f914a3ea8b9cbfdc0c64afb : SUCCESS in 5m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/69a64207a3f24ce581fdc128fc62ef83 : SUCCESS in 21m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1499dad79bfa4607bd6037e4582378a6 : SUCCESS in 18m 42s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ada418e18b97424ab00e427a00aa06d1 : SUCCESS in 21m 18s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/d1d1a8d2407d42338145413beef17f92 : SUCCESS in 21m 25s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9a8a42ac252c4ae4a50b2488def73b47 : SUCCESS in 33m 36s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d631f3e08fcb4f7e881bd3f983499076 : FAILURE in 18m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1db603457b5343e68fdb452b2723108e : FAILURE in 23m 09s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/001a9741fbd14780afccc0eed71e11c9 : FAILURE in 21m 22s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/75adccefbd3c4b50a59630a3e4b03af8 : FAILURE in 18m 52s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d6407b2cbf224cb9be411378f2673e53 : FAILURE in 16m 33s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2c9f3db55632469ba48df7530c47ca62 : FAILURE in 17m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/82b93ee4bb7248b0bdb82098f9db0504 : SUCCESS in 8m 17s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f1dcfbaa30304833aa4b0fb34df1d53e : SUCCESS in 16m 14s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/68f83fece3034c2fa29232f7dadf5b8d : SUCCESS in 5m 55s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/86ce5d4ab16d494aaf473e8526502b19 : SUCCESS in 1h 00m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a8af22e3a1334d0e87e967450fe085b7 : SUCCESS in 48m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ab42dd82ff124e3a94c5aa929c9a690c : SUCCESS in 37m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ce69c2b802db4051b73671d2f25944e5 : SUCCESS in 17m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a1035b1cfd643ceb45babb2c7d8dec2 : FAILURE in 4m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4d88639853a94668b160bc59593b3851 : SUCCESS in 17m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c2a55bfd8a754c039709fe43f0e373cf : SUCCESS in 43m 02s","accounts_in_message":[],"_revision_number":27},{"id":"6ca23492748fb42b4107e6c59daf7a58bf4896c1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 18:05:27.000000000","message":"Patch Set 27:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f4af432756c4c498a7c2f37fae08422\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5db68fb21dc64f8e97286ef13a4509fc : SUCCESS in 51m 54s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6d479027e557420b991f46595773af1e : SUCCESS in 55m 03s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/0b46dcb93670438382095db512c146de : FAILURE in 57m 29s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/004f1d13fbce4f0da226ab6ba8169966 : FAILURE in 1h 17m 37s","accounts_in_message":[],"_revision_number":27},{"id":"e22633d237cf4abb40f93df5238ecae665a22736","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-08-01 19:57:53.000000000","message":"Uploaded patch set 28.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":28},{"id":"aa659bd4f99813bf56759c160c56b1e97f57bf85","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-08-01 19:58:18.000000000","message":"Patch Set 28:\n\n(3 comments)","accounts_in_message":[],"_revision_number":28},{"id":"ee862ad1a52a12fded7d35b17a4daad7a57a6932","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-08-01 20:34:04.000000000","message":"Uploaded patch set 29.","accounts_in_message":[],"_revision_number":29},{"id":"d99c660b47f218213b8462e74df95ce2b59072a4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-08-01 20:58:17.000000000","message":"Uploaded patch set 30.","accounts_in_message":[],"_revision_number":30},{"id":"d80b1f879510aa39632841ebaa68506f0aa3c823","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-01 22:02:14.000000000","message":"Patch Set 30: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cd3bb39942684bf99a466426b8ef2375\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46ff7aec606240a285c41f77ca2b7b39 : SUCCESS in 6m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/7074d034882d492b8b201661fd8aa4af : SUCCESS in 42m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d16ab628b3164c25a55724764496860c : SUCCESS in 33m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e606ee0fead94bb6929f1e07f0aeb805 : SUCCESS in 43m 58s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e2c34755111049b89e821538ba6683fd : FAILURE in 20m 02s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e5e47495242d40479000308c01f320e0 : FAILURE in 19m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9cdbe31797044208a7818543f11e8fc6 : FAILURE in 19m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/32695b3803d94a55b1eb49b7b80b64f1 : SUCCESS in 4m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6d2d269a2d484b558de21ba950d7a049 : SUCCESS in 20m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c4b57eebb09c4e12bd1ee0962ff76c1a : SUCCESS in 17m 44s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/784a1c2fbd4c4275afcbdeaa06ffeb8b : SUCCESS in 23m 03s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/164104c45eca4d80b27cbd684b7dfc25 : SUCCESS in 17m 54s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/372cb9075def4c04ab54c6465aa25189 : SUCCESS in 27m 39s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/12099ebfd64543cb989c1a3c6f122e05 : FAILURE in 19m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1f8bdfbf312546f4b212d9adfbb4f24e : FAILURE in 19m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/329faf3cd4d34b96aa2511184a4f800a : FAILURE in 18m 16s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/132f44acf56a41f5b21b9c36bb533f1e : FAILURE in 17m 41s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/449cf79f539c48d88228ba59edc9fc57 : FAILURE in 16m 42s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a3422624f5e0411db82ae05b8a9334c2 : FAILURE in 14m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/618f8ac22fff44f4838bd3d55e5c67b4 : SUCCESS in 5m 52s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e6a15911e7bb4891a6a966236af2db10 : SUCCESS in 14m 43s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e1dcb80c093149baa4dd505a7284ef3e : SUCCESS in 6m 25s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/94d195d3871b4498b843fd1d7ed10338 : SUCCESS in 44m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f910a5e4a9be45edbd083440ace46639 : SUCCESS in 52m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8f9f31c10da7400fbd253cad65575dc3 : SUCCESS in 57m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9552d7572075420a8a22376dfd68e87b : SUCCESS in 20m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/237d135b1e6147ddbc75397cf51753d3 : SUCCESS in 4m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3ed493f0e5604471807dbab18fd0ed4e : SUCCESS in 15m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ddfd8998fdd74d768be8e981c070bc24 : SUCCESS in 36m 32s","accounts_in_message":[],"_revision_number":30},{"id":"89460e6e3ce70d2b5f996c550df4b5ebfa0dd066","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-01 22:31:05.000000000","message":"Patch Set 30:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8badc87a3cbe437382a1761dfb73c2b1\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e3b4c62e5f80458a9b29dc690039e1ba : SUCCESS in 1h 15m 45s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fc1be65661004281b2f4f44e56cba2dc : TIMED_OUT in 1h 30m 57s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8026b2cf533749fcab3f872daec7724f : FAILURE in 1h 23m 12s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/fb9533250b11422bb2cb0b20ddd65e19 : FAILURE in 1h 17m 50s","accounts_in_message":[],"_revision_number":30},{"id":"e771b92f5a9d5539f7897a76c1b5af65ffe0a161","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-08-13 21:13:38.000000000","message":"Patch Set 30: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":30},{"id":"d4b4909dcf3f94542c9fd7c948759e74a4ec41e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-08-24 12:49:56.000000000","message":"Uploaded patch set 31.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":31},{"id":"ae7939de01c970626e0d9c84f76c88e87b91238f","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-08-24 12:50:34.000000000","message":"Patch Set 31: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":31},{"id":"ee0dc195999a6d6835c1816e2b42ba8ea486231b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-24 13:53:14.000000000","message":"Patch Set 31: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ce03450923e64c47b14f6e4276b0fad2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab1c651547b34c15aab30d9b96c44840 : SUCCESS in 7m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/a8f6f7681ca943d7859defcf5afba2c1 : SUCCESS in 41m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1e71a8bca0f24d629e8ec08a645af763 : SUCCESS in 33m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4201f47cbdaf4691b391f71740d7c698 : SUCCESS in 43m 33s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6358022031f0402ea1ed0447f5d5dd95 : FAILURE in 18m 34s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4dd048bc453b4389b27b710e6e92d15c : FAILURE in 18m 06s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/afb6b9f300db46a18d0e1b7f7895ed45 : FAILURE in 18m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d6350e4bea91416d8197ebf8d62226ac : SUCCESS in 6m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4b2e00e92ce645b3971d53aedd4cef82 : SUCCESS in 14m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bf78b01c91344175ab67a9a234fa9f35 : SUCCESS in 15m 07s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/65a620bf43e94df19ce3277dfafec1ca : SUCCESS in 15m 12s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/beaf4e12df814027ba67951fe26fdb39 : SUCCESS in 14m 24s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/cf3f6cc2bacd4c96856bdd3399e41cfb : SUCCESS in 16m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3e8aa15599d549288d55e4fedb54a504 : FAILURE in 17m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f8f47e11cca74821b20af475756473c2 : FAILURE in 17m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/196b7688560a4655954990ab0e9ce5a9 : FAILURE in 16m 09s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d40ffeac34d04972acf36c95e2df6ebc : FAILURE in 16m 34s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/15f60b885d69421b96c0174c83647673 : FAILURE in 16m 39s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f8da2d66d18f43618d3fc67f965d96af : FAILURE in 18m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7ac8454852da462f891bd356dd06e222 : SUCCESS in 4m 58s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/09380945aac443b790910a5259cd6236 : SUCCESS in 13m 29s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a9d96544dde04aabbc18a0032f767735 : SUCCESS in 6m 02s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6e05a8fe6f3a4e048516ec537fb5f25b : SUCCESS in 50m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6c83c1b230d34df4a2fcc80851c9a522 : SUCCESS in 52m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9f4aadef5c234453ae2c53b5ca82f6dd : SUCCESS in 49m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a971bf218a0344f8827a1d32b95d2b16 : SUCCESS in 19m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a6925ab1000492fbea50946c3197ee3 : FAILURE in 3m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/aa1d2ae1b23e4b0aa1abd7e540e8e484 : SUCCESS in 17m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c72ac03e3b1841e2bf0430f5863fbea8 : SUCCESS in 57m 17s","accounts_in_message":[],"_revision_number":31},{"id":"ccb5453542431526710af70f017803285be577a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-24 16:56:59.000000000","message":"Patch Set 31:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0ef101cce3f492aa767d11b553aeb74\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b0a3f1d467a04eb08622df1fc826326f : SUCCESS in 57m 07s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6dfde3bbac3a43b99ab0f3c2c9a200b0 : SUCCESS in 1h 10m 55s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e933a2877590472c9ec010d029052cbb : FAILURE in 1h 28m 59s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b0b880122d1548ef8266e9047176d4a7 : FAILURE in 59m 56s","accounts_in_message":[],"_revision_number":31},{"id":"2bc1c2a604bacf887b218238bafb645c59b16d97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-08-27 16:20:48.000000000","message":"Uploaded patch set 32.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":32},{"id":"c89537eef94ccb6389a32def21a04aa36122690b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 17:42:18.000000000","message":"Patch Set 32: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7fb3be38d7de44a59ce34c5cc878a9d8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4a766e4738743a9af032ed3e9626db4 : SUCCESS in 8m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/acb39d3a9b80478aac75c4dd6e0c31e7 : SUCCESS in 1h 00m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/425abe76ba464fa4a6ba9b1ae88e173c : SUCCESS in 43m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c4c8271fe72a4361894ef062737f149f : SUCCESS in 51m 00s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e03f9aaf61b24fc4b4edc7afcb3fdc6f : FAILURE in 27m 14s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/97291e31ca1041a2afe8e85af461270b : FAILURE in 20m 54s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/92ca4804314d4960a847d96cad7aece7 : FAILURE in 19m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/dcd0d7a90fdd4d94ae4acf1c2b031757 : SUCCESS in 6m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c631916431f94c7f8b6c05ccaed7506a : SUCCESS in 19m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/286d656a1100437f8f380e8cf8ff1c32 : SUCCESS in 18m 32s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/4e06ebdb2cbe4803a06aadc7ca96ec8f : SUCCESS in 21m 19s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/4199deab1b4440529bfcdc10bbeb64f3 : SUCCESS in 21m 11s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7ab2071eb090410095c456bde3b7dff9 : SUCCESS in 20m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ba6c74461a7748d4a2784717c9836e02 : FAILURE in 19m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/30b0ca39cc50415a8c65f942874d7e72 : FAILURE in 20m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/361b3a8835c54b2194616c8d8ed3cfd1 : FAILURE in 19m 51s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1670dbdfe9f344bd92c7c7f86d2eba34 : FAILURE in 21m 18s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5d320392b4b94adbad76906c6ab5da7f : FAILURE in 18m 41s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c89c893bf1e14be09c1639e8b4fe6890 : FAILURE in 28m 15s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1b847f8511ee40b6a84aa0a34952666f : SUCCESS in 9m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6370cbaf759f4e5597191e785101994d : SUCCESS in 17m 02s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b65d0740691143a7a7ae3a51edc5f752 : SUCCESS in 8m 22s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/cb48865684464098a3fce18dd9103621 : SUCCESS in 54m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/18615115852e4b06bd16ae736aebb930 : SUCCESS in 46m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/51c685f8af054f46a47f8336e4849a28 : SUCCESS in 55m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c674b00761fe4be29b1f76288d926a8d : SUCCESS in 20m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02f792475247487c8b9456959625d649 : FAILURE in 5m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1a394b67b65a4c28bde79ca05771ec98 : SUCCESS in 23m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6ac629763c97443ba5e14f49441a2afd : SUCCESS in 57m 27s","accounts_in_message":[],"_revision_number":32},{"id":"63669c5d280b07ae79a64310afe23ee3df5b2e2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-08-27 19:06:43.000000000","message":"Uploaded patch set 33.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":33},{"id":"b4168f0e9d7b929cfedd2fc1839da8be84200b70","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-08-27 19:07:33.000000000","message":"Patch Set 33: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":33},{"id":"b3629fe608e03defdd2f5bf0c5a8e882195d4a76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 20:26:36.000000000","message":"Patch Set 33: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/66e2426c023b44beaa606f2c8b28373e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a459223011ef4a92b501501c677246c1 : SUCCESS in 9m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/ba28856efb4248a8b83a18f4e80de36e : SUCCESS in 1h 07m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6320235581f146a09b293193cddf3fcc : SUCCESS in 43m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5bcddd5e1ad646d681ffd8c0a14c758f : SUCCESS in 1h 02m 42s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2f75be6d2c5f41a7ad55ef285795e4ad : FAILURE in 23m 06s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c8c7222673da4c7b92a355d1edbf006a : FAILURE in 22m 30s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a51d094f96304bf1b477be54ba041227 : FAILURE in 24m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/055c40cb65024c26955f8e043adf6eca : SUCCESS in 5m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/904fe889735f40c9941eb6897cfd15ca : SUCCESS in 19m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/306fddbb55e64e8fbfcc83d4fe2ee9c0 : SUCCESS in 19m 45s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c9ab344df7a14a3c853d5ac07dcdd559 : SUCCESS in 20m 47s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/0deae0502ff64b46bd35892419b76d1a : SUCCESS in 28m 19s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/887e46f462f744acaf92e0bc40f4de94 : SUCCESS in 25m 18s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c5bd1c9d41c4436ab5b91fb11684bd5d : FAILURE in 19m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/11b69fba89d649e4be6efc415582b0ad : FAILURE in 18m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7afd01bcb6c742a4a995ee82ef936787 : FAILURE in 18m 37s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/854953e48c004c668cd0944c6aab5752 : FAILURE in 20m 31s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/bcb6a598e3b74dbda062105151640c07 : FAILURE in 17m 11s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9559c657b4794ffebe61a8c4005b017c : FAILURE in 18m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fb629b99a8ed4211bfa17f13c6adfb0c : SUCCESS in 6m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/412fa383d36e40bdbb1b159fb56899ce : SUCCESS in 13m 41s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/25df77bbe8a642fb882d47d4ff72f2be : SUCCESS in 6m 48s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/46164baa4c57417eb1145f02e4e0ef25 : SUCCESS in 51m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6cd5c39758c447989a99010afcd33d18 : SUCCESS in 43m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7250b49da18245a59119c6616be7bc42 : SUCCESS in 34m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f2f2638aa6a44df49db430ee7e27e109 : SUCCESS in 18m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9a8e8184f3145c4b8ca6659e0cbe301 : SUCCESS in 4m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fad19e7d07fb4905ae37846c51162ed0 : SUCCESS in 20m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d2a9fa3027fa464aa760f05b79f48183 : SUCCESS in 48m 45s","accounts_in_message":[],"_revision_number":33},{"id":"a4a56a00e776e46b095babd397d30cc107ec9593","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 23:42:07.000000000","message":"Patch Set 33:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9bb0b1d86d634327b596694eaea112b0\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7d2dcc95e6f645dfb148a0efd870ae72 : TIMED_OUT in 1h 30m 53s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c8831bcdaec541cca4283c0fd00744f1 : TIMED_OUT in 1h 31m 13s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/7af750f5be83402b83f6377c876f7978 : TIMED_OUT in 1h 32m 01s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/51a6e5690b574ca889cb08b913a58028 : TIMED_OUT in 1h 31m 18s","accounts_in_message":[],"_revision_number":33},{"id":"92b894d2e49408a6ea967cd11bf672f94ad84bd1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-29 17:13:14.000000000","message":"Patch Set 33:\n\n(3 comments)","accounts_in_message":[],"_revision_number":33},{"id":"0480fbfdab211effacfb2d59eda9f7506169e56b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-29 17:13:27.000000000","message":"Uploaded patch set 34.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":34},{"id":"a9bb22094cf2554fe96db1f7398bcd848eaa3500","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 18:16:04.000000000","message":"Patch Set 34: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/437145b72e6b4aaf85a77a16e2074299\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f4a6b8fb0524640b3d0f3da893b4340 : SUCCESS in 8m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/3bc28d57a51847a385129fb22c644868 : SUCCESS in 56m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a43679bc537243b7b3dbc6ebb9589095 : SUCCESS in 48m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6f43ddf23b6f4569a935fb4e67de43c0 : SUCCESS in 55m 19s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bf289b3d20874bbcb31a6c0dc2732961 : SUCCESS in 18m 14s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8a682c859f1b4b22a2079ff4563639e2 : SUCCESS in 18m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0849711ae1e54276a70fdd0543aebd6d : SUCCESS in 20m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8462ee17499f4390adc5faaec2c825e3 : SUCCESS in 5m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/45f337f842414b2ca0c2a863bc8e4bea : SUCCESS in 17m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a77e28395a60433981ec50504c659eb3 : SUCCESS in 18m 24s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6c9b3ab9a4da498ab88286d8a1630cdf : SUCCESS in 18m 41s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/649a298c5c70415d925df5815a193ac3 : SUCCESS in 18m 52s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1251a00746484a1bb0a29a14b60a6da7 : SUCCESS in 32m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/979a4a2c28434f8980ff4ee2e0c01b12 : SUCCESS in 17m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9c73631de4954556841ac61e9a0d1658 : SUCCESS in 18m 25s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/121725f8bf544b398dde6fa2833e282b : SUCCESS in 20m 03s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/007e368bdda84356b0f7ad8d7a593b8a : SUCCESS in 15m 17s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4df61f6173cc44a5a826ab66111e218d : SUCCESS in 16m 30s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/45568c2c8bb648a7b8918312e91bc72f : SUCCESS in 15m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4b4c4a2e6ccd4828b45dc17e8961ac14 : SUCCESS in 6m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1bf6be9383164cc1aa1f38d8142f9648 : SUCCESS in 13m 19s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2535fc359ab3453eb45e31e5688f4858 : SUCCESS in 6m 17s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c8b903f5b6b540e388f09a11a1db918d : SUCCESS in 48m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8d661d391bff4b21b7d80aebb523ef63 : SUCCESS in 43m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5f7bf454a9d347278e3d80e90b5e7ea7 : SUCCESS in 46m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d087ab7966dd484db990b96f24ce078f : SUCCESS in 17m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f46924b1a9ad41c6811eef235f7541ba : FAILURE in 3m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5b075dec17b04c8dbaa8e17e5b9fb071 : SUCCESS in 17m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ce26464e37fc4edaaa29ded50f379a73 : SUCCESS in 52m 11s","accounts_in_message":[],"_revision_number":34},{"id":"64a39be62d5b22f3d2f1c912188e947d1c1e16c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-29 18:48:27.000000000","message":"Uploaded patch set 35.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":35},{"id":"195433a8398ab6be6832597d064019a84b39c579","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 19:56:54.000000000","message":"Patch Set 35: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55e211dc94cc464188477bd653d4ae61\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/843cade4e0c140509a4f6b116b3ec561 : SUCCESS in 8m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/f6c0ea5a62ae476bba7a349711ff2351 : SUCCESS in 46m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a6abbaec62ef43d98a368abd508d023a : SUCCESS in 42m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/25f4ff52e7c94d69a6626dec42d5968d : SUCCESS in 47m 10s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/42a3f56eca97458f8c29a908e8115f4a : SUCCESS in 20m 04s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4467bc742f364fc083d7b5fb6df38628 : SUCCESS in 22m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ac4e1da807b14cd69ae454b6d558966e : SUCCESS in 20m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a14ab0213c81403ca8ebd40617ee3286 : SUCCESS in 5m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/250307696a8f4aa28d2eb7a005539b80 : SUCCESS in 17m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2cdad8d7ac8044aa92cd797343a7d44b : SUCCESS in 18m 01s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f63654b9f0084061b8ff7b99e76dfbde : SUCCESS in 18m 43s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/ee34bdfd685a4365b4ac0924c7514d6a : SUCCESS in 26m 16s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a67e24211b4f4788b8e6e97ec6f223eb : SUCCESS in 28m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ad8424b2eacb4a9996aaa85c23655181 : SUCCESS in 21m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c217a116baaf4a0d8d83fbdac41fc0c8 : SUCCESS in 19m 09s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c63dab64ea184c5dbdc206a06f5db4db : SUCCESS in 21m 56s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b02a6bcf17964e698806f5375f2369c7 : SUCCESS in 18m 20s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/fbec8d1c7c3b4306813813f66c625f89 : SUCCESS in 21m 12s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ef3e11f37cf24d35aceecda724f3e7f3 : SUCCESS in 17m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9671130e53b1488bb06bc379b9f078ae : SUCCESS in 6m 04s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0d403d8611d142d1b9ce3b60625b0382 : SUCCESS in 15m 20s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/cb52dec8003d44da8c95b829e36574e1 : SUCCESS in 6m 44s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d31ef4d16bcc4b6c88553651abe0f506 : SUCCESS in 53m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c481b49d068042f19078725b1f359854 : SUCCESS in 49m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0ad0fb9dc4b244e8b7fa3a9bc1d4744d : SUCCESS in 46m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3b3a3688538544d8b17de5142b5cbeec : SUCCESS in 20m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfe9dc26cafb453897ab12c359301c82 : SUCCESS in 4m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4ebf133cf1c448c88b253d38133e3d6a : SUCCESS in 20m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/67f2096c80d74c54819098af5f201ed9 : SUCCESS in 58m 46s","accounts_in_message":[],"_revision_number":35},{"id":"6d7a66157598cc65c0dc3bce9c2dbdaecce6979d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 20:49:20.000000000","message":"Patch Set 35:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2050905b8e224a73bb68f6cd6b4d2743\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ef913c60e94145e6888ff8ed6d2a1336 : TIMED_OUT in 1h 30m 50s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4ffbda8378c04aabaa9fd73317c51ddc : SUCCESS in 1h 28m 07s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/bfaec5cef7824a688d989dbf4ab9fe12 : SUCCESS in 1h 30m 21s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8a7e42ffd81e4c1f8bf24525ee623711 : SUCCESS in 1h 07m 00s","accounts_in_message":[],"_revision_number":35},{"id":"0f914c6917f60c1d31fc0552ab9e3457e48d685e","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-08-30 16:12:19.000000000","message":"Patch Set 35:\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"603076c466b41483b57bf5af939457a06c29d3c1","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-08-30 16:13:10.000000000","message":"Patch Set 35:\n\n(1 comment)","accounts_in_message":[],"_revision_number":35},{"id":"c9c821adf077deb439b565810c028e95b6f6ce11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-08-30 16:13:23.000000000","message":"Patch Set 36: Patch Set 35 was rebased on behalf of \u003cGERRIT_ACCOUNT_15343\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"_revision_number":36},{"id":"ba6856cbde2b56b962fa6e6eb9f4f8ffb7f884e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 17:23:09.000000000","message":"Patch Set 36: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc9471d2fbfa4e73b8c71157716061a1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7efd6fa73284ec9b0b5ceeabde9bab5 : SUCCESS in 9m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/9855857e4afa4522b8c861465f370f38 : SUCCESS in 1h 03m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8535dcbadf904e29ab933bd57b9fb56d : SUCCESS in 58m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9c1e1fa29df9430b857c0e831a5f9bc1 : SUCCESS in 1h 00m 29s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a01fa875ced14ee9b4630948aaf71066 : SUCCESS in 18m 18s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/185efdc8c3ac40ba9ceee44f7d43a273 : SUCCESS in 21m 14s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/16a38eb7955c4ca8b2fbd5bbd77eccd8 : SUCCESS in 23m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cf5894d57f6a4e689266eb6cccf45645 : SUCCESS in 6m 43s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/db9c4bd61c534a7699761ff8a7396327 : SUCCESS in 21m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7a3dd2fc25e2480d84171509c8e20c57 : SUCCESS in 17m 23s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/32d7ff2059b948cab53f5ded5af78be1 : SUCCESS in 18m 54s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/1577c1bb50f14b50b039b7e8983c9fc2 : SUCCESS in 21m 35s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/43b47ff4c15f40a59c3e555c9c286ea6 : SUCCESS in 29m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/addb6a9ec7c24dc9849441a758189375 : SUCCESS in 22m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/eef864d41d0546078bb98b2c79829054 : SUCCESS in 17m 40s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/70ef378869554d4da99160a5ba152ffa : SUCCESS in 19m 52s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2de71b6d21744479ad2e206807c182a0 : SUCCESS in 18m 05s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a5d9a642724946edbf95ee7a30f8d92f : SUCCESS in 19m 39s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1b60c9ac90fb4899b267ea5c2323529d : SUCCESS in 21m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8928f8d276294c10bf51f7a4dafba703 : SUCCESS in 7m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fd4205b86e0a4a35bd7b7c32b9aa2bea : SUCCESS in 18m 56s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/cfb078b677f44be7a1defae1c62d33ae : SUCCESS in 6m 25s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8f3af4bf86ed4661938134a27b5b4a5c : SUCCESS in 56m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6f2635a765c144b5804edbdaba640e0a : SUCCESS in 37m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6a09cda0dbfe416b92f9dffd9270121a : SUCCESS in 55m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/de6c8f35bbf04ec9a30acd51e9c07e72 : SUCCESS in 21m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15a8c9f95d36425087121b74f89e600f : SUCCESS in 4m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/084c3b38c6924f1ea2dba85da180c521 : SUCCESS in 25m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e63531013053419c9274bc8c1dbe353b : SUCCESS in 51m 52s","accounts_in_message":[],"_revision_number":36},{"id":"075d076f66cddf6ee2db67a7d97b6479cc9ae6f8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 18:46:20.000000000","message":"Patch Set 36:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fa64450e73f45ec94436e8c0fed4016\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/09268e94ab0d4613ba4c9712b59ec84a : SUCCESS in 1h 23m 16s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c6679f63e78849adafd3ebd62b119874 : SUCCESS in 1h 14m 07s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8e94119a41f44691aa93341925310987 : SUCCESS in 1h 17m 32s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/35e164d437074af3a61970a828565f23 : TIMED_OUT in 1h 30m 57s","accounts_in_message":[],"_revision_number":36},{"id":"a86060fd890875969ff4535a792df910a653f804","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-09-03 15:14:08.000000000","message":"Patch Set 36: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":36},{"id":"13a097587081668f626e2f96b1dfad1ea8167d85","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-04 20:38:41.000000000","message":"Patch Set 36: Code-Review-1\n\n(9 comments)","accounts_in_message":[],"_revision_number":36},{"id":"f6cdf0b8d11398d0456c37885194336043091de8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-09-09 15:08:35.000000000","message":"Uploaded patch set 37.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":37},{"id":"a4b6ecee9dcf14aad4476b46de790a82e938ea3a","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-09-09 15:09:39.000000000","message":"Patch Set 37: Code-Review-1\n\n(9 comments)","accounts_in_message":[],"_revision_number":37},{"id":"0b5dfce1b4c4231f1323b328b3ca149d57eaae87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 16:03:42.000000000","message":"Patch Set 37: Verified-1\n\n(14 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/75e2f9e311f04522a6937215503f164b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf596177e112492ca698913cfe5d9e84 : SUCCESS in 9m 43s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ffd152d2441d420f9e4748ba1139a3a3 : SUCCESS in 24m 56s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/be520cbf5349435398c0692d1aeb7e26 : SUCCESS in 22m 06s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b5eff489eb80452d9ce8fc7c7e3e6c74 : SUCCESS in 21m 26s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3c9d8005bfa54bbc8889d32cad1a01db : SUCCESS in 4m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9907235031c64bdab3d29cd7e5591bca : SUCCESS in 19m 37s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/473e6243e7604ca0884cd3e82bc6fd6f : SUCCESS in 16m 51s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c11373ed252344f1b8b076b47acca934 : SUCCESS in 15m 46s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/651fcdca014a4b6dbb84a21bc8ab087f : SUCCESS in 19m 38s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9eedfab5eec0489e92883f6e39a8abb2 : SUCCESS in 26m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/963c21eaede843b3bd55bf38d6ef8459 : SUCCESS in 16m 50s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8323c9e8d218481885af0abaa27d4e8b : SUCCESS in 20m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/12af944204944967993a63884dc30245 : SUCCESS in 21m 47s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/add650762b56478291507caec28f4826 : SUCCESS in 18m 46s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ce72bb16379c440297124e7aacce49a4 : SUCCESS in 19m 12s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f8f73b16c4d34a89ba1edb9621a617b2 : SUCCESS in 18m 21s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/84e16a286fe94a9f9dcf0cc8fdca774f : SUCCESS in 54m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6ab5cb33f28f4743a9121a9be1f9a736 : SUCCESS in 48m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/086b6aa8c8f14a81bfbf077506def5f7 : SUCCESS in 42m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/82660d8a8596494bac0f828da9b3edb3 : SUCCESS in 27m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d899c46dc00a4178884bbccc406f510c : FAILURE in 4m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0e8a0f97d37040bf98b34c821e5ba732 : SUCCESS in 15m 54s","accounts_in_message":[],"_revision_number":37},{"id":"713e83ca681c23f1c464c4d38fd6f1acd71ac0c4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 17:53:26.000000000","message":"Patch Set 37:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06b892d6862944dcbfa940607e766bed\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/37c477c34b704a95971382b0551d15fd : SUCCESS in 1h 03m 02s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/432a35d271914648b29858e7520aec3d : SUCCESS in 57m 57s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/eab01e2087cd44828db325be6de4ee01 : SUCCESS in 1h 00m 49s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/48d74d46124549039b671b61ad0943ae : SUCCESS in 1h 08m 11s","accounts_in_message":[],"_revision_number":37},{"id":"53e54297e5abeece14c0bc2a2e330c438253509d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-09-09 20:27:11.000000000","message":"Uploaded patch set 38.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":38},{"id":"3430dfd561cb3b64032fe35dcbbc4499f07afd17","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-09-09 21:13:31.000000000","message":"Patch Set 38:\n\n(2 comments)","accounts_in_message":[],"_revision_number":38},{"id":"97b25240033ad1f7b4b263f62e9ef8325f67c9ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 21:28:09.000000000","message":"Patch Set 38: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b460f7f47cb40648e313cb35e65706f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16e030257f824d62bb02519034b6862f : SUCCESS in 6m 11s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e6c460a87bc04978835cf6cdccb2acf2 : SUCCESS in 13m 01s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9b27b31ac8fb4ff8a6f3cbab3dc5c70e : SUCCESS in 25m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4824ddb345674dac93abe49da89bba0e : SUCCESS in 26m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3ca6cf56e9454e70b1cad63a88ba5a66 : SUCCESS in 5m 54s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0671c15f5e0549fcaf8a29dee0af6e14 : SUCCESS in 12m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6a14de12e21140988f25537df6aa98c4 : SUCCESS in 22m 00s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d440fed80f28419685396ab14246e397 : SUCCESS in 22m 41s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/82c5780b2334430caa3cb83cd0bcc9a9 : SUCCESS in 31m 57s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/eb510a8b5a834f8dbcb7728d33a7a0d1 : SUCCESS in 27m 40s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c2b68c1d7cbe4875a4a092666b77ffdf : SUCCESS in 23m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/78e8aa1a7348445cb9214b7b056ed1cc : SUCCESS in 19m 51s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d285f78050504fd1a8be02d1ec556089 : SUCCESS in 22m 59s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3def2fe4d85b4678b157ee5a5ea8a13e : SUCCESS in 18m 29s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a7eb36385bce41dfb8115a7a40214023 : SUCCESS in 21m 46s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f054d3a1fe5c4eb990072d2b3ee32014 : SUCCESS in 17m 08s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/20d05f4eed0f498f829d423de2997d13 : SUCCESS in 53m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d85281f15068433aa3ef4542febea61c : SUCCESS in 47m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fe71b372d5cd45aca64e1b6167867a37 : SUCCESS in 53m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/854053dc22724be98b712cee21636ac0 : SUCCESS in 19m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9e23fa3d7d0419d819cd49c5d51ae94 : SUCCESS in 4m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cabd50a64a304004be18123e47ea7261 : SUCCESS in 17m 52s","accounts_in_message":[],"_revision_number":38},{"id":"5fb291cd27beee6895af887ccc51af774fcd9baa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 00:44:16.000000000","message":"Patch Set 38:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/981a15d6c3fc43f8bb8d5ef960f1aa34\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8b09478fb8704d6785ac17164f88f01b : SUCCESS in 1h 23m 48s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c66d3249012e4891940eb57fdc4a3262 : SUCCESS in 58m 13s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8c5d0e0238444c0ab05c7715a75ba011 : SUCCESS in 54m 48s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/c378a2d938eb4b7e87fbcb492322db06 : SUCCESS in 59m 18s","accounts_in_message":[],"_revision_number":38},{"id":"346eb472e777566856bc17120a0fb4eb799527d1","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-09-11 03:48:09.000000000","message":"Patch Set 38: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":38},{"id":"ea6db4d1dbba6389020059d90d4783ec65710c3c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-11 04:02:08.000000000","message":"Patch Set 38: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":38},{"id":"6458a89bd4c61c91fa7ec94369a41bf0597ae850","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-09-17 19:54:51.000000000","message":"Uploaded patch set 39.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":39},{"id":"d422d142dfe798841f310fe9630bc6d8ad8ab26b","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-09-17 19:56:28.000000000","message":"Patch Set 39: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":39},{"id":"da81bd278b7c433f9e80a02b21c49a02b82faf87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 20:52:28.000000000","message":"Patch Set 39: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eec5a5669dcd41e8a980b7478566f779\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5228fa054c447fb956bb834e48a2551 : SUCCESS in 7m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/22ae22c4c0ad4146a0873badec5dfdde : SUCCESS in 41m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7e3bb5a8d0474643874ce64560aead4a : SUCCESS in 36m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e3e3f8d0212c4af6ac47e9fe12d11547 : SUCCESS in 18m 41s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d9424691391643628edb44ac9037c54e : SUCCESS in 20m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1444758199474e48885a86684d319fd2 : SUCCESS in 21m 26s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0896af2070034c578ed1f8f48656898c : SUCCESS in 20m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/00efc7b42b88451483347a5fdd11f717 : SUCCESS in 5m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/52542657c46249b7a1bc8b359653dd38 : SUCCESS in 11m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e6f4433af78648a2abecc0c0a422b473 : SUCCESS in 18m 53s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/aedffaed9ae24768bddebc33f7e628bc : SUCCESS in 22m 23s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/3b293ed031454670a3df139c549a891b : SUCCESS in 18m 12s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/00d5355876c14ca39fffa66faca8edaf : SUCCESS in 29m 56s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ca09edb5ed7c402788719627a12950c5 : SUCCESS in 12m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/580aae7b8b774e9b9b8f090fe9202e94 : SUCCESS in 17m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/36b181246ec04248bb27475f8c56091e : SUCCESS in 16m 55s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/402b1b4a558b4f309d36819ac391e329 : SUCCESS in 16m 30s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4925ed931fe54caa947aa7d704389f81 : SUCCESS in 18m 02s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ae6180c32f874436959d7d23e6f0d0f7 : SUCCESS in 12m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1de95eba19454454b362e82391f466ee : SUCCESS in 5m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/875ec498ceb34e70b0fa17434b0bc5ee : SUCCESS in 14m 44s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f15e5b70fa464651bf7705e36f8c853c : SUCCESS in 6m 55s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c235cb5d467c452090471e42062e04a9 : SUCCESS in 48m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2ae68145f767490ea436a5cd603b712b : RETRY_LIMIT in 19m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/52a48f73a74e4f36aaa56e897843e54b : RETRY_LIMIT in 18m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/12d9cc50ecfb4625b507108e1f01e193 : SUCCESS in 17m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7aa680357f6b467ca56e64cdd27c351b : SUCCESS in 4m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/48e563d5b45240adb6f153d39cbe5f55 : SUCCESS in 12m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8d01a2e2610541929e44033297015fdc : SUCCESS in 47m 09s","accounts_in_message":[],"_revision_number":39},{"id":"4dd5a1d45945a342f1b5007900fb5acf7edbe124","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 22:35:59.000000000","message":"Patch Set 39:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98e18939d0df4629904079964d1b54b2\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e043807125ab47c7b8463bb82d456cfd : SUCCESS in 1h 22m 25s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cb366869d35f4e1dbf771c027a6e5d15 : SUCCESS in 1h 40m 43s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/06b923b421284d6492b66ffde8f56f08 : TIMED_OUT in 2h 01m 00s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6af3a8aa823248e2b3a444615b7b7865 : SUCCESS in 1h 01m 10s","accounts_in_message":[],"_revision_number":39},{"id":"0d74aae8abf8b0e9204bb888ddaeeef3c4716743","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-09-17 22:56:17.000000000","message":"Patch Set 40: Patch Set 39 was rebased\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":40},{"id":"4f64eb339b2d1ea10946da0d2cc959b7b80e0037","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 00:05:58.000000000","message":"Patch Set 40: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c03d9b296eb94f3697aa1fc0a3c320a9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf348c0bcdf942b1bbfcc2fe794e0364 : SUCCESS in 8m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/527b0e2885ff41bfb21f14880d0c7001 : SUCCESS in 59m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/68918e3c22f441b0a7c703b1d2269ac5 : SUCCESS in 49m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/beefefdebb5a49a3bc01c2906e6a41a4 : SUCCESS in 31m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/416033e5afad4b4c9f7e17fd35f4791d : SUCCESS in 18m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/65595d55c4ca4d2d829f41f8143da5b1 : SUCCESS in 19m 41s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d5f8585e0129443b85d07d66e278676f : SUCCESS in 20m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/27c399e10a5049c78198c856ffaa4aa3 : SUCCESS in 5m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d3ae8ac5db5241509aada92a866ff05b : SUCCESS in 14m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/439230e7f86b4fb090f7fe1580cf2921 : SUCCESS in 13m 05s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/fb4fdcf8e33b4ad683a0a4fb51a30465 : SUCCESS in 18m 18s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/1636df4b90c3437ebf8313864cad939f : SUCCESS in 18m 40s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5a735860ac0640f2bb6babc43fab2934 : SUCCESS in 18m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e57ec737d81e4abe90b2e728f3ca0044 : SUCCESS in 18m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5aa01421093c4b738e1b39e8158d4189 : SUCCESS in 19m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9e81cfb5a9a648099d7e09ca9f44de1b : SUCCESS in 19m 34s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/cedded9a8e974e419813ae71dccab045 : SUCCESS in 11m 11s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/75f6e316ebfc4386a6eaf18a782def3c : SUCCESS in 17m 35s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f39d7024e18241f6bf90d40face54c49 : SUCCESS in 12m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/39a81c90dc874d8b840d962003f3bae3 : SUCCESS in 5m 45s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7a05ef2c798a4e7ab9182a78ab3b9ec4 : SUCCESS in 9m 44s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/97e728f1e1614d6ca91c39846d72b8f0 : SUCCESS in 6m 34s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c93aecd41d8044859bbc2ff0af8f1ca1 : SUCCESS in 55m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/801b3728f5b14bac80d1dc2b5b25752f : RETRY_LIMIT in 11m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/88d69f1086754153878df1d389290f8d : RETRY_LIMIT in 10m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fe588a0b59414ecda94b782eec07e53d : SUCCESS in 23m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/368c7cf3c3e347b0bc6ed11aad1d65eb : SUCCESS in 4m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d8cf079b0f6541d0b03b23f644367295 : SUCCESS in 21m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a4dcc4c01c394a6281c47453de1e89a5 : SUCCESS in 1h 03m 30s","accounts_in_message":[],"_revision_number":40},{"id":"f25819186162dcbd4c01d47a3d3f7d44f8549b9f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 00:55:48.000000000","message":"Patch Set 40:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77a498bf97974608a4f8881b9655816c\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bdf423c12cce443c8bbdf4d45958b5c0 : SUCCESS in 1h 11m 55s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/63065b134d1141ddb12c1823b565348d : SUCCESS in 1h 02m 35s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/0e045e854e7b418382d795e80da7aab8 : SUCCESS in 1h 14m 09s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/953868a43069474b8e2c013a2a09f7c1 : SUCCESS in 1h 12m 48s","accounts_in_message":[],"_revision_number":40},{"id":"acea7d79b78b3386af46b1a439ce5d38ab87b64a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-09-19 04:22:13.000000000","message":"Uploaded patch set 41.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":41},{"id":"e74dedadd43cbd560e774802a67646cd74f312ad","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-09-19 04:23:00.000000000","message":"Patch Set 41:\n\n(3 comments)","accounts_in_message":[],"_revision_number":41},{"id":"2f1c0e53059af0da5405f4c706db966d53c713a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-19 05:36:36.000000000","message":"Patch Set 41: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24ddd4c5335f41e2acb1e695842f7eb3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/120c9001fe684f1d82abfb6b2a1a87f4 : SUCCESS in 9m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/f12b0f2b84b74aaf8112c7d40e8a24cd : SUCCESS in 1h 08m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5ea17f5f9b374bc0abce6cae9aaebd86 : SUCCESS in 53m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/19479d702f6447489871ba456dfa76d1 : SUCCESS in 16m 41s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ebaed64223e147c9adbc134b8a152dec : SUCCESS in 18m 59s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5e408514da8c491f9ea1ff439bdf976d : SUCCESS in 12m 39s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/eb6de453df0047b8a98f142858103a33 : SUCCESS in 20m 18s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ac065828b0c044bbb07baf536ef4b40e : SUCCESS in 5m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ce630ebbecb64b3baa466cf127e30ee4 : SUCCESS in 10m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d7c4268ea888496aabd5ec27bcc74da3 : SUCCESS in 9m 24s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/876eda63ae4b4c44b6ac22128cad6757 : SUCCESS in 15m 09s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/c304b023cd9c44bd80fdc325c448c271 : SUCCESS in 15m 19s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/26f635f5dbe347aba015bf0778a73133 : SUCCESS in 17m 57s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/31fb3b15109f4f649550d37aa2922b37 : SUCCESS in 16m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b51d14c17b214280bad3279388668120 : SUCCESS in 16m 09s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c8839fe9ec1e46baa831838952447f4f : SUCCESS in 16m 56s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/489b5dc7b1094ea2a2ca2e1fd4f15b4e : SUCCESS in 15m 03s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/114ddf03dc104d8a88328a616826121e : SUCCESS in 14m 36s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/bc74a97c1149483fadd4854d1a5b25b0 : SUCCESS in 13m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f766683a47834edf96b84faf8a4a499a : SUCCESS in 5m 55s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e99b0b26333d45079bb33c5d3c060b20 : SUCCESS in 15m 10s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5007242bbc064e8493849b3b6f1c0085 : SUCCESS in 6m 34s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c543d1fe24904a0cac5f5283d859d4e9 : SUCCESS in 35m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ac027a6e4d004988a92ee99dec5a7aff : SUCCESS in 52m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2e44b891204e4fe7bc4696aca63907d0 : SUCCESS in 28m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1379ccb66b6347b9b953352b10129236 : SUCCESS in 20m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b25faa60f2274d87a2b751f8995f64cb : SUCCESS in 3m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/637a29b04825444b804d7cd56fcfde35 : SUCCESS in 10m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9eef50aff3bb445b82f0e9fb1a6f2121 : SUCCESS in 1h 00m 05s","accounts_in_message":[],"_revision_number":41},{"id":"329eccaa829ac9452bf362ae0725807e230ef3c3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-19 08:02:52.000000000","message":"Patch Set 41:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e264940b48f3441ea41c5053460e45e5\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e186560e91494b88b7ee42b426dd3eb5 : TIMED_OUT in 2h 01m 04s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/da65d6f68c724dc3a1e676dda559a763 : TIMED_OUT in 2h 01m 04s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/755d5bdedb604de7baa6b382ca794f21 : FAILURE in 1h 58m 34s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b8fb23d46bd543ed9fb1b33c6b7f87cd : TIMED_OUT in 2h 00m 53s","accounts_in_message":[],"_revision_number":41},{"id":"5c393d1661e6c4e321dd809ff038f4e43891eb38","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-09-19 12:03:27.000000000","message":"Patch Set 41:\n\n(1 comment)","accounts_in_message":[],"_revision_number":41},{"id":"4ac9d6730cd67b97a6a16751343eca94c6de9a94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-19 13:14:52.000000000","message":"Patch Set 41:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78a1f38128c24b63920204de98244c05\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f70f66a7f5b421fa9186a8a1d39c9f2 : SUCCESS in 8m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/9fe3ae775144411cac367cf5a935309a : SUCCESS in 1h 10m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cb89eab8f73844a5b4aacb091b93dd0c : SUCCESS in 57m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e113f03ad42149508bca239b1fb8d0fe : SUCCESS in 22m 05s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/62c4e64f233b40cab6e665818afec963 : SUCCESS in 23m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7d855c91f7c4499dbe3e28a1206f8a2b : SUCCESS in 20m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4379656271a44f2cb345412949ef2430 : SUCCESS in 22m 07s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/36474d7eafec4f948834d55e218bf6a1 : SUCCESS in 6m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0c52cb44daa343e59ca0b2aa412774b1 : SUCCESS in 21m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/47d469db8dd94131862238681fda1b31 : SUCCESS in 20m 20s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b750e03b98554f3b99f94b62d031093a : SUCCESS in 19m 37s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/a3aa48610426494aa512f11ce912a5eb : SUCCESS in 18m 31s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3c91971938be489da00a3ef417c00175 : SUCCESS in 27m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/46cd65c305374e35b625e0701aff53fd : SUCCESS in 12m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1fc5670f89a840aba32a36bd3ddd82f9 : SUCCESS in 17m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2ae833e4824c469a90ac62413fd541cd : SUCCESS in 23m 11s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/866cacc13d2c444fbbbfc18468a7733a : SUCCESS in 16m 05s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7b51f7734a724518b30a7a3388680625 : SUCCESS in 16m 18s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7e874a58888f4667a446af27d3c041ab : SUCCESS in 17m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1650748b63f84388bafae998d39d64dc : SUCCESS in 5m 42s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dfe020ebe504489c91a598945e31ebaa : FAILURE in 20m 36s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c32da719a87040f6909ca4a7267e4e55 : SUCCESS in 8m 46s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5e3352cb63824ff891120c334b6069ea : SUCCESS in 57m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/37120c5f4b3a474592e8d4ad5cca9d4e : SUCCESS in 56m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d3315662edcc46bc8da85efe13e4facf : SUCCESS in 53m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/784c5fb7caf4484e945d6c465a45f09b : SUCCESS in 23m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/796db6a86f2b45ab8bc3396b74c77dae : SUCCESS in 4m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c4ffab0b98b748ca8412b4c69a9a8b8b : SUCCESS in 26m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/85b7ef4d06074c4a9407b5849ec5e8ef : SUCCESS in 1h 02m 00s","accounts_in_message":[],"_revision_number":41},{"id":"a57bd10bd69949710696c0328c6eb0ff7cff6fc5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-19 15:20:22.000000000","message":"Patch Set 41:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22c3406b2e55405ea178840a7004c14f\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9f0df70587794939b5f138ad833f188b : SUCCESS in 1h 32m 31s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c3fc2d1d5667405ea732c0a58deeb14b : TIMED_OUT in 2h 01m 03s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/5515c7179f944cdfb2f49be17b7c2403 : TIMED_OUT in 2h 00m 57s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/bd4a3fcbad684284b2c2df62434f70f1 : TIMED_OUT in 2h 00m 59s","accounts_in_message":[],"_revision_number":41},{"id":"696d224a32dc0f6d0c0f4d1b5b829f65a6e98656","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-09-23 16:00:28.000000000","message":"Patch Set 42: Patch Set 41 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":42},{"id":"79594be15a5d1acb796a001b0c59e1b88bc0a6e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-23 17:16:19.000000000","message":"Patch Set 42: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/feb097d144a6473a97e4fc4a88cd13eb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90ca3457a74e47a88e0675140ed1501c : SUCCESS in 6m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/a06feb3b5c9a4f988ba5a21cffb531dc : SUCCESS in 39m 40s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/61e16b83cba04f7f8d492181bc990d42 : SUCCESS in 39m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/46041dc364764935a25c357e3c3f8cc4 : SUCCESS in 18m 01s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2a3cfdadaf874449b62df0bb409b1300 : SUCCESS in 11m 06s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0f377c39ed78450ebe9e9ae4c743dc6b : SUCCESS in 20m 59s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c3b71cd5e0ed46f683b9c838661074e1 : SUCCESS in 23m 31s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d71539ab3424434aa8e94736bdfdd522 : SUCCESS in 7m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/66f25713eb424d1598c7f34b3d0ea808 : SUCCESS in 24m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6d2394613065472097d03784bcbc85f4 : SUCCESS in 19m 48s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5b6d877f2a77478084209d2ceeac7382 : SUCCESS in 23m 53s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/d640ca464eec46d1be23f8ca5f6dd6a9 : SUCCESS in 27m 00s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/98f9b6daacac44709c1fdfcf77216ed8 : SUCCESS in 23m 44s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8ad63d46cc75451ab5834809cf3b1958 : SUCCESS in 16m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/01608f7533a042a2952926c22b6908b7 : SUCCESS in 19m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fb3d9212de854264b1467c7b32df59da : SUCCESS in 20m 59s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c8d3b77df379436c900ec177734f79e5 : SUCCESS in 15m 13s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b054763c7b6749b9924aac6b9ba27ee1 : SUCCESS in 19m 25s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/510decc3328f47289299f2cc66bc5d14 : SUCCESS in 15m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/62d6f7b8d8ec43328bc90a03c0dce0ac : SUCCESS in 6m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/12383d719eba4e07ba6f4d878be20806 : SUCCESS in 14m 38s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1c08f4bd49444ec6951f1111dbeb374f : SUCCESS in 6m 51s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/fcfca5d40f3949c586a2ea618b2fec46 : SUCCESS in 45m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/640b6d6f07b340e4811b58d53b02bb94 : SUCCESS in 37m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7e5552e795c24433acab4d74c8822731 : SUCCESS in 45m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/067e212d4be8497eb026843f43432882 : SUCCESS in 20m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da3c40b8334a46a6a17ffc3e5897d9a6 : SUCCESS in 4m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1d5ef56b319945d280a6903c1c62f89f : SUCCESS in 18m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1a3da239b97e4ef197a77eabdc93c3f8 : SUCCESS in 1h 01m 47s","accounts_in_message":[],"_revision_number":42},{"id":"80706fc68e8bfa0798e52c4502a3312ced4180cc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-23 20:45:50.000000000","message":"Patch Set 42:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1706a26a2e8b4f048b8944a8b103a7af\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cb91e5c2573d48cdbb75f825a27ec1cc : TIMED_OUT in 2h 01m 05s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d03cee1af363494aac6f2f8842fb9a52 : SUCCESS in 1h 50m 33s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/56812a5bf97749f4961ca8f43a0843d2 : TIMED_OUT in 2h 00m 54s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/866dba4a5ca344b4ba08632205890223 : FAILURE in 1h 36m 46s","accounts_in_message":[],"_revision_number":42},{"id":"f5f6e527d8c8c70be5c0c158824143bfd8ef258b","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-09-24 15:22:02.000000000","message":"Patch Set 42:\n\n(2 comments)","accounts_in_message":[],"_revision_number":42},{"id":"f82a9d101515c896e0dc271f58e6f0beffd2ba18","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-09-24 15:23:47.000000000","message":"Patch Set 42:\n\n(1 comment)","accounts_in_message":[],"_revision_number":42},{"id":"5477b210d1994224f5ddf842051d301d67ae2177","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-24 16:28:29.000000000","message":"Patch Set 42:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8884bb5aa8d849d1b166f7cb6a06da0f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a45d561043fa4528902647707998e225 : SUCCESS in 7m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/4f693fca5d5046b9a469edd653b5c325 : SUCCESS in 52m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a9e458f6998a4e6c8450bdc4bcc86a21 : SUCCESS in 43m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e96659f429d248d39e7e14feaa13b267 : SUCCESS in 29m 13s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/499e0de0f2204ea5b6405cff6a971d47 : SUCCESS in 17m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/60b062d134fb40f6aeb3f10c8eb2f161 : SUCCESS in 18m 42s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9bea528dd8604c7abb5084a394ebe8ee : SUCCESS in 17m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/42eb58df8e9c4fb5a0efeec53b4df52b : SUCCESS in 6m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c415f9046b6f4fe0b68094ce8f687214 : SUCCESS in 19m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fc7743a6fb794660bf3776dc1991f3c2 : SUCCESS in 18m 52s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/05197f72fa4543ceb9caadf221b08ed8 : SUCCESS in 18m 29s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/98622d26b3de443cbcd6d9432033ae35 : SUCCESS in 23m 02s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a86a3b299b8547c3b192225f522dcb25 : SUCCESS in 26m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/82b8821a5c214169bbad8dc04582f576 : SUCCESS in 19m 52s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e1c98f7d81c14d80816cded68a56cdc7 : SUCCESS in 17m 41s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a9b6df2f17ad4b1793138b9cb84c5413 : SUCCESS in 19m 34s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6218838757214167ab249c373e5ea953 : SUCCESS in 15m 12s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/813b97dda9044f9481242a6f35e53caa : SUCCESS in 17m 28s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f44df6d15630423d8a609c971b1c0e8c : SUCCESS in 18m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a30b7300d0af4b0c9eb55820cd30824d : SUCCESS in 6m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4aec2abfa1a641e6a83cc19c4976c2a0 : SUCCESS in 14m 13s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4b2e68559a814ccf8f1ebe747f076383 : SUCCESS in 7m 18s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1b948a3130d1481e82480dcef3389bc5 : SUCCESS in 48m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/aa8ee3bed3734c07be8c7129823a95e4 : SUCCESS in 42m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b3bda35bc041491696fec7e3f3052a48 : SUCCESS in 49m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/81a5e3c2942b451683892147c735f355 : SUCCESS in 17m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af4fa74d5b6b46e38df66eb9c18ebbd8 : SUCCESS in 4m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f5514305e5494eeba470c388c01a98f3 : SUCCESS in 20m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3dcc495fd68d4776922b128dde0d8317 : SUCCESS in 59m 58s","accounts_in_message":[],"_revision_number":42},{"id":"7542dd63e243f61d1fc2cd981cba655a1391aa02","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-24 17:25:02.000000000","message":"Patch Set 42:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8c3a8a90a4f468793c2854ffa6d3b6c\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bab56dd43f464d6f8f8875e9df39a03c : SUCCESS in 1h 25m 41s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/846270d5949943f19d8e999dec3f562e : SUCCESS in 1h 51m 49s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b27d0323cd17434881c81aa8ee2801c3 : FAILURE in 1h 37m 04s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3ed979c274c7426e84affcecfddd21c9 : TIMED_OUT in 2h 01m 03s","accounts_in_message":[],"_revision_number":42},{"id":"174eed34b95c8becd0ffc605cbb916e99f99a6f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-09-27 04:13:24.000000000","message":"Patch Set 43: Patch Set 42 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":43},{"id":"b0d87f28cbe2c6b4925bfb9e2eb3d8008d424fbc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 05:18:28.000000000","message":"Patch Set 43: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d0374891fdd4977bf4df1bfb0da3e0e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efdb649ed16c424e8fddcefd539fc020 : SUCCESS in 8m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/6d32a1ac07e542f8b29977310429a858 : SUCCESS in 38m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2cb18b6075d241c3b97086608a6537f1 : SUCCESS in 56m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6c1a8729cbf24a4194ed2d5a7f7c5c60 : SUCCESS in 32m 13s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f46859d059b748f4bf5712fd1377922a : SUCCESS in 11m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3b05dfd13e6c448f8b38df625d253000 : SUCCESS in 18m 30s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9f4eaba7743d4165a51fb9eef13c17c9 : SUCCESS in 17m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/def5228f9a1e449a8c4e058eb215103e : SUCCESS in 4m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ccecf9c230cb4bbebe4904144894fc6e : SUCCESS in 15m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/448c814af8be4d35a7ffe5de39989e16 : SUCCESS in 14m 48s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/84aed4558d9143c6a5c78de38b69f260 : SUCCESS in 20m 56s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/2e971da0dd6948f68a48421aad6c3388 : SUCCESS in 10m 44s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/80e0518439974701a001883d14ca3be1 : SUCCESS in 29m 20s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/641218594bcb41baa9d109e8c794dcc9 : SUCCESS in 14m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6a5406e4bd19451b83bc45e476a16b7b : SUCCESS in 14m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3991ef91f5e6428d954e1f40bdf54d40 : SUCCESS in 17m 25s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0fd215c7100f42d0961be7472f2f0d5e : SUCCESS in 13m 13s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b1587be8c16047a6b05aca466aaa7c2b : SUCCESS in 14m 22s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/45f362d7358a4c29a69277d9c3c787a6 : SUCCESS in 15m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b6cfe839297246cd85c0e80d4a1c0b50 : SUCCESS in 6m 04s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/24c8bb9b7c8a4720bbc2005b647b6a2d : SUCCESS in 13m 00s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f0367122864a470ebc50810c751c4601 : SUCCESS in 6m 24s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2cf370b1ef29421fad233783e5e122d2 : SUCCESS in 39m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/50be753573ea480cba090a277cd6c61a : SUCCESS in 34m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5042492228fc4295ac2bc1be620abf83 : SUCCESS in 48m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2b550c75f2ef48b691bbf4494c8fcd5a : SUCCESS in 10m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/649d66e181a14b74a6f99a2305af9837 : SUCCESS in 4m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dabe85563a2c4104802110495f7fe5d7 : SUCCESS in 16m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2a3b691cc042490f8df328fea9d63d29 : SUCCESS in 59m 25s","accounts_in_message":[],"_revision_number":43},{"id":"f3f36cfb5cf01d23c2884de633e208afe944308f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-27 05:55:24.000000000","message":"Patch Set 43:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c256d4fa60664c3d94c7da62866e3ae5\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1d27354b3e7f4fd1a7639e3080125df6 : SUCCESS in 1h 40m 09s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7158278c71af487b81928afba292d8b6 : SUCCESS in 1h 31m 27s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/62cc297f89874ed6a4ea675aa3c2d870 : SUCCESS in 1h 04m 39s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/cd7d3f1464a74af58d95e2b23661c72c : SUCCESS in 1h 37m 55s","accounts_in_message":[],"_revision_number":43},{"id":"057c16329248743efdc6382083f8a9a0b4d91f6e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-03 15:12:32.000000000","message":"Patch Set 44: Patch Set 43 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":44},{"id":"d8199a2ffae49d8c59a16daa71d96dcdbbd62533","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 16:22:44.000000000","message":"Patch Set 44: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2f9ab8d002248b5bedad43ac5f5ff4b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35b1a5c21e044b338cafee19f512aad3 : SUCCESS in 8m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/4a8b47f434ac47fbac4120d55036b2e9 : SUCCESS in 1h 07m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/81272f9ee61245779a6917bc2d850b72 : SUCCESS in 39m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/962ad7480f2e484cb377d819b8090ad9 : SUCCESS in 46m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/39a9373480624c2fa474fb7fada0b569 : SUCCESS in 23m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6e96ab7532084f45a8697a8aed910e43 : SUCCESS in 21m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/83a064e74f244d34b6101a5d8d202e34 : SUCCESS in 17m 02s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2a8d13ac03794c5abb58f3ee9a20fcfc : SUCCESS in 19m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7978727853b0439ba3e6fb2bca96a525 : SUCCESS in 6m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a9ecab35c1464d38b2907d0e73374d89 : SUCCESS in 19m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1207526087084989b7b30850ca8ca771 : SUCCESS in 17m 57s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d06169c8505d4d498a327b9d8de53f79 : SUCCESS in 20m 18s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/a0ebea4190f24f4a860deafa8fae769f : SUCCESS in 18m 17s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/07177f9777184490baf47be0fe3ad9ef : SUCCESS in 31m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0fa6392eeb064006a1daa5f4dcf99fa3 : SUCCESS in 16m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4dd3f34827ec4190ab3d8a39477ba302 : SUCCESS in 18m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5ab174731f144588a67d47b761b6480c : SUCCESS in 18m 02s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f80181d795874c33822bffd64c132b21 : SUCCESS in 14m 21s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0c17c1bdeca849bd89e3d38a3f3fc78b : SUCCESS in 15m 51s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4948ce4a0e88418a8c5f29a5fa282116 : SUCCESS in 15m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/086efe068b04464a89898eb8abdd9501 : SUCCESS in 8m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5a33bf3efb8a497cb6002e9f542b51a5 : SUCCESS in 14m 26s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/783120bc74494afaa43412cda2ce9521 : SUCCESS in 8m 21s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e690fcc1e23f440e869e977100a6199f : SUCCESS in 52m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5567c2b7da0842889dda044993db079d : SUCCESS in 48m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8c15de7e430a4cec8112bb489eb7b977 : SUCCESS in 40m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/98eb0570d68c4a1fb853f89160a7e604 : SUCCESS in 20m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01c217d3f20d4e1592309371748f22c3 : SUCCESS in 5m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a7a67cc4e8ff4bbeb89ab326bec52ff2 : SUCCESS in 17m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c0d916446c2940a3a57040bb07995ede : SUCCESS in 55m 16s","accounts_in_message":[],"_revision_number":44},{"id":"d4c45cf34a7b1c3a0f27eecfa4192fb88f92d4ee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 19:07:41.000000000","message":"Patch Set 44:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/742a9cbef38441828be1c35b96ca38a2\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7c4670272f1844fe959b1690c3558811 : SUCCESS in 1h 32m 18s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7acceab0c1314a7893556c9e9cef8b0d : SUCCESS in 1h 20m 04s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a05b6ca5e1ff4763ba6beeb0ab9172d8 : FAILURE in 1h 41m 49s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/faf1f6425b3f41cfb662d737fd3c8af8 : FAILURE in 1h 10m 42s","accounts_in_message":[],"_revision_number":44},{"id":"1e3301512c0a823c82ba7046c7456a8d7c21d0ba","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-03 21:04:08.000000000","message":"Patch Set 44:\n\n(1 comment)","accounts_in_message":[],"_revision_number":44},{"id":"d7c25e54ccbfaac44d0831abab51f949ccbbb20b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-03 22:00:36.000000000","message":"Patch Set 44: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":44},{"id":"c1d99a76cc176e95af431b629313ceb9ff17e2f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 22:15:38.000000000","message":"Patch Set 44: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e464db12d34146918385cd4da50465fc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f1e7de1f2ea4cbfbde4c6ee78d7bc0c : SUCCESS in 9m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/d804eba535f64533a31c262b966bf2a6 : SUCCESS in 1h 06m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5c7000a3748c4da6bf6eaecef5d3dd88 : SUCCESS in 1h 02m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f5dd11c8f981447794f676556e25f2e0 : FAILURE in 46m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c6cdbfef2a9841f290c7e69da5c03584 : SUCCESS in 1h 01m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c58a7d9098ca4b91b85b597d6c14e20a : SUCCESS in 10m 44s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d7635115b49c4f788fd651922739916c : SUCCESS in 17m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7bd1519574d74a89b09dd5f286cb095e : SUCCESS in 16m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8fdc36f6ae494b938259f077ed60f142 : SUCCESS in 5m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0ba556d48c294614988485e536fc3e38 : SUCCESS in 11m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/87fba8d766f04c87957e1e3123e223d9 : SUCCESS in 15m 07s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/fa13057a46c642e799b2e326f7b92ab8 : SUCCESS in 17m 12s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/b461a1ad63d54d39815ab38baf292401 : SUCCESS in 21m 21s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/250b6d98ae804589a4fbc002eb74c8b9 : SUCCESS in 19m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c1006f3ff9434ddc95d3877169c3bf80 : SUCCESS in 18m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f129a77db8bc45d5ab23d66bd5e383ca : SUCCESS in 14m 25s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b575d04913ef4381bd427e472f38060a : SUCCESS in 20m 09s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/17083346e4df45508b0b16182eac7539 : SUCCESS in 10m 43s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/34aebea44f9c4c1ca61a8b6b7a643556 : SUCCESS in 16m 59s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0f8d586d03e8425382b2066f67fcd700 : SUCCESS in 13m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c9599fda57ea4e4191327eadefe3a262 : SUCCESS in 7m 17s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/467b5236b37c4b91bd4a94d16f28c5ee : SUCCESS in 11m 27s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/620c9ab26c524ef0959f29d25aa23782 : FAILURE in 9m 09s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/57077b1b7fdb4bd2897c95d2e5cfbfa8 : SUCCESS in 48m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/108603e802d84a43a11992a5a2f2f5d1 : SUCCESS in 50m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ba7ce597352c41a48edf58520b7c6957 : SUCCESS in 33m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c934c1027b7c42dd8cc58157d6219215 : SUCCESS in 18m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e74a01a2762c4150a09744c3fcaf1c5e : SUCCESS in 5m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f40959a5738f4a5c98bbcc8327a58bc2 : SUCCESS in 24m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1c8f4e7f8ec74354a2587a4d7d4857fb : SUCCESS in 44m 28s","accounts_in_message":[],"_revision_number":44},{"id":"44af2ec347a7e54d7584f6a6d5ed30d4d07910ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-03 23:49:23.000000000","message":"Uploaded patch set 45.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":45},{"id":"91f11506a0dc59242ed94be43831023afbb14f77","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-03 23:49:35.000000000","message":"Patch Set 45:\n\n(4 comments)","accounts_in_message":[],"_revision_number":45},{"id":"5a8dcf037c5e23b331d7d38efc0dbe13859eab49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 00:56:20.000000000","message":"Patch Set 45: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cf6a518616334d2f8e71f74733894ae8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/825cdee7280c4ed1a75e8b80172b64ff : SUCCESS in 7m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/7901a855d887453eb252fd679bbbfb71 : SUCCESS in 49m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4d7a850645724a5f99f5332cf2ea26e3 : SUCCESS in 48m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3ca5ceb20094435bad5a99c96a677064 : SUCCESS in 40m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8d299bf1930f4e3faf8e0eac3e7c0693 : SUCCESS in 52m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e573b36215e844b092f2e28ccf77f0fc : FAILURE in 10m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1194f26c70c14f71b3c74d98a0380719 : FAILURE in 20m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7331891f17f04096a724b1b638352d98 : FAILURE in 19m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f9a2888f5ac64807adc2143d862370d5 : SUCCESS in 5m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/faef0a308db341d89593aa9c63842100 : SUCCESS in 10m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3be8df5e3ac849b5960dd853a88ac171 : SUCCESS in 19m 47s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/712e42cd654f4a939423f764e43f8f3a : SUCCESS in 22m 33s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/e25c95031ff847c598e9eb958ef64088 : SUCCESS in 17m 23s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c534e9fdd994482eaf79b8180726afe2 : SUCCESS in 21m 20s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/938b7d63544141c2bcbcb8e981805b92 : FAILURE in 10m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3924ae94d8f1477f835344e6b8515f04 : FAILURE in 20m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f03228caf10a4c3c9a81c236869404bc : FAILURE in 21m 41s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/44521b6d0efc4a559dcb994504f2c799 : FAILURE in 16m 03s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5d684fa1504b43ef8fb4f11c2991f06d : FAILURE in 9m 17s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1931205c5fe54f2dacfe7a0fdbfa68c9 : FAILURE in 19m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c2cf6a2c41ee4b05a5de1037820524a6 : SUCCESS in 4m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6160ea43bf2b40c1b926c397b190bbce : SUCCESS in 15m 23s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/69a43b2a8ce74ff78a213033d9b53f9f : SUCCESS in 4m 47s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/884a11ad360c4d43ae916af16ffff450 : SUCCESS in 50m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/45f7afb05390438db70f47107cc70456 : SUCCESS in 29m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5f07699cd7064f308f0c62a37ab4f0ab : SUCCESS in 41m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6bc00443d85e4f84aed9c0abc2a0eb35 : SUCCESS in 20m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f595f185b5e54c59a5bd9ffde4dfbde0 : SUCCESS in 3m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ff3ae00ca8494923a3868e1369a5ef38 : SUCCESS in 21m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ed68713e524e40e9a8fbf76efd064323 : SUCCESS in 1h 00m 14s","accounts_in_message":[],"_revision_number":45},{"id":"767c5ccd4616d7687d97f673717f180c3cc58c80","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 02:32:32.000000000","message":"Patch Set 45:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc1b7c620769430488fd8ad3e93275fe\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d50d7fc5cf394e9dbb19105d04893e58 : TIMED_OUT in 2h 01m 00s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0c70bcdd494a4eaba1d97e3a325d8f22 : SUCCESS in 1h 56m 11s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/823212bc3aa849dfb816e0143b335fff : TIMED_OUT in 2h 01m 02s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a127bfbdfb9145de94ae21026b1fa76d : TIMED_OUT in 2h 01m 08s","accounts_in_message":[],"_revision_number":45},{"id":"f16a2fe98378f0a0de148f5e683b818f92116f38","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-04 04:24:38.000000000","message":"Patch Set 45:\n\n(4 comments)","accounts_in_message":[],"_revision_number":45},{"id":"df26888f4c1d8e0e6b80fcc85f913fdf418148ca","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-04 04:27:42.000000000","message":"Patch Set 45:\n\n(1 comment)","accounts_in_message":[],"_revision_number":45},{"id":"ee4c01dfd83829873edf117833b626988659d9b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-16 15:27:16.000000000","message":"Uploaded patch set 46.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":46},{"id":"eddef711bbf65d90a513a71a26dfb5d109c0c897","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-16 16:09:29.000000000","message":"Patch Set 45:\n\n(6 comments)","accounts_in_message":[],"_revision_number":45},{"id":"471884149225972bd7d9526414ac4de3054950dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-16 16:49:51.000000000","message":"Patch Set 46: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9dbe584ac47044a388c82512a851fce2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7207c6fb224348d786e66e8ca88319b4 : SUCCESS in 10m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/5c4540e87c79419ba4fb816f64038f2e : SUCCESS in 1h 09m 29s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/843c75bebcea409d9f1532337b4f2b8f : SUCCESS in 1h 13m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b8e309b138054d059a64d1b75c9982ec : SUCCESS in 54m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d05960e7574a44319e13284b3cc2eeb4 : SUCCESS in 59m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9c36e95f11214c3a908a1592dcdcb97f : FAILURE in 15m 44s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/17d76533c96e498d8bdc880000e3ed1a : FAILURE in 21m 55s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c072fe185c414d06a8047ef5d82a7d45 : FAILURE in 19m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/65baae9330a34b079df8cd2ee1eb45da : SUCCESS in 7m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/717b5418ce14496f8f0bad49a09dd06f : SUCCESS in 20m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d42fe3c91e78416bbca165e03ead9c9d : SUCCESS in 20m 17s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5ef97a6ff08449c1850d9661f95cc906 : SUCCESS in 21m 46s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/1cd7c8c7bf4a40aaac4f2a085c8f2d1b : SUCCESS in 18m 53s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4772d3e0fb324c828e21120891154252 : SUCCESS in 29m 29s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/642e12931fe34743a024bafbb24662ed : FAILURE in 15m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4d956b9bb24c45f096f4c8430bcf4e9a : FAILURE in 14m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1e5bc9443bb147a58e34a86ebe0f5157 : FAILURE in 19m 55s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/33a662aaf8cf4adfb9dfb5ee3abb7ca2 : FAILURE in 15m 57s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/972c8bd262454ca4a12d6e2682c67c5c : FAILURE in 14m 52s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f4e61b4e6f0a4d48939c71860a2d8b6a : FAILURE in 14m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f9f1fb47041344f38a4219e6cc5f72e4 : SUCCESS in 5m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f881936d9c20489f86a8e3c1f68c1157 : SUCCESS in 13m 27s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6f7562abb2804a1aa8258bdb92fbdced : SUCCESS in 7m 39s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e7150b6065cd4f20b04675cec3162c83 : SUCCESS in 59m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b67c201096994cfe8ca424144bae9102 : SUCCESS in 40m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b00ff7890d69485bb699ab1f615fb0fd : SUCCESS in 37m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f4fe600536394616b1f7be720807cd75 : SUCCESS in 16m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad706dc82e0a479bb59343b4887b75c7 : SUCCESS in 4m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/df175ad4c7934994884fdb9ef529dd4e : SUCCESS in 16m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a52e1defb6414a1c8a2c7052a031b7c4 : SUCCESS in 45m 09s","accounts_in_message":[],"_revision_number":46},{"id":"1d858148b66879a816770978b6cbf12650881fbb","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-16 16:58:27.000000000","message":"Patch Set 46: Code-Review-1","accounts_in_message":[],"_revision_number":46},{"id":"1dc88a5b3fb23fe2d45d16a39fc716abc1880e51","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-16 18:37:42.000000000","message":"Patch Set 46:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25bf1cd99cb64defa1dfe71266e1384e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8a5efd66b8bc46cd8228e406d377392f : TIMED_OUT in 3h 00m 54s","accounts_in_message":[],"_revision_number":46},{"id":"8ece2e6d9c4de4e3da8c29eebc1bf46a66b516d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-16 21:00:05.000000000","message":"Uploaded patch set 47.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":47},{"id":"c6dea4aa4e35633e7c7ca3b8fac7073b3e4ce450","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-16 21:00:53.000000000","message":"Patch Set 47:\n\n(1 comment)","accounts_in_message":[],"_revision_number":47},{"id":"95e112b35583c038ff96e7a069dd8bc2bdce5fdf","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-16 21:01:36.000000000","message":"Patch Set 47:\n\n(1 comment)","accounts_in_message":[],"_revision_number":47},{"id":"fe214614eed3ef6f95a85365e97db0dff238195f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-16 21:06:05.000000000","message":"Uploaded patch set 48.","accounts_in_message":[],"_revision_number":48},{"id":"b747b710d0af288d8862101b0b8461e7b8ba1575","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-16 22:01:39.000000000","message":"Patch Set 48: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fcc362c2f58e4577a79b3021f5c2d26d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f368ffa9e1e44529337943826183fd8 : SUCCESS in 7m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/8510c54dfcfc4865829842e50f268d0d : SUCCESS in 47m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d8b764e4eaff47a79c9f1b69a0a53e39 : SUCCESS in 51m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d699caa0bc924c468fbf81e598b4e483 : SUCCESS in 42m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6514ac2828c74511ac77c01a97b2c3d5 : SUCCESS in 49m 05s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/33c2cc400c4c4bf18a7eb97f48a12c1a : FAILURE in 17m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a82dc872b0e748868f193812b78bc534 : FAILURE in 19m 24s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/53fbe49dd66c45dcaa738c36e192b779 : FAILURE in 18m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4caa8b4491c148d9beb9bb70c7d89a7c : SUCCESS in 5m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2aeb6161805a4ee3bf9e06796397157b : SUCCESS in 20m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e1874c56314e48edafcc5b9f735e145c : SUCCESS in 10m 03s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a74b0bbe354345dda8e09fa105fc0302 : SUCCESS in 21m 07s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/97111c9d2543496dbf7dfdd68dd8fbf9 : SUCCESS in 10m 42s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4e4851ab9938433abe83e03feae1570f : SUCCESS in 32m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/771949469ad14f259f124a0aa6082b3c : FAILURE in 11m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fde05e6602a14e68800b629f7e5d8b7d : FAILURE in 11m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0c3de757daa64daa8a161e4ae17bc2d1 : FAILURE in 18m 20s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/eeed77a4b9bb413abc73cc68ef4d3248 : FAILURE in 15m 47s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f4c91b6f3c324e7ebdcb31caaedf571a : FAILURE in 16m 12s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/9966b955794d45ab815e5bf7d8d6f0ec : FAILURE in 15m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8358bd16968d4aff857c6b4e50617485 : SUCCESS in 6m 17s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c3a2dce366014dd68dba8ee21baa9b6c : SUCCESS in 15m 14s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c058dc84f86b4ffabe2f9ef185d2e2df : SUCCESS in 6m 13s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7f90baa29cee47aeb419fb81c63dd189 : SUCCESS in 54m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/95a095fb77934293903c9a6ec1fcee31 : SUCCESS in 40m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/166aa882723c4d4e8d538b3d1c204652 : SUCCESS in 42m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c0653eb83d1642b2aa7571917e52b33a : SUCCESS in 17m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b5eaaba5dc3427c97ee7465bcf0bb3d : SUCCESS in 3m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/77d75409f1cd41599c2f1866fcd4bfe0 : SUCCESS in 18m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c725109fbce34f76a27bd9ffb3dd2eaf : SUCCESS in 41m 23s","accounts_in_message":[],"_revision_number":48},{"id":"5338e1d6e55ee0ac6113f4f1e35ba9439686a50f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-16 23:30:48.000000000","message":"Patch Set 48:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f83f6b758a594d3c8d400442857f4d27\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/125154b3c43a42e9b18e49e683a55a70 : SUCCESS in 2h 21m 52s","accounts_in_message":[],"_revision_number":48},{"id":"e535260a34e020f88c71db5b38a77ce855157896","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-17 16:39:36.000000000","message":"Patch Set 48: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":48},{"id":"54140837b0955e6dd61099ca44aba8f77625e31e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-17 18:19:45.000000000","message":"Patch Set 48:\n\n(3 comments)","accounts_in_message":[],"_revision_number":48},{"id":"7308e2f2ba6b7e6dfc34142b5cb7b414fbe51e29","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-10-17 18:43:53.000000000","message":"Patch Set 48: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":48},{"id":"c00e306ed66904858f66819a3fdcd135d9d4f12f","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-17 20:44:22.000000000","message":"Patch Set 48:\n\n(6 comments)","accounts_in_message":[],"_revision_number":48},{"id":"d23fae9ff9a5cb39c5bb80f29584ceb667dc8ad1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-17 20:45:23.000000000","message":"Uploaded patch set 49.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":49},{"id":"d742f321c152b8465d0b54d06ebb6067b9393eec","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-17 20:45:32.000000000","message":"Patch Set 48: Code-Review+1","accounts_in_message":[],"_revision_number":48},{"id":"3ebabf5f4f4c2af6de51228fba6ac44a52a9f100","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-17 20:46:33.000000000","message":"Patch Set 49:\n\n(2 comments)","accounts_in_message":[],"_revision_number":49},{"id":"39edd01f18fbdc451694b1ba2bc434e6c493a7b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-17 20:47:51.000000000","message":"Patch Set 50: Commit message was updated.","accounts_in_message":[],"_revision_number":50},{"id":"61702992e8e1b63dd09a8d634909c0eb11aa8b81","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-17 21:11:48.000000000","message":"Patch Set 50:\n\n(2 comments)","accounts_in_message":[],"_revision_number":50},{"id":"2a1b6a83bc40873619292c03d0a6607b237e772d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-17 21:30:18.000000000","message":"Patch Set 50: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":50},{"id":"d030a9994d01f24ff778bea81dd87e4c43895e7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-17 21:35:38.000000000","message":"Uploaded patch set 51.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":51},{"id":"975c3bc65b71deec5af7310a90cd5a0af0a9f6f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-17 21:39:26.000000000","message":"Patch Set 52: Commit message was updated.","accounts_in_message":[],"_revision_number":52},{"id":"95d4581153ec675f5e7e079f4b3a91f377b8a3d9","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-17 21:39:51.000000000","message":"Patch Set 52:\n\n(1 comment)","accounts_in_message":[],"_revision_number":52},{"id":"93d6d1f8e9d1c03217520a8b668aace6e85b0c73","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-17 21:46:51.000000000","message":"Uploaded patch set 53.","accounts_in_message":[],"_revision_number":53},{"id":"ad918ecbbe817172f03322d12aa27561b066089d","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-17 21:47:24.000000000","message":"Patch Set 53:\n\n(1 comment)","accounts_in_message":[],"_revision_number":53},{"id":"fd81fc4596805a25e2c23dd371c43b4887b257a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-17 23:03:44.000000000","message":"Patch Set 53: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b25b56af654469cb1b7c7aef51c2b16\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cac12bbc2d434240a37c892646668ee2 : SUCCESS in 8m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/7e2625644cef40bcafa3a729c18a612c : SUCCESS in 46m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/00c4449e60b14119a41822965f59e1bf : SUCCESS in 59m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f07e08b554ec47ea8726702e5a1107b2 : SUCCESS in 41m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c4eaaab7747249148368776afd556793 : SUCCESS in 49m 04s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/390729d094e345088452e95fb86baf1d : SUCCESS in 19m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6cc663c5ade442199233c4494191a5cc : SUCCESS in 21m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/69f15b443e0742bbb9d0a2005268327f : SUCCESS in 20m 56s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/313e58ff09b8426dab3506035ed51cbe : SUCCESS in 5m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/149e14af98e04f2f8aec81c1c8a2a0b8 : SUCCESS in 22m 54s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/44022848488f4f2183355971f280ef49 : SUCCESS in 19m 19s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c5f583fb155543d19324dc43ac45b010 : SUCCESS in 21m 00s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/4b9a968d0b6e41a399c15dc6422f6ec6 : SUCCESS in 26m 52s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/446e0fa4a0df4b2cbe8535b64c06ee04 : SUCCESS in 20m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3144461a2f0b4cbe8ba93982bc35e914 : SUCCESS in 21m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/501ed005b3c641b99e8d05ae5f8da1cd : SUCCESS in 17m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6bf03c3f9b9945259e712a70e530b569 : SUCCESS in 11m 57s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/cef19b0c4bcf4fe2b6ad608fef00224d : SUCCESS in 15m 10s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5dc7f007d0cb48d0bbef2a8239c32e2b : SUCCESS in 17m 54s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a3690c4d632841e1b58a9f90318ecbab : SUCCESS in 10m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1a57a9445d7247d7a97978a7abde1ff9 : SUCCESS in 6m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f904cd7769514264afb5476f02cc3694 : SUCCESS in 17m 55s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/aff6fda4ffef4b52a0dee375f461f777 : SUCCESS in 5m 32s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/16233cfabb03416aa69d7ee63bcb5905 : SUCCESS in 48m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b1147ee351aa42a6b41d4f9a21b17b43 : SUCCESS in 49m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fd3e40d9754c4bcb907531b963c503c9 : SUCCESS in 30m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7dfbb8c1bb974b2cbb8f12b5612bcd86 : SUCCESS in 23m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9db4a5ab20474ef6b7851a35bba4d207 : SUCCESS in 4m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4ea6dc26c5e84fc3ab06c583cae091da : SUCCESS in 12m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/59ad30c545704ff1b0c84b163e559e57 : SUCCESS in 56m 59s","accounts_in_message":[],"_revision_number":53},{"id":"af9f2f19fb99dc49da26bb02a93bb55bdb7da12b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-17 23:05:40.000000000","message":"Patch Set 53:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9548924575d4770b6169ea6ddd4aa00\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1a9e2184b8234dd289da28c65c582cc1 : SUCCESS in 1h 16m 41s","accounts_in_message":[],"_revision_number":53},{"id":"ef78e3ea63452fd5dd90297cc0d85af9d2faf2ae","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-18 16:22:06.000000000","message":"Patch Set 53: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":53},{"id":"da147e8bab6c577213a65db42e2ba80952ec990e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 16:22:43.000000000","message":"Patch Set 53: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":53},{"id":"8e2f4f4836e775e416791bfa59ab5e437ab2fce1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 17:38:56.000000000","message":"Patch Set 53: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/289a4f3d46dd4436a0b4f6b20bdfac24\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/027ff59cb4104585b1427c3aac7c1883 : SUCCESS in 8m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/af6d464c3c6843a4b4f5bf6b6922f389 : SUCCESS in 1h 06m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d177e440eb234eaab2bae552b8caeab8 : SUCCESS in 1h 03m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5b4ab8d5b9ca433d9d2f30bb50b51a0f : SUCCESS in 54m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1b3a3f2b87f144e49b08b2075b50e9d7 : FAILURE in 57m 09s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/644fb3f4bf8a4ff1a00767dfbc0f2ac6 : SUCCESS in 5m 54s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/948b23d2548c4a138ebb17fbaea9bf58 : SUCCESS in 21m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/85d6834cbeec4ffa937d414fd917a032 : SUCCESS in 17m 53s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1b7ae03ffcd640c4a10c7c18c51ee293 : SUCCESS in 19m 34s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/157acd84a4ca4576ac159589a521209a : SUCCESS in 15m 47s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1570a05996724875935dcc4466e8eaa6 : SUCCESS in 18m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ab2e31f93aec411b8ff9cbf30ab0ba0b : SUCCESS in 15m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a132f5e9a7a242feb328b686e7b42cbb : SUCCESS in 16m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e8a15c789e1047db87a191afe1646535 : SUCCESS in 16m 27s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8339d8a978144acfaac336bbce46712c : SUCCESS in 8m 15s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0eb1cee20f6746fd9cc3cb5624066fdd : SUCCESS in 8m 46s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7f6fa7511fbf41c69baec9a913549b95 : SUCCESS in 15m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4afeb8e5bbb5443e81328b64370b99fb : SUCCESS in 4m 27s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2b18896b8c204bfa96f858b4ba139048 : SUCCESS in 6m 59s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8b732bad31804d7bab079fac286a1725 : SUCCESS in 37m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/01e7446233b14a9691ba4a356617c3a6 : SUCCESS in 55m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d2af71ab1a2249c488e6607fe67b7f48 : SUCCESS in 54m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4961277d4d4d4d508218b0a1edab4368 : SUCCESS in 10m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0c062d7dec8415bae481a62e5492cd2 : SUCCESS in 4m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/409aba73b2cf4d1ca9a988a8cb222390 : SUCCESS in 25m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2f24167411364482862987cd3264916f : SUCCESS in 50m 14s","accounts_in_message":[],"_revision_number":53},{"id":"4ee57a99e264a34e79fcc1aa74834999a5f8f524","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-18 18:33:09.000000000","message":"Patch Set 54: Patch Set 53 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":54},{"id":"aa037ca504ce69e00e9e4824bfff4ba162513012","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 19:45:06.000000000","message":"Patch Set 54: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/369a8f194c284c16aced0bbd755013fb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0dcf52b248c4384a4750259f13ea164 : SUCCESS in 8m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/5e62e27e39e24708b15fe68073a52e91 : SUCCESS in 1h 00m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bf73c29dd0de4ed0b55b382b371a9442 : SUCCESS in 1h 06m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bd814e8817a84f82966c813dd57b0d8b : SUCCESS in 1h 01m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2a7f13a92ef24cc097db322f20b86b50 : SUCCESS in 49m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a76bee3872534beaad08346d67549c05 : SUCCESS in 10m 47s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/497bc2c71fd348d98f49307ee38537a1 : SUCCESS in 19m 20s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/50beeaf78f5f43a2b1b28362107b97ea : SUCCESS in 17m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/273990bf28b04e8993a72fb136375b27 : SUCCESS in 5m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/15ccc8c077f04480915ae5c6889d70f2 : SUCCESS in 11m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b30817f8313f48c5b116bce76270f5ae : SUCCESS in 17m 10s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1d78e092e10b4514a7a86cb8fd953db3 : SUCCESS in 11m 14s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/552d2f51763145babe559c339e415fd5 : SUCCESS in 20m 13s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f3e257ef55a846c7bf2b837775212b58 : SUCCESS in 13m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/913de98dad1a47b69c98fa2c88058589 : SUCCESS in 16m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e600a9ad6e0b4b6e89eb2ac4a2600be9 : SUCCESS in 18m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/60ef8fdba08147f3aee0cac5dfad17ce : SUCCESS in 17m 00s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5567c17de6814ded9b318841afa28b2e : SUCCESS in 15m 46s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ba9b104226e4491991bf99474787ef14 : SUCCESS in 12m 22s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/264feb39bc8e4d3a9977cc88a7aeb70a : SUCCESS in 16m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c2b8e65e46324c39965048e05c274da9 : SUCCESS in 5m 47s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/edba264ce8824cc399d5bbd906d7d16b : SUCCESS in 13m 44s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c2ff40af851a4d7182d6d5e6563a1000 : SUCCESS in 5m 00s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/becc91a4791f4e11abc8316dbb4858bb : SUCCESS in 51m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/62e1317129ed4e70b459d3ead7a1055c : SUCCESS in 45m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a8ab60837cd94b9d9fe0da4ef362ffd8 : SUCCESS in 55m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1faa5afe7f2f4dd480c1519ebc5b6a4a : SUCCESS in 20m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4cb6a4cb33f41a9b77c2009bf4c6aa0 : SUCCESS in 4m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/67678a0c129d40768517402028aadf1d : SUCCESS in 20m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5f3c972b87fc4f36a51a8a4ba245d096 : SUCCESS in 39m 40s","accounts_in_message":[],"_revision_number":54},{"id":"3c59a00a692326a45ad59d36d749c6f0129d8743","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 19:54:08.000000000","message":"Patch Set 54:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/926cc185ae6f4b9092d68a946b387024\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/72a7863cd8a3436d983b5472963e7eb6 : SUCCESS in 1h 19m 01s","accounts_in_message":[],"_revision_number":54},{"id":"b15db6e4b916461da4c86eddde19adaabc442677","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-10-18 20:10:17.000000000","message":"Patch Set 54: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":54},{"id":"11aa14b5eca01530e13874d130f66549c6bcbe3e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 20:10:50.000000000","message":"Patch Set 54: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":54},{"id":"a0fd979269ffb90257eb9a2929c2f9ad9378a16b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 21:20:42.000000000","message":"Patch Set 54: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92664e30b25544ccb9787a2e97b1de81\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1da791a174fa434d929e9126c5b208d3 : SUCCESS in 8m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/7cb49308dfe04afe8f91571b5da5020a : SUCCESS in 1h 02m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/638b1b0c0f014cf5a35f6219dbb8a670 : SUCCESS in 1h 00m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/79ba964bfaf440da8b26f27a1f6ba7f0 : SUCCESS in 52m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7006f90122e347658877ced80ec5e95b : SUCCESS in 1h 04m 17s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/2ed98688a97a4e4aa12a23ac17770a47 : SUCCESS in 5m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e5bb39679e284917b7b465af0cae5175 : SUCCESS in 20m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/53337e120d934cbea63519bd07b9b9a1 : SUCCESS in 14m 00s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/92a9261bbe004c3db5964f83ea9f0fed : SUCCESS in 21m 03s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/7347c62721d24f65849e3d0206adf034 : SUCCESS in 15m 26s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e2547c2f1b9743879fe848855c30752c : SUCCESS in 20m 58s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d94c55d14f4d44608b65a48ef3f37b9a : SUCCESS in 12m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c6567e5755ad4a0ea6acb85061ef01d5 : SUCCESS in 14m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c241d1fd098c44a2b83b294e59a5b07c : SUCCESS in 14m 47s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2260f58ade4442aeaf2729a7fb7f6086 : SUCCESS in 12m 43s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/cbefc118266c424bbcf580b25f9ba4d0 : SUCCESS in 8m 39s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6935051829bc4a1eb8b7f900bf5328fa : SUCCESS in 8m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/745a51029b2742a083d039201df1001e : SUCCESS in 4m 26s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e72136cee9784519b503ff06a59503c3 : SUCCESS in 4m 41s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ea786c404e894c5d85fd825a1680be9e : SUCCESS in 51m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9e4bfee70f274523afae89cc6f20bd51 : SUCCESS in 51m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/223a6c014a4041138b6a8718594924b4 : SUCCESS in 56m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/39dcea6e38cb4ae3a4da36e561db9b56 : SUCCESS in 22m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81d419e43a25490ea41c6684e0144769 : SUCCESS in 4m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3d6942f73887425e80e27d1d60f124ea : SUCCESS in 17m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/caef022ffae54f83885094a4dae72fdc : SUCCESS in 57m 20s","accounts_in_message":[],"_revision_number":54},{"id":"df67be57c2ae4530d6509c3d13c067e1ffc6be7e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 21:20:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":54},{"id":"994f872cc5dbcfff6fb0774759383dc6b28963df","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 21:21:44.000000000","message":"Patch Set 54:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9639d9aea7af4fd99ac1c4eed780740d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21a6b9f6da144484af6826f7f5202dcb : SUCCESS in 43s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/5ae6eac1cc334d66a7d0b7c88e7e6de9 : SUCCESS in 42s (non-voting)","accounts_in_message":[],"_revision_number":54}],"current_revision_number":54,"current_revision":"4f69ab3c5dfaa8dcd321bdabb74dc079f7973d82","revisions":{"3cee3016db7f083635f3a800711db7a6d8a5d7d2":{"kind":"REWORK","_number":1,"created":"2024-04-19 22:14:04.000000000","uploader":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"ref":"refs/changes/47/916547/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/916547/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/916547/1"}}},"commit":{"parents":[{"commit":"9c44f5b5da11e97e4996b6f34e23ba3d1f60990e","subject":"support x-open-expired header for expired objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9c44f5b5da11e97e4996b6f34e23ba3d1f60990e"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-19 22:12:31.000000000","tz":-420},"committer":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-19 22:12:31.000000000","tz":-420},"subject":"fix x-open-expired 404 on HEAD?part-number reqs","message":"fix x-open-expired 404 on HEAD?part-number reqs\n\nFixes a bug with the x-open-expired feature where the headers\ndo not get copied when refetching zero-byte manifests that causes\n404 on HEAD requests that use the part-number parameter.\n\nChange-Id: If0382d433f73cc0333bb4d0319fe1487b7783e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3cee3016db7f083635f3a800711db7a6d8a5d7d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3cee3016db7f083635f3a800711db7a6d8a5d7d2"}]},"branch":"refs/heads/master"},"8e5c741dcb05acb1244ceb122202f2cc1907758b":{"kind":"REWORK","_number":2,"created":"2024-04-25 16:39:51.000000000","uploader":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"ref":"refs/changes/47/916547/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/916547/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/916547/2"}}},"commit":{"parents":[{"commit":"9c44f5b5da11e97e4996b6f34e23ba3d1f60990e","subject":"support x-open-expired header for expired objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9c44f5b5da11e97e4996b6f34e23ba3d1f60990e"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-19 22:12:31.000000000","tz":-420},"committer":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-25 16:39:41.000000000","tz":-420},"subject":"fix x-open-expired 404 on HEAD?part-number reqs","message":"fix x-open-expired 404 on HEAD?part-number reqs\n\nFixes a bug with the x-open-expired feature where the headers\ndo not get copied when refetching zero-byte manifests that causes\n404 on HEAD requests that use the part-number parameter.\n\nChange-Id: If0382d433f73cc0333bb4d0319fe1487b7783e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8e5c741dcb05acb1244ceb122202f2cc1907758b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8e5c741dcb05acb1244ceb122202f2cc1907758b"}]},"branch":"refs/heads/master"},"abc713c0c5b88e89bc235684b6cac6a8cc2d8ecc":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-05-01 15:56:50.000000000","uploader":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/47/916547/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/916547/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/916547/3"}}},"commit":{"parents":[{"commit":"51e4b6e279c9a4b50cdecbdb186252c676f2eac0","subject":"reno: Update master for unmaintained/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/51e4b6e279c9a4b50cdecbdb186252c676f2eac0"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-19 22:12:31.000000000","tz":-420},"committer":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-05-01 15:56:50.000000000","tz":0},"subject":"fix x-open-expired 404 on HEAD?part-number reqs","message":"fix x-open-expired 404 on HEAD?part-number reqs\n\nFixes a bug with the x-open-expired feature where the headers\ndo not get copied when refetching zero-byte manifests that causes\n404 on HEAD requests that use the part-number parameter.\n\nChange-Id: If0382d433f73cc0333bb4d0319fe1487b7783e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/abc713c0c5b88e89bc235684b6cac6a8cc2d8ecc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/abc713c0c5b88e89bc235684b6cac6a8cc2d8ecc"}]},"branch":"refs/heads/master","description":"Rebase"},"ea445192a6a7e8d8d689a62ea2277f48ae387aa9":{"kind":"REWORK","_number":4,"created":"2024-05-02 22:45:34.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/47/916547/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/916547/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/916547/4"}}},"commit":{"parents":[{"commit":"51e4b6e279c9a4b50cdecbdb186252c676f2eac0","subject":"reno: Update master for unmaintained/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/51e4b6e279c9a4b50cdecbdb186252c676f2eac0"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-19 22:12:31.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2024-05-02 22:45:22.000000000","tz":-420},"subject":"fix x-open-expired 404 on HEAD?part-number reqs","message":"fix x-open-expired 404 on HEAD?part-number reqs\n\nFixes a bug with the x-open-expired feature where the headers\ndo not get copied when refetching zero-byte manifests that causes\n404 on HEAD requests that use the part-number parameter.\n\nChange-Id: If0382d433f73cc0333bb4d0319fe1487b7783e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ea445192a6a7e8d8d689a62ea2277f48ae387aa9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ea445192a6a7e8d8d689a62ea2277f48ae387aa9"}]},"branch":"refs/heads/master"},"925fbca17acf4b30c149789b6aa3fb34f08676e2":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-05-14 18:06:46.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/47/916547/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/916547/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/916547/5"}}},"commit":{"parents":[{"commit":"d1aa735a37d9d679c51a9eda6131ec3e365976b8","subject":"Merge \"backend ratelimit: support per-method rate limits\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d1aa735a37d9d679c51a9eda6131ec3e365976b8"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-19 22:12:31.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2024-05-14 18:06:46.000000000","tz":0},"subject":"fix x-open-expired 404 on HEAD?part-number reqs","message":"fix x-open-expired 404 on HEAD?part-number reqs\n\nFixes a bug with the x-open-expired feature where the headers\ndo not get copied when refetching zero-byte manifests that causes\n404 on HEAD requests that use the part-number parameter.\n\nChange-Id: If0382d433f73cc0333bb4d0319fe1487b7783e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/925fbca17acf4b30c149789b6aa3fb34f08676e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/925fbca17acf4b30c149789b6aa3fb34f08676e2"}]},"branch":"refs/heads/master","description":"Rebase"},"e07df192d5b001b7e38b78ac41dd2ca5d5c218d7":{"kind":"REWORK","_number":6,"created":"2024-05-15 03:03:51.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/47/916547/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/916547/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/916547/6"}}},"commit":{"parents":[{"commit":"d1aa735a37d9d679c51a9eda6131ec3e365976b8","subject":"Merge \"backend ratelimit: support per-method rate limits\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d1aa735a37d9d679c51a9eda6131ec3e365976b8"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-19 22:12:31.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2024-05-15 03:03:43.000000000","tz":-420},"subject":"fix x-open-expired 404 on HEAD?part-number reqs","message":"fix x-open-expired 404 on HEAD?part-number reqs\n\nFixes a bug with the x-open-expired feature where the headers\ndo not get copied when refetching zero-byte manifests that causes\n404 on HEAD requests that use the part-number parameter.\n\nChange-Id: If0382d433f73cc0333bb4d0319fe1487b7783e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e07df192d5b001b7e38b78ac41dd2ca5d5c218d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e07df192d5b001b7e38b78ac41dd2ca5d5c218d7"}]},"branch":"refs/heads/master"},"7d85d3a95ca4fa3e957c9fc6d1d4873d32c699d1":{"kind":"REWORK","_number":7,"created":"2024-05-15 03:05:00.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/47/916547/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/916547/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/916547/7"}}},"commit":{"parents":[{"commit":"d1aa735a37d9d679c51a9eda6131ec3e365976b8","subject":"Merge \"backend ratelimit: support per-method rate limits\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d1aa735a37d9d679c51a9eda6131ec3e365976b8"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-19 22:12:31.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2024-05-15 03:04:54.000000000","tz":-420},"subject":"fix x-open-expired 404 on HEAD?part-number reqs","message":"fix x-open-expired 404 on HEAD?part-number reqs\n\nFixes a bug with the x-open-expired feature where the headers\ndo not get copied when refetching zero-byte manifests that causes\n404 on HEAD requests that use the part-number parameter.\n\nChange-Id: If0382d433f73cc0333bb4d0319fe1487b7783e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7d85d3a95ca4fa3e957c9fc6d1d4873d32c699d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7d85d3a95ca4fa3e957c9fc6d1d4873d32c699d1"}]},"branch":"refs/heads/master"},"c4e8c16b3ada5a1666748e2bc723852dbf5e242f":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2024-05-15 03:05:08.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/47/916547/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/916547/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/916547/8"}}},"commit":{"parents":[{"commit":"35a00caf3a30b260803dc3bd08fb2b86045093b3","subject":"Merge \"Move old rolling-upgrade jobs to point to EOM tags\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/35a00caf3a30b260803dc3bd08fb2b86045093b3"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-19 22:12:31.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2024-05-15 03:05:08.000000000","tz":0},"subject":"fix x-open-expired 404 on HEAD?part-number reqs","message":"fix x-open-expired 404 on HEAD?part-number reqs\n\nFixes a bug with the x-open-expired feature where the headers\ndo not get copied when refetching zero-byte manifests that causes\n404 on HEAD requests that use the part-number parameter.\n\nChange-Id: If0382d433f73cc0333bb4d0319fe1487b7783e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c4e8c16b3ada5a1666748e2bc723852dbf5e242f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c4e8c16b3ada5a1666748e2bc723852dbf5e242f"}]},"branch":"refs/heads/master","description":"Rebase"},"afbb2486dd727e5a052b333aadf10638fccd4135":{"kind":"REWORK","_number":9,"created":"2024-05-15 19:39:16.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/47/916547/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/916547/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/916547/9"}}},"commit":{"parents":[{"commit":"d1aa735a37d9d679c51a9eda6131ec3e365976b8","subject":"Merge \"backend ratelimit: support per-method rate limits\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d1aa735a37d9d679c51a9eda6131ec3e365976b8"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-19 22:12:31.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2024-05-15 19:38:45.000000000","tz":-420},"subject":"fix x-open-expired 404 on HEAD?part-number reqs","message":"fix x-open-expired 404 on HEAD?part-number reqs\n\nFixes a bug with the x-open-expired feature where the headers\ndo not get copied when refetching zero-byte manifests that causes\n404 on HEAD requests that use the part-number parameter.\n\nChange-Id: If0382d433f73cc0333bb4d0319fe1487b7783e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/afbb2486dd727e5a052b333aadf10638fccd4135"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/afbb2486dd727e5a052b333aadf10638fccd4135"}]},"branch":"refs/heads/master"},"a6d471282d1ea49a500f43c29e58c184edb6d7ae":{"kind":"REWORK","_number":10,"created":"2024-05-16 00:01:19.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/47/916547/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/916547/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/916547/10"}}},"commit":{"parents":[{"commit":"d1aa735a37d9d679c51a9eda6131ec3e365976b8","subject":"Merge \"backend ratelimit: support per-method rate limits\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d1aa735a37d9d679c51a9eda6131ec3e365976b8"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-19 22:12:31.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2024-05-16 00:00:53.000000000","tz":-420},"subject":"fix x-open-expired 404 on HEAD?part-number reqs","message":"fix x-open-expired 404 on HEAD?part-number reqs\n\nFixes a bug with the x-open-expired feature where the headers\ndo not get copied when refetching zero-byte manifests that causes\n404 on HEAD requests that use the part-number parameter.\n\nChange-Id: If0382d433f73cc0333bb4d0319fe1487b7783e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a6d471282d1ea49a500f43c29e58c184edb6d7ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a6d471282d1ea49a500f43c29e58c184edb6d7ae"}]},"branch":"refs/heads/master"},"594e4a9d9241063b8eb22463ee4834984df97605":{"kind":"REWORK","_number":11,"created":"2024-05-16 20:30:21.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/47/916547/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/916547/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/916547/11"}}},"commit":{"parents":[{"commit":"d1aa735a37d9d679c51a9eda6131ec3e365976b8","subject":"Merge \"backend ratelimit: support per-method rate limits\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d1aa735a37d9d679c51a9eda6131ec3e365976b8"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-19 22:12:31.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2024-05-16 20:30:06.000000000","tz":-420},"subject":"fix x-open-expired 404 on HEAD?part-number reqs","message":"fix x-open-expired 404 on HEAD?part-number reqs\n\nFixes a bug with the x-open-expired feature where the headers\ndo not get copied when refetching zero-byte manifests that causes\n404 on HEAD requests that use the part-number parameter.\n\nChange-Id: If0382d433f73cc0333bb4d0319fe1487b7783e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/594e4a9d9241063b8eb22463ee4834984df97605"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/594e4a9d9241063b8eb22463ee4834984df97605"}]},"branch":"refs/heads/master"},"09b4b971d402dcbcd1d9bb5b14b50bc2985630ae":{"kind":"REWORK","_number":12,"created":"2024-05-16 20:45:44.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/47/916547/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/916547/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/916547/12"}}},"commit":{"parents":[{"commit":"d1aa735a37d9d679c51a9eda6131ec3e365976b8","subject":"Merge \"backend ratelimit: support per-method rate limits\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d1aa735a37d9d679c51a9eda6131ec3e365976b8"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-19 22:12:31.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2024-05-16 20:45:36.000000000","tz":-420},"subject":"fix x-open-expired 404 on HEAD?part-number reqs","message":"fix x-open-expired 404 on HEAD?part-number reqs\n\nFixes a bug with the x-open-expired feature where the headers\ndo not get copied when refetching zero-byte manifests that causes\n404 on HEAD requests that use the part-number parameter.\n\nChange-Id: If0382d433f73cc0333bb4d0319fe1487b7783e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/09b4b971d402dcbcd1d9bb5b14b50bc2985630ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/09b4b971d402dcbcd1d9bb5b14b50bc2985630ae"}]},"branch":"refs/heads/master"},"7d05b19ca743149ea3de3afeba3a1ccd212637f3":{"kind":"REWORK","_number":13,"created":"2024-05-20 21:28:58.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/47/916547/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/916547/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/916547/13"}}},"commit":{"parents":[{"commit":"d1aa735a37d9d679c51a9eda6131ec3e365976b8","subject":"Merge \"backend ratelimit: support per-method rate limits\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d1aa735a37d9d679c51a9eda6131ec3e365976b8"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-19 22:12:31.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2024-05-20 21:28:49.000000000","tz":-420},"subject":"fix x-open-expired 404 on HEAD?part-number reqs","message":"fix x-open-expired 404 on HEAD?part-number reqs\n\nFixes a bug with the x-open-expired feature where the headers\ndo not get copied when refetching zero-byte manifests that causes\n404 on HEAD requests that use the part-number parameter.\n\nChange-Id: If0382d433f73cc0333bb4d0319fe1487b7783e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7d05b19ca743149ea3de3afeba3a1ccd212637f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7d05b19ca743149ea3de3afeba3a1ccd212637f3"}]},"branch":"refs/heads/master"},"08b1fb94cb645fa38a77949f586a0e1e591ae09c":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2024-05-20 21:30:18.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/47/916547/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/916547/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/916547/14"}}},"commit":{"parents":[{"commit":"d5c26bb690007c8ffb33a7088fbff7c85abb5e17","subject":"Merge \"test: more test for s3api v4 checksum\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d5c26bb690007c8ffb33a7088fbff7c85abb5e17"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-19 22:12:31.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2024-05-20 21:30:18.000000000","tz":0},"subject":"fix x-open-expired 404 on HEAD?part-number reqs","message":"fix x-open-expired 404 on HEAD?part-number reqs\n\nFixes a bug with the x-open-expired feature where the headers\ndo not get copied when refetching zero-byte manifests that causes\n404 on HEAD requests that use the part-number parameter.\n\nChange-Id: If0382d433f73cc0333bb4d0319fe1487b7783e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/08b1fb94cb645fa38a77949f586a0e1e591ae09c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/08b1fb94cb645fa38a77949f586a0e1e591ae09c"}]},"branch":"refs/heads/master","description":"Rebase"},"30caa51e1f7f2010fdcf5137a619b6742e9206d9":{"kind":"REWORK","_number":15,"created":"2024-05-23 01:18:36.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/47/916547/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/916547/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/916547/15"}}},"commit":{"parents":[{"commit":"d1aa735a37d9d679c51a9eda6131ec3e365976b8","subject":"Merge \"backend ratelimit: support per-method rate limits\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d1aa735a37d9d679c51a9eda6131ec3e365976b8"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-19 22:12:31.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2024-05-23 01:18:23.000000000","tz":-420},"subject":"fix x-open-expired 404 on HEAD?part-number reqs","message":"fix x-open-expired 404 on HEAD?part-number reqs\n\nFixes a bug with the x-open-expired feature where the headers\ndo not get copied when refetching zero-byte manifests that causes\n404 on HEAD requests that use the part-number parameter.\n\nChange-Id: If0382d433f73cc0333bb4d0319fe1487b7783e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/30caa51e1f7f2010fdcf5137a619b6742e9206d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/30caa51e1f7f2010fdcf5137a619b6742e9206d9"}]},"branch":"refs/heads/master"},"8ded7acd304503ec85babbe3b7c72495bec97083":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2024-05-23 01:18:50.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/47/916547/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/916547/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/916547/16"}}},"commit":{"parents":[{"commit":"c620a192d0db733dbb90c71d3c9a7c5f3eba3e56","subject":"Merge \"docs: Update further-reading URL for cross-domain policies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c620a192d0db733dbb90c71d3c9a7c5f3eba3e56"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-19 22:12:31.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2024-05-23 01:18:50.000000000","tz":0},"subject":"fix x-open-expired 404 on HEAD?part-number reqs","message":"fix x-open-expired 404 on HEAD?part-number reqs\n\nFixes a bug with the x-open-expired feature where the headers\ndo not get copied when refetching zero-byte manifests that causes\n404 on HEAD requests that use the part-number parameter.\n\nChange-Id: If0382d433f73cc0333bb4d0319fe1487b7783e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8ded7acd304503ec85babbe3b7c72495bec97083"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8ded7acd304503ec85babbe3b7c72495bec97083"}]},"branch":"refs/heads/master","description":"Rebase"},"738fd00b52c391bb53db91f9bd6c3e92290d31ca":{"kind":"REWORK","_number":17,"created":"2024-05-23 03:37:13.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/47/916547/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/916547/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/916547/17"}}},"commit":{"parents":[{"commit":"c620a192d0db733dbb90c71d3c9a7c5f3eba3e56","subject":"Merge \"docs: Update further-reading URL for cross-domain policies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c620a192d0db733dbb90c71d3c9a7c5f3eba3e56"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-19 22:12:31.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2024-05-23 03:37:04.000000000","tz":-420},"subject":"fix x-open-expired 404 on HEAD?part-number reqs","message":"fix x-open-expired 404 on HEAD?part-number reqs\n\nFixes a bug with the x-open-expired feature where the headers\ndo not get copied when refetching zero-byte manifests that causes\n404 on HEAD requests that use the part-number parameter.\n\nChange-Id: If0382d433f73cc0333bb4d0319fe1487b7783e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/738fd00b52c391bb53db91f9bd6c3e92290d31ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/738fd00b52c391bb53db91f9bd6c3e92290d31ca"}]},"branch":"refs/heads/master"},"994ee167540ff459fd93d3f4d27dda1d0120bf17":{"kind":"REWORK","_number":18,"created":"2024-05-23 16:08:39.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/47/916547/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/916547/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/916547/18"}}},"commit":{"parents":[{"commit":"c620a192d0db733dbb90c71d3c9a7c5f3eba3e56","subject":"Merge \"docs: Update further-reading URL for cross-domain policies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c620a192d0db733dbb90c71d3c9a7c5f3eba3e56"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-19 22:12:31.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2024-05-23 16:08:33.000000000","tz":-420},"subject":"fix x-open-expired 404 on HEAD?part-number reqs","message":"fix x-open-expired 404 on HEAD?part-number reqs\n\nFixes a bug with the x-open-expired feature where the headers\ndo not get copied when refetching zero-byte manifests that causes\n404 on HEAD requests that use the part-number parameter.\n\nChange-Id: If0382d433f73cc0333bb4d0319fe1487b7783e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/994ee167540ff459fd93d3f4d27dda1d0120bf17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/994ee167540ff459fd93d3f4d27dda1d0120bf17"}]},"branch":"refs/heads/master"},"e4eac2c42a16d3b9555ab129f516ec0f72e3fdc1":{"kind":"REWORK","_number":19,"created":"2024-05-23 19:15:18.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/47/916547/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/916547/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/916547/19"}}},"commit":{"parents":[{"commit":"c620a192d0db733dbb90c71d3c9a7c5f3eba3e56","subject":"Merge \"docs: Update further-reading URL for cross-domain policies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c620a192d0db733dbb90c71d3c9a7c5f3eba3e56"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-19 22:12:31.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2024-05-23 19:15:10.000000000","tz":-420},"subject":"fix x-open-expired 404 on HEAD?part-number reqs","message":"fix x-open-expired 404 on HEAD?part-number reqs\n\nFixes a bug with the x-open-expired feature where the headers\ndo not get copied when refetching zero-byte manifests that causes\n404 on HEAD requests that use the part-number parameter.\n\nChange-Id: If0382d433f73cc0333bb4d0319fe1487b7783e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e4eac2c42a16d3b9555ab129f516ec0f72e3fdc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e4eac2c42a16d3b9555ab129f516ec0f72e3fdc1"}]},"branch":"refs/heads/master"},"8f15087359ffcdc9de7087f364c420443b045529":{"kind":"REWORK","_number":20,"created":"2024-06-04 20:25:47.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/47/916547/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/916547/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/916547/20"}}},"commit":{"parents":[{"commit":"c620a192d0db733dbb90c71d3c9a7c5f3eba3e56","subject":"Merge \"docs: Update further-reading URL for cross-domain policies\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c620a192d0db733dbb90c71d3c9a7c5f3eba3e56"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-19 22:12:31.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2024-06-04 20:20:59.000000000","tz":-420},"subject":"fix x-open-expired 404 on HEAD?part-number reqs","message":"fix x-open-expired 404 on HEAD?part-number reqs\n\nFixes a bug with the x-open-expired feature where the headers\ndo not get copied when refetching zero-byte manifests that causes\n404 on HEAD requests that use the part-number parameter.\n\nChange-Id: If0382d433f73cc0333bb4d0319fe1487b7783e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8f15087359ffcdc9de7087f364c420443b045529"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8f15087359ffcdc9de7087f364c420443b045529"}]},"branch":"refs/heads/master"},"20c71be4f43bba3a51213602839039eedd5ac324":{"kind":"REWORK","_number":21,"created":"2024-06-04 22:28:57.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/47/916547/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/916547/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/916547/21"}}},"commit":{"parents":[{"commit":"1d5b1d746e027bb9509ad0e45cc0f2f50d28511f","subject":"Merge \"Lazy import is not needed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1d5b1d746e027bb9509ad0e45cc0f2f50d28511f"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-19 22:12:31.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2024-06-04 22:28:50.000000000","tz":-420},"subject":"fix x-open-expired 404 on HEAD?part-number reqs","message":"fix x-open-expired 404 on HEAD?part-number reqs\n\nFixes a bug with the x-open-expired feature where the headers\ndo not get copied when refetching zero-byte manifests that causes\n404 on HEAD requests that use the part-number parameter.\n\nChange-Id: If0382d433f73cc0333bb4d0319fe1487b7783e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/20c71be4f43bba3a51213602839039eedd5ac324"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/20c71be4f43bba3a51213602839039eedd5ac324"}]},"branch":"refs/heads/master"},"4600fd13a9b6f18572e648e8466c2a4db373ceac":{"kind":"REWORK","_number":22,"created":"2024-06-05 00:04:45.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/47/916547/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/916547/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/916547/22"}}},"commit":{"parents":[{"commit":"1d5b1d746e027bb9509ad0e45cc0f2f50d28511f","subject":"Merge \"Lazy import is not needed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1d5b1d746e027bb9509ad0e45cc0f2f50d28511f"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-19 22:12:31.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2024-06-05 00:04:39.000000000","tz":-420},"subject":"fix x-open-expired 404 on HEAD?part-number reqs","message":"fix x-open-expired 404 on HEAD?part-number reqs\n\nFixes a bug with the x-open-expired feature where the headers\ndo not get copied when refetching zero-byte manifests that causes\n404 on HEAD requests that use the part-number parameter.\n\nChange-Id: If0382d433f73cc0333bb4d0319fe1487b7783e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4600fd13a9b6f18572e648e8466c2a4db373ceac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4600fd13a9b6f18572e648e8466c2a4db373ceac"}]},"branch":"refs/heads/master"},"f21e8ae9c2fd47b5179e95988ad9de1830b19772":{"kind":"REWORK","_number":23,"created":"2024-06-05 15:53:42.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/47/916547/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/916547/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/916547/23"}}},"commit":{"parents":[{"commit":"1d5b1d746e027bb9509ad0e45cc0f2f50d28511f","subject":"Merge \"Lazy import is not needed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1d5b1d746e027bb9509ad0e45cc0f2f50d28511f"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-19 22:12:31.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2024-06-05 15:53:26.000000000","tz":-420},"subject":"fix x-open-expired 404 on HEAD?part-number reqs","message":"fix x-open-expired 404 on HEAD?part-number reqs\n\nFixes a bug with the x-open-expired feature where the headers\ndo not get copied when refetching zero-byte manifests that causes\n404 on HEAD requests that use the part-number parameter.\n\nChange-Id: If0382d433f73cc0333bb4d0319fe1487b7783e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f21e8ae9c2fd47b5179e95988ad9de1830b19772"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f21e8ae9c2fd47b5179e95988ad9de1830b19772"}]},"branch":"refs/heads/master"},"e87cabd8ad9a8f4d817919673f49cdf5e6cd32e6":{"kind":"TRIVIAL_REBASE","_number":24,"created":"2024-06-18 15:12:46.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/47/916547/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/916547/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/916547/24"}}},"commit":{"parents":[{"commit":"46e183df15d0a98d2595e583d2a50c98aba153e2","subject":"Merge \"add bytes of expiring objects to queue entry\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/46e183df15d0a98d2595e583d2a50c98aba153e2"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-19 22:12:31.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2024-06-18 15:12:46.000000000","tz":0},"subject":"fix x-open-expired 404 on HEAD?part-number reqs","message":"fix x-open-expired 404 on HEAD?part-number reqs\n\nFixes a bug with the x-open-expired feature where the headers\ndo not get copied when refetching zero-byte manifests that causes\n404 on HEAD requests that use the part-number parameter.\n\nChange-Id: If0382d433f73cc0333bb4d0319fe1487b7783e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e87cabd8ad9a8f4d817919673f49cdf5e6cd32e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e87cabd8ad9a8f4d817919673f49cdf5e6cd32e6"}]},"branch":"refs/heads/master","description":"Rebase"},"0fc807cf6b387685ea4360eb40e120af4b27f74e":{"kind":"TRIVIAL_REBASE","_number":25,"created":"2024-06-24 16:34:00.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/47/916547/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/916547/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/916547/25"}}},"commit":{"parents":[{"commit":"ff08a649e0bc657e13f346e963ae2d40d0e2f001","subject":"Merge \"Add get_statsd_client function\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ff08a649e0bc657e13f346e963ae2d40d0e2f001"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-19 22:12:31.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2024-06-24 16:34:00.000000000","tz":0},"subject":"fix x-open-expired 404 on HEAD?part-number reqs","message":"fix x-open-expired 404 on HEAD?part-number reqs\n\nFixes a bug with the x-open-expired feature where the headers\ndo not get copied when refetching zero-byte manifests that causes\n404 on HEAD requests that use the part-number parameter.\n\nChange-Id: If0382d433f73cc0333bb4d0319fe1487b7783e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0fc807cf6b387685ea4360eb40e120af4b27f74e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0fc807cf6b387685ea4360eb40e120af4b27f74e"}]},"branch":"refs/heads/master","description":"Rebase"},"8a04bcbcb903d3762c49dcb6f19a1bcb923d23ff":{"kind":"TRIVIAL_REBASE","_number":26,"created":"2024-07-09 01:47:31.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/47/916547/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/916547/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/916547/26"}}},"commit":{"parents":[{"commit":"0fd36e395dbe05fb77a4e76bc6ea4c83597d86ff","subject":"common: move memcached exceptions to the base file.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0fd36e395dbe05fb77a4e76bc6ea4c83597d86ff"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-19 22:12:31.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"nairashwin952013@gmail.com","date":"2024-07-09 01:47:31.000000000","tz":0},"subject":"fix x-open-expired 404 on HEAD?part-number reqs","message":"fix x-open-expired 404 on HEAD?part-number reqs\n\nFixes a bug with the x-open-expired feature where the headers\ndo not get copied when refetching zero-byte manifests that causes\n404 on HEAD requests that use the part-number parameter.\n\nChange-Id: If0382d433f73cc0333bb4d0319fe1487b7783e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8a04bcbcb903d3762c49dcb6f19a1bcb923d23ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8a04bcbcb903d3762c49dcb6f19a1bcb923d23ff"}]},"branch":"refs/heads/master","description":"Rebase"},"17c3559bbca724c059241096a4603f7d1dd9154a":{"kind":"REWORK","_number":27,"created":"2024-07-31 16:45:47.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/47/916547/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/916547/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/916547/27"}}},"commit":{"parents":[{"commit":"e984de864dd0895f4bde81bed56691c48e7cea52","subject":"trivial: assert non_negative_int treats float as valid","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e984de864dd0895f4bde81bed56691c48e7cea52"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-19 22:12:31.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-07-31 16:45:38.000000000","tz":-420},"subject":"fix x-open-expired 404 on HEAD?part-number reqs","message":"fix x-open-expired 404 on HEAD?part-number reqs\n\nFixes a bug with the x-open-expired feature where the headers\ndo not get copied when refetching zero-byte manifests that causes\n404 on HEAD requests that use the part-number parameter.\n\nChange-Id: If0382d433f73cc0333bb4d0319fe1487b7783e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/17c3559bbca724c059241096a4603f7d1dd9154a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/17c3559bbca724c059241096a4603f7d1dd9154a"}]},"branch":"refs/heads/master"},"378c59e4bd475c533e6dfa662814ba012e72b34e":{"kind":"REWORK","_number":28,"created":"2024-08-01 19:57:53.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/47/916547/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/916547/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/916547/28"}}},"commit":{"parents":[{"commit":"e984de864dd0895f4bde81bed56691c48e7cea52","subject":"trivial: assert non_negative_int treats float as valid","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e984de864dd0895f4bde81bed56691c48e7cea52"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-19 22:12:31.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-08-01 19:57:46.000000000","tz":-420},"subject":"fix x-open-expired 404 on HEAD?part-number reqs","message":"fix x-open-expired 404 on HEAD?part-number reqs\n\nFixes a bug with the x-open-expired feature where the headers\ndo not get copied when refetching zero-byte manifests that causes\n404 on HEAD requests that use the part-number parameter.\n\nChange-Id: If0382d433f73cc0333bb4d0319fe1487b7783e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/378c59e4bd475c533e6dfa662814ba012e72b34e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/378c59e4bd475c533e6dfa662814ba012e72b34e"}]},"branch":"refs/heads/master"},"6c65f027b04b66dc3359f05b18aa9a7d242782fa":{"kind":"REWORK","_number":29,"created":"2024-08-01 20:34:04.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/47/916547/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/916547/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/916547/29"}}},"commit":{"parents":[{"commit":"e984de864dd0895f4bde81bed56691c48e7cea52","subject":"trivial: assert non_negative_int treats float as valid","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e984de864dd0895f4bde81bed56691c48e7cea52"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-19 22:12:31.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-08-01 20:33:50.000000000","tz":-420},"subject":"fix x-open-expired 404 on HEAD?part-number reqs","message":"fix x-open-expired 404 on HEAD?part-number reqs\n\nFixes a bug with the x-open-expired feature where the headers\ndo not get copied when refetching zero-byte manifests that causes\n404 on HEAD requests that use the part-number parameter.\n\nChange-Id: If0382d433f73cc0333bb4d0319fe1487b7783e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6c65f027b04b66dc3359f05b18aa9a7d242782fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6c65f027b04b66dc3359f05b18aa9a7d242782fa"}]},"branch":"refs/heads/master"},"ac3b89e11d63c865234aa81f6f4e0fd056d5c824":{"kind":"REWORK","_number":30,"created":"2024-08-01 20:58:17.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/47/916547/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/916547/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/916547/30"}}},"commit":{"parents":[{"commit":"e984de864dd0895f4bde81bed56691c48e7cea52","subject":"trivial: assert non_negative_int treats float as valid","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e984de864dd0895f4bde81bed56691c48e7cea52"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-19 22:12:31.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-08-01 20:58:09.000000000","tz":-420},"subject":"fix x-open-expired 404 on HEAD?part-number reqs","message":"fix x-open-expired 404 on HEAD?part-number reqs\n\nFixes a bug with the x-open-expired feature where the headers\ndo not get copied when refetching zero-byte manifests that causes\n404 on HEAD requests that use the part-number parameter.\n\nChange-Id: If0382d433f73cc0333bb4d0319fe1487b7783e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ac3b89e11d63c865234aa81f6f4e0fd056d5c824"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ac3b89e11d63c865234aa81f6f4e0fd056d5c824"}]},"branch":"refs/heads/master"},"138fca6dcce9e0995914541dd76911dddd878857":{"kind":"REWORK","_number":31,"created":"2024-08-24 12:49:56.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/47/916547/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/916547/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/916547/31"}}},"commit":{"parents":[{"commit":"e6b73612d1dcd0bace2361338e6ae82eb591282b","subject":"FakeStatsdClient: Stop issuing DNS calls for host.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e6b73612d1dcd0bace2361338e6ae82eb591282b"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-19 22:12:31.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-08-24 12:49:50.000000000","tz":-420},"subject":"fix x-open-expired 404 on HEAD?part-number reqs","message":"fix x-open-expired 404 on HEAD?part-number reqs\n\nFixes a bug with the x-open-expired feature where the headers\ndo not get copied when refetching zero-byte manifests that causes\n404 on HEAD requests that use the part-number parameter.\n\nChange-Id: If0382d433f73cc0333bb4d0319fe1487b7783e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/138fca6dcce9e0995914541dd76911dddd878857"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/138fca6dcce9e0995914541dd76911dddd878857"}]},"branch":"refs/heads/master"},"55a88b178d55a98e45ced697e4b4222eb2d661f3":{"kind":"REWORK","_number":32,"created":"2024-08-27 16:20:48.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/47/916547/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/916547/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/916547/32"}}},"commit":{"parents":[{"commit":"e6b73612d1dcd0bace2361338e6ae82eb591282b","subject":"FakeStatsdClient: Stop issuing DNS calls for host.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e6b73612d1dcd0bace2361338e6ae82eb591282b"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-19 22:12:31.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-08-26 23:02:04.000000000","tz":-420},"subject":"fix x-open-expired 404 on HEAD?part-number reqs","message":"fix x-open-expired 404 on HEAD?part-number reqs\n\nFixes a bug with the x-open-expired feature where the headers\ndo not get copied when refetching zero-byte manifests that causes\n404 on HEAD requests that use the part-number parameter.\n\nChange-Id: If0382d433f73cc0333bb4d0319fe1487b7783e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/55a88b178d55a98e45ced697e4b4222eb2d661f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/55a88b178d55a98e45ced697e4b4222eb2d661f3"}]},"branch":"refs/heads/master"},"cc3333b8d3e5d9d93fe38132c30fe47c5147f4e0":{"kind":"REWORK","_number":33,"created":"2024-08-27 19:06:43.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/47/916547/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/916547/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/916547/33"}}},"commit":{"parents":[{"commit":"e6b73612d1dcd0bace2361338e6ae82eb591282b","subject":"FakeStatsdClient: Stop issuing DNS calls for host.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e6b73612d1dcd0bace2361338e6ae82eb591282b"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-19 22:12:31.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-08-27 19:06:35.000000000","tz":-420},"subject":"fix x-open-expired 404 on HEAD?part-number reqs","message":"fix x-open-expired 404 on HEAD?part-number reqs\n\nFixes a bug with the x-open-expired feature where the headers\ndo not get copied when refetching zero-byte manifests that causes\n404 on HEAD requests that use the part-number parameter.\n\nChange-Id: If0382d433f73cc0333bb4d0319fe1487b7783e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cc3333b8d3e5d9d93fe38132c30fe47c5147f4e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cc3333b8d3e5d9d93fe38132c30fe47c5147f4e0"}]},"branch":"refs/heads/master"},"343f49a700e1c82d60e5d829aaed083f112f02ce":{"kind":"REWORK","_number":34,"created":"2024-08-29 17:13:27.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/47/916547/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/916547/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/916547/34"}}},"commit":{"parents":[{"commit":"6a0153f54542da474955802ac01cf174c7ce8573","subject":"Merge \"docs: Simplify reload-process-tree SVGs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6a0153f54542da474955802ac01cf174c7ce8573"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-19 22:12:31.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-29 17:12:58.000000000","tz":-420},"subject":"fix x-open-expired 404 on HEAD?part-number reqs","message":"fix x-open-expired 404 on HEAD?part-number reqs\n\nFixes a bug with the x-open-expired feature where the headers\ndo not get copied when refetching zero-byte manifests that causes\n404 on HEAD requests that use the part-number parameter.\n\nChange-Id: If0382d433f73cc0333bb4d0319fe1487b7783e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/343f49a700e1c82d60e5d829aaed083f112f02ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/343f49a700e1c82d60e5d829aaed083f112f02ce"}]},"branch":"refs/heads/master"},"aefe106beb772496adc6382a14b871b2fad78a0b":{"kind":"REWORK","_number":35,"created":"2024-08-29 18:48:27.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/47/916547/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/916547/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/916547/35"}}},"commit":{"parents":[{"commit":"6a0153f54542da474955802ac01cf174c7ce8573","subject":"Merge \"docs: Simplify reload-process-tree SVGs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6a0153f54542da474955802ac01cf174c7ce8573"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-19 22:12:31.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-29 18:48:19.000000000","tz":-420},"subject":"fix x-open-expired 404 on HEAD?part-number reqs","message":"fix x-open-expired 404 on HEAD?part-number reqs\n\nFixes a bug with the x-open-expired feature where the headers\ndo not get copied when refetching zero-byte manifests that causes\n404 on HEAD requests that use the part-number parameter.\n\nChange-Id: If0382d433f73cc0333bb4d0319fe1487b7783e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aefe106beb772496adc6382a14b871b2fad78a0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aefe106beb772496adc6382a14b871b2fad78a0b"}]},"branch":"refs/heads/master"},"51246ce9a247f6f384b65448e098fcda8229156b":{"kind":"TRIVIAL_REBASE","_number":36,"created":"2024-08-30 16:13:23.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/47/916547/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/916547/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/916547/36"}}},"commit":{"parents":[{"commit":"ec8166be3366865954381572eef68b26d2951a25","subject":"Merge \"tests: Attempt to use configured tmp in xprofile tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ec8166be3366865954381572eef68b26d2951a25"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-19 22:12:31.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-30 16:13:23.000000000","tz":0},"subject":"fix x-open-expired 404 on HEAD?part-number reqs","message":"fix x-open-expired 404 on HEAD?part-number reqs\n\nFixes a bug with the x-open-expired feature where the headers\ndo not get copied when refetching zero-byte manifests that causes\n404 on HEAD requests that use the part-number parameter.\n\nChange-Id: If0382d433f73cc0333bb4d0319fe1487b7783e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/51246ce9a247f6f384b65448e098fcda8229156b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/51246ce9a247f6f384b65448e098fcda8229156b"}]},"branch":"refs/heads/master","description":"Rebase"},"64a2299783128a1d71503878a72c26b23839f937":{"kind":"REWORK","_number":37,"created":"2024-09-09 15:08:35.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/47/916547/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/916547/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/916547/37"}}},"commit":{"parents":[{"commit":"48e087b1cc7ea564d0f205baf5d625cad34be810","subject":"Merge \"object-expirer: refactor creating InternalClient\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/48e087b1cc7ea564d0f205baf5d625cad34be810"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-19 22:12:31.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-09-09 15:07:57.000000000","tz":-420},"subject":"fix x-open-expired 404 on HEAD?part-number reqs","message":"fix x-open-expired 404 on HEAD?part-number reqs\n\nFixes a bug with the x-open-expired feature where the headers\ndo not get copied when refetching zero-byte manifests that causes\n404 on HEAD requests that use the part-number parameter.\n\nChange-Id: If0382d433f73cc0333bb4d0319fe1487b7783e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/64a2299783128a1d71503878a72c26b23839f937"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/64a2299783128a1d71503878a72c26b23839f937"}]},"branch":"refs/heads/master"},"97ce746aeb12fe9b5a15d631f4ff9387d03532d2":{"kind":"REWORK","_number":38,"created":"2024-09-09 20:27:11.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/47/916547/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/916547/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/916547/38"}}},"commit":{"parents":[{"commit":"48e087b1cc7ea564d0f205baf5d625cad34be810","subject":"Merge \"object-expirer: refactor creating InternalClient\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/48e087b1cc7ea564d0f205baf5d625cad34be810"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-19 22:12:31.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-09-09 20:26:55.000000000","tz":-420},"subject":"fix x-open-expired 404 on HEAD?part-number reqs","message":"fix x-open-expired 404 on HEAD?part-number reqs\n\nFixes a bug with the x-open-expired feature where the headers\ndo not get copied when refetching zero-byte manifests that causes\n404 on HEAD requests that use the part-number parameter.\n\nChange-Id: If0382d433f73cc0333bb4d0319fe1487b7783e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/97ce746aeb12fe9b5a15d631f4ff9387d03532d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/97ce746aeb12fe9b5a15d631f4ff9387d03532d2"}]},"branch":"refs/heads/master"},"40f54266fa116f192435901cdc5d1ede66127d3a":{"kind":"REWORK","_number":39,"created":"2024-09-17 19:54:51.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/47/916547/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/916547/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/916547/39"}}},"commit":{"parents":[{"commit":"ffdf96259814e554dc16ea29b3de6b4e37173158","subject":"object-expirer: fix unused _make_internal_client arg","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ffdf96259814e554dc16ea29b3de6b4e37173158"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-19 22:12:31.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-09-17 19:54:41.000000000","tz":-420},"subject":"fix x-open-expired 404 on HEAD?part-number reqs","message":"fix x-open-expired 404 on HEAD?part-number reqs\n\nFixes a bug with the x-open-expired feature where the headers\ndo not get copied when refetching zero-byte manifests that causes\n404 on HEAD requests that use the part-number parameter.\n\nChange-Id: If0382d433f73cc0333bb4d0319fe1487b7783e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/40f54266fa116f192435901cdc5d1ede66127d3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/40f54266fa116f192435901cdc5d1ede66127d3a"}]},"branch":"refs/heads/master"},"cf3154b8e654ca6c56e6b96c6db5eb9645113de8":{"kind":"TRIVIAL_REBASE","_number":40,"created":"2024-09-17 22:56:17.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/47/916547/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/916547/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/916547/40"}}},"commit":{"parents":[{"commit":"e0a9ea496ef6b06301b98d6b6a27b6dc951b9fd2","subject":"Update master for stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e0a9ea496ef6b06301b98d6b6a27b6dc951b9fd2"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-19 22:12:31.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2024-09-17 22:56:17.000000000","tz":0},"subject":"fix x-open-expired 404 on HEAD?part-number reqs","message":"fix x-open-expired 404 on HEAD?part-number reqs\n\nFixes a bug with the x-open-expired feature where the headers\ndo not get copied when refetching zero-byte manifests that causes\n404 on HEAD requests that use the part-number parameter.\n\nChange-Id: If0382d433f73cc0333bb4d0319fe1487b7783e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cf3154b8e654ca6c56e6b96c6db5eb9645113de8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cf3154b8e654ca6c56e6b96c6db5eb9645113de8"}]},"branch":"refs/heads/master","description":"Rebase"},"e085fb0225d600d73ca033b4bfdf578585e8bb28":{"kind":"REWORK","_number":41,"created":"2024-09-19 04:22:13.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/47/916547/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/916547/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/916547/41"}}},"commit":{"parents":[{"commit":"e0a9ea496ef6b06301b98d6b6a27b6dc951b9fd2","subject":"Update master for stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e0a9ea496ef6b06301b98d6b6a27b6dc951b9fd2"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-19 22:12:31.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-09-19 04:21:56.000000000","tz":-420},"subject":"fix x-open-expired 404 on HEAD?part-number reqs","message":"fix x-open-expired 404 on HEAD?part-number reqs\n\nFixes a bug with the x-open-expired feature where the headers\ndo not get copied when refetching zero-byte manifests that causes\n404 on HEAD requests that use the part-number parameter.\n\nChange-Id: If0382d433f73cc0333bb4d0319fe1487b7783e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e085fb0225d600d73ca033b4bfdf578585e8bb28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e085fb0225d600d73ca033b4bfdf578585e8bb28"}]},"branch":"refs/heads/master"},"d23b5401e3231d4ef4ef6860caf064e6ff75ace4":{"kind":"TRIVIAL_REBASE","_number":42,"created":"2024-09-23 16:00:28.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/47/916547/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/916547/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/916547/42"}}},"commit":{"parents":[{"commit":"94dc4cad090390b1826bf6a2fbe5986e1516219e","subject":"Merge \"Quiten boto logging in func tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/94dc4cad090390b1826bf6a2fbe5986e1516219e"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-19 22:12:31.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2024-09-23 16:00:28.000000000","tz":0},"subject":"fix x-open-expired 404 on HEAD?part-number reqs","message":"fix x-open-expired 404 on HEAD?part-number reqs\n\nFixes a bug with the x-open-expired feature where the headers\ndo not get copied when refetching zero-byte manifests that causes\n404 on HEAD requests that use the part-number parameter.\n\nChange-Id: If0382d433f73cc0333bb4d0319fe1487b7783e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d23b5401e3231d4ef4ef6860caf064e6ff75ace4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d23b5401e3231d4ef4ef6860caf064e6ff75ace4"}]},"branch":"refs/heads/master","description":"Rebase"},"438fd3412ce8b183821b544a82f7e60c2e361ab9":{"kind":"TRIVIAL_REBASE","_number":43,"created":"2024-09-27 04:13:24.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/47/916547/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/916547/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/916547/43"}}},"commit":{"parents":[{"commit":"4807999749c7dffc3dce0104bafac3ebfef0baac","subject":"Merge \"func tests: fix etag-quoter insertion in pipeline\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4807999749c7dffc3dce0104bafac3ebfef0baac"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-19 22:12:31.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2024-09-27 04:13:24.000000000","tz":0},"subject":"fix x-open-expired 404 on HEAD?part-number reqs","message":"fix x-open-expired 404 on HEAD?part-number reqs\n\nFixes a bug with the x-open-expired feature where the headers\ndo not get copied when refetching zero-byte manifests that causes\n404 on HEAD requests that use the part-number parameter.\n\nChange-Id: If0382d433f73cc0333bb4d0319fe1487b7783e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/438fd3412ce8b183821b544a82f7e60c2e361ab9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/438fd3412ce8b183821b544a82f7e60c2e361ab9"}]},"branch":"refs/heads/master","description":"Rebase"},"f47631c046ddf6e56237851f6c402eccb27d56c3":{"kind":"TRIVIAL_REBASE","_number":44,"created":"2024-10-03 15:12:32.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/47/916547/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/916547/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/916547/44"}}},"commit":{"parents":[{"commit":"914380cee07031a0844383d175d9c4781e71f583","subject":"Merge \"Add Chinemerem to AUTHORS\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/914380cee07031a0844383d175d9c4781e71f583"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-19 22:12:31.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2024-10-03 15:12:32.000000000","tz":0},"subject":"fix x-open-expired 404 on HEAD?part-number reqs","message":"fix x-open-expired 404 on HEAD?part-number reqs\n\nFixes a bug with the x-open-expired feature where the headers\ndo not get copied when refetching zero-byte manifests that causes\n404 on HEAD requests that use the part-number parameter.\n\nChange-Id: If0382d433f73cc0333bb4d0319fe1487b7783e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f47631c046ddf6e56237851f6c402eccb27d56c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f47631c046ddf6e56237851f6c402eccb27d56c3"}]},"branch":"refs/heads/master","description":"Rebase"},"708ffb85d52629ea952df0d62644fddb17031a86":{"kind":"REWORK","_number":45,"created":"2024-10-03 23:49:23.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/47/916547/45","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/916547/45","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/45 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/45 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/45 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/916547/45"}}},"commit":{"parents":[{"commit":"914380cee07031a0844383d175d9c4781e71f583","subject":"Merge \"Add Chinemerem to AUTHORS\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/914380cee07031a0844383d175d9c4781e71f583"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-19 22:12:31.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-03 23:49:14.000000000","tz":-420},"subject":"fix x-open-expired 404 on HEAD?part-number reqs","message":"fix x-open-expired 404 on HEAD?part-number reqs\n\nFixes a bug with the x-open-expired feature where the headers\ndo not get copied when refetching zero-byte manifests that causes\n404 on HEAD requests that use the part-number parameter.\n\nChange-Id: If0382d433f73cc0333bb4d0319fe1487b7783e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/708ffb85d52629ea952df0d62644fddb17031a86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/708ffb85d52629ea952df0d62644fddb17031a86"}]},"branch":"refs/heads/master"},"86f3686f681828152b65e88c13d4c735de80018c":{"kind":"REWORK","_number":46,"created":"2024-10-16 15:27:16.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/47/916547/46","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/916547/46","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/46 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/46 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/46 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/916547/46"}}},"commit":{"parents":[{"commit":"cfc7f99d43d83b90fe1498e74781ab7027a5156b","subject":"trivial: Fix variable name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cfc7f99d43d83b90fe1498e74781ab7027a5156b"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-19 22:12:31.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-14 14:29:59.000000000","tz":-420},"subject":"fix x-open-expired 404 on HEAD?part-number reqs","message":"fix x-open-expired 404 on HEAD?part-number reqs\n\nFixes a bug with the x-open-expired feature where the headers\ndo not get copied when refetching zero-byte manifests that causes\n404 on HEAD requests that use the part-number parameter.\n\nChange-Id: If0382d433f73cc0333bb4d0319fe1487b7783e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/86f3686f681828152b65e88c13d4c735de80018c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/86f3686f681828152b65e88c13d4c735de80018c"}]},"branch":"refs/heads/master"},"2423db0d1bd009155bcb5041ed0927124c1b1211":{"kind":"REWORK","_number":47,"created":"2024-10-16 21:00:05.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/47/916547/47","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/916547/47","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/47 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/47 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/47 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/916547/47"}}},"commit":{"parents":[{"commit":"cfc7f99d43d83b90fe1498e74781ab7027a5156b","subject":"trivial: Fix variable name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cfc7f99d43d83b90fe1498e74781ab7027a5156b"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-19 22:12:31.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-16 20:59:56.000000000","tz":-420},"subject":"fix x-open-expired 404 on HEAD?part-number reqs","message":"fix x-open-expired 404 on HEAD?part-number reqs\n\nFixes a bug with the x-open-expired feature where the headers\ndo not get copied when refetching zero-byte manifests that causes\n404 on HEAD requests that use the part-number parameter.\n\nChange-Id: If0382d433f73cc0333bb4d0319fe1487b7783e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2423db0d1bd009155bcb5041ed0927124c1b1211"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2423db0d1bd009155bcb5041ed0927124c1b1211"}]},"branch":"refs/heads/master"},"1c9ee0615487b87d25474b06159c9c77d1cf3a70":{"kind":"REWORK","_number":48,"created":"2024-10-16 21:06:05.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/47/916547/48","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/916547/48","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/48 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/48 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/48 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/916547/48"}}},"commit":{"parents":[{"commit":"cfc7f99d43d83b90fe1498e74781ab7027a5156b","subject":"trivial: Fix variable name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cfc7f99d43d83b90fe1498e74781ab7027a5156b"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-19 22:12:31.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-16 21:05:56.000000000","tz":-420},"subject":"fix x-open-expired 404 on HEAD?part-number reqs","message":"fix x-open-expired 404 on HEAD?part-number reqs\n\nFixes a bug with the x-open-expired feature where the headers\ndo not get copied when refetching zero-byte manifests that causes\n404 on HEAD requests that use the part-number parameter.\n\nChange-Id: If0382d433f73cc0333bb4d0319fe1487b7783e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1c9ee0615487b87d25474b06159c9c77d1cf3a70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1c9ee0615487b87d25474b06159c9c77d1cf3a70"}]},"branch":"refs/heads/master"},"a8989d725130e7e0b849bb31c721c746182f3d1c":{"kind":"REWORK","_number":49,"created":"2024-10-17 20:45:23.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/47/916547/49","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/916547/49","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/49 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/49 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/49 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/916547/49"}}},"commit":{"parents":[{"commit":"cfc7f99d43d83b90fe1498e74781ab7027a5156b","subject":"trivial: Fix variable name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cfc7f99d43d83b90fe1498e74781ab7027a5156b"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-19 22:12:31.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-17 20:45:15.000000000","tz":-420},"subject":"fix x-open-expired 404 on HEAD?part-number reqs","message":"fix x-open-expired 404 on HEAD?part-number reqs\n\nFixes a bug with the x-open-expired feature where the headers\ndo not get copied when refetching zero-byte manifests that causes\n404 on HEAD requests that use the part-number parameter.\n\nChange-Id: If0382d433f73cc0333bb4d0319fe1487b7783e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a8989d725130e7e0b849bb31c721c746182f3d1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a8989d725130e7e0b849bb31c721c746182f3d1c"}]},"branch":"refs/heads/master"},"1c4638d4a85d515548df98ec96fd9e322cd89fff":{"kind":"NO_CODE_CHANGE","_number":50,"created":"2024-10-17 20:47:51.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/47/916547/50","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/916547/50","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/50 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/50 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/50 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/916547/50"}}},"commit":{"parents":[{"commit":"cfc7f99d43d83b90fe1498e74781ab7027a5156b","subject":"trivial: Fix variable name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cfc7f99d43d83b90fe1498e74781ab7027a5156b"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-19 22:12:31.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2024-10-17 20:47:51.000000000","tz":0},"subject":"fix x-open-expired 404 on HEAD?part-number reqs","message":"fix x-open-expired 404 on HEAD?part-number reqs\n\nFixes a bug with the x-open-expired feature where the headers\ndo not get copied when refetching zero-byte manifests that causes\n404 on HEAD requests that use the part-number parameter and segment\nGET requests if the segments have expired.\n\nChange-Id: If0382d433f73cc0333bb4d0319fe1487b7783e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1c4638d4a85d515548df98ec96fd9e322cd89fff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1c4638d4a85d515548df98ec96fd9e322cd89fff"}]},"branch":"refs/heads/master","description":"Edit commit message"},"215a77c1d07996539be4dcae2d933fe5fd39a965":{"kind":"REWORK","_number":51,"created":"2024-10-17 21:35:38.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/47/916547/51","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/916547/51","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/51 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/51 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/51 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/916547/51"}}},"commit":{"parents":[{"commit":"cfc7f99d43d83b90fe1498e74781ab7027a5156b","subject":"trivial: Fix variable name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cfc7f99d43d83b90fe1498e74781ab7027a5156b"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-19 22:12:31.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-17 21:35:31.000000000","tz":-420},"subject":"fix x-open-expired 404 on HEAD?part-number reqs","message":"fix x-open-expired 404 on HEAD?part-number reqs\n\nFixes a bug with the x-open-expired feature where the headers\ndo not get copied when refetching zero-byte manifests that causes\n404 on HEAD requests that use the part-number parameter and segment\nGET requests if the segments have expired.\n\nChange-Id: If0382d433f73cc0333bb4d0319fe1487b7783e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/215a77c1d07996539be4dcae2d933fe5fd39a965"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/215a77c1d07996539be4dcae2d933fe5fd39a965"}]},"branch":"refs/heads/master"},"604130e164e7ac2d513ce015b2487ea1b9b481f5":{"kind":"NO_CODE_CHANGE","_number":52,"created":"2024-10-17 21:39:26.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/47/916547/52","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/916547/52","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/52 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/52 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/52 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/916547/52"}}},"commit":{"parents":[{"commit":"cfc7f99d43d83b90fe1498e74781ab7027a5156b","subject":"trivial: Fix variable name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cfc7f99d43d83b90fe1498e74781ab7027a5156b"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-19 22:12:31.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2024-10-17 21:39:26.000000000","tz":0},"subject":"fix x-open-expired 404 on HEAD?part-number reqs","message":"fix x-open-expired 404 on HEAD?part-number reqs\n\nFixes a bug with the x-open-expired feature where our magic header\ndoes not get copied when refetching all manifests that causes\n404 on HEAD requests with part-number\u003dN query parameter since\nthe object-server returns an empty response body and the\nproxy needs to refetch. The fix also applies to segment GET\nrequests  if the segments have expired.\n\nChange-Id: If0382d433f73cc0333bb4d0319fe1487b7783e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/604130e164e7ac2d513ce015b2487ea1b9b481f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/604130e164e7ac2d513ce015b2487ea1b9b481f5"}]},"branch":"refs/heads/master","description":"Edit commit message"},"38e813e731a03917545bd39d9a2a6bfdbc115975":{"kind":"REWORK","_number":53,"created":"2024-10-17 21:46:51.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/47/916547/53","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/916547/53","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/53 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/53 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/53 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/916547/53"}}},"commit":{"parents":[{"commit":"cfc7f99d43d83b90fe1498e74781ab7027a5156b","subject":"trivial: Fix variable name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cfc7f99d43d83b90fe1498e74781ab7027a5156b"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-19 22:12:31.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2024-10-17 21:46:39.000000000","tz":-420},"subject":"fix x-open-expired 404 on HEAD?part-number reqs","message":"fix x-open-expired 404 on HEAD?part-number reqs\n\nFixes a bug with the x-open-expired feature where our magic header\ndoes not get copied when refetching all manifests that causes\n404 on HEAD requests with part-number\u003dN query parameter since\nthe object-server returns an empty response body and the\nproxy needs to refetch. The fix also applies to segment GET\nrequests  if the segments have expired.\n\nChange-Id: If0382d433f73cc0333bb4d0319fe1487b7783e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/38e813e731a03917545bd39d9a2a6bfdbc115975"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/38e813e731a03917545bd39d9a2a6bfdbc115975"}]},"branch":"refs/heads/master"},"4f69ab3c5dfaa8dcd321bdabb74dc079f7973d82":{"kind":"TRIVIAL_REBASE","_number":54,"created":"2024-10-18 18:33:09.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/47/916547/54","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/916547/54","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/54 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/54 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/916547/54 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/916547/54"}}},"commit":{"parents":[{"commit":"8ded39bccd9192b570e24cf039a3a4c8e3ccc216","subject":"Merge \"logging: Add UPDATE to valid http methods\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8ded39bccd9192b570e24cf039a3a4c8e3ccc216"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-19 22:12:31.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2024-10-18 18:33:09.000000000","tz":0},"subject":"fix x-open-expired 404 on HEAD?part-number reqs","message":"fix x-open-expired 404 on HEAD?part-number reqs\n\nFixes a bug with the x-open-expired feature where our magic header\ndoes not get copied when refetching all manifests that causes\n404 on HEAD requests with part-number\u003dN query parameter since\nthe object-server returns an empty response body and the\nproxy needs to refetch. The fix also applies to segment GET\nrequests  if the segments have expired.\n\nChange-Id: If0382d433f73cc0333bb4d0319fe1487b7783e4c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4f69ab3c5dfaa8dcd321bdabb74dc079f7973d82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4f69ab3c5dfaa8dcd321bdabb74dc079f7973d82"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
