)]}'
{"id":"openstack%2Fswift~916657","triplet_id":"openstack%2Fswift~feature%2Fmpu~Ie33664979dfa853e1eea828533f6b267156e85c0","project":"openstack/swift","branch":"feature/mpu","topic":"p-mpu-batches","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-05-02 18:17:14.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-05-02 16:08:29.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Ie33664979dfa853e1eea828533f6b267156e85c0","subject":"mpu-auditor: iterate multiple batches per cycle","status":"MERGED","created":"2024-04-22 18:34:49.000000000","updated":"2024-05-02 18:18:46.000000000","submitted":"2024-05-02 18:17:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":14,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"916657-p-mpu-batches","meta_rev_id":"7424a532e124e673881fd7ed8bb9afd3efb02f89","_number":916657,"virtual_id_number":916657,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-02 18:17:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-05-02 16:08:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-05-02 16:08:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-22 19:53:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-29 16:16:09.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2024-04-29 22:26:40.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"51d5e26c140f8cb8ab72ea427a1c09978e50b1cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-22 18:34:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0bfe80118d1246b3d2e772b94c96329884d40b93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-22 19:53:44.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a13ccc1b5db24410aaafc3e35a647453\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd80a52006ab4a0a8c5c17c7e4633ad4 : SUCCESS in 8m 08s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6d54e2cdc2e04988b989a26e1698ef85 : FAILURE in 53m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9d9e2c34078a43368f51f24cf5443e17 : FAILURE in 57m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c4d67beb11f64da0b715ac2b8b72da99 : FAILURE in 59m 03s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/71f0586f36e14ff68c414507a33cb202 : SUCCESS in 10m 07s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c3e4be12ee44454284316be5ac8d9f31 : SUCCESS in 5m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/94bbc063dd1d4e1e86dd098078018ceb : FAILURE in 20m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8828dde1c68444d2951ffbf687a29a5c : SUCCESS in 13m 14s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/088765f4e74d448c94bac2b9df91c068 : SUCCESS in 16m 41s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/90e9a63ba092405589dea0e6986c66e3 : SUCCESS in 19m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0732a93fe6774a0592fde218ec02d481 : FAILURE in 56m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/728e6b3dd5ed4b7ca2a68af9223d6e15 : FAILURE in 59m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a30053de3df14a38aef1f032390d2579 : FAILURE in 1h 01m 05s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e5923e01949e45eba4bf2b364d3571c1 : FAILURE in 53m 23s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ced3b3a7d9674c4da3c402eb3730ce52 : FAILURE in 49m 32s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2755059ea60541aeada0f50dca2fe4ea : FAILURE in 52m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0b5b07b9becd4338a921985db266bf5b : FAILURE in 5m 47s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b27c82678d794f0dabcc5c4fd1ea8540 : FAILURE in 58m 42s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/dc4e53465fd449f1a4e74edbc5a11f7c : FAILURE in 7m 34s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1748967d87794a4ba2074d100e54b3b5 : SUCCESS in 56m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3ecda23ed6554729bb227bc3c39faec7 : FAILURE in 20m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecdd03192bfc4b03bec292b88593604d : FAILURE in 3m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0efefa7188dc489494ad3a28fa1b0cf5 : SUCCESS in 19m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"72bd542a3a799a64eb87f4daf263f9f09bd072f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-25 09:38:28.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"011def58e802f2415c6ceb3096ca6f3ce9c84308","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 10:52:17.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6532524bf23e43b183582868225bc82d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25e17832d4ce4fe79946117b4fece4cc : SUCCESS in 7m 32s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cd77b8814ba04c40abf108ea10e62f9f : FAILURE in 53m 59s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9ed4dfedfea24d95afc9441eb879b6c9 : FAILURE in 53m 26s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/84179e4e80ad415586ce19a9496f9570 : FAILURE in 54m 50s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5555dc1dd67c4a16aece5e9ce9354c57 : SUCCESS in 9m 46s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f9f7b1845c184ec9a60448cca5c41596 : SUCCESS in 5m 16s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f08c85b87953452790bfe21b5d82dafa : FAILURE in 19m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ee16c3336a2e49498776d1a35673c535 : SUCCESS in 18m 13s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b50a0833ba3b4a72958f9f8a11a76846 : SUCCESS in 19m 31s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/901d6e65cbeb4deea9d364de5f45f69c : SUCCESS in 16m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a75977349dc5462280af3247d1ee83bf : FAILURE in 54m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/48ba33133bb14ba19c7d9e7c49cf9937 : FAILURE in 56m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8ef0289767494428bf8d640bc1c4a14f : FAILURE in 58m 02s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0029a1aea4584fa08f90e41797a1cbd0 : FAILURE in 50m 12s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/15c4f1ba2f2944e18eff973af7bacbdb : FAILURE in 53m 16s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4d02a7194cd5417fac4c191b7ca9ee25 : FAILURE in 51m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/66c23af9575847fe9f60bc41eb1f7810 : FAILURE in 6m 40s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ae98f518fc0c4710a7c8c60c0edca891 : FAILURE in 56m 30s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/00e3efbae6ac4e01a5edc7a69167aade : FAILURE in 7m 49s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/bf58bc3d987344308511af92ca72bfb9 : SUCCESS in 53m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/05fc5fe50b934ff49b05b299694b58bb : FAILURE in 18m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fe054856fe54cb98a15fe4786972dfc : FAILURE in 3m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6174eb4700c2499d887108024a2e8e37 : SUCCESS in 26m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4eb9598052079b5ba3fafec5c2ac1676e095dfd2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 11:12:16.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42d8187f09ae4b49bc3c3687efb211f5\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d669e5613e3d4fa8bccc575a97c5a5d8 : SUCCESS in 1h 28m 01s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6066129f0427407690b4a7d299dbadc9 : TIMED_OUT in 1h 31m 10s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/58d3ba77a2584f14b8c441f7f1b66db8 : TIMED_OUT in 1h 31m 12s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/abca670971fa4bbc9b3b6fd5625370dc : TIMED_OUT in 1h 31m 24s","accounts_in_message":[],"_revision_number":2},{"id":"a3a5d6584714736e7e704afa2c25e65cc29b687b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-25 15:48:06.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"4999597284c4447ffc515563385e70d4d40965f9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 16:14:32.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40a9c597b2b949c89f1a7b1e469f8a95\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7b2661054e9d46538e4d6cf4c90f8678 : SUCCESS in 21m 20s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7b42fbd2403849ee88d15c9b207c328c : SUCCESS in 21m 18s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/87bb3bd9bcd14a9691bc50df3088f2ed : SUCCESS in 25m 04s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/43b6a95592714663bc72b78e79c08ff1 : SUCCESS in 24m 00s","accounts_in_message":[],"_revision_number":3},{"id":"1845304efdc224a3969ac5c5f6ecefff03b3acaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 17:00:58.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c4ca6a65e99844ada3ff75969cbbe564\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a67de5d89ec0469b8d276733c9b6ea69 : SUCCESS in 9m 29s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2bd15ad2b8094d9b84c558ba7436ff98 : SUCCESS in 16m 57s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bdbadef533ab429abd12f66f5bd66aa3 : SUCCESS in 18m 47s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1572a18c67ec42b99f2af46325506880 : SUCCESS in 18m 03s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b0a93ee642e44e3a8d6eadc260f6b009 : SUCCESS in 11m 39s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d6c8a87434f741b9a01d31bd97d9c075 : SUCCESS in 5m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/804352e0641244d185d07f390ff1a632 : FAILURE in 19m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/178c728555084e3fa9ee7bdb3cdc9905 : SUCCESS in 13m 29s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/48097ad14b5740a9851ad59d598866b9 : SUCCESS in 17m 59s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/996ff8f0c24445f69cf26b0ba2cb9c65 : SUCCESS in 17m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4f896541491c475c8f26c2d5a3f6e6c7 : SUCCESS in 16m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a17ac1e907a349e6b66df94c6c53bf4a : SUCCESS in 19m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8cc194ecac1e43b1940313bc7422241c : SUCCESS in 20m 23s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8bfab5dff01749e2b28161b60ba08046 : SUCCESS in 15m 18s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2d1aac0813174d89a33597dbc68ee306 : SUCCESS in 17m 04s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/50800bdb58e34b5891bd3c9fef4d6088 : SUCCESS in 20m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fb9a21b35f0d4a7fbf45b428e5a788a2 : SUCCESS in 6m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d8f01c8dfad84d2593c7f4b534075a04 : SUCCESS in 17m 38s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/916c05b12d4440bf94540d0c0f0732da : SUCCESS in 7m 47s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1dc77d4868bd458490408df87e906b61 : SUCCESS in 1h 07m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/203fa5b40d8044cbb624fab4bb9a3601 : FAILURE in 19m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41ded1991ea24cb397b075a4b526b117 : SUCCESS in 4m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2aa6cb69c2ae44dea1a959c2de9eac31 : SUCCESS in 19m 10s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8a343b42429c8a8c89bd81256cfa288543af1f8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-29 09:16:41.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"197631545a089bd3ba198f701f58ff1b51fa385d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 10:16:36.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66374f38262e44eab4108bee8509bc4f\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a2d2c067f6a14cfcbcb6cc35858544d0 : SUCCESS in 18m 38s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d7e47bd541f94ac2ab23d801d19e8f85 : SUCCESS in 17m 53s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a93e6ea11b5a4c908a0252053d7d412c : SUCCESS in 58m 00s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6b0a4c20284d41feb07ec03c2baf75b4 : SUCCESS in 20m 34s","accounts_in_message":[],"_revision_number":4},{"id":"716b5acbf364d6f36359dc29ec7c3cfaf9e58611","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 10:33:41.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eed0af7d5964442d8b71b4176f299496\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/761f3937be2f4afd8e28b803cfa9d049 : SUCCESS in 10m 05s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/afd2de18b9324b47afb27ccfcefc187b : SUCCESS in 19m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8c4e05167ef84ab1b89a1ebf5f1d1122 : SUCCESS in 18m 35s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0cc42e64c9964c7caf3c584af1c3c767 : SUCCESS in 21m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f9ed6e0ff03541ea9a9d6b69a1471838 : SUCCESS in 10m 18s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9974f27dde284d5ebfcbac8f88de8586 : SUCCESS in 5m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5769751ba69c46379a560b27f2a6569e : FAILURE in 20m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b5ffb38a5a85403e8fa6e024e6a5d7bd : SUCCESS in 17m 11s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f966e9db587945d4b47e1c45b97d9b7b : SUCCESS in 18m 38s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/82c137d7c4a648d7a60e1aff8a7391cf : SUCCESS in 19m 54s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/17f5a08712f54f9595e8171383ad8fc9 : SUCCESS in 16m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/11dcec0ae9624884931d556378f7a52b : SUCCESS in 18m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b311a3420594469da705f0b42b98747d : SUCCESS in 17m 52s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5838da545a5c48a786e90bb223980185 : SUCCESS in 15m 18s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7ad7e423a7bd4cc3b5ed5a0a24efcbf7 : SUCCESS in 17m 23s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3b1fc8e791764360b93f7d66c1adf40f : SUCCESS in 22m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/18938f32221a4bb8abe050acfd716d23 : SUCCESS in 5m 24s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d0fc93ba99044645b1654c07f7f7d8b1 : SUCCESS in 19m 28s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f34576bd5a3249c18e31924a514f0030 : SUCCESS in 7m 49s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e7890e8764024da592098db341b1c424 : SUCCESS in 1h 00m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4287b56a52b54f1cb772d28ada6397d0 : FAILURE in 20m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d3da87fff5b49c6abe9d7f6528dbadc : SUCCESS in 4m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/87e05741d28240c782247d60dac2dc5b : SUCCESS in 20m 45s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f351f627979100349f8b646da246a915afde55cd","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-04-29 16:16:09.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"dbe754960b3a5595d3bac948b2c968186c7b9d77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-29 16:58:39.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"c4bcef1d6ad305cba5db0e53af890cdab123cf77","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-29 16:59:00.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"8ddf8a5b78e532b77bfbec4ce73b0531ddbb4336","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-04-29 17:25:25.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"a2fdb00b2611563404e6431d8c59c925ff6a8974","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 18:01:32.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/48cdb08dc12a4dbfae156f087b17e02e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6610ef87877a43f58290eb97a4ccf72e : SUCCESS in 10m 13s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5d948cc41a0f40ae92ff029ec9fb887e : SUCCESS in 19m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4d6e4b33fbdf4944bad893fbdd850e1a : SUCCESS in 19m 29s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c42dc9f5f3d4418bbb3e8aa0115144f7 : SUCCESS in 19m 49s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e9dc5b71e86a4ba38c2deed8ed0fe53f : SUCCESS in 11m 31s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1196b4fed90041d29bb8ddc0881c7a80 : SUCCESS in 5m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/07e8c7638fb7445ebc9105435524dea8 : FAILURE in 13m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7bf719d87d144a9f82c0f7fca3e495bd : SUCCESS in 20m 09s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/eef2028f8c4d4f059db50e11b96b871a : SUCCESS in 14m 00s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/e6f0f9b1b0d847f992db16df5ed975b3 : SUCCESS in 19m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/002e348157844e35a9640ed7c706be2e : SUCCESS in 14m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c33aa165bc82432585ddc03c448d275c : SUCCESS in 18m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5d21b7e1c74c4d2fbdffddd4e8a54512 : SUCCESS in 18m 24s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/becdb6702b5045d6ba10603499230c86 : SUCCESS in 15m 48s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/521c957928ff4d5c94d5595e8f96e49c : SUCCESS in 22m 01s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ef64028555014f528ca331ae1795090c : SUCCESS in 14m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6329a6824b6a4a76874f8ce07d5c6fdb : SUCCESS in 5m 46s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f3afcfeea8ad4286946ab18fd7773bc7 : SUCCESS in 16m 29s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/99f6829821a1422483599a165bd4f475 : SUCCESS in 7m 22s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2f55b819314149eb8779b33454fd86e0 : SUCCESS in 56m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f6462a96653e402f9af0e27793228fec : FAILURE in 20m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/109dccdffa2c4da09dc679e5ee48a383 : SUCCESS in 3m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0a14e901da224cfe8938a6f1102b38e0 : SUCCESS in 23m 56s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"dde8c4e54021e145e9e7d82fafcb11e950c1b080","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 18:24:43.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6666ca1827bb47008357568c3869c449\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b19848ebb63741abacfbb4efa41f0700 : SUCCESS in 16m 19s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/80f494019fde4468a3baa9222283cc0f : SUCCESS in 1h 23m 59s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b31e6e448e424cde991b84a7df98419b : SUCCESS in 21m 25s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/44a4e17465954a30afa60d793438d496 : SUCCESS in 59m 00s","accounts_in_message":[],"_revision_number":5},{"id":"2a8cbe15038b4fb5dfab62677c97914abda008f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-04-29 22:03:54.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"c42dfd0f2340e275d7935f8244bd872163eb1217","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 22:24:29.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e42f8beea0db4823ab7e954ca0437d5a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b9460fb173e041fa9989e96e9b107a43 : SUCCESS in 18m 37s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8d3c0f2a14cf4585846d8b8d130bbb61 : SUCCESS in 19m 26s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/12436b6ed7034bcbaa1da7be7dd10e6e : SUCCESS in 18m 01s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4d0504bf26fe43f383cb889f2ddb5ce9 : SUCCESS in 18m 26s","accounts_in_message":[],"_revision_number":6},{"id":"c6cf5bb19c5323511f58832d389271357b901c95","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-04-29 22:26:40.000000000","message":"Patch Set 6: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"c08d6177520f7deb39f476f0ff8e0a0115b814e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 23:02:31.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f81782620d1945548a5630eebc570c72\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7431132ea611400688cab31511e31253 : SUCCESS in 9m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6726359f189b42cc950526c71ba551a4 : SUCCESS in 17m 20s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a975509b75e2436a90fd5619450cd8de : SUCCESS in 20m 07s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c1743678450f441f82b60c47bbceac2d : SUCCESS in 15m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1b39697daf124d70b9182009bf7d5f58 : SUCCESS in 10m 27s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9c83bcdd01584ba2bf4a1d28213c794d : SUCCESS in 5m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3e0a0e615744479babebf11aab190bab : SUCCESS in 18m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b6460217cdf442a8bc288b259cc1f90c : SUCCESS in 18m 35s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/af78815785ec4d35b2a6a5b5099aa2dd : SUCCESS in 14m 54s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/e35645c7f1ab4b1b896316ed62cd03da : SUCCESS in 17m 21s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2bba1d26688941aa86db79fed2421b64 : SUCCESS in 16m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d6a3f4da0fc341cea15e52ab58d120b5 : SUCCESS in 18m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0d867d4e19794b77bf59fe82585c6744 : SUCCESS in 17m 41s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9653197b0f3640b5b8a64930d4f33675 : SUCCESS in 16m 19s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4742061bc1e64f41861d7eaae1f5cfc8 : SUCCESS in 20m 17s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/37549354af1c42fab16e3064d3faf8c9 : SUCCESS in 18m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4e44ee03754e44f18c69fb6a9bcb1a80 : SUCCESS in 5m 45s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0dfcd353367f443bb009c09e0c023bcc : SUCCESS in 17m 11s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/916be3d3e62949fd90c545199d1d855e : SUCCESS in 7m 02s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/43efabd86bc3485883692518488bb16f : SUCCESS in 57m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fd70e0d37c8044b0b4c07b39afb08bff : SUCCESS in 19m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d7caffaf9b44e81aab53b590cd4f0e2 : SUCCESS in 4m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5180ab98ced44e6eb2cb3cc5d5809e00 : SUCCESS in 22m 53s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"6e7763632d6750e0a0cd024a8bfdeb906f247a55","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-30 09:28:18.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"170d0b0e8a5b4027f6e5dc30d98824d7de758c4e","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-05-02 09:25:44.000000000","message":"Patch Set 6:\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"acdba1669d210a3038cd6e12dbd167d043198962","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-05-02 09:26:31.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"15b47162f819448e840feae097eb2796d8f01d47","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 10:59:35.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b99fb0b1cada43dda76695269cd19e97\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/083125e603814447b1d1914dc282c2a3 : TIMED_OUT in 1h 31m 15s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b716d5d5d1944d2bbd97bf72d22a315c : SUCCESS in 20m 08s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f0bd47086e2c4b3cac5a7d682e648199 : SUCCESS in 22m 54s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/bf204bb9863449308437b582ef71c69a : SUCCESS in 1h 14m 45s","accounts_in_message":[],"_revision_number":7},{"id":"d6f2b47b503f7298a72a6e100ba23ae4c384dd1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 11:38:36.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5e3cd81d404c49e4b0d3d8b9558323c7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/691e9323368d492d8c1a424935782737 : SUCCESS in 7m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3f888487ac294b8faab1882a49c1c3bb : SUCCESS in 18m 03s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/479dd5108a3242b7b79750a3ed36f13a : SUCCESS in 17m 59s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d08a16a21bd34d469eb960313d3ce19a : SUCCESS in 24m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8f055fea2d5943749b8e41a154ce08dc : SUCCESS in 12m 28s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ab5f86ce370d44098ae0e93262af06de : SUCCESS in 8m 16s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a8aabb5bafd04add93333d5bcdaf524d : SUCCESS in 17m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e3a38806679245e4821f394038355592 : SUCCESS in 20m 05s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/bae832b947ac44548b57e79ce46e7077 : SUCCESS in 16m 04s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/ca0425a560f44ad3a152f96382681746 : SUCCESS in 17m 54s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/857a9d7a8cb749cdbac428e51d2da1cf : SUCCESS in 17m 37s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7229884f1a3e4229a41bd20318d603d9 : SUCCESS in 17m 41s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4e38c1ca39e4473e9768bc15db1c0879 : SUCCESS in 20m 01s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7ee772014933423f9b738eb1b8a29c62 : SUCCESS in 16m 12s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b454fc7cd4924f488b41cf7a8b3efaad : SUCCESS in 16m 02s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/43141d0acbdf469698c94dd9101e143a : SUCCESS in 19m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2d7845ec5f33476899a056f958a786b2 : SUCCESS in 5m 45s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4265bb5148d743b6b5cfa0870c3543ea : SUCCESS in 19m 44s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6d9a1521efd341d28d84454e51310b3f : SUCCESS in 7m 21s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/410a1544a53a49a899389026ca8b390c : TIMED_OUT in 2h 00m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8495fbe95e0c44139ad150fc57d754d2 : SUCCESS in 25m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a1d3be32fbc4d32bac277291cfdb62d : SUCCESS in 4m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0a41a0e6802e43a68da0c59cceb54ee8 : SUCCESS in 19m 33s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"05589af3e03d2141dd95e670265c4c7049b48f6a","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-02 16:08:29.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"e1b24763b014863f7d5aa08a03015edcb098e4de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 17:07:57.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc62e59e49ef431396901a0514b6fb8e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3eea00d37854748a62b49b328f7b852 : SUCCESS in 8m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/48c0882e1f474f96a8dd3b203ee4cf8c : SUCCESS in 25m 17s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3e8c1c29fa484d2b869ba40a3121e985 : SUCCESS in 22m 16s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/60599a965c9943f29f8a9a6cf4f00b39 : SUCCESS in 23m 43s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7b1b75878cfa4af4a81a7887f5b06dec : SUCCESS in 11m 03s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e46c6088f40d41da8d8be945e442e5f7 : SUCCESS in 5m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/43d27145afb94845a172aa66edf11a79 : SUCCESS in 18m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8177ed80cdf4486199b35cfea1c93f3f : SUCCESS in 16m 58s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ed1e2e1c79b94d2bae507b3a54193294 : SUCCESS in 18m 57s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/698280fbb1584717808ca98be91a624a : SUCCESS in 16m 20s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/705f37db2f8843e7bbb89299249a5ea4 : SUCCESS in 16m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e2b541cacfbf410e8f07f8473b9b56ee : SUCCESS in 18m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c0dcb1ea912a497bb3bb40a34e57a485 : SUCCESS in 19m 35s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/dde4772765484f6e970c28e2583ed9d5 : SUCCESS in 16m 58s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1792b713868d4167b84f512099adb644 : SUCCESS in 16m 29s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f6106b31c2054495bfe0e2743ed2a41c : SUCCESS in 15m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c35320a4355646d5a1be7a08227b520e : SUCCESS in 5m 46s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d5615e1c3ab24c1282cb4cb21eb319a9 : SUCCESS in 15m 56s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/eba746ccf2754593b128df18e9e44df8 : SUCCESS in 7m 51s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c263caa098024ff3be8b5beaa206a13b : SUCCESS in 54m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/21f519f1803c4caf8ecd4c042b4e6f61 : SUCCESS in 21m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61ed844c71314f13b217f1fe8fcba875 : SUCCESS in 4m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4dd31214f4e84bdba04798fea82c6ba2 : SUCCESS in 22m 31s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"73e505d497c4eb2d39a4a8a23a8cd31a4923355b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 17:08:19.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"370de767be8f13a6b36105111895ed2611fc4ec9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 17:21:14.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/838a305d580440d5a0f285f0278d8689\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cc0e60d5e0094ee3af42c5892d0488fe : SUCCESS in 15m 15s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7796fb69b7f04a74b274bb3b5ccbbe9d : SUCCESS in 16m 41s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3624504d98034ef199e6298595e3f6a9 : SUCCESS in 1h 10m 55s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/5d3f41c8a55a46af82a0527c3a37d54a : SUCCESS in 20m 45s","accounts_in_message":[],"_revision_number":7},{"id":"d88961e36ca693595798384d7f73056a24a494c1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 18:17:14.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2ed17ca215245ba85073cf32726153b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/730004d0e8a34376a769796493eac64d : SUCCESS in 8m 10s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/b412aa41155a4ae28dab1089a47105bf : SUCCESS in 10m 16s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/b0d05eedbeef4a9ab02c7a1c4e79c7fc : SUCCESS in 5m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ee8a892ccad64b2bbb60da389cfdf850 : SUCCESS in 20m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3440fc318d504803a564ce5497cd4090 : SUCCESS in 20m 11s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a7e2be2bad22453b9abc20346b15a010 : SUCCESS in 19m 09s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/5d7ef5ee7ad04a6fa52726958f8e794d : SUCCESS in 17m 39s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/97fb5143b5b241bb8fabac963cae9651 : SUCCESS in 18m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b9c93c7e17444adf8663fd1908b0e27d : SUCCESS in 20m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/13d7cd1a4070413e83bf478da04941d0 : SUCCESS in 20m 07s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6522f0a8b2d6452e9c0037157ca33056 : SUCCESS in 15m 21s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/267112bb7c3f4a8f877de98b71ea358a : SUCCESS in 16m 44s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/cf616a947cbe4668847dd7750e4243e9 : SUCCESS in 16m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c988a655215549ab8494e18767e2acc9 : SUCCESS in 6m 35s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/12479a03e6e546ceb849feb1d272c486 : SUCCESS in 8m 00s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/10df73aa38eb47f7a419c731346ffcd7 : SUCCESS in 57m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/53c012d8faee43408746931cc63fd48e : SUCCESS in 24m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6a55275657f4ae6bdf58d5b692fcd51 : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":7},{"id":"44c2076e8deed7c36366a5b804fbdb044f07817c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 18:17:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"7424a532e124e673881fd7ed8bb9afd3efb02f89","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 18:18:46.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7704a6bb6614f5cacc2af19f72e82df\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48b19a33cb9740c3a6e78c4a85cae057 : SUCCESS in 1m 10s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/0c8a4fdc04484e6b85ac9534b64f782d : SUCCESS in 33s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/06cb4d6121a346cc9329b3b149e9cf15 : SUCCESS in 36s (non-voting)","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"9eda38ced93274e71ce974251514d80e84426799","revisions":{"ed6f55b893a770faa92b51c575224b4a4a72caae":{"kind":"REWORK","_number":1,"created":"2024-04-22 18:34:49.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/57/916657/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/916657/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/916657/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/916657/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/916657/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/916657/1"}}},"commit":{"parents":[{"commit":"8419a7a0907f9c013cd747e97e98b13ffbc29aee","subject":"Add abort upload support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8419a7a0907f9c013cd747e97e98b13ffbc29aee"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-22 15:15:28.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-22 15:15:28.000000000","tz":60},"subject":"mpu-auditor: iterate multiple batches per cycle","message":"mpu-auditor: iterate multiple batches per cycle\n\nChange-Id: Ie33664979dfa853e1eea828533f6b267156e85c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ed6f55b893a770faa92b51c575224b4a4a72caae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ed6f55b893a770faa92b51c575224b4a4a72caae"}]},"branch":"refs/heads/feature/mpu"},"63c73283ec10524c4481204bdc159cbb174a71af":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-04-25 09:38:28.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/57/916657/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/916657/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/916657/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/916657/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/916657/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/916657/2"}}},"commit":{"parents":[{"commit":"ba28d9ae1598cd842656d235dc31aae0e277acec","subject":"Check for user container existence","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ba28d9ae1598cd842656d235dc31aae0e277acec"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-22 15:15:28.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-25 09:37:52.000000000","tz":60},"subject":"mpu-auditor: iterate multiple batches per cycle","message":"mpu-auditor: iterate multiple batches per cycle\n\nChange-Id: Ie33664979dfa853e1eea828533f6b267156e85c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/63c73283ec10524c4481204bdc159cbb174a71af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/63c73283ec10524c4481204bdc159cbb174a71af"}]},"branch":"refs/heads/feature/mpu"},"c0204a19a8b65f78cb4c954f3cd9aafb38e42836":{"kind":"REWORK","_number":3,"created":"2024-04-25 15:48:06.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/57/916657/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/916657/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/916657/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/916657/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/916657/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/916657/3"}}},"commit":{"parents":[{"commit":"51eb0e75d2aaeba4c78cc4c4e2a6cf550f10caff","subject":"Check for user container existence","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/51eb0e75d2aaeba4c78cc4c4e2a6cf550f10caff"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-22 15:15:28.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-25 12:39:30.000000000","tz":60},"subject":"mpu-auditor: iterate multiple batches per cycle","message":"mpu-auditor: iterate multiple batches per cycle\n\nAlso add some basic exception handling.\n\nChange-Id: Ie33664979dfa853e1eea828533f6b267156e85c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c0204a19a8b65f78cb4c954f3cd9aafb38e42836"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c0204a19a8b65f78cb4c954f3cd9aafb38e42836"}]},"branch":"refs/heads/feature/mpu"},"dea863b7b31f4d87ee21a3e51057f12a21690868":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-04-29 09:16:41.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/57/916657/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/916657/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/916657/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/916657/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/916657/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/916657/4"}}},"commit":{"parents":[{"commit":"de0d2c37a075317eea687769b82f7ba6dfc0a57f","subject":"mpu mw: pre-authorize MPU-handled requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/de0d2c37a075317eea687769b82f7ba6dfc0a57f"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-22 15:15:28.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-29 08:09:46.000000000","tz":60},"subject":"mpu-auditor: iterate multiple batches per cycle","message":"mpu-auditor: iterate multiple batches per cycle\n\nAlso add some basic exception handling.\n\nChange-Id: Ie33664979dfa853e1eea828533f6b267156e85c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dea863b7b31f4d87ee21a3e51057f12a21690868"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dea863b7b31f4d87ee21a3e51057f12a21690868"}]},"branch":"refs/heads/feature/mpu"},"9485d690f1228dc511df7f0823a15722c5ff0a80":{"kind":"REWORK","_number":5,"created":"2024-04-29 16:58:39.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/57/916657/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/916657/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/916657/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/916657/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/916657/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/916657/5"}}},"commit":{"parents":[{"commit":"de0d2c37a075317eea687769b82f7ba6dfc0a57f","subject":"mpu mw: pre-authorize MPU-handled requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/de0d2c37a075317eea687769b82f7ba6dfc0a57f"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-22 15:15:28.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-29 16:57:43.000000000","tz":60},"subject":"mpu-auditor: iterate multiple batches per cycle","message":"mpu-auditor: iterate multiple batches per cycle\n\nAlso add some basic exception handling.\n\nChange-Id: Ie33664979dfa853e1eea828533f6b267156e85c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9485d690f1228dc511df7f0823a15722c5ff0a80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9485d690f1228dc511df7f0823a15722c5ff0a80"}]},"branch":"refs/heads/feature/mpu"},"1d9bab849ca069deb18c019eee5bfaae56123dd6":{"kind":"REWORK","_number":6,"created":"2024-04-29 22:03:54.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/57/916657/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/916657/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/916657/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/916657/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/916657/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/916657/6"}}},"commit":{"parents":[{"commit":"de0d2c37a075317eea687769b82f7ba6dfc0a57f","subject":"mpu mw: pre-authorize MPU-handled requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/de0d2c37a075317eea687769b82f7ba6dfc0a57f"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-22 15:15:28.000000000","tz":60},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-04-29 22:03:04.000000000","tz":-420},"subject":"mpu-auditor: iterate multiple batches per cycle","message":"mpu-auditor: iterate multiple batches per cycle\n\nAlso add some basic exception handling.\n\nChange-Id: Ie33664979dfa853e1eea828533f6b267156e85c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1d9bab849ca069deb18c019eee5bfaae56123dd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1d9bab849ca069deb18c019eee5bfaae56123dd6"}]},"branch":"refs/heads/feature/mpu"},"9eda38ced93274e71ce974251514d80e84426799":{"kind":"REWORK","_number":7,"created":"2024-05-02 09:26:31.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/57/916657/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/916657/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/916657/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/916657/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/916657/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/916657/7"}}},"commit":{"parents":[{"commit":"ba398e15ede580f48464a213361daed749d57f5f","subject":"Merge remote-tracking branch \u0027gerrit/master\u0027 into feature/mpu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ba398e15ede580f48464a213361daed749d57f5f"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-22 15:15:28.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-05-02 09:23:23.000000000","tz":60},"subject":"mpu-auditor: iterate multiple batches per cycle","message":"mpu-auditor: iterate multiple batches per cycle\n\nAlso add some basic exception handling.\n\nChange-Id: Ie33664979dfa853e1eea828533f6b267156e85c0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9eda38ced93274e71ce974251514d80e84426799"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9eda38ced93274e71ce974251514d80e84426799"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
