)]}'
{"id":"openstack%2Fswift~916694","triplet_id":"openstack%2Fswift~master~Idc5d336562a3313467f690583b77218c2994fc9b","project":"openstack/swift","branch":"master","topic":"open-expire","hashtags":[],"change_id":"Idc5d336562a3313467f690583b77218c2994fc9b","subject":"sq: stronger test faking to avoid container request leak","status":"ABANDONED","created":"2024-04-23 00:04:24.000000000","updated":"2024-04-26 15:04:49.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"262c234e0905bfdbfef6e467c8a33c47fce53681","_number":916694,"virtual_id_number":916694,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-04-23 00:52:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-23 00:51:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-23 00:52:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d191ab5e22bb3aecce42cf7e84ab22ea267c5420","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-04-23 00:04:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"91ff13a9f32031831d97dd5085b9fa6ca520634b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 00:51:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b34ad3b5eea64f5db774cd7e1db673be\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ee028d63231e43cea7c8e275f265c419 : SUCCESS in 16m 38s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/842024144644490b9faf963a7ddaac36 : SUCCESS in 17m 02s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8253adfbb1dc44948741bf9185a348aa : SUCCESS in 45m 36s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a0c66fd200a740c9b30f9aca38988cb4 : SUCCESS in 20m 51s","accounts_in_message":[],"_revision_number":1},{"id":"990da6647d0550cbe2996eaed3f9268f1d856f68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 00:52:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ee13e69d7cd4df2bf24f3368c64465e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b297966636f4a2885bd1e2fda585341 : SUCCESS in 8m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/070200d3de714378a2e372fb7a591aeb : SUCCESS in 16m 31s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bee03722dc0347c88ad265e22896dd01 : SUCCESS in 17m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0a6cd18b6c884768861a299cf5d03a8f : SUCCESS in 17m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d4aa4ae6b4714907bfff78ecc3e4e533 : SUCCESS in 11m 56s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/787ae2f0934d4bdcbfe8a273d9ea1b85 : SUCCESS in 4m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a6bc5f5b9c8a43fbad9a7ed2f7984899 : SUCCESS in 19m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/11cdcf80d82447e9acc80ef49ba38dfc : SUCCESS in 13m 23s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5df4130b5a304199a1f8019069fdcf7e : SUCCESS in 17m 25s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/721559b5ad5d4621a841483f35c438dc : SUCCESS in 20m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a20cead156d245e1a51537a846ed0dd1 : SUCCESS in 17m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/07f0ada1f8844d90a9ac138e68904222 : SUCCESS in 15m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8987c21b910543f089487c113e4c03dc : SUCCESS in 16m 52s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f08ab98574aa497ab454a992212ec977 : SUCCESS in 14m 59s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/03fdbd91056e4cae8cb740230b8c61ad : SUCCESS in 15m 53s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ae2c5af0e17a4e39890d56cac51f24ce : SUCCESS in 13m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6e5590b49a474384b4c67dd590bb7b96 : SUCCESS in 33m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8b7c90aee5a8427cb3d0d82902bf1c65 : SUCCESS in 33m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2b09ab8947534523933a58563be2a88e : SUCCESS in 19m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cecca31e90c84abc8810ccfbf68735d8 : SUCCESS in 3m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c0c92a3ca2384ec7b5518daf3a8c81de : SUCCESS in 17m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"262c234e0905bfdbfef6e467c8a33c47fce53681","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-04-26 15:04:49.000000000","message":"Abandoned\n\nsquashed 912496: add bytes of expiring objects to queue entry | https://review.opendev.org/c/openstack/swift/+/912496","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5a6c56a4fb5773aef96f95911dcc3701294fa938","revisions":{"5a6c56a4fb5773aef96f95911dcc3701294fa938":{"kind":"REWORK","_number":1,"created":"2024-04-23 00:04:24.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/94/916694/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/94/916694/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/916694/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/916694/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/916694/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/94/916694/1"}}},"commit":{"parents":[{"commit":"2f4e6bc410e93e7911d994f72af5a53139de3fe7","subject":"add bytes to expirey queue entry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2f4e6bc410e93e7911d994f72af5a53139de3fe7"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-04-23 00:04:16.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-04-23 00:04:16.000000000","tz":-300},"subject":"sq: stronger test faking to avoid container request leak","message":"sq: stronger test faking to avoid container request leak\n\nChange-Id: Idc5d336562a3313467f690583b77218c2994fc9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5a6c56a4fb5773aef96f95911dcc3701294fa938"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5a6c56a4fb5773aef96f95911dcc3701294fa938"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
