)]}'
{"id":"openstack%2Fswift~916907","triplet_id":"openstack%2Fswift~unmaintained%2Fyoga~Id56057262fc37add3a47c50946067f8c873c7515","project":"openstack/swift","branch":"unmaintained/yoga","topic":"fix-zuul-config-error-swift","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2024-05-07 11:36:48.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2024-05-07 11:31:09.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-05-06 15:49:23.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Id56057262fc37add3a47c50946067f8c873c7515","subject":"Clean up centos-7 based jobs","status":"MERGED","created":"2024-04-24 13:39:20.000000000","updated":"2024-05-07 12:47:53.000000000","submitted":"2024-05-07 12:44:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"916907-fix-zuul-config-error-swift","meta_rev_id":"1b6b01be48e3a1610f32d27e9f92fc92a1dd068c","_number":916907,"virtual_id_number":916907,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-07 12:44:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2024-05-07 11:31:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-05-06 15:49:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2024-05-07 11:36:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-24 13:39:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-24 13:39:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-24 17:47:47.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2024-04-25 19:52:38.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2024-05-06 15:49:23.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"e81756ab8e53a4ae1f497b47eba49128375667f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-04-24 13:39:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2ca6949147da69b89e6f20b2ddf243898a76b4fa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 13:39:44.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/swift on branch unmaintained/yoga.  The\nproblem was:\n\n  Nodeset swift-five-nodes does not match existing definition in branch feature/losf\n\nThe problem appears in the the \"swift-five-nodes\" nodeset stanza:\n\n  nodeset:\n      name: swift-five-nodes\n      nodes:\n        - name: test-runner1\n          label: centos-8-stream\n        - name: proxy1\n          label: centos-8-stream\n        - name: account1\n          label: centos-8-stream\n  ...\n\n  in \"openstack/swift/.zuul.yaml@unmaintained/yoga\", line 327","accounts_in_message":[],"_revision_number":1},{"id":"ad3e070fb1c22ff1d0e55fd372d20db5bdc40dd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 13:39:55.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/swift on branch unmaintained/yoga.  The\nproblem was:\n\n  Nodeset swift-five-nodes does not match existing definition in branch feature/losf\n\nThe problem appears in the the \"swift-five-nodes\" nodeset stanza:\n\n  nodeset:\n      name: swift-five-nodes\n      nodes:\n        - name: test-runner1\n          label: centos-8-stream\n        - name: proxy1\n          label: centos-8-stream\n        - name: account1\n          label: centos-8-stream\n  ...\n\n  in \"openstack/swift/.zuul.yaml@unmaintained/yoga\", line 327","accounts_in_message":[],"_revision_number":1},{"id":"62edece8270759c6dac669fe67b4005bfb6d7119","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-04-24 13:53:20.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"fc663baed397d54eccce2788b44c80a985dd97f0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 15:49:27.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/399c985bf06744f39f21821a371076b0\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/350df597ee834fc0a1167770c3162b13 : SUCCESS in 1h 11m 27s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/21da5a6bb8a54146a1ccfa8f00bb319d : SUCCESS in 1h 19m 43s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/90d9e0b2715a4469ab110516a526437c : FAILURE in 1h 54m 57s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d655545bc2484a85a7643ae48d314f09 : SUCCESS in 1h 13m 56s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/2590d5b5c4654f10862e93224d7363b1 : SUCCESS in 36m 16s","accounts_in_message":[],"_revision_number":2},{"id":"e693c0101b7cc500a566414858c490a35c0e15fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 16:02:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/890d6c87229d4d368bf7f3d3884bf19a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1861f03bbed43af87c73378e82fa615 : SUCCESS in 8m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/0067695f745b468bb178fb1d634084e0 : FAILURE in 36m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ec4d427b857a4c6386b51ffd64e584f6 : SUCCESS in 1h 03m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ac2ac7c16fda476e9e7e442b531a9a01 : SUCCESS in 47m 37s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/0e5b759a972b4a52930135438c77b985 : SUCCESS in 21m 19s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/c6b6df13b1fb41ffb197dfb6d61a66fd : SUCCESS in 23m 07s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/d453dfeb28d4446695287be9201bc41f : SUCCESS in 21m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5ae959048c934ff988b70151ea55190f : FAILURE in 3m 45s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0e3939d73582431396f1fcb4ba07acb9 : FAILURE in 4m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/daef796e125a46d587a4b781efe4cd3c : SUCCESS in 16m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0b4285773a0344ffb5db0640bf616a84 : SUCCESS in 15m 40s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/520bdd3da41043f88fe260e40e8807fa : SUCCESS in 12m 48s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f7d08e44df874738be09081b34cec338 : SUCCESS in 16m 52s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/dd3ee8d98253431da5c39cc118704974 : SUCCESS in 17m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f806b7ee553643fda6bb98c8446bc210 : SUCCESS in 16m 23s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/90bc55b6f932439089a0f7720a74210a : SUCCESS in 14m 14s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/fae0fee1546041c6b2bfe9859362466d : SUCCESS in 15m 24s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4f4f93aecf074a99b97b7575f56f2ea2 : SUCCESS in 16m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ac491caa17554941909f467ce51da89f : FAILURE in 6m 31s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/c68b52eb87dd4f2292caa48ac25452e4 : SUCCESS in 53m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/aa24a5d019f6410c8abde190b45774ea : SUCCESS in 46m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/db5b06e09bf644d098a5143a7123ff3d : SUCCESS in 46m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8bc6a510483044909299379ae26a869f : SUCCESS in 15m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74563d295ece471e9e48570b21f654c3 : SUCCESS in 5m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8cd52b59b20f42e3b12ca5f417a88274 : RETRY_LIMIT in 4m 10s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c6778c688386410e8b31d7ab0756d27d : SUCCESS in 36m 37s","accounts_in_message":[],"_revision_number":2},{"id":"a7ac07392d132b15186e5aded32462b08a079567","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-04-24 16:17:28.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"65e04b4adbd661230f6f7eb9be6a9b0f35c10680","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 17:41:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b12ae7614a64f78901abed0af5dbb90\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f520ef191ba4d178e89682302c927a1 : SUCCESS in 8m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/5f54ce5a70314a00b2c6e9eed1b56ce0 : FAILURE in 48m 37s (non-voting)\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9af762d830f743c5a18a4bdde3bcded7 : SUCCESS in 38m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3e42f6e48f2c4c39b9e59989385939b2 : SUCCESS in 42m 06s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/7646efaf70c54661886ae75ed32b13d9 : SUCCESS in 17m 15s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/b1ba8153ecfe4c31b1cf0a36799d8b2a : SUCCESS in 24m 01s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/b96e6f1903c149a0bee03f7197f43da9 : SUCCESS in 21m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/338b8690ae264341b7efa6b5038a92c6 : FAILURE in 3m 40s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0c55d49cb90d4bcd9ba06041381292fe : FAILURE in 3m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b48aea4bff2d463d92deed1a5cf109d4 : SUCCESS in 17m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e97a2cf32bea4eea9cbdd700aa6733b1 : SUCCESS in 15m 51s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a758e9518ee643afbd799bcc10d8aaa3 : SUCCESS in 16m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d7ca1b07b0ab4f36a3377489315e3d3a : SUCCESS in 15m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/97ab7a477f8943a4ba73bcf8f7b1162c : SUCCESS in 15m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4630a676c54e4a81bb6886777670b90f : SUCCESS in 15m 27s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/285903b1895540a880cc215bcde005e1 : SUCCESS in 14m 00s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/05dca9f765114cb091d16f5bf9f46c03 : SUCCESS in 16m 04s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4ad03617027b45019f926531bd88c6f2 : SUCCESS in 16m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fabb67b6716e4c8982a55cfebf6eaf4b : SUCCESS in 6m 23s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/cfac58f88d9842c38956bf412668ae84 : SUCCESS in 59m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d2b97b87fcb1442987da7b0f704c1c03 : SUCCESS in 38m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a4625285f3db4422bf79f708f63f8612 : SUCCESS in 38m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/22179d5300b04ad99b8a56d446e4cb6c : SUCCESS in 15m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b667a2b878184f46a6536b2d5eaf9132 : SUCCESS in 5m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1a889dba07964555bf656601488f30cf : RETRY_LIMIT in 3m 38s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c557be033d014128ae03c1dea89aff97 : SUCCESS in 50m 45s","accounts_in_message":[],"_revision_number":3},{"id":"68bd82b8bad8f6c2a9afbebaaff8300621ca8d90","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-04-24 17:47:47.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c05336af4e15d26edd1714c324f13b20fee4e135","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 18:53:34.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/017e02cc137b4677a3e07f60cfb4f8b5\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f7a35a71b4aa49e38ff2cbe741727935 : SUCCESS in 23m 19s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9338e3c458e247d9816a5d1e5c6f9a4d : SUCCESS in 36m 17s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/ce14f8c9fcb5480db02d2ccf48241a17 : FAILURE in 1h 57m 37s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a3315a63778f45f19ca068c11218125f : SUCCESS in 1h 19m 14s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/c7cd707741d940e1a8bf72480b21f1ef : SUCCESS in 37m 18s","accounts_in_message":[],"_revision_number":3},{"id":"b57697122e511f0c708a8f4f80e8c91edb5788f4","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-04-25 09:41:16.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"d5ccbef865efd3002d58df129a17364f270b249b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-04-25 14:33:20.000000000","message":"Topic swift-zuul-config-error removed","accounts_in_message":[],"_revision_number":3},{"id":"3b02a00f017c4f2b65d6a675d9454a16378f62a5","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-04-25 14:33:28.000000000","message":"Topic set to fix-zuul-config-error-swift","accounts_in_message":[],"_revision_number":3},{"id":"5197915c1890ca34cc7fb9ef65ed301f665f865b","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-04-25 16:00:59.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"abcd64b066c5a73a67fb7b9f0d050f96fe85e39a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 16:01:49.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b4e81b60ee1a7ae46487be17c3199ee8fa1dbf7d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 17:31:45.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3fd4ab848be2441aae1036f5bfebe2c8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1148e64582c2493a88d67af532d6973c : SUCCESS in 8m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/00c2352ce0d347b59f7ec68dca164a44 : FAILURE in 46m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ed114c027be94787a0ad845f4c52f0af : SUCCESS in 50m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5a3185d08c15450e84123f3249cb1037 : SUCCESS in 49m 12s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/60969975ef8444239a282e81511c9cdc : POST_FAILURE in 5m 16s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/c6fe63ba37314c92bcab909f1acd465e : POST_FAILURE in 7m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e2368a7001964370a6b4f5a9c8b8accd : SUCCESS in 17m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0a0cd5ed2024401499a3d568b81b2178 : SUCCESS in 16m 00s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6e55313ba8ea41748ada03830fdb8444 : SUCCESS in 16m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8125ff8ad9174b2fb286073dc54c6fdd : SUCCESS in 15m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/65b1740b8195445ea061e7424e32cddb : SUCCESS in 17m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ee85dcb7051b4d4e919e5e6989c7f1dc : SUCCESS in 17m 56s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9ef86376e09e40f09b4012bbe8702ae1 : SUCCESS in 13m 01s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7de832eb2a0b4d1fbc75d76fb8387cd0 : SUCCESS in 17m 25s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4b34fc6555b749e49f4f75687154d114 : SUCCESS in 17m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cb510f12594243ba9a96010bb2aca4c5 : SUCCESS in 6m 04s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/8f5f12292e18496abfb708b22bc38f2a : SUCCESS in 52m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1695ad359e1c4ffa8c362c040a41b135 : SUCCESS in 38m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/630bfbb7b2074758a953f7054fea0cd9 : SUCCESS in 45m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a481e93b635940cb8e5e99a987a405f3 : SUCCESS in 17m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61ba852d5f4f41f5aec7e4a2f58d280e : SUCCESS in 5m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f1ec26edcbfb43699b28778e549d338a : SUCCESS in 42m 46s","accounts_in_message":[],"_revision_number":3},{"id":"4f1253f397177b460ae64035df47c18cccf7afb8","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-04-25 19:52:38.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a482f95c472bcf809de69fb68ff253d869f28383","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-04-26 08:25:39.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"10ddc7a738a916b02f4d97fb7719c38d4592dba8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-04-26 08:38:01.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":4},{"id":"e092ba7f96217811ba775898b754ddcd350fa669","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-04-26 09:04:17.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ade92b2c44f4f786dc6065846c4b607b4c5ed364","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-04-26 09:08:54.000000000","message":"Patch Set 3: -Code-Review -Workflow","accounts_in_message":[],"_revision_number":3},{"id":"0bb578ff3b862623375f1be83757fb22529ed087","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 09:43:25.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3101a85876ab43b8b8d6b89ecdcfb8d4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab22e0f033924341b4a468c035d48771 : SUCCESS in 8m 33s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/39dd3e4e3197474f8932d585cc260d27 : SUCCESS in 20m 31s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/cff61bc0055e4902aa8e608c386e6c64 : SUCCESS in 21m 12s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/02ab4b9e048e4f1ba4cfa0ffb5fa8a1a : SUCCESS in 20m 28s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/48bec85204324f24a2f95f28e37019c8 : FAILURE in 3m 36s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/50437c1dde144e9ba8f701ecc7b0cd4c : FAILURE in 3m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/84d75d9c6cdc41d1a93dd4d9775c6abb : SUCCESS in 16m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/34dede530c3544dda4d4b9791f67a125 : SUCCESS in 15m 43s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7e9b0d7a8e424a10ad5ab4278c185f69 : SUCCESS in 14m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/20d967db05a04eed98b766b455805528 : SUCCESS in 19m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b4bdf103bdae45f5ba7b72ebf91390dc : SUCCESS in 20m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/609f261733d64427b589ab64d4c496a8 : SUCCESS in 18m 20s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/881f4bcd37c34ad2a42d86f06d0093a3 : SUCCESS in 15m 03s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/cbdeed010cba482fa437264b59a251b6 : SUCCESS in 15m 03s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d7c16aedfed24760a4f9dbc5415b3312 : SUCCESS in 21m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9544465f37db47e99d4c1155346ee299 : SUCCESS in 7m 42s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/4fa2577d32f64e8482d3daf101f7eb81 : SUCCESS in 1h 00m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/06d60def250c498897cebaf559d8c81e : SUCCESS in 32m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0ee5b6fa7f9e42adaf75a37601b60407 : SUCCESS in 48m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3c3df7f9064444bfbdb5d02800ddea36 : SUCCESS in 14m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bab02f9c9ca47c2a856f9079fce7238 : SUCCESS in 5m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9f1f6cd5dc8d4b1390997499b96558ca : RETRY_LIMIT in 4m 09s (non-voting)\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bb0793083047448dbeb59bc0fcc35bc4 : SUCCESS in 1h 04m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/26657de681f541789d3e5f6bb244679c : SUCCESS in 1h 01m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/31a21951a9a34c1680aa37285b7bd634 : FAILURE in 47m 50s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/968a6fe8a2ad46d8913af4f7891edb84 : SUCCESS in 50m 29s","accounts_in_message":[],"_revision_number":4},{"id":"3229ab198aa69aa2e142e3eee84bba1e894e7546","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 10:37:57.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/654a9306241c4d008852acd1c7448c53\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4c38fef6a5524a348d06ea85a9523c39 : SUCCESS in 17m 45s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3abb1c110fe24a59b40d23407df6b7cf : SUCCESS in 42m 05s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/5197194c222d4b428f8eae919fcda711 : SUCCESS in 1h 57m 19s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d535adc420b0400d9f001594b976e317 : SUCCESS in 24m 47s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/534a2ebafbe6499d8ad9be29c8b32705 : SUCCESS in 54m 28s","accounts_in_message":[],"_revision_number":4},{"id":"eb057e8c943894edc89f858efc442a4d0936d541","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-05-03 14:20:23.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2f175a7d6ecf7ee20c855ff619c60e7e8238f015","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-05-03 16:50:28.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"ccb7d3523e8109db92de7edb449b1e26aabbaf93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 18:04:35.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/235a2f93dc71490ba4dbc39564dc7b61\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/645c7a6774fd42bb8112fea4663e6f2e : SUCCESS in 8m 41s\n- swift-tox-func-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/8433caf1d223406f94498686d6596433 : SUCCESS in 20m 40s (non-voting)\n- swift-tox-func-encryption-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/eff5755f33754692a05182c760cb0373 : SUCCESS in 20m 26s (non-voting)\n- swift-tox-func-ec-py36-centos-8-stream-fips https://zuul.opendev.org/t/openstack/build/be9453c1c66d41d88382c00f879cf2ca : SUCCESS in 21m 38s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c613bc76c23b46e0a0fef3dcddc5db60 : FAILURE in 3m 36s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1322b4bf180e42a183a565d9b5cbcdd4 : FAILURE in 3m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2969446d49b341eba73b784ca65d8da8 : SUCCESS in 17m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b4bf7f4bbd0745b89e14ce582bd56d9e : SUCCESS in 12m 41s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/01122d22ecd7486d8b16a73a4eee49a4 : SUCCESS in 16m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9f4c2987be824ec3994a4df7f780457c : SUCCESS in 16m 10s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e7f11a3b871a41579c249c9bdc577635 : SUCCESS in 19m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ffebe75bcf184014bfae844937e9bd6c : SUCCESS in 19m 21s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/96ec93a390bc49b2935fcbc3d33e2cf1 : SUCCESS in 13m 47s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9b862d8e400f404fa5ea6e7f08396ff4 : SUCCESS in 14m 46s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/189570ed18bf44539861ca6cd7e014f0 : SUCCESS in 16m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9d46a4908ba14de2972a3c04f946c861 : SUCCESS in 9m 06s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/04aa3da4ce274297bfb209facb3d2691 : SUCCESS in 1h 03m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4ed22b8f214e4c2da57316cb8bf7f60a : SUCCESS in 42m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a1fcd7092b504792a0df047767c1b62c : SUCCESS in 40m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c0951bcd5d4f4a42a0ed8fa4feb2c42a : SUCCESS in 18m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ee3169852f04787b59f95a13187cf9c : SUCCESS in 6m 44s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/30d0f14aec5e47a3a01b4ae7c948eadb : RETRY_LIMIT in 4m 52s (non-voting)\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/50ac16ef04d14abca7c8a2c49781da7b : SUCCESS in 51m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/85742165d62c4b84927c5b9d4a67d7c2 : SUCCESS in 46m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/5eac9a94cf864b8e84d4b5e0cb54d088 : FAILURE in 44m 03s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ed7c8d1ce6334224b629c2a3648be1c5 : SUCCESS in 49m 27s","accounts_in_message":[],"_revision_number":5},{"id":"4c5a8d0bbec57f96bb4268f32e7ba9db952ae1c2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 18:49:08.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5bab0be7b5bc43948f39442925654b89\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/108e863edcec4dacba0e7c25c80471c8 : SUCCESS in 39m 36s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/38873b4aad064f1cbda113fce44630dd : SUCCESS in 1h 02m 40s\n- swift-probetests-centos-8-stream-arm64 https://zuul.opendev.org/t/openstack/build/34a736fd80a44f358adac56da4a5fd84 : SUCCESS in 1h 56m 16s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d328dda9766a403993a428fd3c29c194 : SUCCESS in 18m 24s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/93775310300d4df28f3f868d98dd559b : SUCCESS in 1h 00m 24s","accounts_in_message":[],"_revision_number":5},{"id":"0985ecbd9925f22518effccfba76e6709ce2fb92","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-06 15:49:23.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"ae4e7095ef75de24b5deced650de8210fef55082","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-05-07 11:31:09.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"6a30f23a8759a0e3c61d208ab9e0d0275b6915de","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-05-07 11:36:48.000000000","message":"Patch Set 5: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f7b4eb8bf6bc86126e8b4b41d72ab1b1021fc35d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-07 11:37:37.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"35c645ffd2c4fb538532abdcc4054b317ec42578","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-07 12:44:17.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c48af0f1b28e4275b94632c8831dc9d9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5db7902b81304dba87c23a273a028877 : SUCCESS in 8m 43s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/d28fa92829d94f8a9661a01d94e7d918 : POST_FAILURE in 4m 50s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/f3a0a9c149794771b84b4fbf606e4e10 : POST_FAILURE in 5m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bd695ca462a146b196d60ba2a274689b : SUCCESS in 16m 35s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/50d77387e15c4ff2a7cdba518cb43b43 : SUCCESS in 14m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/182d2f5ec60a45d2a4dca645ca32eb24 : SUCCESS in 17m 57s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4fc9ff83df8042e8923d91c234340342 : SUCCESS in 19m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/44b6e59c56fe49db8db6699ad25f8994 : SUCCESS in 17m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b150d82b06f64f4dacdfc6087e39e884 : SUCCESS in 14m 48s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d4007f43c9574bf98d74c31680fecb7b : SUCCESS in 14m 18s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3cd7ffbaf8864b8b826a7f5409028ac9 : SUCCESS in 15m 39s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/41a2dff6e3384ff39e17d7b0808b8367 : SUCCESS in 12m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1e8d5ab95ea04dfcbb0acd48518a103f : SUCCESS in 14m 52s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/d2ac8c39f4e34a27aacb762b14798cfc : SUCCESS in 50m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a2747c3d7981433b96597e81324c8199 : SUCCESS in 34m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1b8687164ef74307b0d6ee39149e78f9 : SUCCESS in 47m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3235563bd25f417f94b913706f6b5dfb : SUCCESS in 17m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed1d74fbe17c4c098d69565565e1cba0 : SUCCESS in 5m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/422c1f8825154dab9943ba157b919a71 : SUCCESS in 57m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/baf97e07e94b4036b7e61bce30092321 : SUCCESS in 52m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/60dd860f82a945c99205e4b1b07112eb : SUCCESS in 40m 08s","accounts_in_message":[],"_revision_number":5},{"id":"6478a9f90c4bcf729067b1d2185df257f02bb31c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-07 12:44:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"1b6b01be48e3a1610f32d27e9f92fc92a1dd068c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-07 12:47:53.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/730e2bdc95a34358af84436b38f43262\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/639df7a0962b4be08feb8525b8125af9 : SUCCESS in 42s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/779179e2bcf34a10bcddcef11e9a25d9 : FAILURE in 30s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/708c718c886c4c3bb4ed2cbcec9727fc : FAILURE in 30s (non-voting)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"11fd2ec0f6b6e4fb53b05d031cb706608370b3f3","revisions":{"b4a6e04df5e1941735e88192bca75626d40141be":{"kind":"REWORK","_number":1,"created":"2024-04-24 13:39:20.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/07/916907/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/07/916907/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/07/916907/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/07/916907/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/07/916907/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/07/916907/1"}}},"commit":{"parents":[{"commit":"8733a602276ed1b64a08eaa151619c1b35fb3a77","subject":"Update .gitreview for unmaintained/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8733a602276ed1b64a08eaa151619c1b35fb3a77"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-04-24 13:35:48.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-04-24 13:35:48.000000000","tz":120},"subject":"WIP: Clean up centos-7 based jobs","message":"WIP: Clean up centos-7 based jobs\n\nThere is no Centos-7 nodeset anymore, so let\u0027s get rid of those jobs,\nor use centos-8-stream if possible. Also remove old rolling upgrade\njobs that would run against branches that are already End of Life.\n\nChange-Id: Id56057262fc37add3a47c50946067f8c873c7515\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b4a6e04df5e1941735e88192bca75626d40141be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b4a6e04df5e1941735e88192bca75626d40141be"}]},"branch":"refs/heads/unmaintained/yoga"},"b8c038fe7f52269ed27a266654bd56dff3c3d73f":{"kind":"REWORK","_number":2,"created":"2024-04-24 13:53:20.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/07/916907/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/07/916907/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/07/916907/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/07/916907/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/07/916907/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/07/916907/2"}}},"commit":{"parents":[{"commit":"8733a602276ed1b64a08eaa151619c1b35fb3a77","subject":"Update .gitreview for unmaintained/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8733a602276ed1b64a08eaa151619c1b35fb3a77"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-04-24 13:35:48.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-04-24 13:53:09.000000000","tz":120},"subject":"WIP: Clean up centos-7 based jobs","message":"WIP: Clean up centos-7 based jobs\n\nThere is no Centos-7 nodeset anymore, so let\u0027s get rid of those jobs,\nor use centos-8-stream if possible. Also remove old rolling upgrade\njobs that would run against branches that are already End of Life.\n\nChange-Id: Id56057262fc37add3a47c50946067f8c873c7515\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b8c038fe7f52269ed27a266654bd56dff3c3d73f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b8c038fe7f52269ed27a266654bd56dff3c3d73f"}]},"branch":"refs/heads/unmaintained/yoga"},"9042d1a7ede3f730888aa6864f082f12bbea06ab":{"kind":"REWORK","_number":3,"created":"2024-04-24 16:17:28.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/07/916907/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/07/916907/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/07/916907/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/07/916907/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/07/916907/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/07/916907/3"}}},"commit":{"parents":[{"commit":"8733a602276ed1b64a08eaa151619c1b35fb3a77","subject":"Update .gitreview for unmaintained/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8733a602276ed1b64a08eaa151619c1b35fb3a77"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-04-24 13:35:48.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-04-24 15:10:16.000000000","tz":120},"subject":"Clean up centos-7 based jobs","message":"Clean up centos-7 based jobs\n\nThere is no Centos-7 nodeset anymore, so let\u0027s get rid of those jobs,\nor use centos-8-stream if possible. Also remove old rolling upgrade\njobs that would run against branches that are already End of Life.\nGrenade job is currently broken, let\u0027s remove it from gate and set it\nnon-voting on check queue to unblock gate.\n\nChange-Id: Id56057262fc37add3a47c50946067f8c873c7515\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9042d1a7ede3f730888aa6864f082f12bbea06ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9042d1a7ede3f730888aa6864f082f12bbea06ab"}]},"branch":"refs/heads/unmaintained/yoga"},"6728927a38ed77818fb17b12ec0e7bc5a6dbece4":{"kind":"REWORK","_number":4,"created":"2024-04-26 08:38:01.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/07/916907/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/07/916907/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/07/916907/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/07/916907/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/07/916907/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/07/916907/4"}}},"commit":{"parents":[{"commit":"8733a602276ed1b64a08eaa151619c1b35fb3a77","subject":"Update .gitreview for unmaintained/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8733a602276ed1b64a08eaa151619c1b35fb3a77"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-04-24 13:35:48.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-04-26 08:31:30.000000000","tz":120},"subject":"Clean up centos-7 based jobs","message":"Clean up centos-7 based jobs\n\nThere is no Centos-7 nodeset anymore, so let\u0027s get rid of those jobs,\nor use centos-8-stream if possible. Also remove old rolling upgrade\njobs that would run against branches that are already End of Life.\nGrenade job is currently broken, let\u0027s remove it from gate and set it\nnon-voting on check queue to unblock gate. Grenade is part of the\nintegrated-gate-object-storage template [1], so that has to be removed\nand its jobs have to be added explicitly (definition of job\ntempest-integrated-object-storage is already re-defined (overriden),\nbut openstacksdk-functional-devstack has to be added).\n\n[1] https://opendev.org/openstack/tempest/src/commit/1c3fc43c9dae274a19683b003f78b5f50edbc062/zuul.d/integrated-gate.yaml\n\nChange-Id: Id56057262fc37add3a47c50946067f8c873c7515\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6728927a38ed77818fb17b12ec0e7bc5a6dbece4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6728927a38ed77818fb17b12ec0e7bc5a6dbece4"}]},"branch":"refs/heads/unmaintained/yoga"},"11fd2ec0f6b6e4fb53b05d031cb706608370b3f3":{"kind":"REWORK","_number":5,"created":"2024-05-03 16:50:28.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/07/916907/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/07/916907/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/07/916907/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/07/916907/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/07/916907/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/07/916907/5"}}},"commit":{"parents":[{"commit":"8733a602276ed1b64a08eaa151619c1b35fb3a77","subject":"Update .gitreview for unmaintained/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8733a602276ed1b64a08eaa151619c1b35fb3a77"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-04-24 13:35:48.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-05-03 16:50:10.000000000","tz":120},"subject":"Clean up centos-7 based jobs","message":"Clean up centos-7 based jobs\n\nThere is no Centos-7 nodeset anymore, so let\u0027s get rid of those jobs,\nor use centos-8-stream if possible. Also remove old rolling upgrade\njobs that would run against branches that are already End of Life.\nGrenade job is currently broken, let\u0027s remove it from gate and set it\nnon-voting on check queue to unblock gate. Grenade is part of the\nintegrated-gate-object-storage template [1], so that has to be removed\nand its jobs have to be added explicitly (definition of job\ntempest-integrated-object-storage is already re-defined (overriden),\nbut openstacksdk-functional-devstack has to be added).\n\n[1] https://opendev.org/openstack/tempest/src/commit/1c3fc43c9dae274a19683b003f78b5f50edbc062/zuul.d/integrated-gate.yaml\n\nChange-Id: Id56057262fc37add3a47c50946067f8c873c7515\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/11fd2ec0f6b6e4fb53b05d031cb706608370b3f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/11fd2ec0f6b6e4fb53b05d031cb706608370b3f3"}]},"branch":"refs/heads/unmaintained/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
