)]}'
{"id":"openstack%2Fswift~916941","triplet_id":"openstack%2Fswift~master~Ic67048943dc2c0112b613c390c8262890b923807","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"Ic67048943dc2c0112b613c390c8262890b923807","subject":"sq: unquote account/container names from configs","status":"ABANDONED","created":"2024-04-24 18:07:11.000000000","updated":"2024-05-01 20:15:16.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5f0340fffa3366ce1237b4046350afe8cf92deb4","_number":916941,"virtual_id_number":916941,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-04-24 19:18:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-24 19:18:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"87397f0981348b6394282e70dab8add2d36a2bbf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-04-24 18:07:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0ccf680d3a58e6addeae79781f58594a770e1afd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 19:18:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17ea15d266d94089a4f5f74616b515ca\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c44a57d2a8aa4c9da08e031fee29d8a0 : SUCCESS in 7m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/302419def43c406c8babdf1aa67243f7 : SUCCESS in 59m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/02a3674be8c24d5187bed0a471c4ad20 : SUCCESS in 47m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3225fb12569949b9981457ed16c62205 : SUCCESS in 1h 04m 01s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/51863743c14b44af8c47bc2e4e24ceff : SUCCESS in 21m 28s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5a0ad5556e09461994687ffc32f3b000 : SUCCESS in 19m 16s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f3d4d105ecfb4b8494f1b6d6ae8eedb2 : SUCCESS in 22m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/22f1ff58747e4b2eb49a44e3680c9083 : SUCCESS in 10m 49s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ba829d4809294654a2f78587082b0afd : SUCCESS in 6m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/feea4b99890f43b0902c748e68a4072c : SUCCESS in 19m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c0b0e403f231431ba17f28d9a87056c6 : SUCCESS in 18m 16s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5776c6e4bf214028902d80f3d11a7923 : SUCCESS in 18m 26s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/5e26aa93cedc4b7c918ee3ae228f000b : SUCCESS in 17m 22s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8fdd4cb337ab4900b78e36252c44cf5d : SUCCESS in 19m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e510847b01844b4b882c9b22bb0fbc2e : SUCCESS in 18m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b4e855c9cc5741b7bc1e5c4c5963c4f8 : SUCCESS in 19m 00s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b1205e7ab7c54e048779c8364a4f1f6c : SUCCESS in 15m 43s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ec98b3ced08a4bb5b0b95518576e834c : SUCCESS in 15m 37s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/70c5899a0eba4c88be8e6100bf9210dd : SUCCESS in 18m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a8e1115e698c4d148c012c92cff78fad : SUCCESS in 6m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/560118a4541f44eea080822c2de6d3f2 : SUCCESS in 15m 45s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/fcd31ff1903f48d1b8d72cd86c2285cb : SUCCESS in 9m 44s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d5841110686149b4a52fc8c51f95f83d : SUCCESS in 55m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/74779d488ed143b9b556ec6ba97a89b6 : SUCCESS in 59m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6a55126ceaec4fe197d2cbcac0e228f5 : SUCCESS in 50m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3051463eefb949adadb0255984d73f0a : SUCCESS in 20m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8498614e7bf54e699f7e5f9726580f33 : SUCCESS in 4m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3d7b06f6d5724205a210df6fe7220e04 : SUCCESS in 25m 10s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0d4e21c79c4f48cd8b169d69cdd9b460 : SUCCESS in 1h 04m 23s","accounts_in_message":[],"_revision_number":1},{"id":"4a8a5ea656f6364dca20308bb1816e351e1c0a0d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 19:18:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69cc6e3db4cb4efc901e7f86770033c5\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a0ec4c9e7d4349caa733fc97a74b5742 : SUCCESS in 17m 27s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8f4e6395a81541e8b9f551a160b3ec6a : SUCCESS in 18m 08s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/96b65edd9d24474c9135f91862a92993 : SUCCESS in 1h 09m 12s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/727cab4c6d6f42d489f00c8ee02ad125 : SUCCESS in 1h 00m 38s","accounts_in_message":[],"_revision_number":1},{"id":"5f0340fffa3366ce1237b4046350afe8cf92deb4","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-01 20:15:16.000000000","message":"Abandoned\n\nSquashed into parent","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"53c21237000484eeb80f17d68ff8176222349f56","revisions":{"53c21237000484eeb80f17d68ff8176222349f56":{"kind":"REWORK","_number":1,"created":"2024-04-24 18:07:11.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/41/916941/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/41/916941/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/916941/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/916941/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/916941/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/41/916941/1"}}},"commit":{"parents":[{"commit":"6f7afa83927ea9fb4d06a255112616c5bf9e51df","subject":"expirer: account and container level delay_reaping","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6f7afa83927ea9fb4d06a255112616c5bf9e51df"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-04-24 17:58:17.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-04-24 18:05:05.000000000","tz":-420},"subject":"sq: unquote account/container names from configs","message":"sq: unquote account/container names from configs\n\nChange-Id: Ic67048943dc2c0112b613c390c8262890b923807\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/53c21237000484eeb80f17d68ff8176222349f56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/53c21237000484eeb80f17d68ff8176222349f56"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
