)]}'
{"id":"openstack%2Fswift~917010","triplet_id":"openstack%2Fswift~feature%2Fmpu~I97a958700b38a8e4ff7871b0ba3ec56e168f6626","project":"openstack/swift","branch":"feature/mpu","topic":"p-mpu-mw","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-09-16 10:56:10.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-09-26 13:07:33.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I97a958700b38a8e4ff7871b0ba3ec56e168f6626","subject":"slo: return manifest etag with manifest PUT response","status":"ABANDONED","created":"2024-04-25 09:38:28.000000000","updated":"2024-09-26 13:07:33.000000000","total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"f8cd5bfbce7eb58bb82170dd64070e7b35a13f63","_number":917010,"virtual_id_number":917010,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-09-10 11:28:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-09-11 19:52:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-25 10:49:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-11 19:52:32.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"976c895ea8d867bac4827a6c75db93aeb1364241","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-25 09:38:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6051051572974c0395ec4e8bc684a80cab34e4e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 10:49:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e05d96db40c64e208c0fd77a22d1578a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84f4aee3cace45c584cf6ae3335acf0c : SUCCESS in 8m 15s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/62a24b8ac9e244c8995dcb7b78e53be8 : FAILURE in 1h 00m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/77dffd1ac67140f38ba9569d8c2a02b3 : FAILURE in 56m 03s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8725eb0636a146bbab45b7266e1193c8 : FAILURE in 58m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8bbc2da4a74c443f9e8dde338c2e3187 : SUCCESS in 11m 16s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6fc3cb6d50fc48c384fb09f2adcab94e : SUCCESS in 6m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c2eb5f68fc5d4466bc4af22c21370a57 : FAILURE in 25m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7ffbf96fa0a64a87ad6af966451b0c1f : SUCCESS in 17m 48s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3fa9036746df4602bef54197f82a169c : SUCCESS in 19m 46s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/20114ea3741647dc875912f6e81b9e70 : SUCCESS in 17m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bea44494584b4156a889d59244138b6d : FAILURE in 57m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8dc03eefae3e43d68c9162796bd604cf : FAILURE in 59m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e784aca970d1499fa5e6b11e4cf0db44 : FAILURE in 59m 46s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/43f3318f4f6e4fe98af398a923ee3971 : FAILURE in 51m 51s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/04b65b0eab7b4b3fb4aa6a35fd870a25 : FAILURE in 53m 15s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d74ee12452e74c06bfd4561d5991936d : FAILURE in 53m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/90bcf6fee18c43719ebc45a080543098 : FAILURE in 6m 16s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ed017520634c49c79c224936f4afb2e4 : FAILURE in 57m 25s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/fce13708a31a4ae4901ce29070afe588 : FAILURE in 7m 38s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0f3eab76a35b40ddb133e340b67d17ae : SUCCESS in 58m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b3f9cecd561646c99528a9e33a918df2 : FAILURE in 21m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcf1e6a5f8044402a3835f8b8543082b : SUCCESS in 4m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e4c25db8ebac4b9e8a4bedd9d4b1727b : SUCCESS in 18m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bdf14b58a3d94b603deb0e5b36840438294db224","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 11:24:48.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/230ea5cea0ff4f3eaac4258c14434773\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/29e89e376ffa458d91b95d8402e2f7f3 : SUCCESS in 26m 01s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/31c3aab3a282462cbe836b302618d945 : SUCCESS in 26m 44s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b13353fd263f43f4a5f34b2589ed2659 : FAILURE in 1h 00m 22s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f8a71f431279462aa8aa157844bcdefa : FAILURE in 1h 00m 13s","accounts_in_message":[],"_revision_number":1},{"id":"d7a9f3818c8533430e30f35a3bca4bb3a60e35ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-25 15:48:06.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"49a9d36a802805a794a6b1fb683d9e19760020ca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 16:46:39.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/732c6eb7270b4aabbe089557f235bc8e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f93a601f9c274b1ea7169f723dbe25d5 : SUCCESS in 20m 49s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/15fd6e9623b44ce2a12f7eb52bb898c7 : SUCCESS in 20m 20s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4a1f8d50811c47fb924ab51b2b3c2128 : FAILURE in 56m 28s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6539a2be6c0f44c68d6c447e060ec0ff : FAILURE in 47m 54s","accounts_in_message":[],"_revision_number":2},{"id":"ee54ce163f215cc4f573fc2d4de77ede740dfeb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 16:47:12.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f81d8a636b4e4f299f898e4eebfcf34d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab9440c708ef4607a4bee769ee6d0d06 : SUCCESS in 8m 31s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/12be1d7604e64d9bbd506d1082849246 : FAILURE in 17m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a99f92bf518147f188c95ce06080df1f : FAILURE in 18m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/937a853e6e504c0abc83a304dc2d40c4 : FAILURE in 18m 25s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b90ef18012aa49c3b576f111e2cbac76 : SUCCESS in 11m 24s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d114067428ac463a9f0f2bc58a0d59c6 : SUCCESS in 5m 07s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/961acebb1276427485dce83c91834af4 : FAILURE in 17m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fbc57a6470734bf48a184b412f193c63 : SUCCESS in 17m 01s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1b3799f7cd0240c8a4e248dd80809e68 : SUCCESS in 21m 18s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/6072aeb156824f099dd3936d5f5daa2b : SUCCESS in 16m 58s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b01022583cf5489c834eaca7e5512eca : FAILURE in 19m 08s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a3b726c1aa314e3d897b119ae9f7c18d : FAILURE in 18m 25s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7470ce64137b4aefbdcd5460656f97dd : FAILURE in 19m 31s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f446555c35b14514b36e03eade98719c : FAILURE in 16m 35s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a5bf6f8731564b77b3420fd9cad2de12 : FAILURE in 17m 55s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5f70884e63c54a9db619fc409edae8e5 : FAILURE in 19m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2bbf4682d2bd436783ab44c66f25f4b8 : SUCCESS in 6m 42s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/82c6b27b058447698d66bc3ddb6b22bf : SUCCESS in 14m 51s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e1af6cc955d94e5ea1cf44d2f0a9c25b : SUCCESS in 8m 31s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/fb3e7460a3874ab1869fd43129b1f559 : SUCCESS in 57m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e8dc3d225869463c8cbbdd0931852bdd : FAILURE in 23m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7027b3416b834017a1e74ddfc1fc96c9 : SUCCESS in 4m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f7d0abe221e94c3a918d9c61aa52d78d : SUCCESS in 19m 46s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"33696808874caab794f2714b9aad3754c29aaac5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-29 09:16:41.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"6af51da2752432026bbd032b64f17bd1d25f1e81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 10:22:17.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7b9b1da08d0742a9b9359f6d153016d2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc2497832a814ff3a231e66d5f10645e : SUCCESS in 8m 17s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d94cf6ed851a4fdd8fc27c0747e75fef : FAILURE in 19m 44s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8d98b46c99dc4dbba5e912433081f410 : FAILURE in 27m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2a076194d2944212b3757ca09ea5f1f5 : FAILURE in 21m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/642541ba1fff43f890a841a4044048c5 : SUCCESS in 11m 35s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d7faaf1d9c7b4df1ace2c7c2d363721e : SUCCESS in 5m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/174020bacc724535ac36ec740932fdce : FAILURE in 19m 32s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4957836cae664dcea4046c3e753dea39 : SUCCESS in 19m 01s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a6d0fb27c6774fa892850ae1e2f5f0ba : SUCCESS in 17m 53s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/bc38897ef5604286b6b6da5de76f1e51 : SUCCESS in 17m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1d5caa309a664469a2ba4480ac06f600 : FAILURE in 17m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/22cf7ccdb776477eadba6bf3102e67df : FAILURE in 22m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3e68bb98fcbc43dfb9a21241bc72f4cb : FAILURE in 20m 32s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5e7349300fe649a287630b3850fad532 : FAILURE in 16m 42s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b22a1cb748334a2592aa39f7e320572b : FAILURE in 16m 00s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0711ac7d5c7b4079b5ca4dd2caad1d1c : FAILURE in 15m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0ff400362def4fc8bf7ca571b3ef48c6 : SUCCESS in 6m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/56da5318da534206960fbfd580f4f0ef : SUCCESS in 19m 26s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2b4b45a5da684b85946bbe9e7bb897f4 : SUCCESS in 7m 47s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d16d2c998bc64215b3ce31900c518030 : SUCCESS in 1h 00m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8586fb1bddea4965b61eb79b500023e0 : FAILURE in 22m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce6476a82b0444749a294037e9266c0b : SUCCESS in 4m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7a67e9f5e2bd4f4d9995bab7007987cc : SUCCESS in 19m 50s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"09c63284ace0014245dead660b10b88134fd4793","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 10:50:13.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1b4e1af64584f1ca8e134c3d61f7fed\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7d5a8c4ef6da4467946786f5f4eb9656 : SUCCESS in 1h 04m 51s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fc152a7ad1c14edaa8572d233f643ce9 : TIMED_OUT in 1h 30m 59s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8d60aa905a48475d93971a59d5bcd773 : FAILURE in 21m 57s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/2ef14cea1a9647eab3a49e72cb5f0716 : FAILURE in 23m 07s","accounts_in_message":[],"_revision_number":3},{"id":"455d7b7fe2a7cbf16041a3689c50d298a0a515ec","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-29 17:01:13.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c7eb5c51758bfaa200df257971c2b0081fb67405","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-30 10:30:05.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"2687ea23605b90e3decdec576762e0fd6b2cd669","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 11:23:26.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a2b5ba452c64381b8cf58eeae3205ce\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2fd80fde92af430c8555976df1fd1fb5 : SUCCESS in 22m 11s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a757568143384a2c848325d1a3da13e9 : SUCCESS in 46m 04s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/946f5e2cca3242f3a72a114b2e062b1b : SUCCESS in 21m 32s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8f4e8548b4f242139a5da282832bf2b6 : SUCCESS in 51m 15s","accounts_in_message":[],"_revision_number":4},{"id":"326e141117d2c3d5e2edeaba5b1fd983f17f9d5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 11:34:23.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b6556cbed6841289a2650eefd9e3c78\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4a056d04ff544fe9f586fd3deff7b80 : SUCCESS in 8m 08s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a6403e9d281147b09a3d2dd76cafc856 : SUCCESS in 19m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3cf39fd290d749918401d2eb2ff3af97 : SUCCESS in 20m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dfc0983c0e7a4e0896715fb7463d0f37 : SUCCESS in 21m 32s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ec961cc3addd485ea2ef65fb7ab51842 : SUCCESS in 13m 31s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7e41cfaa7cf34a1789b8348e15056754 : SUCCESS in 5m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/488abe4e36bf4e319bd1cc7a5173f415 : SUCCESS in 16m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/80c7c44ecf3c4021b87adf7793dc2c56 : SUCCESS in 18m 38s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6ad510a60ea646d78344c0e183eee119 : SUCCESS in 15m 29s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/711caa4b81294f4384e436a4d8c33960 : SUCCESS in 19m 40s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/dc709bde768e41ff917dfb14e177bd48 : SUCCESS in 19m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ed897d414f9a4c64a460683639e94e13 : SUCCESS in 20m 10s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5064ec3bec0c49f9895821e9a969b3c9 : SUCCESS in 21m 34s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/97b4e46564d049ae83c8da32435c40fd : SUCCESS in 15m 00s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d28413afd7754dab8920bf4105902f7a : SUCCESS in 19m 31s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1f87db0863154a1c9207af6e12ea9eef : SUCCESS in 17m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4beada68227a4a41809fbef19ff448ab : SUCCESS in 7m 04s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cece38865a8548168e6d9a6824dec025 : SUCCESS in 15m 54s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1c7423419b0746d7b42c090ebe92741c : SUCCESS in 7m 55s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e110910bef5341eab8c02838278f7fe6 : SUCCESS in 59m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a2257e7119a34aa59783f49f87c1cf2a : SUCCESS in 18m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62bded43c673405bb6b283033d9fb7b0 : SUCCESS in 4m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6c79431e77cc41d3acfcdab59363b815 : SUCCESS in 18m 23s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"277b8a86d5a6256db717f7b19e1a5b9baf6cffd8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-05-02 09:19:01.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"01bea4c8be89ea55da160e7ccaee11267fdae223","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 10:26:18.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e53462923dba4fb7ab4c02c6fc5f945b\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/edd835e2ff6b4d58b023e7daadd20da6 : SUCCESS in 21m 10s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7cc01f7e235a472b8e80ef5a4b133be4 : SUCCESS in 1h 00m 29s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/cf9dd428bc4f4bc8812d07975670be36 : SUCCESS in 25m 13s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f826e6bf22c346f0941b6b7960d69ac8 : SUCCESS in 59m 17s","accounts_in_message":[],"_revision_number":5},{"id":"c26505b8024a0fc384328af685fa69bb47ee4ebc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 10:56:27.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/222e578cdee04148bea05f60efa3f8a5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7af784732b6340df919474e4ca44edeb : SUCCESS in 8m 27s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7f25b36eb0f94747a5748709c8f0107d : SUCCESS in 17m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9acaf04f693e46f6b0502e35b0edceb0 : SUCCESS in 22m 40s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/df11ff7a8d68440d8d34f7594796d654 : SUCCESS in 19m 22s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/919bdbd55aae4180863a6d463d8f7ec2 : SUCCESS in 11m 03s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/86e05fbfaf8248288065f106a112f8a5 : SUCCESS in 5m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9c6a6eea052d49d8b70fec97f1c4240d : SUCCESS in 21m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a1b6553544fa49b89720e049914c26be : SUCCESS in 18m 25s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/aedd911a5ecb490c85bbedfa17afc5b6 : SUCCESS in 17m 54s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/3287136e60e848b390bd312c7bb2983a : SUCCESS in 16m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/31b85cbb297d4b2ea5af0129535ee08c : SUCCESS in 18m 30s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/66a4ad6b3f3044f3a111a7df7fe67091 : SUCCESS in 17m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bc2ee56556f54e8086c1e556b5fb5431 : SUCCESS in 19m 31s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1812548759c7487ab813a1cdb224c0d3 : SUCCESS in 16m 17s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/06600c089b1a47eb8e58b887ec174838 : SUCCESS in 20m 45s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/963b57c7ebce4f679e41bf0de80dae6f : SUCCESS in 19m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a0cbf728cce042e2a2e3cbae3bb89afb : SUCCESS in 6m 50s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c98bb4bd57834ce1b47783b977d48185 : SUCCESS in 18m 38s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c4dfab9ad15a46e59eb1fbe21842607d : SUCCESS in 7m 01s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1c4562607a32464db227175dac2c91fa : SUCCESS in 1h 05m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a402c9323aef44758f568fac4b088781 : SUCCESS in 22m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4f70e18c56743d1a939b352af18889a : SUCCESS in 4m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3b31737c7cfc467593c1f635ba06dab6 : SUCCESS in 21m 13s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"edb253f1403497f3f29cf45869b426218fa18654","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-09-10 10:25:48.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"4d94b6a5606dd6011218e54150f7af4ffabe6f83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 11:28:31.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe1f020329ce43ec9e38094050eaaf8e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/708db340dbf044bf98805b13cfc8c156 : SUCCESS in 9m 01s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b23d52f3f8af4fb5b67efad7771aa5c5 : SUCCESS in 21m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5218bc09e3914f1897ebe7bea9591ca3 : SUCCESS in 13m 24s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f974f9e9ad2d4814bb2af0e9330bf74f : SUCCESS in 20m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f9b93a7dbc0f4b5f9a41375c667c897f : SUCCESS in 5m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2e7d53215a3c40dfa12fda8d2421cf14 : SUCCESS in 17m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/28555ea57db949178cc6e384d31fe50a : SUCCESS in 10m 32s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ba346e94e70641d2968db077312a6a68 : SUCCESS in 18m 44s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/1269abb203974bafa1acca48c8556d6c : SUCCESS in 22m 26s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f602f2b91e1e4d5f95790bf8e8f1c995 : SUCCESS in 29m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/372fff628bd5408d8d222cf22a44da74 : SUCCESS in 16m 19s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1f0df6974fd54c3d8790b1d63c8523d7 : SUCCESS in 17m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/dc00c767d2514348b53cdeaa90dfd38d : SUCCESS in 17m 31s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d5473ae476b2494d968f06845c07bf35 : SUCCESS in 17m 30s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/de7d6d84399d4a5fa9e770aff52d6252 : SUCCESS in 13m 45s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/32b21bc419ea498ba7782eef40ecd862 : SUCCESS in 17m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cde4456f2a114506a45a3344eec1ded6 : SUCCESS in 5m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f8e7ec5c8c7d4913955ae703a4eb5505 : SUCCESS in 15m 17s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/467c8272e40d437fb8358d9d86f4f32a : SUCCESS in 5m 39s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d5b449a0fa93452390a942f3f729346b : SUCCESS in 57m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/86960caaaf3342d5aaeb465bd428d08a : SUCCESS in 17m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3da8aff480124a2f839f9c3d700004b6 : SUCCESS in 3m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5803d049d8044fec86ba9d1f1665f51e : SUCCESS in 18m 35s","accounts_in_message":[],"_revision_number":6},{"id":"6887a555ba99eb142f60561bedd8823f621b2f9c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 12:16:41.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03e4c9e84a2b4f52bd337e93131362f3\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0c8eeec03cb842599e711efb95b6e3e2 : SUCCESS in 56m 19s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d69e0904b10f4dd7b6165fcdafd27c40 : SUCCESS in 1h 03m 04s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/9b964c75fe184089acda47c127ccad49 : FAILURE in 1h 01m 06s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7c02f679a7b14de591c5af84d193c3cc : SUCCESS in 48m 35s","accounts_in_message":[],"_revision_number":6},{"id":"2243f1cf2a4e6a6113751f382358b4c50b6355ea","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-11 19:52:32.000000000","message":"Patch Set 6: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"b16924db6292bb90e26881264de89cc349311885","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-09-16 10:56:10.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"f8cd5bfbce7eb58bb82170dd64070e7b35a13f63","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-09-26 13:07:33.000000000","message":"Abandoned\n\nno longer required on feature/mpu","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"08d1298af4bb5ab0a24518fa5ce0de20ada2987e","revisions":{"d72f9f29094b6d93644db0a2a6a08aac1106cf5d":{"kind":"REWORK","_number":1,"created":"2024-04-25 09:38:28.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/10/917010/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/10/917010/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/917010/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/917010/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/917010/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/10/917010/1"}}},"commit":{"parents":[{"commit":"6c712cdc30cc22b8923af696bc2e075e514bcb80","subject":"mpu-auditor: add some basic exception handling","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6c712cdc30cc22b8923af696bc2e075e514bcb80"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-24 10:31:26.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-25 09:37:52.000000000","tz":60},"subject":"slo: return manifest etag with manifest PUT response","message":"slo: return manifest etag with manifest PUT response\n\nWhen possible, include the manifest etag as an X-Manifest-Etag header\nwith the response to a multipart manifest PUT.  When the request has\nheartbeat\u003don, include the manifest etag in the body of the response.\n\nIf you want to create a \"static\" symlink to an SLO manifest by sending\nan \"X-Symlink-Target-Etag: \u003cetag\u003e\" header then the value of \u003cetag\u003e\nmust be the manifest etag rather than the large object etag. However,\nthe manifest etag was not previously included in the response to an\nSLO manifest PUT request.\n\nIn non-heartbeat mode the etags are known when the manifest PUT\nresponse starts, so the manifest etag can be included in response\nheaders. In heartbeat mode neither etags are not known when the\nresponse starts, so they cannot be included in response\nheaders. However, the large object etag is already included in a json\ndictionary in the response body. This patch adds the manifest etag to\nthe response body.\n\nChange-Id: I97a958700b38a8e4ff7871b0ba3ec56e168f6626\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d72f9f29094b6d93644db0a2a6a08aac1106cf5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d72f9f29094b6d93644db0a2a6a08aac1106cf5d"}]},"branch":"refs/heads/feature/mpu"},"48d2c899c4bdde4d74bde54c3662b691944fdfad":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-04-25 15:48:06.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/10/917010/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/10/917010/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/917010/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/917010/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/917010/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/10/917010/2"}}},"commit":{"parents":[{"commit":"c0204a19a8b65f78cb4c954f3cd9aafb38e42836","subject":"mpu-auditor: iterate multiple batches per cycle","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c0204a19a8b65f78cb4c954f3cd9aafb38e42836"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-24 10:31:26.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-25 12:39:56.000000000","tz":60},"subject":"slo: return manifest etag with manifest PUT response","message":"slo: return manifest etag with manifest PUT response\n\nWhen possible, include the manifest etag as an X-Manifest-Etag header\nwith the response to a multipart manifest PUT.  When the request has\nheartbeat\u003don, include the manifest etag in the body of the response.\n\nIf you want to create a \"static\" symlink to an SLO manifest by sending\nan \"X-Symlink-Target-Etag: \u003cetag\u003e\" header then the value of \u003cetag\u003e\nmust be the manifest etag rather than the large object etag. However,\nthe manifest etag was not previously included in the response to an\nSLO manifest PUT request.\n\nIn non-heartbeat mode the etags are known when the manifest PUT\nresponse starts, so the manifest etag can be included in response\nheaders. In heartbeat mode neither etags are not known when the\nresponse starts, so they cannot be included in response\nheaders. However, the large object etag is already included in a json\ndictionary in the response body. This patch adds the manifest etag to\nthe response body.\n\nChange-Id: I97a958700b38a8e4ff7871b0ba3ec56e168f6626\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/48d2c899c4bdde4d74bde54c3662b691944fdfad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/48d2c899c4bdde4d74bde54c3662b691944fdfad"}]},"branch":"refs/heads/feature/mpu"},"9b5fdce262e56759a2ca1cd5fa10702b1ca34eb4":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-04-29 09:16:41.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/10/917010/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/10/917010/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/917010/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/917010/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/917010/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/10/917010/3"}}},"commit":{"parents":[{"commit":"dea863b7b31f4d87ee21a3e51057f12a21690868","subject":"mpu-auditor: iterate multiple batches per cycle","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dea863b7b31f4d87ee21a3e51057f12a21690868"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-24 10:31:26.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-29 08:09:46.000000000","tz":60},"subject":"slo: return manifest etag with manifest PUT response","message":"slo: return manifest etag with manifest PUT response\n\nWhen possible, include the manifest etag as an X-Manifest-Etag header\nwith the response to a multipart manifest PUT.  When the request has\nheartbeat\u003don, include the manifest etag in the body of the response.\n\nIf you want to create a \"static\" symlink to an SLO manifest by sending\nan \"X-Symlink-Target-Etag: \u003cetag\u003e\" header then the value of \u003cetag\u003e\nmust be the manifest etag rather than the large object etag. However,\nthe manifest etag was not previously included in the response to an\nSLO manifest PUT request.\n\nIn non-heartbeat mode the etags are known when the manifest PUT\nresponse starts, so the manifest etag can be included in response\nheaders. In heartbeat mode neither etags are not known when the\nresponse starts, so they cannot be included in response\nheaders. However, the large object etag is already included in a json\ndictionary in the response body. This patch adds the manifest etag to\nthe response body.\n\nChange-Id: I97a958700b38a8e4ff7871b0ba3ec56e168f6626\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9b5fdce262e56759a2ca1cd5fa10702b1ca34eb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9b5fdce262e56759a2ca1cd5fa10702b1ca34eb4"}]},"branch":"refs/heads/feature/mpu"},"b23c34875a0b24c26e9027821b9e8ab380beb397":{"kind":"REWORK","_number":4,"created":"2024-04-30 10:30:05.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/10/917010/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/10/917010/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/917010/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/917010/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/917010/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/10/917010/4"}}},"commit":{"parents":[{"commit":"de0d2c37a075317eea687769b82f7ba6dfc0a57f","subject":"mpu mw: pre-authorize MPU-handled requests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/de0d2c37a075317eea687769b82f7ba6dfc0a57f"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-24 10:31:26.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-30 10:09:07.000000000","tz":60},"subject":"slo: return manifest etag with manifest PUT response","message":"slo: return manifest etag with manifest PUT response\n\nWhen possible, include the manifest etag as an X-Manifest-Etag header\nwith the response to a multipart manifest PUT.  When the request has\nheartbeat\u003don, include the manifest etag in the body of the response.\n\nIf you want to create a \"static\" symlink to an SLO manifest by sending\nan \"X-Symlink-Target-Etag: \u003cetag\u003e\" header then the value of \u003cetag\u003e\nmust be the manifest etag rather than the large object etag. However,\nthe manifest etag was not previously included in the response to an\nSLO manifest PUT request.\n\nIn non-heartbeat mode the etags are known when the manifest PUT\nresponse starts, so the manifest etag can be included in response\nheaders. In heartbeat mode neither etags are not known when the\nresponse starts, so they cannot be included in response\nheaders. However, the large object etag is already included in a json\ndictionary in the response body. This patch adds the manifest etag to\nthe response body.\n\nChange-Id: I97a958700b38a8e4ff7871b0ba3ec56e168f6626\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b23c34875a0b24c26e9027821b9e8ab380beb397"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b23c34875a0b24c26e9027821b9e8ab380beb397"}]},"branch":"refs/heads/feature/mpu"},"0b3e47ebb892f502be5466ff2f83e15e203b46f5":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-05-02 09:19:01.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/10/917010/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/10/917010/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/917010/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/917010/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/917010/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/10/917010/5"}}},"commit":{"parents":[{"commit":"ba398e15ede580f48464a213361daed749d57f5f","subject":"Merge remote-tracking branch \u0027gerrit/master\u0027 into feature/mpu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ba398e15ede580f48464a213361daed749d57f5f"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-24 10:31:26.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-05-02 09:18:04.000000000","tz":60},"subject":"slo: return manifest etag with manifest PUT response","message":"slo: return manifest etag with manifest PUT response\n\nWhen possible, include the manifest etag as an X-Manifest-Etag header\nwith the response to a multipart manifest PUT.  When the request has\nheartbeat\u003don, include the manifest etag in the body of the response.\n\nIf you want to create a \"static\" symlink to an SLO manifest by sending\nan \"X-Symlink-Target-Etag: \u003cetag\u003e\" header then the value of \u003cetag\u003e\nmust be the manifest etag rather than the large object etag. However,\nthe manifest etag was not previously included in the response to an\nSLO manifest PUT request.\n\nIn non-heartbeat mode the etags are known when the manifest PUT\nresponse starts, so the manifest etag can be included in response\nheaders. In heartbeat mode neither etags are not known when the\nresponse starts, so they cannot be included in response\nheaders. However, the large object etag is already included in a json\ndictionary in the response body. This patch adds the manifest etag to\nthe response body.\n\nChange-Id: I97a958700b38a8e4ff7871b0ba3ec56e168f6626\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0b3e47ebb892f502be5466ff2f83e15e203b46f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0b3e47ebb892f502be5466ff2f83e15e203b46f5"}]},"branch":"refs/heads/feature/mpu"},"08d1298af4bb5ab0a24518fa5ce0de20ada2987e":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-09-10 10:25:48.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/10/917010/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/10/917010/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/917010/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/917010/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/917010/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/10/917010/6"}}},"commit":{"parents":[{"commit":"31b3823148bbda17d97a34b1fd7dd5cb389a91bb","subject":"Merge branch \u0027master\u0027 into feature/mpu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/31b3823148bbda17d97a34b1fd7dd5cb389a91bb"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-24 10:31:26.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-09-10 10:16:11.000000000","tz":60},"subject":"slo: return manifest etag with manifest PUT response","message":"slo: return manifest etag with manifest PUT response\n\nWhen possible, include the manifest etag as an X-Manifest-Etag header\nwith the response to a multipart manifest PUT.  When the request has\nheartbeat\u003don, include the manifest etag in the body of the response.\n\nIf you want to create a \"static\" symlink to an SLO manifest by sending\nan \"X-Symlink-Target-Etag: \u003cetag\u003e\" header then the value of \u003cetag\u003e\nmust be the manifest etag rather than the large object etag. However,\nthe manifest etag was not previously included in the response to an\nSLO manifest PUT request.\n\nIn non-heartbeat mode the etags are known when the manifest PUT\nresponse starts, so the manifest etag can be included in response\nheaders. In heartbeat mode neither etags are not known when the\nresponse starts, so they cannot be included in response\nheaders. However, the large object etag is already included in a json\ndictionary in the response body. This patch adds the manifest etag to\nthe response body.\n\nChange-Id: I97a958700b38a8e4ff7871b0ba3ec56e168f6626\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/08d1298af4bb5ab0a24518fa5ce0de20ada2987e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/08d1298af4bb5ab0a24518fa5ce0de20ada2987e"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
