)]}'
{"id":"openstack%2Fswift~917034","triplet_id":"openstack%2Fswift~feature%2Fmpu~I8689da307dec96eec9ff93f6b15938e6374e3b79","project":"openstack/swift","branch":"feature/mpu","topic":"p-mpu-acl","hashtags":[],"change_id":"I8689da307dec96eec9ff93f6b15938e6374e3b79","subject":"slo: use pre-auth\u0027d request to fetch segments from reserved namespace","status":"MERGED","created":"2024-04-25 11:49:14.000000000","updated":"2024-04-26 11:07:41.000000000","submitted":"2024-04-26 11:06:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"917034-p-mpu-acl","meta_rev_id":"bb67266e553796316f0fe6631232d15f556dbe95","_number":917034,"virtual_id_number":917034,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-26 11:06:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2024-04-26 10:03:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2024-04-26 10:03:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-25 12:51:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2a3e59e7844f447905faf51e9e27642730374279","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-25 11:49:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"efcbd4975127fababcf7b8b2979b905b9d8e71a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 12:51:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ca3829b17354c43934cb25909f0bac0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d665b6618bc9421ab724336a94a881e7 : SUCCESS in 8m 28s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c092860b11cc412f91871ba6b66b01df : SUCCESS in 17m 50s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cf16a8c4d21a44f78621eb9038bf5a62 : SUCCESS in 17m 07s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3d1bf88fcf3647149a641400bb118b8b : SUCCESS in 17m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c201d50f2094494bb069a66dea546d8f : SUCCESS in 10m 23s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/05aac2504a4645418674803499c7fef3 : SUCCESS in 6m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f6131e4ad2b34a3096e088638b3d9d40 : SUCCESS in 19m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/444245696c624666ae987f5c09c0d707 : SUCCESS in 16m 51s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f23ef671a3884fd88651967f176c0ee3 : SUCCESS in 19m 40s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/d92e06812c134253995cff51956ab58c : SUCCESS in 18m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/18e8509b22fa4dbab2fed641df53471f : SUCCESS in 18m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/38463f4e983640408bccfa470c0e9f68 : SUCCESS in 17m 37s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4a0940d85e6d4b1c816ef4e9a9f601af : SUCCESS in 19m 27s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4cafa8ec81214b1cb0b1a5311c3541ec : SUCCESS in 18m 26s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c991af1191504962801c0a41c3629cca : SUCCESS in 15m 01s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e66e8f211b0148b7a5f5f4a772a8058e : SUCCESS in 16m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f5eb9d4b3a00465cae161b5176332ae6 : SUCCESS in 6m 30s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3d3724eb036f4b469c270f3788e5c820 : SUCCESS in 19m 41s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4cd44b53a3234aa3a79bad5dbf8463b5 : SUCCESS in 7m 20s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6a11a995f7c24d6b9b02440b237d4084 : SUCCESS in 57m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e280ae2bd9f847e3bd9838ab745414b3 : SUCCESS in 19m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93839a71701642d1a582b047c8b4559f : SUCCESS in 6m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8f2bd7c92c6a4bf081d8795800c2e70b : SUCCESS in 17m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eb7146accff80e7239b860a98c032af06255a34b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 12:55:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5281d4a75b0641d5b9b49e6c38742b41\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4ca2e95b383e4d259b03cddf4abaf929 : SUCCESS in 18m 47s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c62b11ca10bb4895828075b68dcf9d6d : SUCCESS in 1h 04m 29s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f3e6111da05940558480b577242ba625 : SUCCESS in 20m 03s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1433fbfd8e914468b8f1135044e63871 : SUCCESS in 1h 04m 23s","accounts_in_message":[],"_revision_number":1},{"id":"765e462fa87f3dad9187ce008b5f81d793d6c6b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-25 15:35:47.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"91c7abc3d0906216e302309f009eb794a5f9dae9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 16:37:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba0438919ad142589e1ea5c5ed3e5940\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/726c7a4b4e91496292e1a45c7225e641 : SUCCESS in 9m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9952ece78a174529b78e9c52d7246ed5 : SUCCESS in 20m 02s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0eb7bbecdafc49a9a4012cbf04acddbc : SUCCESS in 19m 30s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a592923be3054b1d9876b436ccebb7d1 : SUCCESS in 19m 47s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/62f788dcce5e416ca53bb286c7292432 : SUCCESS in 11m 43s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/19d232a24f2b4e7fba92463b1b4f4f10 : SUCCESS in 5m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/84f6c0fd791e46bd9efea7e713009c99 : SUCCESS in 18m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6dd81707954a4bed904b8b4494be3d7f : SUCCESS in 18m 49s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/562ac57e8ce84e228af4ef14da2ff4a4 : SUCCESS in 19m 32s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/83c96aa709a64fe99ea2efb869c1ae65 : SUCCESS in 18m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/807437da31ab4ca2a8272436467b16b7 : SUCCESS in 16m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b57ccb749eca4c1baa286eea8a53f8cf : SUCCESS in 18m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6938ce12cd4343f8bb8b19f74ebacc41 : SUCCESS in 18m 41s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/26c841af9e7f43c7aa2b04e39ff8638e : SUCCESS in 18m 21s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1212f9175b354a828fb0b72a0aa2d4c7 : SUCCESS in 15m 25s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b4ba9e8c01074e3cb669da36d505a713 : SUCCESS in 17m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fd134027b71b4a0ebcc1834907b405b6 : SUCCESS in 6m 38s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e0486e74cfc54c43a0f1b4f7e8cfd875 : SUCCESS in 15m 05s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/576702f30735448fb61c8c1c6c09d694 : SUCCESS in 8m 54s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1723ec4e114446c581fcbf173a765e82 : SUCCESS in 56m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7ee1415a8b7e469ab2800e5d67b88be8 : SUCCESS in 20m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b004aad286d4438caaaa79a646e617bc : SUCCESS in 5m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1175b182381c4435bdfd2964b9373313 : SUCCESS in 18m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"24ee96d72de6c74ab1a75a290dab47ad0b83a482","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 16:40:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02a947df8e49427faba2f33d2581588c\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6d43f06ea5f24dba86c80830ada2c09c : SUCCESS in 19m 58s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/892c70f8350a4516b4faea8f1d456b91 : SUCCESS in 1h 02m 55s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/62f4c1f0c2604673b17569b40ec6368e : SUCCESS in 24m 59s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8ac81d3c064f4580be1a76be1b8dd9b8 : SUCCESS in 45m 39s","accounts_in_message":[],"_revision_number":2},{"id":"8d4d725ad9e80242b96d17c312c0144236d7ea48","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-26 10:03:12.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"98713edd5c7be6a4ee97db30b6340de07d189e13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 10:03:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6acfb2e5fd13bb951654ed6c2a102f568900ce02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 11:06:35.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3598096b4984718b6036550a117ec9c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52800814aba140ff8f2d9efa2a7050c8 : SUCCESS in 7m 42s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/0be60ea4aacb4ce98deba8bc78816fb6 : SUCCESS in 10m 22s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/fca7f143726243a7b78161144cc2551f : SUCCESS in 5m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0f7ae5391fdf43019d7606fb69e99738 : SUCCESS in 17m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/33b931056d5e467baa51ffe532e2483e : SUCCESS in 15m 49s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/bee52532adfc4541b45c0507604531b1 : SUCCESS in 14m 58s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/926da2de575544fe980c66e193189a04 : SUCCESS in 14m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1c048acff9dd43699eac4d59c1f2555e : SUCCESS in 19m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4a73de71b0b5440086ef9d56b574ccd4 : SUCCESS in 20m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/395b950f101a4854b0e23349a4679b03 : SUCCESS in 22m 55s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/58d15f491dc54d65838bc869f9e7b062 : SUCCESS in 15m 51s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e7dfa349cebb4c3982a9f7e50bc98815 : SUCCESS in 16m 34s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7bdf0b53cbd149bab80cc4148e1fffd6 : SUCCESS in 15m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/adfab84cd78b45a2bf0e33e5edaad502 : SUCCESS in 5m 52s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c9df2cac129344738b095fb9d44a224a : SUCCESS in 9m 39s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d6703e2301bd433e8677776aed0d0153 : SUCCESS in 58m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6dd5697315204c68abf3aa9412ca995e : SUCCESS in 23m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eaa70b2c54a04f47bfc8bb1a63443c3d : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":2},{"id":"d99b8f48b49917e9715e4f7a4adf10aeb2d7aa37","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 11:06:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"bb67266e553796316f0fe6631232d15f556dbe95","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 11:07:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e02b0b44b804f6fa1024d6f9430c902\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82e3cbe5c59f409ab83d86536308372c : SUCCESS in 44s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/de61410dff3843389546bcbc8a3b1076 : SUCCESS in 45s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/51af82c1c0974a57b9be236c2738ed5c : SUCCESS in 32s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0b7ea5acde8a93bdc0d951d0a1859fc7c322d2ef","revisions":{"17f5159eef49ba4e6bc955c1018724ae6178c084":{"kind":"REWORK","_number":1,"created":"2024-04-25 11:49:14.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/34/917034/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/34/917034/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/917034/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/917034/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/917034/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/34/917034/1"}}},"commit":{"parents":[{"commit":"51eb0e75d2aaeba4c78cc4c4e2a6cf550f10caff","subject":"Check for user container existence","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/51eb0e75d2aaeba4c78cc4c4e2a6cf550f10caff"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-25 09:21:46.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-25 11:44:01.000000000","tz":60},"subject":"slo: use pre-auth\u0027d request to fetch segments from reserved namespace","message":"slo: use pre-auth\u0027d request to fetch segments from reserved namespace\n\nThere\u0027s precedent in symlink.py for using a pre-auth\u0027d request to\nresolve a symlink when it points to a target in the reserved\nnamespace. This patch changes the SLO segment iterator to do similar\ni.e. if the SLO manifest has a header indicating that reserved names\nare allowed, then it is assumed that another middleware created the\nmanifest, determined that the segments are to be in the reserved\nnamespace, and takes responsibility for managing access to them.  In\nthis case the SegmentedIterable will use pre-auth\u0027d requests to fetch\nthe segments.\n\nChange-Id: I8689da307dec96eec9ff93f6b15938e6374e3b79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/17f5159eef49ba4e6bc955c1018724ae6178c084"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/17f5159eef49ba4e6bc955c1018724ae6178c084"}]},"branch":"refs/heads/feature/mpu"},"0b7ea5acde8a93bdc0d951d0a1859fc7c322d2ef":{"kind":"REWORK","_number":2,"created":"2024-04-25 15:35:47.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/34/917034/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/34/917034/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/917034/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/917034/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/34/917034/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/34/917034/2"}}},"commit":{"parents":[{"commit":"3621eca0ffad4684cd3b9b70d6a496619ebabfef","subject":"mpu func tests: add cross account tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3621eca0ffad4684cd3b9b70d6a496619ebabfef"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-25 09:21:46.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-25 15:35:37.000000000","tz":60},"subject":"slo: use pre-auth\u0027d request to fetch segments from reserved namespace","message":"slo: use pre-auth\u0027d request to fetch segments from reserved namespace\n\nThere\u0027s precedent in symlink.py for using a pre-auth\u0027d request to\nresolve a symlink when it points to a target in the reserved\nnamespace. This patch changes the SLO segment iterator to do similar\ni.e. if the SLO manifest has a header indicating that reserved names\nare allowed, then it is assumed that another middleware created the\nmanifest, determined that the segments are to be in the reserved\nnamespace, and takes responsibility for managing access to them.  In\nthis case the SegmentedIterable will use pre-auth\u0027d requests to fetch\nthe segments.\n\nChange-Id: I8689da307dec96eec9ff93f6b15938e6374e3b79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0b7ea5acde8a93bdc0d951d0a1859fc7c322d2ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0b7ea5acde8a93bdc0d951d0a1859fc7c322d2ef"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
