)]}'
{"id":"openstack%2Fswift~917070","triplet_id":"openstack%2Fswift~master~I95606aed5a5bd70424fbc51dd965f238fa7f064f","project":"openstack/swift","branch":"master","topic":"open-expire","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-04-26 15:04:15.000000000","reason":"Change was submitted"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-04-26 14:01:03.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}},"hashtags":[],"change_id":"I95606aed5a5bd70424fbc51dd965f238fa7f064f","subject":"object-expirer: test for mix of async and delayed deletes","status":"MERGED","created":"2024-04-25 21:45:28.000000000","updated":"2024-08-01 13:53:22.000000000","submitted":"2024-04-26 15:04:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"917070-open-expire","meta_rev_id":"506f21db548fd09f739e2bc25d4b6424fc2c6fa7","_number":917070,"virtual_id_number":917070,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-26 15:04:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2024-04-26 09:56:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":2,"date":"2024-04-26 14:01:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":1,"date":"2024-04-26 14:01:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-25 22:40:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-04-25 22:45:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-26 09:54:57.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2024-04-26 09:56:28.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"f7b194e08227aeccbc68671a53714d82e36d8a81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-04-25 21:45:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"12f2e5f99ecfa6c73996cf23531c637d58ef43a2","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-04-25 22:00:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"07b362dca6a2c5a6d4f6ea053e5e2d20fd73adf8","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-04-25 22:09:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f5c8a27fb1538075368e74bf59e0e57c5e6036f1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 22:40:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/127aca12ccf444b09e8c2903de0d1fac\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/626ef9896f624b65ae4f41256b903677 : SUCCESS in 16m 11s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b50fe56f8676405ba3f844037272d763 : SUCCESS in 17m 38s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b5fb6bd7f1c540a5b6684ddecb4ba44e : SUCCESS in 53m 40s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1eecf721e461447ab64351fc6fe21d58 : SUCCESS in 18m 45s","accounts_in_message":[],"_revision_number":1},{"id":"efa664268a5a8041d2e6bafff16f727fa6a88780","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-25 22:45:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/614fa937aeff4423b56a3fd107742412\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82c51e34b9bb446b8d94f6a34031582f : SUCCESS in 7m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ee9279653a344192b41474f0501c5f26 : SUCCESS in 18m 16s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/af4cf31e43fc4107b1e385da5e7ad1c9 : SUCCESS in 18m 28s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/53042b7bf8a84e25969ce672ce600944 : SUCCESS in 20m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c926d69cbb5749b1997025ae2c15412e : SUCCESS in 11m 24s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/3d3b537803214b44bea3b29539d1fa72 : SUCCESS in 4m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/80bfc4b4ccb94479b8e092347d522067 : SUCCESS in 20m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0a2d4405c417450d94f4898c5f9a0529 : SUCCESS in 18m 10s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f49ccb3b5cdc4c6181fb9b8935bd1536 : SUCCESS in 17m 19s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/a72ee29a2efe4b388c51820b3f6c7422 : SUCCESS in 15m 56s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cce1b2ec0a524e19b8e339cc10632cc6 : SUCCESS in 20m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/51f22b92addd46dbaeaf8b86dbf5e795 : SUCCESS in 18m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e803508e567b4e389b0d8709e613cb10 : SUCCESS in 19m 15s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a2f091af625b48daa4e6e6f0646ac98c : SUCCESS in 15m 30s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9ddbc49a3d3247448340cceb6126e718 : SUCCESS in 16m 13s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0c0925b6740949d8b0bc656bb5fb087f : SUCCESS in 17m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/702fb62392c5469886c55b4f690eff4c : SUCCESS in 47m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b16713169d1c4c83baef8373116538a1 : SUCCESS in 54m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/930e3b55887543ca9707d69466886289 : SUCCESS in 18m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f916f98f44a4b5083f9e618a801f54c : SUCCESS in 4m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/18041084053e4fc5a425bd7ee7450dfd : SUCCESS in 23m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c2c84942c03d9f68ea0d8501a53006a5fec73970","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-26 09:54:40.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"9034a90fde304ed35fbe548066bad920d011af9d","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-26 09:54:57.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"2d0d5bf085191d29e734c6c7771bbf52c38089b5","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-26 09:56:28.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dedb26f91584e8af7bb6888fb2f6f51ce9c8b164","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 10:51:28.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04d37e106e2f437a85574682e64e4126\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5d7a15c0cccb4669a2491fc6ff9a6c04 : SUCCESS in 15m 38s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5b2ed2709d73447a9f1f1125fe783f45 : SUCCESS in 14m 49s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/5c00c1565e17484ca734eca70c3e0291 : SUCCESS in 54m 54s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8a1e944de8df4ba5bb19fadd695e2622 : SUCCESS in 21m 39s","accounts_in_message":[],"_revision_number":2},{"id":"6e11f23a909e1f39ea7fdb892cca07b081ad6dcb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 10:53:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76d68d2b571444a29923c80d8db4723c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90e95636227448038d50450a731952ed : SUCCESS in 8m 42s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2dd39083b4804eb2835858aca08abcec : SUCCESS in 18m 22s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2be2eaa5ed594765940b6340f6ec81ba : SUCCESS in 18m 54s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/84c240839eb4474c9458c3d749dddfe5 : SUCCESS in 19m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e14921d1c35549d6b8af892f34c15d71 : SUCCESS in 10m 51s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/eea37d1bc7e0436ebee3c8655bbe9cd3 : SUCCESS in 4m 54s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/94bed4ec28c949228e276efe688fac51 : SUCCESS in 20m 32s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/654753f9857b45b996521f43dfb51846 : SUCCESS in 23m 33s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b9757d83a0cf4012b7a1452b2bb72d40 : SUCCESS in 16m 29s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/c417a53ef5824ab18d591d8c6a0b36a4 : SUCCESS in 22m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5f6a61fe09664d3ba36ed169637fed37 : SUCCESS in 13m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/003f118c9ff4460d8d9acdda7b66de95 : SUCCESS in 18m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9f401197d438435fb051805c5d143fdd : SUCCESS in 26m 19s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/aaa47cd4c56447e5b4e3fb828c972005 : SUCCESS in 16m 12s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ea3d68d19fa1494ea1f456ae2840f7aa : SUCCESS in 13m 59s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/60f5c0ffc1034f78933ebb008354fb83 : SUCCESS in 22m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5588dd85ac6e4063850e5a7dff9b98f3 : SUCCESS in 58m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5e53374784ff4a129d2ffd5e06a46bad : SUCCESS in 51m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/92a705f5769a45b7b6f581ac057e8832 : SUCCESS in 18m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34678fef04c14a7abd481435d6fed422 : SUCCESS in 4m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b1bc803fc9bd4614938e6bffee819102 : SUCCESS in 26m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"09c47292aaeacd4739484ffc3229b08fb804e4d6","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-04-26 14:01:03.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d831ec1a72ba58a90b3e124e454de403ac87efd7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 14:01:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d0ef1a412b062f698762381d24f2b03ed6accf22","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 15:04:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56bb8d948d1147c6b78d205f08f5a318\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eebe9766a980423caccbe8917aea0d9a : SUCCESS in 9m 15s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/86de43a15de040639bfb0e406a2bb33a : SUCCESS in 11m 10s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/6dea173e2ac549bf9dfe7033cf3ad81d : SUCCESS in 6m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bf97da7e7ab54b6db77d905c8b6d55cb : SUCCESS in 13m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3a599548c77f4b68a17f9415f78948e7 : SUCCESS in 18m 31s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/654a45d3230c46b88d67b47f355d148f : SUCCESS in 23m 07s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/087d2323b54348bfa11db2395f1cd7f8 : SUCCESS in 13m 44s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/eab865fc74df47f9b3604585a9cbcab6 : SUCCESS in 19m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/528f8d1ea56049b9b6603488f2f816c9 : SUCCESS in 18m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/35cbe527f5c04e19ac2974df35739262 : SUCCESS in 18m 21s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/bf295f7cafb843339b6e12fbf3dfb5ab : SUCCESS in 16m 32s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/498243e84a6140dbad505e0943c99ba3 : SUCCESS in 19m 13s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6423665827c64482b26e37480d5df333 : SUCCESS in 18m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e0434b219bdd4d5aa13b60671f0caddc : SUCCESS in 6m 47s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0a89359699c84e5896f6cd29618c02c3 : SUCCESS in 7m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0250c2c4f96d4ad998cc9b0769421bbf : SUCCESS in 41m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/009c577aa9cc47219425b87e4a7f2659 : SUCCESS in 54m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8d036a1f53fd4431b31d117701a26d2b : SUCCESS in 20m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3fef513faf14427b9740e8bfceca629 : SUCCESS in 4m 26s","accounts_in_message":[],"_revision_number":2},{"id":"fe791c166a0cec308a6d87ea1a531dcb69573508","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 15:04:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"61bac5b9365c0d3fb7a33f8aa28834b9fca93d16","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 15:05:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5851809f5e54dd4ae7b81966e362036\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4445ba179f4b43b783e69498e5be0b7f : SUCCESS in 46s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/d94c0c22b22c4186a9190263e31fc7ca : SUCCESS in 37s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/3e840d0b67b44594b6523a2c1c8900bb : SUCCESS in 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"506f21db548fd09f739e2bc25d4b6424fc2c6fa7","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-08-01 13:53:22.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b490857b475207fd9baf69168d75231e65562032","revisions":{"44c3fa3e829ca194510be90ec3d416f24c9d8de5":{"kind":"REWORK","_number":1,"created":"2024-04-25 21:45:28.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/70/917070/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/70/917070/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/917070/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/917070/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/917070/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/70/917070/1"}}},"commit":{"parents":[{"commit":"5961ba0ca7ddcfe296627fbef01ec4b32d538290","subject":"expirer: account and container level delay_reaping","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5961ba0ca7ddcfe296627fbef01ec4b32d538290"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-04-25 21:45:18.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-04-25 21:45:18.000000000","tz":-300},"subject":"sq? this convined me!","message":"sq? this convined me!\n\nChange-Id: I95606aed5a5bd70424fbc51dd965f238fa7f064f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/44c3fa3e829ca194510be90ec3d416f24c9d8de5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/44c3fa3e829ca194510be90ec3d416f24c9d8de5"}]},"branch":"refs/heads/master"},"b490857b475207fd9baf69168d75231e65562032":{"kind":"REWORK","_number":2,"created":"2024-04-26 09:54:40.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/70/917070/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/70/917070/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/917070/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/917070/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/917070/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/70/917070/2"}}},"commit":{"parents":[{"commit":"5961ba0ca7ddcfe296627fbef01ec4b32d538290","subject":"expirer: account and container level delay_reaping","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5961ba0ca7ddcfe296627fbef01ec4b32d538290"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-04-25 21:45:18.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-04-26 09:53:19.000000000","tz":60},"subject":"object-expirer: test for mix of async and delayed deletes","message":"object-expirer: test for mix of async and delayed deletes\n\nRelated-Change: I106103438c4162a561486ac73a09436e998ae1f0\nChange-Id: I95606aed5a5bd70424fbc51dd965f238fa7f064f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b490857b475207fd9baf69168d75231e65562032"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b490857b475207fd9baf69168d75231e65562032"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
