)]}'
{"id":"openstack%2Fswift~917224","triplet_id":"openstack%2Fswift~master~I0441226e279b36a787ec85a459d6d1b72a2c8c84","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-04-26 20:07:43.000000000","reason":"Change was submitted"},"36763":{"account":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"last_update":"2024-04-26 20:07:43.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0441226e279b36a787ec85a459d6d1b72a2c8c84","subject":"Add Anish to authors.","status":"MERGED","created":"2024-04-26 16:54:57.000000000","updated":"2024-04-26 20:08:49.000000000","submitted":"2024-04-26 20:07:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"917224","meta_rev_id":"e92092990b63320c9b57ca2341a2ff8e02385ad8","_number":917224,"virtual_id_number":917224,"owner":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-26 20:07:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-04-26 18:51:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-04-26 18:51:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-26 17:00:25.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2024-04-26 17:41:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-26 18:51:52.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"b43e4434d97b12a7350af34936ac9460193b31bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"date":"2024-04-26 16:54:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"12759bd0ec570c8fa0c51c04e12f2d63e7b76922","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-04-26 17:00:25.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c2c40a775d723dce34462579a76bf8e558e23633","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 17:41:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/795646d318f14f8ca8c1d9ded6fbeea8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a11c0afd7e94df68270a641c00e3ddd : SUCCESS in 8m 10s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/72b25fb3bbc94466b6db671367c71189 : SUCCESS in 11m 48s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ad739bb69f824cdfba471ecd58991e9b : SUCCESS in 5m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ec66a63779314f1ca2325b737609e8ba : SUCCESS in 19m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1cec0e0ba0924696b52475fabfd21e25 : SUCCESS in 19m 47s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/47afe726ec924869a01976cabebb2bd3 : SUCCESS in 17m 00s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/04033936d0644828be007e92cc108adc : SUCCESS in 17m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0ec6005c47864fd999578b37f3eee5ba : SUCCESS in 27m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/911502f43ede4f91a08be37e8bcbc0d2 : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":1},{"id":"0c5ee0cf83106dfc68dcc285577f01b18dba4ea2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 18:11:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3b31b2797444335982f0e2a09e7569b\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bfcb547d7e2a4bd8a65d2f99831e1591 : SUCCESS in 29m 13s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d96a58f322dc434581423ffb7921673f : SUCCESS in 1h 09m 27s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/64b1714e992942dc9673c012e11fc6ea : SUCCESS in 1h 09m 54s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d58b2e4ef4b44751ab2ac0e593ab1e73 : SUCCESS in 1h 04m 48s","accounts_in_message":[],"_revision_number":1},{"id":"6f8d85a1b4f47ef003c0a7b645153822fd4f793f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-04-26 18:51:23.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"7002ecbeb9d98c66912480ea73848d060aa8f0a4","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-04-26 18:51:52.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"362a0f480b434ae9610a47253d53bb381ed69050","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 19:29:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a245eedf9624a1a8de1026aaef66e2c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f63e5cefc2b64f66a919fc6263084981 : SUCCESS in 8m 45s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bf3f5991bf5949749c914149896fd10e : SUCCESS in 10m 49s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a67589bbe3804b57b05c27eee6d3bc2f : SUCCESS in 6m 42s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2245145ab7634e2bae520e6bea3056f7 : SUCCESS in 20m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fed42cc78e7849bfbe2b0e5b35ec0f46 : SUCCESS in 18m 08s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ce41587cd6d4490eb1c11995b0436c3d : SUCCESS in 16m 27s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/af41d68267b34616a0e8472497ee58c7 : SUCCESS in 20m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d0b8c96e81a44007b2b97c78f38a9cb4 : SUCCESS in 14m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/672be8151f8e435db368a90a577666c4 : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":2},{"id":"67d4938fac371573bd03da0c8606c2bbd51f8cc8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 19:29:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d86f89b4c495fb34f6e98ce8dd580c14cb019866","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 19:58:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f226ca11b1c4aa4a0691ca651bde183\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/99238d3801964461b597584f8eb5343e : SUCCESS in 13m 45s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ed3869e79d104598913ae81c8ad47302 : SUCCESS in 44m 26s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c329220f2b4d4759a76267922b1f444c : SUCCESS in 22m 02s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b2acc78d6aac41a394e7fcddd2371570 : SUCCESS in 1h 05m 45s","accounts_in_message":[],"_revision_number":2},{"id":"994f4d0482b40873ab8d0a1ae6033d1351842b97","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 20:07:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e29b358b0d143b58d8581f4946eb9e0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f0c4eb944e54f548fb227b08039fb26 : SUCCESS in 8m 51s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/36df082212f248f592dcaa22c0069c53 : SUCCESS in 11m 21s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/467e27ba18d94f0aad8b250475b17bbb : SUCCESS in 6m 15s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ae55c895fe074cbeac9fe9f97a1837ce : SUCCESS in 18m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/84b24a9901c544ffb111ee7acedbc7aa : SUCCESS in 17m 26s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5e75c59b88f849638fb29e886e49a4eb : SUCCESS in 16m 55s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/09bd039906a94b17b40666f63c1742a8 : SUCCESS in 16m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/39a61c47822945e9bf7f3b8bceec0564 : SUCCESS in 18m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ef971f46697d4def8e20337fba0865e1 : SUCCESS in 19m 40s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/31b07bfc33ef421cb10cc9253b6c37c5 : SUCCESS in 19m 15s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0bd14c40dc3541b1879f14eb0f9bd411 : SUCCESS in 16m 07s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/50faaf20a9d34dcbbcd2367a4525d7d0 : SUCCESS in 15m 23s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3283b434de6b40e9baa4422899362389 : SUCCESS in 16m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d55db06f195f4e03ad2d9cf34e221e26 : SUCCESS in 6m 20s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2a6c151ab47d498581487df41885823d : SUCCESS in 6m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/caa1bd302a774333881ec87081487aab : SUCCESS in 21m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc5ed0d1bd204311bffb761c06b1e743 : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":2},{"id":"b2ec637323b74a6eee98365b9b754d5b66b1154f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 20:07:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"e92092990b63320c9b57ca2341a2ff8e02385ad8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-26 20:08:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e90dce48fd8f41dd91caae422f12db9f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6af464d2a64c4e3fb1ec10d833a34eb5 : SUCCESS in 46s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/cabd78bdc70b4112a875da855ee81f34 : SUCCESS in 30s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/e55528fc023a482f9c536630d1a0eea2 : SUCCESS in 31s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1d30e8e69689d97e03f8883c66320eb125adae75","revisions":{"fcad5bd7ae7e3c8ebf03e68cadbe83fba09850e9":{"kind":"REWORK","_number":1,"created":"2024-04-26 16:54:57.000000000","uploader":{"_account_id":36763,"name":"Anish Kachinthaya","display_name":"Anish","email":"akachinthaya@nvidia.com","username":"akachinthaya"},"ref":"refs/changes/24/917224/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/917224/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/917224/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/917224/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/917224/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/917224/1"}}},"commit":{"parents":[{"commit":"937af35e6251e559379eb41aa2799388eea7a2de","subject":"Merge \"object-expirer: add example to delay_reaping sample config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/937af35e6251e559379eb41aa2799388eea7a2de"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-26 16:54:32.000000000","tz":-420},"committer":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-26 16:54:32.000000000","tz":-420},"subject":"Add Anish to authors.","message":"Add Anish to authors.\n\nChange-Id: I0441226e279b36a787ec85a459d6d1b72a2c8c84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fcad5bd7ae7e3c8ebf03e68cadbe83fba09850e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fcad5bd7ae7e3c8ebf03e68cadbe83fba09850e9"}]},"branch":"refs/heads/master"},"1d30e8e69689d97e03f8883c66320eb125adae75":{"kind":"REWORK","_number":2,"created":"2024-04-26 18:51:23.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/24/917224/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/24/917224/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/917224/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/917224/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/24/917224/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/24/917224/2"}}},"commit":{"parents":[{"commit":"937af35e6251e559379eb41aa2799388eea7a2de","subject":"Merge \"object-expirer: add example to delay_reaping sample config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/937af35e6251e559379eb41aa2799388eea7a2de"}]}],"author":{"name":"Anish Kachinthaya","email":"akachinthaya@nvidia.com","date":"2024-04-26 16:54:32.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-04-26 18:51:17.000000000","tz":-420},"subject":"Add Anish to authors.","message":"Add Anish to authors.\n\nChange-Id: I0441226e279b36a787ec85a459d6d1b72a2c8c84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1d30e8e69689d97e03f8883c66320eb125adae75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1d30e8e69689d97e03f8883c66320eb125adae75"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
