)]}'
{"id":"openstack%2Fswift~917664","triplet_id":"openstack%2Fswift~unmaintained%2Fzed~If5a087978785399c262f093825cfac3906373159","project":"openstack/swift","branch":"unmaintained/zed","topic":"create-zed","hashtags":[],"change_id":"If5a087978785399c262f093825cfac3906373159","subject":"Update .gitreview for unmaintained/zed","status":"MERGED","created":"2024-04-30 11:06:12.000000000","updated":"2024-08-03 16:09:48.000000000","submitted":"2024-08-03 16:08:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"917664-create-zed","meta_rev_id":"7566d693797a6cfb8d081aed74a733be9d16b5f6","_number":917664,"virtual_id_number":917664,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-03 16:08:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2024-08-03 15:06:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2024-08-03 15:06:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-30 11:09:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-03 13:56:37.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2024-08-03 15:06:52.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"e12403e71b7774b3af6f053bd82ac8c98354c858","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2024-04-30 11:06:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"538e57672abb457571d886affd5544f7aaa735f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 11:09:10.000000000","message":"Patch Set 1: Verified-1\n\nUnable to freeze job graph: The nodeset \"centos-7\" was not found.","accounts_in_message":[],"_revision_number":1},{"id":"24378a72cf6854b8f54e192bc011474d4c3dd568","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 12:17:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b5179efd25245ee9206338ec8eeb04b\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/89415c65f1db428c8a68348c2c94b0c2 : SUCCESS in 21m 13s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c821f68eb01347aeb8e770f41dbb8b58 : SUCCESS in 19m 18s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8f53952b0ff8463080deaa1463a0c8b1 : SUCCESS in 57m 26s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/bfb2f1cd50e64768b116808e10569a14 : SUCCESS in 26m 47s","accounts_in_message":[],"_revision_number":1},{"id":"96cefb1bdbec5b9c4b5b44ff0eac44c29493b03b","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-08-03 13:56:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"56038f55e484c8fdfb260460942eaeb17aca6085","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-03 14:55:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1d2f92be2624707a1908777c5a5e7ff\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/276d8a208d9b46d4a4094b2769377a58 : SUCCESS in 8m 11s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8b2f7239e4064dc0b889c4aa6cfecd88 : SUCCESS in 10m 38s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/125e20cb5fdc48b6bd0de5908d93b74d : SUCCESS in 4m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7aea4783aa1c4e97acd47d873082aa42 : SUCCESS in 16m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/30a08ddb11d4404c8aa1dcedbeb58e49 : SUCCESS in 14m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c2934bd058ea4d81adbc932d0c2b1651 : SUCCESS in 11m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/144714a463ad4881b6933ddd17cdc0e5 : SUCCESS in 15m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/735421ec9bfd48d68c636e26c99cad1f : SUCCESS in 4m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f32bc070b04745b3a696cb1c7183a928 : SUCCESS in 53m 38s","accounts_in_message":[],"_revision_number":1},{"id":"02fa1b6beb88dc146be4e4ee8e589b0d9fd1e497","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-03 15:03:14.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3851455feb6547d5a12a7cc84ed39754\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4e627a04bd584b80a7da7f0a48b01174 : SUCCESS in 46m 19s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6f25b662c6cd4899ab747d5481a883a7 : SUCCESS in 1h 04m 20s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f58c85cae05849c38a0c10e19db49b4f : FAILURE in 50m 55s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/bee4c7ea93134d5ea2a1d775775f80fd : SUCCESS in 40m 30s","accounts_in_message":[],"_revision_number":1},{"id":"49e955a3759e033f458b603c61e9cf327c299fa8","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-08-03 15:06:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"380a2b07b5cee6298d6b82fb705ea9d8702cde97","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-03 15:07:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"aa1b721a9fa6fd1cf3d93e272cf08e97e0928210","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-03 16:08:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14e56dad1c9d47e3a5d1770e62552999\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a8288c155ef415f80c5d69b9c46fc65 : SUCCESS in 9m 07s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/4a669cc0e1994bfe988b2cb7d36d4e66 : SUCCESS in 11m 59s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/d821476e628a42b1a33c4788333b7a41 : SUCCESS in 5m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b850e8163b1449c2a96cdd2e6ee5203a : SUCCESS in 15m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8ec611aa99894054a0569abe1cbb15ea : SUCCESS in 10m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/71c0de9d5f164157b95c79ba995ab6e6 : SUCCESS in 14m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e659cab375a44925a347d0c2c33a464b : SUCCESS in 13m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2132bd47b0674be09e96c715543a2bf9 : SUCCESS in 15m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ee78dc42b7d845c9aef2a1c81b0d62b3 : SUCCESS in 16m 43s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/aee6f8eba0f949f9b961bf2600b0632e : SUCCESS in 15m 50s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/015cd96f50e14711a2dd6553eae3a2a5 : SUCCESS in 15m 15s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/39b1e766f65a43358366f781a4ebbbc3 : SUCCESS in 16m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1a439aec2b0d425c822f2e759862ab39 : SUCCESS in 7m 36s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1ef4bc9f0e1e4e32a893e23e605f3c5e : SUCCESS in 6m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2094d26369f046aa8a6bbcdd6bea43eb : SUCCESS in 18m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbb9f69448f741d1a937c511f9d0bebe : SUCCESS in 4m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a57a1bcfbc824ec184ab73c06df18c77 : SUCCESS in 54m 45s","accounts_in_message":[],"_revision_number":1},{"id":"8f59286712c31ec71f74eb6a027e96cd63dda647","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-03 16:08:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7566d693797a6cfb8d081aed74a733be9d16b5f6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-03 16:09:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2dfc1c9bf55e4062b4cd2753f07092aa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbac4ce45039454a8ca949a2155863e8 : SUCCESS in 49s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/315105e51ff04fe298ac747fe652bf7c : SUCCESS in 32s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/1e19e9daa82d465abcd7fe560a8e853e : SUCCESS in 28s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e46bf8a9b59c94fcadb3eff88e2889089404355e","revisions":{"e46bf8a9b59c94fcadb3eff88e2889089404355e":{"kind":"REWORK","_number":1,"created":"2024-04-30 11:06:12.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/64/917664/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/64/917664/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/917664/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/917664/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/917664/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/64/917664/1"}}},"commit":{"parents":[{"commit":"1217b3bede4b60360b5a5de85abf826b3d1e8778","subject":"Merge \"Update url\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1217b3bede4b60360b5a5de85abf826b3d1e8778"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-04-30 11:06:11.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2024-04-30 11:06:11.000000000","tz":0},"subject":"Update .gitreview for unmaintained/zed","message":"Update .gitreview for unmaintained/zed\n\nChange-Id: If5a087978785399c262f093825cfac3906373159\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e46bf8a9b59c94fcadb3eff88e2889089404355e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e46bf8a9b59c94fcadb3eff88e2889089404355e"}]},"branch":"refs/heads/unmaintained/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
