)]}'
{"id":"openstack%2Fswift~917711","triplet_id":"openstack%2Fswift~master~I5a96626862644fe23a0d4bc53e85989770c8ceec","project":"openstack/swift","branch":"master","topic":"proxy-logging","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2025-05-21 17:14:51.000000000","reason":"Change was submitted"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-05-21 09:34:37.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2025-04-03 23:28:07.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},"36606":{"account":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"last_update":"2025-05-21 17:14:51.000000000","reason":"Change was submitted"},"35790":{"account":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"last_update":"2025-05-21 17:14:51.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-05-20 23:33:46.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I5a96626862644fe23a0d4bc53e85989770c8ceec","subject":"Add labeled metrics to proxy-logging","status":"MERGED","created":"2024-04-30 13:59:30.000000000","updated":"2025-05-21 17:16:00.000000000","submitted":"2025-05-21 17:14:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":263,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"917711-proxy-logging","meta_rev_id":"41f24ecc53dac5148ffcfca0e771b66483b4abf4","_number":917711,"virtual_id_number":917711,"owner":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-21 17:14:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":2,"date":"2025-05-21 09:34:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},{"value":2,"date":"2025-05-20 23:33:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":1,"date":"2025-05-21 09:34:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},{"value":1,"date":"2025-05-20 23:33:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"}],"CC":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-12 14:20:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-12 15:31:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-28 18:38:52.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2024-07-12 12:30:31.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2024-07-16 01:29:41.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"},{"updated":"2024-07-31 19:19:12.000000000","updated_by":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"reviewer":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"state":"REVIEWER"},{"updated":"2024-07-31 22:04:54.000000000","updated_by":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"reviewer":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"state":"REVIEWER"},{"updated":"2024-08-07 04:58:31.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"},{"updated":"2024-08-07 04:58:51.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2024-09-10 16:51:26.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"CC"},{"updated":"2025-05-20 23:33:46.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"7ad79bf68500cb3108cd6a0d8af7136669ca5112","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-04-30 13:59:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5aafeb9728143dc974ad6b803b06a65f4f5e4537","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-06-05 13:59:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d21e9e6090603d70236c2da95a05c66d1266407e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-06-06 13:59:33.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"1c61a697392e649b1c4a44f7c1d1f31c9dffea46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-06-10 16:22:34.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"17892114492bf0ff09b198c1a1275134b522bb9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-06-12 13:59:35.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"19e6e93a4faebf712173156247e712dffb6f79f9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 14:20:55.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7df5af5ad85b49228b0bbbd486bf56c7\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4e4fbb1011ba46fcbb5836cd265eb57d : SUCCESS in 16m 08s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d2c2db9365d949718c4cbf7c5ac4a824 : SUCCESS in 16m 38s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b691e7e717b3465596529975c777a136 : SUCCESS in 18m 29s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6817c164d013499984efb611a1a6e7d6 : SUCCESS in 20m 00s","accounts_in_message":[],"_revision_number":5},{"id":"d0560ade97d0ca11b7f3b26d30a17eb6102cb2ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 15:31:30.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/978ba9e814c84cb499140c0e4639c6b8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40ca2389f1254f87a1c5c04a83c92a5b : SUCCESS in 8m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/2a0c3d0fa3b24acba6fc5dcb33bccb24 : SUCCESS in 59m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cb4108a4cd734f75b438c790a26d71b6 : SUCCESS in 52m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c475b265cd5a4cfcb7d51cddf40aa60d : SUCCESS in 1h 07m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b013e8a8b02b4fee8a617cf3e63a3973 : SUCCESS in 20m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6d4c75da15ce4a8788186e00e4d3276a : SUCCESS in 17m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3d4b2b1b17e04e99a45655e6dd39874a : SUCCESS in 17m 50s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5a17c04bde3848c997dd69fb10245ee2 : SUCCESS in 9m 45s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/309be76571fe4f0e8a3f23b87a53fe71 : SUCCESS in 4m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cb642a5255be4740b4e54d8e6acd661b : TIMED_OUT in 1h 30m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/dd1b51c0957f4b05ab4a710a47b62958 : SUCCESS in 18m 37s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a2eb38978f564a3988d980072acb4426 : SUCCESS in 16m 38s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/542edeabb41c49cfa5fa012d95a5a54d : SUCCESS in 16m 01s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/14126bb6905647a69dad0a99841f24aa : SUCCESS in 25m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fa366cd780a046fdb3db65337650d019 : FAILURE in 6m 37s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/52dc8d0ab2a7446f8cabd025d8331a60 : FAILURE in 4m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4e72c7cd6bcd47e7b3b9bbbc3bc2e03b : FAILURE in 4m 27s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/81cf656d9c0c492fa032721868fb6ab6 : SUCCESS in 15m 00s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f927c625d0dc45cd896d34cdcd27557f : SUCCESS in 17m 19s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/77915086857a40e1ba70a827cc695be1 : SUCCESS in 17m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d64a01e5a94a46fcb6f0a9af9d1c52c3 : SUCCESS in 6m 00s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/441780655391413790151792262879bc : FAILURE in 17m 42s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/24df7233a6314f68b78a15aa4c1b1d64 : SUCCESS in 8m 00s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/aceea7c5f8164de7bbd38e4e920b9bd5 : SUCCESS in 58m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/358292bd89424fc1849ebd7e6745a3aa : SUCCESS in 55m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/93e5478e6e6d4e9cb044f983012e4c25 : SUCCESS in 42m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4916f7a74a9646768a6d1102bbc90cad : TIMED_OUT in 42m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b47b7503102741e8ab48f94d1b2b8d7f : SUCCESS in 4m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bff80ecf3df044deaef5877a5149e1bc : SUCCESS in 19m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3b6b723e772544579ef563f6e73b88e1 : SUCCESS in 1h 04m 06s","accounts_in_message":[],"_revision_number":5},{"id":"c07b8c1d96bcc33fae850ca6b1b672e2d6982e9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-06-13 19:58:36.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"4941ad09d856d2303782b3172a21cd04e5ad9781","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-13 21:26:34.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b879e0baffd54d1a9104217b952a0477\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab9bf59902b74d31a5b74c0dd6d27bbc : SUCCESS in 7m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/aed16218456b45c499c79eef125ce085 : SUCCESS in 59m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/52b3a3dff40b4467a5886dd5bf8e0f8d : SUCCESS in 50m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7f7d3af70fe54285b23e654f37ccde47 : SUCCESS in 1h 07m 12s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4901bc1b9acd47acacce56fcc4caf802 : SUCCESS in 19m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9159a8960c5b45419b43e57a8a897355 : SUCCESS in 17m 43s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fc293f3c71a548a8b5eaac985dba4fae : SUCCESS in 18m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d4f4085ce7f544d4938e00411a794572 : SUCCESS in 10m 38s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/319ee64187554a53b4075fea01a1ef7d : SUCCESS in 5m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/46c1091b81cd4d8f908dcb28ee7ce8bc : FAILURE in 20m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/dd24760db9954343a95db00af479e702 : FAILURE in 18m 28s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9d6cf3a755484293bc6a7cf90bad6cea : FAILURE in 16m 28s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/979e184ce71f4b43a3b96ee62bf09462 : FAILURE in 16m 29s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/107a7e9a62884a58b60942a2a6cb3b2c : FAILURE in 23m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/db6670ed64a74d9eb73a577006faf4ac : SUCCESS in 19m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3c0a4cfda0a645c5852c97b6c3992278 : SUCCESS in 20m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/72b1350e86aa42cd98f79cceae79732b : SUCCESS in 18m 01s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6380293437f544d8b62322f1091d4e3e : SUCCESS in 16m 10s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d3f3fcca36594aecb91cc39d77453073 : SUCCESS in 15m 58s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1c5f05fe8f4745db91d755507e33e0da : SUCCESS in 15m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/13625372723c4e56b758797918d10802 : SUCCESS in 9m 45s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2f3ed09a07d04618bc3314231dc26c64 : FAILURE in 16m 55s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d8b15b6e760c414e94473b841e449cbc : SUCCESS in 7m 04s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0678c52dbd144cd98915747503cd4a92 : SUCCESS in 59m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2c15cf338ded483ab694a9ada63e3c5d : SUCCESS in 55m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/85c2dc05afb14f29bf47099d80308e48 : SUCCESS in 47m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4f552a431b24416f9c4d0a8313d86bbc : FAILURE in 24m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf2ec71344944a60a2dba8542a0fb6b2 : SUCCESS in 3m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1fba495d720048f1813a3c6d03db51be : SUCCESS in 17m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/03a86d36a9ac41f6aa7ee7ee5976c5c7 : SUCCESS in 54m 25s","accounts_in_message":[],"_revision_number":6},{"id":"4705ff071b67f15266c06c828eaca4f5492cf2ab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-13 21:31:59.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9753a54c7bcc4d4293e0b6e7ebbca9ef\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/edb2372ec5b844c286b647b9496217cd : FAILURE in 17m 10s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9af95231b0a849d0a842504cc334406a : TIMED_OUT in 1h 30m 58s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/217bb5249fe64fddba8bb0c358a68af2 : SUCCESS in 20m 46s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/93d871f54203402c99ca311f40c4275a : SUCCESS in 1h 24m 50s","accounts_in_message":[],"_revision_number":6},{"id":"ff7444225be3eb4fecf069ee943e611733a66f16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-06-17 21:22:55.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"8da853b117b333cac3bc50aabe946796d8391f75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-17 22:31:18.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3621c5a85c734b42b365a37dde2cb03f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ea582becf6c45dfa9cb787f9ef292ef : SUCCESS in 8m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/ad9312730f5c4fa5b62312c825eeea92 : SUCCESS in 57m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/01c0d39fb36040c18060ec230981d068 : SUCCESS in 48m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5208bc5b65624dd5926e7d52a3e38056 : SUCCESS in 1h 02m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/186295b74e22434b8fc38725eeceaf59 : SUCCESS in 19m 33s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9bbbe64827664e7fae6a61a2a7dcaf11 : SUCCESS in 22m 03s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fba0b4afa33c472da84943823f7ccb7a : SUCCESS in 19m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d7b201acd6e74a35aa319080326c14c0 : SUCCESS in 11m 23s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/472248eba6c84e63a87d351036be396c : SUCCESS in 6m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b279c1ad30354c7d8aca3b3810c3b9f6 : SUCCESS in 20m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b689e71330074e6994f93985cd8070a5 : SUCCESS in 24m 05s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/fccaf31812144d9bbe8086a7ebdc2483 : SUCCESS in 18m 19s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/34862229246f49de802900fbc0e205e8 : SUCCESS in 18m 07s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/49929ef06d2e4c01a797a65bb98b3dc1 : SUCCESS in 18m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fdf420384ec94ee896e2e11d9193116a : SUCCESS in 14m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/09608004a7264915badb44cba152c8c7 : SUCCESS in 14m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/baae4c29d51547459e63bb5db1eeab39 : SUCCESS in 18m 22s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8a7aefbdf4f44a578af2e00b353a6fd2 : SUCCESS in 15m 18s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a774f1ce4d9040d58ebd4dfda0df7b1b : SUCCESS in 16m 19s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c45ff62aca01427b98ed63f96aaf1430 : SUCCESS in 20m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3ca6424d8f2a4025826b20f5fc74e17b : SUCCESS in 8m 38s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e29a51bd42904c4dbb27fc7e96de7156 : FAILURE in 19m 25s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/eefe37a452014ab39f9a0b4c64c7efca : SUCCESS in 5m 56s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3727b7a7ac174df5bdd55f74775fdde2 : SUCCESS in 59m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bb6ccbc423f04fbeb6eb10d9bb5dd672 : SUCCESS in 49m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/91a133d858ce4a2ca152b2c42bad563c : SUCCESS in 50m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f311e464baed43e4a8fe471c4233043a : SUCCESS in 19m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca8446fdfe9e40368378be5882116f70 : SUCCESS in 5m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9cfe9196d67b4c2abcb5ab226be54569 : SUCCESS in 19m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/57948f8e123d46418a63d05a89eb8dfb : SUCCESS in 1h 02m 48s","accounts_in_message":[],"_revision_number":7},{"id":"e79c9a85e9feabd8c13305fcacb5205509bad0e6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-17 22:34:20.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b15b2d9e39094017960c7f5f5ca0ae56\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1bcfa5bd30314a5f814bba06e421eae8 : SUCCESS in 16m 18s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7a83c8e63d36435b9cc312328b781231 : SUCCESS in 49m 49s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/76bc6d4b9d504f828d663591b56034c1 : SUCCESS in 21m 46s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/fea7774066034aa5be29de93670457b8 : SUCCESS in 1h 09m 48s","accounts_in_message":[],"_revision_number":7},{"id":"e0f200973346717d37dd0724682a7ebaec04122f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-06-25 13:58:34.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"2402b80cf265af9525c28db0c862bca7a4ccf8c4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 14:19:53.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1a90a025e8f450aa1730484cff00202\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2a86a0b3834e415fbb753c3701184fb0 : SUCCESS in 16m 18s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/178612e0e8b34ad4bb08c061edf8ed65 : SUCCESS in 19m 24s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/73f7f93a507d4f06aba8024e1277f442 : SUCCESS in 20m 08s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/12bbcf8177954036904e12c884f0b6b4 : SUCCESS in 18m 34s","accounts_in_message":[],"_revision_number":8},{"id":"6d0f017a83122bb2e9f426f13bc76ba831120b3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 15:13:40.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/585b5b53d98848b9a43705ab9fc3cc17\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fd6a5ca3fde44dd8667a00a15f24ad8 : SUCCESS in 8m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/1a52d485fb764c72838502561a828ab7 : SUCCESS in 1h 05m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4d8ac928595e480da197753909131e3c : SUCCESS in 54m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9cda4724d61b4cfea9367dedf0ff07cc : SUCCESS in 1h 08m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7ec59055aab241db96a5ab370468772f : SUCCESS in 16m 47s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b9904530751c44f8a51013eec56c8191 : SUCCESS in 17m 20s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/65f140c496b349af8d316f056d19767d : SUCCESS in 17m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e3dbb30d974b4c43adceb3daadc2cd19 : SUCCESS in 9m 13s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c42791ffdb7a4ac7ae2558bdc850fd4e : SUCCESS in 4m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5379adc63b2243bea8e623be64eee9e8 : SUCCESS in 18m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a425e36f3b1146478a1fc92d0a341743 : SUCCESS in 19m 17s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/959755154c1a46d7b98b7759c5ee33f7 : SUCCESS in 18m 47s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/9933c6d966fc4b429ffd7e5333dc1a69 : SUCCESS in 16m 43s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4a103cdfef2d452ba92241fd0101388b : SUCCESS in 24m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d780d233478a4aac98a329662e0a464f : SUCCESS in 17m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/42276acf67524cff80692371cfdb309d : SUCCESS in 19m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/96076d3be0ff490586dff88c5c3c29ac : SUCCESS in 20m 07s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4005522ef13c469eb0f636bb287a4727 : SUCCESS in 18m 23s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ce096b531adb499c8285f02d8646a921 : SUCCESS in 15m 54s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ac0440dda29b47a78942ef12dde68f9b : SUCCESS in 16m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3cab5b4e803847a08d58596705643095 : SUCCESS in 6m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a38b566b6e6040278c1143dca5dc84f5 : FAILURE in 4m 23s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b0145818a38047e59a3f48527640b698 : SUCCESS in 6m 41s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2ffa475c25ca45c7872f22928d8fb6eb : SUCCESS in 58m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7ce64081545441aea5794b1166f6ef6b : SUCCESS in 54m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4784f6e0a5ba4beea8f8a30a3b192a6e : SUCCESS in 51m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b3adbab49dc04ec3a8c591153bbfe039 : SUCCESS in 21m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41968bca336f4a6a9e2d0ef9ce449b8a : SUCCESS in 4m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/aa8040935c0a4102824a9f9b3a23987a : SUCCESS in 16m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/304fafdd1d2b4bf086f813915bf32b8f : SUCCESS in 47m 10s","accounts_in_message":[],"_revision_number":8},{"id":"df0079eb2a751842814169ef1a33b449bb56c655","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-06-28 13:58:34.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"6102e841ac5e4553cd82284e91a78246596fc46f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-28 14:19:43.000000000","message":"Patch Set 9:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5f6a338fffe4f37a6716f78cb29a62d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e5e966a44baf4eb8b304b7853c74d315 : FAILURE in 3m 47s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dd6f3040926747e188544f0250865c83 : FAILURE in 19m 55s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/0cc1162523f0444b9fc07ae0ed7025ed : FAILURE in 3m 29s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6fbc39029d4b4c2cbcb76463450f1d9b : FAILURE in 3m 34s","accounts_in_message":[],"_revision_number":9},{"id":"50a34c52304efcd3d42515101a0cd700f4f4a149","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-28 15:12:08.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/620bb7e073034fa79a52ee37586d8b43\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74f822afd83f449180118e3de46ab134 : SUCCESS in 8m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/f47703a5089343868cfc723c7cc8ce3c : SUCCESS in 1h 02m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2122677baa6a4bb88131fd3482eb96f4 : SUCCESS in 46m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c386f8d76b8243c0afa94f0d462c119b : SUCCESS in 1h 05m 08s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/73c6e584551340c2b0ad99732acdf5a7 : SUCCESS in 21m 07s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/51607351c7984f7db0c67022c12decde : SUCCESS in 22m 38s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a0efce1808bd4107b992a61453c5d8bb : SUCCESS in 20m 38s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6c80f16b8f3c48459a8bf3403d34b4d7 : FAILURE in 9m 25s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5df66c78483a47d28e5ca56e7adf8f8e : SUCCESS in 4m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c850765353be42b5af46e8c14bee6c3c : FAILURE in 22m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8cc515afc5d44fc391c825d2cbb12afd : FAILURE in 17m 08s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c448e3cb8e1c4cce9fffefe0e2dc1726 : FAILURE in 18m 56s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/6e084cef08524fb798b078b98d66a5ef : FAILURE in 17m 47s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2e0a2a359f614803859f0d77ffeb7c7c : FAILURE in 27m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ce939c52f076439e8ddf5d2300ba726a : SUCCESS in 18m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a46591fd3ae7486eb1da098ef0de2c34 : SUCCESS in 18m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cc0c8cb158a0494db0436ec043247a9d : SUCCESS in 18m 44s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/30b649ac90044ad9a19ca3bda9dde0e7 : SUCCESS in 16m 13s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/79b620a8597b439f8dd5fea3ebadd900 : SUCCESS in 19m 19s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/cb0af361ad154589a597d44e6e42168e : SUCCESS in 18m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ab2ff6b97efc4e1b8b07a60b01b772a6 : SUCCESS in 6m 11s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3ab343ad78a2493e9e09c10f0e6f568f : SUCCESS in 13m 49s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a8361f3a8f7347439c6d9f4a76637060 : SUCCESS in 6m 41s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/98137dcbc25e4133b08cecb8acc489f0 : SUCCESS in 59m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/295c11b81cce4ffd91a498db6c628038 : SUCCESS in 41m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c2526d9cba174508a9b69a3a069abe53 : SUCCESS in 39m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/458a5534d7ef4c96bf67104b737215f2 : FAILURE in 17m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c9ac67866314466915472f382156289 : SUCCESS in 4m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dd696ff8537240658e33bac368629e4d : SUCCESS in 17m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dbf3aa4303d14d1aaeda0ec54dca3a73 : SUCCESS in 54m 00s","accounts_in_message":[],"_revision_number":9},{"id":"02f0379d164db5ead56340c6ed0416e1b9861952","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-28 18:38:52.000000000","message":"Patch Set 9:\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"752fd6b06fcffda0452a9b881adef8277d80bac1","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-07-02 15:59:30.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"3051efb15f7672583f2e13aca4a23bc9635ac376","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-07-05 19:56:34.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"f1680b05c3d70893fb071fb032424a35803da525","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 21:03:20.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68f509523db24f69b87ac845068395ba\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0326daa42e8e46b383e54adabf5523b9 : SUCCESS in 9m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/b16d8ea879374f1a92d6b33d5fe05aef : SUCCESS in 58m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/da57ef591f524e528b67f76947d30bb8 : SUCCESS in 46m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/73f962e978e34eb3b4ee05b9911c8af6 : SUCCESS in 57m 27s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/93a53c4478d945d69f85c7dee57cc791 : SUCCESS in 21m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/61aca624e05b400c9cbafb938b33ec61 : SUCCESS in 23m 38s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0ab690d8170641909da6472d8bfa232d : SUCCESS in 22m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5c4dc198ed9041dca695e59980f765b3 : SUCCESS in 4m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6ca0255ad082443893d11acf3eeb8602 : SUCCESS in 19m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d806a343cfc94401bcad57153db57537 : SUCCESS in 18m 39s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/69de25fc91eb46c0b31868bcdae18cf1 : SUCCESS in 17m 49s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/f83161aafa03421b8b4b3280a9a76cb0 : SUCCESS in 18m 40s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d68f2c668a024a94ac51ed08ebb1c882 : SUCCESS in 24m 48s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b03ca4fbbc674c0a9aefb0bee81029a6 : SUCCESS in 19m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c710907b841e4516b2ba38e30f57ee4b : SUCCESS in 17m 11s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/270381453e1246ccb2992b14d92f0166 : SUCCESS in 19m 11s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8b03199aadf54568aac1b5c629e46453 : SUCCESS in 16m 55s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/45a55867df374c19a7cc39b173f480ee : SUCCESS in 19m 53s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6767acca7eec4968bef4b6dc2a66a2aa : SUCCESS in 17m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/45a2709064f64348b2e97a3f942392c5 : SUCCESS in 6m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/aed489d668e243f392a71b76662fb747 : SUCCESS in 15m 19s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c20d0e4eea3e4d77a3704927471cd758 : SUCCESS in 9m 01s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/cc1cfe1a0faa4e938629a894e6917dbd : SUCCESS in 1h 06m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4c57869328d648979814156a8d6b9803 : SUCCESS in 51m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bee25d7ee5c04a88a2e0f2100244cb06 : SUCCESS in 55m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5174fb5be20e4a1d8308daca49f67ff0 : SUCCESS in 15m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/588a7eff5d8143898f4166c72e7c5dcd : SUCCESS in 5m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/281309c0af174d32857d15debc300cb4 : SUCCESS in 18m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eceb0f2169a3480b8eff0be0dc730573 : SUCCESS in 56m 30s","accounts_in_message":[],"_revision_number":10},{"id":"e2e6b573ba7b73e1caa8c3f134385c2386e50ad7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 21:29:47.000000000","message":"Patch Set 10:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8fe8897d70e9476baaecf08af7581d6c\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/baae8a3f94954625b12649c1a7c2c69d : SUCCESS in 1h 05m 55s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/847cc5d6297c49a38d2fd4b7209ed6bf : TIMED_OUT in 1h 30m 58s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a9e341424be54d469efc7e29c1e21d83 : FAILURE in 3m 30s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1503bf1d532a477c91c84ed072415e36 : SUCCESS in 1h 24m 09s","accounts_in_message":[],"_revision_number":10},{"id":"319d6b5dae4a4a045328f32f3636873a5483d616","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-07-11 13:58:38.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"b475d7f5a955a7d6813db29569699e718cca490f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 15:07:28.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad6283fd715042279fb404ba7e5fad68\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d0340b9199f4fbfb9c88530a524cd03 : SUCCESS in 8m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/79bba974576c4869931a3641878ba591 : SUCCESS in 50m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e1a2dc7a4bb3443a84dd628b43a081ca : SUCCESS in 46m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4034d8a43e0c4ef89a5779cbc05bae97 : SUCCESS in 1h 01m 26s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b9a7a9991900422c8c4c44b4d38dda9c : SUCCESS in 18m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e38e5f33835a4c4fbfeb7a2eed207510 : SUCCESS in 20m 14s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fa99b6e5a5a44f09828f73c688a3b53b : SUCCESS in 20m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cd3198066b7544bdac5d259654f3cdd9 : SUCCESS in 5m 17s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/16aba709a74840a09ecc98cbd3f82ce9 : SUCCESS in 21m 28s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8de0fbec296743d3b4793ddb850e8dbf : SUCCESS in 19m 47s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c8cc7b789a304b92b3ea6dab5cc7747d : SUCCESS in 21m 45s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/22954fcc3fbd4c08ae2eef5fb0189665 : SUCCESS in 21m 57s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0e35ba3a685644acb06efc722568bab5 : SUCCESS in 20m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0fb4401436b54fc7af482279d2b636c6 : SUCCESS in 20m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/962d2c82503942dd9bf926964e0ed8c6 : SUCCESS in 22m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e47ff91796cf482f9b8634e1bd4958bb : SUCCESS in 17m 37s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/fc5461ffe1354fde9091ea93ff2852d7 : SUCCESS in 18m 41s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1ac52004fcb7475992d757d6ca952e0e : SUCCESS in 18m 31s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/85eab4988fa64b4ea5eec0f25a4c22bb : SUCCESS in 18m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/dd4f920dc8d5408fbd8bf7bdb89db996 : SUCCESS in 5m 30s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5f33fe99a2e54bb6ad117cdea673e764 : SUCCESS in 13m 32s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2ce86e36f4de4650acfc62e74d17c0c7 : SUCCESS in 10m 22s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1fff7299d6b04d3dbdb664d5eb14b508 : SUCCESS in 52m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9e2e482da1cb4a0d94cb75afca85002f : SUCCESS in 54m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4afcfacbde45408c85906efcbe569ff4 : SUCCESS in 37m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d3578c9f053640ccbccdf6c30dba7a03 : SUCCESS in 20m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/276274a09bb2404caf384d651930a4a3 : SUCCESS in 4m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c2e37fb7571a48cc87c4141a4a973de1 : SUCCESS in 19m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/912397d3edb1479ba95e9949b0a2966f : SUCCESS in 54m 43s","accounts_in_message":[],"_revision_number":11},{"id":"7b8af747fb694dd16833c3bab3cf98e6237cb9a0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 15:31:48.000000000","message":"Patch Set 11:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dce55d05404c4a2199ebf423bdc99f92\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8b0a0d46d500415a8509227bc0c0b542 : FAILURE in 2m 56s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ace2ab99ac0f4323a03d8d1990b28641 : FAILURE in 2m 43s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ea45462e9ff44c1d94b2f50640aa9dd0 : SUCCESS in 1h 07m 33s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/c70efc86519043f191b417be5e09100a : TIMED_OUT in 1h 30m 58s","accounts_in_message":[],"_revision_number":11},{"id":"480b0e9953a3fb9943c36a5a8397cd4d188e2e75","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-07-11 16:31:33.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"e83157b224d63c5f89cf0d68cf38ca9642bd1b24","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-07-12 12:30:31.000000000","message":"Patch Set 11: Code-Review-1\n\n(17 comments)","accounts_in_message":[],"_revision_number":11},{"id":"6d641a47bda70b98cf7cdf32bf9796bd19fe84b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-07-12 21:56:34.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"0a0f1a0c6d3be07d3199ab7e951cb2326a8c81dd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 23:15:21.000000000","message":"Patch Set 12:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00ca236ad14f4bcda3a066a3a5112232\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8797a7f71ac44521a771d918b77356b6 : SUCCESS in 1h 16m 52s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c9cc31cac6264fdd8dd4a3f9731a00a0 : SUCCESS in 49m 17s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/834a251897e142dfb0633435380f1f1a : SUCCESS in 21m 41s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/18cbe094b1844a97ab6a19cbd6fd485b : FAILURE in 1h 00m 25s","accounts_in_message":[],"_revision_number":12},{"id":"275308354068f9ce62fac5afe08f6446a2745589","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 23:16:31.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b98c9221c404a438b8c3a4ce9fda680\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecc7d093445440dfbd43adf6de7ad73a : SUCCESS in 8m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/8f59f20b51d34cf78419d0173263c0e8 : SUCCESS in 1h 10m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/afd13926dc9741efaa58cef0048f813a : SUCCESS in 50m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c0fe7495b0974aca88616ae1089acf84 : SUCCESS in 58m 43s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4b519e1e16a64a84a80b76336e1283fc : SUCCESS in 21m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/55e01042b6914918a217af11f007daa0 : SUCCESS in 19m 02s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2440c7651d2940db8d8b58278dc163f8 : SUCCESS in 20m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/98f90713381941aebf5913ea4bc499f1 : SUCCESS in 5m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a551ce9c38df4c73b974bcf1995a3232 : SUCCESS in 21m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0b7ba4feed074693b8a836085544033e : SUCCESS in 18m 17s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/372a7773152744f5a1e71b915a9f76c5 : SUCCESS in 19m 38s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/14d99b08e2274d5ba3cb1391c3dfeb35 : SUCCESS in 26m 25s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4c3cd162fdab45d58bb072e48c8eecae : SUCCESS in 19m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3c85289849fb404bad8acc8ac99b2eb2 : SUCCESS in 20m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ecd8f9b266db4d3da9d7105a06acbca3 : SUCCESS in 16m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/43f065ed538d45c5bbd7e17a8ce474cc : SUCCESS in 20m 27s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/609c08b11a3d43a28fabfb3f53a5f736 : SUCCESS in 17m 11s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d361a956f46845379f01b58e64f33719 : SUCCESS in 16m 49s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c35fa44bcfb640bb8f5f844b14a401e8 : SUCCESS in 16m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/faf326211a48414a80f29437699a700b : SUCCESS in 6m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2787d42e55f5439690727584519907d1 : SUCCESS in 14m 20s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9add4b86ef7b4efc9ec2329f6027b033 : SUCCESS in 6m 41s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/403439db814349f8a785ce18f33a5c8f : SUCCESS in 55m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2007860b83d54facb035dfb2b8ac1e9e : SUCCESS in 45m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fee605df73c548168d2b9b3608338abe : SUCCESS in 34m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/21b7634b1ebe49cd865fe5d2eb885e02 : SUCCESS in 18m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aad041dfea554d5caa47d0fbefb5d674 : SUCCESS in 3m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d5aa7a77cb324dbdb70b94448fdd2116 : SUCCESS in 18m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9fec583753954e59879f4efe5bb87974 : SUCCESS in 49m 53s","accounts_in_message":[],"_revision_number":12},{"id":"70245c8b036398f96570bcd51c915adf29ecab98","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-07-15 17:04:00.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"c4fbc963e1efb9c04fade1d7c938eb45a30d31c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-07-15 19:56:34.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"eb7cae686b97f40ec171ad73982620a1a92c5a3c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-15 20:19:27.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8afada2c17134d7c872d0259320e1ea5\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/606228c8ac554ca2845d3487058ea8ca : SUCCESS in 18m 27s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/343eb20950c5479ea7d23a6b321fa04b : SUCCESS in 18m 36s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4055fde46ec04c3aae9da8c574af200c : SUCCESS in 21m 55s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/866eb7a24f4b4f3d9f93157dd3c6189e : SUCCESS in 18m 22s","accounts_in_message":[],"_revision_number":13},{"id":"97524681210f5023dd49fc08dc01e60bb6d1228e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-15 21:00:07.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7277fd8055594d6ea576540a18e7319e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb2a10c1fdba498ab2edf98c1e8b89be : SUCCESS in 10m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/40bf328698244ceaa56bb62a8e2c51eb : SUCCESS in 1h 02m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/91e4223044e84075afc78d60d784db19 : SUCCESS in 46m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/23a1a4eeec78427ba895cada3fb74591 : SUCCESS in 1h 00m 43s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6b9316d73f5b42c09541e154263fbd4a : SUCCESS in 23m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/15269c7a553243188306020267cb92b6 : SUCCESS in 23m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e4bea2c17e1c4b18a14cadeb048ad0ba : SUCCESS in 20m 52s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b814ef58f0514447936ac5ba8db6e688 : SUCCESS in 6m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f3ec8fa28b1f4bd3b41e1b9fe6b6e816 : SUCCESS in 19m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9f862623c7e54550a182083dd95a357d : SUCCESS in 24m 18s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/03597a8019f74496856776b93bb27dcb : SUCCESS in 22m 29s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/a414598d57f54a4098f5b14e54d847c7 : SUCCESS in 21m 02s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/db0ea4f559d740f5a20841b8f52eedb3 : SUCCESS in 29m 13s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/44f8e6b5af16484dad5db44585a1c1d7 : SUCCESS in 18m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b5d266eb321e4534a494bee36e2dcf37 : SUCCESS in 18m 51s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f0575cd836a54b99b95f401bd9107f4d : SUCCESS in 18m 44s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/90ed8b593b0246e2ae16802054f0dbe7 : SUCCESS in 16m 27s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/588d77c9fbfe4fb5b15b63d9fe614d51 : SUCCESS in 17m 04s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5cde84ecef43447991ff9d4b8a31fe7b : SUCCESS in 18m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f3a5863402924eedba27cd046e804507 : SUCCESS in 6m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/52800af404e940ec8534ea298cf95b68 : SUCCESS in 14m 58s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/02f958dfc30348fca538565d205c7ee6 : SUCCESS in 7m 06s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b5893bc157324ae3b98f70dfe0ea57b7 : SUCCESS in 1h 01m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7f6582f555584bb299d9f3d15b25b5e1 : SUCCESS in 54m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/986450ab78054ced88cbbda58434200e : SUCCESS in 52m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f60f1f1326844fa58a14adf300368915 : SUCCESS in 21m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a99c9a788ce9432ca6bbbde22c56d527 : SUCCESS in 5m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b1dd368778a24fa88127f5d3b4c16294 : SUCCESS in 25m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bd6c3fdb4b2e406ca217dbbfa57d5146 : SUCCESS in 50m 52s","accounts_in_message":[],"_revision_number":13},{"id":"9857eb07cf03d0396d4515685d999e8638af8140","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-07-16 01:29:41.000000000","message":"Patch Set 13:\n\n(15 comments)","accounts_in_message":[],"_revision_number":13},{"id":"92e3069c7c89a0a6b08cc4f31513aeb751884e9b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-07-17 22:00:33.000000000","message":"Patch Set 13:\n\n(4 comments)","accounts_in_message":[],"_revision_number":13},{"id":"64413a17668c1140ff8dc96ddb97779ba075215d","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-07-25 15:30:48.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"99a006c593075b7070bedef098507d8aafda2cd8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-07-29 21:15:38.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"dc415891036fc36fe659e9951d135cb9e9151b8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-29 21:15:56.000000000","message":"Patch Set 14: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/swift for 917711,14","accounts_in_message":[],"_revision_number":14},{"id":"486b434adae87973abd13cdf97cf72f167f48d03","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-29 21:15:57.000000000","message":"Patch Set 14:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/swift for 917711,14","accounts_in_message":[],"_revision_number":14},{"id":"0c3ef67450b908c1235db62ae87e1c918081487b","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-07-30 16:33:18.000000000","message":"Patch Set 14:\n\n(21 comments)","accounts_in_message":[],"_revision_number":14},{"id":"f844957f9eb0a26555ec167c2ae8d1504500974b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-07-30 21:15:35.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"53a0b714dfe8cd8ccda5a47fee5bb38bd0ef5a0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 22:22:58.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a2cfe0cf0724b608d1ce537bf438baf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e0966041b994f9d82519320e667c344 : SUCCESS in 8m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/d9ab14a5daab4bee867091e1d0833813 : SUCCESS in 54m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/57582c176186460daffe91d42047f408 : SUCCESS in 38m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/12c9dbebe83e439ab9aa706deef5dc3f : SUCCESS in 54m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/042a5614084c482f932b3e5690aa88a9 : SUCCESS in 31m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/00a6563b20644374942136866b3357f1 : SUCCESS in 19m 56s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/09509d1d50724da0a8494dce948d6d96 : SUCCESS in 20m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/084d4cf7550a44a4b5ec0c2cc861803c : SUCCESS in 5m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cc135c39462c43aabc23aa0d3be1b046 : SUCCESS in 20m 22s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7400d969d55440cf95d1ac81e2d2c645 : SUCCESS in 15m 29s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/47571a1d83a744a7a402b64d1107a02d : SUCCESS in 30m 58s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/a3c2aa257263471da4f5c7f4333bb963 : SUCCESS in 21m 38s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d11306a8f5434d51b9356fa3cdf4f928 : SUCCESS in 27m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/950fc2e3764a497a9ddef2bb1f8786d7 : SUCCESS in 19m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fabdf03bc9574c0ba215cb6601a49501 : SUCCESS in 18m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/187b8722cf7b49db9c6f36889d90a5ec : SUCCESS in 31m 19s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/35840e3c427048e1a456a98de8677887 : SUCCESS in 16m 57s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/77321c126cac4d378eb633fbe1013427 : SUCCESS in 16m 20s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e4bea9ad23724fa6b9b6e78039073bf7 : SUCCESS in 16m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/06c56268dff8400aa79678ca83af0b4c : SUCCESS in 5m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/563f57507c60437ebfdafad4084d6464 : SUCCESS in 20m 31s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9843b0c252494cbda62367bfbc93e3f6 : SUCCESS in 8m 59s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7dae04bf8d544e9cb7d5fd2fa19b1880 : SUCCESS in 58m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/eda79cdffb4f460aaf427cad185bd036 : SUCCESS in 54m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/49af85cfb7d24029b8ecce8cbe6999df : SUCCESS in 47m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f1f9964dd7ef4d13861ef97901217116 : SUCCESS in 18m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee1bbb59491c4c3184cd4f1c73c933d9 : SUCCESS in 5m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5c367be8c16f44559d0c0296ad9fb584 : SUCCESS in 18m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ae26a9e7c31f4ab3a712d2a9383aff44 : SUCCESS in 57m 01s","accounts_in_message":[],"_revision_number":15},{"id":"bce76d79931668c47912fde5d408571a7c0c9f3c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-30 22:48:32.000000000","message":"Patch Set 15:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13973beab03d4972b51aa7e73d9ffed8\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d3f611083b36450498546e9fc7665bb0 : TIMED_OUT in 1h 30m 56s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d32a09217c0e4906a47b663d2804ead6 : SUCCESS in 1h 08m 17s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/090bca075bf047c3a195b9279515f237 : TIMED_OUT in 1h 30m 55s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e524f1f1efac4baf998214f7de5d058c : FAILURE in 1h 27m 40s","accounts_in_message":[],"_revision_number":15},{"id":"ac22598fb7704230bdf2b9cfb889f98816cf6e6d","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-07-31 19:19:12.000000000","message":"Patch Set 15: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"332d0332c1cd422f981bbe4470303b2dd99fb8d8","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2024-07-31 22:04:54.000000000","message":"Patch Set 15: Code-Review+2","accounts_in_message":[],"_revision_number":15},{"id":"86f4d729dc6f6d30de26e45a47d17bd8a265688e","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"},"date":"2024-07-31 22:05:17.000000000","message":"Removed Code-Review+2 by \u003cGERRIT_ACCOUNT_9625\u003e\n","accounts_in_message":[{"_account_id":9625,"name":"Thiago da Silva","email":"thiagodasilva@gmail.com","username":"thiago"}],"_revision_number":15},{"id":"c09e45471f068a768cc1f8d30822b386bae9ec1d","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-08-02 16:26:58.000000000","message":"Patch Set 15:\n\n(3 comments)","accounts_in_message":[],"_revision_number":15},{"id":"fcf84538a4da55662b92e944bf03a7a40d49a659","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-08-02 16:46:16.000000000","message":"Patch Set 15: Code-Review+1\n\n(16 comments)","accounts_in_message":[],"_revision_number":15},{"id":"8e5a0120d55b5cafabeda3bd9016ddcf841eaabd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-08-02 22:56:05.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"5f6de67d60dec76c387db82c483924eb1536366a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-03 00:05:53.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5dd5faeef444d4abba7cab846708031\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82b0385e1cb74c3c9c1513d273581e9d : SUCCESS in 7m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/2cd8a378bed94d818418b47c4b9767d3 : SUCCESS in 59m 40s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/726d02b2fa5743c9ac268861122f8552 : SUCCESS in 33m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/76f89537af6548dc9f633fa52778a76e : SUCCESS in 40m 02s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cf1e4d19e8b540daa3416a00c17e185f : SUCCESS in 19m 21s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d30f818918304040a890dbc0b6c5b6ba : SUCCESS in 19m 32s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cd61147f595f421892815eaec0064395 : SUCCESS in 20m 33s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/33247a2bac8144d28988d46dfc823a61 : SUCCESS in 5m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0fecd14853104038adeae4b2624320e3 : SUCCESS in 20m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e61e1dc21c7f40c88b98b9f7723f5f45 : SUCCESS in 20m 42s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/14136ff41ac440a9ad756a447d4d0f8a : SUCCESS in 21m 21s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/da8869a6251145f8b58eb1808e29115e : SUCCESS in 26m 51s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9cc342498daf496db9a0e8a1a18285c2 : SUCCESS in 27m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/eaf72b9010024dac99aa2dd151eb4717 : SUCCESS in 20m 24s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/dcc4a083d0bf4710bd38b72b4df21ba9 : SUCCESS in 17m 55s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ae00d0d4f4404c23a7b7eac217bebff7 : SUCCESS in 15m 50s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f9b643510894447498fc04ac38fd949d : SUCCESS in 16m 17s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2372108c39884653a3623a8d8cf5a510 : SUCCESS in 14m 18s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/968c5cef46f64626833ee7404118c165 : SUCCESS in 12m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cc674e709562430f8814087bcfeeb5aa : SUCCESS in 5m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b0a0f6a25d7d464984d7c72cccb7df7a : SUCCESS in 11m 53s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/654a40b78b16499788a5efbb0fb854e9 : SUCCESS in 6m 13s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/43d420c346d34505adecd04f65b72df1 : SUCCESS in 41m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/925cba63ac4b485b8becfd5eaa5461b7 : SUCCESS in 33m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/27dd34bb414d4a4cbd8e6f737570175b : SUCCESS in 53m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5d9f013040864875a04d7002446c17f5 : SUCCESS in 15m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/856469c27f2049dd8371cce1c5943f4d : SUCCESS in 4m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1110516ec68c4177894bbb28529e3040 : SUCCESS in 14m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/881032250f604853a5b7826563411fea : SUCCESS in 59m 23s","accounts_in_message":[],"_revision_number":16},{"id":"a3f967e404a293716bff99b9a4fea3d1fc28976a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-03 00:28:53.000000000","message":"Patch Set 16:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a893686b179047fe90129577bbcc6aff\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d683e3c0bf014cf1b3761da10fe15330 : SUCCESS in 1h 11m 57s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ddf4256070884edaafccbe3dc152740e : TIMED_OUT in 1h 30m 55s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f6a78e1086f04e46b8d863afc4534024 : SUCCESS in 1h 30m 43s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/fa546aec11bf42b9a24bc82cc2140f7d : SUCCESS in 1h 16m 39s","accounts_in_message":[],"_revision_number":16},{"id":"a4fcb79db3ef7a54eb2780eb9beeb0063e3c2446","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-08-05 12:53:33.000000000","message":"Patch Set 16: Code-Review+1","accounts_in_message":[],"_revision_number":16},{"id":"c54459095c2016952fabc832d4b2a304ffbf80ea","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-08-05 15:49:18.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"94d9034b38adb04db0788d061e5a3148171559b2","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-08-06 15:13:58.000000000","message":"Patch Set 16:\n\n(4 comments)","accounts_in_message":[],"_revision_number":16},{"id":"e07190b79cdd40f0ea3d7b227dd74a463364144e","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-08-07 04:58:31.000000000","message":"Patch Set 16:\n\n(6 comments)","accounts_in_message":[],"_revision_number":16},{"id":"1c1dd956c48f04b58b582c1be443cc5b29b1b075","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-08-07 04:58:51.000000000","message":"Patch Set 16: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"11e031e3083b34d57dd868375b11091fd848803e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-08-16 20:08:04.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":17},{"id":"774c01da8decfbf0028cae92e31630388504bd2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-16 21:08:22.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07822abaf4c448e3ba1360375cc7c234\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3480f415b4a0453abd1119146acc5d0a : SUCCESS in 9m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/c26f2afca8ff429988e6c87a7d23ea3c : SUCCESS in 40m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c6830685c53746ed80a4ffc03e0cec48 : SUCCESS in 42m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/87904170d39947d09441a3a9798a8b8c : SUCCESS in 42m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/189448b32e854a2997c73f8fa8ea27d0 : SUCCESS in 20m 01s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b568e7570412439ba0f8ba10bac7e812 : SUCCESS in 19m 16s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ef5f4c9efb3d4e40b0113e05b50063da : SUCCESS in 19m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2eccc338aa584d5fb53ef9e89b17051d : SUCCESS in 5m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2d60e478c04d44c59ad12f620141f911 : SUCCESS in 22m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8c0d60683fa948acb2e8105ee9d7687f : SUCCESS in 18m 45s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d2b7d38d8cda482389f56e9156703f38 : SUCCESS in 18m 30s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/e0706887e4844cbbb8d6b40de9fd35cf : SUCCESS in 27m 03s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8b73449fc6004c0fa2b95d91830cb602 : SUCCESS in 20m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7667c3795ded4deca95a8b57de72965a : SUCCESS in 21m 26s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5b212918cad94f26a5d559924acdc9d8 : SUCCESS in 16m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2041a6cf8f8c474a9269d83e1d66af17 : SUCCESS in 18m 00s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6a62afc9bd3347cd9d9be0c5674af8ab : SUCCESS in 17m 15s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/23cc5df7213e49fe86b65d5b4edd6b79 : SUCCESS in 17m 06s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4ce726d4c7b54a0fab3a5e918398ddc7 : SUCCESS in 14m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/60529874f97b4e08ab16719f448b45fd : SUCCESS in 8m 31s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/394f12f13aa249f9963c0b3b2e4e3212 : SUCCESS in 13m 53s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b6966da98af04030b33d59af618180a4 : SUCCESS in 6m 19s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d0b76f01fce84d1e8b1b13544440b0ae : SUCCESS in 54m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/93fed878dcc1406d9875534235288abf : SUCCESS in 44m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f5c27961bfd74ceab9838556e3d7d9e2 : SUCCESS in 43m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ccb72500b17f4418845523fb53a849ba : SUCCESS in 20m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6b2fed1281c41d2bd9800aee43730b9 : SUCCESS in 4m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/46af0192e6ba43698b942b63ff58ae62 : SUCCESS in 21m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/692e45ea22214b599d1babf26e850afc : SUCCESS in 44m 53s","accounts_in_message":[],"_revision_number":17},{"id":"ebcbf609ae64ef7d0477fa42d18a41f0b58d19d5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-16 22:21:54.000000000","message":"Patch Set 17:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f19bfb9de76b4c91a611aec33a9edde0\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b9c02a47d48441b6a808068362fb9e81 : SUCCESS in 1h 16m 59s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bdfd676a443e4164a2c8cdba8b8d9965 : SUCCESS in 1h 25m 18s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/2b08056d18564e9e942d5ce58c1fa578 : TIMED_OUT in 1h 30m 54s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/36527364558d494ea384c032d5091505 : TIMED_OUT in 1h 30m 55s","accounts_in_message":[],"_revision_number":17},{"id":"e192495ca9ca149ad7f8e40182cd9669050b2351","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-09-05 06:11:12.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":18},{"id":"b08eb6bcfb84cfea9d4042802d6e727dadec4d89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-05 07:20:00.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/375f30406f9b44928abc855bff2dd710\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0bd1092cf7454f5e97bb03d7bc726525 : SUCCESS in 8m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/aea6497a330b497e842d708672607a5a : SUCCESS in 1h 03m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/793a59476bf24708bd961eda42e284a1 : SUCCESS in 46m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/35291968baea40c291b3896f7072b69d : SUCCESS in 59m 01s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/389ed596f84e443f9854fd95939419b6 : SUCCESS in 19m 57s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/74f6fc81690e4a17895d16f87a1be41a : SUCCESS in 18m 50s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/129d9b62ed1842de94d22bd26cda7765 : SUCCESS in 21m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/139043dd9211404c94548548dc93473e : SUCCESS in 5m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b27efa400a5d4eac8fd4495930e02b81 : SUCCESS in 18m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5ee9dda2dd2a44bcb86d038ab78b0818 : SUCCESS in 18m 18s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0b2885cbdac147cbaebe47afd3808e40 : SUCCESS in 17m 19s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/865a6360f8304910a53d4a36535eb653 : SUCCESS in 17m 44s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8331ffaa1c714ed0ab7af8de4eae4650 : SUCCESS in 33m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/36e0124a1db1405590f52415b0b84731 : SUCCESS in 18m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d98c1fa475124450b97d807818a0c577 : SUCCESS in 19m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6d8560a20d3f45b19e7848ca40a0b7ac : SUCCESS in 20m 02s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/293aa0df72c44992943a63e1972a3bac : SUCCESS in 17m 01s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/65c1d1c35e114ce6a241ee25df61d3a4 : SUCCESS in 13m 28s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/175c0562ff004550a49d5eb20e3eaf18 : SUCCESS in 17m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a862aa704c5043148e5aa08a3c31a35d : SUCCESS in 6m 42s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/406b57833c2542fbbc5e58be3910ddab : SUCCESS in 17m 05s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/673e219aebe54df381340b07f46a5898 : SUCCESS in 6m 22s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/cd350fbccf504a67a9863dfd26175dfc : SUCCESS in 49m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fadd43616c564fc68b567578b70dfa31 : SUCCESS in 47m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/04b756b2f6204425ab26a2753f2561f3 : SUCCESS in 49m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9150e86f3da24d42bb645b96eb46da4f : SUCCESS in 20m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9f2ea091b5a406b96e0260cb176cfb8 : SUCCESS in 5m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5432c33b1e7f4614882450cbae99c39f : SUCCESS in 16m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a8944d19c97a48a2b79f47a4cc16cf08 : SUCCESS in 56m 02s","accounts_in_message":[],"_revision_number":18},{"id":"fa99066f8761a37262a06cc0ce6b078d615c3cbb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-05 08:23:41.000000000","message":"Patch Set 18:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9603c1a9a0f487e80d6b44d857dd86c\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/20423972149a4e91be62119689c299bd : SUCCESS in 51m 13s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f30c7d518d7947ecb9b7afe88911fedd : TIMED_OUT in 1h 31m 13s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d910ed5d6ed44eb7b752af326e1ccfe2 : SUCCESS in 42m 06s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b49b9d2ef28441578983f75d7fd14005 : SUCCESS in 56m 10s","accounts_in_message":[],"_revision_number":18},{"id":"cce1117eadafb0309bc6d3ac7e8a42363d3fc5aa","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-09-09 13:07:35.000000000","message":"Patch Set 18:\n\n(6 comments)","accounts_in_message":[],"_revision_number":18},{"id":"adffbf00396c3a533e821f0f9df7bdcca62d9d97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-09-10 13:59:35.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":19},{"id":"7d03264e978d1ad27df3d0fff66bfb85cb986065","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 15:04:47.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3a9761439e64e2aa183149159a4c6db\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18063663c67e428886cdc7541a06cc1e : SUCCESS in 10m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/e802bb8e787d462dbba93106d2c93a77 : SUCCESS in 57m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4b688a14f6d34f9b94ad6fbb2f6726d3 : SUCCESS in 52m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3e932baafff94ae5835f92fd4da27b40 : SUCCESS in 58m 04s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ddc76b4672114e6ba3682730ad744c53 : SUCCESS in 26m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bd99ef56bf054d159d74f35d390ccca6 : SUCCESS in 20m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a2cbcaa6d0ac4bd9976bebd1351005cd : SUCCESS in 19m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1d68866aa7c44e0b99b526ebf36c880a : SUCCESS in 4m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6130d6fdbcdf4fb099b69c2105f7fccd : SUCCESS in 17m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/29dcdfe257ac4fd7ab24ab32322e3a42 : SUCCESS in 21m 49s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f8ded4ba4d9243d89b34d5d3dd7e54ac : SUCCESS in 17m 26s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/42e4a264f5fe46ad875dbc99935ccc21 : SUCCESS in 15m 39s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8ae1b332e68f4817aaa4f79f2577c73b : SUCCESS in 27m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5ea9873a4a51454f9bf65951036cfd68 : SUCCESS in 18m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cbf916becae4464ca526c36b0c8cfd12 : SUCCESS in 19m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3a5c802f70d84f71a180c70a82b37ab8 : SUCCESS in 20m 09s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/dd5c2d9ac5214b759f2309a16bcb553a : SUCCESS in 18m 36s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a8836d7e6b97478aa7ef9a719e1d2cb4 : SUCCESS in 18m 06s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d2827aa0b59a49949238925c1feb27aa : SUCCESS in 22m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ba5c986546884fe388097b136a7a86d6 : SUCCESS in 6m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/de630a33464e457585bc521e53990ef9 : SUCCESS in 14m 02s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/38a26962876d44a49214d42ea7ece9c8 : SUCCESS in 9m 10s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a9565f027b914413bf012d544e8161d4 : SUCCESS in 57m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e893715e195e41fa89e30f106070852d : SUCCESS in 46m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e6350dc348754279879218d5ae60661c : SUCCESS in 47m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/62f8a729fa4944f9b4d5181291e22593 : SUCCESS in 19m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5885d848e5634b3db31a729d774e70e6 : SUCCESS in 3m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/51c6f42f45b949eeb956042cb25ef947 : SUCCESS in 18m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c8047d910f024813895b936f0a270a9d : SUCCESS in 59m 22s","accounts_in_message":[],"_revision_number":19},{"id":"4243597b2cdcda6bbac6dec3a1fa9f0e10e55ba4","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-09-10 15:35:18.000000000","message":"Patch Set 19:\n\n(15 comments)","accounts_in_message":[],"_revision_number":19},{"id":"47955cbe4076011de0b58ae05ecf6fa0da9647d8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 16:47:24.000000000","message":"Patch Set 19:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/82bd585136534fbe9b71bdb9675d83a2\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3b5ffd1015104632a3902210f1922d8d : TIMED_OUT in 1h 30m 59s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/adfba4144af0464aa7622462b915cc53 : TIMED_OUT in 1h 31m 03s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/12ed97ffd33048758726bcff5d7e3821 : TIMED_OUT in 1h 31m 03s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6b57ba3bde394429944871bc890cf39a : TIMED_OUT in 1h 30m 56s","accounts_in_message":[],"_revision_number":19},{"id":"d4855ae9c8aae7dd8955d95d1ac3e3eb263980e9","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-09-10 16:51:26.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"106e5cde59e88fc91f45cb19782cee8d2806866e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 17:57:36.000000000","message":"Patch Set 19:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6f329315ea644af9d6cd22755a969b9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a8db4d002244141b5f75814b3aeeef8 : SUCCESS in 9m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/b35c120a447e4d30b62612ab1e09c4ea : SUCCESS in 53m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c281106da2364dc39dd81f08a35f6bd4 : SUCCESS in 52m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0d0e4f8af5dc49b3a04d7d35ed18f9a7 : SUCCESS in 59m 19s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c872eb4adf7840e284f5487266011940 : SUCCESS in 20m 20s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0db79bd59ad94d72a288fa658b06b52e : FAILURE in 6m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6acfbf0f392a41f58c371b78dcdadb64 : SUCCESS in 15m 38s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e906948876654a478ce5ea34c13c37d3 : SUCCESS in 6m 43s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f4d92c2c24204403bf1d04ae7f47054e : SUCCESS in 20m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8f6a3cc7e01b48ebb833dbf4eb7e962a : SUCCESS in 12m 33s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6b1ff625aff948d49bfb52018fefa299 : SUCCESS in 17m 31s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/dc69aec405e545f9aaee21fadf14d2ec : SUCCESS in 20m 40s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f368039206d24247a59102e0903e8e51 : SUCCESS in 26m 56s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c99fbba588d94bc298e2ce0e9d1a2517 : SUCCESS in 19m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e0efab1010be43f6840a0793c1e8af6c : SUCCESS in 20m 51s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1e5911882a6f4d7fb501fa0839c8f33f : SUCCESS in 24m 50s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/65e01f16eb254020b0ea8e2a5ad7c40c : SUCCESS in 17m 13s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9d017265a4b343e88abcb743f0e6bb8e : SUCCESS in 24m 19s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/df3d63ab74304a2a8b8f5cca66167d84 : SUCCESS in 19m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/114d75036a6b4728a83725c19fa6daf2 : SUCCESS in 4m 57s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e4c74b54f774452584d33f6855004ecf : SUCCESS in 20m 15s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1bbbc6b423064045a2aa71b9bb14f8ad : SUCCESS in 7m 12s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/73b8f4be0fd144d18bd5d6e5aee631c0 : SUCCESS in 44m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/db0eba8f4b344354a7324b00e656500e : SUCCESS in 38m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d7944e90eafe4701a3637cf3ecfd1021 : SUCCESS in 50m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0947e1ca5d2544949f112aac71077b45 : SUCCESS in 20m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/408d351170354a24b88fb112148c1708 : SUCCESS in 4m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a2f26a9ea75848a1922b1e34061f2732 : SUCCESS in 14m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b8819b3b7c2f49d1966a2ed63fd35712 : SUCCESS in 59m 53s","accounts_in_message":[],"_revision_number":19},{"id":"a6fb9b127a7be1d183f200bed3c5e6f2a2dad596","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 20:09:25.000000000","message":"Patch Set 19:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67b03812299d423eb162d8573c0922bb\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f7481244e08544b885a16dbd32f21589 : TIMED_OUT in 1h 30m 57s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dcd8c8ec117f4d42b1834d615f7abd28 : TIMED_OUT in 1h 31m 01s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/caa878184ce647bb87eb2c97af817d7e : TIMED_OUT in 1h 31m 04s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b576c812995b449c96af81544e52906e : TIMED_OUT in 1h 30m 57s","accounts_in_message":[],"_revision_number":19},{"id":"8878b01ed60ed2e08d69bc432e18c96684729417","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-10-02 13:59:57.000000000","message":"Uploaded patch set 20: Patch Set 19 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":20},{"id":"c57d7763e3b87a1c7302c19b33b62f9d5bb5822f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 15:14:12.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/174c2717ded9495690d9c2e7f1dd417c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5273d169ce4e46bfb015dad4120d5f87 : SUCCESS in 5m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/315e97d5431a43b99484c7a67864859a : SUCCESS in 1h 05m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/be48a63a880f4611b8e1c521112fb749 : SUCCESS in 37m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b12913412bc34b8396897f35a962a999 : SUCCESS in 55m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b0734bd23b554c5287b6e0538dd857ea : SUCCESS in 25m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/90dffe25a2554b9bb4dcdb243f0bcabe : SUCCESS in 19m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f0d92ab958db4b33a8e8564571c0d2cf : SUCCESS in 19m 05s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c3a9c80680764996a17de171f3789214 : SUCCESS in 26m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d87494c0d3ce47ec98397b94e8e0c66a : SUCCESS in 5m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/28ce7e0eae9f4eceac94610c6e7b19f7 : SUCCESS in 22m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/78a651e6bec6465a85c55f54bed3e003 : SUCCESS in 20m 31s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9192ae4f78cb46b4b7de87331552c25c : SUCCESS in 21m 14s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/8efd87ce1e704aafac126bfd889d4c1f : SUCCESS in 19m 45s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d5c20411738843b88e9894b6b9ae9ca7 : SUCCESS in 27m 36s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4fa69917a2454f49b8c3850060efa627 : SUCCESS in 15m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8fff30ae0e0c4ff5925ffd6cda8cecc7 : SUCCESS in 17m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9e402d79ebd84c22ab0671fc72c6cb4a : SUCCESS in 10m 36s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c72cbdbcab884f8f8926dd794b856b4b : SUCCESS in 16m 41s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8b1725e10fa84723a199252f3286e8e0 : SUCCESS in 8m 30s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b69717e2bb2449af86c8b17e2347bcc1 : SUCCESS in 16m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/19688c83cce24a02a0d65c02ebdba668 : SUCCESS in 8m 04s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/be025074eba944eab1e7db9f7d3283e9 : SUCCESS in 15m 06s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/818bfcd27eef4f3bba2bb447d0e43eb2 : SUCCESS in 4m 50s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ceaed3feca4b49c9931fac1295d428f0 : SUCCESS in 51m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/25dca5c51c5d4d5b9131f8c664dbaca0 : SUCCESS in 42m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bcbc4937c6944147bda9b9ffbeabec44 : SUCCESS in 51m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/94825ec265934f16a752460453bbcbb1 : SUCCESS in 18m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d924081a38fa40b689d1e7c28cf8fe1c : SUCCESS in 4m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b76ddfd686024800b53f84dc6c55778f : SUCCESS in 11m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/090546d3fd5a439d81496a50a916e100 : SUCCESS in 51m 20s","accounts_in_message":[],"_revision_number":20},{"id":"ab8070ec5f43aa77da6385b196a886671154e662","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-02 16:36:49.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be4cee4836be4a689a1d26317b476212\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/564cd1c87e77413a9c84799394ff160d : SUCCESS in 1h 35m 40s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/301095dee2c0473c9eca05f86475b6f1 : SUCCESS in 1h 31m 04s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d030a552fcfa43109f874670f9b072f1 : SUCCESS in 1h 19m 36s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/461c64ece6444183a51e9ebf3e09cbfb : SUCCESS in 1h 28m 34s","accounts_in_message":[],"_revision_number":20},{"id":"15d8a4452785a099f3abdd3ae8b11db52d11c44d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-10-18 16:59:37.000000000","message":"Uploaded patch set 21: Patch Set 20 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":21},{"id":"1400d297d071645407f7f75c029564639136e2fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 18:12:52.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4936208bc944a2cb2a869c3b2e934d2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ce3d8ea4efb417d91e781fc1b6a4bda : SUCCESS in 6m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/c046d7f67c2442f2a08f3fdcfd6bd3f2 : SUCCESS in 37m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c04c226652a44d179afa7ce4651f7394 : SUCCESS in 58m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/483cf9ddf57e4eb692a24655fdcef3c0 : SUCCESS in 47m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/97ef83dbf6154598a0944112511aaa3d : SUCCESS in 56m 26s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b1f51349c9c14154bcacb792a37fdf5c : SUCCESS in 16m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b61bd134816b49d3b09e793dbb909492 : SUCCESS in 14m 30s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/75916dbf7d34451d8dd341c6fb335822 : SUCCESS in 16m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d31013b2b38242d381ae056ffaa5ee32 : SUCCESS in 6m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/860e3efe5615411b880659caa9d6def2 : SUCCESS in 21m 02s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/481aac8772674f27b92513555c1ba7d9 : SUCCESS in 16m 12s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e8839d27092e46d1a97aafeec145fa6c : SUCCESS in 20m 00s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/c3a181c3caf64e17877a2c598d3f0c58 : SUCCESS in 20m 28s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2bce358716b44dc99bf1f12cbfd8ed9b : SUCCESS in 29m 29s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/bc6ec9bd8fe04015ac5b909a9f57ca5f : SUCCESS in 21m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1b39d49e661749dea6170c79dbdf6704 : SUCCESS in 18m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ca74cef0942543758e4cae8f2cf17d30 : SUCCESS in 17m 31s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/bcf8cf9f5cd74d129cb7ab4e8f660e75 : SUCCESS in 16m 53s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1f7159117a894421936ab34498ce7e03 : SUCCESS in 18m 16s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/89adf580966b4efebb0253f8bcc5c072 : SUCCESS in 9m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cbc17912204444a4a2919968b698591a : SUCCESS in 7m 45s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/467bf1beedb648349d329e0afd024a19 : SUCCESS in 10m 51s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3a9a054b82b244d584f063d0bb6e7945 : SUCCESS in 6m 22s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6447d0a561c145c69211779acb439f80 : SUCCESS in 49m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/09fe1ad8c45e4bf8b279e643863af8c9 : SUCCESS in 45m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4c797cd8eb2043cabca9add4013a9bf3 : SUCCESS in 50m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2ec045c7f87146f0a9a043c7ef4e6055 : SUCCESS in 18m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08746d492bdb4bfbab29da6c11ea1157 : SUCCESS in 4m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6604a43921e6461fafe484300204be7b : SUCCESS in 19m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/143c249977c24cef9125bf89629313be : SUCCESS in 51m 52s","accounts_in_message":[],"_revision_number":21},{"id":"46d1fdee3898c72e6b62a1a463a76b130a6bfedf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-18 18:30:21.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8376af37dea4359937b4ca73a381343\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d87d195a7e2843bab6ff7184f0d99017 : SUCCESS in 1h 28m 58s","accounts_in_message":[],"_revision_number":21},{"id":"b6810fdfe40a9e12c292aebf09c6f17540e653c6","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-27 16:45:25.000000000","message":"Patch Set 21:\n\n(3 comments)","accounts_in_message":[],"_revision_number":21},{"id":"dfb9111e540172a7ac37fb0bfe7c1639e41d23ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-29 14:31:33.000000000","message":"Uploaded patch set 22: Patch Set 21 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":22},{"id":"15c75990bec4ac8f66148ce965585f847765cc90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 15:40:59.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/accbe03567e84784bd04ad215e77a940\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e92d929ee7ab46e1ac89a92317177264 : SUCCESS in 11m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/a9255a78b5024c07acc1e945c6b9bef4 : SUCCESS in 58m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f62e3698af3842b2b7d269f3a99b975d : SUCCESS in 1h 00m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b73536613a864160b669703c95797722 : FAILURE in 11m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5751140912d445a09a77b14e5723c99c : SUCCESS in 1h 06m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/501d845563974ee7baf2ec34f99976ff : SUCCESS in 22m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9d9bec4f3c7948819beefb5d58ca379e : SUCCESS in 22m 50s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3386228a5dfd4a258959296423626dcb : SUCCESS in 27m 24s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7e8dd6f080a3400db52b8634943dc0b9 : SUCCESS in 5m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/09ac5c49ef574e5dabee7d66e0518119 : SUCCESS in 21m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8a923453553c473692491c15c7f9aa6a : SUCCESS in 20m 36s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/eaa2bed139954f6c983a5d992578bcd1 : SUCCESS in 10m 50s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b6dee1ca50a04fd8b2f552320b8cfec2 : SUCCESS in 34m 34s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/582b16208eee49d5a2fc62c723667e80 : SUCCESS in 17m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5e8de0ee37504fb98590e9fdf420360a : SUCCESS in 19m 26s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3769114b0298496185fe8bcde073cb52 : SUCCESS in 9m 56s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/7049e8f4e50c450f824220b84b5c09c4 : SUCCESS in 14m 39s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e21419414bcb490c9d0353c5b5fe5e43 : SUCCESS in 16m 05s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d0620baad9b34c13aad23bb7869b7656 : SUCCESS in 14m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e30433516c77498cb7f475b23c180737 : SUCCESS in 7m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dc6727fe0d6f41f69ca4ca5bad230398 : SUCCESS in 16m 37s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/144d7e99c2a44963ab5a103f55b7d9d7 : SUCCESS in 6m 30s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/cebf0a77e3604274a4b1805b6e149df0 : SUCCESS in 54m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/437886438046478792b25b5907eebce7 : SUCCESS in 44m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/70238e4787ed40c780a22625369d5f8e : SUCCESS in 39m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/98cc44a940e74b1fb4d82578615ff663 : SUCCESS in 17m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1fa71f2938c4adb824bbcdf43b0aad5 : SUCCESS in 6m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ca000c0707e34ac89c4d60a2bf0cad32 : SUCCESS in 23m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7a3e84b221404d93b1b27cb1c5a0cd93 : SUCCESS in 59m 03s","accounts_in_message":[],"_revision_number":22},{"id":"bf707cecce924a53e738fa4d81fe5f453b4792c8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 16:58:10.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8abb6d82893148bca18ab7e25ea530ae\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1021c1e217aa4e6082adeb36c9f45716 : SUCCESS in 2h 09m 40s","accounts_in_message":[],"_revision_number":22},{"id":"8e05848329fd9431660229949b652e73ee1b46b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-12-03 14:58:39.000000000","message":"Uploaded patch set 23.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":23},{"id":"8c10fd99a615be0a9d6da8b8f094d25bbd4aa5dc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 16:05:23.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ecef486b3cc648ac951becaa2384c51e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/940130ea0af043c6804ba00f3d7809c5 : SUCCESS in 1h 04m 33s","accounts_in_message":[],"_revision_number":23},{"id":"9c7afe8f10b95c3ab679592304bb385a27f3ed12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-03 16:11:32.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05897cf1741840c3987d824e5d60a9b3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2329d65a3eb449b94a547a7af0bd8d6 : SUCCESS in 9m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/b892562f420f41208c15a491d4915669 : SUCCESS in 1h 08m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8096210e10124a04a89545f7eeaa88a7 : SUCCESS in 32m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c0d2e90604a24cbe915fe60a073d922e : SUCCESS in 42m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/782d990a9c45470a90c3538c03107929 : SUCCESS in 1h 00m 34s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/71ce416c32d84406b02b7e60aed0c033 : SUCCESS in 20m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2943e0fb58024a89862c9f4a2a046bbe : SUCCESS in 20m 36s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6f92d56dcebd45f1b01e462c1e6f09ae : SUCCESS in 20m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/06eacb0f2ebf465bb6414d030105b934 : SUCCESS in 6m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/23788548bf1b4e62912c17545e5e56a7 : SUCCESS in 16m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/11a96ce283ec43ad872568cde09af95d : SUCCESS in 8m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d66a2c4d50614c3ab23117551da47243 : SUCCESS in 19m 33s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/00e3399f3c7d45eb81f274b486dde8c0 : SUCCESS in 21m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4294574446064dbcb7bd7a19fff27e96 : SUCCESS in 17m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d185091d076d40398199e8ba2703375e : SUCCESS in 17m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0ff5db50e64243a5b7f3894fa6142bb4 : SUCCESS in 9m 38s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e81d1f44cf8947af947ed279eed90c34 : SUCCESS in 15m 25s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9ec11f78dd3147f594e1b87d8360ab7e : SUCCESS in 14m 02s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d8a1c512e79647dca3af5456ecefe6e7 : SUCCESS in 18m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f0275904d0c940b2b4fea944a4f508aa : SUCCESS in 6m 18s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9e1f6c7936bf4850b7a7a52bb3f2caa6 : SUCCESS in 15m 07s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/aabdf05f288c40dca166d8c07a6e28eb : SUCCESS in 9m 27s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/15df100b8032480b94dbd2f69478581d : SUCCESS in 52m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1da6fc8e03fe4cc68c9a6cf023babf58 : SUCCESS in 43m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b9510f7d852f47d1b17c02b6173290f4 : SUCCESS in 37m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9e964f403a3b4b849312da45062a7904 : SUCCESS in 9m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e70f24cf06964e44a498bbc4dede5aef : SUCCESS in 5m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7e1fb77a17494a71b9aa33545b2a0a9d : SUCCESS in 20m 12s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aca4b9f322c64b4bac7f86c9dd86cb8e : SUCCESS in 38m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5f77526354b0404c8484ed00b7fcc15f : SUCCESS in 53m 45s","accounts_in_message":[],"_revision_number":23},{"id":"2e5a31c13547d34a711013fcaabb1a846054b90c","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-12-04 16:22:38.000000000","message":"Patch Set 23:\n\n(2 comments)","accounts_in_message":[],"_revision_number":23},{"id":"0c968afc7e2a4c2406cb7ecca93c83b07f3b116d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"real_author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-01-10 15:09:38.000000000","message":"Patch Set 24: Patch Set 23 was rebased on behalf of \u003cGERRIT_ACCOUNT_36606\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"}],"_revision_number":24},{"id":"6062de73c6fac2f61d0816708528ef4fe6c92a8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 16:35:08.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf55b385b77d4f958e098b18fede2c4e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7eff7c764ed146b0870178d0f611849e : SUCCESS in 8m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/e7392bde288240f38ccce24e11ff0c85 : SUCCESS in 1h 09m 54s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6046fe1765a244c2b504ef7a5b739dbd : SUCCESS in 1h 08m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c7337b1816da4111a0e93d232868dcba : SUCCESS in 41m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/44d7e584cfe04f36b919bf536c75718e : SUCCESS in 1h 03m 32s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ac8058936162487f9b9ac762e6c10209 : SUCCESS in 21m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7f844bd07b674b78bc0f13e9dea0cebe : SUCCESS in 20m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ebee4ad68da84d7ea0ed9cbf94da9983 : SUCCESS in 20m 24s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8e1177267cb3499993016ca81a94e07d : FAILURE in 5m 26s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/478888c52e6f4b939640ee426b5d929f : SUCCESS in 19m 12s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b63ec29b5caf441ebe5b4cb5554093c2 : SUCCESS in 19m 00s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7786d7966edd41a3a0627753e795337d : SUCCESS in 20m 18s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/daa36d72b72145bf8c3fbab15c853e2e : SUCCESS in 20m 06s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/184f6a37f7af474685a89882bf2a3ce2 : SUCCESS in 16m 36s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a02df6ec849845108ab8d5aba6c9de56 : SUCCESS in 16m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f23b9b3eb8aa4802a2cdecf2f8cd88ee : SUCCESS in 5m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e0fc90d1123445c88dccef4fb63aecb9 : SUCCESS in 17m 24s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/03f5a9c07d3648258589cd0d7aa62683 : SUCCESS in 8m 32s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4cc6f695400d4c10941d4f40b0688ba1 : SUCCESS in 59m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d2486d41457443a19c9f1c9c91391819 : SUCCESS in 40m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1c4c46c08aa041b8a9be8ca356766a50 : SUCCESS in 42m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8257bf36795341fe9b2458333030c52e : SUCCESS in 22m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b83caff0467e46f1bb4da8084b040e90 : SUCCESS in 4m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f1e58375d79e409b927fb8819a2f5c2f : SUCCESS in 18m 42s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3a5bfc01ffb343768bd57d971d31c4c3 : SUCCESS in 50m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9655566a65574fb4aac0b69858e32df0 : SUCCESS in 1h 07m 26s","accounts_in_message":[],"_revision_number":24},{"id":"1dc2c57debf9968deee4d849ad380e09b1807a68","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 19:21:01.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7780fd851ce457f8c0368227f44eb06\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b5b033db53814feea858c1214dc56ea2 : SUCCESS in 2h 52m 58s","accounts_in_message":[],"_revision_number":24},{"id":"7b781bebd7a86f87faea9de377844c87f5494b70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-16 12:05:53.000000000","message":"Uploaded patch set 25: Patch Set 24 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":25},{"id":"a8b51be680b07363fdc34123e1af64a76294ea84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 13:08:33.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56e2c8734e354507942ed62b104eb99f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba66d7fb44e14a81a5eecd17a3aba2ed : SUCCESS in 7m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/637f9f1ddbb043f2ab0162753a131be4 : SUCCESS in 49m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/79b5c0100b734e629eabd93addc010a6 : SUCCESS in 55m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c0cc4d2a551443e8aaad37c19cc78821 : SUCCESS in 45m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0d37fa22e5764fab8f2b613a441a9f50 : SUCCESS in 55m 04s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/509f9437c7fc4c96b82aa91075944ad7 : SUCCESS in 20m 42s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7d86c604153949ab84d181bc2fe80b21 : SUCCESS in 18m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0d3ebdf715974233947dbf725ca2af8f : SUCCESS in 21m 58s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c26d997633d24b8b807f397cf4a38ab6 : SUCCESS in 5m 46s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/43e167f6f6d7490aa8eb6742840e1d3a : SUCCESS in 17m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/49fd00bc976c4f98ab408c9bff81a241 : SUCCESS in 18m 14s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1f247c6433c9404db73380e1781ecf95 : SUCCESS in 25m 29s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/1bd4dd8469004685b5dc88b36eb6a9d7 : SUCCESS in 16m 17s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/54ec8a49db9d44be87a64d263c2de694 : SUCCESS in 13m 28s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ed71be34dc1e4f1f847b62ed342a117b : SUCCESS in 16m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/47fa1f0859044a388977b31a0b4078ae : SUCCESS in 5m 56s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c2373809a8dd4535a93fdbe093a83aaf : FAILURE in 17m 35s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2ea553b903c243e09b4d6ff939ff1c0f : SUCCESS in 6m 32s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/af8f0520ace649d3b8440d9ffc64c8ac : SUCCESS in 50m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3cf6467aae114f368084550caaa343d0 : SUCCESS in 50m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9ead22693a3d45d983a589a85276ea03 : SUCCESS in 48m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a894900d3c5f4af8b654efbc6e13d9e5 : SUCCESS in 22m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/694c9252eb5a4e5a91cc8dbf6c55b76f : SUCCESS in 4m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/524bccd11aee48f18fcad41c0e84ce48 : SUCCESS in 24m 24s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8fd47f3f6036434eb7cb7f2d4cb8b65c : SUCCESS in 41m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0a01817a048547dba3793c2f36432a98 : SUCCESS in 51m 23s","accounts_in_message":[],"_revision_number":25},{"id":"7a04d158d6b6fa3f2f4ee4891bd6f09dd2acfad6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 14:40:09.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d374752cdb6d4943be48b433a2adcde6\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f3674e99aa384f8b9bd283a016287e46 : SUCCESS in 2h 14m 11s","accounts_in_message":[],"_revision_number":25},{"id":"4f11de858149eed0740d1ff601b746eceaff3c44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-01-16 19:02:03.000000000","message":"Uploaded patch set 26: Patch Set 25 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":26},{"id":"5522b1fa8ba555ec88535088509d84ccb50bc359","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 20:16:56.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/916e0a720644478da8ec832bf1a4c6b2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98451bd97a8e4a99ba2582f3870d99c7 : FAILURE in 5m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/7254a8539e624963b064235e23c5678d : FAILURE in 53m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a1ce0180b97144dcb74f0f80c699f121 : FAILURE in 52m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c7fd891f4ef4420d964f480bf7105394 : FAILURE in 27m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/17b99ba8f9da41dcbf5c559a1d830617 : POST_FAILURE in 29m 11s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/10e3047f5c1547dc9e8b07574543b145 : FAILURE in 5m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c208ee2f1fb242f3b472ac46fb3d7c42 : FAILURE in 7m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f01a7223e3854e09bc40b9aed57624af : FAILURE in 6m 33s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a1da9f9b714348dc9eb47a1a8f870a3b : SUCCESS in 5m 29s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/afa8c1308673468c941ccadacc23c180 : FAILURE in 4m 22s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/91d82a7a5ba14753bd739027a43b2c62 : FAILURE in 5m 04s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/84d3746545b54851b7e45f78598af5b4 : FAILURE in 7m 42s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/88c48603a22f4f9bba0a9cbe3b143ef0 : FAILURE in 3m 40s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/820129fd0ae24e988a8e6abe7825765b : FAILURE in 3m 42s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/58091d04b6e1404e86b60350149ce8e5 : FAILURE in 4m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a7cf767fecdc4518bd36aa540d451daf : FAILURE in 6m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8aba2cedc5d346e18f55e5b28e35c77c : RETRY_LIMIT in 4m 28s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ce67e4d4c1144ff29510891287191c5f : RETRY_LIMIT in 5m 26s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/468cf9a2e7f44e948cd26ecc21d04e6b : FAILURE in 5m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8a25f10465d6431db322d825b7624ff8 : RETRY_LIMIT in 9m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4010359db72143139c1b390fd63e0a13 : RETRY_LIMIT in 9m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2c91ac1f378d432780559772b14fa0b3 : FAILURE in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d8fae6ea5984e98be75dda457d3d20f : SUCCESS in 4m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/de909444786f4562bb96e7ff9c83c8e6 : SUCCESS in 21m 34s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7bf85d6f016c48fab00ae30fd61c01ca : FAILURE in 15m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e7c4b414b8294923a493d959a56c19af : FAILURE in 26m 35s","accounts_in_message":[],"_revision_number":26},{"id":"a79d4fd2f55fb0c88142c1c7eb9406f981462169","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-01-16 20:27:54.000000000","message":"Patch Set 27: Patch Set 26 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":27},{"id":"be5162cb6ad2d6eff1fba492f3473cf4f4c25324","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 21:59:46.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c294a71dc9cd4a0f85e503a071c36756\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cee44ec3bb94104bfd107bd7be95fca : SUCCESS in 9m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/6634c9664d95431ba6b90d4048544121 : SUCCESS in 1h 04m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/16d913ad700d4b7ea4c296fe568a2fe7 : SUCCESS in 1h 01m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/03d00f32b1d745c88406598daebbeb3e : SUCCESS in 1h 01m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b3d59d5532b947779ba4fd024bd36752 : SUCCESS in 1h 04m 55s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c451664f34f64a5288278ad3a58dca17 : SUCCESS in 22m 28s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ddac5d4c3e63493d87999067e2237ed6 : SUCCESS in 21m 06s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/292c9dfcd8194b0199122c3d804aba94 : SUCCESS in 20m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bf5d956b7d58491d96a5557991b5eeff : SUCCESS in 5m 31s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ac5ac04cdafc4d18a7400687a3d37f55 : SUCCESS in 18m 27s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0f313cf6eeee4bb29dc3f7bc656fea1b : SUCCESS in 19m 40s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/dac0662f03a9477c9852fe8d26b42702 : SUCCESS in 27m 47s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ec7bc568c6814e1d80b3a94b2dbf38b2 : SUCCESS in 9m 56s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/eb9dff9456aa4780a472aeb066bd0d88 : SUCCESS in 17m 45s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/69669eb7c756474e91011f814cd2b398 : SUCCESS in 20m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cf57478de1024f5db2b006ed183c219f : SUCCESS in 4m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/05e7dd8674d14002b0e115340d8ed3af : FAILURE in 20m 03s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1067c8dc8c1e4da28eef9dc0892feba4 : SUCCESS in 7m 01s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ce68b994f52341bbb79cc602d03d2329 : SUCCESS in 38m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3c53d653b0cf4c7889e77b66b8217d85 : SUCCESS in 38m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d32e0e4f9303413ebcc413b6f45e5465 : SUCCESS in 30m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/82119e5cc8954173b5891fa76ecb8f37 : SUCCESS in 20m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a23b57df02342659156c656e8351301 : SUCCESS in 4m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c4a68f5d3dd349b4b4d7d1dedcc26ac6 : SUCCESS in 22m 01s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c5425c7699f24a3c81646ef8c13d990e : SUCCESS in 38m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/edffeb4e22cf4f1ea847f92c87e5874f : SUCCESS in 1h 18m 15s","accounts_in_message":[],"_revision_number":27},{"id":"b397cb53bc440c07950df68b16cbc044bda25722","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 23:08:05.000000000","message":"Patch Set 27:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a78e51bd32d48f68febe7a6da65e47a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/251af55439d84a63ba2458db28e645c6 : SUCCESS in 1h 16m 22s","accounts_in_message":[],"_revision_number":27},{"id":"467d4a951eb2daa18290129339af6186073bb815","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-03-05 14:36:30.000000000","message":"Uploaded patch set 28: Patch Set 27 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":28},{"id":"864d89881badd2d713aad78dc06b0475e5cd5c6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 15:54:37.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a11c41e3e8b04ca8bef0d7352918b7d3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd25366945404ab7961c6c066f31801a : SUCCESS in 17m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/1d0860e6ceb540798d661b15cdce4ffc : SUCCESS in 53m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5ed08544cfc64f8ba8ac7f752537f39e : SUCCESS in 51m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5a655ef9c9a8417ea0d2c4549bab782a : SUCCESS in 1h 06m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a313cdaef79e485e9e369c18b443fc21 : SUCCESS in 55m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/03b50221a5ee4e26b551219583656fbd : SUCCESS in 13m 44s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fa8bb89d2aea4585b79ceb1fe4d5eb1d : SUCCESS in 24m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/93d7c21d88294264a8a13f65c085e2b4 : SUCCESS in 24m 04s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d2968723ba7a44ba90777b9e8ee47943 : SUCCESS in 7m 18s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/61d1fccb351f4b84a8d2b340ed2508cd : SUCCESS in 22m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a68d0d6532694c138c5413b0b1f071ad : SUCCESS in 20m 19s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/141ff93f9fc64c9084984175f305acb6 : SUCCESS in 33m 05s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/2d8b03d088db42e5bb9399a9165b064a : SUCCESS in 17m 27s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/545c919362464314871af8321c25d0f0 : SUCCESS in 16m 20s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6d2d58e3f9e24f62a7625320e79fb7b5 : SUCCESS in 14m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6bd089c1903e47c6b8a1d1cac3246975 : SUCCESS in 6m 45s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5d5bef5af02d494382f94a99209b4653 : SUCCESS in 14m 04s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7de4fbb493794f909873d56399f51bea : SUCCESS in 7m 16s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/35c7d2f691584f92a57e43c6cc1deaac : SUCCESS in 52m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e20d566d2afe4980a1e428df250ed549 : SUCCESS in 46m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/aeecaf67f28e48c7a8cb06bf5de57ed0 : SUCCESS in 58m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a8193fc9f914b5d84da2300f70c8c4b : SUCCESS in 19m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4879e69532ae42c186635a8d726efe8c : SUCCESS in 4m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8a4f6fe3e786467d92f8a462fd5e9875 : SUCCESS in 18m 32s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6345162514ce4351827ffcbc6646107d : SUCCESS in 47m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/525e07e1135947b595aca605515c8f6e : SUCCESS in 51m 23s","accounts_in_message":[],"_revision_number":28},{"id":"2a4b4076729d5e258a18ededda9cb210b95de364","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-05 16:30:40.000000000","message":"Patch Set 28:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/968bdc8228cc495fbcd4eda58256928d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1c8b4f3dab8a4b52899576af76f8e376 : SUCCESS in 1h 47m 11s","accounts_in_message":[],"_revision_number":28},{"id":"0b814d920508faef0d81b5a0112fd647927d4079","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-07 11:58:42.000000000","message":"Uploaded patch set 29.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":29},{"id":"451bf710fd93c80221373914742dc5c328ae14e2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 12:55:18.000000000","message":"Patch Set 29:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9dc736aa407b4e8bb00017fdee44d7fd\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/706c6dfcc0e5442d88dda3635cd882be : SUCCESS in 54m 22s","accounts_in_message":[],"_revision_number":29},{"id":"c05e1200817747b76c6e5e94b8dd7f46b4e86117","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 13:08:28.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/edca0af54cb6461a95560f7ad5735afb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99bf7af5fc2f4e2a9a882740fa251030 : SUCCESS in 8m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/0fd75f0d294d43cfab062757881d63a7 : SUCCESS in 28m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/54f3de1805a94280a33b0b091d5ffb1c : SUCCESS in 59m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2c7b211e8c0c44f2857f9c04a9208663 : SUCCESS in 44m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a377a0a630cb4f68b46f78e90802044a : SUCCESS in 40m 43s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bc074adeb3384ce48e07f1c36d0c97f5 : SUCCESS in 21m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b09e28b0e3274a3dac189065252bfa02 : SUCCESS in 22m 16s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/74249eb5c0a14aaa8b2d8b7d9cc52f3b : SUCCESS in 10m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a39f8c7260cc4bfdafb5113cb59538fb : SUCCESS in 5m 46s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/92add7ebc1434db88b96463a3ae16b5c : SUCCESS in 18m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/99d5a363936743b8a62d8af06016d1ed : SUCCESS in 21m 05s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e24a572395f24839bc7bc421df8f6b4f : SUCCESS in 28m 36s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/8a288a770da3458c9112b900e28ee17d : SUCCESS in 17m 14s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d8fac790ff704d9ea095dad397d4e976 : SUCCESS in 7m 38s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/29beecace1a346ea9450c4812008a4fd : SUCCESS in 14m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3486ae64b0f347709fe745c38e250a7e : SUCCESS in 6m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/300a812e93f740d6b0ac725cfd0824d5 : SUCCESS in 9m 48s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/033ef9acd2964bb19a739d86f21fe8ad : SUCCESS in 8m 41s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/012a8f54761e485ba6b13be87fc7a62e : SUCCESS in 53m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/397b856c32a44a0d8d8eaadb7c9f5383 : SUCCESS in 38m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/10a2c63ef4734b53a3a6ea8edd62bdd2 : SUCCESS in 47m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/289184a274a541bca6e6a636a5dfdfeb : SUCCESS in 19m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e2d7107decf4c6e9109f0543b09b74e : SUCCESS in 4m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cc029ab981c9425caae47e15d0902696 : SUCCESS in 20m 18s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e8502d2033f741f2b5c087ce28320835 : SUCCESS in 47m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8e4e67dbc3ae482db2895edb16251336 : SUCCESS in 51m 58s","accounts_in_message":[],"_revision_number":29},{"id":"5999e3d250c937b8372f81a05d0714c2599d8259","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-07 13:12:08.000000000","message":"Uploaded patch set 30: Patch Set 29 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":30},{"id":"c7efa2a0312c0d318c0a75071b8edf869c849237","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 14:30:49.000000000","message":"Patch Set 30:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7cc302a4c7d54a3193c2c43455c18c5c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/be04350fdc03436a90472801ff67e668 : SUCCESS in 1h 16m 04s","accounts_in_message":[],"_revision_number":30},{"id":"3be780f8086374f1364ca4974d0bf817728fd7e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 14:32:14.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22d4234fa8324cd59eeea1cf1d9b0c77\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3eff6cf5a05d4eb999c6636696de5f45 : SUCCESS in 4m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/589c1dee50c74680b49e81233a83869c : SUCCESS in 1h 04m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/748708927ef94074bebdc0a3c5c634bc : SUCCESS in 50m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f6f0d069511a40c9a5bab30c98c49ab6 : SUCCESS in 45m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/25bf40d6ade94811a65d6bdf6c329741 : SUCCESS in 1h 03m 11s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8ec99821c13a4bd1952b05ba4e2e1970 : SUCCESS in 19m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/64257a8a6b9242cb9a646a6001d29fa0 : SUCCESS in 20m 05s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/27580e23416b479a8bad27b904ff10ec : SUCCESS in 18m 08s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4c3cd0d0cb7745eab6ae42752ce0db6b : SUCCESS in 5m 26s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d9426667efb8454baa5521c411f63cb8 : SUCCESS in 19m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f63e649ffb3e4b03b63dc1c05afcb173 : SUCCESS in 17m 25s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a1942c93ec8a4ae68f044ec6f6238610 : SUCCESS in 25m 13s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c9dbdf94dcac48cfb457540e1aa9fe58 : SUCCESS in 17m 33s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c522e99fe5a446c1924d14f165f11dc8 : SUCCESS in 9m 12s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/64797c4fab74464f86211ccea9c15b50 : SUCCESS in 16m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/206a27eadc1a46009cff79405621b388 : SUCCESS in 6m 07s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/99030872847045b9bc377527638371f1 : FAILURE in 19m 07s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6d0f4e4180c1406a8f510f26a5eecfea : SUCCESS in 7m 10s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/48485dca0a8e4d36b13218b911997bcf : SUCCESS in 56m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ebd80dbcb9df4a82a702b9ad4f068bc8 : SUCCESS in 20m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7e0b9d4ea8dd4ab18d68a15eadad27ac : SUCCESS in 50m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cb10b86a320a49babf57c4910758d48b : SUCCESS in 7m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6d63e79365a4b55b9e38d6942b90986 : SUCCESS in 5m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/091330e01275433b8c3027b9c09fcb47 : SUCCESS in 8m 22s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4f38a2f7f7b24c3b96bdf2c3782eec35 : SUCCESS in 55m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cad249218ae94e6899c7dd7b64db30aa : SUCCESS in 1h 13m 01s","accounts_in_message":[],"_revision_number":30},{"id":"770163ed0f7880078cb77ebf7ade9737d302c19b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-17 11:34:30.000000000","message":"Uploaded patch set 31: Patch Set 30 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":31},{"id":"27892dee8e66ae21b692374ea00d475be092f6b8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-17 12:31:57.000000000","message":"Patch Set 31:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ae9d946a12d42ecb24bcbf933e3f2b4\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b3d51e8d61ec4692bb4fb8bd608a06e7 : SUCCESS in 55m 45s","accounts_in_message":[],"_revision_number":31},{"id":"8dd15cf6a859e4ba616ba80d6bf73742d9374699","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-17 12:41:31.000000000","message":"Patch Set 31: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab2c19b574584f1b86373575a1f33906\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7329fb2eb2da4a51813d4638aa7a372c : SUCCESS in 10m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/9907b8daf86443a694a7cc24fc6a4fb5 : SUCCESS in 58m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3ef235ec29ed485b8a662cc8e6ba50fb : SUCCESS in 54m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/17772f958ebf44169251df4c7ddd66be : SUCCESS in 59m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e3cf93e7c1be42428df920da59f4372a : SUCCESS in 1h 02m 29s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/23a1354f54de4001b7827bfc7e16cdea : SUCCESS in 20m 23s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1ee65bb241bf4542b8cf2bd775314b76 : SUCCESS in 22m 09s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/798eabaace8147a48c433d7a780f5933 : SUCCESS in 25m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0797a7c69a42433fa6e858ee8d205ee6 : SUCCESS in 4m 31s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c6fc9c9e7e314b03bdbe4b95c792b2a1 : SUCCESS in 9m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/abbde7bd96184eafa7b9d05bfd8381f3 : SUCCESS in 21m 02s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d833b40efd9c4eafa8e107a1ef821e11 : SUCCESS in 28m 45s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/1bc2fab57e5248289911b623ef2fcee2 : SUCCESS in 12m 27s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/309f54f04f64436391733695a99af349 : SUCCESS in 18m 12s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4444b7ab4d8644ddb8e2622df50e0629 : SUCCESS in 17m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/90f33a2be0c74e11b4a12fd3d35472cf : SUCCESS in 6m 21s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9ffb68af93954c4ab31337a3b704f0ca : SUCCESS in 17m 43s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2869a2928b6447168f3ab61069a88ae1 : SUCCESS in 6m 40s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e150299066fe424ea3eedfdb53ec2fad : SUCCESS in 37m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0f0e00a7fbb442ccac16e6165d99bc94 : SUCCESS in 41m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/423eb9194d61444986a7d80fe939499c : SUCCESS in 32m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/558d7e3ffa8f4831ab2175b1bec7ec7e : SUCCESS in 20m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2514603560640e5ab259e21df712258 : SUCCESS in 4m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3e41f6f1482e4776a1565fa8087fe6bf : SUCCESS in 20m 19s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/34887a9b58b044d7b99ae7fa8509f5fd : SUCCESS in 44m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/81af9295d3dd4a3cba506c710f1b080c : SUCCESS in 58m 21s","accounts_in_message":[],"_revision_number":31},{"id":"e534bbc2ac5700be2099500aa33294dda5ee5082","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-20 15:24:50.000000000","message":"Patch Set 31: Code-Review+1\n\n(10 comments)","accounts_in_message":[],"_revision_number":31},{"id":"b462bd8c7f9587fdfa675d453f1d7a41fa68ea37","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-20 16:59:40.000000000","message":"Patch Set 31:\n\n(2 comments)","accounts_in_message":[],"_revision_number":31},{"id":"8a4bf1cb1c3b043620441c4f245c6ac36e26ffed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-03-20 19:35:52.000000000","message":"Uploaded patch set 32.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":32},{"id":"d8508ab8c3caa8012638a461490e9d75ebc10a3d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 20:17:34.000000000","message":"Patch Set 32:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d46311a170240d2872ee74cfc72a58a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/993aec8e988c42768343b820167346ee : SUCCESS in 36m 20s","accounts_in_message":[],"_revision_number":32},{"id":"aca03cb1ffd36587f3bbf2821dd600f391654c62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 21:16:29.000000000","message":"Patch Set 32: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9d7ff9e6d004ebbbda50fa86cb49b04\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/887ec29299ad4c66912cb4bd4e241ca0 : SUCCESS in 29m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/60925a89d2244814a510dd97bc0b8020 : SUCCESS in 32m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4215093abdcb466a98ec4ccd295fc459 : SUCCESS in 1h 39m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/55129803e1ee4f6ab138b771d9e4c95a : SUCCESS in 34m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6455911717c34df2993cb061aff44288 : SUCCESS in 55m 02s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/25f012033e5141d3b33d2c6b2d2b6a37 : SUCCESS in 20m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9706ef115e904758bf54cb2265dd048f : SUCCESS in 27m 07s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/09f4c97b8c1b45edb8bdcc49dbc0733d : SUCCESS in 12m 13s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8d733cdd5d5d4a7cb5050119b75530b4 : SUCCESS in 5m 25s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2a19270d52b34f639aea141906c2acfe : SUCCESS in 9m 22s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7c343fbeabae4ddb81afacb49de32b8e : SUCCESS in 25m 00s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/add50d405dc044e2951f1ffc791ffb1b : SUCCESS in 24m 12s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/49820b624417410f8012a6e27a40ca18 : SUCCESS in 20m 54s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ea0e33baad4c4c14ad2508577147f72f : SUCCESS in 14m 40s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d3b4dfd7a7294f42993ad3a4fdad4dcb : SUCCESS in 9m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f12e1d599ba3439899006b7fe882fe1f : SUCCESS in 5m 15s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a940d65dd01d45b1959eaf6d1d52bb9e : SUCCESS in 18m 10s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/72eb0fc9ec864252980aef88c3f5ec27 : SUCCESS in 6m 56s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9866070be4704f008ca522a69d2199de : SUCCESS in 39m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4e14b83d325645a1820a4b84850bad87 : SUCCESS in 48m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b5712fca9f614a47b6c52cb72021baa7 : SUCCESS in 37m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1dcfe0d27c54409da7fb90e03146782f : SUCCESS in 20m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/092c8db31d5144a1b9b4e8c46128c8b3 : SUCCESS in 3m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/93ea19b3dd644d68ba037b2af7784c11 : SUCCESS in 19m 52s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c6434a95d0b84d8f970cbeb76873fb3e : SUCCESS in 44m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eedd40435a394bc0b05b9433c8c4fe37 : SUCCESS in 51m 50s","accounts_in_message":[],"_revision_number":32},{"id":"589075b968f2abde794948846b4e95d7d5c9ace9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-21 15:18:40.000000000","message":"Uploaded patch set 33.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":33},{"id":"60e72020e65dbbb2976aec6a2a934f83a8f01895","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 16:12:12.000000000","message":"Patch Set 33:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d193fce21fa943e2aa917bea1065dab3\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a3986aa7702040eba9753d2c3569ba20 : SUCCESS in 51m 36s","accounts_in_message":[],"_revision_number":33},{"id":"9f3f139c40c4f3b9183c4a346dbda23b69b7725b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 16:25:32.000000000","message":"Patch Set 33: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8bce334d78614063bd72502bb0ee725c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bedba67b4b546c4b28b4cf59f608b03 : SUCCESS in 10m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/221889f89d1b40e38a8fdc2ad3c43874 : SUCCESS in 56m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9f6be0f2fa544a518eed9afa0d61296d : SUCCESS in 54m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/789c9671ebcd4b6cba358b69e81b422d : SUCCESS in 52m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f706f7ce8a5d4b64990e06fca14db31a : SUCCESS in 59m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a6ca25e1b36742d3972b39a3d3b97154 : SUCCESS in 20m 07s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c36ec98b245043f5bce84f6f75a529f8 : SUCCESS in 22m 30s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2a4df012f979449d9d7f78fe7af44a6a : SUCCESS in 20m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6e3a8077e205491cb0b503414ffbd574 : SUCCESS in 3m 25s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c882fdda6bb844a192ab688152df47c5 : SUCCESS in 9m 23s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ff4c88e437884d41aa38048548e4163c : SUCCESS in 21m 04s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/60ec779266974b69a3d6f92ded384709 : SUCCESS in 19m 41s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0e1dc4b74aa24685af589a96ffbfd730 : SUCCESS in 16m 30s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/81842ca14a484ab5bef81298b197dc85 : SUCCESS in 19m 59s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4f957161900a45f890f1844f80fcfce0 : SUCCESS in 13m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/54010592b0574c55b9df435c7cd4a16c : SUCCESS in 3m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fbc7166698c7485e95246ea7cc7718c9 : SUCCESS in 17m 47s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a19460143a094fefba4937d2ecb2816e : SUCCESS in 6m 38s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/910048c2757c46909b147e6860b142ab : SUCCESS in 55m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ff4c76ef98a74357b55f2200749b8ed5 : SUCCESS in 21m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6ab46474034e4a28a7c54279996f3ff1 : SUCCESS in 36m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d9d2b67497b04813af203b2322ff520a : SUCCESS in 23m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1eaa6b38e7c4637ad60cb9006b7e83c : SUCCESS in 4m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9133f521c9034412a5e208e6f6718e0e : SUCCESS in 19m 45s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b16446cdd3114de6a07105f9b05dbecb : SUCCESS in 47m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f1ad74a3741045b88c2443e05b400d7f : SUCCESS in 51m 24s","accounts_in_message":[],"_revision_number":33},{"id":"ef6e1b732d13fa1d6b0825cd89791b1a1991cb00","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-03-21 19:30:42.000000000","message":"Patch Set 33:\n\n(9 comments)","accounts_in_message":[],"_revision_number":33},{"id":"8a6e4a344410be7b3d6809361b50cbf09245ab9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-25 11:32:12.000000000","message":"Uploaded patch set 34.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":34},{"id":"c3d4037266706e918e4b9bbf2978c46f2d8c00da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-25 12:32:11.000000000","message":"Patch Set 34: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45e9587b59de4dda8271f6576a1c6267\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b3dadff4711466d921e8a5664b90450 : SUCCESS in 10m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/5b706abe30734eaca4008a2ae541ae48 : SUCCESS in 47m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ef2e2d5f666d40d6a9b6c30ba1178d62 : SUCCESS in 45m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d22a7c7cabc74ecf829a0278ff2d2bb2 : SUCCESS in 30m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/52c4a86f344c479aaf5152bedd9dbcb7 : SUCCESS in 55m 55s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/22dd8367c50443e69257c73c18b28941 : SUCCESS in 20m 02s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fe366d88feb145d4823a76fcd0d82964 : SUCCESS in 24m 43s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5ee77e1ac6b94ca1a80dd36e3b5c7b11 : SUCCESS in 12m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a3c7521fa5f047ea90fc076db64af034 : SUCCESS in 5m 25s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ab9f93ad423042439e25a3b4e8a32862 : SUCCESS in 24m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7bf06744df5d4e4d840c676bf04093e3 : SUCCESS in 17m 15s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/be3efd1705bb43dcb64155518cd2ab19 : SUCCESS in 24m 08s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/2bf0614cbfd747f998c7f57e58433beb : SUCCESS in 9m 02s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/bcd669260eef4d0e9d252ece0f63507f : SUCCESS in 15m 42s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ffc870d2fdb8445e9805fe835dcf38de : SUCCESS in 9m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6fbc95168eba455692ccc11ed7330318 : SUCCESS in 6m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/76c71463732848f1a1cfea0613534113 : SUCCESS in 14m 36s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/fc13f764b357435c82ba16a31471f6f5 : SUCCESS in 6m 34s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8c7980e87759491784747171b38ec6ba : SUCCESS in 53m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f977e4737c864cdd8a409133f1f805b5 : SUCCESS in 17m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0ba13675fc6b449ca8b7245f9ba7bacb : SUCCESS in 46m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ea154904a9884e03b61a46b737b27c4a : SUCCESS in 20m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0063247953947009c236c6ce26d1f0d : SUCCESS in 5m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/62abc9e81a9f4c62bc529ba1d762af2e : SUCCESS in 19m 35s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/feaf1251df5d444eb0cf47677d36b1f0 : SUCCESS in 39m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e57f697ce1ec4d0da72f7e6c686887d4 : SUCCESS in 44m 09s","accounts_in_message":[],"_revision_number":34},{"id":"386ac416184ef642f9d8726c40e3c7ae11d2fd78","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-25 12:34:30.000000000","message":"Patch Set 34:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37a7daafa04744aa856c71adb880495f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/dfb0ed8a217643b09ea5faeec52af19c : SUCCESS in 38m 48s","accounts_in_message":[],"_revision_number":34},{"id":"2d26b6dbb091a53887b941e5d803aa3df7378085","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-25 18:30:59.000000000","message":"Patch Set 34:\n\n(11 comments)","accounts_in_message":[],"_revision_number":34},{"id":"88db9dab158fd0d4ac0450a38ee6ed12246f303f","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-26 15:08:54.000000000","message":"Patch Set 34:\n\n(10 comments)","accounts_in_message":[],"_revision_number":34},{"id":"4efa68390a5f94a3d29ecc5c947a9c8e50bf1431","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-26 15:09:28.000000000","message":"Uploaded patch set 35.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":35},{"id":"d4632d129faa99a439d46dc059bd989fe6b88d15","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-26 15:11:06.000000000","message":"Patch Set 35:\n\n(2 comments)","accounts_in_message":[],"_revision_number":35},{"id":"6ab1e5a988d4c7655c904e4d40b55f19c2f67e09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 16:17:48.000000000","message":"Patch Set 35: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7685f9fe5f14940bb9a8d6195bbc1c7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a4d782782ec4ea49c980286d4950d30 : SUCCESS in 5m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/f1fe952b0020418d94b5fb698f553372 : SUCCESS in 58m 25s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/50abe44ded4a4a2d986aea68700974d7 : SUCCESS in 53m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/227188cf822d49de816fab19eb82a018 : SUCCESS in 43m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c7314784a2ce416397df1f73ef1cde9e : SUCCESS in 36m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0a100581ce6c4b9a9150ace8010286d8 : SUCCESS in 12m 21s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3856eea4603f45b4b2e7f4f40765b708 : SUCCESS in 18m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/40fd9fb99f81428db595b8163570666b : SUCCESS in 19m 47s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a775d922b9e54ea7806eb5f98168d1f0 : SUCCESS in 5m 51s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/259b5bd7777244fcbac61d6efecfd27f : SUCCESS in 19m 18s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/da020f9b894c4aacbb50a8f36c40e1b2 : SUCCESS in 22m 22s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b79d76bc714a41fca9eb2cfa9aea81c8 : SUCCESS in 37m 05s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/eae003c55ef44aa8823b6df4b713feab : SUCCESS in 14m 25s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/626bdf694b694f168ae2fbc9f7cb34af : SUCCESS in 15m 48s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e6985a72c7804c32b5e446cdd178b91f : SUCCESS in 13m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7690aac1846643e5aa1400eff7f314f9 : SUCCESS in 3m 17s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/08a9d9f54ead4f3594e67301f997281a : SUCCESS in 18m 21s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9a67efd623df4115a911d5b55f206201 : SUCCESS in 7m 55s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8b1748a8f6ae49de9534193aafd0e422 : SUCCESS in 58m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b4ce5cd20b7d4f5eb3fbf2cda9bbdb03 : SUCCESS in 53m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b81c379fb7de4980bc1efd51911723a4 : SUCCESS in 35m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9eb7cb86642447a38997fd3105708b42 : SUCCESS in 17m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e48ab111ca064c61986f6789202c50b3 : SUCCESS in 4m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4155bda0696a4daf9ee15833cb07c32b : SUCCESS in 21m 26s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1cc32137c2ef414899ecedc7fd7c59f2 : SUCCESS in 44m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/97149280d483449fbbc32fef63a53c92 : SUCCESS in 53m 05s","accounts_in_message":[],"_revision_number":35},{"id":"457239f469bb63dd9391f76ed846d7b8560d0cd4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-26 16:19:14.000000000","message":"Patch Set 35:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef69ad920d61443b96e0962aa759ce1c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/583b3665299c421d980d687747a9f6fb : SUCCESS in 41m 37s","accounts_in_message":[],"_revision_number":35},{"id":"6eaa079f04dff3aa834152177b51284c620a8936","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-04-01 13:58:50.000000000","message":"Uploaded patch set 36.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":36},{"id":"cfd5c1b764d5f4ed5ecc38644d96487e8fc752d3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-01 14:45:11.000000000","message":"Patch Set 36:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4461457e53074381b9faef83875840a3\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e0336f95088d4d799c197eb1dfa902b4 : SUCCESS in 44m 33s","accounts_in_message":[],"_revision_number":36},{"id":"dedcac59445f26bd76d78cf83fc1e33b03ed0eab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-01 14:59:13.000000000","message":"Patch Set 36: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/395a0b915bdb46febdf0cfc8f2db8b7b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e76c9584b1346278df2545b65d9e688 : SUCCESS in 8m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/084d980e7fe54023b00684f81a9cc82d : SUCCESS in 49m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5e2e86013a254434b3688a19c2d85a3d : SUCCESS in 57m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/35a47362dfea43d387551a98f66b76d5 : SUCCESS in 54m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/45256a0864cc473eaa648dc9b14be786 : SUCCESS in 49m 28s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/47ff0dc563f24bd48caa4495c3e3e7a9 : SUCCESS in 15m 56s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/060c875850f345248525e3a3d7f7d585 : SUCCESS in 20m 45s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/881757d169b44b2393bcf6af2fa136ea : SUCCESS in 22m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1c203644f37f4ea3abd1c0e586cd01a5 : FAILURE in 4m 34s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/acce93741ab243738e1ae7ba56582ea3 : SUCCESS in 15m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b08d89b0d6ed4002b44bc047dee7f5ae : SUCCESS in 18m 16s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c3d6e05e001f441f8987b98a9d785137 : SUCCESS in 19m 21s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/8d9e4210906d46ba89d51c9cbfd3d3b6 : SUCCESS in 14m 31s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ab127edc875d457380ecf03c21e11977 : SUCCESS in 14m 11s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/be74e6d3369c4eeb9eda79b15312765e : SUCCESS in 15m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2312126eb7d44ae19565b830ac5c4764 : SUCCESS in 8m 24s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/da6ab941508142bb9f98f76c8769a915 : SUCCESS in 15m 10s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/482e5b9316264686a1d11e3e1b5ff95b : SUCCESS in 9m 02s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/fe88bc2e24804eab8bf33c152e7c953f : FAILURE in 53m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/488931b15f0d4dc0a8289f345de103ee : SUCCESS in 43m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ab617d227cf941518735a03ec68ff595 : SUCCESS in 44m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5f9980d14145486ea9dfaeae13f1e328 : SUCCESS in 21m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0157ad38d15404fa180ab392a4929f8 : SUCCESS in 3m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dafee0fe169b40a4ab8f6698c2a3b217 : SUCCESS in 17m 30s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3b0b691570334dd1bfb115dec07948ec : SUCCESS in 37m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/679f0c5a97b94154b04084852b52f596 : SUCCESS in 50m 05s","accounts_in_message":[],"_revision_number":36},{"id":"0fa0c2aa306b791ea911f67b286d82227f4b5b41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-04-02 20:12:50.000000000","message":"Uploaded patch set 37: Patch Set 36 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":37},{"id":"fd71a4ec7f03bb97f3c277952d8df3faa084dece","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-02 20:47:07.000000000","message":"Patch Set 37:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69ad355fa754403080a32dbeda06024f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/36e6d4321edd43768e30f6bd6fccbfc4 : SUCCESS in 32m 29s","accounts_in_message":[],"_revision_number":37},{"id":"73d84a705191028aceead1f89f9c55ba4f765b73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-02 21:46:22.000000000","message":"Patch Set 37: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/74923822669e45ab95a343d6a2922e5b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13ae10aa0aba416db9b2fd2f4f308583 : SUCCESS in 8m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/90afc1064059428382e0afc1e65c4934 : FAILURE in 46m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/dfd46cdc04694693803394697010363b : SUCCESS in 41m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/757bf37480424db5888bf983858cc614 : SUCCESS in 58m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4dd9754f64fb4a3f8f37102bdb6cf989 : SUCCESS in 52m 27s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0154d4049ae54f12958db0cd78337df5 : SUCCESS in 20m 48s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/323a76676fef4352acf3bb222c41d47e : SUCCESS in 20m 33s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e21b197a387b46a58b46e6234dab2ad2 : SUCCESS in 21m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5405698ad8cb42cebec346b9dadb8956 : SUCCESS in 5m 07s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fb8b5e9b308e46b18550ee6d77811666 : SUCCESS in 16m 28s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/13379b40eac5457591dd1fd56cfc4390 : SUCCESS in 15m 24s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ed1f8366458e4d019a8ecc1f1f9c882b : SUCCESS in 18m 53s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/34304b1737de4154b53c27e3e34aa480 : SUCCESS in 16m 55s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e1f2aedbf4c94b3680574f6ffa9aa392 : SUCCESS in 15m 00s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/cc85eb3efa53486abd8639a3e119020c : SUCCESS in 8m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9acbbb0382894306a36fc4dcaddf4e09 : SUCCESS in 5m 56s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5cbd92cd6888401ca7cd4bcb01a91756 : SUCCESS in 14m 47s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8650711dcba54deaa62633d6fdedf235 : SUCCESS in 6m 14s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4847f4b332134aec90f36134b88cbbd9 : SUCCESS in 44m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/47817700ea474ec4aa4433852bc2e6b1 : SUCCESS in 47m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/989c1ae4b30e4084bc11d799960298de : SUCCESS in 32m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c67d17b004ff4496b1346fa0a02b1b8b : SUCCESS in 16m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61bf575d2aeb4c31b4a064eba403569a : SUCCESS in 6m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e93cb8d2727741a1b0efd13372c5e033 : SUCCESS in 20m 50s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0b46bc30c25245fc8fa6dcf7d3be947a : SUCCESS in 31m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7bdf5ab5eaa040998fadc38cf7c33611 : SUCCESS in 54m 56s","accounts_in_message":[],"_revision_number":37},{"id":"efdab0d5fe2e72f908448503ef1decb50d1e2e8e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-04-03 17:00:13.000000000","message":"Uploaded patch set 38.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":38},{"id":"17f1e8faa9520a40efdb4a9021d034de7c327cc4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 17:42:06.000000000","message":"Patch Set 38:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce8525c52f4448fe808e48b4d5668e08\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7210eb73e37c49249c625ce114000347 : SUCCESS in 40m 57s","accounts_in_message":[],"_revision_number":38},{"id":"5b0e85b9c2eb962f12f6e5c817ef9fa63c12894c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 18:01:40.000000000","message":"Patch Set 38: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ba5a2bd8927b49d683714a73222e4bd9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3423c1ec126b4243a87a3137cce9d64a : SUCCESS in 8m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/1b702e75e7e1406f9a23fe7b4cf029b1 : SUCCESS in 47m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/509e945159f349f499c00eea6ece6a6e : SUCCESS in 34m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/168b876c01f8479eacb2e02a86bf3662 : SUCCESS in 43m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6813594b1d1d4aed9d16f33b6ca5c3c3 : SUCCESS in 56m 54s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/04284ee552654ceebf6c4a012a91f19d : SUCCESS in 16m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4617b8f508db423580a4e3d0c826abb4 : SUCCESS in 22m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f0aa9948d1834376a1ea5b58fed09b0a : SUCCESS in 18m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4e43c2eecf8c4f788062bc58fb73775b : SUCCESS in 5m 16s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/31c17bafafc240359f1dfa556e1d66fd : SUCCESS in 16m 38s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1d235a7325bf4653bacdce3a8d9d17f9 : SUCCESS in 18m 07s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/05483f7354dc43678961ae8ce7c640b9 : SUCCESS in 22m 12s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/253abf65869245f8b963c0b6696105cf : SUCCESS in 18m 07s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5eba0844c47a44a496f89a3407fc2601 : SUCCESS in 17m 11s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/83e8aa3f6a9844f68d56e44cf26f65fc : SUCCESS in 14m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d28bc3654ea7486a99d2895db1d64cb2 : SUCCESS in 5m 17s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d672451964864d1dbe8eaaf19de93549 : SUCCESS in 14m 59s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4bd971eab84d4984a1360fd922c4b6d2 : SUCCESS in 7m 08s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a7dba820b76e42899fa1e6726db3c49b : SUCCESS in 55m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6a7fa9b4b53b44fcbea0f1fdbd92e44c : SUCCESS in 45m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/268bc9f791584503ab5a7c10a5dc44fe : SUCCESS in 44m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f184e1678f084cf6ad2481e1edab42d1 : SUCCESS in 19m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4720222674a8474aacaffc01078fcb6b : FAILURE in 4m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3530f79b70024b088670ad9e881c5949 : SUCCESS in 22m 07s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7fad790598ee4a77a1042ee53681e5b7 : SUCCESS in 46m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f56550e6c615415798a0262f8dece82a : SUCCESS in 55m 01s","accounts_in_message":[],"_revision_number":38},{"id":"4e880ce9cb383be8cd66ce62f07947e57ad80b84","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"real_author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-04-03 18:28:12.000000000","message":"Patch Set 39: Patch Set 38 was rebased on behalf of \u003cGERRIT_ACCOUNT_36606\u003e\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"}],"_revision_number":39},{"id":"9c413da68535c68c65ce3513a11589c0981ebd73","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-04-03 18:44:14.000000000","message":"Uploaded patch set 40.","accounts_in_message":[],"_revision_number":40},{"id":"3731f975a34ffef1a3b120109bd3f124fa2cfaca","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-04-03 18:50:23.000000000","message":"Patch Set 40:\n\n(1 comment)","accounts_in_message":[],"_revision_number":40},{"id":"a34438cc67fbdac711d9498f1b90b18ada7a248f","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-04-03 19:23:28.000000000","message":"Patch Set 40:\n\n(8 comments)","accounts_in_message":[],"_revision_number":40},{"id":"d288411a3f19470dd55bd8be0a671d03d45a9ce7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 19:25:50.000000000","message":"Patch Set 40:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3e2453cf6404451b64fb3651470ad8c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9aea28a221c743c495929f4975d62480 : SUCCESS in 39m 34s","accounts_in_message":[],"_revision_number":40},{"id":"d00c51b36e431793a8d1626768c522fa5a670422","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 19:42:43.000000000","message":"Patch Set 40: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc0f3ed5d92b48e18b825ba2de229700\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a22dd3d9faf441ae9355c084f2645e66 : SUCCESS in 5m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/fb304e9fd1cc47b19ffed717478ad639 : SUCCESS in 43m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/138b2f95259143e08245032ba0c007ae : SUCCESS in 42m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bde7726b42e04329b69cac34880786a1 : SUCCESS in 52m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3b9cc82c896b400884f201ba48776793 : SUCCESS in 47m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e340754fbe654aeea2da76aa4fdad04b : SUCCESS in 18m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/32c7a53fc77e49c5a73150c43857ea4b : SUCCESS in 21m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a3ce87a77b004186b5dfd7d46918ea22 : SUCCESS in 19m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4e6cc0eb40c54ef880008e21a8bf60da : SUCCESS in 5m 16s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a8d0eea83e01482196f82b554e04bbb9 : SUCCESS in 15m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/54919dff2fcc4c44b243162fce98319f : SUCCESS in 23m 02s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c94e15845c174095a3aa8f5f627d153f : SUCCESS in 21m 30s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/afb3f02c19d54acca549da75323d5a49 : SUCCESS in 15m 11s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c188919771a94d9ea0a18035ca2f3b43 : SUCCESS in 13m 33s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b0b8c5bca5634d2e9ffe60b57030f8d1 : SUCCESS in 16m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4fa0417007cd4e5fb9850d8fdeddcebb : SUCCESS in 7m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3afa2a0a307b4d0f84d120de367cd526 : SUCCESS in 15m 39s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/39a6e086d95041f1816700e316f28eb2 : SUCCESS in 7m 13s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/71c973e888e5428ca1e602dc3bb9c421 : SUCCESS in 55m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2686187aa5374b40a9c90af4081084cf : SUCCESS in 34m 14s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5c146a1363c346b189b056f14115a8f0 : SUCCESS in 36m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9c31e284751c441d90aa575f9d44b8f4 : SUCCESS in 18m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/030aaa1df9504215a7f0efb2469a812e : SUCCESS in 4m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d559bfd99d0047a2a1798bc5bcf17408 : SUCCESS in 19m 38s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/942317c393754c4d974c27e26de46714 : SUCCESS in 39m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8596736901dc46d2ac2a4c142453d81a : SUCCESS in 44m 18s","accounts_in_message":[],"_revision_number":40},{"id":"a57c58cf59c6bfba5ae81816abe9778afc55ede3","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-04-03 23:28:07.000000000","message":"Patch Set 40:\n\n(1 comment)","accounts_in_message":[],"_revision_number":40},{"id":"3b6b8d0d71e1230f667c57fcdfff76c6e4d215b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-04 14:35:45.000000000","message":"Uploaded patch set 41.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":41},{"id":"25f3b3dd17d8fbfd4eb860321b19927d52568352","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-04 14:42:48.000000000","message":"Patch Set 40:\n\n(26 comments)","accounts_in_message":[],"_revision_number":40},{"id":"1ad95e2103f6fcf40cf7befac12a95306f62f179","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-04 15:16:41.000000000","message":"Patch Set 41:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/365111565e5246d588e75aa929f6f537\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/57d8afa73c18472f965d6f4ed7283ded : SUCCESS in 39m 04s","accounts_in_message":[],"_revision_number":41},{"id":"c9f985d591cf556a5b946b8b9fb46789c8914c37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-04 15:40:40.000000000","message":"Patch Set 41: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa08ad88a94e481a82b6c3d01ed861ff\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85961eead1ac4c8392b61aeb154375e8 : SUCCESS in 10m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/f3b162db33904ef9893b25e40529fc03 : SUCCESS in 55m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c66a60181de84ee597a05ce2842cbd2a : SUCCESS in 51m 34s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8c379feb7adc4115a4991e3513afba10 : SUCCESS in 1h 04m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d6b72237b02249fab9e23e452d0ebf31 : SUCCESS in 53m 04s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/51ddfb54e83e48af97c358927a2dc228 : SUCCESS in 20m 21s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6eaf700bc5604f70bf288b1d38a7710e : SUCCESS in 27m 59s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b36ab20431504678b66504f5707604f3 : SUCCESS in 21m 31s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3f1c83da9888487085784fac2cccfd1a : SUCCESS in 5m 08s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/05907febca244a029c7ee28b86e01bed : SUCCESS in 19m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/cfba172d849f44a4a90a97bed7f96617 : SUCCESS in 21m 17s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/fb180367ed334e7fbbd596ba3f37ea5d : SUCCESS in 34m 50s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d2c13a7648544021b02e60878de8cc07 : SUCCESS in 14m 38s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c80b48eeca8941f6911fa61adb7ca777 : SUCCESS in 14m 44s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6f71bf736c1a45a8b595ff733a8a729f : SUCCESS in 14m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/518837aeaab74b258fa2141a89ebf6b1 : SUCCESS in 5m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b95a668cf66a4d3c8a246a9bfbeef47c : SUCCESS in 17m 48s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/afd8bc80219f47888c9df2a3abf3a710 : SUCCESS in 6m 33s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0209d7c2dfff4521b611c19f447fc02f : SUCCESS in 56m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c8795b8a120a46c5a396a99a123e3f89 : SUCCESS in 25m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ccb3998bb3bd4508b3e46e9c2aaff438 : SUCCESS in 33m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/71c28962ecd4402287a5268a73a9a788 : SUCCESS in 19m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afc4687c63544526b92bf7cbd564408d : SUCCESS in 5m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3a3c476dd99b42c787f0f3825f2c0e36 : SUCCESS in 21m 04s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4680a5e8572c417e9518a7ab4cb4423f : SUCCESS in 40m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bc38c5b8e87a4ac29177760d44e51860 : SUCCESS in 48m 43s","accounts_in_message":[],"_revision_number":41},{"id":"2c3bb991a4c3f0622b2dc26bddafc74d3438f68a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-04 15:51:26.000000000","message":"Uploaded patch set 42: Patch Set 41 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":42},{"id":"11c7d39f29e3598f6fe649537129204c76deea80","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-04 16:30:39.000000000","message":"Patch Set 42:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34a4c631ccb44678ba312877981d60d5\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/56e75c3a60f04a859e5940634f333578 : SUCCESS in 37m 15s","accounts_in_message":[],"_revision_number":42},{"id":"4570b0f3eb5ca76f69f7f0b0a20483d8e8a4f3d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-04 17:00:25.000000000","message":"Patch Set 42: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa419c54dea749eab7df220b172ded3f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40fb6663a4254a779ea09c6b36532493 : SUCCESS in 11m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/8a73688ad3b74d84a458a5f3505308a8 : SUCCESS in 53m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/41d51988c68a467696bfe1c5cb85f7d5 : SUCCESS in 1h 00m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/05701829d0d84f9d9389504b51f73314 : SUCCESS in 47m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8046c0e348ab431b81b53c7c90142bde : SUCCESS in 1h 03m 54s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1436625a12284c2ab134ddaec58d3b71 : SUCCESS in 19m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/67d12b65ee8a47ea9f57dfc0d4b67230 : SUCCESS in 17m 59s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/289d3994016b47babdc23e33fd41ab46 : SUCCESS in 18m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8c8e2da78bb84b63a5afe1409eec6207 : SUCCESS in 8m 25s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5e6c42ac53184d11ab7f49bc4ad87540 : SUCCESS in 24m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3ff9c4498d444ca4bf8af9b42bc45498 : SUCCESS in 21m 16s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/912dab30cf4e4288a96a50736545a79d : SUCCESS in 29m 47s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/61e6d9a7372b40a9a58cc53586184795 : SUCCESS in 9m 25s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3898e8def3884a60a9839a0bcc789fcb : SUCCESS in 16m 51s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/840fdf4fcce043eba04ee5956baba355 : SUCCESS in 13m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bbc3dbfd5542450da872913234094c28 : SUCCESS in 6m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/90038171efd84af89a1e8f9186ad9943 : SUCCESS in 16m 29s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/30ce1ebe093a46e4b0f31c9276e435a9 : SUCCESS in 7m 29s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/09b1f3244cbe47a6a35f222c0e4bfc26 : SUCCESS in 48m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cf24e3690962499489bcff27699510e4 : SUCCESS in 44m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e67fdc66170a43328d327e62adfd4a38 : SUCCESS in 40m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ca46ac076d9348f5874039725a480b53 : SUCCESS in 15m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02c63a82d35c43af81b9cc143b5124a5 : SUCCESS in 4m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/95625a1dfe344d71bb1c7459cc1df088 : SUCCESS in 18m 10s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/87ea3854eb944d8c8ca9cfdc3bc7d7df : SUCCESS in 37m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cdac4176854e402cba5b9f24d715551c : SUCCESS in 56m 15s","accounts_in_message":[],"_revision_number":42},{"id":"b9bc6fb3cba92696bb394ce7da40ced095d967e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-04-04 21:24:18.000000000","message":"Uploaded patch set 43.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":43},{"id":"e79b647c0494da98181f7008ebd9107c6c5b031d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-04 22:10:13.000000000","message":"Patch Set 43:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0f849cc073a42cdbea182d979056c9a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b1e8ceee4025422593db27ce6b87b34f : SUCCESS in 43m 18s","accounts_in_message":[],"_revision_number":43},{"id":"8dd9e27ed1f438be1cc72e11976b52e5465b492e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-04 22:10:54.000000000","message":"Patch Set 43: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cdc63276f3b04a7c82862035064e0b59\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6aa18aaed1534f7a88b8b8be7cb6c50b : SUCCESS in 7m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/47bcf8fc8a4c48d4a2a29690dd4f2e5f : SUCCESS in 41m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3a80885cb04f465f9b09bdb0058883fe : SUCCESS in 44m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/04bf195183264d6a92d5869a5b73be65 : SUCCESS in 36m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d5a5da469bbf47d89d9ddf2acf6f710f : SUCCESS in 45m 09s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9b1fbdf8779a473bbcefad19aa549a1f : SUCCESS in 12m 04s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6889bd1e71bf4a2db073c288dd810ea4 : SUCCESS in 12m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6134a9c88a594a6b884bc84387c69f83 : SUCCESS in 12m 24s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/880789871d9f450f86022d0a1549cf9b : SUCCESS in 4m 03s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f3d55932710e4e92b99116a666bcacfa : SUCCESS in 7m 06s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/94ce96f6e8f24252a6c297efc2b8260b : SUCCESS in 10m 21s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0ac8bbd27ba6437a87a6fb82e439b122 : SUCCESS in 10m 24s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/149da0bbcf6344f09c1aad6a56da3d89 : SUCCESS in 8m 53s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d87cb205932743abb39a8ade963f752c : SUCCESS in 7m 36s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6ae16754a216472095857f40744abb13 : SUCCESS in 11m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6b527a5b98e34d6e89f8116e6bc70a6c : SUCCESS in 3m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/69e7adc113ea4ad19164d19a08e135c5 : SUCCESS in 12m 06s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/711b23c391264035b4b5827e56991e9a : SUCCESS in 3m 48s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/406150d86f264704912223b835715fff : SUCCESS in 33m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/412eeba72ce7432199577a893996b43b : SUCCESS in 17m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7d56a52220224409ae95abe990f69d85 : SUCCESS in 18m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e5c7399dd213417d942201c48e02eeb2 : SUCCESS in 7m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbd22b0523d842a08dae5732907d3353 : SUCCESS in 2m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/460e23836c4846379effc78503f5ae17 : SUCCESS in 15m 10s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/12244ec8e11a40bb887ae04ff82d9ef5 : SUCCESS in 28m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f1f8b91d79dc43dda7ca656b85fc29e9 : SUCCESS in 28m 08s","accounts_in_message":[],"_revision_number":43},{"id":"84a9a82d5e48a6f4c28421ce14d452cd09532ab5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-04-08 13:58:22.000000000","message":"Uploaded patch set 44.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":44},{"id":"fb9e02393846cb8e7d53c01807a9d4b8a41e32d4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-08 14:35:45.000000000","message":"Patch Set 44:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/156e095284c0468b97bc7661f7502f20\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/94f493b3777a4928bfc8a663d9400683 : SUCCESS in 34m 53s","accounts_in_message":[],"_revision_number":44},{"id":"5ba4081c8f58eba35fbc3cb01a2fbf3ec3b3b8ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-08 15:09:08.000000000","message":"Patch Set 44: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9601edfcdddf47739834f538c4242b42\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42a7a78cde8345e1b51acdd433b5cf7d : SUCCESS in 8m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/b3f07ea7d0e4486d9c955bc857f7915c : SUCCESS in 44m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3699e699376541a797544f8f126d1414 : SUCCESS in 48m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/152f2fb8883d4aea8f8126cd7811e609 : SUCCESS in 50m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d0496de4f6f94656bfe4612ac4210540 : SUCCESS in 49m 43s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/05cb9646dd7a45d2a0ddd6af403a1236 : SUCCESS in 19m 35s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/708e873cb7e14cf1a2d9ea1147edb79b : SUCCESS in 21m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b2f4f3e402f84ed3aa0e7e08d66816cd : SUCCESS in 21m 56s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/94a47ed189744b16b97ecdbe7229569f : SUCCESS in 4m 42s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/29503ecaea1b4f09b235e838054c2cef : SUCCESS in 18m 57s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8f1d7aa32d2c4a4c93a94980f652e55b : SUCCESS in 21m 55s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/07e268edaa524ea9b70892b843658849 : SUCCESS in 28m 30s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f0c1a5e2a0ce4f61a4afca67d79e3acf : SUCCESS in 9m 55s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1c24402401e24e9a81314630270ae62f : SUCCESS in 21m 07s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/dfe80b9514d141a1be46f8291cdf736a : SUCCESS in 7m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a82cca9faed745e1bab49b416c2d9c85 : SUCCESS in 6m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/844c76da3133419890f3657220551952 : SUCCESS in 17m 21s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/dcb64823677e41ee94f4fa1fb0c57da6 : SUCCESS in 6m 56s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3240cc8b43be4806b4ceec7e0d60602c : SUCCESS in 58m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8974436741f34731805e393bdea4e782 : SUCCESS in 31m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4102824706284f1792021b2c842b9940 : SUCCESS in 45m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/02e9602e5e554fda9779fbd84bcf0934 : SUCCESS in 19m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad5698c6fd994c98a5b2209738cd10dd : SUCCESS in 5m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f3d77c94e01a4e5d9f65c86f9a7da694 : SUCCESS in 17m 16s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1e431c6632d1445982f1628dc665574e : SUCCESS in 56m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/248b58f1574c4c839e915a9a89423462 : SUCCESS in 53m 49s","accounts_in_message":[],"_revision_number":44},{"id":"4a40126889c46880c4672ab1869d8e9a685c423c","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-04-08 15:19:19.000000000","message":"Patch Set 44:\n\n(25 comments)","accounts_in_message":[],"_revision_number":44},{"id":"74e0ff06de5c860750c0cf0fa0bbc380200423cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-09 13:56:22.000000000","message":"Uploaded patch set 45.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":45},{"id":"9dcb7449251f1ab5c3f385a87fe571ca2b72c5a5","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-09 14:00:17.000000000","message":"Patch Set 45:\n\n(1 comment)","accounts_in_message":[],"_revision_number":45},{"id":"091c16a312f4f4ec7183efd8bab3bf13073db285","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-09 14:34:02.000000000","message":"Patch Set 45:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd732cd550674915b39fc72bfa74d0ab\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/746641bd0c0747d9bba519e066f5742c : SUCCESS in 35m 44s","accounts_in_message":[],"_revision_number":45},{"id":"4f98d28c34051ffb66db2b2fbe0bcd11a166483d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-09 15:01:07.000000000","message":"Patch Set 45: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8eddca8ef8a8472e817e5ab184b3cb1a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6561f3ebf274c5e88c5493b832a7005 : SUCCESS in 8m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/50b7cb82e9974fc5bfcfcf42732ed3e3 : SUCCESS in 45m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/dc816b4d74ca498ab22aeb8c025c556c : SUCCESS in 52m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5e894e79aa994e5f9b4934e0be5f5969 : SUCCESS in 51m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ca34662a0a1442799769b91bfa869463 : SUCCESS in 59m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/95eeb534aa014c74bdca8b47d6aee022 : SUCCESS in 18m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/06361c7ee19a4bb49df511eb24b7af08 : SUCCESS in 18m 46s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6bcc7d0b48614cdeb3d9cfacf43ae671 : SUCCESS in 20m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7d215f641f2947c385b00697a907554a : SUCCESS in 5m 07s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/69d0b94fb46b4b07a8bb6a702404bea5 : SUCCESS in 19m 50s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/84e7fa2068dc4a24ad0f0338c3a3f7b4 : SUCCESS in 21m 45s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a7f61dcb350942ce849a1b6117d4cae3 : SUCCESS in 32m 22s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/07829f2258a44434931799cf1d8bcfe6 : SUCCESS in 8m 02s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/91688588b6ee44cabce3baabef82b99e : SUCCESS in 16m 03s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c425b448503a442aafab8b0e722eba2f : SUCCESS in 14m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ecbbfacf99bb4fe7948154ab5cdb1ae5 : SUCCESS in 8m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/45190eecce514549b99c40318a3fdb12 : SUCCESS in 7m 24s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9f58337897cd474894a0e3c3c3b1b4a7 : SUCCESS in 4m 30s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/99c21581da5e455c9b51ca00398e9e29 : SUCCESS in 57m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0b658020d49d4f69860c0a474eb24810 : SUCCESS in 39m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3b390fe2466d4156a7e4854f71607989 : SUCCESS in 51m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a08d71e79fe945749c74c28e75e436b5 : SUCCESS in 22m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2dc540c310e54807bdcb0076d6a2f834 : SUCCESS in 5m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/deff69e0e3fe49619fcbb2fb8987346f : SUCCESS in 8m 29s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/688f51f460f742c0bab2c6f313132e15 : SUCCESS in 52m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6b313511acec4056aeadbcb4ef28671a : SUCCESS in 54m 48s","accounts_in_message":[],"_revision_number":45},{"id":"4488ff869d15774a446956ae5f9a7e9e8c06154f","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-09 16:01:46.000000000","message":"Patch Set 45: Code-Review-1\n\n(9 comments)","accounts_in_message":[],"_revision_number":45},{"id":"e7dec8f54b76dc99214d53fbbfa9b0a89afa7556","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-04-10 21:23:15.000000000","message":"Patch Set 45:\n\n(8 comments)","accounts_in_message":[],"_revision_number":45},{"id":"ad2d88c31ae9199a258fd13059acaf741a38a6eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-04-10 21:23:34.000000000","message":"Uploaded patch set 46.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":46},{"id":"f06eba65b1536a7440088b3cbfa8e25d97424c64","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-10 22:02:52.000000000","message":"Patch Set 46:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/703c15cad89944a18c8beb7524ba232b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9f9f92cd8ecf487e83fca20258e7ed85 : SUCCESS in 36m 57s","accounts_in_message":[],"_revision_number":46},{"id":"6257ccc36bd779b2cd1bc33d72cf286b7d8ed69c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-04-10 22:13:33.000000000","message":"Patch Set 47: Patch Set 46 was rebased on behalf of \u003cGERRIT_ACCOUNT_36606\u003e","accounts_in_message":[{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"}],"_revision_number":47},{"id":"10f881b6667406ca0dc64e4d29262c2901906209","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-10 22:53:44.000000000","message":"Patch Set 47:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9ed2b16151f4311b486459ceb889349\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/17e42c0487b640e2a34e700791e3cb96 : SUCCESS in 38m 12s","accounts_in_message":[],"_revision_number":47},{"id":"71c8988b826756f953ea85e77a9466e056d7a241","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-10 23:19:40.000000000","message":"Patch Set 47: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1bd97ab39fd460ca00268da661de280\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e152175251140acb1f8052d147593f2 : SUCCESS in 4m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/fce759b21c1b4ac59a56af7ea52b21f8 : SUCCESS in 40m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/606e787dfbf744a996e279bafd324b8f : SUCCESS in 43m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0c55d05d7a77469688b2a6580cac2b42 : SUCCESS in 30m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/105ef3145ae8476185ddfe36d45f90bd : SUCCESS in 49m 00s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7849d89a32d545e3b82db6a1d53e1d50 : SUCCESS in 11m 41s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/056943fc389f412ea49faa3d39d4c18b : SUCCESS in 10m 24s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7330d76ff13e4998a4731815d4db4f1e : SUCCESS in 12m 13s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cd9efaaad7b54e8aa7b3720e9f36e023 : SUCCESS in 3m 15s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4a7c3c8a48bc45409a6e9b389d81daa8 : SUCCESS in 8m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d4540d0a4f3b49ea81e4939ed28772d7 : SUCCESS in 7m 51s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/57bef2b6e75543f8a5acdf4f4c2a2e9e : SUCCESS in 12m 02s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3a8b58822e1f4c4fac9472b87d572567 : SUCCESS in 7m 43s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5a2aa7d59bf34da58e70756da61457a2 : SUCCESS in 7m 54s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0fd3d90b6b9d491e8a1755f569832c41 : SUCCESS in 13m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a8f20f68b1434370b4d77129003052b4 : SUCCESS in 5m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5e5f29fdb3f34b7186564a362849a6a1 : SUCCESS in 8m 10s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ef2884af543d40b68832e040d41ca61b : SUCCESS in 5m 54s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e3f7b016df664382b4752fb3a4410627 : SUCCESS in 48m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/55df13a16c004fb98a45e15eb0af8088 : SUCCESS in 17m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5e03b53b85fb4b98aad0f31d93abb252 : SUCCESS in 31m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5572ff82155747bb95daf799ba035c6a : SUCCESS in 16m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4834d28e0090466687a48f9d26ab303c : SUCCESS in 3m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3e938136bd6949c4858fe382cea8490f : SUCCESS in 15m 21s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ecfb5c03f4164abab8ef23013554d7d0 : SUCCESS in 50m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9d48f2b4e00449178bd482bdae73f414 : SUCCESS in 1h 00m 39s","accounts_in_message":[],"_revision_number":47},{"id":"ab0c59e03d16f918a0f9aec15e3c22f8b397a0f8","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-11 13:09:54.000000000","message":"Patch Set 47:\n\n(1 comment)","accounts_in_message":[],"_revision_number":47},{"id":"841137720f7404b4c8919d9b5924062f7e2cb122","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-11 13:10:10.000000000","message":"Uploaded patch set 48.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":48},{"id":"1e10163171d935299c2fa15661c49010d887378f","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-11 13:11:14.000000000","message":"Patch Set 48:\n\n(1 comment)","accounts_in_message":[],"_revision_number":48},{"id":"bfa531764527c9b90c99dc60e583a78c4f484aa7","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-11 13:26:38.000000000","message":"Patch Set 48: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":48},{"id":"174ab89f36d23cd6f33e8cc0d20bf47400e25d47","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-11 14:02:42.000000000","message":"Patch Set 48:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27aff98b4e364e748e9c5769fdfd0773\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d2df560003b74e6d9735be18053a3255 : SUCCESS in 44m 57s","accounts_in_message":[],"_revision_number":48},{"id":"3f0b35472e502cdacd70b46f9f80ac00026a7ce1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-11 15:04:58.000000000","message":"Patch Set 48: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87a9c6fdb1af456db3a396783da0bfe6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6f61e0c41de4335bba852f41220c061 : SUCCESS in 12m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/e19df516d88b4a63bceeaff209762264 : SUCCESS in 51m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/152441b8d4ac4a7a8a22cd15d139311f : SUCCESS in 1h 15m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/395f2617d6df4bc5b761bd348630561a : SUCCESS in 28m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f8d272f93ca3428fac789189d68b6322 : SUCCESS in 54m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bca79acf73704633aef2dbab46bbceb9 : SUCCESS in 20m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ba653cc23a854ec794a338d2c43fe548 : SUCCESS in 21m 39s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/03e5c92bb5b04d379ebdcd2c0db915dd : SUCCESS in 20m 11s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a6f740633cbc482382c91b977f0a7c04 : SUCCESS in 5m 10s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b56eb7b956cc41888700f21c4d932b58 : SUCCESS in 17m 52s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fb454b85ce0b48c3a9890035981828a9 : SUCCESS in 20m 11s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/19723f1962604f4fa5061f754d99b1be : SUCCESS in 18m 38s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/127bcf6ede2846d5baad503f9dabf857 : SUCCESS in 16m 36s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/735ae1a29b954dd29e0912096842fe62 : SUCCESS in 14m 12s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4137189325e84d04bebc8a0d85422fda : SUCCESS in 16m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1e0c4bb90d2446ceb7eab5ae9f1009fa : SUCCESS in 6m 45s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b9ca58d000d04093a273643f8856ad47 : SUCCESS in 13m 39s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e65b597a190543e294b2e470035581bd : SUCCESS in 7m 42s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/babeec0192c6410793bc8ac8723fa8c4 : SUCCESS in 55m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/176a2826d6644dafbfd75bde69fcb7c8 : SUCCESS in 39m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9e4fc1ff80f2451b9fc55c9b5c2eb910 : SUCCESS in 43m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8d6d12b5e615405a968d68a7a6f217fe : SUCCESS in 20m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0967c5b80774826ae5bf6209ff6ba91 : SUCCESS in 4m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a825356b6f264f96893165334a9973b4 : SUCCESS in 22m 01s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/104fdbe23b644606b8cb8df137cbec91 : SUCCESS in 54m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a09de568021b40caa42f1b571d1913f6 : SUCCESS in 49m 37s","accounts_in_message":[],"_revision_number":48},{"id":"ce4984e34484483f394653e1e24a50465d0738a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-04-14 15:25:24.000000000","message":"Uploaded patch set 49.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":49},{"id":"1c4f19603f58b72923ec4c74947ee52de7fb6e11","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-14 16:17:37.000000000","message":"Patch Set 49:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1459f52511e4e2c9f491b2ec3dd1576\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0d8981448e624216baa3ab2a8157fc31 : SUCCESS in 50m 07s","accounts_in_message":[],"_revision_number":49},{"id":"dd3345a7ed9bf14cbbf1af84d11ada10b79ce214","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-14 16:30:07.000000000","message":"Patch Set 49: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29c19a1a4875421793e4b60699a37969\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d36f53caa91b4d049922d1a9458f288b : SUCCESS in 5m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/003af5783a904f1a842094187f94fefd : SUCCESS in 1h 00m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/33d6e70cb1c742589ce4ca71d88ea406 : SUCCESS in 58m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b2adf932566c48adbef2a319513127da : SUCCESS in 36m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/70da5eb295c6498a9a34b11a81d205ea : SUCCESS in 57m 28s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2cb783224a5a4c8885d79b165d4dca06 : SUCCESS in 11m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2026e73c14bd45568541aaf5c3dc7d5f : SUCCESS in 20m 32s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fb0a1c1a21c34ec39e9e3ddb9d4e5505 : SUCCESS in 10m 38s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7099b728d82c45cdb0b676f79c8b3e7d : SUCCESS in 5m 25s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2c2c3a528e194f63a3a1ef9ef1384897 : SUCCESS in 16m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b9def5c52e33415185fc6c6243bd5857 : SUCCESS in 21m 05s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2f20839ec0bc4694b12e518ff6c35d2a : SUCCESS in 17m 59s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ecde4af50d7d404d86589125401f54e0 : SUCCESS in 14m 08s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e5a2e738f863421280445bfa1e44ee64 : SUCCESS in 15m 11s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3ec3aec3fd6f44388838a5130c1307b0 : SUCCESS in 15m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1aa76ade41424af2b0c6c49d293bc352 : SUCCESS in 6m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dd693d94a48846d1a5eb188b6a54772d : SUCCESS in 15m 10s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9a4555a53a5b4791b97793a71d4a2951 : SUCCESS in 7m 54s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3f4b1f4221b147ba9e5da32634f2d068 : SUCCESS in 56m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/591ec0b5c6e149008b6e69d59c17c786 : SUCCESS in 45m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/08048d55020d428f9dccd9042c01d322 : SUCCESS in 42m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c81ccab49dc04bbcbeccc6d6ba079a78 : SUCCESS in 20m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/044e08ecc85c4cf2888ac54f519cbb28 : SUCCESS in 4m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ae6564b3bdda4f51a9fe86582244ddf1 : SUCCESS in 22m 50s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a6183d9cbacc4678b9059efc1a37ee8d : SUCCESS in 38m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ce39463b4cce42cf9f8febba4d6fd20d : SUCCESS in 42m 10s","accounts_in_message":[],"_revision_number":49},{"id":"102a72bdcf5c3fe22cf13f8feaea9a43d85ff94e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-04-16 19:41:00.000000000","message":"Uploaded patch set 50.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":50},{"id":"e79caab5108066d50bcf64527fddbaba88e104f9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-16 20:24:06.000000000","message":"Patch Set 50:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84f0ba80c381492caa10382564d5599e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/41336d589bdd450cb475ecbe0ae9c4a1 : SUCCESS in 41m 24s","accounts_in_message":[],"_revision_number":50},{"id":"b62844c9dca53b7a45365f58bac831e07dd91273","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-16 20:47:01.000000000","message":"Patch Set 50: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb15d4b8a7af47cc80080b6c7efcd12f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/455dbcb51831430ca9cf43224bdfd5b1 : SUCCESS in 11m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/9dd940f9c7d94fb5aa291b3941a05ef1 : SUCCESS in 56m 58s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d9c5c0c962d74eab92d19c2115f3df78 : SUCCESS in 57m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a464376bdb4a4e6badccd3e9f8bf8313 : SUCCESS in 53m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0a052ccb4765422794781353778bc785 : SUCCESS in 58m 05s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/45334da9fc8041aea95bd06d3655730f : SUCCESS in 20m 00s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ec58a3004ac64cdf89db1d685809d74e : SUCCESS in 12m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8e0bd9ba89d94e868c190af2983fbecb : SUCCESS in 10m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3c7b3945e07e493a905a533777b5c56f : SUCCESS in 4m 35s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2bc59bd9bb12476fb77e4c5e608eb552 : SUCCESS in 15m 12s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3592794d000449d0a25a7e9578fde3b0 : SUCCESS in 16m 42s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/34e66b60b32a42249557913bec5a7083 : SUCCESS in 16m 05s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a21fd4018f704a6f8f240e13374aac5a : SUCCESS in 11m 52s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b690c11788124aa598a981e18ec226ae : SUCCESS in 13m 29s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c2cc1542a92346879987a0966d4619aa : SUCCESS in 14m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/92453e7b55f740de82e60d730780a7dd : SUCCESS in 5m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4b3e60f19cf94a87a540380d1a7527c6 : SUCCESS in 16m 14s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/15b64b93c11c4bc6b603f89e6c9aad2d : SUCCESS in 6m 07s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/07a11d3eeaa74790ab19deaf6c3d29f1 : SUCCESS in 1h 00m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c9eba576e23e46678c157565ac5ff384 : SUCCESS in 36m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/984f6ca6ffd64d61a93668589cf44fb9 : SUCCESS in 57m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/92d0e267f6e5474bb45162a3a92db4d1 : SUCCESS in 19m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46b2a78775884d7cbe47dbdf635d704a : SUCCESS in 3m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/264ae6f1a0754a0c934869fb0786d8bf : SUCCESS in 19m 22s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c22edb45c1154e6eba02efbb4f899ee9 : SUCCESS in 48m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5e342db277e04c11a0d61c7934a195b8 : SUCCESS in 55m 17s","accounts_in_message":[],"_revision_number":50},{"id":"e63538f8eafb1d389d655c7d6b04b40cc534a8d7","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-17 15:43:31.000000000","message":"Patch Set 50: Code-Review+2","accounts_in_message":[],"_revision_number":50},{"id":"ba9d4897f8505ad697cf0853574b0d9d27738bdb","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-04-17 19:23:18.000000000","message":"Patch Set 50:\n\n(1 comment)","accounts_in_message":[],"_revision_number":50},{"id":"e03c2933822e5dead10d4a0a1732ec9292052b63","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-23 16:18:38.000000000","message":"Patch Set 50: -Code-Review\n\n(2 comments)","accounts_in_message":[],"_revision_number":50},{"id":"fc16297369faa2b00483545b4d826a1057eff252","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-04-28 01:18:35.000000000","message":"Uploaded patch set 51: Patch Set 50 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":51},{"id":"50ad3fabba27e1f13841c371ba91ccdf81658359","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 01:52:24.000000000","message":"Patch Set 51:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39b4bb19fb3548ac8fe33427b8f34457\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f73361c9412b4af0b1d6e166ec923d76 : SUCCESS in 31m 31s","accounts_in_message":[],"_revision_number":51},{"id":"f2052e1087c69569e24299efe14970bf2918ae00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 02:30:48.000000000","message":"Patch Set 51: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37f2755dffc2461ab9927004be7e0818\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6aa3aecf253f44388f127c32e75250da : SUCCESS in 5m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/1a0975cac3104674acd4f55479784292 : SUCCESS in 31m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cd728ddc46794d05ae2c3ca311ff9668 : SUCCESS in 28m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c0a8e465285e4a86878eb2e9a7105519 : SUCCESS in 38m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/57dd43523bee4d7c937f6c6b2983e9cc : SUCCESS in 58m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e3ab30da3d7b4d7ab32f3a59f0124e3c : SUCCESS in 18m 22s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8b4b1cfb1b164b4a9f124a0a05a752f3 : SUCCESS in 20m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0ccf1529293242e98901c19f37c82a7e : SUCCESS in 21m 32s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/04a1714c18154bfead403f14262b8f19 : SUCCESS in 5m 06s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6c181f73d2084c5f93daf8f380e44b33 : SUCCESS in 16m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/64faf8b705f945ef8d07f8ccad8c7a71 : SUCCESS in 20m 11s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f80a4997b3714bcf8b616a92464d84da : SUCCESS in 31m 39s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/2a6fbed242db402faff88320bb9d3612 : SUCCESS in 29m 12s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b6a78d1a81d641d49f3ee6cbb68510c7 : SUCCESS in 14m 13s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8aecf2870c3e49909f49386a1cddd4e1 : SUCCESS in 15m 24s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d524394533c44c9795bf091e811ab04e : SUCCESS in 11m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2c95a868ca9040cc9068b456d9f424b5 : SUCCESS in 6m 31s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1c264b048a9b4785b9282d7afb0add6e : SUCCESS in 8m 23s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/8b494f0d3b7a47ae946e627b31b694a7 : SUCCESS in 6m 42s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/57f14ee9afcd4864af26528687725c4a : SUCCESS in 52m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6cf3cc7a22bc455dbae2698c787e3d15 : SUCCESS in 20m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2707170f9f1242959d9a47d7257b87f9 : SUCCESS in 41m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/125329571946493698c96ace1caa5633 : SUCCESS in 21m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6d289b1a18a4c698cec15a161b93004 : SUCCESS in 4m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bf5dffa99feb427bbeea2a0b5a4e8556 : SUCCESS in 22m 46s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6842f53a7bd94d16ab3bfe629c5bfc6b : SUCCESS in 27m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a5a62ce0f79f40fda2a038fbc44fe425 : SUCCESS in 48m 26s","accounts_in_message":[],"_revision_number":51},{"id":"eb6212c69144c3bd632e87e36d7f61f02f78083d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-04-28 19:36:03.000000000","message":"Uploaded patch set 52.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":52},{"id":"8e3222ce7e02adda85429298add949edb17fc9f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 20:46:22.000000000","message":"Patch Set 52: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/112e99c1000c40658105aec0cb704cf0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/759f17f0ef8d4ae388e73670dceee0a4 : SUCCESS in 9m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/b612e9a9e24b411bb93b98c42d33e93a : SUCCESS in 57m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e94a71577deb489ab203cc8a0827231c : SUCCESS in 1h 01m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c7ca8ff730b44ea9a3b746a3f549bd1b : SUCCESS in 52m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4b42546117c548e7b575c81f2744fe6c : SUCCESS in 50m 47s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4d1bf86678b44086a479eb316fa6b85b : SUCCESS in 10m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6c905932d5f64044884ce1258f3cb0cc : SUCCESS in 19m 54s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a16304a882a54bfcaa4626da05f4acea : SUCCESS in 20m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3f8285d67b764ff7b157e75c99614195 : FAILURE in 3m 44s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3d06cbe42b794f1bb4478303ceafb8ac : SUCCESS in 19m 36s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a72da9d379b540b7839de6325a8fa1f0 : SUCCESS in 23m 35s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ef76c2eac54c4fec88c001e0d68b91fd : SUCCESS in 27m 45s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c1cd9b0b06ce4f968a73defeafb33916 : SUCCESS in 24m 56s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e9cab0693a1046a491461b145fbf11eb : SUCCESS in 7m 30s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/dba8503669a94a50a4d5951b28de10c0 : SUCCESS in 13m 18s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/98008e98994846c0b6af013229b06229 : SUCCESS in 14m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/44422cf2427c426b80647dee7c79c756 : SUCCESS in 6m 07s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3aba9c27e2a4458e8b56b9e78556db68 : SUCCESS in 14m 28s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/bb875f505d6c4c50a10e9f6d60b57dc4 : SUCCESS in 6m 59s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e50797725976467782c703a57eb38d6b : SUCCESS in 55m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9cd822987951428aa5720605ec28d572 : SUCCESS in 39m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/42a904234b284496a390588b7510fbac : SUCCESS in 45m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9778f94aa26e498e852ea2140fd44712 : SUCCESS in 20m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b84352bd5aa04dd6b65ffb07bc3b3076 : SUCCESS in 4m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6b5e8084dc164f71a36fc3832b285b8b : SUCCESS in 21m 01s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cf95c043d09b40e5b510374c13d8fad7 : SUCCESS in 51m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/415bcd6bb88d41a086dbd02b71d9e780 : SUCCESS in 59m 58s","accounts_in_message":[],"_revision_number":52},{"id":"fc755cf165dbb7389d453915c9778dfe8102fdcc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 20:57:41.000000000","message":"Patch Set 52:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b1c0b6f23cda47dcbd257b9cad0c0405\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c6b56b8150324da185001f5a831c5f4a : SUCCESS in 1h 18m 25s","accounts_in_message":[],"_revision_number":52},{"id":"51489088e04b904ff16a3321ec9c4ae997f0201b","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-29 09:21:12.000000000","message":"Patch Set 52: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":52},{"id":"684482e897bede2f511fe80685094eccd256f546","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-05-14 08:18:44.000000000","message":"Patch Set 52: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":52},{"id":"371f9ac72f0b81252fbc46b4ae7453aef48bd280","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-05-15 03:01:48.000000000","message":"Uploaded patch set 53.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":53},{"id":"86ac8d205ce17d983670f4636d5acd8d29485b28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 04:04:46.000000000","message":"Patch Set 53: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3bc986a928954fba82db10c95eb56cce\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1ac418c117a4a44b3d2695042b2a296 : SUCCESS in 9m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/33cfcb5244004f169b6daeb1c73af928 : SUCCESS in 49m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/52d9ca0509ae41b38ce5b225dc53ddf3 : SUCCESS in 58m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/dce7434369484a6a96cda0678918e30b : SUCCESS in 56m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ca98d871a61449ada8d57ecb19487b48 : SUCCESS in 55m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/06f9220afddd422989c347c03d8e66af : SUCCESS in 10m 03s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e7b0f8458cc44089b7d06ecbedd5aaeb : SUCCESS in 11m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3a3cdd326ee1440eaf7b0a406b79430d : SUCCESS in 20m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/631d0cee574d43c883ec48dee9e07c49 : SUCCESS in 4m 50s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0e426835551e4f649f5c98e8632c8c80 : SUCCESS in 19m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d404b1373fbf441b91738c189b1a73fd : SUCCESS in 9m 08s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4fb865c63628435d9af38f6175f4b347 : SUCCESS in 35m 30s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/bb6b5d9c8d364d9dbe245538f9a087e1 : SUCCESS in 30m 55s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/504043fa18c243aab337e8e088fff8e5 : SUCCESS in 7m 40s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8f4c3070e000471c8e5a0b2da8bf4bad : SUCCESS in 16m 18s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/826917ab8c2f444f855b237b843489d9 : SUCCESS in 17m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0f317d9a9ed3476a8e192b006ee4c633 : SUCCESS in 9m 47s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b381697b8bbd493e9512151cec0e4010 : SUCCESS in 7m 39s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/ace696c2a94247ab9ec2712f27e17b7d : SUCCESS in 7m 10s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8221fb1547394068814ecb371cc5c6e5 : SUCCESS in 38m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9f16c6dff36142e1985ca7c400096692 : SUCCESS in 43m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/691a794930a44b6496c407663f35de2b : SUCCESS in 40m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dfd54f1ae94d460a85098d8523e66b93 : SUCCESS in 19m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b2d12f9889646f19294fae2bb1a9f1f : SUCCESS in 5m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/536cd2bf14344697b2a68adb667c2d89 : SUCCESS in 20m 01s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8dc2bd310f0647448d29db14caa0606f : SUCCESS in 45m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b73a2030d86d40c1829316595f69819a : SUCCESS in 46m 45s","accounts_in_message":[],"_revision_number":53},{"id":"4ea48532355709469e6f626bf31f28ce931a91f4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 04:07:10.000000000","message":"Patch Set 53:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22c12cbe166f4af19950afff1674a30f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/675adb20cf544b1c998801880de122ca : SUCCESS in 1h 03m 25s","accounts_in_message":[],"_revision_number":53},{"id":"70d4f09a061ef24b6858db03cc23f5f6d4cad001","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-05-15 09:12:23.000000000","message":"Patch Set 53: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":53},{"id":"cae75b407907ba640a579f93d908094f1312cbec","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-05-15 15:33:20.000000000","message":"Patch Set 53:\n\n(1 comment)","accounts_in_message":[],"_revision_number":53},{"id":"3d452d271659e809671de93cead9734c11f7fe1e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-05-20 15:56:54.000000000","message":"Uploaded patch set 54: Patch Set 53 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":54},{"id":"fed81f96aeb6caa0f36117001990591e7fb149c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-05-20 16:49:59.000000000","message":"Uploaded patch set 55: Patch Set 54 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":55},{"id":"2144577ba2587b037412940c8ca7670e5ddf250e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 18:09:09.000000000","message":"Patch Set 55: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6adaa5675a4746e6908b0a9f56e5a5c6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78bbb8233fe0451aa783a329c7c79615 : SUCCESS in 10m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/c1217785a38e41eab63740f7e06d448b : SUCCESS in 51m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f8687bdf4b5b4881b861116e0ed75d6b : FAILURE in 51m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ccf2ce531b094d1f89c6269ee8570ccf : SUCCESS in 40m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/929e5927b94249a981829e6ec41e1575 : SUCCESS in 53m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a9532255b2b741afac449fbb26ddd982 : SUCCESS in 17m 53s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/47e45f714dc34c12950dfd49df67a1f2 : SUCCESS in 18m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/93cd83c8a4eb4d73a1d5c3330044d860 : SUCCESS in 18m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/90c099211a16467ca04a3f3b506721c8 : SUCCESS in 6m 07s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/611e1ed750e2464790f2052f87ff39cc : SUCCESS in 22m 18s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/daf746bc4b7e4017a691d874e7b09ef6 : SUCCESS in 9m 44s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/553f886b7e1d40f2ab5228171530578f : SUCCESS in 21m 17s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/8b4b55f101164e17803246b02c952d44 : SUCCESS in 24m 28s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/dde11597a7ff4ac18aa3203f5fa00a1f : SUCCESS in 8m 49s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/dfe2f02497fb4ab185c7b17acdf1a8b7 : SUCCESS in 13m 09s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/dd003ae256e140cb9165db0574bcb065 : SUCCESS in 22m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/60fedf5d6619422ea7b4cff6dca5c1ff : SUCCESS in 6m 32s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ba29897d9ceb49d195edbefa4e4dbd83 : SUCCESS in 15m 19s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e885a5645b2d431583604016d7a13375 : SUCCESS in 7m 03s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/feb6431f661e44488512ea1d65f15910 : SUCCESS in 50m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6e72aa762ebb4c1cad2d2fd097b80a5f : SUCCESS in 35m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fbaea7353e4147a789d3650665ba7612 : SUCCESS in 35m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/15e297e1b203445fbb5a341c24b1eaa6 : SUCCESS in 20m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57c318522e264256854c339d79a9ca61 : SUCCESS in 4m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c86ca4e38b464d79b14a060e7c45c61e : SUCCESS in 15m 17s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/132ace545a644169ba0fe1f6267e9918 : SUCCESS in 57m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9eac2ed280c048c9b4b48626d7ce80f7 : SUCCESS in 49m 00s","accounts_in_message":[],"_revision_number":55},{"id":"20e9a7d682e181d1e0a551b03d5da498620fe751","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-05-20 18:10:15.000000000","message":"Patch Set 55:\n\n(1 comment)","accounts_in_message":[],"_revision_number":55},{"id":"1171f71e2395672a36b98d59c3fc9ee087ea078d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 18:54:35.000000000","message":"Patch Set 55:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/576bbde6b7b641e39b3f3e4ba58364e3\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/831da3745a28495f9b190a989d0767ad : SUCCESS in 1h 01m 53s","accounts_in_message":[],"_revision_number":55},{"id":"5dc5ca465adcb3919a4e6068a52e045db507c56b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-05-20 19:27:06.000000000","message":"Patch Set 55:\n\n(1 comment)","accounts_in_message":[],"_revision_number":55},{"id":"17e9dae0534b23db3b3cc120c87914b21f4a82aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 20:49:32.000000000","message":"Patch Set 55: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6f0afdaa3534c4a91542f7280b91ba1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/658cb9843f104cb5aca819b115d8ef08 : SUCCESS in 9m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/a6564b38cafc48bc8160bf94da63fb15 : SUCCESS in 57m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/27f93974863848b4b822c10c83b17f3a : SUCCESS in 55m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2b4aff4da9064109918688e30f4debef : SUCCESS in 1h 01m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/baaf5f64aa624acabe49b98f649d55db : SUCCESS in 54m 51s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/551fde65358d4f55a026bc22afc45c79 : SUCCESS in 23m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/363e55ef606c4ec38378491b88e9c750 : SUCCESS in 10m 40s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9622269e81d94810b73108c5e44e909d : SUCCESS in 22m 11s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/97d1bb69e15646528743626ddfcb83f0 : SUCCESS in 5m 04s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f1015cae0411434280508e09ce450b58 : SUCCESS in 18m 57s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/dc749e6d35a44ddb88c01d58223dc847 : SUCCESS in 8m 09s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ad949d300f3b4b789250a3b111c61467 : SUCCESS in 18m 26s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/b9a2f429a8864eb594cbd8e65394fd7f : SUCCESS in 30m 23s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/91667bf4ec734807b0c1aeec08fd3e1b : SUCCESS in 12m 47s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5c04753190ba4165b4024908f53c7023 : SUCCESS in 17m 27s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/bb7cdc29672a4ed18e16502ddda9ce1c : SUCCESS in 13m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ffdca907758a486e8151a7e303c9d875 : SUCCESS in 5m 56s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c03f39c1afc941beb204b4ed20160f51 : SUCCESS in 16m 32s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/a70b93e4b67848068d848a8b049007cc : SUCCESS in 7m 10s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a496cab316374be4b4d343c25dbed5dc : SUCCESS in 34m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/587eee98dcf6448eb1d1d14c3d5c5aa4 : SUCCESS in 36m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d6084965eb7a4c109646e93a7b5e9df0 : SUCCESS in 40m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4c8351203b8b43908d036010d1f926e3 : SUCCESS in 20m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/165cbc95ee1045ff891226c37f8f30e6 : SUCCESS in 2m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/360e103b06fa48a9ac13c2fab783bf08 : SUCCESS in 19m 20s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0dfa683a18304908af3b80e03b5abf63 : SUCCESS in 47m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/095a3db18eb646ccb0330c4cc5d2394b : SUCCESS in 1h 03m 40s","accounts_in_message":[],"_revision_number":55},{"id":"06dcf747688cc8aa38018390cab449f79ac9c50e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-05-20 21:05:00.000000000","message":"Uploaded patch set 56.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":56},{"id":"03adfdddc41810e4039505e142b5014e6db44d8f","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-05-20 21:06:10.000000000","message":"Patch Set 56:\n\n(1 comment)","accounts_in_message":[],"_revision_number":56},{"id":"c1afed714ae7eff2c2e1068d7e767352c1b3f321","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 22:58:48.000000000","message":"Patch Set 56:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f2cb1bd64524536ae3981153a14d541\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b2c57492f6c54f86a161d6227303c6bd : SUCCESS in 1h 19m 14s","accounts_in_message":[],"_revision_number":56},{"id":"249e587c6fde897397fc843fb4861b8191acec1f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-05-20 23:33:46.000000000","message":"Patch Set 56: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":56},{"id":"3998d94123dfb0ab5baa8d44cdc51b62176840e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 00:12:57.000000000","message":"Patch Set 56: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ae90e619c86a414c81c17310a7464787\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1eb79ea4868d42428e38a8d98c86832d : SUCCESS in 9m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/73a597b44e8e4550801f77ac7d52c705 : SUCCESS in 1h 02m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4c89039f40f545be95569041ace9d6e5 : TIMED_OUT in 3h 02m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/af87b234c95e476d8d37fc729dc6eea0 : SUCCESS in 48m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e6ac1a0b295f4708a9ebbf893467a64d : SUCCESS in 54m 51s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/594ab818ed4e4bb8aeb975c5af9d5c43 : SUCCESS in 25m 04s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/28d187e45d234288afa35633bcd454e5 : SUCCESS in 19m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/db96b7381142464fa5f64e5a007f44a9 : SUCCESS in 24m 38s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cbdbe4ae7f634365a646618029f2d28a : SUCCESS in 5m 28s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7860d0f5fe954728bf935ea0cdcd12ab : SUCCESS in 19m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/24d603c4644f4e32a75d2563afff643f : SUCCESS in 22m 31s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/04df50ca297f452499defaebea377f84 : SUCCESS in 23m 20s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/25a40ae2242f477aa13191786eecc8bc : SUCCESS in 22m 42s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/126f8be810a74304b4edd82599a1a696 : SUCCESS in 14m 03s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d3c5024f03da4992b0e9672105e3fbf2 : SUCCESS in 16m 27s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/53e08fb416df4161a20b2350dd15d218 : SUCCESS in 19m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/89a99617ca234670974d269edc628576 : SUCCESS in 6m 20s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/937bc813398540a1b07622855dd131e9 : SUCCESS in 7m 25s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/c65feef58658476e84818199e71fafc6 : SUCCESS in 9m 22s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b7d107dc90a0421fbf6042821ce0cea8 : SUCCESS in 55m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bd15bd5d39114d738a97a3c2e10db4da : SUCCESS in 43m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/eb01bca99fc34fdabe44a92bfed19f71 : SUCCESS in 41m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1b92fef5a8304912bb333959ae6c7378 : SUCCESS in 19m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4fd4770db05435a9738c46fab5cda1e : SUCCESS in 4m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e28616e257d247ea8a265a5897fd962b : SUCCESS in 20m 07s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7dc5e6a89288457a82c054aaddfe2ffc : SUCCESS in 1h 01m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9790d92d8baf404d987e412a655c09fe : SUCCESS in 1h 00m 16s","accounts_in_message":[],"_revision_number":56},{"id":"cd978cea897aa18c51f9602af2ff524ec053c6ec","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-05-21 09:34:37.000000000","message":"Patch Set 56: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":56},{"id":"bcc61fa6406fd2fe87b584fb14f1345897b7f7ab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 10:27:17.000000000","message":"Patch Set 56:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/607db3cbeee94e9f9ad318e1a5d2ac20\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2227712813fb46be8d07750d60ebf739 : SUCCESS in 38m 39s","accounts_in_message":[],"_revision_number":56},{"id":"89f1813013235b9dfbd4fa5b42a7dd1662bcf05b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 10:42:24.000000000","message":"Patch Set 56: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ba208c46e8c4406bc7e62e1c3390c52\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2fe4c3238f94de3a9c915fd8cd3fec3 : SUCCESS in 9m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/7f394756adc14c619d629b06b421c9b6 : SUCCESS in 57m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5f4b7a174aca472c9414318f6ab4b3a7 : SUCCESS in 1h 01m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/df45d5cd1a9b4881a1c9fc6ccdab9678 : SUCCESS in 52m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/28d766021ce14691a4a249fbeb8410bf : SUCCESS in 59m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1e1ff4cc092b47ec9722155e4b2659af : SUCCESS in 10m 04s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/293ffe11728245af9bd36ab0bc852c90 : SUCCESS in 19m 43s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3215e635d33844aea7193722cc5ff915 : SUCCESS in 19m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1fa79a3fc7474ae28d903b56f9aa09a0 : SUCCESS in 5m 04s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5cc12b4521ee485da2d90dd595f06dec : SUCCESS in 21m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4c5907619c444308a04397d5cc9a8ea9 : SUCCESS in 22m 51s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c9c8de2fa8eb4b3fb7a976b17822e70d : SUCCESS in 29m 57s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/23c061ea53ce4bd786099cdabcea7d4e : SUCCESS in 28m 24s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/cad307ae2b344f6a8c1a00dc1e2bc4e5 : SUCCESS in 13m 27s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/15a29c25566340f1aecbb4161f32782c : SUCCESS in 14m 35s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/be3f096a1b9942c5b236fbb2be64d980 : SUCCESS in 15m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2fc3cd64346f44429049e73e82079e8e : SUCCESS in 6m 01s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6d4440ceb6cb4a99bfadfa036d1d247f : SUCCESS in 14m 00s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/76e2c7ee2b074214a70b13bfd52192b4 : SUCCESS in 7m 09s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b59598a1e41c42018f6303f350255e7a : SUCCESS in 52m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/37e04f79ea6e477e9e4dfb948e94707d : SUCCESS in 46m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/406230b3836645ce92bb6ca5fdd5eea7 : SUCCESS in 44m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f5516e5eb0e148d78f672ebab1380333 : SUCCESS in 19m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c93628a9ee74e088473dfe4098e1552 : SUCCESS in 4m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/afd03fa52baa4f358cc25295c24a41eb : SUCCESS in 19m 07s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/306640def46f4d66a01f710f4da9deab : SUCCESS in 49m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/72cdfa130ea14a68b8855e5ee019b76b : SUCCESS in 52m 17s","accounts_in_message":[],"_revision_number":56},{"id":"f25440a1e80b651f0676b1663a0cadfaf79e4e7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 10:42:57.000000000","message":"Patch Set 56: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":56},{"id":"61b5bb71770f7a5e75f520506d976108610310a5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 13:48:22.000000000","message":"Patch Set 56: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/daa45f34bff440dca51cc6c6867ad5d9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a751a00f091c489db93ed0e40835f94e : SUCCESS in 9m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/367438c8aac04c53ae04caa9667dea85 : SUCCESS in 54m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/faa68ca279a449e0a153a002a6ea3a44 : TIMED_OUT in 3h 03m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/27c9c8dd8f1a4cc0bf5bdd59c3076724 : SUCCESS in 51m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6ec5df3f3e6a47198fd740d6e188f4b6 : SUCCESS in 53m 10s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/181f2354f81947db8aa7680fb8c5bf3b : SUCCESS in 5m 42s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/eb9eb3afe5a947b5a840f43289b01cd1 : SUCCESS in 18m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1fa2467a900045b08f9d3a5aa1c5d03b : SUCCESS in 17m 15s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1713fe0cbcc24c90b1a1d64c970ef4c3 : SUCCESS in 26m 55s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f68fb55a0eff4975851d1523897d6077 : SUCCESS in 7m 45s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6a1df1c6612b47768df09df2f677ef1b : SUCCESS in 14m 31s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6562bfd0efa040639f185a43f8a1c569 : SUCCESS in 14m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c4dc196fea2d4206891948b19dc14654 : SUCCESS in 6m 53s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/2626185f28ca4a4094581c6ce841d818 : SUCCESS in 7m 33s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/88db0e92fede4fc88b46fb7b58cba04a : SUCCESS in 51m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ea0ad09e5f07463db8495454570d8738 : SUCCESS in 45m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0339bb3be62d49cab1ffaced128e0fd4 : SUCCESS in 33m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3724cf5807ff4b91acccfbfe557244d3 : SUCCESS in 19m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b57479ab3e5747a08cd6ae1ac418fd99 : SUCCESS in 4m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1d9dbde62e6c48eda291121e1a3b3640 : SUCCESS in 17m 43s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/292ece0ab7ff4a1395705962b6acf4ab : SUCCESS in 50m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/18ce68aabbfc4507a06f9e948d2a56f8 : SUCCESS in 50m 26s","accounts_in_message":[],"_revision_number":56},{"id":"0a4e3514518942630b5d2946f2669c65bad0d907","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-05-21 13:56:22.000000000","message":"Patch Set 56:\n\n(1 comment)","accounts_in_message":[],"_revision_number":56},{"id":"2baa957a338a86c8461fe35bd7e0cc1e399af04e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 14:45:52.000000000","message":"Patch Set 56:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6d7effe65dd4bf897d6d6eb0a7f81af\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0b1e2075f19442db939122e64638a369 : SUCCESS in 39m 13s","accounts_in_message":[],"_revision_number":56},{"id":"608e31c858b106b24bd96272112b68fbb067a6f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 15:13:58.000000000","message":"Patch Set 56: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d6f8f6f18a04364a39151e5ed466141\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42fc38b989a348628644ee383e74c4cd : SUCCESS in 10m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/465bd90077444116af69c0d503de83db : SUCCESS in 52m 46s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/85027150c2664e0f9031deeef43f1fae : SUCCESS in 58m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9b4205548b5048cdb03ed1fd7dcabd0c : SUCCESS in 50m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/077f8ae74f9f4f829b69a5293934cd93 : SUCCESS in 58m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dd017995d9b6487788baf3ba632ccc9e : SUCCESS in 20m 28s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b12f026640504979bf78967befa4ad54 : SUCCESS in 19m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/df18194b806040fd96773085b2a7092f : SUCCESS in 12m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/44f7439a550146589ed95e345e467583 : SUCCESS in 5m 11s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6d160bb916de4fb783d10140d066aafd : SUCCESS in 22m 13s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/08aac6a0c12f4ebdb668d455cce19f7a : SUCCESS in 20m 13s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4f30e39ce653495b958be07a1b887a04 : SUCCESS in 21m 11s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/65329635ce924f3299d6112271effe7f : SUCCESS in 25m 33s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/017600eb54d8430a9c0db0748fadc987 : SUCCESS in 15m 58s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1b3f9561324b4fda82d3aec522031c40 : SUCCESS in 13m 22s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b09bd3697faa41d5ac0546c5956ecc96 : SUCCESS in 12m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7b630f90fd12418bb30ef01be5ba8e92 : SUCCESS in 6m 12s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ee58d17597f241c9b5d733d7409de5f9 : SUCCESS in 17m 43s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/d5d5bfa96354449c9e103719aa32bb62 : SUCCESS in 7m 16s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5b132913204748daa8df45a40dc771a8 : SUCCESS in 53m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b279875972eb4d3abaa332b012c4de44 : SUCCESS in 43m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/82ff6f2960234b4498c12622abd7e842 : SUCCESS in 38m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b3593c4e0b884d6f9aeca779be55d5d4 : SUCCESS in 20m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/864366a8378849ffabc105f2f000d8d3 : SUCCESS in 5m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fcf6f43394d648319229e0ac3ae05875 : SUCCESS in 18m 44s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/55ea9d37e2c04a3e82d9a448ba3eac77 : SUCCESS in 57m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4e30d5f7d6934495ada8c27753383d8a : SUCCESS in 53m 29s","accounts_in_message":[],"_revision_number":56},{"id":"5591282e54b447c3abea8834bc456fe1f5eebac1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 15:14:29.000000000","message":"Patch Set 56: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":56},{"id":"1237dcbc85fb16a54bb6947d40fdd54bf799b95a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 17:14:51.000000000","message":"Patch Set 56: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9c4e08bf7774d309685c907d885ce61\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8605ee1504d42f080cbcc2f2d1251bb : SUCCESS in 10m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/7b32c678ac664ce9b79f4f4ba9dbb6d6 : SUCCESS in 52m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/363baa95359049eda5c2912440bc9ca1 : SUCCESS in 54m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0f3ae611cbc34a90b55d9ecc0fc180aa : SUCCESS in 46m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/37b4bdcba6fe4f8a9271230a6e459c41 : SUCCESS in 1h 04m 05s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/dbd8726fdf86440a8a6a4c69d196f4bc : SUCCESS in 7m 15s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f21542fc348d4d27b8465de85b9e6499 : SUCCESS in 20m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c77772b293d448f6ba51b51736578c0e : SUCCESS in 21m 03s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e38a23abb47c4bc69daa8840964a4681 : SUCCESS in 23m 28s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e412649a51504d52888634e71ef07e3c : SUCCESS in 13m 31s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b12323b730e44ca68fa3eaa4f2c66910 : SUCCESS in 15m 28s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9180a32ba9df48729b871cc6c11b4b8a : SUCCESS in 8m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ed1bffac72e14694880e3d0fc4998fb1 : SUCCESS in 6m 20s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/ac3409d629654cfe80f00534171de0f7 : SUCCESS in 7m 28s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5256a461d6f84f0987580bc151cb5d5f : SUCCESS in 57m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/282a312a24ea43058d3bb2f6713443da : SUCCESS in 37m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5c5b696c987f4c63b7b68696bb52f424 : SUCCESS in 43m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b5b18ed1b7e7406c9bbcd2871ace98a3 : SUCCESS in 21m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6511b3ce5c6b46e6ad7b1d63237410d8 : SUCCESS in 4m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4f5c6721f0f54abf86613e1037be5bfa : SUCCESS in 27m 09s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e591090be06248c9b2cb5cbfa71ec6a6 : SUCCESS in 31m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eca42f6c629248cdbf13db589409570d : SUCCESS in 57m 09s","accounts_in_message":[],"_revision_number":56},{"id":"c8a827e562a6b971ec4c4ec40f64cb035262a599","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 17:14:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":56},{"id":"41f24ecc53dac5148ffcfca0e771b66483b4abf4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 17:16:00.000000000","message":"Patch Set 56:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ffe5097476d04f909c197032ff82aca8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52daffd0ae4e4d62ba4addffeb527eb4 : SUCCESS in 55s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/b28e7724b9d5403cb8668b00e6a5071c : SUCCESS in 36s (non-voting)","accounts_in_message":[],"_revision_number":56}],"current_revision_number":56,"current_revision":"82b19644795ea8c14b5077ab46f0fbd4855bdd70","revisions":{"fa07a55544d6baa4ce905003f33f38a38c177261":{"kind":"REWORK","_number":1,"created":"2024-04-30 13:59:30.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/11/917711/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/917711/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/917711/1"}}},"commit":{"parents":[{"commit":"c5e09667ccab9dba072ab62838cf119be6ecab9f","subject":"stats: API for native labeled metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c5e09667ccab9dba072ab62838cf119be6ecab9f"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-29 14:59:37.000000000","tz":-300},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-04-30 13:55:28.000000000","tz":-240},"subject":"Add labeled metrics to proxy-logging","message":"Add labeled metrics to proxy-logging\n\nModified code to use native labeled metrics API based on Tim Burke\u0027s\npatch for proxy-logging. Thanks Tim.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\n\nChange-Id: I5a96626862644fe23a0d4bc53e85989770c8ceec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fa07a55544d6baa4ce905003f33f38a38c177261"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fa07a55544d6baa4ce905003f33f38a38c177261"}]},"branch":"refs/heads/master"},"1138fa5579c94ae275a1ce672e228e7081d8595e":{"kind":"REWORK","_number":2,"created":"2024-06-05 13:59:31.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/11/917711/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/917711/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/917711/2"}}},"commit":{"parents":[{"commit":"c5e09667ccab9dba072ab62838cf119be6ecab9f","subject":"stats: API for native labeled metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c5e09667ccab9dba072ab62838cf119be6ecab9f"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-29 14:59:37.000000000","tz":-300},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-06-05 13:36:04.000000000","tz":-240},"subject":"Add labeled metrics to proxy-logging","message":"Add labeled metrics to proxy-logging\n\nModified code to use native labeled metrics API based on Tim Burke\u0027s\npatch for proxy-logging. Thanks Tim.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\n\nChange-Id: I5a96626862644fe23a0d4bc53e85989770c8ceec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1138fa5579c94ae275a1ce672e228e7081d8595e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1138fa5579c94ae275a1ce672e228e7081d8595e"}]},"branch":"refs/heads/master"},"c7f98edd2540a8a06a07a64af6c694f5f0f6f49a":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-06-06 13:59:33.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/11/917711/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/917711/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/917711/3"}}},"commit":{"parents":[{"commit":"e0100d965c5e0791b434ce0815c5d521a019b97e","subject":"stats: API for native labeled metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e0100d965c5e0791b434ce0815c5d521a019b97e"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-29 14:59:37.000000000","tz":-300},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-06-06 13:54:42.000000000","tz":-240},"subject":"Add labeled metrics to proxy-logging","message":"Add labeled metrics to proxy-logging\n\nModified code to use native labeled metrics API based on Tim Burke\u0027s\npatch for proxy-logging. Thanks Tim.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\n\nChange-Id: I5a96626862644fe23a0d4bc53e85989770c8ceec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c7f98edd2540a8a06a07a64af6c694f5f0f6f49a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c7f98edd2540a8a06a07a64af6c694f5f0f6f49a"}]},"branch":"refs/heads/master"},"03921f598035e181763d055111f50db9710769e3":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-06-10 16:22:34.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/11/917711/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/917711/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/917711/4"}}},"commit":{"parents":[{"commit":"386c7332f99d2b6f6620546f6de995d987c96de3","subject":"stats: API for native labeled metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/386c7332f99d2b6f6620546f6de995d987c96de3"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-29 14:59:37.000000000","tz":-300},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-06-10 14:01:47.000000000","tz":-240},"subject":"Add labeled metrics to proxy-logging","message":"Add labeled metrics to proxy-logging\n\nModified code to use native labeled metrics API based on Tim Burke\u0027s\npatch for proxy-logging. Thanks Tim.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\n\nChange-Id: I5a96626862644fe23a0d4bc53e85989770c8ceec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/03921f598035e181763d055111f50db9710769e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/03921f598035e181763d055111f50db9710769e3"}]},"branch":"refs/heads/master"},"11c5acc8ced5176fd52c83cf7934b575769218fa":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-06-12 13:59:35.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/11/917711/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/917711/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/917711/5"}}},"commit":{"parents":[{"commit":"f26b113d76c0bf78117364e932353628095f0aee","subject":"stats: API for native labeled metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f26b113d76c0bf78117364e932353628095f0aee"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-29 14:59:37.000000000","tz":-300},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-06-12 13:56:55.000000000","tz":-240},"subject":"Add labeled metrics to proxy-logging","message":"Add labeled metrics to proxy-logging\n\nModified code to use native labeled metrics API based on Tim Burke\u0027s\npatch for proxy-logging. Thanks Tim.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\n\nChange-Id: I5a96626862644fe23a0d4bc53e85989770c8ceec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/11c5acc8ced5176fd52c83cf7934b575769218fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/11c5acc8ced5176fd52c83cf7934b575769218fa"}]},"branch":"refs/heads/master"},"e88f29d7ff2e09030d541c2c36806d444e3e67ad":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-06-13 19:58:36.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/11/917711/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/917711/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/917711/6"}}},"commit":{"parents":[{"commit":"f9d916c2b63ccac04e3f6f981b3d1bb6cd6c4d6b","subject":"stats: API for native labeled metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f9d916c2b63ccac04e3f6f981b3d1bb6cd6c4d6b"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-29 14:59:37.000000000","tz":-300},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-06-13 19:30:57.000000000","tz":-240},"subject":"Add labeled metrics to proxy-logging","message":"Add labeled metrics to proxy-logging\n\nModified code to use native labeled metrics API based on Tim Burke\u0027s\npatch for proxy-logging. Thanks Tim.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\n\nChange-Id: I5a96626862644fe23a0d4bc53e85989770c8ceec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e88f29d7ff2e09030d541c2c36806d444e3e67ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e88f29d7ff2e09030d541c2c36806d444e3e67ad"}]},"branch":"refs/heads/master"},"bb13c29cf0e9ee6145ddcdfeeaed16ae883d900d":{"kind":"REWORK","_number":7,"created":"2024-06-17 21:22:55.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/11/917711/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/917711/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/917711/7"}}},"commit":{"parents":[{"commit":"376c12738ad315dd6cf1a33bebc9a6e1ff30c68b","subject":"stats: API for native labeled metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/376c12738ad315dd6cf1a33bebc9a6e1ff30c68b"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-29 14:59:37.000000000","tz":-300},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-06-17 21:21:48.000000000","tz":-240},"subject":"Add labeled metrics to proxy-logging","message":"Add labeled metrics to proxy-logging\n\nModified code to use native labeled metrics API based on Tim Burke\u0027s\npatch for proxy-logging. Thanks Tim.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\n\nChange-Id: I5a96626862644fe23a0d4bc53e85989770c8ceec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bb13c29cf0e9ee6145ddcdfeeaed16ae883d900d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bb13c29cf0e9ee6145ddcdfeeaed16ae883d900d"}]},"branch":"refs/heads/master"},"10cc867d4151aaf0725703dc53a233ce18ca722e":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2024-06-25 13:58:34.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/11/917711/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/917711/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/917711/8"}}},"commit":{"parents":[{"commit":"e2217f43ea43179d3edd0e513ac669c62a613865","subject":"stats: API for native labeled metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e2217f43ea43179d3edd0e513ac669c62a613865"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-29 14:59:37.000000000","tz":-300},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-06-25 13:35:27.000000000","tz":-240},"subject":"Add labeled metrics to proxy-logging","message":"Add labeled metrics to proxy-logging\n\nModified code to use native labeled metrics API based on Tim Burke\u0027s\npatch for proxy-logging. Thanks Tim.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\n\nChange-Id: I5a96626862644fe23a0d4bc53e85989770c8ceec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/10cc867d4151aaf0725703dc53a233ce18ca722e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/10cc867d4151aaf0725703dc53a233ce18ca722e"}]},"branch":"refs/heads/master"},"2ccdcc708daabbacb91a655a36941232af6da22b":{"kind":"REWORK","_number":9,"created":"2024-06-28 13:58:34.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/11/917711/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/917711/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/917711/9"}}},"commit":{"parents":[{"commit":"691eb58b3c0ec6c0f903eb3b0703b5c78aaf98d3","subject":"stats: API for native labeled metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/691eb58b3c0ec6c0f903eb3b0703b5c78aaf98d3"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-29 14:59:37.000000000","tz":-300},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-06-28 13:56:16.000000000","tz":-240},"subject":"Add labeled metrics to proxy-logging","message":"Add labeled metrics to proxy-logging\n\nModified code to use native labeled metrics API based on Tim Burke\u0027s\npatch for proxy-logging. Thanks Tim.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\n\nChange-Id: I5a96626862644fe23a0d4bc53e85989770c8ceec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2ccdcc708daabbacb91a655a36941232af6da22b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2ccdcc708daabbacb91a655a36941232af6da22b"}]},"branch":"refs/heads/master"},"61e5b6af326dc101d77015effd9ff0cf3343f8f9":{"kind":"REWORK","_number":10,"created":"2024-07-05 19:56:34.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/11/917711/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/917711/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/917711/10"}}},"commit":{"parents":[{"commit":"16dc6b371026e911e0b3175cea1a02d94041dc1b","subject":"stats: API for native labeled metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/16dc6b371026e911e0b3175cea1a02d94041dc1b"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-29 14:59:37.000000000","tz":-300},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-07-05 19:36:44.000000000","tz":-240},"subject":"Add labeled metrics to proxy-logging","message":"Add labeled metrics to proxy-logging\n\nModified code to use native labeled metrics API based on Tim Burke\u0027s\npatch for proxy-logging. Thanks Tim.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\n\nChange-Id: I5a96626862644fe23a0d4bc53e85989770c8ceec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/61e5b6af326dc101d77015effd9ff0cf3343f8f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/61e5b6af326dc101d77015effd9ff0cf3343f8f9"}]},"branch":"refs/heads/master"},"93313ab0e98fd701d0d1a1438d200f83f5b5c66d":{"kind":"REWORK","_number":11,"created":"2024-07-11 13:58:38.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/11/917711/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/917711/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/917711/11"}}},"commit":{"parents":[{"commit":"16dc6b371026e911e0b3175cea1a02d94041dc1b","subject":"stats: API for native labeled metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/16dc6b371026e911e0b3175cea1a02d94041dc1b"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-29 14:59:37.000000000","tz":-300},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-07-11 13:35:45.000000000","tz":-240},"subject":"Add labeled metrics to proxy-logging","message":"Add labeled metrics to proxy-logging\n\nModified code to use native labeled metrics API based on Tim Burke\u0027s\npatch for proxy-logging. Thanks Tim.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\n\nChange-Id: I5a96626862644fe23a0d4bc53e85989770c8ceec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/93313ab0e98fd701d0d1a1438d200f83f5b5c66d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/93313ab0e98fd701d0d1a1438d200f83f5b5c66d"}]},"branch":"refs/heads/master"},"a4c19cd279c1a0e0a6caa56e7e7e2993f2f8ea6c":{"kind":"REWORK","_number":12,"created":"2024-07-12 21:56:34.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/11/917711/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/917711/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/917711/12"}}},"commit":{"parents":[{"commit":"137ce227de8bd03e2493e8e4689975fade1c993d","subject":"stats: API for native labeled metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/137ce227de8bd03e2493e8e4689975fade1c993d"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-29 14:59:37.000000000","tz":-300},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-07-12 21:35:55.000000000","tz":-240},"subject":"Add labeled metrics to proxy-logging","message":"Add labeled metrics to proxy-logging\n\nModified code to use native labeled metrics API based on Tim Burke\u0027s\npatch for proxy-logging. Thanks Tim.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\n\nChange-Id: I5a96626862644fe23a0d4bc53e85989770c8ceec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a4c19cd279c1a0e0a6caa56e7e7e2993f2f8ea6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a4c19cd279c1a0e0a6caa56e7e7e2993f2f8ea6c"}]},"branch":"refs/heads/master"},"0a6f10bf5210231d31c89665d4c0806cba6795c1":{"kind":"REWORK","_number":13,"created":"2024-07-15 19:56:34.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/11/917711/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/917711/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/917711/13"}}},"commit":{"parents":[{"commit":"137ce227de8bd03e2493e8e4689975fade1c993d","subject":"stats: API for native labeled metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/137ce227de8bd03e2493e8e4689975fade1c993d"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-29 14:59:37.000000000","tz":-300},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-07-15 19:55:23.000000000","tz":-240},"subject":"Add labeled metrics to proxy-logging","message":"Add labeled metrics to proxy-logging\n\nModified code to use native labeled metrics API based on Tim Burke\u0027s\npatch for proxy-logging. Thanks Tim.\n\nNew metrics added:\n  - swift_proxy_request_ttfb\n  - swift_proxy_request_body_bytes\n  - swift_proxy_response_body_bytes\n  - swift_proxy_request_body_input_bytes\n  - swift_proxy_response_body_output_bytes\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\n\nRelated-Change: I115ffb1dc601652a979895d7944e011b951a91c1\n\nChange-Id: I5a96626862644fe23a0d4bc53e85989770c8ceec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0a6f10bf5210231d31c89665d4c0806cba6795c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0a6f10bf5210231d31c89665d4c0806cba6795c1"}]},"branch":"refs/heads/master"},"1b8118fe5e36f7aa1eb9d2372f65ec4db1c663c9":{"kind":"REWORK","_number":14,"created":"2024-07-29 21:15:38.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/11/917711/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/917711/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/917711/14"}}},"commit":{"parents":[{"commit":"6c08ab3180050c1fee36cfe38f08964a97983cee","subject":"stats: API for native labeled metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6c08ab3180050c1fee36cfe38f08964a97983cee"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-29 14:59:37.000000000","tz":-300},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-07-29 19:55:17.000000000","tz":-240},"subject":"Add labeled metrics to proxy-logging","message":"Add labeled metrics to proxy-logging\n\nModified code to use native labeled metrics API based on Tim Burke\u0027s\npatch for proxy-logging. Thanks Tim.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\n\nRelated-Change: I115ffb1dc601652a979895d7944e011b951a91c1\n\nChange-Id: I5a96626862644fe23a0d4bc53e85989770c8ceec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1b8118fe5e36f7aa1eb9d2372f65ec4db1c663c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1b8118fe5e36f7aa1eb9d2372f65ec4db1c663c9"}]},"branch":"refs/heads/master"},"cf980c07e3abc703890ea4c56b5c9b80bcb9df4e":{"kind":"REWORK","_number":15,"created":"2024-07-30 21:15:35.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/11/917711/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/917711/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/917711/15"}}},"commit":{"parents":[{"commit":"d561dd39936fe17009f2eeb834e2b6599ffa0dfb","subject":"stats: API for native labeled metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d561dd39936fe17009f2eeb834e2b6599ffa0dfb"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-29 14:59:37.000000000","tz":-300},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-07-30 19:23:08.000000000","tz":-240},"subject":"Add labeled metrics to proxy-logging","message":"Add labeled metrics to proxy-logging\n\nModified code to use native labeled metrics API based on Tim Burke\u0027s\npatch for proxy-logging. Thanks Tim.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\n\nRelated-Change: I115ffb1dc601652a979895d7944e011b951a91c1\n\nChange-Id: I5a96626862644fe23a0d4bc53e85989770c8ceec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cf980c07e3abc703890ea4c56b5c9b80bcb9df4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cf980c07e3abc703890ea4c56b5c9b80bcb9df4e"}]},"branch":"refs/heads/master"},"c83e3bdf1943f6add86ec6ac3f3b2af8cd69c0cd":{"kind":"REWORK","_number":16,"created":"2024-08-02 22:56:05.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/11/917711/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/917711/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/917711/16"}}},"commit":{"parents":[{"commit":"d561dd39936fe17009f2eeb834e2b6599ffa0dfb","subject":"stats: API for native labeled metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d561dd39936fe17009f2eeb834e2b6599ffa0dfb"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-29 14:59:37.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-08-02 22:55:17.000000000","tz":-300},"subject":"Add labeled metrics to proxy-logging","message":"Add labeled metrics to proxy-logging\n\nModified code to use native labeled metrics API based on Tim Burke\u0027s\npatch for proxy-logging. Thanks Tim.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\n\nRelated-Change: I115ffb1dc601652a979895d7944e011b951a91c1\n\nChange-Id: I5a96626862644fe23a0d4bc53e85989770c8ceec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c83e3bdf1943f6add86ec6ac3f3b2af8cd69c0cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c83e3bdf1943f6add86ec6ac3f3b2af8cd69c0cd"}]},"branch":"refs/heads/master"},"cee6b24582370bfa3f5c05d1a176b00b75bc094f":{"kind":"REWORK","_number":17,"created":"2024-08-16 20:08:04.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/11/917711/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/917711/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/917711/17"}}},"commit":{"parents":[{"commit":"2476de5a2c5726456094792803af7ab956889f68","subject":"stats: API for native labeled metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2476de5a2c5726456094792803af7ab956889f68"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-29 14:59:37.000000000","tz":-300},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-08-16 20:07:36.000000000","tz":-420},"subject":"Add labeled metrics to proxy-logging","message":"Add labeled metrics to proxy-logging\n\nModified code to use native labeled metrics API based on Tim Burke\u0027s\npatch for proxy-logging. Thanks Tim.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\n\nRelated-Change: I115ffb1dc601652a979895d7944e011b951a91c1\n\nChange-Id: I5a96626862644fe23a0d4bc53e85989770c8ceec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cee6b24582370bfa3f5c05d1a176b00b75bc094f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cee6b24582370bfa3f5c05d1a176b00b75bc094f"}]},"branch":"refs/heads/master"},"09900db616b35782a7b8615ba2982cad9e3f5364":{"kind":"REWORK","_number":18,"created":"2024-09-05 06:11:12.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/11/917711/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/917711/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/917711/18"}}},"commit":{"parents":[{"commit":"3b3f71dd71f39c4538abbffb889f65b32a8a45fd","subject":"stats: API for native labeled metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3b3f71dd71f39c4538abbffb889f65b32a8a45fd"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-29 14:59:37.000000000","tz":-300},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2024-09-05 06:10:48.000000000","tz":600},"subject":"Add labeled metrics to proxy-logging","message":"Add labeled metrics to proxy-logging\n\nModified code to use native labeled metrics API based on Tim Burke\u0027s\npatch for proxy-logging. Thanks Tim.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\n\nRelated-Change: I115ffb1dc601652a979895d7944e011b951a91c1\n\nChange-Id: I5a96626862644fe23a0d4bc53e85989770c8ceec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/09900db616b35782a7b8615ba2982cad9e3f5364"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/09900db616b35782a7b8615ba2982cad9e3f5364"}]},"branch":"refs/heads/master"},"6fefda865f61bef3e117b89bd65d1d31bcbf676b":{"kind":"REWORK","_number":19,"created":"2024-09-10 13:59:35.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/11/917711/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/917711/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/917711/19"}}},"commit":{"parents":[{"commit":"18dc7b110f29738a5b5dd8de18fdb45644dab6df","subject":"stats: API for native labeled metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/18dc7b110f29738a5b5dd8de18fdb45644dab6df"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-29 14:59:37.000000000","tz":-300},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-09-10 13:58:33.000000000","tz":-240},"subject":"Add labeled metrics to proxy-logging","message":"Add labeled metrics to proxy-logging\n\nModified code to use native labeled metrics API based on Tim Burke\u0027s\npatch for proxy-logging. Thanks Tim.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\n\nRelated-Change: I115ffb1dc601652a979895d7944e011b951a91c1\n\nChange-Id: I5a96626862644fe23a0d4bc53e85989770c8ceec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6fefda865f61bef3e117b89bd65d1d31bcbf676b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6fefda865f61bef3e117b89bd65d1d31bcbf676b"}]},"branch":"refs/heads/master"},"947277842cbc0893b4e8790da611ad96a28fce17":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2024-10-02 13:59:57.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/11/917711/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/917711/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/917711/20"}}},"commit":{"parents":[{"commit":"ac1e00fddb8118caa8ba0ba2abaa1498bb06473b","subject":"stats: API for native labeled metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ac1e00fddb8118caa8ba0ba2abaa1498bb06473b"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-29 14:59:37.000000000","tz":-300},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-10-02 13:59:37.000000000","tz":-240},"subject":"Add labeled metrics to proxy-logging","message":"Add labeled metrics to proxy-logging\n\nModified code to use native labeled metrics API based on Tim Burke\u0027s\npatch for proxy-logging. Thanks Tim.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\n\nRelated-Change: I115ffb1dc601652a979895d7944e011b951a91c1\n\nChange-Id: I5a96626862644fe23a0d4bc53e85989770c8ceec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/947277842cbc0893b4e8790da611ad96a28fce17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/947277842cbc0893b4e8790da611ad96a28fce17"}]},"branch":"refs/heads/master"},"09df3927ca90d4aa215f3c6fb0cbfbdf63ab12fd":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2024-10-18 16:59:37.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/11/917711/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/917711/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/917711/21"}}},"commit":{"parents":[{"commit":"529db2c105bae1cc5228a07e6a98b0d50343c85c","subject":"stats: API for native labeled metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/529db2c105bae1cc5228a07e6a98b0d50343c85c"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-29 14:59:37.000000000","tz":-300},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-10-18 13:39:08.000000000","tz":-240},"subject":"Add labeled metrics to proxy-logging","message":"Add labeled metrics to proxy-logging\n\nModified code to use native labeled metrics API based on Tim Burke\u0027s\npatch for proxy-logging. Thanks Tim.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\n\nRelated-Change: I115ffb1dc601652a979895d7944e011b951a91c1\n\nChange-Id: I5a96626862644fe23a0d4bc53e85989770c8ceec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/09df3927ca90d4aa215f3c6fb0cbfbdf63ab12fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/09df3927ca90d4aa215f3c6fb0cbfbdf63ab12fd"}]},"branch":"refs/heads/master"},"85df8b915d4d8e739e3a2392e39dd38b288b94ea":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2024-11-29 14:31:33.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/11/917711/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/917711/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/917711/22"}}},"commit":{"parents":[{"commit":"5f3d97b5b7c4b267369638eadc903d9ef7e5c3ec","subject":"stats: API for native labeled metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5f3d97b5b7c4b267369638eadc903d9ef7e5c3ec"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-29 14:59:37.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-29 14:29:43.000000000","tz":0},"subject":"Add labeled metrics to proxy-logging","message":"Add labeled metrics to proxy-logging\n\nModified code to use native labeled metrics API based on Tim Burke\u0027s\npatch for proxy-logging. Thanks Tim.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\n\nRelated-Change: I115ffb1dc601652a979895d7944e011b951a91c1\n\nChange-Id: I5a96626862644fe23a0d4bc53e85989770c8ceec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/85df8b915d4d8e739e3a2392e39dd38b288b94ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/85df8b915d4d8e739e3a2392e39dd38b288b94ea"}]},"branch":"refs/heads/master"},"ee01a3293167ef6faa2bb323815210fd6a6bd915":{"kind":"REWORK","_number":23,"created":"2024-12-03 14:58:39.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/11/917711/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/917711/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/917711/23"}}},"commit":{"parents":[{"commit":"5f3d97b5b7c4b267369638eadc903d9ef7e5c3ec","subject":"stats: API for native labeled metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5f3d97b5b7c4b267369638eadc903d9ef7e5c3ec"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-29 14:59:37.000000000","tz":-300},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-12-03 14:35:44.000000000","tz":-300},"subject":"Add labeled metrics to proxy-logging","message":"Add labeled metrics to proxy-logging\n\nModified code to use native labeled metrics API based on Tim Burke\u0027s\npatch for proxy-logging. Thanks Tim.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\n\nRelated-Change: I115ffb1dc601652a979895d7944e011b951a91c1\n\nChange-Id: I5a96626862644fe23a0d4bc53e85989770c8ceec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ee01a3293167ef6faa2bb323815210fd6a6bd915"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ee01a3293167ef6faa2bb323815210fd6a6bd915"}]},"branch":"refs/heads/master"},"a2748293d30a3795835c76ee75becddfd20196e2":{"kind":"TRIVIAL_REBASE","_number":24,"created":"2025-01-10 15:09:38.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"real_uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/11/917711/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/917711/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/917711/24"}}},"commit":{"parents":[{"commit":"bc5febd78fc33fbbe1358c4ef17e43925d315eac","subject":"stats: API for native labeled metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bc5febd78fc33fbbe1358c4ef17e43925d315eac"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-29 14:59:37.000000000","tz":-300},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-01-10 15:09:38.000000000","tz":0},"subject":"Add labeled metrics to proxy-logging","message":"Add labeled metrics to proxy-logging\n\nModified code to use native labeled metrics API based on Tim Burke\u0027s\npatch for proxy-logging. Thanks Tim.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\n\nRelated-Change: I115ffb1dc601652a979895d7944e011b951a91c1\n\nChange-Id: I5a96626862644fe23a0d4bc53e85989770c8ceec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a2748293d30a3795835c76ee75becddfd20196e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a2748293d30a3795835c76ee75becddfd20196e2"}]},"branch":"refs/heads/master","description":"Rebase"},"709190008d69a1c486922cc4a94743285f6c8509":{"kind":"TRIVIAL_REBASE","_number":25,"created":"2025-01-16 12:05:53.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/11/917711/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/917711/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/917711/25"}}},"commit":{"parents":[{"commit":"a402b46a1380c4fdcada1be75b281e3a183e0e9e","subject":"stats: API for native labeled metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a402b46a1380c4fdcada1be75b281e3a183e0e9e"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-29 14:59:37.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-01-16 12:00:04.000000000","tz":0},"subject":"Add labeled metrics to proxy-logging","message":"Add labeled metrics to proxy-logging\n\nModified code to use native labeled metrics API based on Tim Burke\u0027s\npatch for proxy-logging. Thanks Tim.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\n\nRelated-Change: I115ffb1dc601652a979895d7944e011b951a91c1\n\nChange-Id: I5a96626862644fe23a0d4bc53e85989770c8ceec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/709190008d69a1c486922cc4a94743285f6c8509"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/709190008d69a1c486922cc4a94743285f6c8509"}]},"branch":"refs/heads/master"},"c4bb40151c64ee541b01035c5e9f988983d3ce98":{"kind":"TRIVIAL_REBASE","_number":26,"created":"2025-01-16 19:02:03.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/11/917711/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/917711/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/917711/26"}}},"commit":{"parents":[{"commit":"ae6300af86c292b4ded1b4606f7e978bf7474f8a","subject":"wsgi: Reap stale workers (after a timeout) following a reload","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ae6300af86c292b4ded1b4606f7e978bf7474f8a"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-29 14:59:37.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-01-16 19:01:36.000000000","tz":-300},"subject":"Add labeled metrics to proxy-logging","message":"Add labeled metrics to proxy-logging\n\nModified code to use native labeled metrics API based on Tim Burke\u0027s\npatch for proxy-logging. Thanks Tim.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\n\nRelated-Change: I115ffb1dc601652a979895d7944e011b951a91c1\n\nChange-Id: I5a96626862644fe23a0d4bc53e85989770c8ceec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c4bb40151c64ee541b01035c5e9f988983d3ce98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c4bb40151c64ee541b01035c5e9f988983d3ce98"}]},"branch":"refs/heads/master"},"714cd71ca4665962e63eacd1242a6281ff6f1851":{"kind":"TRIVIAL_REBASE","_number":27,"created":"2025-01-16 20:27:54.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/11/917711/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/917711/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/917711/27"}}},"commit":{"parents":[{"commit":"a402b46a1380c4fdcada1be75b281e3a183e0e9e","subject":"stats: API for native labeled metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a402b46a1380c4fdcada1be75b281e3a183e0e9e"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-29 14:59:37.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-01-16 20:27:54.000000000","tz":0},"subject":"Add labeled metrics to proxy-logging","message":"Add labeled metrics to proxy-logging\n\nModified code to use native labeled metrics API based on Tim Burke\u0027s\npatch for proxy-logging. Thanks Tim.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\n\nRelated-Change: I115ffb1dc601652a979895d7944e011b951a91c1\n\nChange-Id: I5a96626862644fe23a0d4bc53e85989770c8ceec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/714cd71ca4665962e63eacd1242a6281ff6f1851"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/714cd71ca4665962e63eacd1242a6281ff6f1851"}]},"branch":"refs/heads/master","description":"Rebase"},"6f89f8f9e279e8ae7b9b9351cbd5c529d9128ab1":{"kind":"TRIVIAL_REBASE","_number":28,"created":"2025-03-05 14:36:30.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/11/917711/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/917711/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/917711/28"}}},"commit":{"parents":[{"commit":"c9da7202cff70a2b3d8ad15e79dc83c76f00cb60","subject":"stats: API for native labeled metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c9da7202cff70a2b3d8ad15e79dc83c76f00cb60"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-29 14:59:37.000000000","tz":-300},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-03-03 20:38:36.000000000","tz":-300},"subject":"Add labeled metrics to proxy-logging","message":"Add labeled metrics to proxy-logging\n\nModified code to use native labeled metrics API based on Tim Burke\u0027s\npatch for proxy-logging. Thanks Tim.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\n\nRelated-Change: I115ffb1dc601652a979895d7944e011b951a91c1\n\nChange-Id: I5a96626862644fe23a0d4bc53e85989770c8ceec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6f89f8f9e279e8ae7b9b9351cbd5c529d9128ab1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6f89f8f9e279e8ae7b9b9351cbd5c529d9128ab1"}]},"branch":"refs/heads/master"},"eb097be036baedb4e4fa9e8bdb42eb470bc563ee":{"kind":"REWORK","_number":29,"created":"2025-03-07 11:58:42.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/11/917711/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/917711/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/917711/29"}}},"commit":{"parents":[{"commit":"70b873837eaa42647194a4653c62c8ed2dd4f971","subject":"stats: API for native labeled metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/70b873837eaa42647194a4653c62c8ed2dd4f971"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-29 14:59:37.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-03-07 11:54:17.000000000","tz":0},"subject":"Add labeled metrics to proxy-logging","message":"Add labeled metrics to proxy-logging\n\nModified code to use native labeled metrics API based on Tim Burke\u0027s\npatch for proxy-logging. Thanks Tim.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\n\nRelated-Change: I115ffb1dc601652a979895d7944e011b951a91c1\n\nChange-Id: I5a96626862644fe23a0d4bc53e85989770c8ceec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eb097be036baedb4e4fa9e8bdb42eb470bc563ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/eb097be036baedb4e4fa9e8bdb42eb470bc563ee"}]},"branch":"refs/heads/master"},"2dc520a6e2224fc138fa5a9e3d87a62ed4c53e91":{"kind":"TRIVIAL_REBASE","_number":30,"created":"2025-03-07 13:12:08.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/11/917711/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/917711/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/917711/30"}}},"commit":{"parents":[{"commit":"376d893bbd58d81ffc5a190dfceb117915f73454","subject":"stats: API for native labeled metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/376d893bbd58d81ffc5a190dfceb117915f73454"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-29 14:59:37.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-03-07 12:52:51.000000000","tz":0},"subject":"Add labeled metrics to proxy-logging","message":"Add labeled metrics to proxy-logging\n\nModified code to use native labeled metrics API based on Tim Burke\u0027s\npatch for proxy-logging. Thanks Tim.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\n\nRelated-Change: I115ffb1dc601652a979895d7944e011b951a91c1\n\nChange-Id: I5a96626862644fe23a0d4bc53e85989770c8ceec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2dc520a6e2224fc138fa5a9e3d87a62ed4c53e91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2dc520a6e2224fc138fa5a9e3d87a62ed4c53e91"}]},"branch":"refs/heads/master"},"d97da4438126ee49c6e295af24e28ed2e421132b":{"kind":"TRIVIAL_REBASE","_number":31,"created":"2025-03-17 11:34:30.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/11/917711/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/917711/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/917711/31"}}},"commit":{"parents":[{"commit":"e6000ffb4174cf9cd6bf467d6f85e547545f7003","subject":"stats: API for native labeled metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e6000ffb4174cf9cd6bf467d6f85e547545f7003"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-29 14:59:37.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-03-17 11:34:08.000000000","tz":0},"subject":"Add labeled metrics to proxy-logging","message":"Add labeled metrics to proxy-logging\n\nModified code to use native labeled metrics API based on Tim Burke\u0027s\npatch for proxy-logging. Thanks Tim.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\n\nRelated-Change: I115ffb1dc601652a979895d7944e011b951a91c1\n\nChange-Id: I5a96626862644fe23a0d4bc53e85989770c8ceec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d97da4438126ee49c6e295af24e28ed2e421132b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d97da4438126ee49c6e295af24e28ed2e421132b"}]},"branch":"refs/heads/master"},"38319c6c11b4e210c3816038c333c839b79c286e":{"kind":"REWORK","_number":32,"created":"2025-03-20 19:35:52.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/11/917711/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/917711/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/917711/32"}}},"commit":{"parents":[{"commit":"e6000ffb4174cf9cd6bf467d6f85e547545f7003","subject":"stats: API for native labeled metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e6000ffb4174cf9cd6bf467d6f85e547545f7003"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-29 14:59:37.000000000","tz":-300},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-03-20 19:23:36.000000000","tz":-240},"subject":"Add labeled metrics to proxy-logging","message":"Add labeled metrics to proxy-logging\n\nModified code to use native labeled metrics API based on Tim Burke\u0027s\npatch for proxy-logging. Thanks Tim.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\n\nRelated-Change: I115ffb1dc601652a979895d7944e011b951a91c1\n\nChange-Id: I5a96626862644fe23a0d4bc53e85989770c8ceec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/38319c6c11b4e210c3816038c333c839b79c286e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/38319c6c11b4e210c3816038c333c839b79c286e"}]},"branch":"refs/heads/master"},"1f43f8da3bfdc43ddc31ed5a7df3b83125e7d08d":{"kind":"REWORK","_number":33,"created":"2025-03-21 15:18:40.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/11/917711/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/917711/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/917711/33"}}},"commit":{"parents":[{"commit":"529a6d8a96639ba97755a587e0116cdd6f0f127f","subject":"sq? fixups for labeled statsd api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/529a6d8a96639ba97755a587e0116cdd6f0f127f"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-29 14:59:37.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-03-21 15:12:55.000000000","tz":0},"subject":"Add labeled metrics to proxy-logging","message":"Add labeled metrics to proxy-logging\n\nModified code to use native labeled metrics API based on Tim Burke\u0027s\npatch for proxy-logging. Thanks Tim.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\n\nRelated-Change: I115ffb1dc601652a979895d7944e011b951a91c1\n\nChange-Id: I5a96626862644fe23a0d4bc53e85989770c8ceec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1f43f8da3bfdc43ddc31ed5a7df3b83125e7d08d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1f43f8da3bfdc43ddc31ed5a7df3b83125e7d08d"}]},"branch":"refs/heads/master"},"18edf67852dc5a11700b29a0a53fbf343b59b3ab":{"kind":"REWORK","_number":34,"created":"2025-03-25 11:32:12.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/11/917711/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/917711/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/917711/34"}}},"commit":{"parents":[{"commit":"d0047a8fe3f8eb702758de2d1773ba28c8c291e8","subject":"stats: API for native labeled metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d0047a8fe3f8eb702758de2d1773ba28c8c291e8"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-29 14:59:37.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-03-25 11:31:10.000000000","tz":0},"subject":"Add labeled metrics to proxy-logging","message":"Add labeled metrics to proxy-logging\n\nModified code to use native labeled metrics API based on Tim Burke\u0027s\npatch for proxy-logging. Thanks Tim.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\n\nRelated-Change: I115ffb1dc601652a979895d7944e011b951a91c1\n\nChange-Id: I5a96626862644fe23a0d4bc53e85989770c8ceec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/18edf67852dc5a11700b29a0a53fbf343b59b3ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/18edf67852dc5a11700b29a0a53fbf343b59b3ab"}]},"branch":"refs/heads/master"},"40bb3e852cc9621b986a014f7411568b9ba3832d":{"kind":"REWORK","_number":35,"created":"2025-03-26 15:09:28.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/11/917711/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/917711/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/917711/35"}}},"commit":{"parents":[{"commit":"0f497ca385703d9b898833ea767a41c5de6408a1","subject":"stats: API for native labeled metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0f497ca385703d9b898833ea767a41c5de6408a1"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-29 14:59:37.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-03-26 14:51:21.000000000","tz":0},"subject":"Add labeled metrics to proxy-logging","message":"Add labeled metrics to proxy-logging\n\nModified code to use native labeled metrics API based on Tim Burke\u0027s\npatch for proxy-logging. Thanks Tim.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\n\nRelated-Change: I115ffb1dc601652a979895d7944e011b951a91c1\n\nChange-Id: I5a96626862644fe23a0d4bc53e85989770c8ceec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/40bb3e852cc9621b986a014f7411568b9ba3832d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/40bb3e852cc9621b986a014f7411568b9ba3832d"}]},"branch":"refs/heads/master"},"ca58eb934b282d6273ce50932dfec70d3b5a5c3b":{"kind":"REWORK","_number":36,"created":"2025-04-01 13:58:50.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/11/917711/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/917711/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/917711/36"}}},"commit":{"parents":[{"commit":"0f497ca385703d9b898833ea767a41c5de6408a1","subject":"stats: API for native labeled metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0f497ca385703d9b898833ea767a41c5de6408a1"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-29 14:59:37.000000000","tz":-300},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-03-31 19:25:51.000000000","tz":-240},"subject":"Add labeled metrics to proxy-logging","message":"Add labeled metrics to proxy-logging\n\nModified code to use native labeled metrics API based on Tim Burke\u0027s\npatch for proxy-logging. Thanks Tim.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\n\nRelated-Change: I115ffb1dc601652a979895d7944e011b951a91c1\n\nChange-Id: I5a96626862644fe23a0d4bc53e85989770c8ceec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ca58eb934b282d6273ce50932dfec70d3b5a5c3b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ca58eb934b282d6273ce50932dfec70d3b5a5c3b"}]},"branch":"refs/heads/master"},"12fded3227b3392ba614e0681afd4dc8eb02d2c8":{"kind":"TRIVIAL_REBASE","_number":37,"created":"2025-04-02 20:12:50.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/11/917711/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/917711/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/917711/37"}}},"commit":{"parents":[{"commit":"b2b3f36512e52b7e810d40ad2b6f9b7bb0bff0ce","subject":"stats: API for native labeled metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b2b3f36512e52b7e810d40ad2b6f9b7bb0bff0ce"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-29 14:59:37.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-04-02 20:04:27.000000000","tz":-300},"subject":"Add labeled metrics to proxy-logging","message":"Add labeled metrics to proxy-logging\n\nModified code to use native labeled metrics API based on Tim Burke\u0027s\npatch for proxy-logging. Thanks Tim.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\n\nRelated-Change: I115ffb1dc601652a979895d7944e011b951a91c1\n\nChange-Id: I5a96626862644fe23a0d4bc53e85989770c8ceec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/12fded3227b3392ba614e0681afd4dc8eb02d2c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/12fded3227b3392ba614e0681afd4dc8eb02d2c8"}]},"branch":"refs/heads/master"},"ec6fadd2120eeec1770910fd7255e26353f0c06d":{"kind":"REWORK","_number":38,"created":"2025-04-03 17:00:13.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/11/917711/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/917711/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/917711/38"}}},"commit":{"parents":[{"commit":"b2b3f36512e52b7e810d40ad2b6f9b7bb0bff0ce","subject":"stats: API for native labeled metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b2b3f36512e52b7e810d40ad2b6f9b7bb0bff0ce"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-29 14:59:37.000000000","tz":-300},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-04-03 16:54:42.000000000","tz":-240},"subject":"Add labeled metrics to proxy-logging","message":"Add labeled metrics to proxy-logging\n\nModified code to use native labeled metrics API based on Tim Burke\u0027s\npatch for proxy-logging. Thanks Tim.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\n\nRelated-Change: I115ffb1dc601652a979895d7944e011b951a91c1\n\nChange-Id: I5a96626862644fe23a0d4bc53e85989770c8ceec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ec6fadd2120eeec1770910fd7255e26353f0c06d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ec6fadd2120eeec1770910fd7255e26353f0c06d"}]},"branch":"refs/heads/master"},"7cdb7b8fc38fb78a2eae41bf8eeee8aa2a8b6fe4":{"kind":"TRIVIAL_REBASE","_number":39,"created":"2025-04-03 18:28:12.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"real_uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/11/917711/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/917711/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/917711/39"}}},"commit":{"parents":[{"commit":"7e5235894b4f590024d572f64ab0b3ba2b02c68b","subject":"stats: API for native labeled metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7e5235894b4f590024d572f64ab0b3ba2b02c68b"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-29 14:59:37.000000000","tz":-300},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-04-03 18:28:12.000000000","tz":0},"subject":"Add labeled metrics to proxy-logging","message":"Add labeled metrics to proxy-logging\n\nModified code to use native labeled metrics API based on Tim Burke\u0027s\npatch for proxy-logging. Thanks Tim.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\n\nRelated-Change: I115ffb1dc601652a979895d7944e011b951a91c1\n\nChange-Id: I5a96626862644fe23a0d4bc53e85989770c8ceec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7cdb7b8fc38fb78a2eae41bf8eeee8aa2a8b6fe4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7cdb7b8fc38fb78a2eae41bf8eeee8aa2a8b6fe4"}]},"branch":"refs/heads/master","description":"Rebase"},"832a31fab8e970901b805f1bcfc140b7216a8b0f":{"kind":"REWORK","_number":40,"created":"2025-04-03 18:44:14.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/11/917711/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/917711/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/917711/40"}}},"commit":{"parents":[{"commit":"7e5235894b4f590024d572f64ab0b3ba2b02c68b","subject":"stats: API for native labeled metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7e5235894b4f590024d572f64ab0b3ba2b02c68b"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-29 14:59:37.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-04-03 18:42:32.000000000","tz":-240},"subject":"Add labeled metrics to proxy-logging","message":"Add labeled metrics to proxy-logging\n\nModified code to use native labeled metrics API based on Tim Burke\u0027s\npatch for proxy-logging. Thanks Tim.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\n\nRelated-Change: I115ffb1dc601652a979895d7944e011b951a91c1\n\nChange-Id: I5a96626862644fe23a0d4bc53e85989770c8ceec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/832a31fab8e970901b805f1bcfc140b7216a8b0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/832a31fab8e970901b805f1bcfc140b7216a8b0f"}]},"branch":"refs/heads/master"},"bafc8a874cc72eb10c12028ac2c86b62aad2bb63":{"kind":"REWORK","_number":41,"created":"2025-04-04 14:35:45.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/11/917711/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/917711/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/917711/41"}}},"commit":{"parents":[{"commit":"0e2791a88a9199006880bbbe31ff4e6d27c58e88","subject":"Remove deprecated statsd label_mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0e2791a88a9199006880bbbe31ff4e6d27c58e88"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-29 14:59:37.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-04-04 14:35:20.000000000","tz":60},"subject":"Add labeled metrics to proxy-logging","message":"Add labeled metrics to proxy-logging\n\nModified code to use native labeled metrics API introduced by the\nRelated-Change.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nRelated-Change: I115ffb1dc601652a979895d7944e011b951a91c1\n\nChange-Id: I5a96626862644fe23a0d4bc53e85989770c8ceec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bafc8a874cc72eb10c12028ac2c86b62aad2bb63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bafc8a874cc72eb10c12028ac2c86b62aad2bb63"}]},"branch":"refs/heads/master"},"491d5d9df878e8c20734c64bea49092289a75775":{"kind":"TRIVIAL_REBASE","_number":42,"created":"2025-04-04 15:51:26.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/11/917711/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/917711/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/917711/42"}}},"commit":{"parents":[{"commit":"769e482ea862e6416f3eff893da7edf18e36f6cb","subject":"tests: provide FakeLabeledStatsdClient for debug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/769e482ea862e6416f3eff893da7edf18e36f6cb"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-29 14:59:37.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-04-04 15:50:46.000000000","tz":60},"subject":"Add labeled metrics to proxy-logging","message":"Add labeled metrics to proxy-logging\n\nModified code to use native labeled metrics API introduced by the\nRelated-Change.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nRelated-Change: I115ffb1dc601652a979895d7944e011b951a91c1\n\nChange-Id: I5a96626862644fe23a0d4bc53e85989770c8ceec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/491d5d9df878e8c20734c64bea49092289a75775"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/491d5d9df878e8c20734c64bea49092289a75775"}]},"branch":"refs/heads/master"},"22f2bb9b3d495cd2dd144b554c41e56e6cbd1ad9":{"kind":"REWORK","_number":43,"created":"2025-04-04 21:24:18.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/11/917711/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/917711/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/917711/43"}}},"commit":{"parents":[{"commit":"769e482ea862e6416f3eff893da7edf18e36f6cb","subject":"tests: provide FakeLabeledStatsdClient for debug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/769e482ea862e6416f3eff893da7edf18e36f6cb"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-29 14:59:37.000000000","tz":-300},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-04-04 21:16:54.000000000","tz":-240},"subject":"Add labeled metrics to proxy-logging","message":"Add labeled metrics to proxy-logging\n\nModified code to use native labeled metrics API introduced by the\nRelated-Change.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nRelated-Change: I115ffb1dc601652a979895d7944e011b951a91c1\n\nChange-Id: I5a96626862644fe23a0d4bc53e85989770c8ceec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/22f2bb9b3d495cd2dd144b554c41e56e6cbd1ad9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/22f2bb9b3d495cd2dd144b554c41e56e6cbd1ad9"}]},"branch":"refs/heads/master"},"a68a0b44a063309b6c080c29451e81b7ff3322ce":{"kind":"REWORK","_number":44,"created":"2025-04-08 13:58:22.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/11/917711/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/917711/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/917711/44"}}},"commit":{"parents":[{"commit":"769e482ea862e6416f3eff893da7edf18e36f6cb","subject":"tests: provide FakeLabeledStatsdClient for debug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/769e482ea862e6416f3eff893da7edf18e36f6cb"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-29 14:59:37.000000000","tz":-300},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-04-08 13:33:10.000000000","tz":-240},"subject":"Add labeled metrics to proxy-logging","message":"Add labeled metrics to proxy-logging\n\nModified code to use native labeled metrics API introduced by the\nRelated-Change.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nRelated-Change: I115ffb1dc601652a979895d7944e011b951a91c1\n\nChange-Id: I5a96626862644fe23a0d4bc53e85989770c8ceec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a68a0b44a063309b6c080c29451e81b7ff3322ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a68a0b44a063309b6c080c29451e81b7ff3322ce"}]},"branch":"refs/heads/master"},"c54335dfbb5d1b794f68fd274c6dfbace7051d4b":{"kind":"REWORK","_number":45,"created":"2025-04-09 13:56:22.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/11/917711/45","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/917711/45","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/45 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/45 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/45 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/917711/45"}}},"commit":{"parents":[{"commit":"7bef99cfd1ae2a25d2e5f531341cb772a885652f","subject":"Merge \"tests: add LabeledStatsdClient sample_rate kwarg coverage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7bef99cfd1ae2a25d2e5f531341cb772a885652f"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-29 14:59:37.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-04-09 13:56:13.000000000","tz":60},"subject":"Add labeled metrics to proxy-logging","message":"Add labeled metrics to proxy-logging\n\nModified code to use native labeled metrics API introduced by the\nRelated-Change.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nRelated-Change: I115ffb1dc601652a979895d7944e011b951a91c1\n\nChange-Id: I5a96626862644fe23a0d4bc53e85989770c8ceec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c54335dfbb5d1b794f68fd274c6dfbace7051d4b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c54335dfbb5d1b794f68fd274c6dfbace7051d4b"}]},"branch":"refs/heads/master"},"ca34f5e945e50516cb4b6f9ade3c022281f988ab":{"kind":"REWORK","_number":46,"created":"2025-04-10 21:23:34.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/11/917711/46","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/917711/46","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/46 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/46 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/46 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/917711/46"}}},"commit":{"parents":[{"commit":"7bef99cfd1ae2a25d2e5f531341cb772a885652f","subject":"Merge \"tests: add LabeledStatsdClient sample_rate kwarg coverage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7bef99cfd1ae2a25d2e5f531341cb772a885652f"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-29 14:59:37.000000000","tz":-300},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-04-10 21:15:49.000000000","tz":-240},"subject":"Add labeled metrics to proxy-logging","message":"Add labeled metrics to proxy-logging\n\nModified code to use native labeled metrics API introduced by the\nRelated-Change.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nRelated-Change: I115ffb1dc601652a979895d7944e011b951a91c1\n\nChange-Id: I5a96626862644fe23a0d4bc53e85989770c8ceec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ca34f5e945e50516cb4b6f9ade3c022281f988ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ca34f5e945e50516cb4b6f9ade3c022281f988ab"}]},"branch":"refs/heads/master"},"bc3f550342635c40943520c5cdaf7aaf4e7b289b":{"kind":"TRIVIAL_REBASE","_number":47,"created":"2025-04-10 22:13:33.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/11/917711/47","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/917711/47","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/47 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/47 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/47 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/917711/47"}}},"commit":{"parents":[{"commit":"c7f93f0f63841140d75c21a833647073fb3f4eb3","subject":"Merge \"Add skip reason next to test name in zuul output\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c7f93f0f63841140d75c21a833647073fb3f4eb3"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-29 14:59:37.000000000","tz":-300},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-04-10 22:13:33.000000000","tz":0},"subject":"Add labeled metrics to proxy-logging","message":"Add labeled metrics to proxy-logging\n\nModified code to use native labeled metrics API introduced by the\nRelated-Change.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nRelated-Change: I115ffb1dc601652a979895d7944e011b951a91c1\n\nChange-Id: I5a96626862644fe23a0d4bc53e85989770c8ceec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bc3f550342635c40943520c5cdaf7aaf4e7b289b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bc3f550342635c40943520c5cdaf7aaf4e7b289b"}]},"branch":"refs/heads/master","description":"Rebase"},"10deeb5e164260b9dcdc69a8bc9eb8c423726b7e":{"kind":"REWORK","_number":48,"created":"2025-04-11 13:10:10.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/11/917711/48","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/917711/48","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/48 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/48 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/48 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/917711/48"}}},"commit":{"parents":[{"commit":"c7f93f0f63841140d75c21a833647073fb3f4eb3","subject":"Merge \"Add skip reason next to test name in zuul output\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c7f93f0f63841140d75c21a833647073fb3f4eb3"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-29 14:59:37.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-04-11 13:08:41.000000000","tz":60},"subject":"Add labeled metrics to proxy-logging","message":"Add labeled metrics to proxy-logging\n\nModified code to use native labeled metrics API introduced by the\nRelated-Change.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nRelated-Change: I115ffb1dc601652a979895d7944e011b951a91c1\n\nChange-Id: I5a96626862644fe23a0d4bc53e85989770c8ceec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/10deeb5e164260b9dcdc69a8bc9eb8c423726b7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/10deeb5e164260b9dcdc69a8bc9eb8c423726b7e"}]},"branch":"refs/heads/master"},"7cc364f61b64d55b2b28a3a3ddedc691906b11c5":{"kind":"REWORK","_number":49,"created":"2025-04-14 15:25:24.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/11/917711/49","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/917711/49","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/49 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/49 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/49 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/917711/49"}}},"commit":{"parents":[{"commit":"c7f93f0f63841140d75c21a833647073fb3f4eb3","subject":"Merge \"Add skip reason next to test name in zuul output\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c7f93f0f63841140d75c21a833647073fb3f4eb3"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-29 14:59:37.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-04-14 15:24:43.000000000","tz":-240},"subject":"Add labeled metrics to proxy-logging","message":"Add labeled metrics to proxy-logging\n\nModified code to use native labeled metrics API introduced by the\nRelated-Change.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nRelated-Change: I115ffb1dc601652a979895d7944e011b951a91c1\n\nChange-Id: I5a96626862644fe23a0d4bc53e85989770c8ceec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7cc364f61b64d55b2b28a3a3ddedc691906b11c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7cc364f61b64d55b2b28a3a3ddedc691906b11c5"}]},"branch":"refs/heads/master"},"7788ea0d82ea4a8b34b6bab47756468733528907":{"kind":"REWORK","_number":50,"created":"2025-04-16 19:41:00.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/11/917711/50","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/917711/50","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/50 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/50 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/50 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/917711/50"}}},"commit":{"parents":[{"commit":"c7f93f0f63841140d75c21a833647073fb3f4eb3","subject":"Merge \"Add skip reason next to test name in zuul output\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c7f93f0f63841140d75c21a833647073fb3f4eb3"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-29 14:59:37.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-04-16 19:39:58.000000000","tz":-240},"subject":"Add labeled metrics to proxy-logging","message":"Add labeled metrics to proxy-logging\n\nModified code to use native labeled metrics API introduced by the\nRelated-Change.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nRelated-Change: I115ffb1dc601652a979895d7944e011b951a91c1\n\nChange-Id: I5a96626862644fe23a0d4bc53e85989770c8ceec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7788ea0d82ea4a8b34b6bab47756468733528907"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7788ea0d82ea4a8b34b6bab47756468733528907"}]},"branch":"refs/heads/master"},"870c7f0899d7b76744b15096031e53264e161744":{"kind":"TRIVIAL_REBASE","_number":51,"created":"2025-04-28 01:18:35.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/11/917711/51","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/917711/51","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/51 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/51 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/51 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/917711/51"}}},"commit":{"parents":[{"commit":"f882ba76dda5e6f0df374d8a628accc1f19c6e93","subject":"Merge \"Remove tags from README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f882ba76dda5e6f0df374d8a628accc1f19c6e93"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-29 14:59:37.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-04-28 01:10:37.000000000","tz":-240},"subject":"Add labeled metrics to proxy-logging","message":"Add labeled metrics to proxy-logging\n\nModified code to use native labeled metrics API introduced by the\nRelated-Change.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nRelated-Change: I115ffb1dc601652a979895d7944e011b951a91c1\n\nChange-Id: I5a96626862644fe23a0d4bc53e85989770c8ceec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/870c7f0899d7b76744b15096031e53264e161744"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/870c7f0899d7b76744b15096031e53264e161744"}]},"branch":"refs/heads/master"},"a5160909acb1339923039fd10ca4cf0e95253b4c":{"kind":"REWORK","_number":52,"created":"2025-04-28 19:36:03.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/11/917711/52","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/917711/52","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/52 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/52 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/52 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/917711/52"}}},"commit":{"parents":[{"commit":"f882ba76dda5e6f0df374d8a628accc1f19c6e93","subject":"Merge \"Remove tags from README\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f882ba76dda5e6f0df374d8a628accc1f19c6e93"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-29 14:59:37.000000000","tz":-300},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-04-28 19:23:38.000000000","tz":-240},"subject":"Add labeled metrics to proxy-logging","message":"Add labeled metrics to proxy-logging\n\nModified code to use native labeled metrics API introduced by the\nRelated-Change.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nRelated-Change: I115ffb1dc601652a979895d7944e011b951a91c1\n\nChange-Id: I5a96626862644fe23a0d4bc53e85989770c8ceec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a5160909acb1339923039fd10ca4cf0e95253b4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a5160909acb1339923039fd10ca4cf0e95253b4c"}]},"branch":"refs/heads/master"},"ccf85e8e8a2ede228d7c8ba9d665edb8f64717ea":{"kind":"REWORK","_number":53,"created":"2025-05-15 03:01:48.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/11/917711/53","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/917711/53","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/53 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/53 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/53 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/917711/53"}}},"commit":{"parents":[{"commit":"c12aa812c9f434d4584e65d48c8f77b6c8c37862","subject":"Merge \"Drop support for py36\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c12aa812c9f434d4584e65d48c8f77b6c8c37862"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-29 14:59:37.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-05-15 02:55:15.000000000","tz":-240},"subject":"Add labeled metrics to proxy-logging","message":"Add labeled metrics to proxy-logging\n\nModified code to use native labeled metrics API introduced by the\nRelated-Change.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nRelated-Change: I115ffb1dc601652a979895d7944e011b951a91c1\n\nChange-Id: I5a96626862644fe23a0d4bc53e85989770c8ceec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ccf85e8e8a2ede228d7c8ba9d665edb8f64717ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ccf85e8e8a2ede228d7c8ba9d665edb8f64717ea"}]},"branch":"refs/heads/master"},"d7b014e23e3abfeab1c749567e9f84ed36552f1d":{"kind":"TRIVIAL_REBASE","_number":54,"created":"2025-05-20 15:56:54.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/11/917711/54","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/917711/54","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/54 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/54 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/54 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/917711/54"}}},"commit":{"parents":[{"commit":"41712d4749a8b8fc8737626b9c447eacd228e3dc","subject":"obj: skip to async_pending when root is busy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/41712d4749a8b8fc8737626b9c447eacd228e3dc"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-29 14:59:37.000000000","tz":-300},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-05-20 14:57:34.000000000","tz":-420},"subject":"Add labeled metrics to proxy-logging","message":"Add labeled metrics to proxy-logging\n\nModified code to use native labeled metrics API introduced by the\nRelated-Change.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nRelated-Change: I115ffb1dc601652a979895d7944e011b951a91c1\n\nChange-Id: I5a96626862644fe23a0d4bc53e85989770c8ceec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d7b014e23e3abfeab1c749567e9f84ed36552f1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d7b014e23e3abfeab1c749567e9f84ed36552f1d"}]},"branch":"refs/heads/master"},"8df07b9917912ada421a9287b92284e35d84625f":{"kind":"TRIVIAL_REBASE","_number":55,"created":"2025-05-20 16:49:59.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/11/917711/55","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/917711/55","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/55 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/55 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/55 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/917711/55"}}},"commit":{"parents":[{"commit":"b735b3d034b11e1472f7b471919649d165c447e0","subject":"object-server: return 503 not 404 if meta or data file unlinked","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b735b3d034b11e1472f7b471919649d165c447e0"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-29 14:59:37.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-05-20 16:49:51.000000000","tz":60},"subject":"Add labeled metrics to proxy-logging","message":"Add labeled metrics to proxy-logging\n\nModified code to use native labeled metrics API introduced by the\nRelated-Change.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nRelated-Change: I115ffb1dc601652a979895d7944e011b951a91c1\n\nChange-Id: I5a96626862644fe23a0d4bc53e85989770c8ceec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8df07b9917912ada421a9287b92284e35d84625f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8df07b9917912ada421a9287b92284e35d84625f"}]},"branch":"refs/heads/master"},"82b19644795ea8c14b5077ab46f0fbd4855bdd70":{"kind":"REWORK","_number":56,"created":"2025-05-20 21:05:00.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/11/917711/56","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/917711/56","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/56 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/56 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/917711/56 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/917711/56"}}},"commit":{"parents":[{"commit":"b735b3d034b11e1472f7b471919649d165c447e0","subject":"object-server: return 503 not 404 if meta or data file unlinked","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b735b3d034b11e1472f7b471919649d165c447e0"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2024-02-29 14:59:37.000000000","tz":-300},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2025-05-20 21:04:36.000000000","tz":-240},"subject":"Add labeled metrics to proxy-logging","message":"Add labeled metrics to proxy-logging\n\nModified code to use native labeled metrics API introduced by the\nRelated-Change.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nRelated-Change: I115ffb1dc601652a979895d7944e011b951a91c1\n\nChange-Id: I5a96626862644fe23a0d4bc53e85989770c8ceec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/82b19644795ea8c14b5077ab46f0fbd4855bdd70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/82b19644795ea8c14b5077ab46f0fbd4855bdd70"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
