)]}'
{"id":"openstack%2Fswift~917790","triplet_id":"openstack%2Fswift~master~I68974338f8e0284ed77960048a83f72855b93348","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-05-02 11:43:07.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-05-02 11:43:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I68974338f8e0284ed77960048a83f72855b93348","subject":"tests: Use mock.patch more","status":"MERGED","created":"2024-04-30 21:15:29.000000000","updated":"2024-05-02 11:44:09.000000000","submitted":"2024-05-02 11:43:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"917790","meta_rev_id":"1249fc3f52fdca5e2619b9f0cf3611850eb5064c","_number":917790,"virtual_id_number":917790,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-02 11:43:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2024-05-02 09:16:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2024-05-02 09:16:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-30 22:01:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-01 19:50:43.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2024-05-02 09:16:51.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"052d3d3f76e003dbe844967662e8a24bd833f89d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-04-30 21:15:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1ad918405f81368ec4e793e13971da04bdf90b39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 22:01:28.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ac010405cde743239f6c16a21b6cbf42\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e44fa125ad8a43198b72130e9aec86b1 : SUCCESS in 8m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f0e0128d027b4709ae99b55897dc31d8 : SUCCESS in 19m 14s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d60abf4193894819aebd64ca7d4dfb43 : SUCCESS in 18m 39s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fcaf933a5c1e4a8d92093c93a9daa410 : SUCCESS in 16m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7532809be8724ad8ad39d7bf9c125ed4 : SUCCESS in 10m 08s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/616935d7b05e4b46bc721f576400916c : SUCCESS in 4m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fd1fcc4fabb6441791178e74b609b453 : SUCCESS in 14m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f170c937e5194e2fa291e761873e373d : SUCCESS in 16m 42s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f60567dbe70249e19c365f9c7dec49a7 : SUCCESS in 24m 50s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/401609d6e7c64d13a7234dce74ba07ea : SUCCESS in 21m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d2ec1a1f09604d47b716b7ee91fb76d0 : SUCCESS in 14m 27s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f7bdd28969f641268bd544d0b66b7ac6 : SUCCESS in 19m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0c1e71c2bdb346de943d5b37be773d4c : SUCCESS in 15m 03s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/78920e7b15494b7aac6e2c009cc5c42c : SUCCESS in 21m 39s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/bb5c67d0d53f4a10ac147123539a104b : SUCCESS in 25m 34s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6732b8431f4844d9bbf487b1ebcc81cf : SUCCESS in 16m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2b6b0ad94a8c483f9368c9c2febf31a7 : SUCCESS in 35m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f29762e05524443d954737ac176b6502 : SUCCESS in 43m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/04fb13eccc024f5da130ae4f56085d3a : SUCCESS in 18m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aeeb47104d0f43b08746b7e42303249b : FAILURE in 3m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/211229e09968484093c9ec62e0c134d2 : SUCCESS in 23m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"872e34506de2af9635c5ddaa3269e1b876dc7a5a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 22:06:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a4d55730273480282e6b80acfa4c8bb\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6b7efae903884c0d9d734bfc4edc3e98 : SUCCESS in 16m 15s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8e2949db60484eae9e9b74845a3ec349 : SUCCESS in 16m 16s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d61bfa162e274c4eaccdf8a2132fa0ed : SUCCESS in 49m 00s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/24939511234c48cf8083aeb0687fddc5 : SUCCESS in 18m 58s","accounts_in_message":[],"_revision_number":1},{"id":"0e1d90be6fb08792f76c05f6b7282797f700eba5","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-05-01 19:50:43.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c94ab4ff6a306f59d8f0d8bca9e4d3c9b7d04ea1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-02 00:33:17.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"1c0febedf88a27a5cd9ffabcaa5007472bfb2db1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 01:15:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3aa33dc1723a449a956f3e9aa24e8010\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b611a61a58d04f6b81adb7b71e3d7133 : SUCCESS in 39m 07s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5ad14c257e3b418499467ed49b3a61ec : SUCCESS in 16m 43s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a14585a73d9b48b78686b6a09ef8d25a : SUCCESS in 40m 51s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/c2ac092ab84c482e8bebe22e2e453fa6 : SUCCESS in 18m 35s","accounts_in_message":[],"_revision_number":2},{"id":"13413b6934afe1f5cb6a87384744235e964291af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 01:56:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a32fd9abe27848cc95d1a04c9ff7ab25\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f44d100399ec4194941c4e2f22c95c61 : SUCCESS in 8m 17s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6c4eeea78b1147558caddd820d44b6d1 : SUCCESS in 21m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/99f6d7a00e3b436ba58cadc92f7b047e : SUCCESS in 16m 47s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8ef62e75a0fa4909ab8419b4e22d6fd7 : SUCCESS in 18m 43s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6e1e7082b5c046719bea87b6d21b9c36 : SUCCESS in 11m 51s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6ca90b3470ac495bb87ae7bdfdabadad : SUCCESS in 4m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1299262c909c499592d00acaec05eb85 : SUCCESS in 17m 37s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b22bb98b6f6a47b7a980d25358a27443 : SUCCESS in 19m 23s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/2fdcddc7d2b74d6690daa575fcb12c4c : SUCCESS in 19m 01s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/f4468c319eea40748c1e3cde65c40858 : SUCCESS in 11m 58s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/63c7448b0e69435880dfc4a7a72cc400 : SUCCESS in 14m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7ba74d96a08b4d3ebbdb7157403771e8 : SUCCESS in 15m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/391a98ac1ca34984a05e84fbfaecfe66 : SUCCESS in 19m 28s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a19b07311da146f4b7d5ed7c5e7eec07 : SUCCESS in 14m 57s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a3cb5096873747b3b769ca5c34283700 : SUCCESS in 15m 39s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c59042e5b4dc49cdb219a3af7bd0ef2c : SUCCESS in 13m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1b76cc64a9404014bb581626596be90a : SUCCESS in 51m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d7080846aca842bdbf3befae87569901 : SUCCESS in 51m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/99bc55e94df041c98daca8fca5603ae6 : SUCCESS in 19m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ac814fa25684595a45fcf0b445927a5 : SUCCESS in 5m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d54a7cb9e1b146f8b1f3607a5ce9b3f1 : SUCCESS in 24m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d4f87e268b5b57ba6e5838afbb948f83476b82c6","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-05-02 09:16:51.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"efc6e111fcf6557543ce5c93d1fa7f32de1d42e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 09:17:17.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3fe2c9b112ed1adf6a2fb7b63006ff4cc7c30abd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 11:43:07.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85feae3e821d4be0ac57fa80c7136730\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45c9ae2f8494460284101e76e7d5176a : SUCCESS in 7m 13s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/42e3646071af438a84fd3b72dfdab03e : SUCCESS in 9m 50s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/5c51da8da13d4fd18f2490ab73b96705 : SUCCESS in 6m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/40dc9bb6f7ed465bb00ba5ad743ca40b : SUCCESS in 20m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a0280ed007314c7787122f805a97ea48 : SUCCESS in 21m 05s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/bf652b97407841be8ff132a6c502e67a : SUCCESS in 14m 09s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/a695df525c5245308d15a226cd7228ac : SUCCESS in 16m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5820b0a0db8c4dd897d2f0416d51abdf : SUCCESS in 17m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2cd73d22e5be423a9ef007696b31ebf6 : SUCCESS in 21m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/142da1d8c9ba447aba8dd7c0ffa0c7f8 : SUCCESS in 16m 32s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/14e1d3498e364b44a17638032fc359e2 : SUCCESS in 16m 55s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e7b1ad73d4b64b8aa5171275ca83d525 : SUCCESS in 15m 56s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/87f9513eb32d47f0af8cd170e990e73e : SUCCESS in 16m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b3cc4608edfe42aab2b6e52500d9e15f : SUCCESS in 5m 58s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9d2b63d89ad24c14ae020aa86e30b337 : SUCCESS in 7m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/017a7b707f4447f99b04d3769195af6e : SUCCESS in 53m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/81e50550b61e426290db796a2a5c9441 : SUCCESS in 52m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1da90a092c484e9b8e9dec7a34650614 : SUCCESS in 18m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39bab2969e494025a4ec9c5bec6d908b : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":2},{"id":"d7542773dddcd016d8e8ad24329084535a145ba1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 11:43:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"1249fc3f52fdca5e2619b9f0cf3611850eb5064c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 11:44:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91929d1233ef47518d2f02903addcb4d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec41762919fd429885fd26d3597778ba : SUCCESS in 44s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/ae61c64a22f843a5b89cdb966bc6ffc4 : SUCCESS in 36s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/f82fb67e426042fd90fc9d02bbd6ae26 : SUCCESS in 34s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"761d9196772bc150fd4923fb04367b919a521d59","revisions":{"6594ccc28f1eab100baf55f231ed652966d725c0":{"kind":"REWORK","_number":1,"created":"2024-04-30 21:15:29.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/90/917790/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/917790/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/917790/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/917790/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/917790/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/917790/1"}}},"commit":{"parents":[{"commit":"d4a03039198652af9928abcf3bbbee3d11ad9046","subject":"tests: Use @with_tempdir more","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d4a03039198652af9928abcf3bbbee3d11ad9046"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-04-30 21:14:30.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-04-30 21:14:30.000000000","tz":-420},"subject":"tests: Use mock.patch more","message":"tests: Use mock.patch more\n\nChange-Id: I68974338f8e0284ed77960048a83f72855b93348\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6594ccc28f1eab100baf55f231ed652966d725c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6594ccc28f1eab100baf55f231ed652966d725c0"}]},"branch":"refs/heads/master"},"761d9196772bc150fd4923fb04367b919a521d59":{"kind":"REWORK","_number":2,"created":"2024-05-02 00:33:17.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/90/917790/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/90/917790/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/917790/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/917790/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/90/917790/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/90/917790/2"}}},"commit":{"parents":[{"commit":"b4dddb740639401f8166ee167b4173fc51ed03a0","subject":"tests: Use @with_tempdir more","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b4dddb740639401f8166ee167b4173fc51ed03a0"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-04-30 21:14:30.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-02 00:33:11.000000000","tz":-420},"subject":"tests: Use mock.patch more","message":"tests: Use mock.patch more\n\nChange-Id: I68974338f8e0284ed77960048a83f72855b93348\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/761d9196772bc150fd4923fb04367b919a521d59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/761d9196772bc150fd4923fb04367b919a521d59"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
