)]}'
{"id":"openstack%2Fswift~918106","triplet_id":"openstack%2Fswift~master~Ib2e702745fe9f975dbb957c0a9f444eeb0cae5aa","project":"openstack/swift","branch":"master","topic":"py312","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-05-07 08:59:58.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ib2e702745fe9f975dbb957c0a9f444eeb0cae5aa","subject":"sq: WIP: use a ClosingMapper","status":"ABANDONED","created":"2024-05-03 16:55:31.000000000","updated":"2024-05-07 08:59:58.000000000","total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"549d3d19c8502ece2a3cb7c2615c4cbf808f54d8","_number":918106,"virtual_id_number":918106,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-05-03 18:04:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-05-06 15:45:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-03 18:04:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-06 15:45:30.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"535a8fd37d07cb7aa7c50b6a3e8558ea29896a10","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-05-03 16:55:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a80dcc858e9aeb6ce252cc845d5ccb723d3122b2","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-05-03 16:58:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7958e141786bdde05f231e072209735907c787b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 18:04:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/67cac0db05c64a43b07198913bc8b506\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe2b08bda75f4f3fa8e7eb2e8cd3a104 : SUCCESS in 8m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/bafde46021f240b3afe77cf70e02e92a : FAILURE in 46m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2f41a6f6be7d452193b4c3bf1f8219a5 : SUCCESS in 38m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9d3cd76ad7e743fabc65129b3195a51c : SUCCESS in 1h 03m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b30973ded93640c4be925c57dcbf4019 : SUCCESS in 17m 01s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/02c95b4f24794617ba94be2276c6a6e5 : SUCCESS in 19m 24s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5e1d07d64df14728a0bcaee23ed02851 : SUCCESS in 20m 04s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0c493404a4b14163923bf919fbebe8cf : SUCCESS in 12m 11s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6bcbf508582248e088abc7465ee72059 : SUCCESS in 5m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4811d499c72e4dda9030e6fab73b1015 : SUCCESS in 21m 21s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5c7d197bbf3c42dd87148b9e8d46d5ce : SUCCESS in 17m 51s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d6ad4a5b449d440fa8db89b556bb2f02 : SUCCESS in 15m 36s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/cfd8746716a44e7db84e6df5e5d15dca : SUCCESS in 19m 20s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6c39f04936844338ad44d9afbc8693ba : SUCCESS in 16m 30s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5347994e65bb4d32936be0deb077b858 : SUCCESS in 21m 40s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6adab387c02a493485c1067609a33a64 : SUCCESS in 18m 42s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d95e2df5ef454a35a29f183711a2ab7c : SUCCESS in 15m 51s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b585e1f935824321b4bb810430cc69c0 : SUCCESS in 15m 31s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/96ca121748ae4decb20d30c914e30c2b : SUCCESS in 16m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/67b4df4747694aa1a5d21f2bfa8f78f2 : SUCCESS in 6m 16s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a62f79a9df314498ad12fbb80f82df90 : SUCCESS in 14m 18s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0205d6660bc645d0a9044cad68987591 : SUCCESS in 7m 19s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/437b8551548e416483bf7c7841cbba1e : SUCCESS in 1h 00m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e00eb7b8c4e549eaac8a2f4be7e3ca0f : SUCCESS in 55m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6ae961d378cc4581a86d60e1ad08fb5b : SUCCESS in 40m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a3b07e02dfa3435f887cf6fbd3c1f152 : SUCCESS in 19m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4ad4752e83f4fb787fd1f86816c98b6 : SUCCESS in 4m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/00a726618c2b4ef19a722019314ee90a : SUCCESS in 26m 13s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/721c4e06aff5496f8c5f3933de6d46e3 : SUCCESS in 51m 22s","accounts_in_message":[],"_revision_number":1},{"id":"b10395cb470494cd9eeeac025fa7d86012f8d29c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 18:28:32.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/596b4cd1bbad4c67be214752dd3d1c4e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b603201ac6654bcb99827c18510ae9f3 : TIMED_OUT in 1h 31m 02s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9351200851944dbca0a2bbdd38d8e2aa : SUCCESS in 51m 59s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d60181baac1c4428966747f97490d650 : SUCCESS in 1h 07m 18s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ef559b8211404f6bb5b4864d8fe0cdec : SUCCESS in 45m 23s","accounts_in_message":[],"_revision_number":1},{"id":"aa20560ade232566a55364386f927958c7c483f8","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-06 15:45:30.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"549d3d19c8502ece2a3cb7c2615c4cbf808f54d8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-05-07 08:59:58.000000000","message":"Abandoned\n\nsquashed into parent","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"10061d310c5a47d2229fd9b4c92a67a7da32d151","revisions":{"10061d310c5a47d2229fd9b4c92a67a7da32d151":{"kind":"REWORK","_number":1,"created":"2024-05-03 16:55:31.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/06/918106/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/918106/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/918106/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/918106/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/918106/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/918106/1"}}},"commit":{"parents":[{"commit":"0442c2351063979823e728338c51345fe33cf1bc","subject":"Ensure client disconnects are logged promptly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0442c2351063979823e728338c51345fe33cf1bc"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-05-03 16:54:43.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-05-03 16:54:43.000000000","tz":60},"subject":"sq: WIP: use a ClosingMapper","message":"sq: WIP: use a ClosingMapper\n\nit\u0027s like map() but it closes the iterable\n\nChange-Id: Ib2e702745fe9f975dbb957c0a9f444eeb0cae5aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/10061d310c5a47d2229fd9b4c92a67a7da32d151"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/10061d310c5a47d2229fd9b4c92a67a7da32d151"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
