)]}'
{"id":"openstack%2Fswift~918141","triplet_id":"openstack%2Fswift~master~I4dfb6b0534c9e4bca34fa0112b61d84d7a28b22a","project":"openstack/swift","branch":"master","topic":"db-insertion-order","attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-05-03 20:55:53.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-05-13 07:26:37.000000000","reason":"\u003cGERRIT_ACCOUNT_7233\u003e replied on the change","reason_account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I4dfb6b0534c9e4bca34fa0112b61d84d7a28b22a","subject":"container: Ensure a consistent DB insertion order","status":"NEW","created":"2024-05-03 20:09:56.000000000","updated":"2024-05-13 07:26:37.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"fb76a3ee2ecc94c8f2ce181556d02a5a4c42db75","_number":918141,"virtual_id_number":918141,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-05-03 22:35:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2024-05-13 07:26:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-03 20:55:53.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2024-05-03 22:11:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-03 22:35:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-13 07:26:37.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"24fbb9a0f0c7b13327c344c1e9f1b3d29be5073f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-03 20:09:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ea2abbf5db02e22606d07f70512f7808109fd8c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-03 20:52:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5ab617c5fe92cee89a005418676c5c23774fe05e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-03 20:55:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1d1841cbe62168da4c2eaf7c5e6b44d3640b9a12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-03 21:22:03.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"da7870821bd636dc0f07a645647d4015744be2e3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 22:11:42.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/afde9bf526e54ffcb6644f0f64eec6a7\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6fc979a56afd4590bc13529d1e20b1dd : SUCCESS in 14m 13s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5c489bec13e84055a08fbca3a10af42d : SUCCESS in 39m 00s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e6e1f8a365d64da49bef4bf6b4f0e61d : SUCCESS in 23m 38s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7373a10f2ada49e0aaaf26b015cee874 : SUCCESS in 48m 03s","accounts_in_message":[],"_revision_number":3},{"id":"f15056504e99cd4e897bc4638fdb5378bbbf92fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 22:35:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7cecea31d7574da1a8874ffa60b38240\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1e70a99373c469aac46b25aa9d91ed8 : SUCCESS in 9m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/5ec6c7b3112d4342a9b79ac6e1d55917 : SUCCESS in 1h 05m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7d385a45cbc4483888afd04732f24a00 : SUCCESS in 54m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d9d8593046b8466da8bc32b5d54e40cf : SUCCESS in 1h 06m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9edf1aec7d1d4eb8a6f04ef6430aed12 : SUCCESS in 17m 36s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fc5f6ff7762c453d96507e816ac039d7 : SUCCESS in 19m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4ceb9e69e009413d90f5c871417b5991 : SUCCESS in 20m 08s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/261c762bda7d4bf087e130089bf02720 : SUCCESS in 11m 18s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8d663d4a9b6f4f318515b2264d71dcb6 : SUCCESS in 5m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9cff77afb7d14e618f8bc2b38539d317 : SUCCESS in 20m 35s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/efa65fbb9c2c4fe284939a52fb805f03 : SUCCESS in 19m 39s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/01cd1dab08284353946f21b22b2b8df9 : SUCCESS in 13m 39s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/18b3346237f6469283f6fe9b4fa03da5 : SUCCESS in 19m 58s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/28370c5876ed483c805f45fdd2e46f97 : SUCCESS in 17m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/df92419254134e63ac02c280568c26dd : SUCCESS in 19m 27s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f0964a6d6f8441f8b1026901a2e20231 : SUCCESS in 15m 48s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1be6a350b7624ec7b93755b26a2eb28a : SUCCESS in 22m 37s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d0f798b5d02a4904ae75e7fd1e135073 : SUCCESS in 15m 46s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/047d3bb703444f869984d8429672d2ff : SUCCESS in 19m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/07382f74de4f4e748ddc31a5e9ed26c0 : SUCCESS in 8m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/34d1527f1ce24aff89d0abafc2b4d27a : SUCCESS in 13m 51s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/35d5f580275849d09d9faaa9c49e599d : SUCCESS in 7m 29s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4a7b95b9acc9477fbe795bdef6a690ba : SUCCESS in 48m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d36d5408bd6346f4b3d43827f63d8a35 : SUCCESS in 35m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/49b4d62271e248fd92ff49fa8f7228b3 : SUCCESS in 36m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b9817a4dc5f84d7eb30dfb822abafed6 : SUCCESS in 18m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0560434dd02a41dcb4f01bc548f80a63 : SUCCESS in 5m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1a85333245894ec6a3d783baf48a30a7 : SUCCESS in 19m 55s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/36de926466944e3db22ba2af153864d8 : SUCCESS in 39m 25s","accounts_in_message":[],"_revision_number":3},{"id":"fb76a3ee2ecc94c8f2ce181556d02a5a4c42db75","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-05-13 07:26:37.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a42fc110f749fa9bb270a9764b930f8cb75e477f","revisions":{"aab45c5c7ad9366102cd90de6e0efd28dbcf28d6":{"kind":"REWORK","_number":1,"created":"2024-05-03 20:09:56.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/41/918141/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/41/918141/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/918141/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/918141/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/918141/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/41/918141/1"}}},"commit":{"parents":[{"commit":"761d9196772bc150fd4923fb04367b919a521d59","subject":"tests: Use mock.patch more","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/761d9196772bc150fd4923fb04367b919a521d59"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-03 20:04:03.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-03 20:05:27.000000000","tz":-420},"subject":"container: Ensure a consistent DB insertion order","message":"container: Ensure a consistent DB insertion order\n\nChange-Id: I4dfb6b0534c9e4bca34fa0112b61d84d7a28b22a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aab45c5c7ad9366102cd90de6e0efd28dbcf28d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aab45c5c7ad9366102cd90de6e0efd28dbcf28d6"}]},"branch":"refs/heads/master"},"42d2ee22af92f7437c73a0d4a8120470da601589":{"kind":"REWORK","_number":2,"created":"2024-05-03 20:52:33.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/41/918141/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/41/918141/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/918141/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/918141/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/918141/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/41/918141/2"}}},"commit":{"parents":[{"commit":"ec3224cc18879c8130421e629134704b6562534f","subject":"Merge \"Clean up watchdog threads\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ec3224cc18879c8130421e629134704b6562534f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-03 20:04:03.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-03 20:52:25.000000000","tz":-420},"subject":"container: Ensure a consistent DB insertion order","message":"container: Ensure a consistent DB insertion order\n\nChange-Id: I4dfb6b0534c9e4bca34fa0112b61d84d7a28b22a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/42d2ee22af92f7437c73a0d4a8120470da601589"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/42d2ee22af92f7437c73a0d4a8120470da601589"}]},"branch":"refs/heads/master"},"a42fc110f749fa9bb270a9764b930f8cb75e477f":{"kind":"REWORK","_number":3,"created":"2024-05-03 21:22:03.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/41/918141/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/41/918141/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/918141/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/918141/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/41/918141/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/41/918141/3"}}},"commit":{"parents":[{"commit":"ec3224cc18879c8130421e629134704b6562534f","subject":"Merge \"Clean up watchdog threads\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ec3224cc18879c8130421e629134704b6562534f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-03 20:04:03.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-03 21:21:32.000000000","tz":-420},"subject":"container: Ensure a consistent DB insertion order","message":"container: Ensure a consistent DB insertion order\n\nChange-Id: I4dfb6b0534c9e4bca34fa0112b61d84d7a28b22a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a42fc110f749fa9bb270a9764b930f8cb75e477f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a42fc110f749fa9bb270a9764b930f8cb75e477f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
