)]}'
{"id":"openstack%2Fswift~918694","triplet_id":"openstack%2Fswift~master~I6463b54daa51152b3166357c743c3d662204d4bf","project":"openstack/swift","branch":"master","topic":"cors-credentials","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2024-09-12 01:26:23.000000000","reason":"Change was abandoned"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-09-12 01:26:23.000000000","reason":"Change was abandoned"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-09-12 01:25:44.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I6463b54daa51152b3166357c743c3d662204d4bf","subject":"cors: Add ability to allow requests with credentials","status":"ABANDONED","created":"2024-05-08 18:58:29.000000000","updated":"2024-09-12 01:26:23.000000000","total_comment_count":17,"unresolved_comment_count":10,"has_review_started":true,"meta_rev_id":"1037fe8585d55b0a082b4309a5205e80eafb1e47","_number":918694,"virtual_id_number":918694,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-05-15 20:27:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-08 20:13:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-08 20:16:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-13 06:51:05.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2024-05-16 14:21:27.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"},{"updated":"2024-05-23 01:20:57.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"CC"}],"messages":[{"id":"ff4fc7147bdbee3e9e77a6b534d76a469c3b4fc3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-08 18:58:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5edde326149de0e6adc41803feaf0887dc21e49c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-08 19:01:24.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"37cb2e9d88e3ea10de0137860f82dde1bc0878ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 20:13:42.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7241c68ce2c409c8ad8758939ab853f\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7dc13e63f59449a1bcc7408f39b715b0 : SUCCESS in 52m 42s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cea240d466d541378aba58b1bf302c6d : SUCCESS in 1h 12m 39s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4f56acb77f3c49ab90110f91372edcf0 : FAILURE in 1h 10m 54s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ca8f783a92884cb7b0aba3206f8c5a44 : SUCCESS in 19m 19s","accounts_in_message":[],"_revision_number":1},{"id":"d3b4c05a49c10c735b22683c420d10448798fde8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-08 20:16:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7999b016049a4bb3a5e7f98c13803680\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d80b3cad38244812921961441b77544b : SUCCESS in 9m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/b9c41e3a5b6b4f75a70d883e94fcec23 : SUCCESS in 1h 01m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1741ad85879246fabdc6cc7ce48d39cf : SUCCESS in 42m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8e1f9767c6074bd2908c9c2d4f9be34a : SUCCESS in 57m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2dcb955f37334d32b666c6f1a3fad142 : RETRY_LIMIT in 4m 53s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fae1adff9d624a06bb5d5d0a8a4bd228 : RETRY_LIMIT in 4m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8d8bd7d5ed0045afb50ea2cf21d71477 : RETRY_LIMIT in 4m 49s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2b352204d7574ae590cf63fe69def72e : SUCCESS in 11m 18s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f5a1150677ca48d8a4cec465a1303476 : SUCCESS in 5m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a0e4a072cba24ddc86ac06da24b66a73 : SUCCESS in 16m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/344c488ccc774a4b9b772f69040d245b : SUCCESS in 19m 21s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6738ead3090847cea083f7d87f94f123 : SUCCESS in 15m 22s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/8df8b1b0662d4ad39aff83dbff3d35eb : SUCCESS in 13m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7e1524534248443b86082452d826c223 : SUCCESS in 14m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1f72189e9f17425ab1e4277ad6fb9865 : SUCCESS in 14m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/17171524eaa8461ca602bcfa16d8d873 : SUCCESS in 15m 59s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9eedab0364a542519679a8f9c4364881 : SUCCESS in 18m 10s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9270d0b0468b40ed88befe00d924a1fa : SUCCESS in 14m 57s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1213a5ca05b54de089eed65c2440a422 : SUCCESS in 18m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f0f6c8c34b8546089922d855121c598a : RETRY_LIMIT in 2m 55s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4d6f204f9d534891a1097c381dec6760 : SUCCESS in 14m 17s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/70229b07899e4edeb91dc8aab668b910 : SUCCESS in 7m 22s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d1ca06bbbe49499086065b6855f5dd1f : RETRY_LIMIT in 2m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/df0aa8dd4989470e88ed6dd5b32bd490 : SUCCESS in 52m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2fcb13b82fc24e4ba07f907055e9e598 : SUCCESS in 1h 11m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e5d8caadcd0846b2aaba48086e9c7f0d : SUCCESS in 21m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ca3922145c5f4689affcc29e4252cb1d : SUCCESS in 3m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/eaa8cb9fe6064b50beaa7fa3002cfabc : SUCCESS in 21m 50s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8c9fb0f8e651433daa37f1f07bff2295 : SUCCESS in 1h 04m 59s","accounts_in_message":[],"_revision_number":1},{"id":"42c2aab7812b880ef84ebb292c910b102ca22004","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-09 16:35:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d0738d97d22bf73d8653387cc60d2203529ac3b2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 17:24:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/001195f4735f4c8487ba743cefd68bd7\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a61a3423e84b4e9490ec67a00995f6a0 : SUCCESS in 15m 43s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1782449c03b74580b1f8f337fcb0d26d : SUCCESS in 46m 11s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b4a5c9bdde85473b8c16d8765734632c : SUCCESS in 22m 33s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3ee897210d544549a6838c1f0b734588 : SUCCESS in 47m 22s","accounts_in_message":[],"_revision_number":1},{"id":"5eba7a3f6d17f1c46d644c461f0ec28102bd822c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-09 17:33:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3dcd860e3d7745c38e03220bcec5ed6d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1db611f08ea409c839959103546a41b : SUCCESS in 7m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/7c762ffb9277450895b39e9c2962e688 : SUCCESS in 51m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c7829756acde431ab53ad0750cb6dec9 : SUCCESS in 42m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4a6daec38551499da9f600c83da26665 : SUCCESS in 50m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ee55476dcefc491e97b47e568652c3fd : SUCCESS in 25m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d8659b57d2114a4eb3571721fb889672 : SUCCESS in 26m 13s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7f418773b3c144b5b46dedb9035388c8 : SUCCESS in 21m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/655a38b28a494a8e86e62470e24b2477 : SUCCESS in 11m 19s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/fe0de6f105d845368e34ee8d758e0d72 : SUCCESS in 5m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e461b9ba879546e7b9d4dde41d8bb152 : SUCCESS in 19m 40s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f14c11f7f8d74019806c01106d239513 : SUCCESS in 19m 44s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/28d3de21194d42408348faa004e95919 : SUCCESS in 13m 58s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/806c0b4fbec94d4087a21ee50b91a10b : SUCCESS in 17m 40s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/abb2a3b156f548be9cd58f4bb426000c : SUCCESS in 18m 19s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d02bd7fbec7e4551b11ddc2bb8b78ee9 : SUCCESS in 19m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a3795a4b2c40414bb986fa9279258af1 : SUCCESS in 20m 11s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/46662eed9ba74c68a9a85c8450f82e94 : SUCCESS in 15m 42s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/74739a0085864bb684548fb1475731fa : SUCCESS in 16m 32s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/462dacfc52ca4a69a8b29f4d8a81c8b8 : SUCCESS in 16m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f2d76171bc7040359d5d0e09db5a297a : SUCCESS in 6m 16s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7376839394ee491d9ed5cecec29ee3af : SUCCESS in 16m 49s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a42fa2627a54476e8fd27d60ca7daf3a : SUCCESS in 8m 03s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f1a3a0f9b3584aa28d8046124a90f4bb : SUCCESS in 52m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0a7f94ac9b674f9897e128185c7e17f1 : SUCCESS in 45m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5642f0fec8bc468ca7e05028a8d62523 : SUCCESS in 41m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b17d62df0bbe417f9b599f7310d6be38 : SUCCESS in 19m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47c9cae4ecc04a0ab27ee972028ca1f1 : SUCCESS in 4m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5a4399df3d254770a93739e78823b021 : SUCCESS in 21m 47s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0a9e43aa7e0742a4bbeeb58bec389323 : SUCCESS in 49m 06s","accounts_in_message":[],"_revision_number":1},{"id":"cc30e2c032b1bc133230cb5c860ce9ea724b6152","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-05-13 06:51:05.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3f1967a99dac74ac1dbb62d73808f78c3903d5af","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-15 16:08:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6e8dcae286a15d31f98278ec67b7573c561acb40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-15 18:14:14.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"966ad876987e2ffd9c40efc987219fa504af93af","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-15 18:33:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b0c3b3f88a8bf93f288760097e1af9ad0a848548","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-15 19:12:13.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d87aa5d271c8cfed35fa58ed77a09f56c17806ca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 20:18:41.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2865850bc0a449aa800be3dc654cfd8\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6ec0315139174e4881a82ee2f7749d9c : SUCCESS in 46m 38s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/54c557dbe976472c8f4eb11eb1f09d3a : SUCCESS in 16m 40s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e8cc30a0d58d454b919c6d327369cbdb : SUCCESS in 1h 04m 44s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e176283da312491c93da9a9f2b5956bc : SUCCESS in 19m 33s","accounts_in_message":[],"_revision_number":3},{"id":"6606e01429a7e6d89d774dd19f91e0abfadc095e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 20:27:49.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a6a7c19ea1c422595c6e3e57935edfd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b5a0e46b88843b7b894f76545774886 : SUCCESS in 7m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/37ced77d5eb146dda9120ab4be325283 : SUCCESS in 1h 06m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/146fa35c3075439f9f41166c84209ae7 : SUCCESS in 50m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/967eefa419384ab3a5e0fd63781200a2 : SUCCESS in 1h 03m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f50c219dc4c7495ca22a70a733314ea6 : SUCCESS in 17m 53s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f85ce05b61274fd2a83cde766456ba75 : SUCCESS in 22m 28s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e04970d2c8cc47ec85a435469a629567 : SUCCESS in 23m 08s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4f309fc5c5e94d0dbe1b884c05e8ad54 : SUCCESS in 11m 03s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/c5f0fe260d0e4344a79851f601f58cee : SUCCESS in 7m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a0f3f8f855104f37a8701b8a7ef52adc : SUCCESS in 20m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6cd17576c3694938a1a96e186863b1d1 : SUCCESS in 17m 24s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/98813bd0e04e4f7688ea080175bc4b89 : SUCCESS in 16m 11s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/a72ae3c794cc4e6282bd594ea79bf9de : SUCCESS in 15m 04s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/984e3c36a62b4b2cbda6f04aca0fd1f7 : SUCCESS in 19m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5c5ceae7f30a49598b5d54c50b3fa34b : SUCCESS in 16m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f9034ef82d82435d8b36276ce3c6a93d : SUCCESS in 22m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bd7044b4a0a14c248b05deffa943ea59 : SUCCESS in 24m 51s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/81f03d2e8f5645b88a48e0f2d62a97e9 : SUCCESS in 15m 38s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6264f6af2c894cb3b4f414a74ceaf223 : SUCCESS in 14m 47s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/74ef6220943d48eb895a6f956ea0e8d0 : SUCCESS in 20m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2ccd3ed287014930a86bbb8e17f9e822 : SUCCESS in 5m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/806c2ae6ff3941ca827dbab026fce570 : SUCCESS in 14m 34s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/708b494e5d3e437787ba474ffcbaedb0 : SUCCESS in 8m 26s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3e806b1a72914527b2b6a2e82915b25c : SUCCESS in 54m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c4781554d1a6440f9c219f670b0edb44 : SUCCESS in 47m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bd1342c21201422c8115632d9103aa10 : SUCCESS in 58m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/efa30131712e42049e605663cd6240f5 : SUCCESS in 20m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/087ef2e4c6f64daa9f62518a71d5bb22 : SUCCESS in 4m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/33f1fdc0e68d4faf875a140dc83cfc46 : SUCCESS in 20m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/71e22290287b4488a65d8d3a66be4872 : SUCCESS in 53m 42s","accounts_in_message":[],"_revision_number":3},{"id":"a2829a0bf64f7fe4e41b63a97306b3b301a965f2","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-05-16 14:21:27.000000000","message":"Patch Set 3:\n\n(10 comments)","accounts_in_message":[],"_revision_number":3},{"id":"f2797b5941c56411d3fad06e06f46b7bc85ae8c9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-12 01:25:44.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1037fe8585d55b0a082b4309a5205e80eafb1e47","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-12 01:26:23.000000000","message":"Abandoned\n\nSeems to have been an app error.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"87c2dfb3dc45bc02b017bbe4a6d9623a285b2040","revisions":{"893f50c1681843ad51dfef908e27acab1fd53d2b":{"kind":"REWORK","_number":1,"created":"2024-05-08 18:58:29.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/94/918694/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/94/918694/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/918694/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/918694/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/918694/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/94/918694/1"}}},"commit":{"parents":[{"commit":"9da22bb5fedd900bbe004aa76fb163abfb769bd7","subject":"Move statsd testing to its own module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9da22bb5fedd900bbe004aa76fb163abfb769bd7"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-08 18:36:16.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-08 18:57:19.000000000","tz":-420},"subject":"cors: Allow requests with credentials","message":"cors: Allow requests with credentials\n\nSome applications expect to be able to send credentials as part of a\ncross-origin request; browsers will block access to the response\nunless a `Access-Control-Allow-Credentials: true` header is included\nin the response.\n\nAlways allow such requests -- Swift does not use the credential\nmechanisms CORS is designed around anyway.\n\nChange-Id: I6463b54daa51152b3166357c743c3d662204d4bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/893f50c1681843ad51dfef908e27acab1fd53d2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/893f50c1681843ad51dfef908e27acab1fd53d2b"}]},"branch":"refs/heads/master"},"cfc32914a0e66426e773fb20e43fe4b737c2f7bc":{"kind":"REWORK","_number":2,"created":"2024-05-15 18:14:14.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/94/918694/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/94/918694/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/918694/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/918694/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/918694/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/94/918694/2"}}},"commit":{"parents":[{"commit":"9da22bb5fedd900bbe004aa76fb163abfb769bd7","subject":"Move statsd testing to its own module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9da22bb5fedd900bbe004aa76fb163abfb769bd7"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-08 18:36:16.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-15 18:12:47.000000000","tz":-420},"subject":"cors: Add ability to allow requests with credentials","message":"cors: Add ability to allow requests with credentials\n\nSome applications expect to be able to send credentials as part of a\ncross-origin request; browsers will block access to the response\nunless a `Access-Control-Allow-Credentials: true` header is included\nin the response.\n\nSee https://fetch.spec.whatwg.org/#cors-protocol-and-credentials for\nmore information.\n\nLook for X-Container-Meta-Access-Control-Allow-Credentials on the\ncontainer to decide whether to allow such requests, similar to how\nother CORS behaviors may be configured.\n\nChange-Id: I6463b54daa51152b3166357c743c3d662204d4bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cfc32914a0e66426e773fb20e43fe4b737c2f7bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cfc32914a0e66426e773fb20e43fe4b737c2f7bc"}]},"branch":"refs/heads/master"},"87c2dfb3dc45bc02b017bbe4a6d9623a285b2040":{"kind":"REWORK","_number":3,"created":"2024-05-15 19:12:13.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/94/918694/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/94/918694/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/918694/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/918694/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/94/918694/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/94/918694/3"}}},"commit":{"parents":[{"commit":"9da22bb5fedd900bbe004aa76fb163abfb769bd7","subject":"Move statsd testing to its own module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9da22bb5fedd900bbe004aa76fb163abfb769bd7"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-08 18:36:16.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-15 19:12:06.000000000","tz":-420},"subject":"cors: Add ability to allow requests with credentials","message":"cors: Add ability to allow requests with credentials\n\nSome applications expect to be able to send credentials as part of a\ncross-origin request; browsers will block access to the response\nunless a `Access-Control-Allow-Credentials: true` header is included\nin the response.\n\nSee https://fetch.spec.whatwg.org/#cors-protocol-and-credentials for\nmore information.\n\nLook for X-Container-Meta-Access-Control-Allow-Credentials on the\ncontainer to decide whether to allow such requests, similar to how\nother CORS behaviors may be configured.\n\nChange-Id: I6463b54daa51152b3166357c743c3d662204d4bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87c2dfb3dc45bc02b017bbe4a6d9623a285b2040"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87c2dfb3dc45bc02b017bbe4a6d9623a285b2040"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
