)]}'
{"id":"openstack%2Fswift~919219","triplet_id":"openstack%2Fswift~master~Iba4a9271910b032b8726c0dc605d95cf8da05e44","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"Iba4a9271910b032b8726c0dc605d95cf8da05e44","subject":"Move old rolling-upgrade jobs to point to EOM tags","status":"MERGED","created":"2024-05-10 19:42:10.000000000","updated":"2024-05-14 20:43:17.000000000","submitted":"2024-05-14 20:40:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"919219","meta_rev_id":"7e5ca9aa47efececcf35e860d7306dd061be2a9c","_number":919219,"virtual_id_number":919219,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-14 20:40:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-05-14 19:37:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-05-14 19:37:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-10 20:42:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-10 20:49:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"33b6e7d220f119931d2755027f0f172b33be4419","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-10 19:42:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e51e62b906bebf43a0af852e9d8b15997295823b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-10 20:42:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4eee55f3e444e2581f44f06e76aed89\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b928012291bd4c45bfb064c2a5372c3d : SUCCESS in 16m 11s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f94e3a3f0fc941389359ff087fb2440c : SUCCESS in 18m 26s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4793faa2620541c6980324fea88be121 : SUCCESS in 20m 57s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1fbd1ec344c14fb9b1db95fd3b82a762 : SUCCESS in 58m 04s","accounts_in_message":[],"_revision_number":1},{"id":"8db41bb6c28d2594451494f6c0a323519668675f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-10 20:49:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f50f0fe3774d451ba960e13afef6c3c1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecab5c471c9449bebb55d8b1bb7bb939 : SUCCESS in 8m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/8b6d31f920c14871946cb26fda6f7c4d : SUCCESS in 53m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e1940d44862f4428b74c7e7bdcea0cc2 : SUCCESS in 40m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/59b36b9b74a144378028110df37cd4e2 : SUCCESS in 56m 32s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a8394bc2315d4b5a8c0c1739144d3d60 : SUCCESS in 18m 21s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/995ee202ce5a48ff850374b4f6bf2fbc : SUCCESS in 20m 09s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3f963e268a904cdab068550b0ac05220 : SUCCESS in 18m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5626384112e64218a11cce7221352918 : SUCCESS in 10m 16s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a6fcc1ed29fa4759923c986481d67b7e : SUCCESS in 7m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/66dd24b8dded458aac76776883118f18 : SUCCESS in 21m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fc8b958b49f04cf6a3a793fe582401a7 : SUCCESS in 18m 07s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0a7fa233c3834e46bdde78a51e0e3260 : SUCCESS in 18m 33s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/ce75dbde85024fca8a687707ca669e4d : SUCCESS in 17m 23s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4b485247abae4b29aded3c006de7c508 : SUCCESS in 25m 58s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/741dda2dea9b4475a2523cc0c4c66995 : SUCCESS in 17m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ba2384d637c2486eb53900efe40ba62c : SUCCESS in 20m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f75a4b63a08f4525acf49cb553da8e11 : SUCCESS in 24m 28s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/127324f3c3a94c34a624a02e1a181c01 : SUCCESS in 15m 43s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0c3af544c3104feab37f273f15c228e9 : SUCCESS in 16m 12s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/557bb4f440704a50a05a7cfed25c651a : SUCCESS in 21m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/500e4abfb20c40ddb10839a4fa75bcfd : SUCCESS in 7m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9a5e638b9e97419ea15077cbdfa60475 : SUCCESS in 17m 11s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f36429111988436d9a547fd6b188b903 : SUCCESS in 6m 18s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5a3dd5c43067475c8546059e26ef3719 : SUCCESS in 52m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/11908479392544a1bf095fb350d6b620 : SUCCESS in 54m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/842270f475f5490abb7cce9f989c337f : SUCCESS in 36m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/81406863f3a2461aa7f28979f35961e3 : SUCCESS in 17m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03077a59edf944289b0cce1432936229 : SUCCESS in 4m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/55eaf6ebb4374fcb9098144b89336722 : SUCCESS in 17m 57s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7c29e9106b2a4ba48717806fce315e80 : SUCCESS in 1h 00m 09s","accounts_in_message":[],"_revision_number":1},{"id":"209930e6383d5badf6d39ba844d7c00bfeb9ea7b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-10 23:42:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"486f621a42112621ca09a05af8d5440bb49a10fa","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-11 00:15:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bdbd213e10e9479ab524f406715be823\n\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8721d069ae26474ca04356a7a6d0f58f : SUCCESS in 16m 55s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1e8b25c482b34588a558d14212a7120f : SUCCESS in 17m 10s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/4577e466d5e042dd8c8740d19dfee36c : SUCCESS in 17m 06s\n- swift-tox-py36-centos-8-stream https://zuul.opendev.org/t/openstack/build/9f18c104008e42098de12dee99b709c1 : SUCCESS in 21m 23s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/0059c89cfbca4863bc0423f49363ca84 : SUCCESS in 15m 47s\n- swift-tox-func-encryption-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/495a6a0373aa4b908b73cf96edd3b8d4 : SUCCESS in 17m 17s\n- swift-tox-func-ec-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/9b2dd8e689e0489b8bbf108731683ef0 : SUCCESS in 18m 41s\n- swift-multinode-rolling-upgrade-victoria https://zuul.opendev.org/t/openstack/build/55ceaa473d924fb1929e482ab7e27125 : SUCCESS in 19m 33s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/cc52d8b5e0e2425daa3c7652b78cb239 : SUCCESS in 19m 20s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/475b55a36de94c1285a8c165f6df4f16 : SUCCESS in 24m 23s\n- swift-multinode-rolling-upgrade-yoga https://zuul.opendev.org/t/openstack/build/f069d3fa7fec4c5791102ccdc9dd27ec : SUCCESS in 24m 19s\n- swift-multinode-rolling-upgrade-zed https://zuul.opendev.org/t/openstack/build/15099ae9b06048b38893bdec9ef301d6 : SUCCESS in 23m 52s\n- swift-multinode-rolling-upgrade-antelope https://zuul.opendev.org/t/openstack/build/10ba151962f0451684ebb7af82baa9b9 : SUCCESS in 19m 23s\n- swift-multinode-rolling-upgrade-bobcat https://zuul.opendev.org/t/openstack/build/96d9adedd9e644e3bb8e0449caf85e58 : SUCCESS in 23m 59s\n- swift-multinode-rolling-upgrade-caracal https://zuul.opendev.org/t/openstack/build/10460de7aa5b4a1e91ea55df8c50bd32 : SUCCESS in 19m 13s\n- swift-multinode-rolling-upgrade-master https://zuul.opendev.org/t/openstack/build/542bb7ec9d0d4488985c2cbfcbf03874 : SUCCESS in 16m 01s","accounts_in_message":[],"_revision_number":1},{"id":"e99f930c471820dda3df69e7a297573409f67b9e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-14 19:37:45.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5ec427ae52d7fba4c42e032ba78975cb605a4fa4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 19:38:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4dc299b033ee3f42bdb355d6174619bde0989cfa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 20:40:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d636d0408c494ac0a1613f44349efd09\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78d47cceaae44393a98794ebc4ecd24f : SUCCESS in 7m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/f90b3aa1546a43929ec0a5c86fea043e : SUCCESS in 45m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/223c3adaffda4a0998876cfd1d52e21f : SUCCESS in 36m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0816594c91714f308ed00be83b2ee387 : SUCCESS in 47m 10s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/5a04aa4f070f425f9028a423316b0b5c : SUCCESS in 10m 34s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/fb87ff1556714841ac23af914c5aee8b : SUCCESS in 5m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/355eb42d13e24f66a0f5ca695a2ef5bc : SUCCESS in 18m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/609fc467242548ed9ef0347b1a0bd464 : SUCCESS in 15m 59s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d1630917eece412395c2bce3ef57c860 : SUCCESS in 18m 11s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/00590812ad59467888a261edac07cedf : SUCCESS in 14m 59s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/390e8418aaa7420baf09417ee04d6914 : SUCCESS in 17m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/386c4c3e91f243e6b38c868b4f6ad58e : SUCCESS in 17m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6e95cc7f37404240b34dab6703e0e9f5 : SUCCESS in 17m 45s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9c296b5045b249ed8f81f0934be0adcd : SUCCESS in 20m 48s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8df55d1c2c6e43b0bc36d32b7115d121 : SUCCESS in 15m 27s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/831f8638f2b64a1d9a16dfeddc918537 : SUCCESS in 14m 01s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2f21c8e855e048a3a03e64dd1925f413 : SUCCESS in 15m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3525da11959d402e88d1c64236c84ea0 : SUCCESS in 5m 56s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/510da0534d3640d2832923389882f63d : SUCCESS in 7m 06s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/04dfdafbbca742908e25da70f6c8ff44 : SUCCESS in 57m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e672fdea4c5f4b47a3e430fa498c873a : SUCCESS in 40m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3279a543561a4659a160c7b204ab5dd6 : SUCCESS in 52m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9603bb83eb3d483eadb0ef96b2ddbc44 : SUCCESS in 20m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81bd06a964804fcc9fac82ca88103a52 : SUCCESS in 4m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dde73e99b1464a69bfe6c0e1580edd21 : SUCCESS in 21m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/310e69c85ffe42729eee7058553ac867 : SUCCESS in 1h 02m 26s","accounts_in_message":[],"_revision_number":1},{"id":"8188ccb98b4c208fff9c6cc921bd03ccc86ad474","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 20:40:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7e5ca9aa47efececcf35e860d7306dd061be2a9c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 20:43:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ed2e33903de414e8a115f243f47198c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cc9a29e46254648ac744019a0c5f7b4 : SUCCESS in 47s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/6d05076111d744ccacc94ddaedf27f0a : SUCCESS in 31s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/a9d3b22a880f4729b650e43bb38b56ed : SUCCESS in 41s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0444a056e7c9276cab626eda9bd1e74bf3bff0ac","revisions":{"0444a056e7c9276cab626eda9bd1e74bf3bff0ac":{"kind":"REWORK","_number":1,"created":"2024-05-10 19:42:10.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/19/919219/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/919219/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/919219/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/919219/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/919219/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/919219/1"}}},"commit":{"parents":[{"commit":"47cde0b28b53531927cd8e52f11d55fb3d97a8b3","subject":"Merge \"Test under py312\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/47cde0b28b53531927cd8e52f11d55fb3d97a8b3"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-10 19:42:00.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-10 19:42:00.000000000","tz":-420},"subject":"Move old rolling-upgrade jobs to point to EOM tags","message":"Move old rolling-upgrade jobs to point to EOM tags\n\nChange-Id: Iba4a9271910b032b8726c0dc605d95cf8da05e44\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0444a056e7c9276cab626eda9bd1e74bf3bff0ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0444a056e7c9276cab626eda9bd1e74bf3bff0ac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
