)]}'
{"id":"openstack%2Fswift~919609","triplet_id":"openstack%2Fswift~master~Ia83ab181277e04e473e33680b3f2d686f534a084","project":"openstack/swift","branch":"master","topic":"lower-constraints","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-05-16 20:24:59.000000000","reason":"Change was submitted"},"34892":{"account":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"last_update":"2024-05-16 20:24:59.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-05-14 18:48:18.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Ia83ab181277e04e473e33680b3f2d686f534a084","subject":"bump lxml to 4.2.3","status":"MERGED","created":"2024-05-14 18:05:00.000000000","updated":"2024-06-07 19:31:21.000000000","submitted":"2024-05-16 20:24:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"919609-lower-constraints","meta_rev_id":"3f4508617f1a36619f72364090608d5c099e95fe","_number":919609,"virtual_id_number":919609,"owner":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-16 20:24:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-05-14 18:48:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-05-16 19:20:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-14 18:06:16.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2024-05-14 18:06:16.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2024-05-14 19:06:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-14 20:11:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"78c2a9a0fdf95535c34ff8ca6b708443d84af440","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-05-14 18:05:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"21dffe1e89c1445f05bc3f25abee06446ea765e1","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-05-14 18:06:16.000000000","message":"Patch Set 1:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"1e728233157669a53e8a694b81b87cad6cd58b17","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-14 18:48:18.000000000","message":"Patch Set 1: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"42e8f4ebb780573697bbaf4c457afd8a4832d319","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 19:06:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/568d104e126143ee9517fb76dda75fed\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6069d204fe56461ebe2f1fe4796195d7 : SUCCESS in 15m 54s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/40c29e150f91417081d31c079b948274 : SUCCESS in 15m 25s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/eecad2325e474b569a8acc8d05b3363a : SUCCESS in 59m 45s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b44672e70cc24dc4ab10c09b196d9abf : SUCCESS in 19m 45s","accounts_in_message":[],"_revision_number":1},{"id":"5b755c076707ef7fb07ffcd975fb4c95077769ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 20:11:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7ea5bbb5e039460bb04acd9b35ad0063\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69defc8d04a5433b85edc5953693e9e9 : SUCCESS in 8m 39s\n- requirements-check https://zuul.opendev.org/t/openstack/build/cca55cdfd0a94ed38af53d53fc53336c : SUCCESS in 2m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/e3378db4d7ec4aef8df3788d03562f24 : SUCCESS in 55m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/66bc634fdf9844bf996d3fe4ae7a0fbd : SUCCESS in 39m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0473154d4b8e4e6cbfd9ba1161d4afc9 : SUCCESS in 51m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2234774ea57f47a5bb01e8d05255955b : SUCCESS in 22m 15s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a02faf4a0067408883ad088fbfe03780 : SUCCESS in 23m 29s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/837239bd184b4bcc905d72082e92b36f : SUCCESS in 24m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fb99ad8fc6ce485a9e2d7c652278be9c : SUCCESS in 10m 56s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/03ad295faf2d4c358db73f1c946b7e2c : SUCCESS in 5m 38s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3b36e6b6d73b4357b625d22fcba08928 : SUCCESS in 21m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d25f57ee8025456fa96b8b1f198e3c2e : SUCCESS in 19m 05s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/caf92b4e8cf44033b8cc548bd61aed1f : SUCCESS in 22m 14s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/a28cb76d158043509a589975c5277fb0 : SUCCESS in 19m 46s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7aaa84edec054160925e014696ee8419 : SUCCESS in 23m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/676bf4ed07c24958a59b02e94b5e3656 : SUCCESS in 18m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/57f76ea85ae84f8c90eb87c876e71548 : SUCCESS in 18m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1932a87a1ce24f1d934a1705c6fab7ea : SUCCESS in 22m 28s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3de34a633767417d85d2d0edf266726a : SUCCESS in 17m 32s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6350bdf70f544cb8a62f4a99db75bd2f : SUCCESS in 16m 33s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/01a476ce158f41e19e7b802b1001f0d0 : SUCCESS in 18m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2501f5e0a98a4db98d84dd459e96ef1e : SUCCESS in 6m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/27577eaa9a834029a786451689e3a6b1 : SUCCESS in 15m 47s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/28870d7013ff48fba05577a3f091aaaf : SUCCESS in 7m 12s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b92001e019f04b24a2d16711ffd011d5 : TIMED_OUT in 2h 00m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/258eb004a1bb4270bd7f5928e2d3ecdc : SUCCESS in 55m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/609b182e7b9f48a4afab5c9b44928318 : SUCCESS in 52m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a72c86eacbff4afd8cf405c163f6965d : SUCCESS in 20m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bf3c9e6baf34db5b25c041ccf2fa3c8 : SUCCESS in 5m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/073e760f66e14c1aa620c9e364957472 : SUCCESS in 18m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eeb976cefbf3455fbfa2d931845709bf : SUCCESS in 56m 28s","accounts_in_message":[],"_revision_number":1},{"id":"48f648482676882e06a242a48ecc30deb2d35581","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-15 16:00:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"40322cbec38335022bf4a02eaf80c8eb44997188","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 17:13:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8734be15737d43a68f7580f5d8de9847\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c7da4f7f99049018a2d302d20402ffb : SUCCESS in 8m 12s\n- requirements-check https://zuul.opendev.org/t/openstack/build/d6cc31ba37d441ba87bd0ad9e1fd0a1e : SUCCESS in 3m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/8f40b1cef0174a849bb92505bf3dbf10 : SUCCESS in 51m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5ddea3118ede4db6b6a0d63da3b9f36a : SUCCESS in 1h 07m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b35bac36f09047ff8b81363ede791f61 : SUCCESS in 50m 09s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ab9ba8fcf6684f9f8656e0f1962a6486 : SUCCESS in 22m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/926589439f4947c2b0c6d972f4261677 : SUCCESS in 19m 43s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/61277f67a3684269ae43f222727db9bd : SUCCESS in 19m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fdac88d5684240d8ae756293feabb5a5 : SUCCESS in 12m 03s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/44a5c2fd30914f15a814902ef0d0d386 : SUCCESS in 5m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1e91468a75de4541b5b2d02544e7226f : SUCCESS in 16m 32s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3fa98eb7bca74cf8bf14cb31433e6291 : SUCCESS in 19m 49s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b86864504e9749b28ed0c282e058188b : SUCCESS in 18m 17s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/83c02ddf7a07471c96e8d3e6734b73ee : SUCCESS in 19m 10s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/71b445fa71644b0d87b1e1ee0715177b : SUCCESS in 17m 42s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/330dcc84d1d74ca3badd853d69196b93 : SUCCESS in 18m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/45ea25e63b04459fb0d74eaf37bd85b5 : SUCCESS in 19m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f3c312ad1f9c476c8d2594a80c805393 : SUCCESS in 21m 36s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6199cf1393884a1d8d46fa6265dd4b61 : SUCCESS in 16m 36s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e5f6366bc59b4a54806c48f422b089eb : SUCCESS in 15m 49s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/581fb79d6c7d447fa2bd6754dadc74b8 : SUCCESS in 14m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cba72cfb3ed34e1dae67b0d93e5c216b : SUCCESS in 6m 36s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cbb51476781b4a0cb892abf443a4057e : SUCCESS in 14m 18s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/eff939492bb2410f9532654f0879344a : SUCCESS in 8m 01s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0b23c1ee496c4986ad0ce06ef4bc3887 : SUCCESS in 57m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/246d7f20b50e4fb9a9c0215539f9060d : SUCCESS in 51m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a83d204154b242bb92fc9680215aa566 : SUCCESS in 54m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/759cb11cd4a444cc98502b5d24a282e8 : SUCCESS in 15m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/756eb466b956454990a7508bbcaee18e : SUCCESS in 5m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2748191dfc814de6b355b999affb4ce1 : SUCCESS in 19m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/68ec3134eaaf4699874c578e161020b1 : SUCCESS in 57m 24s","accounts_in_message":[],"_revision_number":1},{"id":"a67f12f8e6ebc7008209b04cec2f9c9b921314e3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 17:14:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fef6e3663e024fd4bf7b82765487946d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b4828a6d0acc42a0aa8be0a8bdd30de6 : SUCCESS in 1h 11m 02s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1db6842895d943eba6c0e040889eb4b6 : SUCCESS in 16m 56s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c0ff4d00522a451b8dc4a161a94e37bf : SUCCESS in 1h 12m 56s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b9d9be06aa9d4c3d8183b8f5948db491 : SUCCESS in 21m 53s","accounts_in_message":[],"_revision_number":1},{"id":"84f98b9615c10d624c809c7b774b285f72682561","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-16 19:20:20.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"44921c0e5d832553bf3982002ba49309619ef931","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 19:20:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"36a1903bae125db5fe1af03964fe2904f1e4e7e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 20:24:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6beeff65241343d094b1a66f6bdc5612\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63e2f9e8597f4ae2ae217daec3638d36 : SUCCESS in 8m 48s\n- requirements-check https://zuul.opendev.org/t/openstack/build/84646fa4ec254aef8b4c254f2e1b4c2e : SUCCESS in 2m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/b68460c8d9e34fa6853733fb35e4955d : SUCCESS in 1h 02m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/075ca21f345349ea80eeb5a5692ae14b : SUCCESS in 53m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/27aad78ff4c446968d5884223ad5a7c9 : SUCCESS in 58m 28s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/9081d9d581704a63804837d6df0ed32f : SUCCESS in 13m 51s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/f03d49ec97f4446e9266e9c80e1eed38 : SUCCESS in 6m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/710d12585fb0422ea4407a0f153b421b : SUCCESS in 13m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2ed548ac802b44a5bb75e98a8c6c3e66 : SUCCESS in 16m 21s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0874eff6db2740b1a51ea86ed7f480b6 : SUCCESS in 13m 49s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/5d9040ecd44f42f6867c652c160a9a2d : SUCCESS in 18m 55s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/aab695c2288b4ed6887f2d2f8f82d8ec : SUCCESS in 23m 31s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/dfd9615f5149410cb2b7ff61215c05ca : SUCCESS in 18m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/04940b7f5f76484c88b8d11fa3883fd1 : SUCCESS in 18m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cad6eb9da96544aebb2a3cf9c4b81ba3 : SUCCESS in 18m 37s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5550e5f79a9a4b60959d268d281ca1b6 : SUCCESS in 16m 26s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/68108a2664d04e528503a50ea9825c15 : SUCCESS in 14m 24s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8c3a6b82f8c94ec3b5d13873af86beca : SUCCESS in 17m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/344132cd8697471f9a40f7e3d39e53aa : SUCCESS in 7m 05s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2561ec876c12480ab082a75f04759091 : SUCCESS in 7m 42s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/40e87ebf434d4e46866e0a908f790449 : SUCCESS in 1h 03m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/aa245a1af7b24f32845070e90c269836 : SUCCESS in 53m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/964cfdfeb9ea4ea4b3011ba4d9b2dc63 : SUCCESS in 48m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d71062fe6fe44c58ad8778c006e055ee : SUCCESS in 18m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3c7bcb41109495bb76e636f34294da8 : SUCCESS in 4m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/47075937a59848ebb9a28d8c852ae68c : SUCCESS in 17m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b3a2e4eb894a422fbd92b3936fe9faed : SUCCESS in 45m 16s","accounts_in_message":[],"_revision_number":1},{"id":"9c19b99ac0333fc83a6332daf70bcf4775c2ff81","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 20:24:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ced2922779707f02c15f0051d3de1e59f36f4ec8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 20:26:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37e59988b1164194858f3690efa58a1c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a32c9d13dfb4e4bad14211a953bfd42 : SUCCESS in 53s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/a5c2a4620e0c457ea9f13f108543d395 : SUCCESS in 41s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/f9c90e8c981b462f98f595ce94821ad5 : SUCCESS in 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1eefda6838b8d5d22ddf866a881685e1595eaa4b","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-05-16 20:48:25.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"3f4508617f1a36619f72364090608d5c099e95fe","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-06-07 19:31:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"54e13ee9b5e46209a62e3a13bc03b6edbd225105","revisions":{"54e13ee9b5e46209a62e3a13bc03b6edbd225105":{"kind":"REWORK","_number":1,"created":"2024-05-14 18:05:00.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/09/919609/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/09/919609/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/919609/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/919609/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/919609/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/09/919609/1"}}},"commit":{"parents":[{"commit":"5ba1fbf86b961915b09f89ef4a2a437cf8d4df5c","subject":"Merge \"proxy_logging: unit test first-byte.timing metrics\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5ba1fbf86b961915b09f89ef4a2a437cf8d4df5c"}]}],"author":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2024-04-26 17:21:21.000000000","tz":-420},"committer":{"name":"indianwhocodes","email":"nairashwin952013@gmail.com","date":"2024-05-14 18:03:51.000000000","tz":-420},"subject":"bump lxml to 4.2.3","message":"bump lxml to 4.2.3\n\nChange-Id: Ia83ab181277e04e473e33680b3f2d686f534a084\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/54e13ee9b5e46209a62e3a13bc03b6edbd225105"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/54e13ee9b5e46209a62e3a13bc03b6edbd225105"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
