)]}'
{"id":"openstack%2Fswift~919636","triplet_id":"openstack%2Fswift~master~I551266dafb33688507196c75d98eced09e084d8b","project":"openstack/swift","branch":"master","topic":"parallel-task-iter","attention_set":{},"removed_from_attention_set":{"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2024-05-17 15:13:31.000000000","reason":"Change was abandoned"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-05-17 15:13:31.000000000","reason":"Change was abandoned"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-05-17 15:13:25.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}},"hashtags":[],"change_id":"I551266dafb33688507196c75d98eced09e084d8b","subject":"sq: drop MAX_TASK_CONTAINER_TO_CACHE","status":"ABANDONED","created":"2024-05-15 02:23:22.000000000","updated":"2024-05-17 15:13:31.000000000","total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e5d569080b87623575ac0f7ac55a10357eca9712","_number":919636,"virtual_id_number":919636,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-05-15 03:44:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"all":[{"value":-1,"date":"2024-05-17 05:33:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":1,"date":"2024-05-17 11:57:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-15 03:44:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-17 05:33:14.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"REVIEWER"},{"updated":"2024-05-17 11:57:00.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"f0b687f0c22be4406e8e2b4a48feac7432395351","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-05-15 02:23:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e733c56ee0a39e574a29b13cc6db065fa0dfcfd7","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-05-15 02:30:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c65e1bb0b4a544b79f93a9f96bf43393a25c3cbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 03:44:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e136c47cea804f3fba0d5d1c7950c30d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/515edfe59bda4c9cbfecd99868d7dcc0 : SUCCESS in 8m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/648fc439ae8441eeaaab86594538153d : SUCCESS in 53m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d664065db4fc4fa0a91780b4ac3051ca : SUCCESS in 46m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/960950dc9ba347ada9a763614a42ff44 : SUCCESS in 1h 00m 50s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7b67030acb15433397983bd1966d121b : SUCCESS in 18m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/675d99f320f1475cb97972749fdc99c3 : SUCCESS in 30m 22s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e0f7e31d80b243ad9a4025e7db493759 : SUCCESS in 20m 18s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5153c65726c74b06b63fdd8ec214c968 : SUCCESS in 10m 54s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/814768027c3e4e64af1c5093d791d08e : SUCCESS in 6m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4881c2824360410588cf6caeefc5d93c : SUCCESS in 20m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/94d6a18641d448b39a40630c3e8367cf : SUCCESS in 17m 49s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/eb6151c5b9504da09975a957c565a50f : SUCCESS in 19m 15s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/102c64d33be745ea91bceed8c14e5f64 : SUCCESS in 16m 05s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/334ffd2a089e4d7ea730026316bd2a1b : SUCCESS in 23m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4002781d1e9c4d408f8fa8cc1dfe7aa5 : SUCCESS in 23m 30s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f374d314d2b9496eb9dc417d4ce9b5f0 : SUCCESS in 17m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/314286e217d14ecf98bd2d5d18f7cb85 : SUCCESS in 23m 24s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/eeaf10f33d1f41a2863fa2ce204b541e : SUCCESS in 17m 35s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/30b8ba248fa44b7ab9d1f0af2c16b5f9 : SUCCESS in 18m 53s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7ca539b0e0f84a8cb2f233d41a8cc720 : SUCCESS in 17m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9f306db5423f4a0b9956f1caf43af62e : SUCCESS in 6m 42s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f9c894a6078a439797dbdfc082728136 : SUCCESS in 21m 54s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/cf4c84dcb2ff4762a78f59062ec7ae0f : SUCCESS in 10m 53s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/35149a97a68b446dbe8212a311e5862e : SUCCESS in 56m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c372f51c809e45ce971cdb82382ea057 : SUCCESS in 46m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cf1ee327323a4c4992b1ec31e477fe86 : SUCCESS in 51m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4be56fe2a5034f609e1bd3bb85e58213 : SUCCESS in 21m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57c71367e116412d984c674e7963247b : SUCCESS in 4m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/141f3bb26deb4f5aa3dcd8d1a0000c02 : SUCCESS in 21m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8173ec5a1a274db5bdd6ee1bcc90deed : SUCCESS in 1h 09m 21s","accounts_in_message":[],"_revision_number":1},{"id":"dc34e67c9bdf0ddb169eafb44abc90f1503edc39","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 03:45:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2de34819dd6d45f6805c669ed6ac40b1\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/51b0219e7d4b495cb9e0764de3d46850 : SUCCESS in 21m 16s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/91aad48a86d3460ca3d0907c9ce5c271 : SUCCESS in 1h 00m 08s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b6a6b1cdfce241938758b4d16dee49e5 : SUCCESS in 24m 04s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e6a14b4edd854648a00e7f1d681c1394 : SUCCESS in 1h 18m 56s","accounts_in_message":[],"_revision_number":1},{"id":"c6442d4a341edc60ae21d5df09c9561cda5138e8","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-05-17 05:33:14.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ef1c70b24b11bffb431e4c9de920b262854b5b4f","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-05-17 11:57:00.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6f4dbcf0d24dfe181f0306de9a5a0a6d3eaba954","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-05-17 15:13:25.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e5d569080b87623575ac0f7ac55a10357eca9712","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-05-17 15:13:31.000000000","message":"Abandoned\n\nsquashed 914713: DNM expirer: new options to control task iteration | https://review.opendev.org/c/openstack/swift/+/914713","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"935f431bb8e49cf7f42a52c5eb035d43fbf9e50a","revisions":{"935f431bb8e49cf7f42a52c5eb035d43fbf9e50a":{"kind":"REWORK","_number":1,"created":"2024-05-15 02:23:22.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/36/919636/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/36/919636/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/919636/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/919636/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/919636/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/36/919636/1"}}},"commit":{"parents":[{"commit":"014058097b89fe6bb8e89b63e10ccc4517b4e9a7","subject":"sq: strategy is randomized","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/014058097b89fe6bb8e89b63e10ccc4517b4e9a7"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-05-15 01:18:36.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-05-15 01:22:35.000000000","tz":-300},"subject":"sq: drop MAX_TASK_CONTAINER_TO_CACHE","message":"sq: drop MAX_TASK_CONTAINER_TO_CACHE\n\nChange-Id: I551266dafb33688507196c75d98eced09e084d8b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/935f431bb8e49cf7f42a52c5eb035d43fbf9e50a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/935f431bb8e49cf7f42a52c5eb035d43fbf9e50a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
