)]}'
{"id":"openstack%2Fswift~919637","triplet_id":"openstack%2Fswift~master~I318bc3ed6d665f2117fd9573d6a487b361623a72","project":"openstack/swift","branch":"master","topic":"parallel-task-iter","attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-05-17 15:54:08.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-05-17 15:54:08.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}},"hashtags":[],"change_id":"I318bc3ed6d665f2117fd9573d6a487b361623a72","subject":"prefactor: rework process config","status":"NEW","created":"2024-05-15 02:23:22.000000000","updated":"2024-05-17 15:54:08.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":20,"unresolved_comment_count":7,"has_review_started":true,"meta_rev_id":"b153c0fe30c5bea8e7550b81ad6cbe2e6e5b2d1d","_number":919637,"virtual_id_number":919637,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-05-15 04:01:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2024-05-17 14:09:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":-1,"date":"2024-05-17 15:54:08.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-15 02:50:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-15 04:01:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-17 14:09:06.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"d3472844cfffac59e0088bb95a53e0bd11fb1757","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-05-15 02:23:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e9d16feebc3b6a272c3c78701c3d412853650535","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-05-15 02:34:07.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"62b73be019c07661ce54298aee4176369a853c79","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 02:50:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31505a7dda6f465eb83885ee97cd9898\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/33539c6929ee43b682b8bd8d54f87a6a : SUCCESS in 20m 35s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/eab825d6d1d3466ea886694bf93c0ee0 : SUCCESS in 21m 23s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/2bbd175d14f9405c9160500114e93baa : SUCCESS in 23m 48s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/44d17a5dee9b4b34b222bdd2a66aa822 : SUCCESS in 23m 54s","accounts_in_message":[],"_revision_number":1},{"id":"58c17d22da40b5874e6330450f2bdc9328b2df12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 04:01:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74cc5881d3194978a467fec70d5b615f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/161deebacaf54cebb7a4152a8cadef96 : SUCCESS in 8m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/678a4e5dfbd54352a892e72e83034aca : SUCCESS in 1h 05m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d18695af8c674c278175f3510688e81e : SUCCESS in 55m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/80736e20896a40dca0964174b603df22 : SUCCESS in 54m 43s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0b66ffc64df940d798a3523a3f04685f : SUCCESS in 16m 39s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b30a95bedf1d471b8a63466965d135ec : SUCCESS in 21m 02s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1158b3775ada41f28bf32660364b31c8 : SUCCESS in 22m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ae8af576d28b4fc988658bfc92343a54 : SUCCESS in 11m 01s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/780140a595914345b7d293f9ef1ec033 : SUCCESS in 7m 20s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cc66b01804794dee947e0d775642c4e8 : SUCCESS in 26m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0e05ea940da745af950fa2d18cf6a88d : SUCCESS in 23m 37s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/50c73b2df9be412fb3fa40573c5d14ac : SUCCESS in 17m 51s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/62e941674fca49879f2741b73b1055f4 : SUCCESS in 17m 59s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e5f5ab5c24a34f9c8b500e8cc757400d : SUCCESS in 31m 02s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f3eafdefae064c5399c8c16926fb23b6 : SUCCESS in 21m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/954ffd0d9cc34725aa1b5fbdf6a77236 : SUCCESS in 22m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/478d858cd36e4d8084292b06d7021720 : SUCCESS in 19m 04s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a18cf5ac3cba49c9a61d2fdecdea7099 : SUCCESS in 14m 39s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/dc9beb657c1a4f50a8ae3ca418b69f7f : SUCCESS in 15m 11s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e9254679940d4de586f629636e0c42bb : SUCCESS in 17m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0c21a11decd04c21b6a74c2a716eb0cc : SUCCESS in 8m 06s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/204822e4b6f5406e8235efa048ea258b : SUCCESS in 16m 15s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/be1a4c5ebc4647c181be373043aae63e : SUCCESS in 7m 28s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f1365b0104d94e36b72fcafff2d8fa20 : SUCCESS in 1h 02m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/03a9f630f1f9497cae29459d37f4dbf4 : SUCCESS in 42m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0e29f4c6565946d8827f65091bfbad61 : SUCCESS in 48m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4b1809e4b1d64f0abf94ebcac91a9daf : SUCCESS in 18m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/223cbbdf8dd5456090ebec66dc05ff02 : SUCCESS in 4m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a4822783cb4b44e58d0f041603fa4d98 : SUCCESS in 20m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b1f08be6e7894bdfb2b7cb7ade13a41c : SUCCESS in 1h 01m 52s","accounts_in_message":[],"_revision_number":1},{"id":"ff14ef3ff799a1cb07bf5b6574c2a16e06e8d6a2","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-05-17 14:09:06.000000000","message":"Patch Set 1: Code-Review+1\n\n(8 comments)","accounts_in_message":[],"_revision_number":1},{"id":"b153c0fe30c5bea8e7550b81ad6cbe2e6e5b2d1d","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-05-17 15:54:08.000000000","message":"Patch Set 1: Workflow-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c3ee927dc5c173a0e248056316c54584d524914d","revisions":{"c3ee927dc5c173a0e248056316c54584d524914d":{"kind":"REWORK","_number":1,"created":"2024-05-15 02:23:22.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/37/919637/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/37/919637/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/919637/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/919637/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/37/919637/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/37/919637/1"}}},"commit":{"parents":[{"commit":"935f431bb8e49cf7f42a52c5eb035d43fbf9e50a","subject":"sq: drop MAX_TASK_CONTAINER_TO_CACHE","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/935f431bb8e49cf7f42a52c5eb035d43fbf9e50a"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-05-14 23:37:54.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-05-15 01:22:42.000000000","tz":-300},"subject":"prefactor: rework process config","message":"prefactor: rework process config\n\nChange-Id: I318bc3ed6d665f2117fd9573d6a487b361623a72\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c3ee927dc5c173a0e248056316c54584d524914d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c3ee927dc5c173a0e248056316c54584d524914d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
