)]}'
{"id":"openstack%2Fswift~919638","triplet_id":"openstack%2Fswift~master~Ie5d8b555489d28c3b901e5bbebdcecbde7bb3367","project":"openstack/swift","branch":"master","topic":"task-container-per-day","attention_set":{},"removed_from_attention_set":{"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2024-05-31 22:10:40.000000000","reason":"Change was submitted"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-05-31 22:10:40.000000000","reason":"Change was submitted"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-05-31 22:10:40.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-05-31 20:17:23.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Ie5d8b555489d28c3b901e5bbebdcecbde7bb3367","subject":"tests: realistic task_container names","status":"MERGED","created":"2024-05-15 02:23:22.000000000","updated":"2024-05-31 22:11:42.000000000","submitted":"2024-05-31 22:10:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":53,"unresolved_comment_count":9,"has_review_started":true,"submission_id":"919638-task-container-per-day","meta_rev_id":"4d8fc6643014bf7dcf8a25b53ce57b642cb5f365","_number":919638,"virtual_id_number":919638,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-31 22:10:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-05-31 20:17:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-05-31 20:17:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}],"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-15 03:52:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-15 03:54:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-17 15:13:17.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2024-05-28 19:06:51.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2024-05-29 15:51:51.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"REVIEWER"}],"messages":[{"id":"c556942847997489fc2a6fcac08761b54f1a5369","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-05-15 02:23:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"55561fd5ad306ad210000aa378264455a3451995","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-05-15 02:41:56.000000000","message":"Patch Set 1:\n\n(6 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e733b90da0aa7fc534a5880452002ab7923b879d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 03:52:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bfe6a0fdb7764099997c14a8591f3f6f\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ed6ad078b69d46a5a8bea6a4e7ec35df : SUCCESS in 16m 54s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9a8a0658e6884f91ab6bd23086e14172 : SUCCESS in 1h 17m 56s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e8c54865316045a4b79bf06c451dc4e0 : SUCCESS in 1h 21m 31s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/983400fecbd245ce8e9a300452d9d4b5 : SUCCESS in 1h 25m 12s","accounts_in_message":[],"_revision_number":1},{"id":"b3932b7536b9bd7540f595881a02d55a712a0ae5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 03:54:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33fe27a91179497bb93323ce9b4ee1b8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f7a00797b064b999b1141df13ea5257 : SUCCESS in 9m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/3414c6ead13841ba822d27276077124d : SUCCESS in 1h 06m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4a84bc4f4e614891b482b9aa6878c45c : SUCCESS in 43m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e99a3486950f4695ad12c116aeeb8b05 : SUCCESS in 48m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/73559d2013ea440d87d75a0ebad57829 : SUCCESS in 20m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fee23b90319b407cb2fb9df21bfd965a : SUCCESS in 18m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a6fa068611694f739531afa7494eb104 : SUCCESS in 21m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6339b73770454c588d69eba331ededa4 : SUCCESS in 10m 13s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8831a6dd27004b698335eff31ad3e85a : SUCCESS in 8m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/15e41849269b499180c572b0f7601599 : SUCCESS in 19m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/23a0504f68d54396858aaaa5ba1b51aa : SUCCESS in 18m 38s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/2aa5e03c13424c7ca96fac48aee29914 : SUCCESS in 19m 46s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/b895f71d83b848df8952dabeab3f544f : SUCCESS in 13m 40s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4dfcda57da66478f82e46b391f55a857 : SUCCESS in 24m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d92f6efc103f4c12b78047e3affabe65 : SUCCESS in 20m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/56a6dc66922740f39e8eb3ad02270994 : SUCCESS in 18m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/350e9df7b5cd4153b6360da5a41fb6eb : SUCCESS in 19m 38s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/cedbf730cb8a4b3cbda8a956343853a5 : SUCCESS in 14m 49s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/783e51af81f946b880424b5b2932d8dd : SUCCESS in 16m 25s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e51fe67bb89e4d46961b6a59ac800181 : SUCCESS in 17m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cd94881544db475f8547d6122cf3c89a : SUCCESS in 7m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cee36a113a9e4dc584565caf3de448d5 : SUCCESS in 15m 23s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7c11bcc5dd5c467ba1ea8a526c88145b : SUCCESS in 7m 27s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f228cab229284858a3a71a9701c0c63f : SUCCESS in 51m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/656119f40f274d96b313e6188766b0c1 : SUCCESS in 56m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ba9f844a2ef74b4fa230ff8115214c14 : SUCCESS in 52m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2d49a467d95743b3a3d4f4d78c66ee24 : SUCCESS in 22m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0e7ef026ef04052884f15b91c0ed025 : SUCCESS in 3m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/98ff3692e5b54614929e7319ab293f4a : SUCCESS in 21m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3743f0da78384ae1a3edd4f030fd02a0 : SUCCESS in 1h 02m 07s","accounts_in_message":[],"_revision_number":1},{"id":"ba00a7b68e9792ab8ab3c60c6ef4dd105d828707","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-05-17 15:13:17.000000000","message":"Patch Set 1:\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"7710f76f289c1f7ed9803db3553027ab703abdd5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-05-25 13:36:02.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"4f9edf1aca45a41a11cc8c2e148f4dedfbd3e436","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-25 14:45:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1dd5ee3d5c114555bc2d4c11c2f1eba7\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f7991d64923f4f57b5aefa666fed3745 : SUCCESS in 20m 53s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2c6382b187c944d9b12d918f262a7971 : SUCCESS in 17m 04s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e2c70eebbc4d409582a783f1b7467350 : SUCCESS in 1h 06m 40s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/704ebcf1a45e4b56aac731317217d532 : SUCCESS in 18m 56s","accounts_in_message":[],"_revision_number":2},{"id":"f978c6b950cf42e9f4755a0de84a96f61374859a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-25 14:46:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d58f6cc24f444d98a639294b52158b63\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d588f642b8346658c4c1806b56d7a8a : SUCCESS in 9m 34s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/890eb15cf96740d585bcc8105c6f503c : SUCCESS in 18m 19s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7d4ee3258b0b407bafdf6687b99b986b : SUCCESS in 19m 06s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b9ca3f7780584b5e9c3259c1472c13cb : SUCCESS in 18m 10s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1beab2a74ef149de8ea1bdfefda0d2e4 : SUCCESS in 11m 59s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6ba4b07a9bca4365b597194f7b708f00 : SUCCESS in 5m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/360e037a70f242dda7dc5cdf6f59c1f4 : SUCCESS in 20m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9338d38f585d45ca9785297d9efe604e : SUCCESS in 17m 54s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f36d5ef9232c41a68014526391a0f7c9 : SUCCESS in 18m 31s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/cce641c5adbd4082b503985d9774cb51 : SUCCESS in 16m 30s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c917be2e9d324a39ab7bf30ab2ed6638 : SUCCESS in 22m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c6dfd678310743fd9fa04c39dca190a3 : SUCCESS in 18m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/84b662a90e08437597db4fb91ff5bbd6 : SUCCESS in 18m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ec567f1a8020481e8a215fbb3486524c : SUCCESS in 20m 26s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/15cb0381d0f94e018f5167bd567aa3fd : SUCCESS in 15m 53s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/04022c68dda6495bab65789c4eefa2bc : SUCCESS in 14m 46s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a4205762875b4aae89cedd229681a461 : SUCCESS in 16m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0c237ab8d8af47baa7147f2682ac5683 : SUCCESS in 55m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/641b628bced549aabf4fd928173c60e2 : SUCCESS in 53m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8a652993d03344baa34e3c43970e90fb : SUCCESS in 19m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed6753cf5ffc4102bef1709f79fa1f62 : SUCCESS in 4m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/473d2fb0bce04a3e953554fe6e4f248f : SUCCESS in 19m 52s","accounts_in_message":[],"_revision_number":2},{"id":"d28a5412c55a56bb607ebe5954d9fcdeb89b1849","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-05-28 17:00:37.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"c71de421c67a792eab25c4b2db0389f36313bb32","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 17:52:26.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/460291413b84466cb5225d43a52fc081\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6e54f3613e1c48b4b0f8a95273dc4d15 : SUCCESS in 21m 24s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/40b839b85825421a9f358601d3f0ac25 : SUCCESS in 45m 10s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ea3f41271af54185b1892112c37c0291 : SUCCESS in 26m 39s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f8056061e34b459d8b77f50e88ac397c : SUCCESS in 49m 28s","accounts_in_message":[],"_revision_number":3},{"id":"8849b67b47c3588884ea68bc21d3268e591d75a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 18:12:17.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7d19056e5ada47e7901b802091be06f3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f1d69fae5064f4a9c312ddd3d905796 : SUCCESS in 6m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ccc79a8ec99c4835ac0e7f712694a59a : SUCCESS in 20m 57s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ada84aabe800435f9b8d5e1e2a0ea5ee : SUCCESS in 19m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9a629dfc66a047fba97b80ef5d8b9bd4 : SUCCESS in 21m 50s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ae180046212b4d97a185d196c867d1bb : SUCCESS in 11m 33s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f76d76af810044f9ae35c027d06908a6 : SUCCESS in 6m 50s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b84832d9a23b4c488d291a6c8ea75804 : SUCCESS in 19m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/651fbeef322b4c92b1d61d47aa3a0aa7 : SUCCESS in 20m 51s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/65a8454398b24fbdb751572719b7c81c : SUCCESS in 17m 55s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/331914e8356e43d88801335128b2eddf : SUCCESS in 19m 41s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f15b6f39342c42119957798d7111517b : SUCCESS in 23m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2dc9f5a0ccbd4f30aa01a1ece9c3e406 : SUCCESS in 17m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/674da3802bbc4b93a04628035ce3ad42 : SUCCESS in 18m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f9dcca4330c74b16ac3ccb89a03174a9 : SUCCESS in 19m 27s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e83592a3710f4cfaa00cc7e8a1d6a59b : SUCCESS in 16m 43s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c59098e1ee974628ad736c4a893562fa : SUCCESS in 16m 33s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2f131c1690284ca68451d8acd12afe8a : SUCCESS in 18m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7f63e89c0f174ff6a6b99454d21bbeb6 : SUCCESS in 45m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/97df4eb1bc2240d1bc805a336303141c : SUCCESS in 39m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/03013468c1b04d2f9e55aa955702210f : SUCCESS in 22m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd3d53a2d9c14692a1c96dbbeab0ad91 : SUCCESS in 3m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f3d1b7944d69434facbafcab84c5f0a1 : RETRY_LIMIT in 2m 07s","accounts_in_message":[],"_revision_number":3},{"id":"c2337c37714f31b37b171490a8c2e39b2ce11f4b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-28 19:06:51.000000000","message":"Patch Set 3: Code-Review+1\n\n(12 comments)","accounts_in_message":[],"_revision_number":3},{"id":"0cb6a1b3765b1ab417522c35216d221a9c2ec3d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-05-29 13:42:28.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"ccdf31ae7c1fc4bb289bfc3b6b89e39456b1354d","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-05-29 13:42:45.000000000","message":"Patch Set 4:\n\n(19 comments)","accounts_in_message":[],"_revision_number":4},{"id":"9cba79888aca28aa562d891b22e059c883b33f99","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-29 13:52:19.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2de37975398a41c69c1ad61fda4d472a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/787880ac4f544bd3882a86e5f52ec97f : RETRY_LIMIT in 1m 27s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9e629f6e07c14f0fa4ebfc296c56f00e : RETRY_LIMIT in 1m 28s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/2f68c917c54e4e0fa99ff71094681524 : RETRY_LIMIT in 1m 06s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/772efb6026324e8c9ba5d04278dbb916 : RETRY_LIMIT in 1m 45s","accounts_in_message":[],"_revision_number":4},{"id":"d66b0a33a278bdbb8260030d40cbc2655ba68e57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-29 14:49:25.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1412ad01172f4dcca8f70e0b7d0d54c1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bcbac5e7a5a4586b180b685b5ba1939 : SUCCESS in 9m 29s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/64eee357cc664d09a4bff9be0fde2db7 : SUCCESS in 20m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1dc2d3a1206f40359f2934f9d4061cdf : SUCCESS in 16m 24s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a076232e45854a2a8ae5f4e0d22f477d : SUCCESS in 20m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/51e9e99d7f924be7b8eaf337b2c070b0 : SUCCESS in 11m 38s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9fd556d866384a4fb2831b43082d5c53 : SUCCESS in 5m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/77edd778f26c44fcafd61c6211e0a655 : SUCCESS in 20m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c8e605b53f6f466ba96369e159618436 : SUCCESS in 18m 53s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/61dc27d8eb104579b1b7bb292d1f3eed : SUCCESS in 18m 00s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/2f58c1b9e9e943b5b1b6b7fc6b30259f : SUCCESS in 18m 44s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/91a3cf6a0f864a4fb79d06fd8b9c7b85 : SUCCESS in 17m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9813423ed8cb482a884cbb0a9e9aa78e : SUCCESS in 18m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5d20983aa898463c9373c9d9712b0ab4 : SUCCESS in 17m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9c4f18c0b99644cf8a3d5c31ff462460 : SUCCESS in 19m 21s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/24decc16a54f45038d9237819dcbebd1 : SUCCESS in 15m 35s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0af66e1444e447348a231bde83c6f5cf : SUCCESS in 16m 10s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c4a74a532f1b479a81150dac6da3b795 : SUCCESS in 14m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fd248e5bb8414a1797d5fadf5344231a : SUCCESS in 50m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f9cfca5f4a934de4bf006eed6b65cf28 : SUCCESS in 52m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a40cd358ead84079a1bc2381b9f8c4b1 : SUCCESS in 17m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc3340dec5024b2489b0fc91df897220 : SUCCESS in 4m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/623772b7576d4717a32f5455890a95fb : SUCCESS in 20m 37s","accounts_in_message":[],"_revision_number":4},{"id":"cd589523a784a6844589a30cc542671a701489be","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-05-29 15:51:51.000000000","message":"Patch Set 4: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"b58dfc018ec6cc2c86a489d06d415231bcf82e80","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-29 16:40:18.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(5 comments)","accounts_in_message":[],"_revision_number":4},{"id":"83d81e49d6ac8778ec686cef6e7a0f5b3afdcc99","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-05-30 14:57:21.000000000","message":"Patch Set 4:\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"9a6e8ae973b25b03a98dbca5902195aecebdc0be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-05-31 20:10:27.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":5},{"id":"8f6c0744b72a3d78ce6fa35ca3ff52a00546ad39","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-31 20:17:23.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"af402a4d548479f8afb58d3434cb5471b6355f73","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-31 20:36:56.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/febf2413146f440e9d6cf63d6af1ead8\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3591ff9c9a7a4391b11a67357f707898 : SUCCESS in 21m 13s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3e4bc2329c1c4d07b181b6a4d0e96dd9 : SUCCESS in 18m 16s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ac58a6a626604d038912d03a50508f40 : SUCCESS in 25m 14s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d2487ed228db421c9c83a245af248e65 : SUCCESS in 20m 29s","accounts_in_message":[],"_revision_number":5},{"id":"a05017ef06c4c783012a4d7c8d9c135dce827f56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-31 21:11:28.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08731c5040db4c20bd016ebc32ca41e6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9af3883492ed4c00adfcddbc65f00c76 : SUCCESS in 9m 43s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/018750db7771455896aabdf9410c4ec6 : SUCCESS in 20m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bb1df05cfa4b484e9abe6a4625903975 : SUCCESS in 21m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/20ffd98f3f0247f9ae714aeba8ed008c : SUCCESS in 21m 52s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b213ea1bd9314fbcaa11cb03ef064d82 : SUCCESS in 11m 30s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f3a9405dc45e4d869e0310a97af1ded8 : SUCCESS in 5m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7cbd299742b14fbda3e1886d38a2c68c : SUCCESS in 20m 28s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6abb57dae4fb40848a2d2c97f376cda7 : SUCCESS in 16m 13s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/871f545761df492a859f5ebc7925a755 : SUCCESS in 15m 41s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/11a2564a5f7c470aae847830d937ef7b : SUCCESS in 17m 01s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a35e50b96d6e4c28a25ca504ecdba391 : SUCCESS in 16m 59s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9c66dc0603f6483b9aa08fda5e72c161 : SUCCESS in 23m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7399c446af0346df9d451ae1baccd6b9 : SUCCESS in 17m 55s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3dbd49628808417983be13273f8456e8 : SUCCESS in 18m 55s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ec2f4999df914b46824f421a07295380 : SUCCESS in 17m 46s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f45ef52b18784d1b8a7a5e328dcf25dd : SUCCESS in 14m 21s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f09e7c2356cc437daf0733f132073ca6 : SUCCESS in 16m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5b4a492158604a6d832eeeee1b53eb16 : SUCCESS in 55m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/51ad7e1c5df74715996ea5441a0992e9 : SUCCESS in 44m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/18964d9a12f64025a35a0e105a0959e5 : SUCCESS in 18m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aedff8d104d246ef982eff005da8c037 : SUCCESS in 3m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/91b8c572c1844910a1d7d8d4cea88974 : SUCCESS in 25m 38s","accounts_in_message":[],"_revision_number":5},{"id":"a1fde1018d1fed9687f399078b8b64520e93d6dc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-31 21:11:49.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"f34a9629800574c657536bec90626b8b0b623077","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-31 22:10:40.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46c76231142a42d88d52aae6c9827137\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/216d2d431f3340388b35d148f300bcaf : SUCCESS in 7m 02s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/065481e59cf043a798a79c688a4b9fe7 : SUCCESS in 8m 31s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/e8b11099a2744ac0aabaf9b225526ef9 : SUCCESS in 5m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/70fd7fb2b8134eaaa98efac57cb3643b : SUCCESS in 21m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ee6dd6b0e99548f49313cba4f2f8cc03 : SUCCESS in 18m 50s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/fb086fbcf8b04199814c16cea137aca4 : SUCCESS in 13m 42s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/b634e88de2644e7dbb32855bf78a1cc6 : SUCCESS in 13m 18s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b64f624e60f84f6596c88dc539218d38 : SUCCESS in 16m 27s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/688a7a2e09134821bdc07733515c531d : SUCCESS in 17m 07s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/29a54bf9b34a4a30a5a4a887b5214293 : SUCCESS in 15m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1f722826b6ad401697d083a71a30eb0d : SUCCESS in 19m 24s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3754c0ae5a214207b66155401b35434a : SUCCESS in 13m 36s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/be8bc569e00d40108987f4f9c5218d0b : SUCCESS in 16m 52s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f147c759056a47389af8286225da4e7e : SUCCESS in 16m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ca153a0d50ef4942bde05771a3115a80 : SUCCESS in 7m 29s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b2ddf5878138421796a246e8a512bcc8 : SUCCESS in 7m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a7b9da23f10b40c9aa36bde62f5f3989 : SUCCESS in 53m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b69f3e7e46204c9890ee63bcbef1f915 : SUCCESS in 34m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2ae57f42eba04e8faba267560586fe83 : SUCCESS in 15m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26eed80a3ca546448a252681652d0bf2 : SUCCESS in 3m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d76ebcb72dcb43eba789c0f13b9b2c3d : SUCCESS in 20m 57s","accounts_in_message":[],"_revision_number":5},{"id":"e5d5b9cdd472251186ce6570ab1db80a433627db","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-31 22:10:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"4d8fc6643014bf7dcf8a25b53ce57b642cb5f365","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-31 22:11:42.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d02dfb0550f472e9a433a18d8cb51ad\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05b97dad7c1f4d368f89a682af49a867 : SUCCESS in 43s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/d4d3dca747384aa7ad24afcea9b21894 : SUCCESS in 37s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/75b1ad313bee46518cdf78e8098d9986 : SUCCESS in 34s (non-voting)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"99c629edb813c7d9fcd736f5d34b3a47d224a457","revisions":{"49ba4be1eb04a305140a5fd918389a2a211ccfc6":{"kind":"REWORK","_number":1,"created":"2024-05-15 02:23:22.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/38/919638/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/38/919638/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/919638/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/919638/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/919638/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/38/919638/1"}}},"commit":{"parents":[{"commit":"c3ee927dc5c173a0e248056316c54584d524914d","subject":"prefactor: rework process config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c3ee927dc5c173a0e248056316c54584d524914d"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-05-14 23:43:34.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-05-15 02:20:08.000000000","tz":-300},"subject":"tests: realistic task_container names","message":"tests: realistic task_container names\n\nChange-Id: Ie5d8b555489d28c3b901e5bbebdcecbde7bb3367\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/49ba4be1eb04a305140a5fd918389a2a211ccfc6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/49ba4be1eb04a305140a5fd918389a2a211ccfc6"}]},"branch":"refs/heads/master"},"6915f80641d45c9b4ac234885c32fc0d0b2794dc":{"kind":"REWORK","_number":2,"created":"2024-05-25 13:36:02.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/38/919638/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/38/919638/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/919638/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/919638/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/919638/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/38/919638/2"}}},"commit":{"parents":[{"commit":"616ea0129250a4adc062bb953096221259687cee","subject":"DNM expirer: new options to control task iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/616ea0129250a4adc062bb953096221259687cee"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-05-14 23:43:34.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-05-25 13:31:06.000000000","tz":-300},"subject":"tests: realistic task_container names","message":"tests: realistic task_container names\n\nChange-Id: Ie5d8b555489d28c3b901e5bbebdcecbde7bb3367\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6915f80641d45c9b4ac234885c32fc0d0b2794dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6915f80641d45c9b4ac234885c32fc0d0b2794dc"}]},"branch":"refs/heads/master"},"663d866e3453e7dee54cbb205afd4b74a6f79840":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-05-28 17:00:37.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/38/919638/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/38/919638/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/919638/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/919638/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/919638/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/38/919638/3"}}},"commit":{"parents":[{"commit":"521ab60b8e0c22250af8b8a67f59174d4f233850","subject":"DNM expirer: new options to control task iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/521ab60b8e0c22250af8b8a67f59174d4f233850"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-05-14 23:43:34.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-05-28 17:00:22.000000000","tz":-300},"subject":"tests: realistic task_container names","message":"tests: realistic task_container names\n\nChange-Id: Ie5d8b555489d28c3b901e5bbebdcecbde7bb3367\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/663d866e3453e7dee54cbb205afd4b74a6f79840"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/663d866e3453e7dee54cbb205afd4b74a6f79840"}]},"branch":"refs/heads/master"},"c9108823e03f4b9e86f1fea5e6775b1808576609":{"kind":"REWORK","_number":4,"created":"2024-05-29 13:42:28.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/38/919638/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/38/919638/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/919638/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/919638/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/919638/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/38/919638/4"}}},"commit":{"parents":[{"commit":"521ab60b8e0c22250af8b8a67f59174d4f233850","subject":"DNM expirer: new options to control task iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/521ab60b8e0c22250af8b8a67f59174d4f233850"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-05-14 23:43:34.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-05-29 13:35:30.000000000","tz":-300},"subject":"tests: realistic task_container names","message":"tests: realistic task_container names\n\nChange-Id: Ie5d8b555489d28c3b901e5bbebdcecbde7bb3367\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c9108823e03f4b9e86f1fea5e6775b1808576609"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c9108823e03f4b9e86f1fea5e6775b1808576609"}]},"branch":"refs/heads/master"},"99c629edb813c7d9fcd736f5d34b3a47d224a457":{"kind":"REWORK","_number":5,"created":"2024-05-31 20:10:27.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/38/919638/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/38/919638/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/919638/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/919638/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/919638/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/38/919638/5"}}},"commit":{"parents":[{"commit":"82debd2cda0a3f2727150be37e001daee857d292","subject":"Merge \"docs: Add tempurl/formpost metrics\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/82debd2cda0a3f2727150be37e001daee857d292"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-05-14 23:43:34.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-05-31 19:09:32.000000000","tz":-300},"subject":"tests: realistic task_container names","message":"tests: realistic task_container names\n\nChange-Id: Ie5d8b555489d28c3b901e5bbebdcecbde7bb3367\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/99c629edb813c7d9fcd736f5d34b3a47d224a457"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/99c629edb813c7d9fcd736f5d34b3a47d224a457"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
