)]}'
{"id":"openstack%2Fswift~919667","triplet_id":"openstack%2Fswift~master~I6fcec6748c3fd0f22baac69f7ceb8261c5bb5d63","project":"openstack/swift","branch":"master","topic":"fix-rebase","attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2024-07-22 22:41:50.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2024-06-21 22:58:13.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-04-10 23:24:56.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"6968":{"account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"last_update":"2026-06-24 16:22:44.000000000","reason":"\u003cGERRIT_ACCOUNT_6968\u003e replied on the change","reason_account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2025-04-10 23:24:56.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-04-10 23:24:56.000000000","reason":"Vote got outdated and was removed: Code-Review+1"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I6fcec6748c3fd0f22baac69f7ceb8261c5bb5d63","subject":"obj: skip to async_pending when root is busy","status":"NEW","created":"2024-05-15 07:27:32.000000000","updated":"2026-08-07 17:46:40.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":73,"unresolved_comment_count":22,"has_review_started":true,"meta_rev_id":"772c176bcea799d477c1d02a612deda8e464b491","_number":919667,"virtual_id_number":919667,"owner":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-02-24 09:47:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-15 08:37:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-15 16:32:23.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"},{"updated":"2024-05-15 19:02:40.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2024-06-03 16:41:53.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"CC"},{"updated":"2024-07-15 20:32:00.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2024-08-05 12:03:06.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2026-06-24 16:22:44.000000000","updated_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"reviewer":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"state":"REVIEWER"},{"updated":"2026-08-07 17:46:40.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"CC"}],"messages":[{"id":"05fddcf5388945a70eaf47ca86643483f4280fa1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-05-15 07:27:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"34ef353e99b45813b2e00f81bb14d54357007387","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-05-15 07:31:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"894b18e4cca3782d319c19d75dca582ca4789606","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-05-15 07:33:05.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"b0e874aadaa32fd5969899f61061b80a04a8942b","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-05-15 07:34:24.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"a567756a2aab557ef215db038e329dd6e2ef1e76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 08:37:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fda92d2b469f4b6d8243112fc2607da8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7638fcb33b1f414b9dc0b119b50d2dff : SUCCESS in 6m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/f2e6c65c5a77447a8215012fe5265087 : SUCCESS in 1h 00m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2216c23ee8a643568ee8a7e95bfd6d1c : SUCCESS in 39m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5df1b9db3d03447c8b52629dac976825 : SUCCESS in 54m 00s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fc20389fc76b46598f1806b1e60e01ea : SUCCESS in 18m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6a8db5173b7a408cbdca47cd33ce2fad : SUCCESS in 17m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/896e05e55a904aeeb32d205d51b4039b : SUCCESS in 20m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1632f7d167754fe7a7197587443a0be8 : SUCCESS in 10m 36s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/33e2010b79cd4193b54650055077352d : SUCCESS in 6m 15s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/63d55a39ff7245739b67d23a6ec8b901 : SUCCESS in 19m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5818b17f37df43a491d41397875fe243 : SUCCESS in 18m 04s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ecbf91b89eea4689b0c6af9e9ae24772 : SUCCESS in 18m 45s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/2067b5f6a3254cf4bad0d4e78ea5a45a : SUCCESS in 18m 01s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c7c7c4b2c04f4c26906bdbe60ac22dc3 : SUCCESS in 22m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1e4c238477ca48b08a5abfea5b7c7422 : SUCCESS in 17m 24s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5ac6d80587cb4fe0badcd57eed8d5852 : SUCCESS in 20m 08s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b9fda8c3519d47d39f2de56bb2966d5e : SUCCESS in 20m 42s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0a3855c4a2e1428b87e9fda85aa3ca81 : SUCCESS in 16m 51s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/076998e39ab047698ae7fc982cf44c50 : SUCCESS in 20m 54s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c5e1cf0760074e36b0c838c41469f8b6 : SUCCESS in 18m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/becf7cadb600421a86d450ec07156927 : SUCCESS in 6m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3df872fbc04e4db0bd1786cd385e6112 : SUCCESS in 16m 12s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e03fada20e8a457fb95125af0dfa3ce7 : SUCCESS in 8m 16s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0eca38a2f2f44e3abb0c9a355933bfa9 : SUCCESS in 55m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bcf99b28d8ee45d3bf486a78c10c2b28 : SUCCESS in 52m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/67e62c5e058b4af085cc567745d1d83c : SUCCESS in 42m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/292c60a6696f4236b13af22426f0e778 : SUCCESS in 19m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f442646f87d4c738d88184687836604 : SUCCESS in 4m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/08cd1fac0755400f8eb2655064dedf4d : SUCCESS in 19m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/099f16ee052d4dbf9a45d01272bb579a : SUCCESS in 47m 12s","accounts_in_message":[],"_revision_number":2},{"id":"23c611cfb6474bcedb94e97182d90249464a012f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 08:48:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b735468bf374bb19ea47198eecf04a2\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/75b7ef83187b4849b610c53983bff9c2 : SUCCESS in 1h 00m 32s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dd11c5bb8388436eb8b68669eac8e756 : SUCCESS in 15m 00s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ef81d9bf83ae4d77a1813843f059b040 : SUCCESS in 1h 14m 43s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/eb2ff9a12d50400883f4efb784299754 : SUCCESS in 17m 37s","accounts_in_message":[],"_revision_number":2},{"id":"a82f09902606c8760d571e865a5daee8e31e8c72","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-05-15 16:32:23.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"79bc01e0dd8700d2b75e8a4a7c04c10628db7fde","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-15 19:02:40.000000000","message":"Patch Set 2: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"829cf4e025d1372cb192ec9cf7c7ebebc9ae644b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-05-16 06:23:04.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"8020e829d63db9399bbfd6a3b1f92b23e0c74459","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 07:28:38.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9baa315b383e458688c400146147f2c1\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1fa1c43ba74a4a0b9a79955ef27a58ea : SUCCESS in 15m 14s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/654bc6c5628a4cd5ae8b75d5ad15f06b : SUCCESS in 52m 28s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/deb595d68246495d89095301f817ddf8 : SUCCESS in 19m 44s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/9cdcd6e0a6704d8d8e02ea1571670c9b : SUCCESS in 1h 03m 31s","accounts_in_message":[],"_revision_number":3},{"id":"0b0e7c08885c8bd63da1dd6c19323b812a1d1fa8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 07:48:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a99489e913b47ceb0128eb7f1fc93a8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c31fc291e61d4d88b3f0d995f0acbeb0 : SUCCESS in 8m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/7cf720a8c8f9461595fcd1bbffd65b76 : SUCCESS in 1h 06m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3f15832ca9634bddaf273417923aba2a : SUCCESS in 56m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/af316ce21e204b34b10955398a64e21b : SUCCESS in 55m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0f9b3de622904334a3134a3dfdbf226a : SUCCESS in 19m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/77949fe2d13f4a8fa20838f853e47079 : SUCCESS in 19m 41s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/77c1e8f3ee5d486cb1306f9fbff9eaf4 : SUCCESS in 18m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/97f50e8708eb4c9f81aedc711b8c67b0 : SUCCESS in 10m 39s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8aab44e5f74247f38df8b643ef4d5a86 : SUCCESS in 4m 42s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3ed0b67fbd154da38091320777bd6abb : SUCCESS in 20m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5e45eb216bdf420c8e347c6c8eabd8a1 : SUCCESS in 22m 54s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c0e838c8fbb54f48896bb80b252947bd : SUCCESS in 17m 13s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/cd2c45bc7a6e4635b64483790f513a9f : SUCCESS in 24m 46s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1649d8baaff84425812cdd7df414d0c7 : SUCCESS in 23m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f95948656a9d4adc8735a4ae70517945 : SUCCESS in 18m 58s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a4401de12b54433daea42770d44c26dd : SUCCESS in 19m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ebc1b4b576184d0b887ee5f928b0dd9f : SUCCESS in 19m 37s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/945f49ebd33440c9bada418ccef019a6 : SUCCESS in 16m 24s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/05f511cc2df846ba93cc60b661a7db00 : SUCCESS in 16m 30s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1e2c00eed5274fcbb584225b09edb59e : SUCCESS in 16m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/09b8dd222d9846d7ba5df7a2cf2aeb31 : SUCCESS in 5m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/09698c17139647f5a5d059abdcf5b570 : SUCCESS in 14m 48s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8aa1c4037fe34d2b94fcc2041a1765a2 : SUCCESS in 8m 37s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b35a005afde94c6cbbbfdad9f4a43263 : SUCCESS in 1h 05m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6f851fac60184125aaff7caf9481c499 : SUCCESS in 53m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/dcc47248395e4231bdb26bd0cda09539 : SUCCESS in 42m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ebe20f2b25514d47bb82642970e1cdc2 : SUCCESS in 21m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98f0c550b41940fb80e9a0021f8b1ea5 : SUCCESS in 3m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/60d62ec781f84573b62017f46a12aece : SUCCESS in 19m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b3c0311e70b5490399b37f14cc13d602 : SUCCESS in 1h 00m 28s","accounts_in_message":[],"_revision_number":3},{"id":"03aa87bef264fc6620a8ad42c17d3c809de1c5fd","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-05-17 04:24:33.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"359b0ec83698fd17fbf4cb000c87e0d34bb2782b","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-05-20 19:46:26.000000000","message":"Patch Set 3:\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"51662e6db5e06e4176b1192076e5ef93003b6e03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-05-22 07:05:57.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"af0d0bf6680d676d298b2a724c40043607c228bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-22 08:23:48.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9af04dc2bd7547fb92618f714957182b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/625d7edc29df47548b6feffe8f5eb0f8 : SUCCESS in 8m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/b50c8c88c2ff4cd081933d50dbd9e6f1 : SUCCESS in 1h 05m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/53c3bec2baed41e0af23abd9e848ab69 : SUCCESS in 39m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3fe7b9adf97e45b5bcbda37585d35dfd : SUCCESS in 1h 10m 28s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a4406e5a5c1d4fa2b1eb4ddb929748c7 : SUCCESS in 18m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/71a1b9206b9243db93114999cda8dfcc : SUCCESS in 23m 03s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2a7bdd0c5dcf43f5a982801f971fd490 : SUCCESS in 20m 03s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/401318cd740e4e6caad4e4de64aa67d4 : SUCCESS in 10m 44s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/557ee5dd7e324cba9872ef519079b4ef : SUCCESS in 5m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/08aca76f95a044aab79208952b86ae78 : SUCCESS in 21m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/864b16b8234c431280a65c497ada9fb2 : SUCCESS in 17m 21s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c8e926f5b2c34fdf8d283de94cbe6a68 : SUCCESS in 20m 03s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/90bb151601b14c859ff78eaab1eb0ef2 : SUCCESS in 17m 44s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b07b174f819346a281dccd62a3405e02 : SUCCESS in 17m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/06bc924ce87e431795f4f0d1a0b6c644 : SUCCESS in 16m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/64fd641c61ac4cf6bdfa0ae90d726406 : SUCCESS in 23m 40s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3e15e078e78c468081c2f3e8fa6e3da3 : SUCCESS in 21m 54s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f4e98a41c1d74ad6833f6575ced25759 : SUCCESS in 14m 20s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5c038f5ede3745e987544fa376ae4e77 : SUCCESS in 17m 15s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/90f349e3fc074a4ea0ea71f74a8c7d8a : SUCCESS in 21m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9f631d3213bf46d0b640d2b67870ab19 : SUCCESS in 7m 36s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4e9201d18d3648db8580ad3f28fe7abe : SUCCESS in 18m 30s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8412bccd67b246bc8f69e91ac46738de : SUCCESS in 7m 09s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/de4fcc64112646c9b0271df2fdf00658 : SUCCESS in 1h 04m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d7b585ad03d3422890ef13859cf741e1 : SUCCESS in 43m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9730a39654c24791a9034576e01f674b : SUCCESS in 55m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f33d39959a694759953a712c6f01b700 : SUCCESS in 19m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a4ed7bb73a44418be0317111ca6f177 : SUCCESS in 4m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/739c5563139149169f5819daa4eac3f3 : SUCCESS in 19m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9bb5c5769d1548f98fe7d24d4615c7c0 : SUCCESS in 54m 50s","accounts_in_message":[],"_revision_number":4},{"id":"c9f0600eecf7bfdfc06f59cceb6a1cca4148e35f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-22 08:33:51.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85dca7dcbb444a81a9a512812cea8c6f\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/bacd1a104f4a4bbe9c867d016d2475ec : SUCCESS in 1h 04m 54s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a0dcd1cae13c4d53b2dabde5a15095b8 : SUCCESS in 1h 26m 01s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/81dfa047211a43e6a5cec2fc62b1ba64 : SUCCESS in 28m 44s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/fb07c3f8e1ee411fa4186aba9acb5a5e : SUCCESS in 56m 05s","accounts_in_message":[],"_revision_number":4},{"id":"3819845bf50e889095a615b11f8c2992df1a596d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-05-24 06:39:04.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"0dc4ffef0f66f5ba438dfac79d8a6028e8fbb273","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-24 07:04:10.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2aaf89aeb55744159591f606c693c07f\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1775bc17c6b6411cb8508107cde48461 : SUCCESS in 16m 06s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9ea91adaef7543bd9c80d4eaec9fc364 : SUCCESS in 15m 50s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/6e133b84175d4319af67420d12e21f70 : SUCCESS in 23m 56s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/bd751722d58a45e5af9ef2c555a2b6e2 : SUCCESS in 17m 42s","accounts_in_message":[],"_revision_number":5},{"id":"0f1bed557900bde9ee8acde38cca9063e3319b0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-24 08:09:34.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fbf30d60e924505a7339db2777ffaa2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86f78e3449c746eca1d681bbb919a637 : SUCCESS in 8m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/8b61d82650eb4295ad962c3d1e9e1235 : SUCCESS in 1h 02m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/31e9abec6ec14ea7b0f08e5bdb651910 : SUCCESS in 1h 00m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/854f093ef5234733b6fea1e8d9322fc1 : SUCCESS in 1h 02m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7c521953ed334a8ab635d23f8bf33a41 : SUCCESS in 22m 47s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/27e0d7d9dd6645a6a2123786d40a19bd : SUCCESS in 19m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/afacc5d3d62743cb9d2452a399fde065 : SUCCESS in 22m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7e84b2ec0c5846a8bcde0c8381c8e498 : SUCCESS in 8m 44s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1e51519ee0364dbfa382b46e1968d997 : SUCCESS in 5m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/778bc676b69d4d60acc0c0c36b32d594 : SUCCESS in 16m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/42bda2003dfd499aabb8bfa7f71bd37e : SUCCESS in 16m 05s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ffca1ec84080466cad76960fd3488144 : SUCCESS in 16m 01s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/b52ed5cb3029435ba03ae66c8cab867a : SUCCESS in 17m 35s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/62bb2e6a28ea48c28ef2de54b552ad9e : SUCCESS in 14m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c7f68b19239d4f3c91d1b172a1fc7478 : SUCCESS in 17m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/44c674ad9ba74f0c8148198f25e83ab3 : SUCCESS in 17m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/009adad983ab4a3692ff09875ac7c378 : SUCCESS in 17m 44s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/aa2f846e2ade4205bddc38bc5ceabad9 : SUCCESS in 14m 36s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/38a6cc27025542b590993c6bc43f88b4 : SUCCESS in 16m 09s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/96ba0d576f9749ff8713950cde37dfd5 : SUCCESS in 16m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/37feac8e73d948c8b046907de42c55f5 : SUCCESS in 5m 38s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d6cf7382d9f94812985dc6e7348132df : SUCCESS in 14m 04s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/060ea8d8ba8b41f38a0ee442f0b0f599 : SUCCESS in 6m 26s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a271987bda7840ee8aeff85b9af5e041 : SUCCESS in 59m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e3e9325267b749e98d5b954d96284bba : SUCCESS in 35m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3ccb77f2b8e94d83be917b9acdfe5a5e : SUCCESS in 35m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7905e7ddd97e406d8ab2913fb30cafbc : SUCCESS in 17m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/389fb86ea3164a1b8c36191c7c23e2b8 : SUCCESS in 3m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1d215530db0446b688409487144eb964 : SUCCESS in 16m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5680077eb61447f08b9698c3a9ad9705 : SUCCESS in 1h 01m 07s","accounts_in_message":[],"_revision_number":5},{"id":"c8fe56b5afd7859e77d9640253cda72482173ad3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-05-27 06:31:03.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"1093ad9678c5d225f70d8f17441c22cfe99228ab","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-05-27 06:32:56.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"04d64c2a420970b36dffcf642bfb88b6aa25e02f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-27 06:56:47.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f91fef68d29247fda3a6885d32d35b4e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/10b9be5f6f1749a18ab06a9bef82139b : SUCCESS in 16m 51s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e935ac5ca56a4632af65b70ebf054b4c : SUCCESS in 16m 47s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4eaf41abc76c482f988dbc16c096ca49 : SUCCESS in 20m 04s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/571adc1af95f4e549bdf338d6487271a : SUCCESS in 24m 34s","accounts_in_message":[],"_revision_number":6},{"id":"f1147020ae642096a6cdbb4dae98e62434e068b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-27 08:37:02.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ba0dd32e074b4530b48b5c9c4c400196\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11651ea9c8044d969719fd14ab5c5682 : SUCCESS in 8m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/4696d13869364c87a698b7f0d2909569 : SUCCESS in 40m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4bf8a477521044468c8df15b6e2aa415 : SUCCESS in 37m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fc3e359ed44f422588fefa8e37bc0113 : SUCCESS in 47m 47s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ede300e2117d4d74a9f4c6c1663abc21 : SUCCESS in 17m 48s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e33d24ecec584e3ba1c30a291bfaf4b5 : SUCCESS in 18m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/97600821d5a5440490acc03d4be6c9a3 : SUCCESS in 19m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/575f32b6a4f9404587a0bba4e950d9fb : SUCCESS in 10m 38s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/54a17cebc476458fac0cda0e8fcb0709 : SUCCESS in 4m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/64affad4baf342479c57b9ae6cd832b9 : SUCCESS in 19m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/04e13f97289c4951b0df1657ee49ef4b : SUCCESS in 20m 50s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/28685b0cedd84066aaa2a1f02f614f2f : SUCCESS in 16m 57s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/97e5e00f91eb4978b83cfb48cef993d3 : SUCCESS in 16m 00s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a113e8f5a68e401c9f639a44ebc0e2be : SUCCESS in 15m 18s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9f7c4d38f7d645269e29b65d11369078 : SUCCESS in 17m 35s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8816e99544204a7c86ed8cc1d9b7d48d : SUCCESS in 18m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/809b29ca26084107a379ed8bbadbe37f : SUCCESS in 18m 44s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2ff2091c7ef6494d94f000dafabc4f20 : SUCCESS in 16m 49s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/981af3c825624708beb526f933737c35 : SUCCESS in 16m 47s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c965ad24731b4c24b45702e407f1d3f3 : SUCCESS in 13m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a19ddb8143f5427bb9da3890f674f8b0 : SUCCESS in 5m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b6301760a626428ba3c60f4ce8d45380 : SUCCESS in 19m 12s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/18a46cc37c8249438510df1f981b162e : SUCCESS in 6m 57s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ff6e0c729f8a418097cdcc8109be047a : TIMED_OUT in 2h 00m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bf0b685c7ad24022be19c89ccd2081ee : SUCCESS in 55m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b7e5e236b4864fe8ad38de93ee104183 : SUCCESS in 36m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b1ea4ec2442a49ae926e78b4a6e4b282 : SUCCESS in 14m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e826d57fdcf48bb8a242cb1d237b249 : SUCCESS in 3m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/74d5beea8ca04ce3ae2e3000e024a4b4 : SUCCESS in 15m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3177c777111b4d0ebf087b51991ba1bf : SUCCESS in 1h 04m 00s","accounts_in_message":[],"_revision_number":6},{"id":"a38eff255fa617978cedf0761126a0f10a0be248","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-05-28 00:48:15.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"74cb9a51a66d7cd50f13c8d682837d85c7593fc8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 01:15:55.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4968615e8df24fcb8dabb1832486c8bd\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/65306aa9198449578dd258e7204b9ef7 : SUCCESS in 15m 09s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c5ea3db4d16c4e1faf3b5060f743f3c1 : SUCCESS in 17m 48s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/870dcf70d69d4d40b22bcf0f38ec66ae : SUCCESS in 26m 12s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/c39aa8bcbbb34758a02e0548a6ef55e7 : SUCCESS in 19m 02s","accounts_in_message":[],"_revision_number":7},{"id":"56f8a9c62ef63f9395b2c60b9890d5e235c28f90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 02:00:20.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aade5f0834f6487b84767220e8f04329\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/135ca66037644436b8cad3b847c8fc4a : SUCCESS in 6m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/da4ba7ac806c4cabb989d0f97087fe2a : SUCCESS in 41m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/218642f4e40344a79e46f6010fdb64d3 : SUCCESS in 54m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a435a5c4786342b7bc9eac843574c2e5 : SUCCESS in 49m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/46a179d335304a3596ad40f8bae54463 : SUCCESS in 18m 44s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/50690cd9731046a8be743445402ec5f3 : SUCCESS in 19m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/94851ae90cdd45989a8fbe9030c59adc : SUCCESS in 19m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/726bea6ede5b4d3cbbc4593c141e78d4 : SUCCESS in 8m 26s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5cdda487a29b46d7a70958f623724bd6 : SUCCESS in 4m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8a055b8aa8fc467c9739d90c9c3b4c72 : SUCCESS in 13m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0e34e73875324abdb340fd6ef36cfbfd : SUCCESS in 12m 50s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/eeb18c40c51b4140aae520873c8e7752 : SUCCESS in 13m 02s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/4c0bdf23f90340ec83650d35c134e5b5 : SUCCESS in 18m 33s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/45d3196231874e84971cb12d46101632 : SUCCESS in 23m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8d909b1d24864398baea293ce7607ff6 : SUCCESS in 17m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fc195fe32dc241afa7341c1991da8f9f : SUCCESS in 18m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/868d6858b8e342bdafe0f45f94e4d0fb : SUCCESS in 19m 47s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/199baa91295d4ce4939a7ddcf319b496 : SUCCESS in 14m 42s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/33212970c87c457b8b8fc66713c88ac6 : SUCCESS in 17m 18s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6e7f2a456ece4c5a93ab5da7ae01df9d : SUCCESS in 16m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ac209ac266a3426a986f5cf9d48516d5 : SUCCESS in 6m 04s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a0f766a989b44f538e2e821e8d941a7a : SUCCESS in 16m 42s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f040b720a8404b60949b9988bb8e18e3 : SUCCESS in 7m 37s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1215ce48ac644630a83a267de7651927 : SUCCESS in 49m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3898cc283b9a4030af6008eb6c386521 : SUCCESS in 52m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d5f893a030ed48f8bf6d3d712b522c9f : SUCCESS in 48m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0eabc46637b04359be2152b397d25300 : SUCCESS in 18m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31ba01a8b75b4494893dcbc20e4c20c2 : SUCCESS in 4m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8c913e7e9a614676b0690d55b540a7c1 : SUCCESS in 20m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d2ddfec086124fb2ad04e4e7db709953 : SUCCESS in 1h 06m 10s","accounts_in_message":[],"_revision_number":7},{"id":"baad32ce2f1f22f3178c7ba1649fd7d88038ca0d","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-06-03 16:41:53.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"f3fd2584fe984e3d555f61f974b3a8879c11fced","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-21 22:58:13.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"1301b7d3f1660aa2f7f431da4238380547747e45","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-06-25 00:49:57.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"e7d9a2c86f484de935c6d5106650cd0ca1bac440","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-06-25 05:05:40.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"c05271057a73af54ce5db67e4a925129d6b0ee0d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 05:28:47.000000000","message":"Patch Set 8:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc56fc3fb75a4c8f8649264a72850717\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4567c7998ddf485585a0e00a5f1a985d : FAILURE in 16m 07s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b257d6b818c44bbda10de1ef6d200ba9 : FAILURE in 16m 20s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/96df1be67138457cbf6a8c32b18b8988 : SUCCESS in 21m 59s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/fd7ef80b062542ffbd7e2725c29c9b2f : SUCCESS in 20m 08s","accounts_in_message":[],"_revision_number":8},{"id":"83b14be4411ad50d3dd234a000717b72ac5ace0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 06:28:00.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/096f3f2ce7ff427aaf3cc8d86c18385f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4461c8eb7dd3439f951163c6e4f2778a : SUCCESS in 7m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/72e6254636f3465c8a4491e49cb2a671 : SUCCESS in 1h 02m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/666f6f9df0874e02a06a45d1a64382ae : SUCCESS in 50m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/89a997b1dcd44fea8096b11bc8799d6c : SUCCESS in 50m 33s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/911687a57da24293bb64f9d51b52c1a9 : SUCCESS in 18m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/34325ce166d048f1b66a73bf53cc6ee5 : SUCCESS in 18m 39s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/67b22269ca0042a2ac242148135377a3 : SUCCESS in 24m 47s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/31bdf659cc1649b0a10f04cff170f25c : SUCCESS in 10m 50s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b72e6fd0614b409cb110e72b1d075a0b : SUCCESS in 4m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d88347687a6149c6a3bb21d302bfc8c8 : FAILURE in 19m 04s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0bca71590500441ba8bcfb0129db1dbf : FAILURE in 18m 06s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/4c9ff8f6e07f45258536110668a4fc19 : FAILURE in 17m 57s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/4184d8f44ac247fb9043bb1c9c13baf9 : FAILURE in 16m 22s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a24d6167cc4c4780a246af438e567a09 : FAILURE in 23m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9bd04163e2b84e10af8758896745da17 : SUCCESS in 18m 25s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/78424ed80c5745a18a64bd86a01827fd : SUCCESS in 17m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3d9dff45f4894299904d1620f26f1232 : SUCCESS in 18m 45s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/728f56a10c1c403e9ee7a33f0d5c19c5 : SUCCESS in 15m 33s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/44c4360ccee94043bf4169efae74a5de : SUCCESS in 15m 38s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8fad21206fe248a7856705599265d885 : SUCCESS in 16m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6dc6c410cf0b447bafcf5805db2791b9 : SUCCESS in 13m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/660bf6f5f4aa4640a838c5ac9c2aef59 : FAILURE in 4m 22s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b2661021d3944330b4cd22074f33e7b4 : SUCCESS in 7m 01s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/125e714351584549bf0be3808bbe2877 : SUCCESS in 55m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/22a89182b6244b9a95e12478e0fd3066 : SUCCESS in 54m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cfe4f05640f846dda27e8c857d793623 : SUCCESS in 47m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fae74b58d0064b52b84a49ad83036e14 : FAILURE in 16m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/003f74e8cd484e5e81a0cee7238fdace : SUCCESS in 4m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/730e0d4a88b44481893fcf2805ba10e1 : SUCCESS in 17m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/94f7fe619ec64304bf8bfbb7ec1a1639 : SUCCESS in 1h 01m 17s","accounts_in_message":[],"_revision_number":8},{"id":"33ec8ca0b6c21d8e5a674cbf3e5e043c80c1d712","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-07-11 03:29:09.000000000","message":"Patch Set 8:\n\n(4 comments)","accounts_in_message":[],"_revision_number":8},{"id":"3488b1eb46372362d6c0205eac8a1b8f314477f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-07-11 07:31:10.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"e7e2f3b2acccd80129774521a53b65dd7402b4fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 08:44:48.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f832aa6d3974198913e2d43278438a3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d026e0172bd433fb84b73ba45f9bbfc : SUCCESS in 10m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/b8e6c88855424594ae73b46a9ed292c5 : SUCCESS in 1h 03m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e0b37f94c55b48eaa3c39cc34bf6a81a : SUCCESS in 39m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cd3417f1175c431ca295dbfd90fb8e66 : SUCCESS in 54m 33s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b82e53b0826c49158aecb0830dbc61a7 : SUCCESS in 16m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bf9b8bb28b1641929ab730fd55d9b349 : SUCCESS in 19m 55s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f91e7722784f47bd8acd794ab515b87f : SUCCESS in 21m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c9b88beaff494933929bdb7b8c3b7586 : SUCCESS in 5m 06s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c7b6034b88c24bff85d8fe523a48fdec : SUCCESS in 19m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/00755c56023f4cf38bb760cdded9b652 : SUCCESS in 19m 54s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c88526b072c64824b3bcb4cc197b720b : SUCCESS in 24m 33s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/838f1791fef94a15aa2a1d65aec5493c : SUCCESS in 27m 11s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/817c2e4a6c0d45f7bad96c2cdfa0d9fd : SUCCESS in 27m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/69eee53778774cf79e0ebae4dd356686 : SUCCESS in 17m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/90df986045954765a9c4178c4f400fa6 : SUCCESS in 25m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/04baac56e4334422a4a47fcba470567d : SUCCESS in 18m 38s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d33aa40b13bf47a6a51ce773b92b2776 : SUCCESS in 23m 19s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f75e6cdcfc814cb3a86c07fbf089d64b : SUCCESS in 16m 27s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/54718e46824d45ee90ab6c6b261557fd : SUCCESS in 17m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7e45e6dab3bc400d911a9b78579505ef : SUCCESS in 5m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/40053c6cf99f491f82830a14acc26d7c : SUCCESS in 17m 56s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3e1d3c84bc7a43ff90740394ca1684f2 : SUCCESS in 9m 25s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7e25fdf185174e968f5276c8304f8674 : SUCCESS in 57m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/90f229f234144a6d8b8574fc375d0b98 : SUCCESS in 52m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c71ea987662541a09d7f9afedaa83df0 : SUCCESS in 51m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/810d988495934997b7eedd0419b57a74 : SUCCESS in 18m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e354baeb2a34fc787c376b69ec4fba9 : SUCCESS in 5m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a07c548054d9471996715b91bafeaf06 : SUCCESS in 25m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f9e04c91f2d64751bb8a95548c18abcd : SUCCESS in 58m 47s","accounts_in_message":[],"_revision_number":9},{"id":"2de96f230756bd2bc96abfad08cd8d390a4ff656","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 08:52:20.000000000","message":"Patch Set 9:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f28fe0d8ee894aa5a0d2dab917a6964b\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/093ed36fe90e4bc2b842c817d2a74c8a : FAILURE in 3m 29s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/68683be2ea584899a1d41ba5f66ddc2e : FAILURE in 3m 03s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c998f932c4614baeb7b5b76c00903166 : SUCCESS in 1h 19m 20s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3eb0dc9dfa5f446981fd30a40e92ca0c : FAILURE in 2m 46s","accounts_in_message":[],"_revision_number":9},{"id":"146b36233a278c36c29bee5396ea356be872d2c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-07-15 20:25:55.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"a35feb473a15682749caab9587ee7ef0255ff14c","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-07-15 20:32:00.000000000","message":"Patch Set 10: Code-Review+1\n\n(21 comments)","accounts_in_message":[],"_revision_number":10},{"id":"110cd964ee6ffabb5e82c0db050b5f6aaed8369f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-15 21:59:04.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4e859d1ccd041238ee00d67c4132c9e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/89e73bb6d05c4414844e564dcc98d146 : SUCCESS in 17m 08s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/84f5f6b74301439fbfc8450863521788 : SUCCESS in 1h 31m 00s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/02cbdce72e4947638f718308733a1d74 : SUCCESS in 21m 54s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/55f6084a5eb141ae8135774d41cd5c56 : SUCCESS in 1h 13m 29s","accounts_in_message":[],"_revision_number":10},{"id":"d60a3ab8307da5625b7b8f6e41d0cd7ecdaae3e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-15 22:09:40.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a39e64ab57c949b1b5ae83aa42003bf3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79b6fe155e23462fa4d6ccec529ea3e2 : SUCCESS in 10m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/0fcbe4d95d054db399990923ca3ab743 : SUCCESS in 50m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0546637e25434c599289ef6b27894c9e : SUCCESS in 56m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/98a6e40322074ee2a535ae6f0dab5da6 : SUCCESS in 57m 26s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b351c5d8debd4b2c9a5f39734569e35e : SUCCESS in 20m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bdad30480d1f482ca9301b511e761dcb : SUCCESS in 21m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9e42478362be4da9a545d11c8b8a91ce : SUCCESS in 19m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/423c2f099e3e4e0a8b07c56f9671c9fc : SUCCESS in 6m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/289f2f6c9ec44f4095263290db4d9f68 : SUCCESS in 21m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1a4d2f7009ca4844b9294c30ebfc1b6c : SUCCESS in 20m 59s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b98a99a3f4d44e049a33507605744db6 : SUCCESS in 22m 02s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/3671799edccc4bc59a49c6ec7426ab97 : SUCCESS in 15m 41s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7683272bb7fb4119a08f021b88e3e015 : SUCCESS in 25m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7957f073a85f49ba8d31dae31b6d1c58 : SUCCESS in 22m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/96f1e935fb9a4c4a861623b43b034bdf : SUCCESS in 18m 24s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/533140ec47ee47a4ba31effc04e6a7c1 : SUCCESS in 18m 54s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b7f26855ac4a425292dc9d919ad9a63a : SUCCESS in 20m 45s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/fc06b8506270405a9fcab07c58f56df5 : SUCCESS in 21m 20s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a0871bcfc90c49ba8851fae82f791e4a : SUCCESS in 19m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/81d0e566234e4353a1374f2188b6753c : SUCCESS in 7m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7dbf4eaafcdb4f08875aeaa7f450da11 : SUCCESS in 13m 44s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/773dec2b2530403daa9503c2591f608e : SUCCESS in 6m 17s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a5d851dede0948bc893039693b870e1a : SUCCESS in 1h 12m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/292ce8dd3c254891855b132b5ab2003e : SUCCESS in 51m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/23133a00c06b4c24b70d00857242bad2 : SUCCESS in 46m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/876e26fd47c043c1bfe75c3b1f62be46 : SUCCESS in 18m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd81585b01714d47b2e14141747206a8 : SUCCESS in 3m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6c4555a577e6449b91899417b8800b3f : SUCCESS in 24m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/95d90147446641a3a7d79bce529c91b9 : SUCCESS in 1h 01m 17s","accounts_in_message":[],"_revision_number":10},{"id":"56c4a857c99162c2e93811f8aec1b95febbe02e6","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-07-17 08:05:33.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"af5c47d5bfe12031ac4bcf591248a15396c4e755","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-07-22 22:41:50.000000000","message":"Patch Set 10: Code-Review+1\n\n(7 comments)","accounts_in_message":[],"_revision_number":10},{"id":"4f5019ead56d8283c17457456e34f32b7251eaf4","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-08-05 12:03:06.000000000","message":"Patch Set 10: Code-Review+1\n\n(9 comments)","accounts_in_message":[],"_revision_number":10},{"id":"e2a50d2a97ac8ac925eec380e824ba3af612878a","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-01-15 17:47:04.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"f625b1f561f78668850a65337a731de84c7b0cff","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-04-10 00:21:33.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"2a8501246377360cd12fe5df6b8f9bfb64f2b5ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-04-10 23:23:50.000000000","message":"Patch Set 11: Patch Set 10 was rebased on behalf of \u003cGERRIT_ACCOUNT_1179\u003e\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}],"_revision_number":11},{"id":"191334b06bc61631b0084c07397f3bcd0a0c6ffa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-04-10 23:24:56.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":12},{"id":"78699d3a38d1f7d1d097f374ac51fc2545ae2772","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-11 00:04:29.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b34b6c1bd19d4673bfa8f8974aa0deda\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/56af14b3d9ef4557a95f3cff636f21c8 : SUCCESS in 37m 42s","accounts_in_message":[],"_revision_number":12},{"id":"baa9b41c73511e344f1f6a81378130e4bd9f1cdc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-11 00:36:54.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7391b7673f1c422aa644ef277dd62b15\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f8a9f6dfb96491c987ee8538799e67c : SUCCESS in 8m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/e7208b61e52a4112963803fd431e1ada : SUCCESS in 46m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/46a085b7a3714ec78054acbb61eef1f9 : SUCCESS in 42m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/62ca7d9a35bc42ef9033e8e0fc4fae21 : SUCCESS in 36m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9f0593d33fd445e2abfbf7378f6c4307 : SUCCESS in 46m 24s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f60b0faa79a54001a8622a1f275080dc : SUCCESS in 19m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5a0b0fceb8bc42d6986125462eecf70c : SUCCESS in 15m 28s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/61183ebadaff4a1a855b08e40e403fa0 : SUCCESS in 17m 50s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/54d3b0518b74492989781b81fe0141a4 : SUCCESS in 4m 31s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/101866e0816f45489ae2665bcfe52264 : SUCCESS in 12m 51s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5d7c3d98a50e4808b26bb6939d76ea42 : SUCCESS in 7m 48s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7abb50d6998748fa886794c2fcd2b8cb : SUCCESS in 17m 11s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a81af9dc3ba14e5aab232a4ae1bff980 : SUCCESS in 9m 03s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f97ae33088b94b729b53cc4fded49147 : SUCCESS in 7m 43s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6d195357723e4cc08ecdf3210de01205 : SUCCESS in 14m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/735ec484b5f646c29bde725c387d9b81 : SUCCESS in 8m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3fe7b9da313945329ac84ed168453714 : SUCCESS in 15m 58s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c9432a6a42c5475fa747665dfa787b3f : SUCCESS in 4m 24s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0c4a8447ab6c44af897bbb07971241f8 : SUCCESS in 32m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0924f4697126489796cc32cfcd5a9b82 : SUCCESS in 40m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7fc843aa293941628b3ddfe7280e465c : SUCCESS in 51m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/95b640d311c847cab5a48689cb64f7b5 : SUCCESS in 23m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73b24b447a76412c95586f071031ab9b : SUCCESS in 2m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/881d712a19964b4f93afed6400023a92 : SUCCESS in 20m 28s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/795c7f0f09c14c1ea9aea74d47821842 : SUCCESS in 45m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/56b11bf4c8a24e5eb724923e62754ee5 : SUCCESS in 25m 43s","accounts_in_message":[],"_revision_number":12},{"id":"f65303b402ce785ac4fcb3d877c3a0ab2ae1a5a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-05-20 15:56:54.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"574b4c628316cb718d6f6f5421b4a28d1d3bbffd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-05-20 17:01:45.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.","accounts_in_message":[],"_revision_number":14},{"id":"92bab8e863ccc2fc0cfec399427609cf707baea1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 18:16:36.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bcf9c91e04954ba2880b913404152084\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d15fbcd802e4869a732cdf01414039e : SUCCESS in 10m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/118d6a8b4ccb44c0ad968faaa47d0ea1 : SUCCESS in 1h 01m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/370dfbe85d154c618f2f257edddd0ecb : SUCCESS in 57m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ac7480b716a545899a1792c5235b0d13 : SUCCESS in 45m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/560feec1a15c4716851ab228767ad0a7 : SUCCESS in 55m 23s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/37fd1e5be060468e99e82828d799e3a2 : SUCCESS in 16m 17s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0a0ec80d42f74599ae05b165076b6c84 : SUCCESS in 28m 02s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/70495db1af684c3a8849416a073c0414 : SUCCESS in 27m 07s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fb867455cb25422686066d50fc155292 : SUCCESS in 5m 01s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c5e3f8027d604ba5ba07b7098d754e39 : SUCCESS in 8m 21s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/498b811b39b443e4bac76aa8cf0361cb : SUCCESS in 24m 38s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8248a3aefa7c4f68ab3509029634c6c0 : SUCCESS in 10m 28s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/059983d766f54edfb019c3bc140b0c1d : SUCCESS in 34m 00s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d3edffcedcdc40d6a833144e6fbea040 : SUCCESS in 13m 21s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e17e652e5ac343ffbd43969c5855f6f9 : SUCCESS in 13m 17s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ed164d9d924c4be09205dd5c3426156a : SUCCESS in 22m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/53ff11f30e194f019da8ece46e453ce0 : SUCCESS in 3m 53s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/869e0910078e41dcb54aa1fd50633bf0 : SUCCESS in 8m 11s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/13c204fb02d44937a9e94d355511ae74 : SUCCESS in 7m 37s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ad0650d61aec4ac7a301c81e35e16a73 : SUCCESS in 50m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/912d5d980b6e41e7b062478beb7117a9 : SUCCESS in 19m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ebd447daa1734ba8a9d5a5f4fbcea462 : SUCCESS in 48m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7f7fc8c84ff642548f8ebc6eae2d4047 : SUCCESS in 23m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e85fa27f5ab4300873a1e5901b61d62 : SUCCESS in 5m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6ca4608ca06b4707995370e723ab7c1e : SUCCESS in 19m 36s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d11c8c50a62f4399865584f2f613d103 : SUCCESS in 56m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b780e4addf404990be525718782888d7 : SUCCESS in 1h 05m 00s","accounts_in_message":[],"_revision_number":14},{"id":"c34a144939161e99ec3188c7ecaa28914d572641","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 20:16:13.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8acc8816c1f449158f54066dcf16de34\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/382757a11bb24eb280b75125160d1936 : SUCCESS in 44m 34s","accounts_in_message":[],"_revision_number":14},{"id":"94db2f30d70c30f78686bfbb24f90c8ee9302cde","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2026-02-24 08:39:39.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"581ef43b7e9caab9ae26cecc1996b353b52b065c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 09:10:16.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39651f7067634627a81e75b926411d0a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ac4dda15852a486581c77327683ae497 : SUCCESS in 28m 47s","accounts_in_message":[],"_revision_number":15},{"id":"0bcfe648e5e811ea61229e4beecf1a8524760d25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-24 09:47:11.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44c80f148d2e4e538bcd1c29ebd6bc5f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07953c681f70483db6193cc21e68e101 : SUCCESS in 5m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/24edd095d4a74192918dd07d851b5671 : SUCCESS in 50m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0a4f3719c0b148bfaa22df74b1462306 : SUCCESS in 59m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d17e3af1730e4331a793286c299fec46 : SUCCESS in 34m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cd4b2c129aa449759643310c3aa529b3 : SUCCESS in 53m 23s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3de359722f55432cb0ba864aaa98da8c : SUCCESS in 18m 14s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6d69f617b21e4f32b8f33d68ef6e28a7 : SUCCESS in 18m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2f62c8c35eed4e6e98c2889a07d07b33 : SUCCESS in 10m 19s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/98f94f84f09f473b8b06338d4e2c0562 : SUCCESS in 6m 17s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e286f0b42044459c84fe848e9b5789cd : SUCCESS in 10m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/165286c123754a50a6c88998b902aeed : SUCCESS in 19m 11s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/faf1c45c1abb4335b87eb42748bb1d27 : SUCCESS in 12m 45s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/5b3591dc8baa4fe89ade13a9fb8af6b7 : SUCCESS in 21m 12s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/7a0f6b00cca54fb4a5885d61eada17d6 : SUCCESS in 14m 14s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e00e283189bc4fa5bec122470cb44f75 : SUCCESS in 8m 12s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/db7010be21814a45b23afc8dccfedfa7 : SUCCESS in 8m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d2af10c379b04caa843e8ee2fac90512 : SUCCESS in 8m 32s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b2ea777bba5d4ba8a4cc1d716af3f382 : SUCCESS in 17m 11s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/97ad48033615444f9177ac1be184546c : SUCCESS in 7m 42s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3cc8797dbb91470bb120143349855c0a : SUCCESS in 59m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0a2849eef73d45ee93824513acd1c5f4 : SUCCESS in 24m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e8821266e0ca4c41b1560b11c22903ce : SUCCESS in 22m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/46635c812a884ff6b3f90cb47cd3ade5 : SUCCESS in 12m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0be1f1c2d9b8442b9c6cd07b6c6afe43 : SUCCESS in 5m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c0003808c4144fbcb1176d260e1adc36 : SUCCESS in 17m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0e1012001a894124a045a368186a57ee : SUCCESS in 1h 00m 12s","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"106202a019bd9787f2968e40474aec287515a59f","revisions":{"e383da3baacf23e9a00a905e83ce525dfb12af9e":{"kind":"REWORK","_number":1,"created":"2024-05-15 07:27:32.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/67/919667/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/919667/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/919667/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/919667/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/919667/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/919667/1"}}},"commit":{"parents":[{"commit":"fc646045fb9b5b83389fc5b3dd09d2a4e85dc898","subject":"Pass db_state to container-update and async_pending","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fc646045fb9b5b83389fc5b3dd09d2a4e85dc898"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2024-05-15 06:32:12.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2024-05-15 07:27:09.000000000","tz":600},"subject":"obj: skip to async_pending when root is busy","message":"obj: skip to async_pending when root is busy\n\nThe proxy will send the shard to update to the object server normally\nunless it failed to get that information. It would usually fail to get\nthis information when the root container server is busy and not\nresponding in time.\nThe proxy also sends through the current db_state of the root container.\nIt knows this from the container_info.\n\nKnowing this, the db_state is SHARDED but we don\u0027t have a shard, this\nindicates the root container isn\u0027t responding in a timely manner. As\nsuch it probably isn\u0027t a good idea to fire questions hoping for a 301\nredirect response and overload it more. Instead let\u0027s skip the\n\"synchronous\" update part and write directly to async_pending and let\nthe object updater deal with it later.\n\nHopefully this will buy the root container some extra time to get hold\nof itself.\n\nChange-Id: I6fcec6748c3fd0f22baac69f7ceb8261c5bb5d63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e383da3baacf23e9a00a905e83ce525dfb12af9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e383da3baacf23e9a00a905e83ce525dfb12af9e"}]},"branch":"refs/heads/master"},"d9e0703829fea641685e349ad7bf6fd1128ed947":{"kind":"REWORK","_number":2,"created":"2024-05-15 07:31:07.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/67/919667/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/919667/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/919667/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/919667/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/919667/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/919667/2"}}},"commit":{"parents":[{"commit":"fc646045fb9b5b83389fc5b3dd09d2a4e85dc898","subject":"Pass db_state to container-update and async_pending","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fc646045fb9b5b83389fc5b3dd09d2a4e85dc898"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2024-05-15 06:32:12.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2024-05-15 07:30:52.000000000","tz":600},"subject":"obj: skip to async_pending when root is busy","message":"obj: skip to async_pending when root is busy\n\nThe proxy will send the shard to update to the object server normally\nunless it failed to get that information. It would usually fail to get\nthis information when the root container server is busy and not\nresponding in time.\nThe proxy also sends through the current db_state of the root container.\nIt knows this from the container_info.\n\nKnowing this, the db_state is SHARDED but we don\u0027t have a shard, this\nindicates the root container isn\u0027t responding in a timely manner. As\nsuch it probably isn\u0027t a good idea to fire questions hoping for a 301\nredirect response and overload it more. Instead let\u0027s skip the\n\"synchronous\" update part and write directly to async_pending and let\nthe object updater deal with it later.\n\nHopefully this will buy the root container some extra time to get hold\nof itself.\n\nChange-Id: I6fcec6748c3fd0f22baac69f7ceb8261c5bb5d63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d9e0703829fea641685e349ad7bf6fd1128ed947"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d9e0703829fea641685e349ad7bf6fd1128ed947"}]},"branch":"refs/heads/master"},"67f7400931a0e1804cb4663a0f7aaed8dfae1e9c":{"kind":"REWORK","_number":3,"created":"2024-05-16 06:23:04.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/67/919667/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/919667/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/919667/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/919667/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/919667/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/919667/3"}}},"commit":{"parents":[{"commit":"fc646045fb9b5b83389fc5b3dd09d2a4e85dc898","subject":"Pass db_state to container-update and async_pending","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fc646045fb9b5b83389fc5b3dd09d2a4e85dc898"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2024-05-15 06:32:12.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2024-05-16 06:22:47.000000000","tz":600},"subject":"obj: skip to async_pending when root is busy","message":"obj: skip to async_pending when root is busy\n\nThe proxy will send the shard to update to the object server normally\nunless it failed to get that information. It would usually fail to get\nthis information when the root container server is busy and not\nresponding in time.\nThe proxy also sends through the current db_state of the root container.\nIt knows this from the container_info.\n\nKnowing this, the db_state is SHARDED but we don\u0027t have a shard, this\nindicates the root container isn\u0027t responding in a timely manner. As\nsuch it probably isn\u0027t a good idea to fire questions hoping for a 301\nredirect response and overload it more. Instead let\u0027s skip the\n\"synchronous\" update part and write directly to async_pending and let\nthe object updater deal with it later.\n\nHopefully this will buy the root container some extra time to get hold\nof itself.\n\nChange-Id: I6fcec6748c3fd0f22baac69f7ceb8261c5bb5d63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/67f7400931a0e1804cb4663a0f7aaed8dfae1e9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/67f7400931a0e1804cb4663a0f7aaed8dfae1e9c"}]},"branch":"refs/heads/master"},"ca0d8ccccc5fda0ca88cb207da7cd7ba727216a5":{"kind":"REWORK","_number":4,"created":"2024-05-22 07:05:57.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/67/919667/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/919667/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/919667/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/919667/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/919667/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/919667/4"}}},"commit":{"parents":[{"commit":"fc646045fb9b5b83389fc5b3dd09d2a4e85dc898","subject":"Pass db_state to container-update and async_pending","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fc646045fb9b5b83389fc5b3dd09d2a4e85dc898"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2024-05-15 06:32:12.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2024-05-22 07:05:36.000000000","tz":600},"subject":"obj: skip to async_pending when root is busy","message":"obj: skip to async_pending when root is busy\n\nThe proxy will send the shard to update to the object server normally\nunless it failed to get that information. It would usually fail to get\nthis information when the root container server is busy and not\nresponding in time.\nThe proxy also sends through the current db_state of the root container.\nIt knows this from the container_info.\n\nKnowing this, the db_state is SHARDED but we don\u0027t have a shard, this\nindicates the root container isn\u0027t responding in a timely manner. As\nsuch it probably isn\u0027t a good idea to fire questions hoping for a 301\nredirect response and overload it more. Instead let\u0027s skip the\n\"synchronous\" update part and write directly to async_pending and let\nthe object updater deal with it later.\n\nHopefully this will buy the root container some extra time to get hold\nof itself.\n\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nChange-Id: I6fcec6748c3fd0f22baac69f7ceb8261c5bb5d63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ca0d8ccccc5fda0ca88cb207da7cd7ba727216a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ca0d8ccccc5fda0ca88cb207da7cd7ba727216a5"}]},"branch":"refs/heads/master"},"0b5b8689af127a0cbcc3cb83a5fb945cfc7173ee":{"kind":"REWORK","_number":5,"created":"2024-05-24 06:39:04.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/67/919667/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/919667/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/919667/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/919667/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/919667/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/919667/5"}}},"commit":{"parents":[{"commit":"fc646045fb9b5b83389fc5b3dd09d2a4e85dc898","subject":"Pass db_state to container-update and async_pending","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fc646045fb9b5b83389fc5b3dd09d2a4e85dc898"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2024-05-15 06:32:12.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2024-05-24 06:38:45.000000000","tz":600},"subject":"obj: skip to async_pending when root is busy","message":"obj: skip to async_pending when root is busy\n\nThe proxy will send the shard to update to the object server normally\nunless it failed to get that information. It would usually fail to get\nthis information when the root container server is busy and not\nresponding in time.\nThe proxy also sends through the current db_state of the root container.\nIt knows this from the container_info.\n\nKnowing this, the db_state is SHARDED but we don\u0027t have a shard, this\nindicates the root container isn\u0027t responding in a timely manner. As\nsuch it probably isn\u0027t a good idea to fire questions hoping for a 301\nredirect response and overload it more. Instead let\u0027s skip the\n\"synchronous\" update part and write directly to async_pending and let\nthe object updater deal with it later.\n\nHopefully this will buy the root container some extra time to get hold\nof itself.\n\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nChange-Id: I6fcec6748c3fd0f22baac69f7ceb8261c5bb5d63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0b5b8689af127a0cbcc3cb83a5fb945cfc7173ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0b5b8689af127a0cbcc3cb83a5fb945cfc7173ee"}]},"branch":"refs/heads/master"},"2290a8ae79b66d0377185fdecbb2b9234a18ae07":{"kind":"REWORK","_number":6,"created":"2024-05-27 06:31:03.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/67/919667/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/919667/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/919667/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/919667/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/919667/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/919667/6"}}},"commit":{"parents":[{"commit":"fc646045fb9b5b83389fc5b3dd09d2a4e85dc898","subject":"Pass db_state to container-update and async_pending","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fc646045fb9b5b83389fc5b3dd09d2a4e85dc898"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2024-05-15 06:32:12.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2024-05-27 06:30:07.000000000","tz":600},"subject":"obj: skip to async_pending when root is busy","message":"obj: skip to async_pending when root is busy\n\nThe proxy will send the shard to update to the object server normally\nunless it failed to get that information. It would usually fail to get\nthis information when the root container server is busy and not\nresponding in time.\nThe proxy also sends through the current db_state of the root container.\nIt knows this from the container_info.\n\nKnowing this, the db_state is SHARDED but we don\u0027t have a shard, this\nindicates the root container isn\u0027t responding in a timely manner. As\nsuch it probably isn\u0027t a good idea to fire questions hoping for a 301\nredirect response and overload it more. Instead let\u0027s skip the\n\"synchronous\" update part and write directly to async_pending and let\nthe object updater deal with it later.\n\nHopefully this will buy the root container some extra time to get hold\nof itself.\n\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nChange-Id: I6fcec6748c3fd0f22baac69f7ceb8261c5bb5d63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2290a8ae79b66d0377185fdecbb2b9234a18ae07"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2290a8ae79b66d0377185fdecbb2b9234a18ae07"}]},"branch":"refs/heads/master"},"da8ca404d552aa9d53090d8209a57fa820eda17a":{"kind":"REWORK","_number":7,"created":"2024-05-28 00:48:15.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/67/919667/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/919667/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/919667/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/919667/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/919667/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/919667/7"}}},"commit":{"parents":[{"commit":"fc646045fb9b5b83389fc5b3dd09d2a4e85dc898","subject":"Pass db_state to container-update and async_pending","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fc646045fb9b5b83389fc5b3dd09d2a4e85dc898"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2024-05-15 06:32:12.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2024-05-28 00:47:50.000000000","tz":600},"subject":"obj: skip to async_pending when root is busy","message":"obj: skip to async_pending when root is busy\n\nThe proxy will send the shard to update to the object server normally\nunless it failed to get that information. It would usually fail to get\nthis information when the root container server is busy and not\nresponding in time.\nThe proxy also sends through the current db_state of the root container.\nIt knows this from the container_info.\n\nKnowing this, the db_state is SHARDED but we don\u0027t have a shard, this\nindicates the root container isn\u0027t responding in a timely manner. As\nsuch it probably isn\u0027t a good idea to fire questions hoping for a 301\nredirect response and overload it more. Instead let\u0027s skip the\n\"synchronous\" update part and write directly to async_pending and let\nthe object updater deal with it later.\n\nHopefully this will buy the root container some extra time to get hold\nof itself.\n\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nChange-Id: I6fcec6748c3fd0f22baac69f7ceb8261c5bb5d63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/da8ca404d552aa9d53090d8209a57fa820eda17a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/da8ca404d552aa9d53090d8209a57fa820eda17a"}]},"branch":"refs/heads/master"},"b672c83229c6ad2d9795e04e747fa54ac9d324f5":{"kind":"REWORK","_number":8,"created":"2024-06-25 05:05:40.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/67/919667/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/919667/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/919667/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/919667/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/919667/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/919667/8"}}},"commit":{"parents":[{"commit":"fc646045fb9b5b83389fc5b3dd09d2a4e85dc898","subject":"Pass db_state to container-update and async_pending","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fc646045fb9b5b83389fc5b3dd09d2a4e85dc898"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2024-05-15 06:32:12.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2024-06-25 05:05:14.000000000","tz":600},"subject":"obj: skip to async_pending when root is busy","message":"obj: skip to async_pending when root is busy\n\nThe proxy will send the shard to update to the object server normally\nunless it failed to get that information. It would usually fail to get\nthis information when the root container server is busy and not\nresponding in time.\nThe proxy also sends through the current db_state of the root container.\nIt knows this from the container_info.\n\nKnowing this, the db_state is SHARDED but we don\u0027t have a shard, this\nindicates the root container isn\u0027t responding in a timely manner. As\nsuch it probably isn\u0027t a good idea to fire questions hoping for a 301\nredirect response and overload it more. Instead let\u0027s skip the\n\"synchronous\" update part and write directly to async_pending and let\nthe object updater deal with it later.\n\nHopefully this will buy the root container some extra time to get hold\nof itself.\n\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nChange-Id: I6fcec6748c3fd0f22baac69f7ceb8261c5bb5d63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b672c83229c6ad2d9795e04e747fa54ac9d324f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b672c83229c6ad2d9795e04e747fa54ac9d324f5"}]},"branch":"refs/heads/master"},"d32da227e3b5751f803826e390b21b993394db53":{"kind":"REWORK","_number":9,"created":"2024-07-11 07:31:10.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/67/919667/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/919667/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/919667/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/919667/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/919667/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/919667/9"}}},"commit":{"parents":[{"commit":"8f2a66c5440efc4352fb150fd644c9cc19e8c821","subject":"Pass db_state to container-update and async_pending","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8f2a66c5440efc4352fb150fd644c9cc19e8c821"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2024-05-15 06:32:12.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2024-07-11 07:30:52.000000000","tz":600},"subject":"obj: skip to async_pending when root is busy","message":"obj: skip to async_pending when root is busy\n\nThe proxy will send the shard to update to the object server normally\nunless it failed to get that information. It would usually fail to get\nthis information when the root container server is busy and not\nresponding in time.\nThe proxy also sends through the current db_state of the root container.\nIt knows this from the container_info.\n\nKnowing this, the db_state is SHARDED but we don\u0027t have a shard, this\nindicates the root container isn\u0027t responding in a timely manner. As\nsuch it probably isn\u0027t a good idea to fire questions hoping for a 301\nredirect response and overload it more. Instead let\u0027s skip the\n\"synchronous\" update part and write directly to async_pending and let\nthe object updater deal with it later.\n\nHopefully this will buy the root container some extra time to get hold\nof itself.\n\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nChange-Id: I6fcec6748c3fd0f22baac69f7ceb8261c5bb5d63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d32da227e3b5751f803826e390b21b993394db53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d32da227e3b5751f803826e390b21b993394db53"}]},"branch":"refs/heads/master"},"14696fdb2954e2a37bda4773645655c62e75201d":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2024-07-15 20:25:55.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/67/919667/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/919667/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/919667/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/919667/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/919667/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/919667/10"}}},"commit":{"parents":[{"commit":"e8affa7db5817e50cd82130a8f4c250725b80884","subject":"Pass db_state to container-update and async_pending","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e8affa7db5817e50cd82130a8f4c250725b80884"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2024-05-15 06:32:12.000000000","tz":600},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-07-15 20:25:41.000000000","tz":-300},"subject":"obj: skip to async_pending when root is busy","message":"obj: skip to async_pending when root is busy\n\nThe proxy will send the shard to update to the object server normally\nunless it failed to get that information. It would usually fail to get\nthis information when the root container server is busy and not\nresponding in time.\nThe proxy also sends through the current db_state of the root container.\nIt knows this from the container_info.\n\nKnowing this, the db_state is SHARDED but we don\u0027t have a shard, this\nindicates the root container isn\u0027t responding in a timely manner. As\nsuch it probably isn\u0027t a good idea to fire questions hoping for a 301\nredirect response and overload it more. Instead let\u0027s skip the\n\"synchronous\" update part and write directly to async_pending and let\nthe object updater deal with it later.\n\nHopefully this will buy the root container some extra time to get hold\nof itself.\n\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nChange-Id: I6fcec6748c3fd0f22baac69f7ceb8261c5bb5d63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/14696fdb2954e2a37bda4773645655c62e75201d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/14696fdb2954e2a37bda4773645655c62e75201d"}]},"branch":"refs/heads/master"},"6ae8aff350c8358b20e77cc31668f9da01f91532":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2025-04-10 23:23:50.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/67/919667/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/919667/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/919667/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/919667/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/919667/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/919667/11"}}},"commit":{"parents":[{"commit":"c7f93f0f63841140d75c21a833647073fb3f4eb3","subject":"Merge \"Add skip reason next to test name in zuul output\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c7f93f0f63841140d75c21a833647073fb3f4eb3"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2024-05-15 06:32:12.000000000","tz":600},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-04-10 23:23:50.000000000","tz":0},"subject":"obj: skip to async_pending when root is busy","message":"obj: skip to async_pending when root is busy\n\nThe proxy will send the shard to update to the object server normally\nunless it failed to get that information. It would usually fail to get\nthis information when the root container server is busy and not\nresponding in time.\nThe proxy also sends through the current db_state of the root container.\nIt knows this from the container_info.\n\nKnowing this, the db_state is SHARDED but we don\u0027t have a shard, this\nindicates the root container isn\u0027t responding in a timely manner. As\nsuch it probably isn\u0027t a good idea to fire questions hoping for a 301\nredirect response and overload it more. Instead let\u0027s skip the\n\"synchronous\" update part and write directly to async_pending and let\nthe object updater deal with it later.\n\nHopefully this will buy the root container some extra time to get hold\nof itself.\n\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nChange-Id: I6fcec6748c3fd0f22baac69f7ceb8261c5bb5d63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6ae8aff350c8358b20e77cc31668f9da01f91532"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6ae8aff350c8358b20e77cc31668f9da01f91532"}]},"branch":"refs/heads/master","description":"Rebase"},"cd731e43cd86ed33da05f214067a4a53cb3daebb":{"kind":"REWORK","_number":12,"created":"2025-04-10 23:24:56.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/67/919667/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/919667/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/919667/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/919667/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/919667/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/919667/12"}}},"commit":{"parents":[{"commit":"c7f93f0f63841140d75c21a833647073fb3f4eb3","subject":"Merge \"Add skip reason next to test name in zuul output\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c7f93f0f63841140d75c21a833647073fb3f4eb3"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2024-05-15 06:32:12.000000000","tz":600},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-04-10 23:24:49.000000000","tz":-420},"subject":"obj: skip to async_pending when root is busy","message":"obj: skip to async_pending when root is busy\n\nThe proxy will send the shard to update to the object server normally\nunless it failed to get that information. It would usually fail to get\nthis information when the root container server is busy and not\nresponding in time.\nThe proxy also sends through the current db_state of the root container.\nIt knows this from the container_info.\n\nKnowing this, the db_state is SHARDED but we don\u0027t have a shard, this\nindicates the root container isn\u0027t responding in a timely manner. As\nsuch it probably isn\u0027t a good idea to fire questions hoping for a 301\nredirect response and overload it more. Instead let\u0027s skip the\n\"synchronous\" update part and write directly to async_pending and let\nthe object updater deal with it later.\n\nHopefully this will buy the root container some extra time to get hold\nof itself.\n\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nChange-Id: I6fcec6748c3fd0f22baac69f7ceb8261c5bb5d63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cd731e43cd86ed33da05f214067a4a53cb3daebb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cd731e43cd86ed33da05f214067a4a53cb3daebb"}]},"branch":"refs/heads/master"},"41712d4749a8b8fc8737626b9c447eacd228e3dc":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2025-05-20 15:56:54.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/67/919667/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/919667/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/919667/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/919667/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/919667/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/919667/13"}}},"commit":{"parents":[{"commit":"d06f01728b1df0f95493ac39edffed3dedacbb2c","subject":"db_auditor: add vacuum support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d06f01728b1df0f95493ac39edffed3dedacbb2c"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2024-05-15 06:32:12.000000000","tz":600},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-05-20 14:56:45.000000000","tz":-420},"subject":"obj: skip to async_pending when root is busy","message":"obj: skip to async_pending when root is busy\n\nThe proxy will send the shard to update to the object server normally\nunless it failed to get that information. It would usually fail to get\nthis information when the root container server is busy and not\nresponding in time.\nThe proxy also sends through the current db_state of the root container.\nIt knows this from the container_info.\n\nKnowing this, the db_state is SHARDED but we don\u0027t have a shard, this\nindicates the root container isn\u0027t responding in a timely manner. As\nsuch it probably isn\u0027t a good idea to fire questions hoping for a 301\nredirect response and overload it more. Instead let\u0027s skip the\n\"synchronous\" update part and write directly to async_pending and let\nthe object updater deal with it later.\n\nHopefully this will buy the root container some extra time to get hold\nof itself.\n\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nChange-Id: I6fcec6748c3fd0f22baac69f7ceb8261c5bb5d63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/41712d4749a8b8fc8737626b9c447eacd228e3dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/41712d4749a8b8fc8737626b9c447eacd228e3dc"}]},"branch":"refs/heads/master"},"134d1b6afbdf40de92a0dc3c559c6fde33961db4":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2025-05-20 17:01:45.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/67/919667/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/919667/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/919667/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/919667/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/919667/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/919667/14"}}},"commit":{"parents":[{"commit":"b735b3d034b11e1472f7b471919649d165c447e0","subject":"object-server: return 503 not 404 if meta or data file unlinked","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b735b3d034b11e1472f7b471919649d165c447e0"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2024-05-15 06:32:12.000000000","tz":600},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-05-20 17:01:38.000000000","tz":-420},"subject":"obj: skip to async_pending when root is busy","message":"obj: skip to async_pending when root is busy\n\nThe proxy will send the shard to update to the object server normally\nunless it failed to get that information. It would usually fail to get\nthis information when the root container server is busy and not\nresponding in time.\nThe proxy also sends through the current db_state of the root container.\nIt knows this from the container_info.\n\nKnowing this, the db_state is SHARDED but we don\u0027t have a shard, this\nindicates the root container isn\u0027t responding in a timely manner. As\nsuch it probably isn\u0027t a good idea to fire questions hoping for a 301\nredirect response and overload it more. Instead let\u0027s skip the\n\"synchronous\" update part and write directly to async_pending and let\nthe object updater deal with it later.\n\nHopefully this will buy the root container some extra time to get hold\nof itself.\n\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nChange-Id: I6fcec6748c3fd0f22baac69f7ceb8261c5bb5d63\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/134d1b6afbdf40de92a0dc3c559c6fde33961db4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/134d1b6afbdf40de92a0dc3c559c6fde33961db4"}]},"branch":"refs/heads/master"},"106202a019bd9787f2968e40474aec287515a59f":{"kind":"REWORK","_number":15,"created":"2026-02-24 08:39:39.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/67/919667/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/67/919667/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/919667/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/919667/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/67/919667/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/67/919667/15"}}},"commit":{"parents":[{"commit":"51b068309a626d2228ab69b5158ee857e96d101d","subject":"tests: more timestamp iter refactoring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/51b068309a626d2228ab69b5158ee857e96d101d"}]}],"author":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2024-05-15 06:32:12.000000000","tz":600},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2026-02-24 08:39:23.000000000","tz":660},"subject":"obj: skip to async_pending when root is busy","message":"obj: skip to async_pending when root is busy\n\nThe proxy will send the shard to update to the object server normally\nunless it failed to get that information. It would usually fail to get\nthis information when the root container server is busy and not\nresponding in time.\nThe proxy also sends through the current db_state of the root container.\nIt knows this from the container_info.\n\nKnowing this, the db_state is SHARDED but we don\u0027t have a shard, this\nindicates the root container isn\u0027t responding in a timely manner. As\nsuch it probably isn\u0027t a good idea to fire questions hoping for a 301\nredirect response and overload it more. Instead let\u0027s skip the\n\"synchronous\" update part and write directly to async_pending and let\nthe object updater deal with it later.\n\nHopefully this will buy the root container some extra time to get hold\nof itself.\n\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nChange-Id: I6fcec6748c3fd0f22baac69f7ceb8261c5bb5d63\nSigned-off-by: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/106202a019bd9787f2968e40474aec287515a59f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/106202a019bd9787f2968e40474aec287515a59f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
