)]}'
{"id":"openstack%2Fswift~919881","triplet_id":"openstack%2Fswift~master~I2a4581e408e3012a44bfb0cd58563b82040e23fc","project":"openstack/swift","branch":"master","topic":"input-checksum-mismatch","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-05-18 00:34:18.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-05-17 21:31:01.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I2a4581e408e3012a44bfb0cd58563b82040e23fc","subject":"test: more test for s3api v4 checksum","status":"MERGED","created":"2024-05-16 15:45:14.000000000","updated":"2024-05-18 00:35:34.000000000","submitted":"2024-05-18 00:34:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"919881-input-checksum-mismatch","meta_rev_id":"0758456a74966e8d6e14f3e1365d51c1957fa665","_number":919881,"virtual_id_number":919881,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-18 00:34:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2024-05-17 21:13:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-05-16 19:18:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-05-17 21:31:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-16 16:50:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-16 16:51:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-16 19:18:24.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"a43aa9195621e7aace82e67efdd7a18557b597fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-05-16 15:45:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2126df8857b177fe8197734532aa3742badc8021","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 16:50:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01f324b7a3b549ef9d2100f64b9e846a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a10e8e3da1bd4e54ae87950361470ae7 : SUCCESS in 50m 47s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/56932a8f352c41759b398542a7a09dda : SUCCESS in 42m 42s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/5a8fd987d79c4e82be98e6b929405fb4 : SUCCESS in 1h 00m 17s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/94557df49c2f4fbf8ce720b18545dbde : SUCCESS in 1h 02m 31s","accounts_in_message":[],"_revision_number":1},{"id":"fcc0d23c19f2f6ade1086e505952a333344c17ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 16:51:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f8a24f02f2e46b8afccfdc003403ee2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df9b10906a2d4f019925544ad0589832 : SUCCESS in 10m 04s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/818694f3e50240b88049180cf965095b : SUCCESS in 19m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ceb99671d7bb46d58749470bcb2620b2 : SUCCESS in 15m 59s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6aa18e9dbc704d14878331799dbaabce : SUCCESS in 20m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7c0d8c63dbfc4075bfc7aedb1fb94916 : SUCCESS in 10m 32s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/fa3a94c1bae54f2b8d598cfc2f654c23 : SUCCESS in 6m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8441242ffb124a5ea1f36fe08fbcb59a : SUCCESS in 21m 03s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/228380171ec44970afcc5e98c47e4a6c : SUCCESS in 17m 04s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0d7f95c3049b432bbcbe63712944c081 : SUCCESS in 15m 19s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/e303adf1ee9a4a23b4aa9da3eeb5aae0 : SUCCESS in 20m 10s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1a0413ec0532457d8102d33c2e2cb362 : SUCCESS in 21m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/18d3c33d00664076a7c93317d7fbe8f9 : SUCCESS in 23m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a7d32724d259470b838c9d554b6b8c95 : SUCCESS in 18m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/099b5fc0e81b4f1d90904ce0f10c75e9 : SUCCESS in 17m 32s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f59f427dbaf2446ab1c00414a79b0013 : SUCCESS in 18m 13s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/457996dc3ed6420c9d6986b02e79e625 : SUCCESS in 15m 05s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a3444d976e91457e83f57be1ffa592ea : SUCCESS in 18m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/15d19e79365445d28442caf36cc08ea2 : SUCCESS in 47m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cf69baa27f8743fab9f0a45823f8ba4d : SUCCESS in 59m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d3fabc4e6d0944cc96c598a254a2c78c : SUCCESS in 18m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42a0a4daf6e74caabd0a96cd9a8fb75f : SUCCESS in 4m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/55cd37c3cfd3455fb5690de9452da3a9 : SUCCESS in 26m 23s","accounts_in_message":[],"_revision_number":1},{"id":"fc9a1b17987592d8cb0e9b6297b8012fac90c69c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-16 19:18:24.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"009fd7a26b77237d7cc5046288b224460e881896","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-05-17 21:13:12.000000000","message":"Patch Set 1: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f8fa5d7508194d49b3e4021fc32db9e8dacd429c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-17 21:31:01.000000000","message":"Patch Set 1: Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"3431a690a65677c10e812494738929b9a4b3e411","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-17 21:31:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"67cebd8d8e5cd6e179034ed9e38f550909eeed8d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-18 00:34:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12799faa597d4e54bb20024da8182476\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b3cb972e9bb48d4b05f0762b0713e2e : SUCCESS in 8m 45s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/58bbdedc4fb34282aac25f0d07c6d9a7 : SUCCESS in 12m 30s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/f539c51063024b9c9f902d164dd2c758 : SUCCESS in 6m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e6e2723e850d45bb89e9f2d7482e611a : SUCCESS in 20m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/77d5329a59d94b7f894d91fb1792a7a0 : SUCCESS in 19m 10s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/370b074e812b4f38a451e2e09476a0fc : SUCCESS in 17m 43s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/146d95916c6f44af85a899ac943c95d1 : SUCCESS in 14m 07s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/001feb81bec3420d966b46ac0593ce14 : SUCCESS in 21m 21s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d55ba0cb0b814c8e8f9a46d490eea896 : SUCCESS in 14m 50s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7b90ad96b7094e85b711bb9ff9197c76 : SUCCESS in 21m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/feacea30843548abb1a884fdef49c5e0 : SUCCESS in 17m 41s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/cedb5d1211d640d790972ea08dc25d5c : SUCCESS in 16m 51s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/07a450df6d884506a4e356412d3ab9c5 : SUCCESS in 15m 42s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/363ddc30fcef46b2becaf8e00af25114 : SUCCESS in 19m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/430768f92ce84f5ab8362d96d46c784b : SUCCESS in 7m 28s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/19a4588845c142a980ed2c92b5ec3c7d : SUCCESS in 7m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b2a58d9e0129415ca3b4405c2228eed4 : SUCCESS in 43m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/20a1c74a08d6446fafa673dfd0f9fda8 : SUCCESS in 56m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e3fa4cd575724eb790406a0486d22d29 : SUCCESS in 21m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bc3b21402f244b78f61407ad7aac5ae : SUCCESS in 3m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/aa4ea60f5f3a4d9484cd011e53e88945 : SUCCESS in 20m 37s","accounts_in_message":[],"_revision_number":1},{"id":"56b225fda3ad28975fd526c5db9f9ba252b4198a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-18 00:34:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0758456a74966e8d6e14f3e1365d51c1957fa665","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-18 00:35:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/786e6aecedcc47bb98bf7e82beb908fb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33c595c1f9d744ab8043ed4634f6d839 : SUCCESS in 56s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/8a84641050254db0ad82862185b60d24 : SUCCESS in 43s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/446eeae5ac2146e3b4f842fc4ac6641e : SUCCESS in 33s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"893deca2743cc7590dbea90d5b0adfcf999bbc8b","revisions":{"893deca2743cc7590dbea90d5b0adfcf999bbc8b":{"kind":"REWORK","_number":1,"created":"2024-05-16 15:45:14.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/81/919881/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/81/919881/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/81/919881/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/81/919881/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/81/919881/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/81/919881/1"}}},"commit":{"parents":[{"commit":"f5e422eb7858f789066a20b7718a316ea951659f","subject":"s3api: Improve checksum-mismatch detection","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f5e422eb7858f789066a20b7718a316ea951659f"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-05-16 15:44:19.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-05-16 15:44:55.000000000","tz":-300},"subject":"test: more test for s3api v4 checksum","message":"test: more test for s3api v4 checksum\n\nRelated-Change: I9924ff3b8d7d246631fe61b916823e028e2c01f2\nChange-Id: I2a4581e408e3012a44bfb0cd58563b82040e23fc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/893deca2743cc7590dbea90d5b0adfcf999bbc8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/893deca2743cc7590dbea90d5b0adfcf999bbc8b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
