)]}'
{"id":"openstack%2Fswift~919897","triplet_id":"openstack%2Fswift~master~I17bed216f9bb7d97f6ab50431f45048220294c3d","project":"openstack/swift","branch":"master","attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2024-10-04 16:33:24.000000000","reason":"\u003cGERRIT_ACCOUNT_34892\u003e replied on the change","reason_account":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-05-17 01:41:38.000000000","reason":"\u003cGERRIT_ACCOUNT_7233\u003e replied on the change","reason_account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I17bed216f9bb7d97f6ab50431f45048220294c3d","subject":"Patch in orjson if available","status":"NEW","created":"2024-05-16 19:58:43.000000000","updated":"2024-10-04 16:33:24.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"469532fb585124e2c8e20b46619d974cbcd43a19","_number":919897,"virtual_id_number":919897,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-05-16 21:08:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-2,"date":"2024-05-16 20:02:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-16 19:59:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-17 01:41:38.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"},{"updated":"2024-10-04 16:32:52.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"CC"}],"messages":[{"id":"4a1b985eccdad20c8999872db32024a7d8cab3e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-16 19:58:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8a2cd74c93a57733612c768d3b865acc09b52a2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 19:59:30.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/swift for 919780,3","accounts_in_message":[],"_revision_number":1},{"id":"2c790d1f197475d97cf2c1669b2fa94e2d6e8fd9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 19:59:33.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/swift for 919780,3","accounts_in_message":[],"_revision_number":1},{"id":"2278703cf7906553934f5e2b0ae12334a1321778","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-16 20:00:41.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"1f4af6599ac6e53787a9d417c5bfad02ce8b513d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-16 20:02:25.000000000","message":"Patch Set 2: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e99a9c1b46ba085df58288f4089e79b6c349f9c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 21:08:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8701a86ba5e44b9b7a41ee4e6369191\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a5213d095ca44389aacbb5c7b496a9e : SUCCESS in 8m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/d1c94c6d7c1e4b699f4d4e707b8a5510 : SUCCESS in 46m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/aed7429e371442e48d138a28ffeacbad : SUCCESS in 35m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bd191e019e6442e6bd069efe6c37729e : SUCCESS in 46m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/793e5ed852f347869b98a50cc1de3738 : SUCCESS in 19m 42s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f093546f9e7c45cfa3d91616078eed71 : SUCCESS in 20m 13s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/943ecc94bb6c4f8693a9ca9091c8c44a : SUCCESS in 22m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5b5591e515b74ecd86633e6e24bb4074 : SUCCESS in 11m 49s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5f93077c79c242f1b8434675efe45604 : SUCCESS in 5m 42s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/266987f147ce46eda3b4b7794ff69dcb : SUCCESS in 20m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5bbf1dba8eb7404daf781754b8c286e0 : SUCCESS in 17m 44s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ff2ec4799c5045b684cb004788dc658b : SUCCESS in 18m 12s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/36da67e3bb70483ca67a0e214fb6acaa : SUCCESS in 18m 32s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7f886cbac85d4e0cbcafb0ee515c4198 : SUCCESS in 22m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5670796ce356449e91296d3c6260fab3 : SUCCESS in 16m 29s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/11d9075fccbb4ce797393350d922041d : SUCCESS in 17m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2a3f902db06f496495baae6fea9d0c70 : SUCCESS in 19m 23s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8856c0e5761d452c8a465f9f8f411697 : SUCCESS in 15m 10s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1017c2b531984879b03ef93992e74c2c : SUCCESS in 15m 32s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b352a1149d8c467cabc4de5e2951bee1 : SUCCESS in 22m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8f8a94bbc89e4bd68cc5a16d7b7d4296 : SUCCESS in 7m 05s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3eac0ddae2514ee490e9ca07a838f811 : SUCCESS in 17m 10s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/700d92ae0be543208badef2e1dd2fc8b : SUCCESS in 7m 29s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9aae8c1cffe74805ae4f183d1d457be0 : SUCCESS in 52m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/47e43d1f7c8a45b790293b3437e69529 : SUCCESS in 42m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7239ae6b565e4a38bee6e39b59261631 : SUCCESS in 41m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fa25b4377d434421bed80d3ba092451d : SUCCESS in 19m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/025e56380f424f27ad8c7d0747160615 : SUCCESS in 4m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/eefcd92901b545929b2998435df233ab : SUCCESS in 19m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7111924057184ac1a1cecb1b9bb5bad1 : SUCCESS in 51m 56s","accounts_in_message":[],"_revision_number":2},{"id":"31f32d07260128fea2ab4d717951fbc5016b0de9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-16 21:17:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8c2726d06914407b6c3411dbe5a992e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8963bc920de74f71b28e9b2a97cc6f67 : SUCCESS in 14m 46s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ee8c723414934fc4a21fa2b524a76938 : SUCCESS in 1h 06m 29s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/1a9cca8b40774e35bb0a8f3dee138a18 : SUCCESS in 22m 14s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b90d9b3c2d61444fa66dda9f8463dcdb : SUCCESS in 1h 15m 34s","accounts_in_message":[],"_revision_number":2},{"id":"11892ddad129e2a02daf72d2f2607363db7dd1d0","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-05-17 01:41:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"469532fb585124e2c8e20b46619d974cbcd43a19","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-10-04 16:33:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5caf9649e386e9d5e722b3d05513af45a8d0cda6","revisions":{"3f08c6f1a7dd1f160d0321f81a03aba38294c545":{"kind":"REWORK","_number":1,"created":"2024-05-16 19:58:43.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/97/919897/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/97/919897/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/919897/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/919897/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/919897/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/97/919897/1"}}},"commit":{"parents":[{"commit":"63d765e02b30739e2c6a97246dd16dc53bc73a5a","subject":"Allow StatsdClients to no-op if no host provided","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/63d765e02b30739e2c6a97246dd16dc53bc73a5a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-16 19:55:22.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-16 19:55:22.000000000","tz":-420},"subject":"Patch in orjson if available","message":"Patch in orjson if available\n\nChange-Id: I17bed216f9bb7d97f6ab50431f45048220294c3d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3f08c6f1a7dd1f160d0321f81a03aba38294c545"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3f08c6f1a7dd1f160d0321f81a03aba38294c545"}]},"branch":"refs/heads/master"},"5caf9649e386e9d5e722b3d05513af45a8d0cda6":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-05-16 20:00:41.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/97/919897/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/97/919897/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/919897/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/919897/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/97/919897/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/97/919897/2"}}},"commit":{"parents":[{"commit":"d47811fc3eceacc11c66c2028cad0ce8be41075e","subject":"Merge \"make statsd_client more explicit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d47811fc3eceacc11c66c2028cad0ce8be41075e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-16 19:55:22.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-16 19:59:18.000000000","tz":-420},"subject":"Patch in orjson if available","message":"Patch in orjson if available\n\nChange-Id: I17bed216f9bb7d97f6ab50431f45048220294c3d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5caf9649e386e9d5e722b3d05513af45a8d0cda6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5caf9649e386e9d5e722b3d05513af45a8d0cda6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Code-Review\u003dMIN"],"failing_atoms":["label:Code-Review\u003dMAX"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
