)]}'
{"id":"openstack%2Fswift~919911","triplet_id":"openstack%2Fswift~master~I4bbd28a4c8554cba742a68028ad1aa24d897146f","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"6968":{"account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"last_update":"2026-07-21 14:35:41.000000000","reason":"\u003cGERRIT_ACCOUNT_6968\u003e replied on the change","reason_account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-07-21 17:13:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4bbd28a4c8554cba742a68028ad1aa24d897146f","subject":"obj: Have handoff_delete reduce replication to handoffs","status":"MERGED","created":"2024-05-17 01:09:40.000000000","updated":"2026-07-21 17:13:29.000000000","submitted":"2026-07-21 17:13:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"919911","meta_rev_id":"f6f09e18027dfa1970bff0c4f64d162e47664e98","_number":919911,"virtual_id_number":919911,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-21 17:13:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"all":[{"value":2,"date":"2026-07-21 14:35:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"all":[{"value":1,"date":"2026-07-21 14:35:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-17 02:31:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-10 14:31:04.000000000","updated_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"reviewer":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"state":"REVIEWER"},{"updated":"2025-04-10 14:31:45.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"CC"}],"messages":[{"id":"210a970a03d0cdfb057f041ae97408d1518e46e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-17 01:09:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ecf4117a7e3db310c46c6c55fb13bd2e8d4664d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-17 02:31:59.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/094b98de12cf4191b565cb8a222b7df9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/432db9aaaeca41149ae67c279bc3a67c : SUCCESS in 7m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/1139442533f34103836a4dc611316d53 : SUCCESS in 43m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b67a1207c83745eeb621d6f917c54f20 : SUCCESS in 39m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/97f77a27db2f4f0896bf0d549961ba79 : SUCCESS in 45m 23s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3c8ee8286e984b51b9c34a73945e7ca5 : SUCCESS in 16m 51s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/96d7d85b4c754b799ea0614312e5b54f : SUCCESS in 18m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c5f336a1568944a4896f8c285980ebcd : SUCCESS in 22m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b50c2a17ea274736a1eb934e3d29239c : SUCCESS in 11m 33s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/cc24bd364bcf4b4d9039b0d9f71cb62f : SUCCESS in 4m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/047ddf64922c4498bb71bef2601edf15 : SUCCESS in 16m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/92d1aa9577f64d1ea12b8e484a85410c : SUCCESS in 14m 54s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6008e8cdce794bc1bfe67960a5926e63 : SUCCESS in 17m 54s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/913f9fa95f064c418f04101b773c16e6 : SUCCESS in 17m 42s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/61fece679bad4a3aa94796791c00278e : SUCCESS in 21m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/23e442e78ba54c3dbfe8ff98540c6ae1 : SUCCESS in 16m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/25fcda49df4b4fe496dbfc68ac3a529e : SUCCESS in 16m 26s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ee3fcb86a64e4790bdb283bb8d5e2bd8 : SUCCESS in 18m 06s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8e2ec7ec81624c479657702fda1ea20d : SUCCESS in 16m 07s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/30512b836c8d4178a7ba96f4c194e92b : SUCCESS in 15m 49s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/30782fbc7a20410680ce9be3dda13eb0 : SUCCESS in 15m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cc5a51dd7c0f43c585f30fb447aa29bd : SUCCESS in 7m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/976d7965dac54607918fa73f65b89a0f : SUCCESS in 18m 24s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2b6ba5e73e7e4a408b7dec6727353807 : SUCCESS in 9m 51s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d78ae544a8ec491c882ab3091f085e11 : SUCCESS in 1h 02m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6f7f2aae006b4f9cb5ddd945690ef870 : SUCCESS in 51m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/12e5debc7d5040ac92bd34299c649a83 : SUCCESS in 41m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/53affa5c7fdd417fa667793d06a5791c : SUCCESS in 16m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8bddd84870e849578ff1e48fdbf90331 : FAILURE in 3m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/707172c979fe438ab8d23565961e0841 : SUCCESS in 16m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/60ec988f615a4ffe81d570536e26218b : SUCCESS in 43m 38s","accounts_in_message":[],"_revision_number":1},{"id":"c65f0621d20ea585981e74abf7044282ac0db972","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-17 02:41:07.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef2b52f84b7a4bdb815ee21c10eea2f0\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/202848e7b06746fa9fc3ca1f0fcb3cfc : SUCCESS in 15m 42s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/950ba55bc2a942beac91d5c65d602e27 : SUCCESS in 14m 54s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/0a21a7eacee94a90bfcdccf413e320f4 : FAILURE in 1h 29m 52s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/cdaaddabb79045cfb35c9616191a8330 : SUCCESS in 20m 44s","accounts_in_message":[],"_revision_number":1},{"id":"43281562e09d011f048b44aa1cc072997550b8c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-17 15:10:51.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"b109f3d92e60d2f94bfeaae433d53e11fcd5c270","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-17 16:05:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25ec29ed543041019efb53dc9611156e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e865c095437547bfb69e4b0d220d128b : SUCCESS in 24m 39s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/556ac4e7a08e41eb8f54982964f8a49d : SUCCESS in 53m 10s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8e63f2bf202c4a228dee6aa776bb7946 : SUCCESS in 21m 04s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/bef3efb2d2d44c9b900012b45442e2cc : SUCCESS in 44m 10s","accounts_in_message":[],"_revision_number":2},{"id":"a4a0f72091299c77c655638a1a425afb8b0bb0c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-17 16:25:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1a1a2dcecd646878d6f8295e85c4247\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2812e521294411eaa2db1f356982e09 : SUCCESS in 8m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/9f7410b8e91d48e29ce05f8a9ad155f1 : SUCCESS in 1h 02m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f47886ca1d2b49ddbb264ba1e6209412 : SUCCESS in 46m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8d674a9a8dd443e69ad0db93a7032c0d : SUCCESS in 50m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ca2c254b1aea49eb8b23631292f998c9 : SUCCESS in 17m 16s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2c0a330ef9894023918269874a3cd58b : SUCCESS in 18m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0ab282eb7e9042f780600399e880ef4a : SUCCESS in 20m 26s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/de4af1577dfc486ca19767b25072fbff : SUCCESS in 12m 38s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/74a9789a90854e8685125718f489de8b : SUCCESS in 6m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/981aecaed9b64c2b842a98a489274968 : SUCCESS in 19m 14s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4ed62185db7c493fb01b4ccedb6d7c0b : SUCCESS in 20m 04s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b31980ff5495432ab2e64a8325ba714c : SUCCESS in 19m 06s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/b048f8c60e1949248b54d608d9c0f093 : SUCCESS in 16m 37s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/553c06ac459a4f2da18e5dacf3861354 : SUCCESS in 17m 37s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fa9927f7673a4f0fb711492bd156b8dd : SUCCESS in 18m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c07b80255c704733a7cecc118bad8caf : SUCCESS in 19m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d5561f926377464a9ffe9a9d5130f0d0 : SUCCESS in 18m 41s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c66a76b290dd4179b4681864f62bce7d : SUCCESS in 15m 57s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/82b884000f2a4326a92808ebf9d5bc26 : SUCCESS in 18m 50s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5af7614be88840f6a44c52fcbb1b59ed : SUCCESS in 20m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/308a90e50c1c4e36aa1e4b2746868efa : SUCCESS in 7m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/21a907d527314bba9e8a6b9b195f2c88 : SUCCESS in 18m 02s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0809555776fb4b7194efc75b51b88975 : SUCCESS in 7m 41s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ba1809065fc14584b4342e479f9cfab8 : SUCCESS in 1h 04m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7f3528c7f8314f25bd7f12c6539d3184 : SUCCESS in 58m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/495e9819fc8c4a0cba7e28e8a46abe87 : SUCCESS in 53m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5c074e0b8afb4af9bc834237218b55d4 : SUCCESS in 15m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1cfba96187f4bb1a75c266e09b953ce : SUCCESS in 4m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/42f8661eea6b4944aaa899ea9b7ef850 : SUCCESS in 18m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8996547677804cbca7def18ed490bb8c : SUCCESS in 50m 38s","accounts_in_message":[],"_revision_number":2},{"id":"f464e07e064a7119d814e8c34e9bcc5d111fdeff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-28 19:47:16.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"7ed0d33014a9e7181723ad567a39a778173cfd47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 20:54:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5acc91fc054241469c27f00a44102059\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3ea7fd47e9e4747845c171f7dfd7b98 : SUCCESS in 8m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/a702046eb4d14265b15f0da8c607eede : SUCCESS in 1h 04m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6d1ccea1bc7f49169bb0e0a56c50b247 : SUCCESS in 1h 01m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/886128f8d3e14442be66dffc7a77383a : SUCCESS in 59m 47s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0909405f07a741d6ad6ccebe464a3134 : SUCCESS in 26m 47s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c09d2d3c600d4efc9dc69880fbf1ddff : SUCCESS in 24m 22s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8bddd642453d4764bbe41bb1653010e6 : SUCCESS in 24m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cbe049d7f0324a2a818bf5335d6d57c7 : SUCCESS in 5m 41s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1c632a6586044b5397ee730c52154576 : SUCCESS in 24m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/223259fe61ca48159f4185bfd1034763 : SUCCESS in 21m 29s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0ef2cf81f7284b6d9998bc9395cda8b0 : SUCCESS in 22m 21s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/63643e44298b4e7dade90a810db482e9 : SUCCESS in 19m 13s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e3d965f7a6934793922fdd89a8783ae6 : SUCCESS in 27m 41s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d5f8c378300f4b2bbffc6a4d7631e3c3 : SUCCESS in 19m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/08fd9fbe75f94b6ab2452aed7b60be7e : SUCCESS in 17m 50s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9d8da16609cd4158aab78e58d4a5bb86 : SUCCESS in 19m 24s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b77c1f88f8034b7bb142164db5ac8373 : SUCCESS in 15m 33s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b12c09db5156432cbb0030802cfd2255 : SUCCESS in 15m 13s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e339e34dc1154553a3bce6823c6f2323 : SUCCESS in 20m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/eb9acf019a9f454ab61e3688096529bf : SUCCESS in 5m 40s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6dbe937a647b4f8d87c43e89858b36bb : SUCCESS in 12m 48s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2cf11b2fd1d64105bd4f43bd88ce210e : SUCCESS in 6m 29s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/48f29ffa0b8140da97e183a9a445d988 : SUCCESS in 55m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1779f7685d3843f6bf646ec28c101ebb : SUCCESS in 37m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5839a64da78140589251bd47d76a85db : SUCCESS in 48m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f3b260d5e44f42d7bcd67ed3c235ac57 : SUCCESS in 15m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bed31fe4dfff4eaa8e1d4cfc4d6600b6 : SUCCESS in 4m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6b3cad792ea6438fbf5af4e8cc166d1b : SUCCESS in 26m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cbc2b7066c0c4151b4b240d0a5e33ac6 : SUCCESS in 52m 27s","accounts_in_message":[],"_revision_number":3},{"id":"171ecc8e44276787361fb6d1b219c9c6b24cebb1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 21:39:42.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0516659c7cea46b1bccb2eec9ac4bd4b\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/665995954ebc4979a196bf201146e635 : TIMED_OUT in 1h 31m 12s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4eb25e37bfe3405c9ef696fe5fa7cc5d : TIMED_OUT in 1h 31m 08s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/2b5bab3265dd45109d5471f3158fc90d : TIMED_OUT in 1h 30m 49s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1b3290ccddbb48b88c934052b244b053 : TIMED_OUT in 1h 30m 50s","accounts_in_message":[],"_revision_number":3},{"id":"6819046938d422bce3912d303788bcc3c8f675b9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2026-07-16 20:12:00.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"206ff6a9a69760081279c44cde4b4dce9fa943fa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-16 20:13:53.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e1dd4ce3425644b4801b43b23097e564\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d70382ca80e147538286ce072d6e95ea : NODE_FAILURE Node(set) request 7aa6e4b0c69f442ebd850ba6bfbd2b22 failed in 0s","accounts_in_message":[],"_revision_number":3},{"id":"3e5e58cb29931590a4ae653d0c4431a2fa968bd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-16 21:22:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b73fd674a88464890ebb34de15f9e64\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a5eaad94b2848e0940131833b3eccb2 : SUCCESS in 12m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ac29a7e74594637ba815c62e5be8dfd : SUCCESS in 8m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/cca85a4ddb284ada953c83c77f003145 : SUCCESS in 1h 05m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/01958b690c24468c98ab0de36634d8f0 : SUCCESS in 43m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d40fce1607754265b06edf69a4453d6a : SUCCESS in 35m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dee87efc4d944389a2636ab817d63f26 : SUCCESS in 54m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b2c53cfe67b548ffa1fa83bbf2671112 : SUCCESS in 16m 42s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9f1e431b70304e13b80d253897d6970e : SUCCESS in 17m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7728c5ca6ef046ad9ff0a28ec9989757 : SUCCESS in 20m 22s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5a0b6b6eb3c345658ac9f78d88e869fb : SUCCESS in 9m 10s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c931d6ffd75b4fc1950cc19bbb1548f9 : SUCCESS in 20m 13s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/7e693abc9fa34370883148d26cb4b805 : SUCCESS in 21m 00s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/68d5aee6b4f94f169fe423f2f5de829a : SUCCESS in 11m 23s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/ca9bf091551c43f9b0ba1a4b7fca626e : SUCCESS in 19m 52s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/1876064228ee4ee0ac7419e03aa279a0 : SUCCESS in 12m 58s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ea7479e9cd344c868fa9e9d35f9a21b9 : SUCCESS in 15m 22s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6ab034a9242848bd9a77f9deded46153 : SUCCESS in 15m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f0f09fbe09164cc4bcd955e5eae75680 : SUCCESS in 6m 06s (non-voting)\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9d99d29c50914cf0b3400866b299345a : FAILURE in 4m 06s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/67a4a29ef0104273985bcc3790d5e1de : SUCCESS in 8m 53s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c654777349c84737b99fa37c4765aa44 : SUCCESS in 50m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/92acdfba721f4ada91b76027b5dd4c09 : SUCCESS in 37m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/61435a56c199431bb106d45706b54415 : SUCCESS in 53m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/72316448ac414e0eab6459dbd79a687e : SUCCESS in 22m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4052311751af4b809fb7f55e39254eba : SUCCESS in 4m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b6fa1583ae6f4d8285ec2b03b6f0ad5e : SUCCESS in 18m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b977240d381f4c6cbb46935f9f3e7c7a : SUCCESS in 1h 01m 45s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":3},{"id":"f3117a5aa06df7d34ce4162f8516819cc8d69cb8","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-07-21 14:35:41.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0a2e2076edc135b1c7c2ea8767f1f5fa0c3dce0f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 14:36:41.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"d8f7314a5000af8adf4eee9e18e8c94a66aa2ca7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 17:13:29.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10d9ad50716f4b5ba0d27653b04407b1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2abed5ba0cc34cedb371ff962cf8ce4f : SUCCESS in 10m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/051b49c041034e768e37a26ea2b0dc2e : SUCCESS in 10m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/60d039e38a2c4540808de1a7eb18369a : SUCCESS in 39m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b53457f9a2124fb4ab86c5674a5f2565 : SUCCESS in 1h 01m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e86ddfa0c5774eafbcde53498d6af4ce : SUCCESS in 44m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/49287ba079dd42c99fc1f44ba88d448d : SUCCESS in 1h 04m 20s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/2e0f99d46e2040e89f09deb29b6f613b : SUCCESS in 7m 36s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ccf9158828b4496197f39b9f6a65073f : SUCCESS in 25m 32s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/2d179eb941e449d6b16db204a4b1cfdf : SUCCESS in 23m 10s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/b29b3432a03d4d618e2d4e7ace02e1c7 : SUCCESS in 17m 25s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/18aa03235ce548789cf411010d3f9712 : SUCCESS in 10m 05s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/313d7a0b6b264fe3a349cf40560e533c : SUCCESS in 11m 35s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e8751e6505ef41d29b983d829fd96c77 : SUCCESS in 16m 43s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6df2aa0be2b74d3f8bf511f101a4eb4e : SUCCESS in 16m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/991aacf6ccbd45c4a7ad1795a18d85bb : SUCCESS in 6m 56s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/5c344d38b195421d9057a61ddcbdc5e5 : SUCCESS in 10m 00s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e6a9ed0091634014a1997fc9a07d81f3 : SUCCESS in 1h 01m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c0643d16be0644da82e3a2870452823c : SUCCESS in 24m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b879dacd2cea40ad87c05cdebb9639ec : SUCCESS in 31m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4b7f728b14a9499bad2a4d1655e56738 : SUCCESS in 30m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/471ede4f73b94355aeefdbf0c71ac407 : SUCCESS in 5m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8d76c7ca11b64976aa6205ab0f11f514 : SUCCESS in 20m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/40feeb90900d4554b3d53226d2aaf07c : SUCCESS in 28m 02s","accounts_in_message":[],"_revision_number":3},{"id":"f6f09e18027dfa1970bff0c4f64d162e47664e98","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-21 17:13:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"5e5abf6406624c62d582d1bdc61a8df86efffd97","revisions":{"037afd12b0a0982d6eb85db26f06c95347c3f7ac":{"kind":"REWORK","_number":1,"created":"2024-05-17 01:09:40.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/11/919911/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/919911/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/919911/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/919911/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/919911/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/919911/1"}}},"commit":{"parents":[{"commit":"d47811fc3eceacc11c66c2028cad0ce8be41075e","subject":"Merge \"make statsd_client more explicit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d47811fc3eceacc11c66c2028cad0ce8be41075e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-16 23:19:46.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-17 01:09:30.000000000","tz":-420},"subject":"obj: Have handoff_delete reduce replication to handoffs","message":"obj: Have handoff_delete reduce replication to handoffs\n\nSuppose you\u0027ve got a full cluster. Some time after bringing in more\nhardware and setting handoff_delete\u003d1 and handoffs_first\u003dtrue but\nbefore you\u0027re all the way out of the woods, you notice one of the\npartitions has one primary on fresh hardware and at least one primary\nwith a full disk.\n\nThe fresh-hardware primary should have few if any handoffs; as a\nresult, it goes back to syncing mode. For every 507 it gets back, it\nfinds a handoff to sync to. Those handoffs are going to try to clean\nthat up as quick as they can, and with handoff_delete\u003d1, they only\nneed to sync with the fresh-hardware primary in order to delete --\nall we\u0027ve done is waste some rsync connections and network bandwidth,\nand the fresh-hardware primary will try to do it again.\n\nNow, only be willing to sync to at most handoff_delete-1 handoffs.\nThis respects the operator\u0027s stated willingness to tolerate reduced\ndurability for the sake of faster replication and ensures that no\nhandoff should receive a bunch of data just to immediately delete\nit the next chance it gets.\n\nNote that the intentionally-create-handoffs behavior can now be\ndisabled entirely via handoff_delete \u003d 1.\n\nChange-Id: I4bbd28a4c8554cba742a68028ad1aa24d897146f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/037afd12b0a0982d6eb85db26f06c95347c3f7ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/037afd12b0a0982d6eb85db26f06c95347c3f7ac"}]},"branch":"refs/heads/master"},"c9e6a1e127192ad9b50251f767215d51f1c4f09e":{"kind":"REWORK","_number":2,"created":"2024-05-17 15:10:51.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/11/919911/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/919911/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/919911/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/919911/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/919911/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/919911/2"}}},"commit":{"parents":[{"commit":"d47811fc3eceacc11c66c2028cad0ce8be41075e","subject":"Merge \"make statsd_client more explicit\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d47811fc3eceacc11c66c2028cad0ce8be41075e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-16 23:19:46.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-17 15:10:43.000000000","tz":-420},"subject":"obj: Have handoff_delete reduce replication to handoffs","message":"obj: Have handoff_delete reduce replication to handoffs\n\nSuppose you\u0027ve got a full cluster. Some time after bringing in more\nhardware and setting handoff_delete\u003d1 and handoffs_first\u003dtrue but\nbefore you\u0027re all the way out of the woods, you notice one of the\npartitions has one primary on fresh hardware and at least one primary\nwith a full disk.\n\nThe fresh-hardware primary should have few if any handoffs; as a\nresult, it goes back to syncing mode. For every 507 it gets back, it\nfinds a handoff to sync to. Those handoffs are going to try to clean\nthat up as quick as they can, and with handoff_delete\u003d1, they only\nneed to sync with the fresh-hardware primary in order to delete --\nall we\u0027ve done is waste some rsync connections and network bandwidth,\nand the fresh-hardware primary will try to do it again.\n\nNow, only be willing to sync to at most handoff_delete-1 handoffs.\nThis respects the operator\u0027s stated willingness to tolerate reduced\ndurability for the sake of faster replication and ensures that no\nhandoff should receive a bunch of data just to immediately delete\nit the next chance it gets.\n\nNote that the intentionally-create-handoffs behavior can now be\ndisabled entirely via handoff_delete \u003d 1.\n\nChange-Id: I4bbd28a4c8554cba742a68028ad1aa24d897146f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c9e6a1e127192ad9b50251f767215d51f1c4f09e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c9e6a1e127192ad9b50251f767215d51f1c4f09e"}]},"branch":"refs/heads/master"},"5e5abf6406624c62d582d1bdc61a8df86efffd97":{"kind":"REWORK","_number":3,"created":"2024-08-28 19:47:16.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/11/919911/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/919911/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/919911/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/919911/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/919911/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/919911/3"}}},"commit":{"parents":[{"commit":"6a0153f54542da474955802ac01cf174c7ce8573","subject":"Merge \"docs: Simplify reload-process-tree SVGs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6a0153f54542da474955802ac01cf174c7ce8573"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-16 23:19:46.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-28 19:46:04.000000000","tz":-420},"subject":"obj: Have handoff_delete reduce replication to handoffs","message":"obj: Have handoff_delete reduce replication to handoffs\n\nSuppose you\u0027ve got a full cluster. Some time after bringing in more\nhardware and setting handoff_delete\u003d1 and handoffs_first\u003dtrue but\nbefore you\u0027re all the way out of the woods, you notice one of the\npartitions has one primary on fresh hardware and at least one primary\nwith a full disk.\n\nThe fresh-hardware primary should have few if any handoffs; as a\nresult, it goes back to syncing mode. For every 507 it gets back, it\nfinds a handoff to sync to. Those handoffs are going to try to clean\nthat up as quick as they can, and with handoff_delete\u003d1, they only\nneed to sync with the fresh-hardware primary in order to delete --\nall we\u0027ve done is waste some rsync connections and network bandwidth,\nand the fresh-hardware primary will try to do it again.\n\nNow, only be willing to sync to at most handoff_delete-1 handoffs.\nThis respects the operator\u0027s stated willingness to tolerate reduced\ndurability for the sake of faster replication and ensures that no\nhandoff should receive a bunch of data just to immediately delete\nit the next chance it gets.\n\nNote that the intentionally-create-handoffs behavior can now be\ndisabled entirely via handoff_delete \u003d 1.\n\nChange-Id: I4bbd28a4c8554cba742a68028ad1aa24d897146f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5e5abf6406624c62d582d1bdc61a8df86efffd97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5e5abf6406624c62d582d1bdc61a8df86efffd97"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
