)]}'
{"id":"openstack%2Fswift~919936","triplet_id":"openstack%2Fswift~master~I68848cc8489bef430637eb9e42e7f7a77175de1a","project":"openstack/swift","branch":"master","topic":"noop-stats","hashtags":[],"change_id":"I68848cc8489bef430637eb9e42e7f7a77175de1a","subject":"sq? test StatsdClient directly","status":"ABANDONED","created":"2024-05-17 10:05:28.000000000","updated":"2024-05-17 17:16:19.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a83c8d6e6cc35c2fae7487b0a088bb4dcf7b30d5","_number":919936,"virtual_id_number":919936,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-05-17 11:39:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-17 11:31:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-17 11:39:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a0dcbcb68f189c54b917083c3562013c783d5f0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-05-17 10:05:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"92671f4b7b9102a0461ab83cd903a2a632935428","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-05-17 10:30:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b1659053249ca3345a606579e7e81716dc7a7d8f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-17 11:31:55.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78fd72d892054d2f9ef00863c04ea071\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c4ca49fa3e414c9f84b439b676f8abc6 : SUCCESS in 15m 12s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/21c2d3c1b3634e64a45529d2af5bd6db : SUCCESS in 49m 40s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/de7744875fa04e8ba49586918f2eb281 : SUCCESS in 22m 10s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/eb73b1bceb9e46ee814b96461cca78a1 : SUCCESS in 59m 21s","accounts_in_message":[],"_revision_number":2},{"id":"a94a5e5385931fae841f6a64ca9f0e1ade984008","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-17 11:39:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97f70875938e458a861ef933815b44cc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6a13e8d661243d3969fdab917724898 : SUCCESS in 8m 51s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e2f8fb2ca0974615968f2c7ab3e2c01d : SUCCESS in 19m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e36d4cf9a6424163bd41215ffc714275 : SUCCESS in 19m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0be7794be38044238398c466a2d98824 : SUCCESS in 19m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cfee0d776b4a416e824b56e7c496740f : SUCCESS in 10m 46s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a869339e69e04ad4ade9f3e0aabdf138 : SUCCESS in 5m 31s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/38b5789938304c06ba19fa8825761641 : SUCCESS in 15m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b6cadb5fd77a4de8b43f231356915697 : SUCCESS in 18m 43s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a7b4dfc8bc864e9ea47700d5926b2e72 : SUCCESS in 12m 03s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/e1bf4a79ef63400ebbfcbec6928b6236 : SUCCESS in 12m 11s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3ee80a686ef74504820afda0a058817b : SUCCESS in 22m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/58f8e5e36cfd460895ac80a971079cc8 : SUCCESS in 17m 42s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9e6adfe8a1554537bb3a990529337c43 : SUCCESS in 19m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/dd55fced820c4623ade696cb61c394e0 : SUCCESS in 19m 12s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7671f3d2b0a04791baa6081b3ee581b4 : SUCCESS in 15m 13s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/06206552c252436494cde2e37f31c76d : SUCCESS in 17m 37s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e943598b0c3f41a599fac0ec02e93edf : SUCCESS in 14m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d4397654f0cf4121af91632359fec53d : SUCCESS in 1h 01m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/527e0d3d57694aa1bb9796ef26b39822 : SUCCESS in 46m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f4bde1aff6c941ec8cf3ef4b19c717ad : SUCCESS in 14m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d1eaffd870148f6b6e6dc9eb1ed4304 : SUCCESS in 4m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6c1ca87c38b74e2ab2ccd2e0defb2dcd : SUCCESS in 18m 12s","accounts_in_message":[],"_revision_number":2},{"id":"a83c8d6e6cc35c2fae7487b0a088bb4dcf7b30d5","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-17 17:16:19.000000000","message":"Abandoned\n\nSquashed into parent","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2a6363c57a90c04cd3ff19aea780e4462b25dc56","revisions":{"5c63b02fdb31ac9b7ee7b4eb3923a6d71cdddfcb":{"kind":"REWORK","_number":1,"created":"2024-05-17 10:05:28.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/36/919936/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/36/919936/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/919936/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/919936/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/919936/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/36/919936/1"}}},"commit":{"parents":[{"commit":"f8d6eaade58afb4e2da69314030e34c45675e402","subject":"Allow StatsdClients to no-op if no host provided","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f8d6eaade58afb4e2da69314030e34c45675e402"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-05-17 10:03:28.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-05-17 10:03:28.000000000","tz":60},"subject":"sq? test StatsdClient directly","message":"sq? test StatsdClient directly\n\nUnit tests currently cover StatsdClient indirectly via\nget_logger(). Start to add tests that actually call StatsdClient().\n\nChange-Id: I68848cc8489bef430637eb9e42e7f7a77175de1a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5c63b02fdb31ac9b7ee7b4eb3923a6d71cdddfcb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5c63b02fdb31ac9b7ee7b4eb3923a6d71cdddfcb"}]},"branch":"refs/heads/master"},"2a6363c57a90c04cd3ff19aea780e4462b25dc56":{"kind":"REWORK","_number":2,"created":"2024-05-17 10:30:54.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/36/919936/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/36/919936/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/919936/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/919936/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/36/919936/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/36/919936/2"}}},"commit":{"parents":[{"commit":"f8d6eaade58afb4e2da69314030e34c45675e402","subject":"Allow StatsdClients to no-op if no host provided","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f8d6eaade58afb4e2da69314030e34c45675e402"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-05-17 10:03:28.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-05-17 10:30:43.000000000","tz":60},"subject":"sq? test StatsdClient directly","message":"sq? test StatsdClient directly\n\nUnit tests currently cover StatsdClient indirectly via\nget_logger(). Start to add tests that actually call StatsdClient().\n\nChange-Id: I68848cc8489bef430637eb9e42e7f7a77175de1a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2a6363c57a90c04cd3ff19aea780e4462b25dc56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2a6363c57a90c04cd3ff19aea780e4462b25dc56"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
