)]}'
{"id":"openstack%2Fswift~920063","triplet_id":"openstack%2Fswift~master~Id3afe301d0c143080e9a8d79caabf8abef6d4ea8","project":"openstack/swift","branch":"master","topic":"db_state_in_async","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-05-28 02:07:55.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Id3afe301d0c143080e9a8d79caabf8abef6d4ea8","subject":"sq? move stuff around","status":"ABANDONED","created":"2024-05-20 19:43:54.000000000","updated":"2024-05-28 02:07:55.000000000","total_comment_count":5,"unresolved_comment_count":4,"has_review_started":true,"meta_rev_id":"414994e8480243ed6a918653417e80341a0f4daa","_number":920063,"virtual_id_number":920063,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-05-20 20:46:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-20 20:46:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-21 06:59:36.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"}],"messages":[{"id":"a1d1f4be0e9a0009550be7f1094468c83d9c6ea2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-05-20 19:43:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"62b90c240edf917e2d3791eb4a29ed16f8501f33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 20:46:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f2339203de34474a992abbbe9e7a947\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f4f58e13d024894ad187be4718c4571 : SUCCESS in 8m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/bd0fb0f42de64afbae41ed4740678404 : SUCCESS in 50m 45s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/529d6e866460423095db17275fadf5e7 : SUCCESS in 38m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9a8043c2ab9745d1b0013b75f9ed02a6 : SUCCESS in 53m 24s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d919b744d84a46a7887fe0bda518e127 : SUCCESS in 21m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/70f385af6e5c4280bd91b830ca8cfd3a : SUCCESS in 19m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/38ca535c748245a3b4f62c516f4d30c7 : SUCCESS in 17m 34s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/13402682231548acb5a43cc1aa0bd615 : SUCCESS in 9m 15s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/34b73cfba1db4d55be8a5c200e448806 : SUCCESS in 5m 08s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/be5a62d466f44ff2808347c4697945ce : SUCCESS in 15m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4b913669c93f4e9b82085da07a1dcc47 : SUCCESS in 14m 15s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6facba4455914970bdeab0b6003a6744 : SUCCESS in 26m 14s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/bddad6e3f4a44acb94f36ae309593323 : SUCCESS in 13m 29s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a57248d413b244378f271deef9f5cc11 : SUCCESS in 27m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d28d12fdfd944e4a9464faba6b27060a : SUCCESS in 16m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3533a5276bde4aff8696125086706c38 : SUCCESS in 17m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0616efa59e214f43b90cb83eb978d476 : SUCCESS in 21m 15s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ed90741c174d435db6af89e902ced401 : SUCCESS in 17m 13s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/98e68e1e79fc4f8ab28a0f252bf1d7dc : SUCCESS in 18m 36s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d30732d9645c46b69e77f5b0f4083c79 : SUCCESS in 20m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/779fd67f158547bfa6f22795a62d3978 : SUCCESS in 6m 00s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c5393121a75a4ae2bfa611e0ad883349 : SUCCESS in 18m 00s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d37f8a6ebb2b4778a2662bcea53e80b9 : SUCCESS in 7m 06s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8608ef31ed1a443284497f5066e889cc : SUCCESS in 57m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/32a82ec505984ce0b9b1ea8f06077994 : SUCCESS in 48m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9a4d5a9b9c0f4d0aaf4b7dd6b3611856 : SUCCESS in 49m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b122f3e5be8f460a9e83c8f9c4afda10 : SUCCESS in 20m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/680b7f4749da4971ab1f53450fb63399 : SUCCESS in 5m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8fee10ea74e84ba286a9e59158138f3d : SUCCESS in 19m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8ebfbb70611a42908fd26c70e8cbc334 : SUCCESS in 56m 14s","accounts_in_message":[],"_revision_number":1},{"id":"be7346d20234da5457fa385fa273b737b5052fe5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 21:10:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3a1b4b3e0984e81a3adceae57ee094e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ca77760e7d21443f9fb3dd10c918c9e8 : SUCCESS in 14m 47s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a515f4984e374fd4a23af2bf81406cbb : SUCCESS in 17m 40s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a1ffc551c7054b7ab282eba7728c09da : SUCCESS in 1h 25m 04s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/c9f2e307c1134f18a3c2a5a2d9910c16 : SUCCESS in 22m 10s","accounts_in_message":[],"_revision_number":1},{"id":"1a355404d66bf6e3b0e054b1fd8af836330e7166","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-05-21 06:59:36.000000000","message":"Patch Set 1:\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"414994e8480243ed6a918653417e80341a0f4daa","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-05-28 02:07:55.000000000","message":"Abandoned\n\nsquashed into parent","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0f97594f8bdcb292aac311906bb09ea6b6b92db7","revisions":{"0f97594f8bdcb292aac311906bb09ea6b6b92db7":{"kind":"REWORK","_number":1,"created":"2024-05-20 19:43:54.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/63/920063/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/63/920063/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/920063/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/920063/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/920063/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/63/920063/1"}}},"commit":{"parents":[{"commit":"67f7400931a0e1804cb4663a0f7aaed8dfae1e9c","subject":"obj: skip to async_pending when root is busy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/67f7400931a0e1804cb4663a0f7aaed8dfae1e9c"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-05-20 19:42:24.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-05-20 19:43:25.000000000","tz":-300},"subject":"sq? move stuff around","message":"sq? move stuff around\n\nRelated-Change: I6fcec6748c3fd0f22baac69f7ceb8261c5bb5d63\nChange-Id: Id3afe301d0c143080e9a8d79caabf8abef6d4ea8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0f97594f8bdcb292aac311906bb09ea6b6b92db7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0f97594f8bdcb292aac311906bb09ea6b6b92db7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
