)]}'
{"id":"openstack%2Fswift~920206","triplet_id":"openstack%2Fswift~master~I06c2a16126886bab8807d704294b9809844be086","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2024-09-04 18:11:48.000000000","reason":"Change was submitted"},"597":{"account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"last_update":"2024-09-04 18:11:48.000000000","reason":"Change was submitted"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-09-04 12:45:28.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-09-04 18:11:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I06c2a16126886bab8807d704294b9809844be086","subject":"s3api: Clean up some errors","status":"MERGED","created":"2024-05-22 19:12:38.000000000","updated":"2024-09-04 18:15:51.000000000","submitted":"2024-09-04 18:11:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":99,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"920206","meta_rev_id":"dc2823882a101ef15990ffa9c6d949a776354653","_number":920206,"virtual_id_number":920206,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-04 18:11:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":2,"date":"2024-09-04 12:45:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":1,"date":"2024-09-04 16:45:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}],"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-22 19:35:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-22 19:50:15.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"CC"},{"updated":"2024-05-22 20:17:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-31 07:29:28.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"CC"},{"updated":"2024-06-03 05:50:14.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2024-07-25 16:15:06.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"CC"},{"updated":"2024-08-13 16:21:45.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2024-09-03 13:48:42.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"37448fb56d8d1700cb5e79fc9a3d27d998c7af59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-22 19:12:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"03df072e7d562f3608a5fd412293142c76db199b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-22 19:35:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a21c1fa046e4dcb8630ca2b9bd489e8\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/03bd6837df214583a018b065fa2a3884 : SUCCESS in 19m 20s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4672887dd9c146d38ceb849bed5b1a14 : SUCCESS in 14m 35s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/bc554f9d939a413499b1ff4d4730a1d8 : SUCCESS in 21m 38s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1d9b13b9c5094e1f9e95beef8c6aa5fa : SUCCESS in 18m 58s","accounts_in_message":[],"_revision_number":1},{"id":"aa5a7126c1c2a883d2a35f7ccf3edf1c27ee8658","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2024-05-22 19:50:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d14cd2ae3fe4d12110a93b8c8ec1a33fa4fbbad5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-22 20:17:01.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/33f33ed57d1c49668e49bf8a3e102daf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b0e1925912e49c6a070d35397d76b09 : SUCCESS in 10m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/3f521671d1374d4aa86edf2c56febb84 : SUCCESS in 52m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6f99c6bcca1c48c7ad86e6cd9000b2a4 : SUCCESS in 45m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6b0a2f803d4048d38ad03f7bc2410a34 : SUCCESS in 54m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/09aba901d1534ded898761f0e8c52a5d : SUCCESS in 19m 13s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2e1e8b9f4b4649f6b8d147b2077c7603 : SUCCESS in 19m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3eae0ec409b441d28db2d0fdc2d8cffc : SUCCESS in 20m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5ccc118836c24864a04b9579ae479bae : SUCCESS in 9m 30s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d2555f59c7994c43a46b6718d7f0710a : SUCCESS in 6m 07s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4880a0fac4ce472da91602bbfb6bc398 : SUCCESS in 14m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0f2169acb9274981a50b167eacf57885 : SUCCESS in 17m 04s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a7b71a61e0e046edb26352d6956fc2bc : SUCCESS in 13m 04s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/92ca92c3e1d24d38a2b4c99440bfe43f : SUCCESS in 16m 56s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ddccfbcf5da04fc1a21c5af82282b2a8 : SUCCESS in 22m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/982cfb25135b4dc68529fcafed266e06 : SUCCESS in 14m 52s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0793555e9bc14e49a56a6e221a05d398 : SUCCESS in 19m 40s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/60c453c0bd004c93909178248b611184 : SUCCESS in 19m 54s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d9d6ee75940b425dae00318c43635c05 : SUCCESS in 12m 19s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/4b0938f5558d44af940cbb50041668a9 : SUCCESS in 17m 08s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/bebde2521935484ca4a95061c3e8fc9b : SUCCESS in 14m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/324b243c39b04b9f8182aa6ae3b3338b : SUCCESS in 5m 24s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/26db67a446404a19ac93078d9793766f : SUCCESS in 14m 05s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/dfc08c9598984380ac1241eff451ac98 : SUCCESS in 7m 37s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9ec6be89cd0a4350b29254934d8ae21e : SUCCESS in 51m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cc140cc21f1340128b63df7d0ab3497e : SUCCESS in 47m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0168a49a6f0b4e1daeef575db2bc4909 : SUCCESS in 37m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f69b227a141349949df451a6d805b9c2 : SUCCESS in 21m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d816801e585d4a13b6165599b66d5513 : FAILURE in 4m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1ac82987040f41f3a99f67f3fe078c75 : SUCCESS in 19m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4c94919139b148ce82648add66a8d800 : SUCCESS in 59m 36s","accounts_in_message":[],"_revision_number":1},{"id":"a25096adc6133ff5c0187d47fef5a188f9441fde","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-22 21:32:21.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"c71d30217b82f6a0417ae976b68946874ab0aa69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-28 22:36:50.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ccab4cbdd44c01e2a63adce40791bf662ffa1e47","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 23:06:30.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3c3c7bb06fa467998ee14d21e774875\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3eb1795da18843d1899cab5ecf6ea6c7 : SUCCESS in 17m 49s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/df64983b392649b09fdb365c5c6e35af : SUCCESS in 18m 40s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/10ebb8979ccf4f1399c4427ae69d35ec : SUCCESS in 27m 23s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e63bfa3679f641d8ad63e4da4427d54f : SUCCESS in 26m 57s","accounts_in_message":[],"_revision_number":2},{"id":"23d7916c457e62f4ec63d57f1cda86844380ce22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 23:49:51.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fe94ae85fc924a53adb6d529119295b0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54893951765e4d86a7039cbf1c6b13a5 : SUCCESS in 8m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/9f4462f37bad47f585c3d12703329c73 : SUCCESS in 53m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1c0d9416c7c040c0982db767bfe0aec8 : SUCCESS in 57m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9327efbf754044c2a606bad2a4255a18 : SUCCESS in 48m 53s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1676c77530cc46faa6f52a4a00de9792 : SUCCESS in 18m 52s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/da0187672bad4274bf9d0f7a440c253e : SUCCESS in 18m 15s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/14f00c3cd3ba4ced8d7eb2cd00f899f8 : SUCCESS in 19m 25s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e7ea9e1254654238a5481e107b533eb7 : SUCCESS in 10m 16s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/658154b03ca64fe5a0baddadf3f7a54b : SUCCESS in 4m 54s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8c2e9d96175c4d638b253a5a91a97d90 : SUCCESS in 19m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/982208cf18574e2fafd1f95367778f88 : SUCCESS in 19m 09s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d514c1f74dce4ce98ba610b024b38889 : SUCCESS in 15m 53s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/8e3204bbe89f47919a42e2d3eb7caec6 : SUCCESS in 20m 14s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4094dbf694474a4b9bbeaca41390bd1e : SUCCESS in 15m 08s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4b8c99ec94b64b13adfbb39ed54bfd50 : SUCCESS in 18m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/682681d19a074979ba4fc7646fc23bf4 : SUCCESS in 20m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3b88befa90564d21a3086f1d6da748ed : SUCCESS in 20m 44s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1558c0a0fad741019cd245532e015365 : SUCCESS in 12m 22s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7ecb13da11644917b3a99b57df03e9b7 : SUCCESS in 15m 11s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/712c8b8ea6d3483aabc7ced5d311361a : SUCCESS in 16m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/63351fa59d4746648f0c20ed0b522226 : SUCCESS in 5m 59s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ef585dd94c6c46db91a9fe92079c9d2b : SUCCESS in 14m 04s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/25564302b05b4e2489654e8e2bd239f1 : SUCCESS in 8m 01s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/80e54bc170c749e6ab0490cd50527622 : SUCCESS in 58m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e4a0dfb7f7de404f8b12c5a6c23309e4 : SUCCESS in 34m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c201b2ef25b94642932360ea6cef6fcf : SUCCESS in 49m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/702fe9c3f91742d4b2fad6fa7b28f774 : SUCCESS in 20m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed12e3ea9ea945548d04feeaacb7eac4 : FAILURE in 4m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a2d57ad2a9d142b1bc5a2c4b639a1535 : SUCCESS in 26m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cb251aa5d8164291b98b104c49926668 : SUCCESS in 1h 02m 18s","accounts_in_message":[],"_revision_number":2},{"id":"a3ab21ed4c8432b6ead4b79593c4ce0a767cea3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-29 21:03:41.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"d237dda122e6b85a397eed602be02c2bd085b4fc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-29 21:12:11.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c818598f6b0c43fb8de1c03735f2c6ac\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/51cab101c50845aabbe4290d90c3535c : RETRY_LIMIT in 1m 34s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/73fdaca4d4b346ebacb290bd73ea5e89 : RETRY_LIMIT in 1m 08s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3ddf029984e54f37bac19bbacdb60bdc : RETRY_LIMIT in 1m 24s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f57d40a369bb4efb97e560ee951560c5 : RETRY_LIMIT in 1m 26s","accounts_in_message":[],"_revision_number":3},{"id":"e00061506cc675f2446ccc65c99cd9cbaf0b4f36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-29 22:04:47.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a2a5cb86d25d4af083dbb1df4ec84e53\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/545ab6b4265643ff9021e8686a1c9e3c : RETRY_LIMIT in 1m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/0ef03474b41740d58c531a699de98075 : RETRY_LIMIT in 2m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ae5aae4cd444483aa6f54fcc16579064 : RETRY_LIMIT in 1m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/90ba24408369438ab7073d881209728d : RETRY_LIMIT in 1m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0f34582dff6a43cd98f50b5df5fba54f : SUCCESS in 18m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ad3b3e4a228542dbaaa52ec02ce45b2b : SUCCESS in 19m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3bc1bd0829544091a7d6e7da685669d0 : SUCCESS in 21m 04s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/de36e9a752a0442ab4c18f85e0c9a085 : RETRY_LIMIT in 1m 33s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5245735effa44085b9315d2751650830 : RETRY_LIMIT in 1m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ca3df6547cba48a3a6243b706d4f03d1 : SUCCESS in 18m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f6bc07589aee48b6b26ed266a11768ac : SUCCESS in 19m 11s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/10d1f6e51d3d4eca9e00a0610700c09a : SUCCESS in 18m 14s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/9709ba01138246d5909a8a34f5bc0ccd : RETRY_LIMIT in 1m 35s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/52538ce15f09475091be8880b930d18f : SUCCESS in 15m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/445986c7a7344fed98462a03607e8281 : SUCCESS in 19m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c3d1f66fb86b4e0e89d773c0f9dde653 : SUCCESS in 19m 23s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/65b12c4bffd24c4882031942b58515f7 : SUCCESS in 22m 29s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9b470dc046e1492d839af0627ef3ad13 : SUCCESS in 15m 36s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e9171e2191304ff9bec6f307b9a714c7 : SUCCESS in 16m 31s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/861d2c5c817d47ec8c9d0aa0d31d5df6 : SUCCESS in 17m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/068393ced9914107a9065824a59d40db : SUCCESS in 7m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/16810310ef354e6a9d376d2e18af214f : SUCCESS in 18m 50s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8d8e2aa0cc734c4b8d82538a0173409a : SUCCESS in 9m 51s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f8e7ee9b46e541c9bbbb73bbcaf8435c : SUCCESS in 53m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/551a190894a643d7a7d67d2bcb6dd565 : RETRY_LIMIT in 1m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/dc935866d0454e569e53c0dbd155f7c2 : RETRY_LIMIT in 1m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ddd6ccaa9ca34b518b43cdbbadac3b7f : RETRY_LIMIT in 1m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89ee66ea047b4cd5937476ed4fec2bcf : RETRY_LIMIT in 1m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/389e1d624fd94241acf6aa376bc487b9 : SUCCESS in 19m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/56b7a0564f7647b1a8b473ea85cf3986 : RETRY_LIMIT in 1m 20s","accounts_in_message":[],"_revision_number":3},{"id":"81959f6fd09177003c8c992420fc38299fb661a5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-05-30 16:11:31.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e7c0b21ecd91aeb170dca4348c4d35f838c895bc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-30 16:19:54.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd92499a3e3c4fa2986521391a040551\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/18e769f66c8741afad77a4c8ad4fd2c9 : RETRY_LIMIT in 1m 10s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e20f0f6172ab4691a3b3666979b35af1 : RETRY_LIMIT in 1m 27s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d391fa7dd82a4bfdbb3c2df836352015 : RETRY_LIMIT in 1m 17s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1ec57b5eeaad48e3845630bf9ced1253 : RETRY_LIMIT in 1m 04s","accounts_in_message":[],"_revision_number":3},{"id":"b9af438c037d76ab414f575278c5ca5fc35eebd8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-30 17:25:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a5824b104fe4dbeb782d6538002aafa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cbdedbd030444e2850b75657e6277ba : SUCCESS in 8m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/6ccb5525bf164ee49628d130b62f814d : SUCCESS in 50m 40s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a11514939775461392965ca3d15fd29b : SUCCESS in 39m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/968de32b92a444c3aff504abc655ec81 : SUCCESS in 54m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d9cd68bd748f46b98d961e04508e349b : SUCCESS in 15m 16s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b7d44d7376df40adb2be5a8049596c9b : SUCCESS in 19m 36s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6738c4b019f1403b8fae6dd01c6c825a : SUCCESS in 20m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c87d19f35453424997732df43e662894 : SUCCESS in 10m 29s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0b5abab210244c6191431208569380dc : SUCCESS in 7m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/da2c5864b8d4420aba552656835b5537 : SUCCESS in 20m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/db878fa407b247d5b4e0de65da0f8e30 : SUCCESS in 18m 06s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0fc89933cfd94652b0dfea294be3ec8e : SUCCESS in 17m 13s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/88fb263d4728450d872f7ca78877e367 : SUCCESS in 17m 42s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/73c28727eccb40a6a275d89f486f3fd5 : SUCCESS in 22m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c0e23e431a4e43fb80e947e9d85a512a : SUCCESS in 18m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/296a0f2f5fb8496c9bec592f615e6dc0 : SUCCESS in 21m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8ea28f8718b2442fb1fe05d5f1dfd4e8 : SUCCESS in 19m 46s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b79cd38710444377848e962bff4f3571 : SUCCESS in 15m 20s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/684dd1e5427b474b82896f1b1cdf8469 : SUCCESS in 14m 38s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d56f9662fac94677b64dd2da38359db4 : SUCCESS in 15m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4045fc0ff13e44b3936c21c45eef4f42 : SUCCESS in 8m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0028ac3f6bad4eaa9f680cc4ae8314bb : SUCCESS in 15m 47s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/780f8fe5178940df9b958e7f85c09a5d : SUCCESS in 8m 43s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a6aca819b7814659a776cc974b4cec4f : SUCCESS in 48m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/de9cd1565ca146a2928d68b094adbfa6 : SUCCESS in 37m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d799a9d497264a1d967792caeb04f0e6 : SUCCESS in 41m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8231951a548143aa8a0d447a4ef0f5de : SUCCESS in 18m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c571379a5e5b4c28b74c4ce271e39d34 : SUCCESS in 4m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5e3c2e4c64b044b4b367d24d75aada2a : SUCCESS in 21m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4d2cf35922274f538c6ce5465d137cd7 : SUCCESS in 1h 03m 51s","accounts_in_message":[],"_revision_number":3},{"id":"8c486fef4177be5aed623f50280f4734cff4bd67","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-05-31 07:29:28.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"da55c4909517902d58977a726e4692e4174d4f93","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-06-03 05:50:14.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b7537c16b2055891f0cd9b4ed0c498534aa9c0c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-07-02 00:25:00.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"9c5562ded8297d1622a7e3c28ba75ad3ed84cf20","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 00:49:43.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b559224ee0ed4defaa221f69400017b6\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/60c0034d086a4fedb4bb7b42ac3ef57b : SUCCESS in 18m 35s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/70735489f8b743e899ce9102deea35d7 : SUCCESS in 19m 34s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e996d21babc94e12ba74dfa17267d362 : SUCCESS in 22m 23s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d2fe96a2be724e7c975b4e0b9d128fad : SUCCESS in 22m 55s","accounts_in_message":[],"_revision_number":4},{"id":"d4edecf96b89e723d00cb667ecd509063d18c3eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 01:39:12.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/79efc2b478b34e1db76bed869091580c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3423b42a843e49b28c5c766a24a7c3cb : SUCCESS in 9m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/f4228516875b401c9f308a01d414335c : SUCCESS in 1h 02m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0a01eeba68fa4ad780d67a843538e63d : SUCCESS in 56m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ea390abe8ca84f479e6b332e17c1985f : SUCCESS in 1h 03m 58s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e2cde0aa8d8343a38eae06289fe3be75 : SUCCESS in 19m 34s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3d1aa3159ce34a50aefa3fbfb45d46fb : SUCCESS in 19m 24s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c11c38e1af8245a191d8ccdf823f6108 : SUCCESS in 20m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a5de09628f96451fa98c0c9afe1da526 : SUCCESS in 5m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9cc665ce337e41868d03e3055bd68042 : SUCCESS in 21m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e44085d03908415fa8d795945d8372ad : SUCCESS in 18m 52s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d47b0e506efc469bb1072826ad8d21ce : SUCCESS in 17m 46s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/79c7c83ba17b4ff2927be03f19ef4451 : SUCCESS in 13m 44s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/cfac8c2d467046b6bb33046f32c66d2c : SUCCESS in 20m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e49f3e20c33249329b08ccdab2f1715e : SUCCESS in 18m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/eed4a7152a6b42db97636e9e85482c3d : SUCCESS in 15m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/125ddacde1444bfe8f52785cf8afbfef : SUCCESS in 19m 08s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/56ca1c4f7b514caab8740c2a5968075f : SUCCESS in 16m 51s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/3438112cb36647079a43c5a5f131b716 : SUCCESS in 17m 45s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5f1b6c6f12b5445381497a0523632f38 : SUCCESS in 18m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d11b3808ab464855a8bd76e6eaa1ebcc : SUCCESS in 6m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dc5ee971a06b4bbf9c78713d73c45029 : SUCCESS in 15m 36s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3a42ad76f1fa424c8b9c3a1546f65951 : SUCCESS in 6m 39s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6a0b32f842c04f729a3b5181ad907665 : SUCCESS in 58m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7b0ec97b99f04b7fa66420aed8e0dc33 : SUCCESS in 50m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a92a453aedd74fe2acabd7c2100f6a43 : SUCCESS in 47m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/56d6a3ad8e36474f936344459e493ad4 : SUCCESS in 16m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69942cf0037342aba455bebca988ff05 : SUCCESS in 5m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c5cc9995e89449aa9c390ee1aa83cf81 : SUCCESS in 18m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/be9e5a68d0f148cd88389120602caaaf : SUCCESS in 1h 01m 55s","accounts_in_message":[],"_revision_number":4},{"id":"33b58853c104ede1cb75af7cafddb2df14b939b2","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-08-13 16:21:45.000000000","message":"Patch Set 4:\n\n(9 comments)","accounts_in_message":[],"_revision_number":4},{"id":"4252a7abd0af69d57c37d17b14f9f916099c8539","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-13 21:54:53.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"73f2616fceb4a4678a7ebdb2eb623f595f38e0bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-16 00:11:36.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"fb4e6c2db82166f9f12cdeb78cc4d48be43abb24","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-16 00:18:15.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"25202b8f8e35417ac8601492dfd58a8a3feaf552","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-16 01:15:55.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/39f2647c11a246b69910f859862e34fc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb65cf58b3724c118be746b9dbf82df7 : SUCCESS in 9m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/e2f66d8d2fce47f3b50bb176a1e759d2 : SUCCESS in 1h 03m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0423fbbc8ee34efbaf9760dcd15bfee5 : SUCCESS in 57m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/27479d7ecccc434bbdad0b8cc7fb1821 : SUCCESS in 45m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/af2e530afce54459a3cdafda7c7926f0 : SUCCESS in 18m 56s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4b4e6e5199454578a7b2bda323c744f8 : SUCCESS in 20m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c709901aae6240ebbb0953b8a5cd395c : SUCCESS in 20m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/aebe0e9cf23c483191e8b8a95fcbb646 : SUCCESS in 5m 56s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/41b99f2793ce47a195096fa300c94484 : SUCCESS in 20m 21s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2d24d9b488314c1891d93639212802f7 : SUCCESS in 17m 42s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/874018683abe450983c9f1078a593d48 : SUCCESS in 20m 16s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/134a9956f89246e99cc0e39409c3d8e6 : SUCCESS in 22m 00s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/60f4b98e97c248a5b72eed1ab00bf300 : SUCCESS in 27m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5e8c422685b04773af20ac438426d57f : SUCCESS in 18m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6b9316aa604944c3b3671089aa097243 : SUCCESS in 21m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6b141a7b79bc4e8fa2fb7de7fc1c1d4b : SUCCESS in 19m 21s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a28aa3e4502c457f922d90d4f4f92fcc : SUCCESS in 19m 27s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/dbaa7254d58e400798d696012ec5b91d : SUCCESS in 16m 55s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b32c36b93f0648af950ead2e285089fc : SUCCESS in 20m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5ce97cb917164bd7a8fcc5bdb6b1fd65 : SUCCESS in 6m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e4c02c4dc05343c68d3c6709332c4f9e : SUCCESS in 16m 28s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5c51f757e8b444a0a88d2751e0539859 : SUCCESS in 5m 56s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/07b6edec54a24e88b38330d66fd4de69 : SUCCESS in 47m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9d1b6546025d41f9ae269f29f8113450 : SUCCESS in 45m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/59dc956b41f046e98653d321d478cd81 : SUCCESS in 45m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3858734a7c35451993615dca29f150c2 : SUCCESS in 18m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f5dee1bb9e245ea9020583482bb7d13 : SUCCESS in 4m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bd9607ab2fe54928bb6a2c83e925956c : SUCCESS in 17m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3dbe422550c54365bd0857cef68f1ea5 : SUCCESS in 1h 01m 57s","accounts_in_message":[],"_revision_number":5},{"id":"ce2f87aebc7843bc8618c15c5b0eb656da72df7c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-16 01:23:28.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b88f49f4b6c24c8997ac41670d57651a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/baba42f211b44ed68ac8eaebed5e4e2e : SUCCESS in 1h 03m 03s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/eda98f445143414ba1d17a3572e8a9e4 : SUCCESS in 1h 04m 23s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/7908e69be0564b6f8991741c21909942 : SUCCESS in 1h 08m 53s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6525768cb8534dbca38df14f0c398d1f : FAILURE in 1h 09m 36s","accounts_in_message":[],"_revision_number":5},{"id":"945f7758675565c28d9dc619e73554e1ca5415c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-22 00:06:36.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"f6a87352e413d990710282bbe38945edb7bdd0f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 01:12:23.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5061a19f83040bfa7108ee3e9828020\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd08234c6d2a4ceb8128d18ad4a4d753 : SUCCESS in 8m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/07d9287ec5fc4a9485f8ad10a12e4f7d : SUCCESS in 44m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/69078be56a8d442bbc17f830be5bc56a : SUCCESS in 38m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6eb6f206f7d44fba96f5c250bba65197 : SUCCESS in 47m 54s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b676135f4b894702909b6f68dfe252a9 : SUCCESS in 21m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dec61a9c05b74f09acbb1c25a5dbbf3b : SUCCESS in 21m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2a599139b63c46d79e8ff5df9837e9e7 : SUCCESS in 23m 43s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/af34d0bf78804bc681fbf966ec9dc01b : RETRY_LIMIT in 2m 07s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0dc892187975456ea681a048b1a8d49f : SUCCESS in 21m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b2eb15df1bc04121a7a26961757c37d1 : SUCCESS in 20m 58s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/bc48a90b1a96488a8f9c373a8b6079e3 : SUCCESS in 20m 02s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/2bdde85d4b834314bd5a959ce443bbb4 : SUCCESS in 21m 19s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/28bff89632d54cff9cdd0520e19fd5c1 : SUCCESS in 28m 27s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/77df44bffc2c4887b187ad57b2a7f41e : SUCCESS in 20m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/64b6b0d11358449d80f0bca6e1337ed9 : SUCCESS in 18m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/719e21086ec14c7ba62bc4f7fee58e7d : SUCCESS in 21m 14s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c1191dc0b824451b8f2075704ef3c839 : SUCCESS in 15m 44s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/fc8d036236724816b04be959eadd2d37 : SUCCESS in 17m 46s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4853cee7e561429f8bac90e8d9ba678e : SUCCESS in 17m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/29ce5639a4b6432db340b3484287db14 : SUCCESS in 7m 30s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3885425804b247f59441844da39cd870 : SUCCESS in 14m 53s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/07ebabc8ae53426b8a62c62245697627 : SUCCESS in 6m 49s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1ea3eff4d3ab42ccbccbb3cf0c4fe08a : SUCCESS in 49m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6e9816de478340eb88491519fbbcf919 : SUCCESS in 54m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/51ffdf52c6c843a68db629051e1374c4 : SUCCESS in 51m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/82968fc11ba2454c811136d0c7b66777 : SUCCESS in 17m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/583b9b13419b471196bf1c142e8e473c : SUCCESS in 4m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2518c07ee40f4a6980ed7e46ecb75ae4 : SUCCESS in 16m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6264b46c1e4049c693f1b47e382f8e85 : SUCCESS in 1h 00m 01s","accounts_in_message":[],"_revision_number":6},{"id":"8c68a5079ae333e1f2cdcd0200e8ea82db0919c6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-22 03:20:22.000000000","message":"Patch Set 6:\n\n(6 comments)","accounts_in_message":[],"_revision_number":6},{"id":"38649b3722574fcc08a30a9c34ed890294125420","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 03:53:08.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d990fd663d748b08b495f979cb6b3ae\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/89e0aec34d9c4ab2990926b9f401c727 : TIMED_OUT in 1h 31m 03s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bc4f6e548c6d42ca8f2f6fbe5578fe09 : SUCCESS in 1h 06m 42s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/653caa579c46434ba835495c764644bf : FAILURE in 1h 20m 08s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/80d4b9e56bca4b53b874f392a96b4505 : TIMED_OUT in 1h 31m 00s","accounts_in_message":[],"_revision_number":6},{"id":"768adcef6ac641b3d7ef225f81d156928cc76511","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-22 22:20:50.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"e9c4bcc5af99b3cee6229de016b5d44e29136afd","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-22 22:21:15.000000000","message":"Patch Set 7:\n\n(6 comments)","accounts_in_message":[],"_revision_number":7},{"id":"dd197f60d7110c6567c685137caf471a5495e13d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 23:27:19.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c98fd23e1c849a18a896af50e1f2d33\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e41ffe7e1fa14959a204237cec44fe59 : SUCCESS in 8m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/3790a797bf5b42128a7cb3e1a0cc5b85 : SUCCESS in 1h 00m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/084c44fe308243c7a16dc38332dc74e4 : SUCCESS in 37m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fb902b44315f4c4ba95ce5bba84f694a : SUCCESS in 46m 19s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f6050f9bccfa4ab59ab378d62887d6b9 : SUCCESS in 23m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/aa17495fd5dd4cc9be204b58bfc5c80f : SUCCESS in 24m 05s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c2c20c5361f4442e8d4990201ad89e1b : SUCCESS in 22m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/87622fabfcd044e68929a24c08ba62e1 : SUCCESS in 5m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cc36c2ee6132476c8800245fe2711183 : SUCCESS in 20m 54s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1ff377426d144a03bac9f2b86e7954b1 : SUCCESS in 19m 45s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/4523a2ed33834d54afba4f2ca321f466 : SUCCESS in 23m 14s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/e0eac99ef30b4638b0a938e18c6ae344 : SUCCESS in 16m 09s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8383dd305dc24d59b4733c79b3655e59 : SUCCESS in 36m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b98921e58c2143ee8ceb731c21e3a6d5 : SUCCESS in 18m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a88f9347c13c4d86a8a4f16d91289c3c : SUCCESS in 18m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3fe48364f99d440a972b49fa8a12595b : SUCCESS in 17m 48s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6ec0efd04d354c18a34dd460739bb81f : SUCCESS in 17m 32s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/49864f86d93c4578b3fd8f2f2e8d19b9 : SUCCESS in 17m 57s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2be293add99d41fca652ccaded70413b : SUCCESS in 16m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/feac33be09e342768de2a5660ce80a81 : SUCCESS in 6m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/470910a7e2154be490e135c5340caa2d : SUCCESS in 14m 40s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1f40c7b79e0d48b7b5de617cc3349b24 : SUCCESS in 7m 40s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/43d62476d06e4055908a8a92174535e4 : SUCCESS in 52m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f850b9b2ba034bbab0a059f7629edc52 : SUCCESS in 52m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e731ecfd69ea4264849710360276d18f : SUCCESS in 47m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d16b1586d9ea46c9b085f9300dfef4bc : SUCCESS in 19m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c15cab8ddc404fc2818518f5d5aa5d1b : SUCCESS in 4m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/00cd2c7c70eb4460b339241f831add05 : SUCCESS in 18m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6f5c54054bac4aefa09350f3548dc664 : SUCCESS in 1h 03m 00s","accounts_in_message":[],"_revision_number":7},{"id":"467447c2d47b92bc892867c38ec3ca08f9727a97","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 10:53:00.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4512aeab9c934529ad3c5bf87f7c4df1\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/49327734e30d4ff9a12bb73fec9cee9c : TIMED_OUT in 1h 31m 03s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/634e85579bcc4f3c87eea6c7cabef15e : FAILURE in 1h 16m 42s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/0275f8481f4d4beaa2b95db089a38979 : TIMED_OUT in 1h 31m 36s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e14f4321184e4c3aacb2c59e54d22f2c : TIMED_OUT in 1h 31m 15s","accounts_in_message":[],"_revision_number":7},{"id":"791f2220cbf6efb778893adf673f9e629d3f5e48","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-08-23 17:28:30.000000000","message":"Patch Set 7:\n\n(17 comments)","accounts_in_message":[],"_revision_number":7},{"id":"cf54c3b2ac8bffa4c14d9438c403ad9c993e2360","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-23 18:17:36.000000000","message":"Patch Set 7:\n\n(5 comments)","accounts_in_message":[],"_revision_number":7},{"id":"20cc9f5469e0a1b35175e4d2921c67f5c3f4bb72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-23 19:22:17.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"a1148723fde88d21632348bb9c347eaa50b1c02c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-23 19:51:15.000000000","message":"Patch Set 8: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":8},{"id":"1cabb5d59f20222a67850c0f4be3031f0b9afedb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-23 20:26:05.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5cafaa6100be40baa65bf55b7047a3bb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2741ab6af26a4b0285fd50f8abcc1781 : SUCCESS in 7m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/2a68f40890b54ac5ab2291a22ff024b1 : SUCCESS in 46m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/296a3db8132641c7bbf6f6cce6f3ca89 : SUCCESS in 39m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d96f01014b514ed586cf00c9fd3c7ba3 : SUCCESS in 47m 11s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/788822bcbd2b4713b994c1f9edd2d23a : SUCCESS in 20m 49s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/70c41789725e4ed3950c741198569b06 : SUCCESS in 21m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/075fb35ad8a54490a8f6564f8a600ed0 : SUCCESS in 21m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9b06573407564990a178b0d7724affa3 : SUCCESS in 5m 58s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/479b28c620ed4f2c9b1f971a45d3898f : SUCCESS in 19m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1d117b06afd24d209011a7acae0ffd8a : SUCCESS in 20m 20s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/385e80455da4494b83483dcaff3f2717 : SUCCESS in 18m 32s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/2f03eeb9aa704ae4b967e06f25bc5415 : SUCCESS in 19m 07s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/44d58126be6e4d20beb73aedc4c34fdc : SUCCESS in 18m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2e37ad49736f4639b86b995947413507 : SUCCESS in 17m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/84660212b2674d1ba70b9c95216ef221 : SUCCESS in 18m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/790b4f83d3eb498f8ffe41d75eec4ae9 : SUCCESS in 20m 51s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/aba8a151d6314b188bd591c06bafa3e3 : SUCCESS in 16m 14s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/fb852f571af846d7820f3aee3d31a3bb : SUCCESS in 17m 15s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/da8b49be62c641a7951e617ce7dfcbe6 : SUCCESS in 18m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/905d46cdd5df48668a6455f22bcba889 : SUCCESS in 6m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1cbb06909abd4364995d938c68cba38f : SUCCESS in 15m 26s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9bb7d787886d419a95ba129bd1dfad76 : SUCCESS in 6m 07s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/63c3c0eaa54a4dc6a76e545dfdf24061 : SUCCESS in 49m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/79358b81c1bb49eb82c46a6817946068 : SUCCESS in 44m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/216e2569c106455aa4fd286cc1b73f4c : SUCCESS in 44m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/313b3eb67d2b4eadb28ef0b53c09c4eb : SUCCESS in 19m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5474aef168244e84a8499bb322b086a8 : SUCCESS in 4m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/49094d602bf148c9808eb8ace13bedb2 : SUCCESS in 17m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9f21371e7c5b45079d4ac6acff903028 : SUCCESS in 56m 00s","accounts_in_message":[],"_revision_number":8},{"id":"30ed26a031c64efd6bf39b5222572c288017b5b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-23 23:42:54.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"edeaa4777faced04268788960329048eaeb920c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-24 00:54:28.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33a8549d21794ff1bc2135b4e5e5cc0a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e482eee17ea4cf0934e6d0da46b33f7 : SUCCESS in 8m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/2e5b4aba05014323a9075d293e050662 : SUCCESS in 1h 00m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/01a03952dbf443338822e1e92f7fbd34 : SUCCESS in 54m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4b2c22c018b044f497241aa2505d294e : SUCCESS in 1h 04m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/121dae1f591c4249a3877bbc8f518357 : SUCCESS in 19m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9eb99eda9f124f69859ca4e057801fc0 : SUCCESS in 21m 55s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a926ca3c2b064e0fbc859044addc3e16 : SUCCESS in 18m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/154a315578954893ae8b7c676e4b7c72 : SUCCESS in 7m 54s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d4917ed81043442dabade32c483b59be : SUCCESS in 20m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b6c4e57f83144b23843f1af23f1c0b77 : SUCCESS in 20m 38s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/53d92e63ac34434397fa8d46f2f26953 : SUCCESS in 21m 22s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/e1ba8557508740dab093676a8a95aa94 : SUCCESS in 22m 03s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/fd86783818e64f41b004e32573663d38 : SUCCESS in 21m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/82f2c0cc340a407982540aa2835357d5 : SUCCESS in 17m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/04e4007e1f82493999bbcdc36d900d43 : SUCCESS in 19m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/91cbb3da44754907928e3d043624e56e : SUCCESS in 18m 16s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6f6a02ac63fe4a97822ca22744817710 : SUCCESS in 18m 36s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c3074dbf69fd4761b2ff54d9cee85e18 : SUCCESS in 16m 18s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d2a10887e01e4d7999de40051a9e92c5 : SUCCESS in 19m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f16eff4bf9f94924899a390e54315f25 : SUCCESS in 6m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ff73455fa9b34116958a9a406e797c87 : SUCCESS in 19m 47s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8ed5dacd79cb414eaf5ecc1c7cdbc9b7 : SUCCESS in 6m 40s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7ba0f39d45254b91b9806d84853870b5 : SUCCESS in 50m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3e549addec22407994c71c77d299574e : SUCCESS in 48m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ba6158575e4d4474b0bf0293ce94ad33 : SUCCESS in 52m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7efecd8f3bb84ee99a0d327c7a2a2aff : SUCCESS in 20m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bc0e392f65c45d9962a4aa7e893f7c3 : SUCCESS in 4m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1ab7b3dd716240d1a7d883623400d9f3 : SUCCESS in 18m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0baaca11acdc428a89072b2adbe0595b : SUCCESS in 41m 17s","accounts_in_message":[],"_revision_number":9},{"id":"5c669ccf8b2736bb9a0a4a8e19b80de8a05ca1ff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-24 12:27:55.000000000","message":"Patch Set 9:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d833873bd81042ac9250b496a3f4cad1\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/af0fada314b5407aa27880356ceb89ea : SUCCESS in 1h 17m 19s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fbd32352bd7d450789fbedb7245f34ac : TIMED_OUT in 1h 30m 53s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/6fab4c86b8ca4e0090fff4ff861c6168 : SUCCESS in 1h 27m 19s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/bbc020b72fa0435bb50cfca8f35d7b79 : TIMED_OUT in 1h 31m 03s","accounts_in_message":[],"_revision_number":9},{"id":"9a4d1483aeca7ca5900a4080bd9c5f0047ed34fa","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-09-03 13:48:42.000000000","message":"Patch Set 9: Code-Review+1\n\n(20 comments)","accounts_in_message":[],"_revision_number":9},{"id":"35c86e5538340edcc00a03fdb6194ba82df5afe2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-03 16:00:47.000000000","message":"Patch Set 9:\n\n(4 comments)","accounts_in_message":[],"_revision_number":9},{"id":"1e1f64a4e75f05c811baaca9fe15df7ed2c1ec64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-03 19:24:48.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"0fe21196cb919b2ecc4e3b6f6fce9bd2430ca0be","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 20:42:02.000000000","message":"Patch Set 10: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a8b6f7b2c15849c188e327b8f01d031f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae375dcd7bfe422997b34e34173eb353 : SUCCESS in 8m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/afcd7ac29cde47cd8e9ba5156b92214e : SUCCESS in 1h 00m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a107963ecb2b474ab8fad71aad2d731d : SUCCESS in 1h 02m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/faafca27b85e45f984364be4e4686414 : SUCCESS in 53m 27s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/52216bb35dfb4ba9871eb2f794711687 : SUCCESS in 19m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5f11d2b690bd4162ab17c6f003a4ad32 : SUCCESS in 18m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/adb3ae64ae394e38aacb2a08c19ecc84 : SUCCESS in 20m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/459481d09f274f96a5d32fbbcf231e07 : SUCCESS in 8m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d26046224c3444e6ab3ef84c3357e570 : FAILURE in 22m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/287482ef8696408ab84e2862c98d9695 : FAILURE in 20m 30s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0b0540b4eb5f47b59772ef1536e98b8b : FAILURE in 19m 57s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/919fb6d3080e42e28e82ca542726dc7e : FAILURE in 21m 37s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1e53b30ee3414e38a4093479c5a0ba0d : FAILURE in 26m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3b7fd09873304b698ace3bec898e254e : SUCCESS in 19m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c201a13b674a48019e2f4902adac73dc : SUCCESS in 22m 17s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/d75c064fa11c4e27b17561493d13d499 : SUCCESS in 22m 05s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9f9e0a11dc5f471d9d9d0d736021adea : SUCCESS in 15m 16s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e4e14886982248eeb4bbcccf1c2ddd4d : SUCCESS in 18m 15s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ad31dceda0eb41bfb9180f1841586e6f : SUCCESS in 17m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3b54db0f5a2c4f169a1c2c2bd84e8eb2 : SUCCESS in 8m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fd567614ff95427aaa8b5c8f466d78a7 : SUCCESS in 14m 04s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8f879c462a3c4a3bafa91d306a37b706 : FAILURE in 6m 56s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7dfa0694ba554046a6f06d799f399aec : SUCCESS in 49m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/53530e023f294f95a9ae8b7517de98b1 : SUCCESS in 45m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8568f0ac8a6f49f89198f9359e6ad002 : SUCCESS in 46m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6ccd62a314494666ad74baefb23bcc26 : FAILURE in 20m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac5960e08a2c42e4b2cf7161c349e24a : FAILURE in 4m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/360d3a06bdfb423098c6752a539164d1 : SUCCESS in 16m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a6e5c08edb304000b08e0e8669eb366d : SUCCESS in 47m 34s","accounts_in_message":[],"_revision_number":10},{"id":"b4c2d63af2a86b0cbe003f9612f7aac33bf759c2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 20:58:44.000000000","message":"Patch Set 10:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4390c7fa0ad42baaee073599507b2f0\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/914b92c37eb2450fa26bed6f18442ae4 : FAILURE in 1h 16m 33s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/88c3522a92504f34bc9274539b1b58bc : FAILURE in 1h 09m 28s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/749835eb558f468d91cbe2dba41ad800 : TIMED_OUT in 1h 30m 54s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8421bf5b12a1457f9a9f5659c8a037f7 : TIMED_OUT in 1h 31m 11s","accounts_in_message":[],"_revision_number":10},{"id":"c27625b6a5d3eb1c371d25053a5149deeb91e4ac","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-03 22:55:11.000000000","message":"Patch Set 10:\n\n(8 comments)","accounts_in_message":[],"_revision_number":10},{"id":"7150d140649bd80863919978ca7f6e16ab6d32ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-04 00:27:19.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"bd5665856dabbee8cea69bfe8eeaa1a8c4752dd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 01:32:52.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c473f7fb1c1044a8836db2ac0527e5eb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29f4c9c709a0439ebc8b7a5106edcb74 : SUCCESS in 8m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/e08e82e7411c4253bfa5dc8892d6a8e0 : SUCCESS in 1h 00m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6772ac46f4ee42a58cfbaa18248d0ae2 : SUCCESS in 51m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/365563a51ede48e18ee7e2bff881685f : SUCCESS in 59m 55s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/35029e21778847ee980075ab78cee92d : SUCCESS in 19m 16s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0c267f96a5614a2fb7cdc65f1264e39d : SUCCESS in 19m 56s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c38b272274f24e55a0cc373def4930f9 : SUCCESS in 22m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d1e846f4f2ce4aca8e22758e5069b4b8 : SUCCESS in 5m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/2a6b07ebaa464ecf973edb969a03f7ef : SUCCESS in 19m 40s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/27717d26bf524bc0a063bfbf3ec5f0e8 : SUCCESS in 20m 25s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/109cd08fb8814910b0250dc2c3ac26b3 : SUCCESS in 21m 24s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/d07108e8709d476f8353d5026d7e12ae : SUCCESS in 16m 38s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/447d60a34d50488b8d0cffb6e6a039f7 : SUCCESS in 20m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f13c131a1714437f9b46a80c85a80b1d : SUCCESS in 17m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/881a5bb6b4a446ca85b28b0f71c10c1e : SUCCESS in 16m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b918e21defbe42f384b5924bf6d6209f : SUCCESS in 18m 05s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e3da648e216b4a798571563d8c2f8b22 : SUCCESS in 16m 01s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/24a9dd4e2b174b1a9ef7640b55dee1fe : SUCCESS in 16m 55s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/db9d1fcfd85e44a6b78655edbc40540c : SUCCESS in 21m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fe9b8cc58cff4c35a00562e69a000e36 : SUCCESS in 7m 00s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/aa08bc4b212a4ec9bfd8f1b1d8709eb9 : SUCCESS in 14m 55s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8744fed9b6db49748b18cc7c644fc501 : SUCCESS in 7m 30s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/73f0a615424a4b19b1217ace16e30984 : SUCCESS in 48m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5f00ac1c71b94f08b3c7defade88a3f0 : SUCCESS in 42m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9f44bc7a7ba64768bd2f32379c4a5b74 : SUCCESS in 50m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1fe82c7c49f64e4194c06d724c91ce7d : SUCCESS in 21m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3afb7742e484fe9a70b43c08019888a : SUCCESS in 4m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/17be0a0a63914c8b9b529045df33b7fb : SUCCESS in 18m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f9bdf5a34445453e8ff8557c3d77a049 : SUCCESS in 43m 12s","accounts_in_message":[],"_revision_number":11},{"id":"102e3da74d58443bcd7edc5c48f0899b8ed8eedd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 01:54:05.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9313d12d91c6444ea10a05d4ed4a70a5\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8b9ac6d92b794e1f8c386084bc4823d1 : SUCCESS in 1h 25m 01s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/43387307bc19419fa3e999a1de49eb96 : SUCCESS in 1h 08m 58s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/5a64e75258604ab686918a0a9a935a83 : SUCCESS in 57m 00s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7301ce218bdb450fa5889639d6c888e1 : SUCCESS in 1h 02m 57s","accounts_in_message":[],"_revision_number":11},{"id":"4507afc6fe4223969df37dd27d0aa7e9a81a611b","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-09-04 12:45:28.000000000","message":"Patch Set 11: Code-Review+2\n\n(6 comments)","accounts_in_message":[],"_revision_number":11},{"id":"272873d0e12229a52e05f77f7ede85cb92ae66dc","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-09-04 14:21:41.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"ba4a3fded9136ef89f62aeccf60a3f72fcd37f38","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-09-04 16:45:32.000000000","message":"Patch Set 11: Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"6a039273c3330ecc78908fb10ab1ad31c10ed8f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 16:46:22.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"3d413f7e2d169b772d540fa702d091c15212aef0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 18:11:48.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/839ac223677b49d09225504270cf2565\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f353df16a7c547518fccf963cf6a7fe1 : SUCCESS in 8m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/8f580143af824b5ba0f55ad9b4b5ac25 : SUCCESS in 1h 02m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e94bcb3d5c8e423e95bde9aa618e86f8 : SUCCESS in 47m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7b4f5cf5f166482b9969d24301a4a729 : SUCCESS in 56m 54s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/e67054b6a6cd49a9a3452702cf64655d : SUCCESS in 9m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/fb6cbe30078b48a49a002f58a509db71 : SUCCESS in 20m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/dccc5702150f40b38d3841c0a66e234c : SUCCESS in 16m 46s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/7e507baca0634f78b05ef1fc75accb4c : SUCCESS in 21m 24s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/2989d465f5ba49bfa5ba79b5ad6d299b : SUCCESS in 27m 26s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/243314ecd3644c6fbb40c39a46998738 : SUCCESS in 33m 27s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9a33f55e4bd2442e84bcb53a5fd1761c : SUCCESS in 18m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/174c3eb7b2ec4722b63d0bea39a44f43 : SUCCESS in 19m 12s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7913848b9e7a481986db5e18ddf12199 : SUCCESS in 19m 11s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b09603c214954d8983b38af34bc05b65 : SUCCESS in 17m 36s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/15955237bcb44e58ae2a32bc09dabec7 : SUCCESS in 17m 39s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1d83980fadae412eae841ddd7355c298 : SUCCESS in 19m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/557f4b3308cb4e618846089e29d93c45 : SUCCESS in 5m 58s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/042453c128ed4a62905a8dcfa6a9c332 : SUCCESS in 6m 25s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/fa68650d3544470e9420dcd2fa082b22 : SUCCESS in 46m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/17f2e583054048f1b66d93c7c9c312e1 : SUCCESS in 44m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/890a02c7a27e4bcb91358e7fe54fb672 : SUCCESS in 52m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3135eeeb61204d6cab9b213e9a5ec164 : SUCCESS in 18m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2746a2651184af196ced27bda894149 : SUCCESS in 4m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/35bfb053ab39406dbb6a3a64eb28ec62 : SUCCESS in 18m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/37f47991093c422195eaa405cdca385c : SUCCESS in 47m 16s","accounts_in_message":[],"_revision_number":11},{"id":"23b065fe7094e05e40442efc98fc84d956fe85ca","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 18:11:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":11},{"id":"dc2823882a101ef15990ffa9c6d949a776354653","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 18:15:51.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/edb6163c2cf040f5abf4dcef7e60de33\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1988037cfd8e48f4abcf7491e192bcb6 : SUCCESS in 1m 12s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/fe57404636434e27a28b6d46654508fa : SUCCESS in 39s (non-voting)","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"7bf279779995be5081e3ff9da7e113f20efe4446","revisions":{"1c5b8ab21dfb85957615b40821cb4d3343735a7a":{"kind":"REWORK","_number":1,"created":"2024-05-22 19:12:38.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/06/920206/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/920206/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/920206/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/920206/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/920206/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/920206/1"}}},"commit":{"parents":[{"commit":"337079f21f66c4f288c5ffa736a177f446472e5e","subject":"Merge \"replicator: Ensure handoffs can clear with large handoff_delete\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/337079f21f66c4f288c5ffa736a177f446472e5e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-22 18:21:01.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-22 19:11:41.000000000","tz":-420},"subject":"s3api: Clean up some sha256 validation/mismatch errors","message":"s3api: Clean up some sha256 validation/mismatch errors\n\nChange-Id: I06c2a16126886bab8807d704294b9809844be086\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1c5b8ab21dfb85957615b40821cb4d3343735a7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1c5b8ab21dfb85957615b40821cb4d3343735a7a"}]},"branch":"refs/heads/master"},"ad420f552a306140ad479db7b5a6eb6e1911a84e":{"kind":"REWORK","_number":2,"created":"2024-05-28 22:36:50.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/06/920206/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/920206/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/920206/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/920206/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/920206/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/920206/2"}}},"commit":{"parents":[{"commit":"82debd2cda0a3f2727150be37e001daee857d292","subject":"Merge \"docs: Add tempurl/formpost metrics\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/82debd2cda0a3f2727150be37e001daee857d292"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-22 18:21:01.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-28 21:06:19.000000000","tz":-420},"subject":"s3api: Clean up some sha256 validation/mismatch errors","message":"s3api: Clean up some sha256 validation/mismatch errors\n\n- SHA256 mismatches should trip XAmzContentSHA256Mismatch errors,\n  not BadDigest. This should include ClientComputedContentSHA256 and\n  S3ComputedContentSHA256 elements.\n- Requests with a v4 authorization header perform a\n  looks-like-a-valid-sha256 check, rejecting with InvalidArgument\n  on failure.\n- v2-signed requests can still raise XAmzContentSHA256Mismatch errors\n  (though they *don\u0027t* do the looks-like-a-valid-sha256 check).\n\nChange-Id: I06c2a16126886bab8807d704294b9809844be086\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ad420f552a306140ad479db7b5a6eb6e1911a84e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ad420f552a306140ad479db7b5a6eb6e1911a84e"}]},"branch":"refs/heads/master"},"fa8eab42ce7b87f0a9d43d3512d51a92df7325f2":{"kind":"REWORK","_number":3,"created":"2024-05-29 21:03:41.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/06/920206/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/920206/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/920206/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/920206/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/920206/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/920206/3"}}},"commit":{"parents":[{"commit":"82debd2cda0a3f2727150be37e001daee857d292","subject":"Merge \"docs: Add tempurl/formpost metrics\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/82debd2cda0a3f2727150be37e001daee857d292"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-22 18:21:01.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-29 17:10:27.000000000","tz":-420},"subject":"s3api: Clean up some sha256 validation/mismatch errors","message":"s3api: Clean up some sha256 validation/mismatch errors\n\n- SHA256 mismatches should trip XAmzContentSHA256Mismatch errors,\n  not BadDigest. This should include ClientComputedContentSHA256 and\n  S3ComputedContentSHA256 elements.\n- Requests with a v4 authorization header perform a\n  looks-like-a-valid-sha256 check, rejecting with InvalidArgument\n  on failure.\n- v2-signed requests can still raise XAmzContentSHA256Mismatch errors\n  (though they *don\u0027t* do the looks-like-a-valid-sha256 check).\n\nChange-Id: I06c2a16126886bab8807d704294b9809844be086\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fa8eab42ce7b87f0a9d43d3512d51a92df7325f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fa8eab42ce7b87f0a9d43d3512d51a92df7325f2"}]},"branch":"refs/heads/master"},"53e5615eb8f616c463ac08db48e6182c9cff9eb1":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-07-02 00:25:00.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/06/920206/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/920206/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/920206/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/920206/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/920206/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/920206/4"}}},"commit":{"parents":[{"commit":"68adc5bff358cde91544c63f5ae21277cba3cc29","subject":"CI: up-rev a few py2 constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/68adc5bff358cde91544c63f5ae21277cba3cc29"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-22 18:21:01.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-07-02 00:24:47.000000000","tz":-420},"subject":"s3api: Clean up some sha256 validation/mismatch errors","message":"s3api: Clean up some sha256 validation/mismatch errors\n\n- SHA256 mismatches should trip XAmzContentSHA256Mismatch errors,\n  not BadDigest. This should include ClientComputedContentSHA256 and\n  S3ComputedContentSHA256 elements.\n- Requests with a v4 authorization header perform a\n  looks-like-a-valid-sha256 check, rejecting with InvalidArgument\n  on failure.\n- v2-signed requests can still raise XAmzContentSHA256Mismatch errors\n  (though they *don\u0027t* do the looks-like-a-valid-sha256 check).\n\nChange-Id: I06c2a16126886bab8807d704294b9809844be086\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/53e5615eb8f616c463ac08db48e6182c9cff9eb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/53e5615eb8f616c463ac08db48e6182c9cff9eb1"}]},"branch":"refs/heads/master"},"f6cd1cf49de4d19eef3fd2b71088abaf75968a6f":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2024-08-16 00:11:36.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/06/920206/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/920206/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/920206/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/920206/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/920206/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/920206/5"}}},"commit":{"parents":[{"commit":"e6b73612d1dcd0bace2361338e6ae82eb591282b","subject":"FakeStatsdClient: Stop issuing DNS calls for host.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e6b73612d1dcd0bace2361338e6ae82eb591282b"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-22 18:21:01.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-16 00:09:40.000000000","tz":-420},"subject":"s3api: Clean up some sha256 validation/mismatch errors","message":"s3api: Clean up some sha256 validation/mismatch errors\n\n- SHA256 mismatches should trip XAmzContentSHA256Mismatch errors,\n  not BadDigest. This should include ClientComputedContentSHA256 and\n  S3ComputedContentSHA256 elements.\n- Requests with a v4 authorization header perform a\n  looks-like-a-valid-sha256 check, rejecting with InvalidArgument\n  on failure.\n- v2-signed requests can still raise XAmzContentSHA256Mismatch errors\n  (though they *don\u0027t* do the looks-like-a-valid-sha256 check).\n\nChange-Id: I06c2a16126886bab8807d704294b9809844be086\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f6cd1cf49de4d19eef3fd2b71088abaf75968a6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f6cd1cf49de4d19eef3fd2b71088abaf75968a6f"}]},"branch":"refs/heads/master"},"49f18d7bc13e176e3716cd219aa9239b64bd3852":{"kind":"REWORK","_number":6,"created":"2024-08-22 00:06:36.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/06/920206/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/920206/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/920206/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/920206/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/920206/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/920206/6"}}},"commit":{"parents":[{"commit":"e6b73612d1dcd0bace2361338e6ae82eb591282b","subject":"FakeStatsdClient: Stop issuing DNS calls for host.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e6b73612d1dcd0bace2361338e6ae82eb591282b"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-22 18:21:01.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-21 22:50:55.000000000","tz":-420},"subject":"s3api: Clean up some sha256 validation/mismatch errors","message":"s3api: Clean up some sha256 validation/mismatch errors\n\n- SHA256 mismatches should trip XAmzContentSHA256Mismatch errors,\n  not BadDigest. This should include ClientComputedContentSHA256 and\n  S3ComputedContentSHA256 elements.\n- Requests with a v4 authorization header perform a\n  looks-like-a-valid-sha256 check, rejecting with InvalidArgument\n  on failure.\n- v2-signed requests can still raise XAmzContentSHA256Mismatch errors\n  (though they *don\u0027t* do the looks-like-a-valid-sha256 check).\n\nChange-Id: I06c2a16126886bab8807d704294b9809844be086\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/49f18d7bc13e176e3716cd219aa9239b64bd3852"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/49f18d7bc13e176e3716cd219aa9239b64bd3852"}]},"branch":"refs/heads/master"},"de0dcd8f697d1cbd6399498b76143559a8dc9953":{"kind":"REWORK","_number":7,"created":"2024-08-22 22:20:50.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/06/920206/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/920206/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/920206/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/920206/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/920206/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/920206/7"}}},"commit":{"parents":[{"commit":"d830703a32d16d4618a30684da92b78bf48deff6","subject":"Setup proxy object GET and HEAD tests correctly","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d830703a32d16d4618a30684da92b78bf48deff6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-22 18:21:01.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-22 22:20:35.000000000","tz":-420},"subject":"s3api: Clean up some sha256 validation/mismatch errors","message":"s3api: Clean up some sha256 validation/mismatch errors\n\n- SHA256 mismatches should trip XAmzContentSHA256Mismatch errors,\n  not BadDigest. This should include ClientComputedContentSHA256 and\n  S3ComputedContentSHA256 elements.\n- Requests with a v4 authorization header perform a\n  looks-like-a-valid-sha256 check, rejecting with InvalidArgument\n  on failure.\n- v2-signed requests can still raise XAmzContentSHA256Mismatch errors\n  (though they *don\u0027t* do the looks-like-a-valid-sha256 check).\n\nChange-Id: I06c2a16126886bab8807d704294b9809844be086\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/de0dcd8f697d1cbd6399498b76143559a8dc9953"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/de0dcd8f697d1cbd6399498b76143559a8dc9953"}]},"branch":"refs/heads/master"},"2f82bd0b3cc5d8af3c86285c122e222f5e63e6c3":{"kind":"REWORK","_number":8,"created":"2024-08-23 19:22:17.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/06/920206/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/920206/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/920206/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/920206/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/920206/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/920206/8"}}},"commit":{"parents":[{"commit":"92eebe24c6837ef36302c801547c2de2e2726b2a","subject":"Improve test coverage for proxy object DELETE and POST","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/92eebe24c6837ef36302c801547c2de2e2726b2a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-22 18:21:01.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-23 19:21:38.000000000","tz":-420},"subject":"s3api: Clean up some sha256 validation/mismatch errors","message":"s3api: Clean up some sha256 validation/mismatch errors\n\n- SHA256 mismatches should trip XAmzContentSHA256Mismatch errors,\n  not BadDigest. This should include ClientComputedContentSHA256 and\n  S3ComputedContentSHA256 elements.\n- Requests with a v4 authorization header perform a\n  looks-like-a-valid-sha256 check, rejecting with InvalidArgument\n  on failure.\n- v2-signed requests can still raise XAmzContentSHA256Mismatch errors\n  (though they *don\u0027t* do the looks-like-a-valid-sha256 check).\n\nChange-Id: I06c2a16126886bab8807d704294b9809844be086\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2f82bd0b3cc5d8af3c86285c122e222f5e63e6c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2f82bd0b3cc5d8af3c86285c122e222f5e63e6c3"}]},"branch":"refs/heads/master"},"70948481f92a26466c23fa2d2cc072b77c6e0e67":{"kind":"REWORK","_number":9,"created":"2024-08-23 23:42:54.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/06/920206/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/920206/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/920206/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/920206/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/920206/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/920206/9"}}},"commit":{"parents":[{"commit":"92eebe24c6837ef36302c801547c2de2e2726b2a","subject":"Improve test coverage for proxy object DELETE and POST","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/92eebe24c6837ef36302c801547c2de2e2726b2a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-22 18:21:01.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-08-23 23:41:18.000000000","tz":-420},"subject":"s3api: Clean up some a bunch of errors","message":"s3api: Clean up some a bunch of errors\n\n- SHA256 mismatches should trip XAmzContentSHA256Mismatch errors,\n  not BadDigest. This should include ClientComputedContentSHA256 and\n  S3ComputedContentSHA256 elements.\n- Requests with a v4 authorization header require a sha256 header,\n  rejecting with InvalidRequest on failure (and pretty darn early!).\n- Requests with a v4 authorization header perform a\n  looks-like-a-valid-sha256 check, rejecting with InvalidArgument\n  on failure.\n- v2-signed requests can still raise XAmzContentSHA256Mismatch errors\n  (though they *don\u0027t* do the looks-like-a-valid-sha256 check).\n\nChange-Id: I06c2a16126886bab8807d704294b9809844be086\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/70948481f92a26466c23fa2d2cc072b77c6e0e67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/70948481f92a26466c23fa2d2cc072b77c6e0e67"}]},"branch":"refs/heads/master"},"09ca01e9d8a379b538119c4966056be061d265e9":{"kind":"REWORK","_number":10,"created":"2024-09-03 19:24:48.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/06/920206/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/920206/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/920206/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/920206/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/920206/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/920206/10"}}},"commit":{"parents":[{"commit":"ec8166be3366865954381572eef68b26d2951a25","subject":"Merge \"tests: Attempt to use configured tmp in xprofile tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ec8166be3366865954381572eef68b26d2951a25"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-22 18:21:01.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-03 19:24:34.000000000","tz":-420},"subject":"s3api: Clean up some errors","message":"s3api: Clean up some errors\n\n- SHA256 mismatches should trip XAmzContentSHA256Mismatch errors,\n  not BadDigest. This should include ClientComputedContentSHA256 and\n  S3ComputedContentSHA256 elements.\n- BadDigest responses should include ExpectedDigest elements.\n- Fix a typo in InvalidDigest error message.\n- Requests with a v4 authorization header require a sha256 header,\n  rejecting with InvalidRequest on failure (and pretty darn early!).\n- Requests with a v4 authorization header perform a\n  looks-like-a-valid-sha256 check, rejecting with InvalidArgument\n  on failure.\n- Invalid SHA256 should take precedence over invalid MD5.\n- v2-signed requests can still raise XAmzContentSHA256Mismatch errors\n  (though they *don\u0027t* do the looks-like-a-valid-sha256 check).\n- If provided, SHA256 should be used in calculating canonical request\n  for v4 pre-signed URLs.\n\nChange-Id: I06c2a16126886bab8807d704294b9809844be086\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/09ca01e9d8a379b538119c4966056be061d265e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/09ca01e9d8a379b538119c4966056be061d265e9"}]},"branch":"refs/heads/master"},"7bf279779995be5081e3ff9da7e113f20efe4446":{"kind":"REWORK","_number":11,"created":"2024-09-04 00:27:19.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/06/920206/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/920206/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/920206/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/920206/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/920206/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/920206/11"}}},"commit":{"parents":[{"commit":"ec8166be3366865954381572eef68b26d2951a25","subject":"Merge \"tests: Attempt to use configured tmp in xprofile tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ec8166be3366865954381572eef68b26d2951a25"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-05-22 18:21:01.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-09-04 00:26:41.000000000","tz":-420},"subject":"s3api: Clean up some errors","message":"s3api: Clean up some errors\n\n- SHA256 mismatches should trip XAmzContentSHA256Mismatch errors,\n  not BadDigest. This should include ClientComputedContentSHA256 and\n  S3ComputedContentSHA256 elements.\n- BadDigest responses should include ExpectedDigest elements.\n- Fix a typo in InvalidDigest error message.\n- Requests with a v4 authorization header require a sha256 header,\n  rejecting with InvalidRequest on failure (and pretty darn early!).\n- Requests with a v4 authorization header perform a\n  looks-like-a-valid-sha256 check, rejecting with InvalidArgument\n  on failure.\n- Invalid SHA256 should take precedence over invalid MD5.\n- v2-signed requests can still raise XAmzContentSHA256Mismatch errors\n  (though they *don\u0027t* do the looks-like-a-valid-sha256 check).\n- If provided, SHA256 should be used in calculating canonical request\n  for v4 pre-signed URLs.\n\nChange-Id: I06c2a16126886bab8807d704294b9809844be086\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7bf279779995be5081e3ff9da7e113f20efe4446"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7bf279779995be5081e3ff9da7e113f20efe4446"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
