)]}'
{"id":"openstack%2Fswift~920452","triplet_id":"openstack%2Fswift~master~I5ea9e6dc8b44c8c5f55837debe24dd76be7d6248","project":"openstack/swift","branch":"master","topic":"expirer-409-tests","attention_set":{},"removed_from_attention_set":{"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2025-02-08 01:07:26.000000000","reason":"\u003cGERRIT_ACCOUNT_34930\u003e replied on the change","reason_account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-02-08 02:26:03.000000000","reason":"Change was submitted"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2025-02-08 02:26:03.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-02-08 02:26:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5ea9e6dc8b44c8c5f55837debe24dd76be7d6248","subject":"Deprecate expirer options","status":"MERGED","created":"2024-05-25 13:36:02.000000000","updated":"2025-02-08 02:28:08.000000000","submitted":"2025-02-08 02:26:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":213,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"920452-expirer-409-tests","meta_rev_id":"a93a8431923f386215b141901e5ff8757f73eef0","_number":920452,"virtual_id_number":920452,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-08 02:26:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"all":[{"value":2,"date":"2025-02-07 16:37:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-02-07 21:25:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"all":[{"value":1,"date":"2025-02-08 01:07:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-25 14:57:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-25 15:23:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-05-30 00:17:07.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"CC"},{"updated":"2024-06-04 21:26:14.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"REVIEWER"},{"updated":"2024-06-19 16:31:59.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2024-07-19 14:05:59.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2024-07-26 16:49:45.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"9750eaf4d55de314ea677417e7e22f4e840c64e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-05-25 13:36:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4114ffc0042510db418cc5fc7680d9bd9e68f561","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-25 14:57:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf5d0eea2d8949af902fc83bf92caf04\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/83a5a3874dae41158eeca5d7ceed7a4f : SUCCESS in 1h 02m 36s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/22aceb24237e4167bb03cc42ad871aca : SUCCESS in 16m 45s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/9263cbb965764812a9305f22981cb9bd : SUCCESS in 1h 18m 59s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e479d9c9a16b41ecb3d17944de4c75e9 : SUCCESS in 19m 03s","accounts_in_message":[],"_revision_number":1},{"id":"a1f3000ba54f56a75cbd3960fb423c22854aca39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-25 15:23:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0852ea089d53424bac93eb5d0c0e2186\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab070cb7877c4b83a20bcde06ad714d3 : SUCCESS in 8m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/42acfccca4b64331bb3e38f4594f607f : SUCCESS in 1h 10m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/33c0135e2af442a0b28a33e5942c2f39 : SUCCESS in 41m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6c7e45ba0bd844d6afad5e674cc6c697 : SUCCESS in 54m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a03a2a22ba1f4303bb319166d3f9d186 : SUCCESS in 18m 50s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/25f0123eeea14d00953e3e2d841e6f1f : SUCCESS in 21m 54s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ed7db08419ee4f8a90ee0d655c3448b1 : SUCCESS in 19m 08s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/171a02e971b64e90a7810260732c7eeb : SUCCESS in 11m 19s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ef3542201f56475d9cd750fb8533b3ad : SUCCESS in 7m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7a34e789dc92477cb16d7810dd1f3152 : SUCCESS in 18m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/da6bf9496e754856a846b011d87d300d : SUCCESS in 19m 59s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d5ad67ddbac24ce8b1ab09e36149a564 : SUCCESS in 18m 57s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/827b89b5cbec488a9a2fba8477205fdb : SUCCESS in 15m 23s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/96ecfeb9e6ac4d90b433fbaf9768b109 : SUCCESS in 22m 56s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/896892f826d248c39e882624dc32b9a7 : SUCCESS in 17m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7d9b1e7ebf3a44afbc8549320a2e1441 : SUCCESS in 17m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/93b4b6336c164ec986441af8a5037d60 : SUCCESS in 19m 39s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b8628e74bda348428e1492a697c41577 : SUCCESS in 15m 57s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/12d9583c5c404cf5b2e53dbf6bb4b9af : SUCCESS in 15m 51s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/fec4fe635c944bff851965f7947eea67 : SUCCESS in 16m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9662e07cfc334634be1a54a566655774 : SUCCESS in 6m 45s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/389bae672b3f4f368609c70b6de5d910 : SUCCESS in 18m 51s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9d151aef797e46b59feccfec7d6c5718 : SUCCESS in 7m 12s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4202b2ea536a40c6af824b640eb696e7 : SUCCESS in 59m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/118dac5c17f34b95818b045a273b82c0 : SUCCESS in 53m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/76301acf5ab047f3acd76b92aeae0e15 : SUCCESS in 51m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0b80eb7fe5bb46a5a57e32a0ddcba57a : SUCCESS in 20m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ce034fe5e6f4ce4930bb942cdc4fa59 : SUCCESS in 5m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f8b7dde896314138b800da1b6000bba5 : SUCCESS in 20m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/72bb0e8d0f2d4c6a827950a79c1aed84 : SUCCESS in 1h 06m 51s","accounts_in_message":[],"_revision_number":1},{"id":"2f0fff90f80d5c61c26e042e81185be3f82bc663","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-05-28 17:00:37.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"ad3b50be1b93345ce8f07feba7385aaf84341c06","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 18:09:35.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/deff1731b6a140b1a476797695de5f92\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e9ba8d73ac7e4a088325e16cbcfe4c0e : FAILURE in 20m 59s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e7a54567372840f898f1e8106177cda2 : FAILURE in 1h 06m 35s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/837afdf75db24a469da6270f72cb1058 : SUCCESS in 26m 21s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/43983a5a7d4743adb255a8c7afbc6418 : SUCCESS in 53m 20s","accounts_in_message":[],"_revision_number":2},{"id":"7b761804c7d56df249da872c13827c91ed1ac19a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-28 18:27:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/79fc39cd27654129ac554f6bbb380afa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee92eebba367403ca8a3b2d3bb31d908 : SUCCESS in 9m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/63d52a346e044b9692dcf309e11a7a6c : SUCCESS in 1h 10m 34s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cc7d451c8471416395c65d29b3e1320e : SUCCESS in 56m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/787bc4f41a054819836cf6d250a47a94 : FAILURE in 1h 00m 28s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d4f12d1fc17949cbbb6d1186db54fdae : SUCCESS in 19m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c9466dfdc7d949318d148d952911b86d : SUCCESS in 20m 03s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/543b439ce47f4a6882d8539c1fe0f751 : SUCCESS in 16m 26s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b06f08a0df6e47f39afe3f52ca28692d : SUCCESS in 10m 37s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d8abb309a1f64bdd8f65949ad6b95379 : SUCCESS in 5m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/707543d571474ea0a33fe814fdf34f1e : FAILURE in 20m 10s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a41f00ae6c504644bff684fcb8287d6c : FAILURE in 17m 11s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/814243e6f5c34c63880655ee89268f24 : FAILURE in 17m 32s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/730eaa83738e4e049c7aef29f32dbde8 : FAILURE in 17m 22s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e03946c9fe1843fa861959c605b04165 : FAILURE in 20m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0eb29c43fabb44ddbf282444cc028931 : SUCCESS in 19m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/96182d678438408380f1f8dc53779254 : SUCCESS in 19m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/113513065149420abbf416bc4e9808d4 : SUCCESS in 20m 27s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/fb2b894bc2f64933afd08ee38ba00af9 : SUCCESS in 17m 37s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/6a36f95eec0b40ce9a35cc3de5cfc7c4 : SUCCESS in 16m 55s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/bc7cf1a0f8e34d81a91520cadcb5a797 : SUCCESS in 16m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d1e24d5355e947fc9ec301df54cd576c : SUCCESS in 8m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9b5c310459694de5aea88b1ce92323d4 : RETRY_LIMIT in 1m 37s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f6f616465fa3405fb9984b1f1a7c9f5d : RETRY_LIMIT in 1m 44s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/03b0385a087b416ebb098e6fe46b2fbe : SUCCESS in 53m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/75eb504ce5f8450c85798b398bb9d5c1 : SUCCESS in 1h 01m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3f93721ab58a4eaeba3b8705287fc9b3 : SUCCESS in 53m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/422189b91d3a4f919c9990ae4f4f0a5e : FAILURE in 20m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e863be96f7a84bbe9aa4bad9d4c79dad : SUCCESS in 4m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ec29a4a5f5d84825b650ecfed4cb9909 : RETRY_LIMIT in 1m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/90c046a815bb41dda227e17f038af251 : SUCCESS in 59m 16s","accounts_in_message":[],"_revision_number":2},{"id":"b91e903cfd508950baec77b281e22c02d531521f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-05-29 13:42:28.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"03ea67ba85dbb56847f83b66a0ac20f8bfe931d8","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-05-29 13:42:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7eb1026d88bfa45d7563595a3cb18f6dad1978e9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-29 13:51:24.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f69dbed5340e439fa64b7a8a2885f917\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/14be1a5233a7428e8b98f8e4a84a668b : RETRY_LIMIT in 1m 06s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4bbe3816a56c4fe799ea4ea0a47a1738 : RETRY_LIMIT in 1m 31s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e12eccb28a514eb4974833b34f976c37 : RETRY_LIMIT in 1m 26s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/eba334b35eb5488ca05f5e2394215df9 : RETRY_LIMIT in 1m 05s","accounts_in_message":[],"_revision_number":3},{"id":"1ec72479d67d389faf317c9c6a5339cf381615a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-29 15:03:18.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d8121b0bd164785955f0816af3de7dc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9539bd75b724d77a1c7c952570685a5 : SUCCESS in 8m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/ef331f13ec9244c1a9dbf0f9363db77a : SUCCESS in 1h 14m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/832c303e9e8b4fa9b211834f99f6a27a : SUCCESS in 54m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/658b092a920a44938fb7816f6b57980f : SUCCESS in 1h 03m 31s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9e45a72e84fb4fd5981831e53d62221b : SUCCESS in 17m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/034f738a8616439da8219aac8f084161 : SUCCESS in 19m 37s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/666d788a90114f8495ad3ef1285f5f7e : SUCCESS in 19m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3729e4757f5e476484f0251b79194451 : SUCCESS in 10m 43s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8f91965ec38d48da88ad6eedb3327353 : SUCCESS in 5m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0567cdb04770492a84c5f56cbac4a8e5 : SUCCESS in 18m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/536478a13f1f4f8b95e5bb250ffe9ce4 : SUCCESS in 20m 07s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/bd03632d61cd47b8866e51de32e81cd0 : SUCCESS in 17m 45s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/43af030858ee4830a39634a78b1de835 : SUCCESS in 15m 51s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/44ae12b3eb3d4c40a4093614c8d1cf90 : SUCCESS in 22m 34s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ce3143d62a8d4211b7c4a3da239e7072 : SUCCESS in 14m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fa21df48c51a42cbb9911e4e68a7c4ab : SUCCESS in 19m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/15c817727bcc4457af188aaf73e5dd79 : SUCCESS in 20m 55s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e00cb5b19b6240b09ebd54ea95a54680 : SUCCESS in 17m 46s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b9ad58c6fe1949b4bd7fa75bdccfa77f : SUCCESS in 18m 05s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ee36fd1eb02e46ae854c58f77ea0a55a : SUCCESS in 18m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c093e5c3ea9b4e97842fc549bddd39b8 : SUCCESS in 14m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bd989e94ab664a8abfc6784c5c70f332 : SUCCESS in 19m 44s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b68fc6c6c50242d2ba8b56ea3377b19d : SUCCESS in 10m 09s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/efd9094271bf44ef9608ed675c1caca6 : SUCCESS in 1h 06m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bfe76f2e1510423bae3e1fc8b9b421c3 : SUCCESS in 52m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/de9a42d3807b4beb91e4631a977fa38d : SUCCESS in 50m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/49e53e55e6094f70a7c4effbea9d626d : SUCCESS in 18m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/547bdc8ec7c748adb1afb829f020e640 : SUCCESS in 4m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6ab535a635b44c2a96e5b5b0b610ca6c : SUCCESS in 16m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/643d611fa05540b08497cd4f898e8468 : SUCCESS in 47m 42s","accounts_in_message":[],"_revision_number":3},{"id":"2c8140e677772aa6c82d0a76410358dfb5e12ecd","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-05-30 00:17:07.000000000","message":"Patch Set 3:\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"8897b9d48b6dd910c839cd6ceefc4294a3b9aca1","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-05-31 20:45:21.000000000","message":"Patch Set 3:\n\n(6 comments)","accounts_in_message":[],"_revision_number":3},{"id":"3f7f8ebe70bab8f74fb7c8470cb23ae5e3d99d00","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-05-31 20:45:30.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"fed1a1e8e9e5b459e3788010ab7d16bde8e3039b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-31 21:10:40.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5d1678e0df743309b21223e0d341461\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/251fd094b0624339968b14608df15144 : SUCCESS in 15m 45s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/edc88f8a11964938b944bdbb02ceb8b6 : SUCCESS in 14m 56s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/69d08092e6094180b25373fbf8793a92 : SUCCESS in 24m 01s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/cf1044482b444b718be386c2df6a4532 : SUCCESS in 17m 15s","accounts_in_message":[],"_revision_number":4},{"id":"71cda4a6eb516f8d634c3e5d4813aeac58737c1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-31 21:57:48.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a574ff8bbc3443ada81bfc66d0f9d339\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be7f051847c04c76a60f6e1ae5bb818d : SUCCESS in 8m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/40bde9811502492ab355d2cce0f13615 : SUCCESS in 1h 00m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e9b86b3ec97347ecb5386428bcd6a1f5 : SUCCESS in 52m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b0eb74429f224e0dbc4342ce3e11e99b : SUCCESS in 50m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d3dc0b6a333c4edd8ec90fa8fb0097f7 : SUCCESS in 18m 22s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/13cddf48e1b843b3b1e45d0efacafd4e : SUCCESS in 19m 05s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/50abaf5800a942a1959d27c1d5302af5 : SUCCESS in 20m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/dbd7b4323f2e4b57b9d4e908c4f4bb56 : SUCCESS in 11m 23s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/582020fe588240948425cdfba36a09ef : SUCCESS in 6m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/858eafd218f144b79753361b72d1f824 : SUCCESS in 13m 28s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/25baf5ff41a847c7a91832d584e971c9 : SUCCESS in 12m 14s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8ecc1d425cf048d6b9c5e97ae0ab3472 : SUCCESS in 17m 45s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/fb0e7f159ddf40ea97d456b689672a36 : SUCCESS in 15m 59s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2767598218424f2ebe3fbeb78d621959 : SUCCESS in 23m 21s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2c0bed77982342678545980b5ad138f7 : SUCCESS in 17m 40s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/99a3f3cec7c34aef801c265dd4cbe0da : SUCCESS in 15m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/015621de7eb946a0a658114fe66f885b : SUCCESS in 17m 57s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3e67ff0b918b4b48a4715d4c6b809951 : SUCCESS in 12m 59s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/605b965600234cb9b2e0083e00ea342d : SUCCESS in 16m 04s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4b590888e83d455180e82dcbb7c78c6f : SUCCESS in 16m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8ed32e506f7747c5aa564c01892c2102 : SUCCESS in 5m 46s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e896011542a243c3860dc7cc851f0f22 : SUCCESS in 17m 37s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/163aa9dc310e445a9d4e6528172363c5 : SUCCESS in 7m 20s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0d7b4507c0d447f7a8adef9a60a9cb7f : SUCCESS in 58m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/67746ed6f5aa4fec989db2054c58c9fc : SUCCESS in 40m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9f8885a9766346ad8afe16044943a0bd : SUCCESS in 56m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5b01898114774e15ad67df8f0d1ef08f : SUCCESS in 17m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87b79605f3a74ba285b2adfbec140f08 : SUCCESS in 4m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/10b2ce62589643d48679c769e355b2fe : SUCCESS in 18m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6921efa860a14e3ba816ff3612a62dae : SUCCESS in 57m 17s","accounts_in_message":[],"_revision_number":4},{"id":"dfb7b828a9732738e619104ce0ece2e198aa1760","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-06-04 05:51:58.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"d4d98be089487a3f2912ab65b69414b354ca0194","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-06-04 16:30:20.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"d959d5a9bd1de78dcfa8765aa0e309875bbe8b94","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-06-04 16:30:40.000000000","message":"Patch Set 4:\n\n(5 comments)","accounts_in_message":[],"_revision_number":4},{"id":"cbefd80bdb9b7d6b0a8e9f2f05e437e302ac1fb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 17:39:19.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/accce1a2b2a84e6788caa5d2d27eda2c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2146d984b5143e089ef94cdca257f44 : SUCCESS in 8m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/848061bd13db4030a29585728a24982d : SUCCESS in 55m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e82171c0695646a68517da2a8ca65e25 : SUCCESS in 38m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6d387618cfe0443eb92caa8545d3f900 : FAILURE in 1h 04m 26s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2f87f5a495cf4875a43851691063ad41 : SUCCESS in 21m 00s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9fbff09cbed249d5afe64149ec0a5612 : SUCCESS in 22m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7e9cf94fce4241c6bf74001d0b84f6bf : SUCCESS in 22m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4f21688738c84bbc9c8bc20be28a05ef : SUCCESS in 10m 08s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/b30a88fe79044fa2a523255a55fa166d : SUCCESS in 4m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1ff44569676145e2b4f57ddb8f903720 : SUCCESS in 20m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2a8a051776274223945cbcab52b7348d : SUCCESS in 18m 02s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c4e41f60ac1342f39c1d5efed1c23ded : SUCCESS in 18m 15s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/291f354c8f22407cadb26b676c58b8bf : SUCCESS in 17m 13s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/25c8c2c3189e4e5295080d4b0ee015cd : SUCCESS in 18m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/582ef3c4096c41a8aa21a04e8a55f39c : SUCCESS in 19m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f2016afb96794996b77ead0a79ae62e4 : SUCCESS in 19m 48s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c01c53c2e1bd435d98f70cf517e188f9 : SUCCESS in 19m 13s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a50e9d752ad24cacb19282b2e25d5060 : SUCCESS in 16m 12s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c2f5a313f0394ef1be924c9d078061e7 : SUCCESS in 14m 49s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/abb0e50567544db7b8c4eab0c8abde43 : SUCCESS in 17m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c75f1a9a6a7d49e291855fcf0d3694af : SUCCESS in 13m 38s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6d97b1d9566b4cbf9e424b605e273c51 : RETRY_LIMIT in 2m 52s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0c08dd8af64d4b6ab4e13c3530efb667 : RETRY_LIMIT in 2m 08s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d4517bbc4ed644eba784dff0e5c52dd5 : SUCCESS in 59m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/942ccb1d02664cb48aaf0966e0959adf : SUCCESS in 55m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6adbb50ce8a04f56a32ea15de6e69296 : SUCCESS in 58m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/704c00ce85204f69aefa71767107ef77 : SUCCESS in 21m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aee1bc5e95634940b0d53a14ea164dc3 : SUCCESS in 4m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/80c8221aa71d4ff3953283f2bdab4112 : RETRY_LIMIT in 2m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2a352ba8aaf447b9b2f50373cadd4f53 : SUCCESS in 53m 12s","accounts_in_message":[],"_revision_number":5},{"id":"76476c5380e3b8f69a9f86f998ded925ecdfff0f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 17:47:15.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe7efde98aef45c389d872aba8993fd1\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0213180d37bd42b08633f9e53e803526 : SUCCESS in 17m 59s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b2ead979eb2b42d7ae2ba1735364b964 : SUCCESS in 18m 55s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/0f779d7608e0417ba62f04e7784a5de3 : SUCCESS in 1h 14m 51s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/dbc2ea9f51a74e4bb4c357e1c9fdcc67 : SUCCESS in 26m 10s","accounts_in_message":[],"_revision_number":5},{"id":"8bc79c22db28289c7ef0aa424316b73e359c3cb2","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-06-04 21:26:14.000000000","message":"Patch Set 5: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"05be9aea03faab4ec307a0224fdfc2a5c70ed197","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-06-13 23:07:27.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"bcb0df66ace3e65eaf9e64e0831fa2f1f2fad522","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-14 00:26:15.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91fa43c2a5924157afb66e65fd2ef1cc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b852880758cb44efa87cfec06d7cb285 : SUCCESS in 8m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/4b5cac97d4cc40c8b127269eda57a7ed : SUCCESS in 42m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/89b32ff37a5241ff94878416f5cf2bd3 : SUCCESS in 50m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d53f1de9a8b8408ab0d64bab36916004 : SUCCESS in 1h 11m 54s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/87cf7ba596534a3f8b1bac652756b441 : SUCCESS in 21m 56s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/976591c9dd5443a5959f2ddb3b9048cd : SUCCESS in 20m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cba0067850804e3d83c20e770a722ff3 : SUCCESS in 17m 24s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e14a771b2a01471bbeb14d59102beca6 : SUCCESS in 11m 28s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a75cc8966a814c56bac7f3115d14aa42 : SUCCESS in 4m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/95c489ad1c3e4974b026b9b0147ec203 : SUCCESS in 20m 20s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c8100dd3027b4f8d8e70f8db85591a63 : SUCCESS in 21m 49s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/33ccb35ab72d44418c862d5a3d5c3cf9 : SUCCESS in 12m 42s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/49415af3334a487abe02b5ff26d576f0 : SUCCESS in 20m 12s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/bed9ac8d4afe4ec6810a4786d0397053 : SUCCESS in 25m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/027440e4d4ac46889482feca16e58844 : SUCCESS in 16m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/df8eb17492204db19fbd7436f15f0f87 : SUCCESS in 22m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b44c5e49bbb1422ca80ddb02d76be448 : SUCCESS in 19m 03s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9060da751f2e410c8073df3ad2ae61ad : SUCCESS in 18m 39s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/72f20f6b0dec4762a2ba05d6e593e4f4 : SUCCESS in 13m 34s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/236211e37abc4f2cbdd91dc78681c19e : SUCCESS in 16m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d8dcc9e1b4bb44b9a3df861b8e66b56c : SUCCESS in 5m 30s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d593bfea9114499b9a23def34516ccb3 : FAILURE in 16m 25s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5d6a00baa3f14636b4e562574ad57afe : SUCCESS in 8m 20s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/22a942b37d3b457d8cd851f44bfe6117 : SUCCESS in 53m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e52886790d574f1198d759cb264d2da2 : SUCCESS in 53m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/80245c73c7b440539d8160ed07726f0a : SUCCESS in 38m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c94c906ea343457b8e7d3c61d537c547 : SUCCESS in 21m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68eb10fbd6fc4d9ba600cfd5863038fc : SUCCESS in 4m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1aa0ceb2955043fb9f7716425108fe82 : SUCCESS in 16m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4a48540f25b34fb48865a9a23fe121f5 : SUCCESS in 53m 56s","accounts_in_message":[],"_revision_number":6},{"id":"89e57960055a0be79f17880a68fedad7096910d9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-14 00:40:56.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5345bfcf2b7843a59a498aa7d4cc63ac\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a707260938af4fed8c87f5272e29b66b : SUCCESS in 24m 11s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8a46b7e0a1384f15a2958a957ad79899 : TIMED_OUT in 1h 31m 01s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4b16aeab89194d9eaf5770a052ac2578 : SUCCESS in 29m 25s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6f4094474b8d46799ce2574610009e44 : TIMED_OUT in 1h 30m 56s","accounts_in_message":[],"_revision_number":6},{"id":"6fa418271671f06f1a35aca6b0e4e7ad430aed17","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-06-14 01:27:04.000000000","message":"Patch Set 6:\n\n(13 comments)","accounts_in_message":[],"_revision_number":6},{"id":"b8c63f308be42cefb0ecd26273eb849c03d956d7","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-06-19 16:31:59.000000000","message":"Patch Set 6:\n\n(15 comments)","accounts_in_message":[],"_revision_number":6},{"id":"8a950a2bd9797f8b4d56bebfd23ae35cc239372d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-06-19 16:32:12.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"33adc3bde26369da862706bc0a1f289481d33c11","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-06-19 16:36:10.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"60494738200374693bed9fda31f54c00a2e7e366","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-19 17:36:47.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73aed8fd9db44ff483400d795a886ad3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45868e7d365d4f8c99eb868d406c3479 : SUCCESS in 9m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/92e03be86f8741e1989787639b3b8d6c : SUCCESS in 55m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/dbd965917f8c4a1b8464f7b6acb509d9 : SUCCESS in 50m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/edba015f1af44c1da69b6532c050637c : SUCCESS in 1h 00m 03s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e85e4316c1a847bdb0cc1eb583bca8e6 : SUCCESS in 18m 33s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a6e662db05cf44b0b8b312ee529c80b3 : SUCCESS in 21m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2e5dbbd1e1fe4e27b803737ee883609b : SUCCESS in 22m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5ff402fde921412392cf68030c0969e1 : SUCCESS in 11m 32s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ef593a862eb148f4847ccd3637ceadf3 : SUCCESS in 5m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7fb18d2743f649abb4d4ada4bc5ab018 : SUCCESS in 19m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c1f3f47317354707928b7920ba3d2c24 : SUCCESS in 18m 28s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/332e429aa9bd4c33969c2c252d7cdce3 : SUCCESS in 17m 57s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/23719a188633481ca33d2827e29d8631 : SUCCESS in 17m 47s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b232c681e67842a48797cff1c3dec7cb : SUCCESS in 22m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ae92a68210da437fbe72a9611d90a542 : SUCCESS in 17m 56s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/caf146c01e7c4cd6922d887efa257f26 : SUCCESS in 18m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a3671fe81a954cc5be9b0c3257d06aa0 : SUCCESS in 20m 13s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/fa5a9fd0d24549a0857be67ed53b5a23 : SUCCESS in 16m 58s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b997fea177ac4720b4ddfefe0ce235ed : SUCCESS in 16m 50s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/16689e294adb4871aaa7168eac8bd35d : SUCCESS in 17m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5d5f210da7894cef89083d460e345b66 : SUCCESS in 6m 54s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d7eec0ee3e3d4066aaa4dc425ff49ccd : SUCCESS in 17m 56s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f6d081f88ae04182b0c27a65fd47bd7b : SUCCESS in 8m 37s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f494b70295144528a2b5774dffebc4fc : SUCCESS in 51m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ed67d82ba4c04a51b92cb290db32a164 : SUCCESS in 49m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/da2f1237246b4c7aa2298b225b067f0b : SUCCESS in 54m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fe57564e4eaa4c4d9049f901cd6c70a9 : SUCCESS in 18m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46ae810cd4d34c4e8e3c8dc39cacc326 : SUCCESS in 5m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6a10417fd7a4493ea1862a1756fe49ac : SUCCESS in 19m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/35585ad6ba4c45c0b286828616fd9826 : SUCCESS in 44m 33s","accounts_in_message":[],"_revision_number":7},{"id":"2cdd18875e94407565688c66f98f4e8db16756c3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-19 17:43:10.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b52547b5fe54de5abbd33b9d6141912\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3be65206ab2c486097593e942cf68c22 : SUCCESS in 15m 42s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ce6aae3555dc475695ce87a57d48e3cb : SUCCESS in 18m 27s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e0345a2dea3a48858835e101401facd7 : SUCCESS in 21m 06s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/71c6ef975c8c4938b170e78e54dd5848 : SUCCESS in 1h 09m 08s","accounts_in_message":[],"_revision_number":7},{"id":"deb7b0fb11e435af1a79694882c472f73219e440","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-06-24 15:45:31.000000000","message":"Patch Set 7: Workflow-1\n\n(14 comments)","accounts_in_message":[],"_revision_number":7},{"id":"c9a5c54b6b1c7d9c2cffbe8fa88b9b41618fc425","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-06-25 00:34:28.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":8},{"id":"1a0c0f742c0c608e1e574c9dc9b1fa1187827f01","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 01:44:54.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5223b4a4e3434bafa3fab5b1543bd6ce\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/93acdc4d4eeb4a719a2f8adfc0ae9416 : SUCCESS in 1h 01m 56s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9b9bd7a3f25a4676baffad82cf39f89d : SUCCESS in 1h 07m 56s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/bef3e628ce0d4716b125833299b24884 : SUCCESS in 19m 15s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b00095097bb344cea79d6e4324e2a835 : SUCCESS in 21m 29s","accounts_in_message":[],"_revision_number":8},{"id":"a3798c760b1dcc061ead24671667bf995e94171e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 01:56:23.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a87c74782d204e24bdfce88dd7e757ea\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/525f8813d4b84ebdb56074cb77f81d6f : SUCCESS in 9m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/4bdb48b7bbb043eb84c4abc622b37843 : SUCCESS in 56m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d83ed9b61a9446eb8c35e297afcc52e8 : SUCCESS in 38m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/685c467b6b7b438cb5cbfd0c57a7e9c2 : SUCCESS in 1h 14m 02s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f2416780e0e14ad28716dd3df7f35f6e : SUCCESS in 20m 53s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/be4dc4599c984e4ea796517a294590b2 : SUCCESS in 20m 42s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/54df495c96ae444c99c9273a794a8328 : SUCCESS in 16m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a012b1bda79c490283bf7b4b88c1c82c : SUCCESS in 10m 31s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/57473188bae346b59f2a49ccbadd9f46 : SUCCESS in 8m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c532cf5a8f1e4fb5be35c923841c136b : SUCCESS in 19m 52s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bce57e1cb29841189613450ab0c4562a : SUCCESS in 18m 30s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/2d3b516a8c9e4ec18e11e170d276cc47 : SUCCESS in 18m 21s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/70a5d414c9c540628a7b6d423925576c : SUCCESS in 23m 52s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6679c0581fd44fc3906684832196270c : SUCCESS in 21m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8a0d31b4d6304262801ff18ba070bfc4 : SUCCESS in 19m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b01558676466404e9567e374acf78db7 : SUCCESS in 17m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e821ebab2b8b45b68fe70e67ed31bf4e : SUCCESS in 18m 31s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/86329710ca6140509e8d3a7c1a8449f1 : SUCCESS in 16m 06s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/098a7c419b6043f7bdc716f547baec62 : SUCCESS in 15m 45s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0528f8e3279b43688d9e15c7262bf011 : SUCCESS in 14m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2576bdb0e5c9485e954854b0be865926 : SUCCESS in 6m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/aa855dbd3b8a48f9b846488cea8dd0ba : FAILURE in 4m 39s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5bded801b29b45639db294c25ed0360e : SUCCESS in 7m 57s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c109c9715ef74a62a4f057c5e935d2ad : SUCCESS in 58m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d1158c217946471686ee843ed85c0d69 : SUCCESS in 51m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/202ca7e2c95942fe8b57a3a1d96895bd : SUCCESS in 40m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c0380b8177864b849eefa9fcebe50273 : SUCCESS in 19m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0dd1b3d2cd7d4c569b10e83b823f5bb9 : SUCCESS in 3m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b5722952a483448eaee60148ff7156bc : SUCCESS in 21m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2ab763b2b52f4d1592ef4634b1112b9f : SUCCESS in 50m 36s","accounts_in_message":[],"_revision_number":8},{"id":"9c41f65ac41d4160540504afe712f99da6b071ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-07-09 23:58:46.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"741dbae4879c5c177c949a7a4758c964f0674311","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-07-10 00:02:17.000000000","message":"Patch Set 9:\n\n(14 comments)","accounts_in_message":[],"_revision_number":9},{"id":"4cf6611542bd9f7ef40374204d95acc0732788b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-10 01:06:23.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/90d7dfd610464312b0dd14fcc0804122\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02382c9ada5a43ce94a9e0e8b023369f : SUCCESS in 9m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/3ef90899c2ba4736bdb325f63caca83a : SUCCESS in 1h 05m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/11892ec30c3744e9b95f135dcba1d054 : SUCCESS in 46m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a2594b5cafc04f5383832b3f281d24d8 : SUCCESS in 59m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5cecb8c87c4444288d9b2d45cff92f95 : SUCCESS in 18m 10s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/22b80667387f484da876c2e952019b10 : SUCCESS in 19m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d14a5e1901f14e1581c6f23b4b3ee7df : SUCCESS in 19m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f3103d41236d40b1a2f1d3706437e728 : SUCCESS in 6m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/dae88673573b459c92e31e1c2cccb86c : SUCCESS in 18m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5f6d1d6d17b342c6ba0596b36acd7075 : SUCCESS in 19m 04s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b28bb8398d354f1f9ef869b56ac1062e : SUCCESS in 18m 17s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/c3e75bbcb9a24387a363cd45360e0929 : SUCCESS in 19m 48s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7f8bb37e27f64c37ac0130a1a96b1359 : SUCCESS in 18m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7229d9162f2d4ef5a9843a9c077eb218 : SUCCESS in 18m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/462f00b038a749b1b6c631e73131e1be : SUCCESS in 18m 57s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0264cc987ec5403cbe0b3548d3435b7b : SUCCESS in 19m 26s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/589bb04ccc6c4c4fbdcceca5322c9572 : SUCCESS in 17m 38s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/089546032b7e473e825019297fd1ea0d : SUCCESS in 15m 46s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e3fb94beda89436ea2c803ff54db82e0 : SUCCESS in 17m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/af1f53065ad648d98e47dd5193fa403f : SUCCESS in 6m 37s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/63692cb4590e4b4294869835f7f14151 : SUCCESS in 15m 25s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/620df1c017d3498887f1f6de29dd1520 : SUCCESS in 7m 01s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/36b87d79ad3f4b32948e63c479b668b1 : FAILURE in 58m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1d1c5014009d43df9a9d19cb339cb12d : SUCCESS in 48m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fc0410cf1a814bdcad45f2b714350303 : SUCCESS in 37m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/03dff3244496468b81a7d921e08f2236 : SUCCESS in 20m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/657476ad8bb44584bb895e3d17286ce5 : SUCCESS in 4m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5be923e301f64c1ea5abd636f7ba65bd : SUCCESS in 17m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/02313161048444eb9bc4e8b483dec4f2 : SUCCESS in 47m 53s","accounts_in_message":[],"_revision_number":9},{"id":"4a265720de0a9f2661dbf5634b00b1959a7234eb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-10 01:32:23.000000000","message":"Patch Set 9:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35b608547b984b2692cd31109f4f68ac\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6e76e81a92ae48e69f658e643a60d377 : TIMED_OUT in 1h 31m 08s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3aff7f997b394d1da331be2594ec383c : FAILURE in 4m 03s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/32ebd64a53604dc79f99601f43487f34 : SUCCESS in 1h 19m 25s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/13aeade170944b24bbd5580578d58d63 : FAILURE in 3m 37s","accounts_in_message":[],"_revision_number":9},{"id":"7a319f3d37e591d2c517479b5c3988480a4a9ac3","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-07-10 13:56:47.000000000","message":"Patch Set 9: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"b24527820846696e9e7b7869638761cc8920deac","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-07-10 16:31:27.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"32b20bf2500351b5ce7e6f684a305a0aaa897a8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-07-10 23:01:25.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":10},{"id":"bc883e57897a6815d9b5ffe0f2124ec2b92ccf7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 00:29:56.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/145ed685027c4fb3a528011a4bd19d66\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d9e765452494dc2b3fec7dd46e8a2a6 : SUCCESS in 8m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/4ceee3d691584327acc3f492b201a4e9 : SUCCESS in 1h 06m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a80dc0f06c4a4d848648ae7180ac06a9 : SUCCESS in 57m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2fd3f6a7d54d459b826d7c21899bb523 : SUCCESS in 58m 03s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/af8f40772334449999862c4c1389b845 : SUCCESS in 19m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ebda732aac1640a6837c52cc22137e3e : SUCCESS in 22m 06s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/889ec7c3798f4f948cf27ad25810f6f4 : SUCCESS in 22m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1d1c5852730b4ef29a38664987316c37 : SUCCESS in 6m 01s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/31747b2566c245f3933ef9632623d2de : SUCCESS in 17m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6b03947f5f454c8babc4c1921461d9d3 : SUCCESS in 17m 04s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/5c7ef115fdfb45c7bd9933b96bfa3fbf : SUCCESS in 21m 04s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/f70cb02e8ad04e16befc312c369a24a9 : SUCCESS in 18m 33s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5f4fb92627114b20943f2759f38b0c9d : SUCCESS in 24m 53s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/886d38708ed24f33aaca9df0d23f77d3 : SUCCESS in 18m 01s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/902703aaee0f471e877f9c0f624e7aa0 : SUCCESS in 20m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9c76dc1f98c344eab66be8d50681e6a0 : SUCCESS in 21m 39s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/45d44a41aa6d4adf8d35417ed1206079 : SUCCESS in 16m 41s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/29836b5e72ec4cc9a9eb2f22bdb5afec : SUCCESS in 15m 26s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ea80fd41965443529aa050bd0879da17 : SUCCESS in 17m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/be81c982c6b749cca54cdbabb7755396 : SUCCESS in 6m 42s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7c4e1046026b4524877218e460eef8f3 : SUCCESS in 13m 23s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8f0cb791d0934ebe9e73b5d25e9a10ff : SUCCESS in 7m 06s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5691eb57f8ce460a8221ab04798361a5 : SUCCESS in 56m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3799e429faa04725939f0757dc1a07a9 : SUCCESS in 40m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f67ce81afe264fa790c600eca0006f52 : SUCCESS in 46m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/94dfe81a15694391a60a3595bbdfec62 : SUCCESS in 21m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/728d8b63bb3d49f0a33695ffbaf952cd : SUCCESS in 5m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c78d8cafb1e3497d989a05f9eae7c2a1 : SUCCESS in 22m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5645811198bd4a9ba324f274cec749c4 : SUCCESS in 56m 53s","accounts_in_message":[],"_revision_number":10},{"id":"98d1aac000690830cd75489fa6335ff9647bc76f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 00:44:14.000000000","message":"Patch Set 10:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1944ce761e6d46a8b2dcd5f909a70a6d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/600c816ef7d74140b87fa3e2634d48b0 : TIMED_OUT in 1h 31m 01s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b755d99efefd47d98ecb037a6ebcfb37 : FAILURE in 4m 00s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ccb1ba74ded54e61946bf94bd50603ac : TIMED_OUT in 1h 31m 48s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ba913a5a891b41fe92818ee620b10ae4 : FAILURE in 4m 21s","accounts_in_message":[],"_revision_number":10},{"id":"9e1341653645bc630e2746759431a3fd177b5398","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-07-18 16:52:31.000000000","message":"Patch Set 10:\n\n(21 comments)","accounts_in_message":[],"_revision_number":10},{"id":"fd664db0dc294555150f74f1b93f28f8bde4ee3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-07-19 13:45:09.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"296beca32dc11b900b395aa306e293bbda879e69","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-07-19 14:05:59.000000000","message":"Patch Set 11: Code-Review+1\n\n(10 comments)","accounts_in_message":[],"_revision_number":11},{"id":"05c8e95c62e970256de63a19f0315e4018274168","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 14:22:43.000000000","message":"Patch Set 11:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58d67224b3d24d43b21c3f0e412c2a23\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f59433d400f84344ab163aa7e4b58cfb : SUCCESS in 25m 36s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/14ef9e5c67c34669ba9cf0a4bce00610 : SUCCESS in 29m 18s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/2edeb465079f4d9092a41cab4d555ee3 : SUCCESS in 34m 34s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b72e4366cd244d9787bfdd2ad7b4fead : RETRY_LIMIT in 2m 09s","accounts_in_message":[],"_revision_number":11},{"id":"28eb4bb467775ef561cc9329e1b05bfdc35ff3bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 14:48:43.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15bc797029494bbbaffd4cd822124c64\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77290857844642f3a292b6fe85ca10a6 : SUCCESS in 11m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/b1e19faeda4145c5a2f9cc90259293b3 : SUCCESS in 58m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fa7544e058c84b7f8f5cdfa9b4fb5385 : SUCCESS in 47m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9f59342988454615943aefc7dd48e6f8 : SUCCESS in 50m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9ca5f842fee94824b48381a662b39ad2 : SUCCESS in 21m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/731fc03ecadd489790ca807777867f32 : SUCCESS in 20m 16s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5483b54d6f9244f9a010bb8a5ac54351 : SUCCESS in 22m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8fdb96a1e34541b1aec81339efab3009 : SUCCESS in 5m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6a8d4426a1624dbf9d852b31c6968e28 : SUCCESS in 19m 48s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/04781a9927e24800ab562d90bca0bd27 : SUCCESS in 16m 12s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c5528a970aef4a189f9399819f8dd06f : SUCCESS in 20m 20s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/7bd38651bc7d45cfac5b14e4cb831565 : SUCCESS in 21m 31s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ee958cb7a65b4fa3bd82f7a18bbc6f95 : SUCCESS in 28m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/07877d4757464a30b0ad16a02af3cfa7 : SUCCESS in 19m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0008e67d7b1f469fb4b5adf7d4a1269d : SUCCESS in 17m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/44e09d376acc43a9b86bf996935d37d3 : SUCCESS in 22m 04s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/de429f02dec946e4a4d9008503959f82 : SUCCESS in 16m 18s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9c1e3f8d952440279fa9cf55a358b56b : SUCCESS in 20m 24s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/7dc0f9c1842549beb2f98729e536fdfc : SUCCESS in 17m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a78c3788e7dc4ebaa5558adcf41a03cc : SUCCESS in 6m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/016b6bcaa8f0490c9266adc1907feb84 : SUCCESS in 20m 32s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/07f2b64705834f428e78450343bae118 : SUCCESS in 10m 24s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ce1ad849685a4dd68257944bc95d02b7 : SUCCESS in 55m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1b7adc4323144f758e7d6c2630f1465a : SUCCESS in 55m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bc4a0172c16046b094616633c7d77530 : SUCCESS in 44m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/45c34edfe5b446b6b95f34ddc94ce3ea : SUCCESS in 20m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e1d18a120bd43efaa760333af7d48dc : SUCCESS in 4m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b1b56b49bfcd4a1fb9d437accd0f4c96 : SUCCESS in 17m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2c5db724561b47198fd381c811ada0e9 : SUCCESS in 54m 54s","accounts_in_message":[],"_revision_number":11},{"id":"128d125a164ae290ae2c383ff7a6adef97492210","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-07-24 10:24:51.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"787293013807d88a85eccaf3b7482f09d7afbe4c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 11:54:24.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4b405e4db6e74cbcb4600f36addc26ae\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efb17e41846e4c46bb3220966ba51b51 : SUCCESS in 7m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/e5677db27f684adda29665acdecbef94 : SUCCESS in 1h 04m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/eaa6bd0ee38b4bcdace8890fd4428970 : SUCCESS in 44m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e357c017d82e4da996ca484a4f7e14aa : SUCCESS in 1h 01m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1ed8dccd9dc0402381a8f3131fb3da09 : SUCCESS in 19m 59s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/053552694cd44b8391aea22530a78a1f : SUCCESS in 18m 24s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4792c1087c684d3eb30fff7a9f2e2a25 : SUCCESS in 19m 49s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4df46e278f6a450db31d36917bc9bc20 : SUCCESS in 6m 12s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/219bd086f0134fc7866181e0d4b82a6a : SUCCESS in 18m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a88ec062b3a144559c52e88e51616381 : SUCCESS in 18m 29s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d8ad600698054669a2834904d3dcc6b2 : SUCCESS in 20m 07s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/510bb4a088554c5a9c33982fae8b6565 : SUCCESS in 20m 02s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/09bf48e1049e4086949f46caab11bdcb : SUCCESS in 25m 31s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7373ae53785f43388995a14a64d26bdb : SUCCESS in 18m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/708f9c1b2d0f4a7dabf79141ff3f9724 : SUCCESS in 18m 21s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1694e154ab2d43f5b10a5f377d6ea2bc : SUCCESS in 17m 36s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7e4fb29836d5467d98d53c321a1a938d : SUCCESS in 15m 23s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f7e0b9fa86db4aa19ee3a4783adec214 : SUCCESS in 19m 38s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/54661762c4ab4f599f03328c1baeba3c : SUCCESS in 19m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/89d2e4443c7f4c5d8c0f0bce15bf27d2 : SUCCESS in 8m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/76af121dc46b411f8b27b040f7b471b9 : SUCCESS in 15m 08s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1c0f5da9bda14cf9917942da6e2c0c8b : SUCCESS in 6m 19s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c86be103cfd349ebae394c79617b417e : SUCCESS in 47m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1a8fd1252b1c4391a975e4f8efaaf041 : SUCCESS in 46m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/408095c9c0b34ed4aba1da9b30f27a53 : SUCCESS in 46m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/06a46f8180934eab92804d26f76118ea : SUCCESS in 18m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c685f38a9354ef6b252ecebcb7ac632 : SUCCESS in 4m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a04a24ecf920401186c4320e1d604caf : SUCCESS in 25m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/90337599a4764317aac98a5cf170acdc : SUCCESS in 53m 41s","accounts_in_message":[],"_revision_number":12},{"id":"5ce45616c37c960278481cdf48e54d5c7e503be7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 11:58:12.000000000","message":"Patch Set 12:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fd443be9c2e747c296f0f66d96b30b55\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/14421bc029f741bb89a2226c3abeb069 : SUCCESS in 1h 19m 30s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f8ced807ef6a4db9983e5e3ade525199 : SUCCESS in 21m 12s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/01509648e2484c7fa41cd8fbba61aed9 : TIMED_OUT in 1h 31m 09s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/cfee5c00f3aa4ecb8537021e22bcdb4e : SUCCESS in 21m 22s","accounts_in_message":[],"_revision_number":12},{"id":"d7e781cbb2ac6890076882b5b87a704427760843","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-07-26 16:49:45.000000000","message":"Patch Set 12: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"dca26744e33cd58985e8b45a2ccca76de4ba4607","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-07-31 14:06:46.000000000","message":"Patch Set 12:\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"59c705d3b2c0f925ae9c207c4a44bba52423aca7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-07-31 14:10:17.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"58cda213e1bf73ed48d55e38e95c82c27924fa94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 15:26:02.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e7d0ba65116469793e875d943ae8003\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c524524bca8644879ab7bf603511704e : SUCCESS in 8m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/b0df95de12fb447ea717444a59bc3b8e : SUCCESS in 1h 02m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/47685232928f423bab9be548e57ea0db : SUCCESS in 42m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/67650b5f4ddd4d239b023b44145470dc : SUCCESS in 54m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/75d9f539b3e84da9b919abb041afa16a : SUCCESS in 18m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3df114d6834844b18d35b356cb8c05cc : SUCCESS in 20m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a6a8c0a619a34285a7d9c9abc5f42871 : SUCCESS in 23m 03s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/886f02e1a27b415aab46e82270601166 : SUCCESS in 5m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/194de09d05144378967e0cda3c9ccac4 : SUCCESS in 19m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e395d5735c3d4f6099b76fb25f85dfb4 : SUCCESS in 21m 03s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6fc9dd0d0ce24953911b957252d1eb46 : SUCCESS in 22m 29s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/ac445bb936214a2abcb16f741e73fe35 : SUCCESS in 19m 44s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/76beafef9c6d4ec4b72f8818f4242eeb : SUCCESS in 46m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/aa08d48d88764025bf17be3d76b10aac : SUCCESS in 19m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3db4867f7b404be68e37b27249e539d2 : SUCCESS in 18m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/df42d22d4d0647ea9dd3a9626305a90b : SUCCESS in 22m 37s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/bb5fc6461aec4826b20b42bf6e79cd7e : SUCCESS in 22m 32s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/076bc0888c0e4a67a16b231ba237c405 : SUCCESS in 19m 47s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c2d4fb79badb482d88dc374053dc6732 : SUCCESS in 24m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0c878bfd913b44398f5abf0b11e7d8ff : SUCCESS in 5m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5b09682a6ac54c209c9d0456d36ec48f : SUCCESS in 18m 24s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e2e4c9874f9340ffb711a419a93d1f1d : SUCCESS in 8m 11s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2e5883d346574438877b2d805db322e1 : SUCCESS in 50m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cb4ab0f58bdc4f5bb00334608b2f40f1 : SUCCESS in 46m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3be30fc5e0224092a2b452c7a5430dea : SUCCESS in 46m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ecc8deac7eca4cc8b0c2ffb67a4c815c : SUCCESS in 18m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5558d8cc93d46379001b204136a5009 : SUCCESS in 4m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/57b93597d8ba4a5a91932845845a92cc : SUCCESS in 18m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0766df4d06244101a4291f55dc825c5c : SUCCESS in 1h 00m 57s","accounts_in_message":[],"_revision_number":13},{"id":"581199f7e813d60c4792542fc011f9c0f05995be","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 15:43:40.000000000","message":"Patch Set 13:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8151dbdf825421a9017d2065bf79198\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/af3f97f405584d1db527114891f2904a : TIMED_OUT in 1h 31m 07s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a229789176d1415191556a9b8b7467c5 : TIMED_OUT in 1h 31m 00s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/1938ec69dbbc4de280d32bf7861a87b9 : TIMED_OUT in 1h 30m 50s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/93d3f773c1f643cd8f56191bc81d8995 : SUCCESS in 1h 25m 58s","accounts_in_message":[],"_revision_number":13},{"id":"3c753577924d66ca4dbe4fe8e6634ada827d5d1b","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-07-31 15:51:16.000000000","message":"Patch Set 13: Code-Review+1\n\n(19 comments)","accounts_in_message":[],"_revision_number":13},{"id":"099f1830eb89d29fd3fec3dd0cb5001dbde97ed2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-08-02 11:36:31.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"3e86ee4f4647e37f712530e8cf7e7621b7b28ea1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-02 12:50:25.000000000","message":"Patch Set 14: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/733f1c9c89544f1a870cf91a66da6b5e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31eaff0426c24f56b6084c952c7b4f90 : SUCCESS in 7m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/80794fa19a35463cbe42b0c832f13855 : SUCCESS in 1h 07m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/02d0fc2536d14abe8d916605d2a1445c : SUCCESS in 50m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f2fb236fca23449ba5a3fbd370004eec : SUCCESS in 49m 22s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3adfda6db4f34a668349afa06d609227 : SUCCESS in 21m 02s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8d24ebe1c6754a41ae9c83b1d3f15d65 : SUCCESS in 20m 38s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7f327973873e4c7c830bf9dc49f82b17 : SUCCESS in 20m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/423329569d8f4c3d8e25006d4c4b759e : SUCCESS in 5m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/00d47fd0610a45ee8d3df90b0c06b932 : SUCCESS in 19m 55s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3dc12562c9af4c7ba3ed0231abad53f7 : SUCCESS in 22m 23s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d672c8d3e85645b28b7c0536db351621 : SUCCESS in 22m 14s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/c4b58e0756ce4df6b05e5837f79c6bea : SUCCESS in 30m 31s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7972c1bc476b42aea7cf6861edb9f86f : SUCCESS in 33m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b78d5ccfd57f490b94f7b70e4c17927f : SUCCESS in 19m 00s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/42a8774d27d2420bb1e8108cc5456f8a : SUCCESS in 19m 51s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/73c6fe34e5814e7abbecdc8a39d2e600 : SUCCESS in 18m 39s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3b24cc04f343475ba38d3a54b46a3025 : SUCCESS in 16m 59s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/bc24697b1f894909905a58bb97ddd488 : SUCCESS in 17m 00s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/44e2da38d0ec4923bf59c91eb7ca6ba5 : SUCCESS in 17m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a7379f5fde154adda49b4f699fe2310d : SUCCESS in 6m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0233dae3d52444f1974806953db3f44e : SUCCESS in 12m 28s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/87f3f6a0e6984036b29af064882768f0 : SUCCESS in 6m 32s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5b7ab57447384f8b937a31cdfdae1936 : SUCCESS in 46m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3b97a11f7b224b3cb8413921e76f3852 : SUCCESS in 40m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a7d54acbfd594e219dd32f8e41c68ab6 : SUCCESS in 39m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8323f9abc94143b48c576806da85ce21 : SUCCESS in 21m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02f0efabee47469d9022093400e6e4f6 : FAILURE in 4m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e7dd6b62baab4789ab4bb3843199cc7f : SUCCESS in 17m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d6c7243b719f4f49bbff3f1d4064aeec : SUCCESS in 56m 45s","accounts_in_message":[],"_revision_number":14},{"id":"9d9a45b53fce86398445de4da0eb4883cf57abef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-02 12:58:18.000000000","message":"Patch Set 14:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b80d90cdfcd0480f9732552b4183e6d4\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/126ca7b837604de2b206976bbf4f78fd : SUCCESS in 36m 43s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5c2130a31f234ac2b01b5f624c252436 : SUCCESS in 46m 26s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/0555d33ac7974dfc9914f23ad4adacde : FAILURE in 58m 25s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/db331e8ff5a24f72979a9b1bcdee6a05 : SUCCESS in 57m 25s","accounts_in_message":[],"_revision_number":14},{"id":"07b49712478f72c69735c0571423fae4e1917c00","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-08-02 13:30:58.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"6e38b8d2f9b623239bda36d4c8fecdf5dc396092","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-02 14:48:20.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4a53838bc84549d592ad2a5c23e3abcd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3feac4a4afbd4ebab8f826df20b46d73 : SUCCESS in 10m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/371e242d9b2d46cdae17b92038da6e51 : SUCCESS in 1h 10m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/76d73bd83bab46fa946737a61a7a5b32 : SUCCESS in 52m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cfe0c1236e0a44cfafb146c0366b5c9c : SUCCESS in 1h 04m 04s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e99b8879d6944c6b9040475ec82519f9 : SUCCESS in 18m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ee40bc6825a44fd08428ab84a9f0b78f : SUCCESS in 23m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8bbd43e0271347279a2a943c0a532b0c : SUCCESS in 18m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8a0fe4f9c7df42e4af8ec8c11c2a5093 : SUCCESS in 5m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/dc55ac5b96df47bdae5d631aeea8e260 : SUCCESS in 20m 30s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c1cfb29f5a6f40bd8c4470d031358758 : SUCCESS in 17m 26s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/4b1f45b78a364d67b41cc38d50d7650c : SUCCESS in 20m 16s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/8e1b29e3c4b24660a8e720adfa22b2df : SUCCESS in 23m 19s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/fc6b778a527e45d99459e91e7c6e17b5 : SUCCESS in 24m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/31eb9639e1d449308e73d4778c349477 : POST_FAILURE in 27m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4d7a93719740452a89b887837e45b48a : SUCCESS in 18m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ec74fa3b54aa429480d489434cf120b2 : SUCCESS in 19m 19s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8b6009852a194578a361cf8819dcd249 : SUCCESS in 16m 34s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ed522f0afb9140b7aeb6f5168c45c0c2 : SUCCESS in 17m 39s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/08fc74a576bf4ceda999b10b7c342204 : SUCCESS in 18m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6345b03897e846ada0843808af8e7e1a : SUCCESS in 7m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b4ec86f35f3e4e80a3e5be7ee3edf48d : SUCCESS in 14m 19s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ffb0d752826f4235a40a8742506850f7 : SUCCESS in 6m 47s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/51505bfe4df547bfbb9f96b4a9c35428 : SUCCESS in 50m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c54d16c130644f9aa6641323587fb6c8 : SUCCESS in 49m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a04cbadc3d6041ddbb8793e50ab3ae66 : SUCCESS in 40m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/de949084ccc84a3e9b1f9b08e5c87b60 : SUCCESS in 21m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2386892fea3e4ce88af9c73dc7d580f4 : SUCCESS in 5m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d7457a5ff7d448b9ab6b0b5485110179 : SUCCESS in 19m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fdf7b6df826c4b80878b214e7548ff9d : SUCCESS in 43m 24s","accounts_in_message":[],"_revision_number":15},{"id":"a2cc99d00d766f1a7533191b90cab97cfcf55947","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-08-02 15:44:13.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"02c8cf44c59c9fcc82da58a324ee7965ee41d58e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-02 16:05:04.000000000","message":"Patch Set 15:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4cf3ab5740ba41a7a2396fd5cdf48e0e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6394dbf2687649bcaa62fc3f43635791 : TIMED_OUT in 1h 30m 59s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/48683f34c2be43d58a0d88b3d4bf260b : SUCCESS in 1h 10m 53s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ce5dc8b4d1354c98bf224a65aa4eccc0 : SUCCESS in 53m 56s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/92783d46a3324d1f821c3d1a1aa6494a : SUCCESS in 1h 27m 21s","accounts_in_message":[],"_revision_number":15},{"id":"3fd5c5b2a6bbb23507b4bb9b4a6bca97d5633aa2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-02 16:54:32.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9eb1769ae964ab5a6d57b6b315bf28c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c48027a032148f2835b8e508117d039 : SUCCESS in 8m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/947076f4dca5483aaae371e38d100ca1 : SUCCESS in 48m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a7b1ae32342143afbb7146656e91ff0f : SUCCESS in 50m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/abf8494b6d9341a880b0798a817d4c71 : SUCCESS in 59m 24s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/073abfab069f465482868ec137b9de77 : SUCCESS in 20m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/734ddd42f90e4b7788db34a9652695fc : SUCCESS in 20m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/aebe55835f14420fb2f79246b83c9b53 : SUCCESS in 23m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8c0e40e4647e4efe85f3cf04d98e8493 : SUCCESS in 5m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/af151b3a5ed94f30b248e39e35f48798 : SUCCESS in 20m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2d6a8b10a3d243bb8c93dce41553e88a : SUCCESS in 19m 00s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/42e96037cf5c44d3beda49af53afabec : SUCCESS in 20m 58s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/bd6609e3dd2647409ebce581e07ca4ad : SUCCESS in 27m 41s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/172300e82e574bb18e15b7927f63f81d : SUCCESS in 27m 09s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/358e0b5fdb454902a757be26b21e365a : SUCCESS in 19m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1f377a9eee30457fa2668d280bfd59f6 : SUCCESS in 18m 32s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6559d429cbb943eebed32b0614a35c5f : SUCCESS in 18m 29s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e4e07114856c48c9a5a230eba8d1c9bd : SUCCESS in 16m 41s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5265aab2bb94492a93a0e869ce50c26c : SUCCESS in 17m 06s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/eec5e818add14b519263442e50e802fa : SUCCESS in 16m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0b5471aedb364c71a109ce1e500c126a : SUCCESS in 6m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8198c62ce4904b17bda706174c87b5d6 : SUCCESS in 14m 54s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/38d4d59ec6fe4117856a6ecc6065bd49 : SUCCESS in 6m 25s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/07d5c151e47e4bf19a2dd444880019c7 : SUCCESS in 52m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/486353279f9f4a96af9053eb76065e79 : SUCCESS in 50m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4c08699e3911422584ed3f28a473a120 : SUCCESS in 52m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d0b6b715273448e68850fb25651b17df : SUCCESS in 20m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/449a5d98fbc74199874c18c05c6e360f : SUCCESS in 4m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fc17eeedf496494b80a5a23b59fe82c7 : SUCCESS in 16m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fb3d6c5bbab2478f9221facddb798b9b : SUCCESS in 1h 02m 00s","accounts_in_message":[],"_revision_number":15},{"id":"d18b6bac9aaf3bb8de8ad702a14bfbb15e8bb830","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-08-05 13:00:48.000000000","message":"Patch Set 15: Code-Review+1","accounts_in_message":[],"_revision_number":15},{"id":"3979dab066e7621ddb59400efbdb7d4fbca26767","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-08-07 13:26:44.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"24b6b02ec44d15adafe543846437c0832c565e0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 14:31:28.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3488c90109a4ba6a66fcf3a349785c5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0210b48defb45ecbf813ea6c61c84d7 : SUCCESS in 8m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/8351ff2cd42c4c80ab5d5065e234a786 : SUCCESS in 48m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a8f890c4804d49dfbe80316be8a3c55f : SUCCESS in 39m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b78c06a54b2947429be63a73a87ea5cb : SUCCESS in 49m 20s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/27e7908cb6674714b696ed65655b3d51 : SUCCESS in 20m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e0681b8193bf46b0b0164b13ea563195 : SUCCESS in 20m 24s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2ddf416d83d045ff9bfec33630ba092d : SUCCESS in 19m 43s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7b995d2d7e8143939805ee7879c87546 : SUCCESS in 6m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f8ae106fde4046fd8b10ee636e22eb51 : SUCCESS in 19m 45s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6f41309ded074cdf9356fab206c5dd8b : SUCCESS in 19m 50s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/bf8d5a40440b4bf59c36c4093922484d : SUCCESS in 20m 03s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/c3003cdba811481bb1fc0354ae090f62 : SUCCESS in 18m 29s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ef27c01a00dd48f98f023325ba6ee6bf : SUCCESS in 25m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3bd1b865d9fb4eaab2fdf24d5a98e6ed : SUCCESS in 17m 49s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a0c81ba3089a4fbb833abc44ce0f6636 : SUCCESS in 17m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e25da9aac8fd4e3fafa09335544927cd : SUCCESS in 17m 56s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4e692c2cc0fe4929bc387907d9065dcc : SUCCESS in 15m 33s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7d523e998c2548748ecdc11585eac379 : SUCCESS in 16m 16s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1afffaf530e34357bfb6feb2c730fca8 : SUCCESS in 17m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/70745d8cc62a4acc9756726806ff5309 : SUCCESS in 6m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/34d4f2ab4fc74669959b1e97c7c8d602 : SUCCESS in 15m 26s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8b99ebbea3dc4976873f3afc23423589 : SUCCESS in 5m 44s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/728d55d526234922a77bd1118f732806 : SUCCESS in 53m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9ac72a8286de4e1f9c19ea2a07c3a006 : SUCCESS in 51m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2e6c9ca1939146a5bd7e3f685180ad61 : SUCCESS in 46m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7aac8829b38846949ad9850e3de17072 : SUCCESS in 16m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28d070b622ba4ae9bbdf042ee211815d : SUCCESS in 4m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b6232da2834b4931a327c93a4c4a5f0c : SUCCESS in 17m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5c39873cc817401c81760b2dc0687b31 : SUCCESS in 57m 54s","accounts_in_message":[],"_revision_number":16},{"id":"76aa63a8702ee4f7ebd582b278b25c6c4d2db596","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 16:16:13.000000000","message":"Patch Set 16:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e39e831d1179496c89a5c691e555f382\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ba45ec15443247fe9b33ea5061b804e2 : SUCCESS in 1h 30m 58s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0c95f0ea22a84e97944eb1dae8598aa1 : SUCCESS in 1h 27m 55s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/85ea98fcbabc4ce5a061be00b33c7273 : TIMED_OUT in 1h 30m 58s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/86855e10ffe545fcb510bef951d09f86 : TIMED_OUT in 1h 31m 12s","accounts_in_message":[],"_revision_number":16},{"id":"084ae09ed9dd7727d14175b8ed2c5e9d659e13fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-09-11 08:45:14.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":17},{"id":"99128eeed3a57e637eddb00d81ca2abc449f98ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-09-11 08:48:50.000000000","message":"Uploaded patch set 18: Patch Set 17 was rebased.","accounts_in_message":[],"_revision_number":18},{"id":"ad9946b6cd5666d3f540457f381f1ff394060de2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 09:50:22.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7fdc197cafe84ebda6160757b4da2079\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20e6d3f7879942a0a2ced36522305350 : SUCCESS in 7m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/6713fbb2b5ce462e90eb1fa478236a88 : SUCCESS in 44m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bde077345cb14b498af2c13fbc5edf54 : SUCCESS in 40m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/827c66a9687049da9ddaa0dc26e06dcf : FAILURE in 36m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6eaea206bd274d0c9a9706dc6094e817 : SUCCESS in 12m 49s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fa0bd37bbb9f488b908ba822965b4e06 : SUCCESS in 18m 13s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/185be463477e4cb3a82b9cbcd04f52e6 : SUCCESS in 18m 04s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ceed014c51b340138cf8869a66a0f29f : SUCCESS in 4m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/881b67b51b3141c99189d996502c8e98 : SUCCESS in 18m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/23042d58e51047979e370ed1f8c43983 : SUCCESS in 18m 31s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3908e0d49b9f43f4a11c423fb0a783a6 : SUCCESS in 17m 09s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/cfb5b954847641e984ce3ee34df983a2 : SUCCESS in 19m 50s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b7200587d9674bde868c50b1c25dda9e : SUCCESS in 16m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5b3a2fc242df4a1cbc8de5977abe9b26 : SUCCESS in 17m 12s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ccc6e962c12546029beb766d571adcf1 : SUCCESS in 12m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c4f37b3a545f4bcbb08719a4cd7ae583 : SUCCESS in 15m 48s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/44f91d71fbee473f94d21be8bf730c44 : SUCCESS in 17m 18s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8533386c74244dea856ea2c927517c85 : SUCCESS in 15m 31s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/89594834677b421dadc10888f1faaa4d : SUCCESS in 11m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bbbbc4395cce46a3a528ba817aa79e8f : SUCCESS in 6m 11s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ff6c8a5aedaf42699ba2ffdfeaa627a4 : SUCCESS in 12m 59s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f481f8e5da43430ab982bb8a4fd22e1b : SUCCESS in 6m 43s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3692d655fbbb48029d22c0298ff0f5d6 : SUCCESS in 52m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/614f536a35c74c138c90f30ea1c6db1c : SUCCESS in 53m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ae3afe2bb1cb486588b2a7a4c603c44a : SUCCESS in 47m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3791cb6d891c4ad695f64b76b17de6f8 : SUCCESS in 19m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c6e9ba2a30c4117b1ce0633cddab7cf : SUCCESS in 4m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/59689c7ce2d34b65a10b515e1787dc33 : SUCCESS in 15m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ce3ff806003d4729896557e213fe90c3 : SUCCESS in 52m 19s","accounts_in_message":[],"_revision_number":18},{"id":"2086a1f4525099d2e94c428e7fe19afa8fdf9a39","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 11:09:38.000000000","message":"Patch Set 18:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8e9c1e13d3b44378b27eaaae6ad6670\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0d0f40911cc941e7b2531de6b39ead02 : SUCCESS in 46m 09s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/efa82c5108494697a3fcd4f0779be794 : TIMED_OUT in 1h 30m 54s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/aadacd699ebf4518aa7257e0bfc27953 : SUCCESS in 58m 01s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/9a98e3a7589b4f97a2b7ceac9a46a092 : SUCCESS in 1h 15m 10s","accounts_in_message":[],"_revision_number":18},{"id":"a6d1ab6198a8d93c528bf2d3c86c5c83dbabe5cf","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-09-11 13:05:11.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"a1595c06da497d984f44d5b2ce37dc31d7c0c2e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 14:10:53.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/060bb5070e1f4b21bdc8b3d26ca57e28\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd70f7166f164ec9800378aacda3515c : SUCCESS in 8m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/d96e5e5d4ffa4fb9a3e6b9739ac1ed6e : SUCCESS in 56m 40s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bca56610f30f42c8b249ab2495108fa0 : SUCCESS in 47m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e08596d7c8a0421ea031f8bbed04bb00 : SUCCESS in 58m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e567e77538314857a65cae8435ee8d46 : SUCCESS in 23m 47s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/33d8609ef90c47deb53b99d32e36c932 : SUCCESS in 20m 29s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cb459863d6564ae9941319382c843b6f : SUCCESS in 23m 47s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/69465900b47e4558bce08a35a1dd6120 : SUCCESS in 8m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/512ad9f1c97243bfb59bd0210f1d4d55 : SUCCESS in 19m 15s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/851c59a40b3d42568ec0f4d842deccf9 : SUCCESS in 19m 52s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c4a7881800a542d4af9d6f5b445ce772 : SUCCESS in 22m 33s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/3b10e36da4cb4b37a328a9609e709e61 : SUCCESS in 22m 46s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/57a0e6aa92ee4582b6a4f758811f0a6f : SUCCESS in 21m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8bb8aaf037724138a21f9bfbd04edf24 : SUCCESS in 18m 53s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8a1e7ddf818844ba8058e1434864fd90 : SUCCESS in 18m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f9abb1bcfe374ae79a9be542ef9085ff : SUCCESS in 24m 07s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2bff9258be3d4301a9dd20c633d88c8a : SUCCESS in 15m 29s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5de97da2188f47ebb8aee5ee59039c03 : SUCCESS in 17m 34s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8c21c4f3109649dd8acecb203ea32901 : SUCCESS in 15m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/480407fce45042f4a2e02821934542e7 : SUCCESS in 6m 15s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/902af4bfbe9f40ec994bfc491015bf09 : SUCCESS in 15m 38s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ec4bbe2db99848a9bafac1a9c55904d6 : SUCCESS in 5m 14s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0e8a7893be1a414d9ffbb10afa0492e8 : SUCCESS in 42m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/829c2906ec8a4ca987595281e1a70cd4 : SUCCESS in 41m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/17c4decfd1c4405cb3e4fe82bc484314 : SUCCESS in 30m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/20d56b9f4aff40b3834aa75b568d18d9 : SUCCESS in 11m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63652ca536bd48e8a44581df252f4b59 : SUCCESS in 4m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ebb0f5dbc9ac424097680232ac0384c0 : SUCCESS in 21m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/decf2d222a2f451c8db48fd2a042835d : SUCCESS in 58m 18s","accounts_in_message":[],"_revision_number":18},{"id":"673a972099d2f651f3c8bcfc328513e619e1cd2a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 16:04:35.000000000","message":"Patch Set 18:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d042bdbe20c4981b401e824f5d2a07b\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2661195785a2454cb7c30e654d98bd00 : TIMED_OUT in 1h 31m 05s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d148a2a464b4498c82031cc051190096 : TIMED_OUT in 1h 31m 07s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/6129d58653da4805ac3d696fb684d2cf : SUCCESS in 1h 14m 02s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d5a52a5197fd404f8b64130a55f42514 : TIMED_OUT in 1h 31m 20s","accounts_in_message":[],"_revision_number":18},{"id":"00f60b2cc6a7575c012e188f536a3a2a13c62981","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"real_author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2024-09-11 22:00:32.000000000","message":"Patch Set 19: Patch Set 18 was rebased on behalf of \u003cGERRIT_ACCOUNT_7847\u003e\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}],"_revision_number":19},{"id":"704127f86b19b7d9d808823b6530eaf3733f6e55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 23:11:39.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/baa316ff39374f989b0e905f61d95db8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d80c5c452bc1420897c87f0a3676cd52 : SUCCESS in 9m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/3286d6ea62a346c69867edc7fdf5758b : SUCCESS in 1h 04m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/446acd785a994ce2bc3e7e82d66029bc : SUCCESS in 45m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0b2ceb034aa84366a9fbc5379dedf1ad : SUCCESS in 56m 01s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/acdc3f60559348bea71b4da1b7f4121f : SUCCESS in 17m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/03381300cede45a4b1d6fad046183101 : SUCCESS in 21m 16s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cad8e1160e054c33bf64655cf4e3bbda : SUCCESS in 18m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d5d55e86e9ce4c95b1e321d282f0bbd9 : SUCCESS in 5m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e3ad754c726b4466be325cc6bfadf4a7 : SUCCESS in 20m 17s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/81238760a57a419187c39195db5c5fdc : SUCCESS in 10m 06s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/2f00b7a7199048d1b5e06f1c3aaea59e : SUCCESS in 16m 22s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/56942efd4ac148a097f499b5e9d16296 : SUCCESS in 19m 39s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5b6e9cdbcf8d447f8a198e5d3e8aa9d7 : SUCCESS in 13m 52s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4f747ebf6dcf4d03a8fc7855429ba96e : SUCCESS in 18m 27s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/97712582c52c4349a4b0cf03355a7155 : SUCCESS in 16m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0ef009c01e6647299ffee0f094b32f11 : SUCCESS in 21m 54s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5ac040dc1f744841814e3ad7ae2f0bc7 : SUCCESS in 15m 25s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/56a998444cb143bd8c96dc4c18747c42 : SUCCESS in 19m 39s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1ce764ea49d94d8495610b220a38ec9c : SUCCESS in 11m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a762cf084e1848f79298ae49014797fe : SUCCESS in 6m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7f85189e40a94332b6368cf5fb63029b : SUCCESS in 9m 45s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/976d388112ca48d1b9a023776826a9b3 : SUCCESS in 5m 33s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5a117029821449038b7df35d4e5565b3 : SUCCESS in 50m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0fbb76bcdad848dba60c650d286d86f0 : SUCCESS in 53m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f6425bb591e54fe589e704329ea9a3be : SUCCESS in 48m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e131e4d623b04d16ae6191fc056d4bd2 : SUCCESS in 21m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33d0af72873c478dbfa805e2ac1f8127 : SUCCESS in 4m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9c940fe6b00b4fcb927204506f629df5 : SUCCESS in 23m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/046419785a2d4163bb38753fe5c85732 : SUCCESS in 1h 02m 56s","accounts_in_message":[],"_revision_number":19},{"id":"151a056eab7ed9fa06be292f693aaf99d8d56703","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 00:57:18.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/11e75d65a7ee48e392900a665adf44dc\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/91db2140645f4121a7314783644b6a48 : SUCCESS in 1h 02m 16s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ea3071e9340a4d90974ce7177134f6d7 : SUCCESS in 1h 04m 44s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/36e7166a9b2e46cea8798383c91cc467 : SUCCESS in 1h 13m 29s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/499294454c394f2f851b3e358cf83b09 : SUCCESS in 1h 12m 05s","accounts_in_message":[],"_revision_number":19},{"id":"ec4b8927a38c54afc6d1bcf4256f1d2e18d40f92","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-10-02 23:08:47.000000000","message":"Uploaded patch set 20: Patch Set 19 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":20},{"id":"82f6c88d4ff9f6c75ee73cabc113fc0be9982a6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 00:23:56.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/efad6d07e0a949588e21fd134a633b89\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4e0258ec1564e929ccc9261ec3cdf53 : SUCCESS in 8m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/1c6f9f93928840b3bf375741355fa205 : SUCCESS in 1h 07m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/906c74c890c64b31978fd13c3b0f7a19 : SUCCESS in 49m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/678e1da2f0874ea8ba97d73732e3b65c : SUCCESS in 44m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cc95ba7360b045f4aa7f390bfff86e0e : SUCCESS in 24m 58s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1fa287afe33a4c518c4e5001651a6dd1 : SUCCESS in 10m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1b085c370d2843cb884f884e3530c9f1 : SUCCESS in 18m 25s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0e039f127eba43159c3917fce40dd834 : SUCCESS in 17m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b0ce76d5df7a4de7886048228d349e3f : SUCCESS in 8m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/98297416b8ad49779b6c36563da128b3 : SUCCESS in 19m 54s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/37cf7c7688ef43c0be00de7fd42a1ac4 : SUCCESS in 9m 26s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/55bed0d66d144570bd4a5e2a18c19d55 : SUCCESS in 22m 53s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/e79eb08f35394557b57bb65ea01fb78c : SUCCESS in 20m 28s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/dc2a568b35a6426e9cbe97d88f6aba13 : SUCCESS in 27m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1c78cd4194944f97b3f2451946941560 : SUCCESS in 15m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c4dc8a809e6546a4a09b9e042f1bffdc : SUCCESS in 15m 35s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a86b0b299a034ebb8c2716bad5cb9baa : SUCCESS in 16m 17s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/75793a7bf4de4c75b21a4ff44f652e82 : SUCCESS in 16m 23s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/82476e0ca12f4401b5460a4f5b592828 : SUCCESS in 17m 08s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1d750764a90d491280dc12dd6d0b368d : SUCCESS in 13m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/086d2a12cf3a4ef9b0c9c11b1bb079bd : SUCCESS in 6m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/48630ecba5684717a9426728c4413ca5 : SUCCESS in 12m 53s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d1b6200c47bc4e34ae38591fd552f661 : SUCCESS in 7m 15s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c4a37c2e64464fd19a00b5445a1a2651 : SUCCESS in 49m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5687beb4287942f9926451f1743c3599 : SUCCESS in 28m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fbc4b3dc780f453d8b20ff99cd2e9566 : SUCCESS in 46m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4a3ecd91482142689d2fac6c760c7f4c : SUCCESS in 17m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd42fd51985d46d8afd5d5c7f0f4462b : SUCCESS in 5m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8de790b118a24868984b74ee5375f505 : SUCCESS in 16m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ad3c0f925b7b481c8ae3ab18db723096 : SUCCESS in 45m 10s","accounts_in_message":[],"_revision_number":20},{"id":"32009c7213e5814e58fd1f9c502b767b9cf71eff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-03 03:54:38.000000000","message":"Patch Set 20:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8297e29bed7a456cb85491054e39e10b\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dbf5c4fec72440da92e4b96ee6018c59 : SUCCESS in 1h 25m 42s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f22abff1a2574acba3a72fba4168eaec : SUCCESS in 1h 35m 17s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4c9b875bce5e4fcca80f7856a2117506 : FAILURE in 1h 24m 36s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ef20917b873e4444ab397a7c180a57b1 : FAILURE in 1h 21m 55s","accounts_in_message":[],"_revision_number":20},{"id":"988657af3e53b07a2e9dc7059fd41aa9612245e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-10-04 21:01:35.000000000","message":"Uploaded patch set 21: Patch Set 20 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":21},{"id":"62cd85cacc27b97f42e4cf8af4f91044b71e9cfa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 22:02:42.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a7f59c43cfa41c392bbc1c1244b3703\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf5071dc529f44c58f75fea3c5f3eac4 : SUCCESS in 8m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/29aa343c3ba2484e88448f797977f725 : SUCCESS in 53m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d0070fbbe54941d0904ab2695cbd7804 : SUCCESS in 50m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/51e3de0512634ab5a062aed80e61dbb0 : SUCCESS in 41m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c61478c21ee94ba6b8bbb8ea3b7b7880 : SUCCESS in 49m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2f959663735a40e3b1dff90d5245baef : SUCCESS in 16m 04s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ddb8609ef780414e8c7c275edc458cb9 : SUCCESS in 17m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c1e438368845430bb3aaf21f060d5b09 : SUCCESS in 16m 38s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d44ff5cc8323415d89a5bdd6fc1ee46b : SUCCESS in 5m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5e78ed3dd08845748d6f0ef152401984 : SUCCESS in 19m 24s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2c250f5f591241418af622a2eb21f6c7 : SUCCESS in 18m 00s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/dd779637cb9d43d6bf568286e1571136 : SUCCESS in 21m 10s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/b1c6676b5f9f46c297158c60d0bd4eb0 : SUCCESS in 22m 44s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/cd0a5557ed1540b98793618c4eef4491 : SUCCESS in 22m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8575015171f3446f93cc3a6e10ff8ad9 : SUCCESS in 19m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bbcc08297b364ebba8063d5838673fd2 : SUCCESS in 21m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/80e9acb8e39d4c7598fb0bdd7acac461 : SUCCESS in 17m 36s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c454850393074e7787e6f4962aadc9b9 : SUCCESS in 18m 47s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ea5dd5f816f24e84940278afa56e4f45 : SUCCESS in 15m 06s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/22a23e8e89234f3eb077f588f7c8f12d : SUCCESS in 17m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/774ce8d5980e455d8c699cb767772e32 : SUCCESS in 6m 26s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a2b0c27037cd46219b4040568f787889 : SUCCESS in 13m 46s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/37d5df100184418286edd04c2f2b35b6 : SUCCESS in 7m 32s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1fa0e03d368c432d842610835dd3cf58 : SUCCESS in 56m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ab0359e7763743cb8df2060c025ff5b7 : SUCCESS in 40m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/619f526a0a8846fea68da829a443825c : SUCCESS in 29m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7aa90983fd314318941929136d64fb12 : SUCCESS in 20m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91c9397bd22948829f81b2228f68ec81 : SUCCESS in 4m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4b8856c5abbe45ccbb0cd9a396225905 : SUCCESS in 17m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/96c2c9d642644f89922f3e43c846c042 : SUCCESS in 48m 49s","accounts_in_message":[],"_revision_number":21},{"id":"52bf642d8a6c32d4372ba4d27cbbeefb035d5e92","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-05 00:10:29.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/680e274d6cc84157bf5eed3a25fa3635\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6700dc745a9f4eefbc0c5db85007738c : SUCCESS in 58m 49s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/717d08cf15304a91a398b99eaf2ebedb : SUCCESS in 1h 11m 14s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/012ef2ab5c4a4e4e8572aed1bff4e06b : SUCCESS in 1h 35m 53s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/2091f741c7ae437ab360cfa12a6d925b : SUCCESS in 1h 45m 09s","accounts_in_message":[],"_revision_number":21},{"id":"986c4767d80f1dc5d5871f08f9af24af09c32113","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-10-24 21:11:26.000000000","message":"Uploaded patch set 22.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":22},{"id":"2b61eba18f4e9ba4826ccabe3ef1e91d3289a888","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 22:20:34.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/80278c08a4e2472089e4526412d73f7a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d23ed5fdb70442199f14722198e6a1c0 : SUCCESS in 8m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/ec28ba32a4284359922b578b9e460ceb : SUCCESS in 1h 02m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1f32edd475714b978612805eb6b75fc7 : SUCCESS in 56m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b96017a039464075a0e23a34f5d86214 : SUCCESS in 50m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/601999202a7c426fbacfffbdaa5cd27f : SUCCESS in 1h 00m 29s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dd555523ad8f445491018ca4fcdeca7e : SUCCESS in 21m 20s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/69ce82b9d64a46eab8a49048502cbdd4 : SUCCESS in 17m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/737fe278ea5140b5bd006f77db31be2a : SUCCESS in 18m 28s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7d822ce95207493b9acadae80ab32489 : SUCCESS in 8m 41s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1904d10ff2d9474fab5c67103fe6bd2f : SUCCESS in 21m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/185d2f30ebb64b29b6c8539ceb00e170 : SUCCESS in 17m 58s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/86f6fedf92f24126b219f8a548f2a2db : FAILURE in 3m 05s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/8dc09544b54b46c8a74469fd3e7bf6bf : SUCCESS in 11m 37s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c0604d1295574a8e8b03250e9d122e55 : SUCCESS in 29m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/24c59868111a4a46802e98b325fa7c62 : SUCCESS in 11m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/efe73d7602204bb1b3bb0ca8dde67c9f : SUCCESS in 16m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0c4dd780de984a17a22c6a57598b24b4 : SUCCESS in 12m 11s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b20997d8197d44679783fa11ebad0627 : SUCCESS in 13m 22s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f77b57e1c96d43e6880174569f11c3f0 : SUCCESS in 13m 56s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/26ea59fbfe9f4dcbacf5ba8d4823d9da : SUCCESS in 10m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9f8f0e6069f14643a2d9ed4b92b78b0f : SUCCESS in 6m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3e0665e63fe84a6d8fd812cc07ac3cc4 : SUCCESS in 19m 12s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c30985f9301f444ea6dd71cc84a9ecf8 : SUCCESS in 6m 29s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6552abb7e4f0469d8ed643a197d3a456 : SUCCESS in 43m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/55e73bf0a5544b80849e37aeaa7822ed : SUCCESS in 32m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d13e91f6991647548290d37a908d434c : SUCCESS in 55m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/44541964c06543d6a4980860aba1c99b : SUCCESS in 19m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2b77f541d10482d8fca87985afe9a47 : SUCCESS in 4m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/099585d14ed34db3ac8ee9791418025e : SUCCESS in 23m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/08e2eae93bfb4d68bf8198488fcb731d : SUCCESS in 42m 36s","accounts_in_message":[],"_revision_number":22},{"id":"b95155e015adaba2dc375fcac711f05231be5c6f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 22:36:31.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/419a61e8bc984afe8ff87945935726a9\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e71ec8725c3a4f53ab50c4e83b43b33f : SUCCESS in 1h 23m 00s","accounts_in_message":[],"_revision_number":22},{"id":"c02205ac2d10423d7033e84dfc99ef779c6c882d","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-28 12:41:32.000000000","message":"Patch Set 22: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":22},{"id":"804102f9ccc56893bda6a85c14f55ef7de54b59e","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-10-28 22:11:05.000000000","message":"Patch Set 22: Code-Review+1\n\n(6 comments)","accounts_in_message":[],"_revision_number":22},{"id":"b3eb3ea52ea7432c237ec85d11f4fb89cb6c4733","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-10-28 22:11:18.000000000","message":"Uploaded patch set 23.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":23},{"id":"ac97baaf35cbacf348e6aee80a3815f3c48534ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 23:25:39.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f60417178d14f488cee72f72e88ebd0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2208761ed9fc4b2eb3efe184c2812d14 : SUCCESS in 7m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/607c482c51b3433d8fabd544917b61d3 : SUCCESS in 38m 33s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/34093b42f0394f48bbbb23547e3667f4 : SUCCESS in 37m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9476260777ba49c698c4f077219d9e75 : SUCCESS in 1h 01m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/65d9429a59d4486089063e844144b2d5 : SUCCESS in 1h 07m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/31fe1f2471d34d708e5c044a5f86677a : SUCCESS in 16m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ca3b7fd03ba9436881d731f9d03230d0 : SUCCESS in 18m 24s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/591b90740740444198e01b467f4e9feb : SUCCESS in 18m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a81078bd3e5c4d04b04862b1a1ddbc21 : POST_FAILURE in 34m 48s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b748a79f6f6a462ca6faf89f4a862c4c : SUCCESS in 16m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5f9ead864cb243c8acaac15a0c7e26bb : SUCCESS in 15m 11s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9a14006c11d54398a41e2e930fac8bee : SUCCESS in 14m 47s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/a55ccd95ccb94507aba5ee51eb97ec62 : SUCCESS in 22m 33s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d729b9fd4e614c219568943d8bc8de84 : SUCCESS in 13m 16s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/826292e07a5a40bf84b578e3cfec57a1 : SUCCESS in 15m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2fbea6ead86f44738fcf297413ca446d : SUCCESS in 10m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c5b2e8681c784061b07a2c4a76112e1d : SUCCESS in 15m 52s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e220ce9e06274b13b1aee720cece4818 : SUCCESS in 8m 15s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f3c7237ae8d94c1c90873e4a9f02788f : SUCCESS in 13m 36s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/172ff6a67d804ca0884b48304c2dddf8 : SUCCESS in 8m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/89a25bfa70c14fd085248352b86589fa : SUCCESS in 5m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f35f014e530c4c88abc29b7cfed0808c : SUCCESS in 9m 24s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e16d517945cf4629b647594c37505529 : SUCCESS in 5m 49s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6d55ef968b5e4210af4b5538449af72f : SUCCESS in 33m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e19e7adacf1b4ad39e3afe71119aaebd : SUCCESS in 28m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/dc4eb222e19f4c56b7bf002f63bd1532 : SUCCESS in 33m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da21614085154a21af2f4014aac3a4bc : SUCCESS in 9m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ff51bd253704a3690f508e9bea1d21e : SUCCESS in 3m 48s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f2cfaec8d0e043e5b0f27a16a6a907ee : SUCCESS in 10m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/827953009b58428d88bba5509bca4048 : SUCCESS in 57m 59s","accounts_in_message":[],"_revision_number":23},{"id":"9889d51b6fd9551ff1e4f9ca3ed598c591df7f0e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 23:47:51.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e66ccb295745477785a0d5479a66aecd\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ee03291cd48c4eb6a8435c400455e964 : SUCCESS in 1h 34m 44s","accounts_in_message":[],"_revision_number":23},{"id":"51d25ef595a99b441fb515378de96111e2edcd68","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-29 09:26:48.000000000","message":"Patch Set 23: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":23},{"id":"832a0989df50c98bfdea0deae4dd97e98cb67334","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-10-30 14:06:36.000000000","message":"Patch Set 23: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":23},{"id":"4c31a2582e7aae5b2bb5159a48a2fa0f7142abb5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-31 11:15:16.000000000","message":"Uploaded patch set 24.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":24},{"id":"0e05111f20bd14b2960faea0b331283101a9161b","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-31 11:18:06.000000000","message":"Patch Set 23: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":23},{"id":"3d7840efac61b954a47d0dc0b8591892eab768b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 12:40:48.000000000","message":"Patch Set 24: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/54923aee18b94ad8969df018c2236e14\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfff8fab4f9c4741a3b010662a608582 : SUCCESS in 8m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/e5d18877ef774428bf87e46bde73d205 : SUCCESS in 54m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9f3bcc20ff6d426eb6247220e9b50bfc : SUCCESS in 35m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b9c7e13c4fa64e0799f8d253f5c9ad2e : SUCCESS in 50m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d03c660ea5ca4f2184e560fe59dea505 : SUCCESS in 47m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/68928b4e176c43079f2ff65e5bc55527 : SUCCESS in 10m 02s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c090ce5fc62346e4abd00c99398aebfe : SUCCESS in 19m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/00a9cb0cacd64122b1bd31698423f892 : SUCCESS in 10m 32s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/826c0adb80ac48eeb7a07de3ceb3b319 : SUCCESS in 4m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1de66a5f06ff46d2b5d40eb09c198783 : FAILURE in 22m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/24d5293c99d0420eb00e43d2afb70ad1 : FAILURE in 19m 35s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b1845f086e714563b193148b1514f374 : FAILURE in 17m 00s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/8d2bc4343aca4aef968690c341a334af : FAILURE in 9m 54s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7f636e1ba2bc4527a5e8b5e5234c27d1 : FAILURE in 37m 44s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/7322f941e1424faca89b28d11ee3a14b : SUCCESS in 10m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/da0ae0e0bd0b419db28236b34d5ffa33 : SUCCESS in 17m 13s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bc9c81af67cd4f15a96236fd8d95c191 : SUCCESS in 10m 04s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/3cd352b597ff44d398fa726facf6323b : SUCCESS in 18m 56s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a9ab02b6e3ea44e3a427373684a4af8b : SUCCESS in 13m 58s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/748c9ee7106e41baa675214577890d00 : SUCCESS in 13m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f4dc0874bfb3497d96d88d357a80adf8 : SUCCESS in 6m 16s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ae4e570739a44e4d9da1df3b03e52511 : SUCCESS in 18m 28s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c7a9676b35a14615b4e68def4595589a : SUCCESS in 6m 21s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6c2d2d3d73734b218c963f23b4c1e4c1 : SUCCESS in 52m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4fa185193cef42b6b9466227df46cb8f : SUCCESS in 36m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/74a7db5ad2924579814023b7e67c65df : SUCCESS in 28m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c433fc55153f45afb40cea2cc0c2c258 : FAILURE in 20m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c47623b877fc4f848203323ebb9f637b : SUCCESS in 4m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/05457bb5c4af48548e61f5518a9f5137 : SUCCESS in 11m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/467d365753d548d38edb3e6f5f7b52b4 : SUCCESS in 52m 36s","accounts_in_message":[],"_revision_number":24},{"id":"7ccaf542263849e07a0d79ff38fc31e68c08101e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-31 13:31:35.000000000","message":"Uploaded patch set 25.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":25},{"id":"7e8b5245107293fbd1912dd0ea8014d83b851fe5","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-31 13:32:55.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"403870d35ca385636a1207005ecbce67138fe3ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-31 15:21:18.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b6918e591f043dcb7cd018cd4e7e91c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce17e4f197834056aa6b9d52acb297d2 : SUCCESS in 9m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/4ff3e40087c44adb9f4445fe76135649 : SUCCESS in 58m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/663308636dfb48c3b2092035c52a651b : SUCCESS in 56m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9e065be854604b3b90f2c14b462a5e4d : SUCCESS in 46m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b4eebda66d5843f4953cc9d0d73732ad : SUCCESS in 1h 02m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e40ff9a0ca864b29a5ff7bc85acb9319 : SUCCESS in 17m 42s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d5eae765c0b240008c71758ea0b68d91 : SUCCESS in 19m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/807a5260dac8460095dad90f786ea2c1 : SUCCESS in 17m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b6082f2e17704e8db8d14be995089ed4 : SUCCESS in 5m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5629f4e5af1c48a781c339010b390ade : SUCCESS in 20m 27s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ce6cf119dbc94ad0b810283468ddfa67 : SUCCESS in 19m 10s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9723b6f64ea4464fb8d546037a776bad : SUCCESS in 18m 27s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/88bc078360574a228a6dbe6e9ff96e9c : SUCCESS in 9m 50s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/cb0a7cc957d146c684f4999aea7ec025 : SUCCESS in 30m 28s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9cfe582c43d04ba2ac27fb4e58af12b7 : SUCCESS in 11m 29s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a6f189fd46d24411b9092c6a7d9ab4b6 : SUCCESS in 9m 51s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6410d96510f440c384ce3f22cd4068a3 : SUCCESS in 23m 21s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/e30058f13e4d4c1684587bb2d51f6b8d : SUCCESS in 15m 40s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ebee2f1b6206408c916bfb54ad972390 : SUCCESS in 14m 17s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/946431102aa540acb969112970b84077 : SUCCESS in 14m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/425becc5fe5f423ea9e7a70a366b7b95 : SUCCESS in 8m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/da2198e2e80c4ce0a0344c7fa2075654 : FAILURE in 22m 37s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b2baec5b0ee543669f486befbfc3e2eb : SUCCESS in 9m 14s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/eeb88347001f406592b7e30f7ed2b212 : SUCCESS in 42m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1e7a783d289a4dc5ad2ec8fe4fb3993c : SUCCESS in 35m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/976c7d28c56c4506b62bf1b262695b8a : SUCCESS in 29m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7bbb1387b92041b2b47ab47266c50891 : SUCCESS in 19m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efc8fadc23cb42c19a6a1be9cc741da3 : SUCCESS in 4m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f922c016f5684e0591cc6ab45d63772f : SUCCESS in 21m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/db28ff6113df44089a933da324bc5249 : SUCCESS in 58m 32s","accounts_in_message":[],"_revision_number":25},{"id":"870b3ebf0be2d4452c444edd03fd99d3c8cd3bed","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-10-31 15:41:05.000000000","message":"Patch Set 25: Code-Review+2","accounts_in_message":[],"_revision_number":25},{"id":"0c0cf07b27a1b36cefcfcce4febc193f45ea7ad6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-01 07:16:53.000000000","message":"Patch Set 25:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d1429f8e2204a7d848f7d643f3134b3\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bfddd27493d04f2f881019024394f82a : TIMED_OUT in 3h 01m 17s","accounts_in_message":[],"_revision_number":25},{"id":"0effe31e4dda1b1e6ecef6fa755a3493c602a218","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-01 09:55:05.000000000","message":"Uploaded patch set 26: Patch Set 25 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":26},{"id":"3119e6bea287f665318002525ff030e53a60863a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-01 11:13:35.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/649aef1e347e4a15a1accdfdff400bd9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fbb69417f9a4f8997e4c3451ecc0d8d : SUCCESS in 9m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/9d2d18e0cf6145de87a969e9fd38817b : SUCCESS in 59m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/abd9746fc28e497b86ece02a2e1a6a6f : SUCCESS in 50m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a1cd405bda444a009fe20fdf73b064d4 : SUCCESS in 45m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6dac1ad4e9424fdf969a0cca3958bc30 : SUCCESS in 1h 01m 02s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d3a2e7cb2e724ae6b196c0ce21e2562f : SUCCESS in 19m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/66801e85d393414c823fe04a917610d1 : SUCCESS in 21m 59s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d2bba0ca64a64eb0a79950832374a96f : SUCCESS in 21m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e4415037dedb4f62b46ffeaadc9aed87 : SUCCESS in 4m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/37ee3d737cec44e79c37fa68952bc5a7 : SUCCESS in 20m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4d8914271e5941809b49ed0abf63da26 : SUCCESS in 18m 32s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/bd337552fa254845ad00c4c75dd79694 : SUCCESS in 21m 19s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/95d4e3c1536b4e97968e3ec4e70ed230 : SUCCESS in 13m 52s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0bf8aa6c551c4353969a16c205194729 : SUCCESS in 27m 36s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6ec02cb6d7774eaaa79b6c8d8ad58c05 : SUCCESS in 16m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bfb239cbfa834d61940f54d3eaee88a4 : SUCCESS in 18m 39s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/33fd81227b2e4e03b1a68e0fb22f0d96 : SUCCESS in 11m 34s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/fb3ad0ecca5941ee858aee8c7d1e6206 : SUCCESS in 15m 21s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/180a413838ba49a795b240e59d2c9335 : SUCCESS in 14m 00s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/710e7e1721704d328838b9b8eb4cd55a : SUCCESS in 16m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cf7b0ecf8c0947629cf44e9cb9db9194 : SUCCESS in 5m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8cbf43cc6bcc45c2890f09e397286a5f : SUCCESS in 18m 28s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f26cf0711b9d4944bd9473faad5fd10a : SUCCESS in 8m 16s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4004a59813d04bd68f1d51f9587da1d5 : SUCCESS in 38m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/016427c16f0f49f3b5953437341011a1 : SUCCESS in 51m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f885a036567542598e71e6a5cd502335 : SUCCESS in 49m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f72cfbafd9564b6887b250485fede44c : SUCCESS in 18m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/451cac3a463248dd8d558f5d1add74b6 : SUCCESS in 4m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/98341eea17af4118adbc75d189ac51e5 : SUCCESS in 12m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a7b80ee7a05e4cdda6174ebf615ced64 : SUCCESS in 1h 01m 09s","accounts_in_message":[],"_revision_number":26},{"id":"43f0578289fc811a3c9d56e5e8751e78a7c50f9a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-01 13:44:46.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1cc0fb5ac7f430a9e8873cd89517771\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/37471db0cffd4c26960641518cb009a0 : SUCCESS in 2h 32m 45s","accounts_in_message":[],"_revision_number":26},{"id":"9659ebf4b5d4583f5683964e95ec98388c7675e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-11-01 20:13:43.000000000","message":"Uploaded patch set 27.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":27},{"id":"92719d7ea1286e5af0e755a911f9fa926b3013eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-01 21:19:00.000000000","message":"Patch Set 27: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/425d258b7d874f8a9ee6251859fa2435\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7378580e383742c4b6e1db75ce2e9208 : SUCCESS in 8m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/3154df01fd95412ab323399c4c518b1a : SUCCESS in 50m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/aee4a08bc99b4afcaba752ea801b620d : SUCCESS in 47m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f72e6b14203f47fb901b77a86601bc01 : SUCCESS in 39m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/16d96fd2ea8d4a03a9f15f2c4486cea6 : FAILURE in 34m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/45731920493a42fb9234b2f8ab024616 : SUCCESS in 21m 19s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9d1be81373be4cc4a96d6001107be58a : SUCCESS in 21m 43s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3b09ba575b9648d7bdc984480f954227 : SUCCESS in 23m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5a76e07025c9488a9e80134d716f3d6c : SUCCESS in 4m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f3a14534ddd047f2996e1f73371fbf49 : SUCCESS in 21m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b89316a570a940309c09d3ab9c6db038 : SUCCESS in 18m 19s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/844a272dcf604fccac19c30842845b16 : SUCCESS in 9m 24s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/830c60e3981648a19ba39807f1a791a7 : SUCCESS in 21m 01s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2fc50aebfbce4511bacb82e3c4f4bf20 : SUCCESS in 20m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3817795e5b4c462da49b609abeef818e : SUCCESS in 19m 01s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e8f30aa731cf47be9eb1f6ef0cac10c1 : SUCCESS in 16m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8dea0e6d2bdf4c1f87a5e784bbfd3811 : SUCCESS in 11m 26s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/86a346f6c643465cb0fd47703642372f : SUCCESS in 14m 34s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/2007defdc29344329f1f23869cbe2364 : SUCCESS in 17m 57s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/99a4b46855fd4c2f87982d33f3bd8b6f : SUCCESS in 15m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/11f77ab1ec134c108ef355bd5a34d05c : SUCCESS in 4m 30s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7139676e3e374aea8e73776c27a8cf3e : SUCCESS in 17m 33s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ac0e77be878f427190734d466ea2e922 : SUCCESS in 4m 54s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/568ed9572ab74d308877fb19b06b0869 : SUCCESS in 36m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3a76ad3f3a7c4576a9c91b36da5eb90a : SUCCESS in 39m 42s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4463217248fb4997bfc7c64a6c2e62c5 : SUCCESS in 56m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f11c5c0fb705445b9ea77cfbc1c49b4d : SUCCESS in 11m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9be03c05f17744ffad2b14c3020a4a54 : SUCCESS in 4m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/02e9a2506de24c4ea77585f051853a5e : SUCCESS in 19m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/244e6452f64a442da8a934606e7c24be : SUCCESS in 38m 47s","accounts_in_message":[],"_revision_number":27},{"id":"12adff2a771a5e59e495ec6d7ea55e28557d5245","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-11-01 21:27:32.000000000","message":"Patch Set 27:\n\n(3 comments)","accounts_in_message":[],"_revision_number":27},{"id":"ed2cb843f69c57e0cb32cf6ce4b2a7d8403bf05d","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-11-01 21:39:57.000000000","message":"Patch Set 27: Code-Review+1\n\n(12 comments)","accounts_in_message":[],"_revision_number":27},{"id":"2c7aeb0328ca5a34da2657434fc46086ddd092a0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-01 21:58:08.000000000","message":"Patch Set 27:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff3a39ced9c9453cab824d36de82d9a9\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9c5392e5ff6541b696e30560e4fe5140 : SUCCESS in 1h 41m 13s","accounts_in_message":[],"_revision_number":27},{"id":"376fa6dc03036bace58062c3a9712659a7fc4c8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-01 22:43:12.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2dbb263ee0b1414592f69d5acbd5c23e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a22c82a1881407289404e33cf259086 : SUCCESS in 8m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/b6e8a1d63db9407eb506ec1134cce778 : SUCCESS in 59m 42s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/56c304a22bc441939fd00f3a67fd065e : SUCCESS in 1h 03m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6e6db704d0df49648f98d61634161149 : SUCCESS in 47m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cb072d8490b5463c9b0cd9d0f4e71b1b : SUCCESS in 49m 55s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6993641fef1343948f23946e65772e9d : SUCCESS in 19m 21s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/96667d9881f745309bc0a67fdf2c04e7 : SUCCESS in 12m 43s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/78f09dcb582e42ba9f7e8614f6b36684 : SUCCESS in 19m 52s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/222f7dda92ed4998bf38ab0bdd4c8390 : SUCCESS in 5m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0af8064a74c3484cb6490108e87a32a1 : SUCCESS in 20m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7ad72af1614747c8b708edc8d3328ce4 : SUCCESS in 16m 38s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/83336da92ca5412e88a8df46942e40a9 : SUCCESS in 15m 01s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/ec2ca135f01b4718a6ebc65add44d14b : SUCCESS in 28m 27s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7616eade1ff5472bb35e98c402e95983 : SUCCESS in 31m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f0a9e412a3b34fb6b9c4d40ec18ab43b : SUCCESS in 15m 44s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ecb00709fdb34b76bda8891dfa1e73cd : SUCCESS in 15m 04s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9c08af81d04d4c8294adf9b62c63753e : SUCCESS in 11m 14s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3f60fb0911af4541a4a32117a30887da : SUCCESS in 14m 20s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5cc7a317392c4d95b97e5209a022e2b9 : SUCCESS in 15m 43s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/dabeb4751d764d449caf326a56c2b127 : SUCCESS in 18m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fbe0663060b648edb0cf0428222dad15 : SUCCESS in 7m 36s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4ff89f85673a40ddacf16fe2d49854fd : SUCCESS in 15m 37s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2a6dd0e8d38d493b80886cb60d399049 : SUCCESS in 5m 56s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/dcff60f94ddf48d9acd427c475e0d6dc : SUCCESS in 38m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3f03eb2b8cf540959aae785cb9ca137c : SUCCESS in 39m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/539278bd53e247c4ad44a68148d82d22 : SUCCESS in 50m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/143edf3ac2534123b9cc7ed049eb3b7d : SUCCESS in 18m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc2fb11d7fc344afb315d249181fb2b4 : SUCCESS in 5m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b8bacf5b585b42a49c876a23acb9f83f : SUCCESS in 18m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/33fe8abb7e7841768b4938e83777dc5a : SUCCESS in 46m 37s","accounts_in_message":[],"_revision_number":27},{"id":"b8a5331d4c780ae61da2189c4136848dc90fada1","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-11-04 00:03:10.000000000","message":"Patch Set 27: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":27},{"id":"73fa08aaf7890b98534281b04f025c10c89cefbc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-11-04 00:06:02.000000000","message":"Uploaded patch set 28.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":28},{"id":"036e2d36fc260aa18dbdf9734bc06a8168c92dc0","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-11-04 00:06:52.000000000","message":"Patch Set 28: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":28},{"id":"1650c87c8696a7c2e81a45b5d5fe606cdf1335a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 01:03:44.000000000","message":"Patch Set 28: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a953f62637cb44e6a3966974cea9c5cf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9205b18cb040495c908d34fcedb88278 : SUCCESS in 7m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/235134ec301b46c3b70b95cac0214cbf : SUCCESS in 42m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6aaeca26b6224ad98b67f6edf9c02eb7 : SUCCESS in 42m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/26ab3942f4b04e9facf26fa6306b609e : SUCCESS in 34m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8c5f1b3bc7ef4e37bd3c49d0335b48e5 : SUCCESS in 43m 17s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c9b58df63ecc40a488f934284f370880 : SUCCESS in 19m 15s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/19884b1a5bbb43ebabd80d4f5d3ae3d5 : SUCCESS in 20m 37s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b4b5c0cd1915418fa0921df201cd4c49 : SUCCESS in 19m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d1b2349c29c240cbafff13091ac4c37f : SUCCESS in 5m 28s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6e737a762ea34ad192fc9573ab931483 : SUCCESS in 18m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/52b47e26bc674ce9b69da7c377597be0 : SUCCESS in 17m 13s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/055512dbc3b449a1a25622fb75c15ccc : SUCCESS in 10m 05s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9a1d3496220341428a23acfa70c8f97a : SUCCESS in 13m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/25f90289633145d5aa29f19fae5e34c6 : SUCCESS in 12m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8f19e719127f4959b6474b250fd7b011 : SUCCESS in 10m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/264c4c3c987943918f014c88c472fe71 : SUCCESS in 10m 07s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e798ad8c0f044b4d8c89dd38d27ef82d : SUCCESS in 15m 41s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/31a464e2afa5425d871c414cd768129a : SUCCESS in 9m 21s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/77128d7049a74a34b408ea5d52969c84 : SUCCESS in 16m 00s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b9907928563c47bc80e05c2089c57511 : SUCCESS in 4m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/48314843c9fe46f1b6079732ef13921d : SUCCESS in 15m 09s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/af81aa6ce5b041a181fcceb081061017 : SUCCESS in 6m 35s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/72bde6914214460ea13adb7c43aaa532 : SUCCESS in 50m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/28b9ae39815144b88a5d8eac12a8ea0b : SUCCESS in 27m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/41661e6332da4e6e84c4c11d0a8ed1d6 : SUCCESS in 49m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/51c211b280654d4aa0369faeb89b9df5 : SUCCESS in 21m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b7cb4c9069343c4988e0c943404080f : SUCCESS in 3m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/104c5c9e019b43079849a5cff29058ae : SUCCESS in 19m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a0a7901f430d476cbeca53941facd3e7 : SUCCESS in 34m 44s","accounts_in_message":[],"_revision_number":28},{"id":"9b4528198c3e5cbf65e82303cf71634070162e0e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-04 01:21:36.000000000","message":"Patch Set 28:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6789d3afc59e44129a58e89008978ac2\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/000709a4294b4022b4bb731fcc104ae7 : SUCCESS in 1h 13m 04s","accounts_in_message":[],"_revision_number":28},{"id":"9664c57560203c8368f8d704181d42c90c92f52b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-11-05 00:21:14.000000000","message":"Uploaded patch set 29.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":29},{"id":"d0f98cfd362e46a234f338320f8ee19044d5e31d","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-11-05 00:21:48.000000000","message":"Patch Set 28: Code-Review+1\n\n(11 comments)","accounts_in_message":[],"_revision_number":28},{"id":"7be801daaf97e91b35529d4bb942c5853d80967a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 01:34:59.000000000","message":"Patch Set 29: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e74e8cf3f9c4946a0015fe5d6eaf97a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06a35be646f34343b94d74f16e3259fe : SUCCESS in 8m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/ab036ea8f6ca49e5b68259709ae7d5a1 : SUCCESS in 1h 00m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/776a7619c680465580d014a4d0d0108b : SUCCESS in 1h 02m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/66b1936fcf01460b980f42f3305a7d39 : SUCCESS in 49m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0f8758ab022142c8b23a87ab18c7d582 : SUCCESS in 1h 02m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9f765ec98dcf4ced8742847844cbc221 : SUCCESS in 19m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d28f04321cd140a7a3a48347270c9926 : SUCCESS in 20m 33s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2b85c2a01dd044799a84b09525727442 : SUCCESS in 21m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/46695b30ac584e3bba09f21191d32866 : SUCCESS in 5m 21s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/efbedc9144a046c28667ec453f55c369 : SUCCESS in 15m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7e06b75e6d5b4eb89e3d5fcab3f3f83a : SUCCESS in 15m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/cceff9e063194e58a761e7d21a1cc0c9 : SUCCESS in 19m 55s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/760a608fb8324982b47a0fb090535cac : SUCCESS in 19m 27s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c4229d7fe9f74e458f1ed0d6ecc7d0e4 : SUCCESS in 18m 30s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f589e56d373f48d4984b81f7f0b3693d : SUCCESS in 18m 09s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5ed7704dbbc548799b3ee637139b6e67 : SUCCESS in 18m 04s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/7599b3eafa064e488223d61d1ef75224 : SUCCESS in 8m 59s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8bc760977863407fa202776461100f77 : SUCCESS in 14m 37s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/07594848be19462da59a85c438f3cf73 : SUCCESS in 16m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3819db9d68834ddfa3dd4231ddf98a82 : SUCCESS in 7m 45s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6b448ab89add4533b8b7cc52ff78b4e1 : SUCCESS in 18m 46s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f86944eed9d5464e89d84e5b28e47fa6 : SUCCESS in 8m 19s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3b06f87066d844ec8044aec0c449876e : SUCCESS in 54m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/eadbe504fc3d4965bd36e17d9a511dec : SUCCESS in 51m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c862f5388136411183d1840d0b83f561 : SUCCESS in 50m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e1cb33f6570543dabaf6ff7093790b27 : SUCCESS in 13m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/002572b938934dcfb7d2392f8f28ce26 : SUCCESS in 4m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7a076a38b52b41a5bb44d8f5af80b89a : SUCCESS in 15m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dd4b9e2598f84c77b6e2ff01df3eba84 : SUCCESS in 56m 09s","accounts_in_message":[],"_revision_number":29},{"id":"b0c043b96a266246c35339d53e4db7e631c5da88","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-05 01:41:29.000000000","message":"Patch Set 29:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/efda76aefa2f4b3187467bb1c7b5f5c6\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b90f0df37af64a02b189f3f22afcba58 : SUCCESS in 1h 18m 02s","accounts_in_message":[],"_revision_number":29},{"id":"e707e3dcdb5f7a43ad3027ebcc729e8b7baf3f4d","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-05 10:10:19.000000000","message":"Patch Set 29: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":29},{"id":"88ef1711f527fcd160696ba077bd2fed6a54ae9e","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-05 11:32:00.000000000","message":"Patch Set 29:\n\n(1 comment)","accounts_in_message":[],"_revision_number":29},{"id":"6ccd5fe6d8f8c27c32d4de767c61254d28e63437","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-11-05 22:52:54.000000000","message":"Uploaded patch set 30: Patch Set 29 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":30},{"id":"0336a3ab5d33d700814e9fdc4b4dc032c164cae6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 00:12:43.000000000","message":"Patch Set 30: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/662da88723db4012b63a9b2db72605f7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60c6440b493a415dba77621c197c0962 : SUCCESS in 9m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/36afe11d71d04391ab47ea2276072aff : SUCCESS in 1h 07m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7ca8ee010f8047e4a82dac969475162b : SUCCESS in 1h 03m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d3db3070938449cea547a320c619bc7e : SUCCESS in 38m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/df74e46cd5b64223a2c1392c7fccacf6 : SUCCESS in 1h 10m 14s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/99411c7570ef4914a2fabb9ef74329e1 : SUCCESS in 16m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4d309b0b1ea341c7b848200efcac1e79 : SUCCESS in 18m 09s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ce31b10196c3409ea209cabe71d61043 : SUCCESS in 21m 24s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ec13375ef2424377804329d27b95d3f6 : SUCCESS in 5m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/88f6b25c2ecf4ae1a34e35bd367510c7 : SUCCESS in 26m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7c967ee377bf43e5b7f588caddad8fb9 : SUCCESS in 19m 28s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4933aa9a96d3496f8ff524ed245082cf : SUCCESS in 21m 05s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7d54bd8efa5446f5aa317ecf1be53df4 : SUCCESS in 32m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/75f7991447d843c4bdf8078db1a264aa : SUCCESS in 18m 29s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7224fbaa7838487fbab2d7c0453fee5a : SUCCESS in 17m 05s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/9bfadec37b414c098f17f2a562846729 : SUCCESS in 17m 27s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ac71d9bb64e446e98998256fc890308e : SUCCESS in 16m 35s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/94617d622589497ca3f8fe7ff0685584 : SUCCESS in 15m 54s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/47d83e43d3724fd59a2c7b61740b3c99 : SUCCESS in 17m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4d65cbddfd6c4c76ac073810d05b96a0 : SUCCESS in 8m 14s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/799f801373e54fc68a6768dfb3ddc32e : SUCCESS in 14m 19s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/656722fe3d424dc2be407990b42bb3eb : SUCCESS in 9m 12s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d880fb1f17834eb8bd41ef2cb20a4a1f : SUCCESS in 53m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c841c165d9e94e5d8e836b99a4cf1560 : SUCCESS in 50m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2f0a0d517ff545c7b2348fec0c5a8f8f : SUCCESS in 49m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cbd537b420a24f7fa1693fa2292194f8 : SUCCESS in 20m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eba85cfec8404e3fbcd136c704bf07e3 : SUCCESS in 4m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/990edf0b1e71465b94d6c03afbd481b2 : SUCCESS in 19m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ca13888efc5a47e388684ad1c44cfc65 : SUCCESS in 1h 06m 19s","accounts_in_message":[],"_revision_number":30},{"id":"da73601e1ea36c4b0c2ff9dc713a592fba97a9ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-06 00:28:36.000000000","message":"Patch Set 30:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a4e97a0ec9f4755a4f25df0a2b352fd\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bf70bd9414764cffb8f255713cd96aa6 : SUCCESS in 1h 33m 48s","accounts_in_message":[],"_revision_number":30},{"id":"65ebddde8dea62ca74eff82793290c6952e140b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-11-12 17:02:51.000000000","message":"Uploaded patch set 31: Patch Set 30 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":31},{"id":"5a1777fb0611b907f55e28adfcaa0499f24a3a48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 18:05:42.000000000","message":"Patch Set 31: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b8a245102e664e10be2223d3a0a3168d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccaf47656d324cdfb423aeb464cf6327 : SUCCESS in 9m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/6ad301f8f4754dc68628d423bed21a8a : SUCCESS in 41m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c940325b18c04d848b12416e8132c6b8 : FAILURE in 23m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/baf9ca3b42ac41b0b6c5aef6bdf300da : SUCCESS in 42m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3322110e7186424980e935feeb697e7e : SUCCESS in 49m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/95a8db08f6f148c18e2538e7c4f5eaac : SUCCESS in 14m 16s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dc2683efd16b418e836ccfc7de9b8038 : POST_FAILURE in 19m 59s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/74028ec31e5b45b08e2bd59976c5b785 : POST_FAILURE in 18m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/55959e2cbc85486e9f3a4aba41e1bd2f : SUCCESS in 4m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1535d653b19849bfb88cf50b3ac86a6d : SUCCESS in 20m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f29a5a7f6e1143feaa7251228476e7f1 : POST_FAILURE in 20m 16s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/024ca3e11dd5421dbf6b9966affd115f : SUCCESS in 16m 13s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c5d1374856bf4022b0296b92a7cdb28c : POST_FAILURE in 26m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/c691bab6d91b4e6281721da3fee4c9f5 : SUCCESS in 15m 08s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/040e67b84b5b49889703b9ac5290b1bf : POST_FAILURE in 18m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cd6df4edd76a4a129e78d31352f33afc : SUCCESS in 17m 44s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9a7ab8a208064599ab4e97ff93d74187 : SUCCESS in 15m 23s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/71c9ff6a7d2c44318a9d52d01d363bfd : SUCCESS in 16m 03s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/49811f0106b34fbe80cf510107b11d17 : SUCCESS in 10m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/33cbaf0a7d5f4457902ce1b296624999 : SUCCESS in 10m 02s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c1af2f9c83bf4b068f039fa6b032202a : POST_FAILURE in 16m 00s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1c073c44cfbe442e8f5d8fc27b134a45 : POST_FAILURE in 6m 15s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/647d7fbd483446cbbef1cbe47cd8d133 : SUCCESS in 55m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/622d8908f36947799f799e9485f6279a : SUCCESS in 36m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b2fdf387d2ea432a8f9885c299f03c14 : SUCCESS in 52m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bd43e68a17b249bfaef525bcb08326c7 : SUCCESS in 20m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a94f91e375514af2a039fd3bc32c1337 : SUCCESS in 4m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b704658653034f37aa8f0771b3b80b47 : POST_FAILURE in 17m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5477e4fde7d24fdb8f015f7dde5394ba : SUCCESS in 54m 46s","accounts_in_message":[],"_revision_number":31},{"id":"c40157b6b06cf22582e927fc41c66db4a766ad07","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 18:25:36.000000000","message":"Patch Set 31:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc4b2847714b48809d4c110d63d20446\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/41a68d8faf464fb58ea4fdb4c4d5983e : SUCCESS in 1h 20m 07s","accounts_in_message":[],"_revision_number":31},{"id":"5d05a266ac18f5468d9044bf4e0ec1f381030012","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-11-15 18:52:07.000000000","message":"Patch Set 31: Code-Review+1","accounts_in_message":[],"_revision_number":31},{"id":"4de36b762aabaf24a6d20ea57249d548fec2af34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-11-15 22:58:23.000000000","message":"Uploaded patch set 32: Patch Set 31 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":32},{"id":"9ee2db4d1e5d19664888a42e1fd8da7ffaa7aa95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 23:49:26.000000000","message":"Patch Set 32: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/15e9447d33ac4dcb8e5af7364b3d3186\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f711ef2fc0894c3bb4275448766bfe0a : SUCCESS in 8m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/18ef860f654e45dbb1798c7f63760c44 : SUCCESS in 45m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9ca5b634c3904be087b923ba34bff7f9 : SUCCESS in 44m 40s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/438fbaf87232420c92a6002cd1fd6ee1 : SUCCESS in 35m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/14c4b76a50fb40d597612029d8ec158f : SUCCESS in 47m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ea71cd30673a480a989fbb72481bccae : SUCCESS in 19m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d39f4b7f51fe46c9916acf7fa5f9e1fa : SUCCESS in 20m 02s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/91f04ae3592242c3bcd198b7b5535e08 : SUCCESS in 13m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4b45a15ad77a4a5a89340ea0889d8f66 : SUCCESS in 5m 40s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/79216de1621148f1bdfd832139156f4e : SUCCESS in 11m 09s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ec5c467388e847e08d040d30b2fa7c9f : SUCCESS in 17m 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0c8eaac131a04df1bec9e4ccf384548e : SUCCESS in 22m 14s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4a701e77829940e38f861cdc28f4a672 : SUCCESS in 14m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5500ae115895406386d992881b448083 : SUCCESS in 17m 10s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2693578b49ff42a489b18207fc4a6dad : SUCCESS in 16m 51s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/21568cc2a6fb40c8b3d0c716d5b641f2 : SUCCESS in 11m 43s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/fbf066f21efa4f9ea1e922ae19854456 : SUCCESS in 14m 56s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5d058ef2d0b846b885c94799ad57d7da : SUCCESS in 17m 23s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1a2ed12020d840a7b05f4600eb2a2025 : SUCCESS in 10m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1e30bd637c044c22aa5bf316aab63735 : SUCCESS in 5m 42s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1d0b4e8ee4774334a5145f17a89eb5ae : SUCCESS in 14m 49s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e6c35649b0f247ad929ee647a8488e0e : SUCCESS in 6m 39s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/67320c100b2444ec89a5e27a7087b60b : FAILURE in 42m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bdeeb6efba0c47b2b836e8519a5a19b7 : SUCCESS in 37m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3757b7dcaa924a9d9ab7b54c7f6f39e3 : SUCCESS in 37m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4418c02ad8d349108c0dd8ae01afe73d : SUCCESS in 15m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf5949f7277d4b42b1d61d27004d0106 : SUCCESS in 4m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ecb292ed7ab6488fb152d2eeb70a7068 : SUCCESS in 11m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/31e1fc7f34824bd4864c807b88430f33 : SUCCESS in 42m 42s","accounts_in_message":[],"_revision_number":32},{"id":"fbc7f7aebdd8a9df5a1fd05bdc2b2b292f7b5368","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-16 00:25:20.000000000","message":"Patch Set 32:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d40e9bb1adf64fb093854011cb2195d4\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ab72bd975cdb4bc38d72b58f751baced : SUCCESS in 1h 24m 54s","accounts_in_message":[],"_revision_number":32},{"id":"8770fff8b26a1de2cad0405e65c8e4c9e05ed324","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-11-18 19:06:00.000000000","message":"Patch Set 32:\n\n(5 comments)","accounts_in_message":[],"_revision_number":32},{"id":"54b53d91019589c3b6a62e7412c41d42878ef9e0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-11-18 19:09:46.000000000","message":"Uploaded patch set 33.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":33},{"id":"7f3c6f727ff15b3650745057031ac97d7fa8d55b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 20:21:52.000000000","message":"Patch Set 33: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/893e1acf81344daead619a81f3d712f6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd2f0f34ac5348f6a071232f4b145fd4 : SUCCESS in 8m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/a669414f5bc242ca8653c9ca2273fd03 : SUCCESS in 1h 01m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/23bb52d7d95e4229ab19924ef1dccd2c : SUCCESS in 1h 03m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/be9b8ba9f99246b6b642ef2323fb5c87 : SUCCESS in 43m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/acebfc368a784a25841f151da8c15069 : SUCCESS in 1h 01m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d7cbafaad9454435903c55eec0c810cf : SUCCESS in 16m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2473423dba404ca3bbe39b26421ee1cf : SUCCESS in 18m 26s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f9bd6e6503894e2b8761f23927397176 : SUCCESS in 20m 32s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7279adf6c9344e11a793ec60561994d0 : SUCCESS in 5m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8dc98887308f461e832fc465d9b784a5 : SUCCESS in 19m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d588f95384c84941a5f544f8f0160ed9 : SUCCESS in 11m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/17bf806e0bc74fa893d3840cadfd4c4e : SUCCESS in 24m 33s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ae388011ba5c4f969b9a1ba5835e759a : SUCCESS in 19m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/88c7738b06e24a99a72d10fb29aab561 : SUCCESS in 13m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/700ba11a4c41443fb3993250dcc94f81 : SUCCESS in 17m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/54e541ab7d30442eb6bdd2b4b7387dfa : SUCCESS in 11m 54s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e6fd2b8841e34b91bb315fedd749a83d : SUCCESS in 19m 08s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/fa86a94ca674459f9cdc6bfacce431f3 : SUCCESS in 10m 16s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0aa1add0fd2143a8b57a37608711abfa : SUCCESS in 20m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/01869271dd144e12b0e32abb0b8c44e3 : SUCCESS in 4m 33s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a777ee69d75b45cfa08c5b02ebdda30b : SUCCESS in 23m 38s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/51c9254708ef486296ec2863a1c594d0 : SUCCESS in 6m 50s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e7191868cef04e31bb9fc192d8d0731e : SUCCESS in 40m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/954f059161ae4e18b5cb19899f5d1efc : SUCCESS in 58m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1276f6dade61455eab93c71375c540e3 : SUCCESS in 52m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/23fa8292d60a4e45a9de0cecb70f4275 : SUCCESS in 26m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/423f1a68f1204e6cbffe293dbd11909f : SUCCESS in 5m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/549fc8427eb74a438e198b515a43fcb8 : SUCCESS in 24m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6bbfa2fd3cf3418fa1b13fb1ed71d2ca : SUCCESS in 44m 46s","accounts_in_message":[],"_revision_number":33},{"id":"381ac9d5dbccb28c726cf182ec3812c89d46da94","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-18 20:41:34.000000000","message":"Patch Set 33:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a59c040bfdc64e2291844ae8ddcd2dc5\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a1e72bec4f4649768303c977d225533f : SUCCESS in 1h 29m 15s","accounts_in_message":[],"_revision_number":33},{"id":"c9209141f1eac96d164f145fa4f02d8a07cb104f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-11-29 14:47:58.000000000","message":"Uploaded patch set 34: Patch Set 33 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":34},{"id":"e9f37559e0de2f74afbb5db528635fd9a3c96160","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 15:57:46.000000000","message":"Patch Set 34: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01377394fbb440bda2fffd4101cc715f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af132fc8c86940619a1c98a3a667a50b : SUCCESS in 8m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/644429378952486f8fc9ad2d1900b785 : SUCCESS in 52m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/23e0f2a243984d679c29f5e18cdb87bd : SUCCESS in 33m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/221584ae895c44f19207c9e9c64cf177 : SUCCESS in 34m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b76e3e5abd404b76af23036738c205b9 : SUCCESS in 56m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e25096fc2b9f41189684eb49ae7174ab : SUCCESS in 19m 03s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f283ca606e5b49bea3655717704ba62e : SUCCESS in 19m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5a9ae94ba70a4d4e9580f2350e38769b : SUCCESS in 22m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b792b93bb43a49979a17da78ecb0ef4b : SUCCESS in 6m 04s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1a2474f0fd5c4d9eb70212dd86adad47 : SUCCESS in 18m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a0e439e5873641c381e2cffb7a32f7e7 : SUCCESS in 20m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f627573ac967435d834d398acc6f86d2 : SUCCESS in 21m 41s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6f3a15dfd10c422683fabd208dc62270 : SUCCESS in 36m 44s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6dcfdaba762546cd8b78fe9d76869319 : SUCCESS in 16m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fe35c6887d074bd8b80e4bed5bcc53e8 : SUCCESS in 16m 38s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/43ab1ad099d64e3f9a7584f796def1c9 : SUCCESS in 22m 08s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0e93beff7532462eba539b9d6f0cdb7b : SUCCESS in 15m 54s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7f18868740ef460abe127b0062dfed17 : SUCCESS in 16m 08s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a0d72cd5e7c743a8b112412b9ae67e0a : SUCCESS in 23m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a445ca012a94428e9ae9726fe236ef79 : SUCCESS in 7m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3326075391ed463e962ebad847df7e24 : FAILURE in 15m 10s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/44a5b67748c64c6a862f5ef12b13035b : SUCCESS in 6m 44s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6fbcfcedba3f4307a6b7d4879201af25 : SUCCESS in 1h 06m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/72b9db1c4ec1406095b2e9edd55c9b26 : SUCCESS in 27m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9f2da99d22fc484e9e20be844678c9f7 : SUCCESS in 53m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9e0bf0d7c398414eb48c738f976d3dd5 : SUCCESS in 19m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae225838516741219a93846a181d59d3 : SUCCESS in 4m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a7858282160c4c7bbe2e55dc42768fed : SUCCESS in 23m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/69127c6531074f74b0169071bfee7e22 : SUCCESS in 52m 14s","accounts_in_message":[],"_revision_number":34},{"id":"45e8ef5bec3316f32aa5a09b9c7d48d3f0ca7061","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-29 18:52:33.000000000","message":"Patch Set 34:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6b01d6e51c2411c95d3fcfe99195ba1\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/109e818a911b4a21a07429d1dfc8a92e : TIMED_OUT in 3h 00m 58s","accounts_in_message":[],"_revision_number":34},{"id":"e1ecf17f20a86415a2e69cb1ba17ae654ea7dbe8","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-01-03 22:15:04.000000000","message":"Patch Set 34: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":34},{"id":"d9dd6bb03dd331a3f9a9d4004604cbaf3165738a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-01-16 12:23:57.000000000","message":"Uploaded patch set 35: Patch Set 34 was rebased.\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":35},{"id":"d01b4014ee13ebf07f4c0db6f2cde82d953ed4bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 13:25:49.000000000","message":"Patch Set 35: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e12e9fc3ef714d438f1e22224f195f2d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70dd4a2b7977435b8590b8166ca68240 : SUCCESS in 9m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/825e194683aa4fdb9b8f7480afff0137 : SUCCESS in 41m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/22d0a6fcc20246b1aa6ae3441a45f1fb : SUCCESS in 53m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a6524d17fb6647b499ed5197f569a64b : SUCCESS in 42m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c31cd6ec295a46109e1c72864666788b : SUCCESS in 54m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/49f7405687b04e39b3300e8a842205dd : SUCCESS in 22m 04s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4e5d5f4f5d164d4092610d618a3ed623 : SUCCESS in 24m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ae8d111e0e3a4c98af0debdb404c3b29 : SUCCESS in 21m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7d86c284cfc346c3940768bfa5459920 : FAILURE in 3m 38s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cc1861d956d34b61abb6b5f3a950ed41 : SUCCESS in 8m 52s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/63c02f82d62e4e9a9f00cf4061f3ef36 : SUCCESS in 19m 59s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d32751bd68194b2eb4a2951c070ac04f : SUCCESS in 22m 19s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e0280895af314087a2964379a36065c1 : SUCCESS in 15m 14s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9c705b51552044a5bbdc97e5ef33916f : SUCCESS in 9m 10s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6d001942a8134483a915063eaa94da67 : SUCCESS in 9m 40s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cf4f0f7b723042f596c30430974fe605 : SUCCESS in 4m 18s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e92ef4c1cbdb47feb95cc54675d20c4c : FAILURE in 20m 22s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/28c45f4b43aa4b0ebbfee39b2ed34970 : SUCCESS in 8m 16s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/93763d3d1e06425586b782da354bb48e : SUCCESS in 57m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7bb8485c68f94f75b992e45638448a96 : SUCCESS in 38m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/95783d600eb24f72ad2e592b4a8d1a82 : SUCCESS in 31m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2847554466e1411fa0447845e449c321 : SUCCESS in 24m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40defdf703074544ac24b1aab97df3cb : SUCCESS in 4m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f08eebbe0a0149f7adc65c29c7d940b0 : SUCCESS in 22m 39s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a0d863ccae9940418007fdc522b88e37 : SUCCESS in 44m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f04379f08bc14be1b7a79b2c8d171300 : SUCCESS in 56m 49s","accounts_in_message":[],"_revision_number":35},{"id":"1f1bffaafb42f41c5f7b98feee9f35882ab6ddb4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-16 16:57:04.000000000","message":"Patch Set 35:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/487804c93f5f40a68e14628b19621a4d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/484a78b903e047389a619c57195be678 : SUCCESS in 2h 42m 34s","accounts_in_message":[],"_revision_number":35},{"id":"425e732b6f877f94ac783081835e2e4225d537f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-02-07 16:33:48.000000000","message":"Uploaded patch set 36.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":36},{"id":"5b95077b5d52024580ecaada3582383bf9970fbc","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-02-07 16:37:06.000000000","message":"Patch Set 36: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":36},{"id":"69a99af2e4ea1e8e9f74f29c93d1bafc97c9823d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 17:45:10.000000000","message":"Patch Set 36: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5103b352c05f40c796d7953976580daf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1158b26a3b4543f6a43197eccbb09417 : SUCCESS in 19m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/7257d48e30204a5da0f6e46339374824 : SUCCESS in 48m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/56d5d1348ecf49c5930b7b62d420301b : SUCCESS in 54m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0ad31309359f462fb62cc8e711357565 : SUCCESS in 50m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/56ef2d0a8f49477c9f7c11e169adb32b : SUCCESS in 1h 05m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/93d7c52130bc48e39ca7a0767ea0564f : SUCCESS in 21m 04s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1961276fdbbf47bfbb057e8c28b49d6c : SUCCESS in 24m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/851399d5ad3141b1aa38e4e892ca0316 : SUCCESS in 19m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/80afb2aa70b946c4b20558c486ec651b : SUCCESS in 5m 49s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/62eb46aa39ef4224ac6263343d2f3ff8 : SUCCESS in 18m 15s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/53806b4a77c047f2a34d1c3f3c91eb6d : SUCCESS in 27m 21s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/60e5e995d7cf40bc96d4c0e279ff10cd : SUCCESS in 26m 56s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/7c38e849f43243728d7dc7d3eaeead77 : SUCCESS in 15m 33s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/cc5e3da5403b44c3a216f6ebd3cd9fbb : SUCCESS in 15m 48s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4f7c25dedb964865a19a8c10ac16f0e0 : SUCCESS in 19m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/67a7742b9eff4b639aa9421c991df678 : SUCCESS in 7m 38s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/18a43899d4644768a4e22b63bfd82d54 : FAILURE in 20m 34s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7840023982bd412d8d8dac5be3ce7ee2 : SUCCESS in 6m 43s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e12ab06b118449c998baff13d5dfe7d4 : SUCCESS in 48m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2754fc63d2654c16ae7b90cb5033a6c5 : SUCCESS in 36m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7607aaafbb1945048f222d9ea3b8f85a : SUCCESS in 44m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aaff498c5bf34d31b908fd7f8c320afd : SUCCESS in 27m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a52db066e564e21ade760b59ebc873b : SUCCESS in 4m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1064fb58cd9f4059bdcada199b735fa2 : SUCCESS in 21m 47s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f9b20e91b22046d78be5540321063a6d : SUCCESS in 47m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c2d7aa9adc464649bf527c669259fe95 : SUCCESS in 52m 10s","accounts_in_message":[],"_revision_number":36},{"id":"5df7fc01adeaad583aba9afbb603cb81f45ba42d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 19:01:00.000000000","message":"Patch Set 36:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee93af2969d04200b26eb02fa2f50713\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a7695424c5314947b72b238a623b42c1 : SUCCESS in 1h 38m 34s","accounts_in_message":[],"_revision_number":36},{"id":"e661a73d674cf561ab30863a6abc5747d5478408","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-02-07 21:25:30.000000000","message":"Patch Set 36: Code-Review+2\n\n(5 comments)","accounts_in_message":[],"_revision_number":36},{"id":"cc53537bdc2eabaf6787d0c01d044aaf13dc8d71","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-02-08 01:07:26.000000000","message":"Patch Set 36: Workflow+1\n\n(5 comments)","accounts_in_message":[],"_revision_number":36},{"id":"f0261892048adf14b33c59cad57f125b574ea816","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-08 01:07:59.000000000","message":"Patch Set 36: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":36},{"id":"f1e9b8508efbba148ae5b7736d9c0113e0261bdd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-08 02:26:03.000000000","message":"Patch Set 36: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ee137ac87844c108bdfa75c16a95725\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c8a3d96016c43b182f19bc6268e2cf7 : SUCCESS in 6m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/176dd9634bf549b8830094572e1be3f6 : SUCCESS in 44m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1f5a0b7d7e954d19a1781a6ee97e6491 : SUCCESS in 42m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/de3960457c6c41d5bb4b1bddc53156bb : SUCCESS in 36m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2937f56920534f318e19c3162c7094b7 : SUCCESS in 43m 41s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/a3ffabcb76e74ca1b5a9aa21835795fc : SUCCESS in 4m 39s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ee9d65a0befd4cebb56e30bbc923aad8 : SUCCESS in 9m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5f6f1e0653f64f2cbfbb0964b14f07af : SUCCESS in 9m 21s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3e3dd05d121d43aeab9f400017874c94 : SUCCESS in 29m 19s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/07df88be2a514867903e04d07e501d17 : SUCCESS in 14m 10s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/90a665e23774460c909d3d97c8b51675 : SUCCESS in 9m 05s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/724eff77081d479893d0c84759d19d25 : SUCCESS in 14m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/259e37d5bdec4c749321658477d48c57 : SUCCESS in 5m 39s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2d81b9ef0aef4647b39e0942bef73fb2 : SUCCESS in 4m 55s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/20c81d74cf054c80a4fe1bba83ebad54 : SUCCESS in 54m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/28c4c9fbd9cd4a8dbb9ff0a405f88468 : SUCCESS in 48m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/dfeb1581f47d433b8348c9e666eb28fb : SUCCESS in 32m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a91ba72a99154a4cabe18a3fcbe7de3e : SUCCESS in 20m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7338fa7619d14026b87b54c5a6fdee49 : SUCCESS in 2m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c70f88b5fb054e46b7882213943928cf : SUCCESS in 16m 42s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ca8d5522094d4b40bb8fa0db952d5590 : SUCCESS in 35m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0ae00ec374b9456a93a12df178dc33d9 : SUCCESS in 1h 06m 42s","accounts_in_message":[],"_revision_number":36},{"id":"4f9dbdba2572cba6b5b1ef1e2bc0151d8611e9bd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-08 02:26:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":36},{"id":"a93a8431923f386215b141901e5ff8757f73eef0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-08 02:28:08.000000000","message":"Patch Set 36:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/164e30aba25d4384bdc6a8ccad9e4354\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3669e968e53d4defaca6b156f4272372 : SUCCESS in 1m 12s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/0baa92d5fe0e4ca9a5a4f5d0fbc95c74 : SUCCESS in 37s (non-voting)","accounts_in_message":[],"_revision_number":36}],"current_revision_number":36,"current_revision":"b69a2bef45b100a2bc06b7fbb2eacd06380b117b","revisions":{"517e20ffae91c9302eb467996b271073bbadd1a0":{"kind":"REWORK","_number":1,"created":"2024-05-25 13:36:02.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/52/920452/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/920452/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/920452/1"}}},"commit":{"parents":[{"commit":"6915f80641d45c9b4ac234885c32fc0d0b2794dc","subject":"tests: realistic task_container names","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6915f80641d45c9b4ac234885c32fc0d0b2794dc"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-05-24 19:27:26.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-05-25 13:35:51.000000000","tz":-300},"subject":"wip: make per_day configurable","message":"wip: make per_day configurable\n\nChange-Id: I5ea9e6dc8b44c8c5f55837debe24dd76be7d6248\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/517e20ffae91c9302eb467996b271073bbadd1a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/517e20ffae91c9302eb467996b271073bbadd1a0"}]},"branch":"refs/heads/master"},"45c067912d6b37cbeb41caaaf33cf5e664b5b939":{"kind":"REWORK","_number":2,"created":"2024-05-28 17:00:37.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/52/920452/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/920452/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/920452/2"}}},"commit":{"parents":[{"commit":"663d866e3453e7dee54cbb205afd4b74a6f79840","subject":"tests: realistic task_container names","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/663d866e3453e7dee54cbb205afd4b74a6f79840"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-05-24 19:27:26.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-05-28 17:00:22.000000000","tz":-300},"subject":"configurable object_expirer_task_container_per_day","message":"configurable object_expirer_task_container_per_day\n\nChange-Id: I5ea9e6dc8b44c8c5f55837debe24dd76be7d6248\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/45c067912d6b37cbeb41caaaf33cf5e664b5b939"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/45c067912d6b37cbeb41caaaf33cf5e664b5b939"}]},"branch":"refs/heads/master"},"8480d21430ed80f324c899397d991e9506853c29":{"kind":"REWORK","_number":3,"created":"2024-05-29 13:42:28.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/52/920452/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/920452/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/920452/3"}}},"commit":{"parents":[{"commit":"c9108823e03f4b9e86f1fea5e6775b1808576609","subject":"tests: realistic task_container names","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c9108823e03f4b9e86f1fea5e6775b1808576609"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-05-24 19:27:26.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-05-29 13:41:15.000000000","tz":-300},"subject":"configurable object_expirer_task_container_per_day","message":"configurable object_expirer_task_container_per_day\n\nChange-Id: I5ea9e6dc8b44c8c5f55837debe24dd76be7d6248\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8480d21430ed80f324c899397d991e9506853c29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8480d21430ed80f324c899397d991e9506853c29"}]},"branch":"refs/heads/master"},"b61cf5170d46fb78d2f56e6f0ff8537bc162c466":{"kind":"REWORK","_number":4,"created":"2024-05-31 20:45:30.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/52/920452/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/920452/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/920452/4"}}},"commit":{"parents":[{"commit":"9d6ab9ddffdfb76430e842056082b46b82243625","subject":"DNM expirer: new options to control task iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9d6ab9ddffdfb76430e842056082b46b82243625"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-05-24 19:27:26.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-05-31 20:32:19.000000000","tz":-300},"subject":"configurable object_expirer_task_container_per_day","message":"configurable object_expirer_task_container_per_day\n\nChange-Id: I5ea9e6dc8b44c8c5f55837debe24dd76be7d6248\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b61cf5170d46fb78d2f56e6f0ff8537bc162c466"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b61cf5170d46fb78d2f56e6f0ff8537bc162c466"}]},"branch":"refs/heads/master"},"c6e36d1e962e8eba623bdb086b5522e40368339e":{"kind":"REWORK","_number":5,"created":"2024-06-04 16:30:20.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/52/920452/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/920452/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/920452/5"}}},"commit":{"parents":[{"commit":"689ae8e8a1e7223ff20b44afa5c1790187751bfd","subject":"DNM expirer: new options to control task iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/689ae8e8a1e7223ff20b44afa5c1790187751bfd"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-05-24 19:27:26.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-04 16:29:48.000000000","tz":-300},"subject":"Configurable object_expirer_task_container_per_day","message":"Configurable object_expirer_task_container_per_day\n\nIn order to better scale the object-expirer cluster operators may\nconfigure the object_expirer_task_container_per_day to increase the\nnumber of task_containers created per day beyond the default value of\none hundred to any value that better matches their clusters topology and\nuserage of the x-delete-at feature.\n\nUpgradeImpact: Two existing expirer options have been deprecated\n\n * expiring_objects_container_divisor\n * expiring_objects_account_name\n\nThe upstream maintainers are not aware of any clusters where these have\nbeen configured to non-default values.\n\nOperators are encouraged to remove their \"container_divisor\" setting and\nuse the default value of 86400 which will make more sense with\n\"task_container_per_day\".\n\nIf a cluter was deployed with a non-standard \"account_name\" operators\nshould remove the option from all configs so they are using a supported\nconfiguration going forward, but will need to deploy a stand-alone\nlegacy expirer config to clean-up old expiration tasks from the\npreviously configured account name.\n\nChange-Id: I5ea9e6dc8b44c8c5f55837debe24dd76be7d6248\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c6e36d1e962e8eba623bdb086b5522e40368339e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c6e36d1e962e8eba623bdb086b5522e40368339e"}]},"branch":"refs/heads/master"},"c503d937bdf0898e739f4febf2b5f416d3af87e9":{"kind":"REWORK","_number":6,"created":"2024-06-13 23:07:27.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/52/920452/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/920452/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/920452/6"}}},"commit":{"parents":[{"commit":"3637b1abd9240f7feeebc7f299d42fe8815d7dcd","subject":"add bytes of expiring objects to queue entry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3637b1abd9240f7feeebc7f299d42fe8815d7dcd"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-05-24 19:27:26.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-13 22:03:18.000000000","tz":-300},"subject":"Configurable object_expirer_task_container_per_day","message":"Configurable object_expirer_task_container_per_day\n\nIn order to better scale the object-expirer cluster operators may\nconfigure object_expirer_task_container_per_day to increase the number\nof task_containers created per day beyond the default value of one\nhundred to any value that better matches their clusters topology and\nusage of the x-delete-at feature.\n\nUpgradeImpact: Two existing expirer options have been deprecated\n\n * expiring_objects_container_divisor\n * expiring_objects_account_name\n\nThe upstream maintainers are not aware of any clusters where these have\nbeen configured to non-default values.\n\nOperators are encouraged to remove their \"container_divisor\" setting and\nuse the default value of 86400 which will make more sense with\n\"task_container_per_day\".\n\nIf a cluter was deployed with a non-standard \"account_name\" operators\nshould remove the option from all configs so they are using a supported\nconfiguration going forward, but will need to deploy a stand-alone\nlegacy expirer config to clean-up old expiration tasks from the\npreviously configured account name.\n\nChange-Id: I5ea9e6dc8b44c8c5f55837debe24dd76be7d6248\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c503d937bdf0898e739f4febf2b5f416d3af87e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c503d937bdf0898e739f4febf2b5f416d3af87e9"}]},"branch":"refs/heads/master"},"43cc17025cd094dd9a0c866887e7b35de08a3bc4":{"kind":"REWORK","_number":7,"created":"2024-06-19 16:32:12.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/52/920452/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/920452/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/920452/7"}}},"commit":{"parents":[{"commit":"3637b1abd9240f7feeebc7f299d42fe8815d7dcd","subject":"add bytes of expiring objects to queue entry","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3637b1abd9240f7feeebc7f299d42fe8815d7dcd"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-05-24 19:27:26.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-06-19 16:30:21.000000000","tz":60},"subject":"Configurable object_expirer_task_container_per_day","message":"Configurable object_expirer_task_container_per_day\n\nIn order to better scale the object-expirer cluster operators may\nconfigure object_expirer_task_container_per_day to increase the number\nof task_containers created per day beyond the default value of one\nhundred to any value that better matches their clusters topology and\nusage of the x-delete-at feature.\n\nUpgradeImpact: Two existing expirer options have been deprecated\n\n * expiring_objects_container_divisor\n * expiring_objects_account_name\n\nThe upstream maintainers are not aware of any clusters where these have\nbeen configured to non-default values.\n\nOperators are encouraged to remove their \"container_divisor\" setting and\nuse the default value of 86400 which will make more sense with\n\"task_container_per_day\".\n\nIf a cluster was deployed with a non-standard \"account_name\" operators\nshould remove the option from all configs so they are using a supported\nconfiguration going forward, but will need to deploy a stand-alone\nlegacy expirer config to clean-up old expiration tasks from the\npreviously configured account name.\n\nChange-Id: I5ea9e6dc8b44c8c5f55837debe24dd76be7d6248\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/43cc17025cd094dd9a0c866887e7b35de08a3bc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/43cc17025cd094dd9a0c866887e7b35de08a3bc4"}]},"branch":"refs/heads/master"},"ed495a6bc8f40fdd6363cfefa90e49765f2674b9":{"kind":"REWORK","_number":8,"created":"2024-06-25 00:34:28.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/52/920452/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/920452/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/920452/8"}}},"commit":{"parents":[{"commit":"9c50e0e8291c4828abadb98d6683bb0b65e992db","subject":"DNM expirer: new options to control task iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9c50e0e8291c4828abadb98d6683bb0b65e992db"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-05-24 19:27:26.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-25 00:21:58.000000000","tz":-300},"subject":"Configurable object_expirer_task_container_per_day","message":"Configurable object_expirer_task_container_per_day\n\nIn order to better scale the object-expirer cluster operators may\nconfigure object_expirer_task_container_per_day to increase the number\nof task_containers created per day beyond the default value of one\nhundred to any value that better matches their clusters topology and\nusage of the x-delete-at feature.\n\nUpgradeImpact: Two existing expirer options have been deprecated\n\n * expiring_objects_container_divisor\n * expiring_objects_account_name\n\nThe upstream maintainers are not aware of any clusters where these have\nbeen configured to non-default values.\n\nOperators are encouraged to remove their \"container_divisor\" setting and\nuse the default value of 86400 which will make more sense with\n\"task_container_per_day\".\n\nIf a cluster was deployed with a non-standard \"account_name\" operators\nshould remove the option from all configs so they are using a supported\nconfiguration going forward, but will need to deploy a stand-alone\nlegacy expirer config to clean-up old expiration tasks from the\npreviously configured account name.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I5ea9e6dc8b44c8c5f55837debe24dd76be7d6248\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ed495a6bc8f40fdd6363cfefa90e49765f2674b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ed495a6bc8f40fdd6363cfefa90e49765f2674b9"}]},"branch":"refs/heads/master"},"9a682528608b1cc47286d9566db15aea4cd00909":{"kind":"REWORK","_number":9,"created":"2024-07-09 23:58:46.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/52/920452/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/920452/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/920452/9"}}},"commit":{"parents":[{"commit":"c94c2ee05018ad1267795575de1571c84483a3a7","subject":"DNM expirer: new options to control task iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c94c2ee05018ad1267795575de1571c84483a3a7"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-05-24 19:27:26.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-07-09 22:02:52.000000000","tz":-300},"subject":"Configurable object_expirer_task_container_per_day","message":"Configurable object_expirer_task_container_per_day\n\nIn order to better scale the object-expirer cluster operators may\nconfigure object_expirer_task_container_per_day to increase the number\nof task_containers created per day beyond the default value of one\nhundred to any value that better matches their clusters topology and\nusage of the x-delete-at feature.\n\nUpgradeImpact: Two existing expirer options have been deprecated\n\n * expiring_objects_container_divisor\n * expiring_objects_account_name\n\nThe upstream maintainers are not aware of any clusters where these have\nbeen configured to non-default values.\n\nOperators are encouraged to remove their \"container_divisor\" setting and\nuse the default value of 86400 which will make more sense with\n\"task_container_per_day\".\n\nIf a cluster was deployed with a non-standard \"account_name\" operators\nshould remove the option from all configs so they are using a supported\nconfiguration going forward, but will need to deploy a stand-alone\nlegacy expirer config to clean-up old expiration tasks from the\npreviously configured account name.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I5ea9e6dc8b44c8c5f55837debe24dd76be7d6248\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9a682528608b1cc47286d9566db15aea4cd00909"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9a682528608b1cc47286d9566db15aea4cd00909"}]},"branch":"refs/heads/master"},"6fe94c3c0074b7e4d74d7cf23a198b869e816ce7":{"kind":"REWORK","_number":10,"created":"2024-07-10 23:01:25.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/52/920452/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/920452/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/920452/10"}}},"commit":{"parents":[{"commit":"feaa5c3057806353792bcef7ac29a5ec16638fa2","subject":"DNM expirer: new options to control task iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/feaa5c3057806353792bcef7ac29a5ec16638fa2"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-05-24 19:27:26.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-07-10 21:34:02.000000000","tz":-300},"subject":"Configurable object_expirer_task_container_per_day","message":"Configurable object_expirer_task_container_per_day\n\nIn order to better scale the object-expirer cluster operators may\nconfigure object_expirer_task_container_per_day to increase the number\nof task_containers created per day beyond the default value of one\nhundred to any value that better matches their clusters topology and\nusage of the x-delete-at feature.\n\nUpgradeImpact: Two existing expirer options have been deprecated\n\n * expiring_objects_container_divisor\n * expiring_objects_account_name\n\nThe upstream maintainers are not aware of any clusters where these have\nbeen configured to non-default values.\n\nOperators are encouraged to remove their \"container_divisor\" setting and\nuse the default value of 86400 which will make more sense with\n\"task_container_per_day\".\n\nIf a cluster was deployed with a non-standard \"account_name\" operators\nshould remove the option from all configs so they are using a supported\nconfiguration going forward, but will need to deploy a stand-alone\nlegacy expirer config to clean-up old expiration tasks from the\npreviously configured account name.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I5ea9e6dc8b44c8c5f55837debe24dd76be7d6248\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6fe94c3c0074b7e4d74d7cf23a198b869e816ce7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6fe94c3c0074b7e4d74d7cf23a198b869e816ce7"}]},"branch":"refs/heads/master"},"17faac823373f58cde95365c4fae359ae6648267":{"kind":"REWORK","_number":11,"created":"2024-07-19 13:45:09.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/52/920452/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/920452/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/920452/11"}}},"commit":{"parents":[{"commit":"951108c769df25ce7479a3e921e64168f8e8aabe","subject":"DNM expirer: new options to control task iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/951108c769df25ce7479a3e921e64168f8e8aabe"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-05-24 19:27:26.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-07-19 13:18:34.000000000","tz":60},"subject":"Configurable expiring_objects_task_container_per_day","message":"Configurable expiring_objects_task_container_per_day\n\nIn order to better scale the object-expirer cluster operators may\nconfigure expiring_objects_task_container_per_day to increase the number\nof task_containers created per day beyond the default value of one\nhundred to any value that better matches their cluster\u0027s topology and\nusage of the x-delete-at feature.\n\nUpgradeImpact: Two existing expirer options have been deprecated\n\n * expiring_objects_container_divisor\n * expiring_objects_account_name\n\nThe upstream maintainers are not aware of any clusters where these have\nbeen configured to non-default values.\n\nOperators are encouraged to remove their \"container_divisor\" setting and\nuse the default value of 86400 which will make more sense with\n\"task_container_per_day\".\n\nIf a cluster was deployed with a non-standard \"account_name\" operators\nshould remove the option from all configs so they are using a supported\nconfiguration going forward, but will need to deploy a stand-alone\nlegacy expirer config to clean-up old expiration tasks from the\npreviously configured account name.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I5ea9e6dc8b44c8c5f55837debe24dd76be7d6248\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/17faac823373f58cde95365c4fae359ae6648267"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/17faac823373f58cde95365c4fae359ae6648267"}]},"branch":"refs/heads/master"},"d0e7bf1cfbf7370d244c5ea92c7856d60695d95d":{"kind":"REWORK","_number":12,"created":"2024-07-24 10:24:51.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/52/920452/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/920452/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/920452/12"}}},"commit":{"parents":[{"commit":"ab5703d3f5edfbc92a1f775a48072cb85416a6e6","subject":"DNM expirer: new options to control task iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ab5703d3f5edfbc92a1f775a48072cb85416a6e6"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-05-24 19:27:26.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-07-24 10:22:25.000000000","tz":60},"subject":"Configurable expiring_objects_task_container_per_day","message":"Configurable expiring_objects_task_container_per_day\n\nIn order to better scale the object-expirer cluster operators may\nconfigure expiring_objects_task_container_per_day to increase the number\nof task_containers created per day beyond the default value of one\nhundred to any value that better matches their cluster\u0027s topology and\nusage of the x-delete-at feature.\n\nUpgradeImpact: Two existing expirer options have been deprecated\n\n * expiring_objects_container_divisor\n * expiring_objects_account_name\n\nThe upstream maintainers are not aware of any clusters where these have\nbeen configured to non-default values.\n\nOperators are encouraged to remove their \"container_divisor\" setting and\nuse the default value of 86400 which will make more sense with\n\"task_container_per_day\".\n\nIf a cluster was deployed with a non-standard \"account_name\" operators\nshould remove the option from all configs so they are using a supported\nconfiguration going forward, but will need to deploy a stand-alone\nlegacy expirer config to clean-up old expiration tasks from the\npreviously configured account name.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I5ea9e6dc8b44c8c5f55837debe24dd76be7d6248\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d0e7bf1cfbf7370d244c5ea92c7856d60695d95d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d0e7bf1cfbf7370d244c5ea92c7856d60695d95d"}]},"branch":"refs/heads/master"},"c9c0e033e02dcedefdaff1b40094a07d14ab379d":{"kind":"REWORK","_number":13,"created":"2024-07-31 14:10:17.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/52/920452/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/920452/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/920452/13"}}},"commit":{"parents":[{"commit":"ab5703d3f5edfbc92a1f775a48072cb85416a6e6","subject":"DNM expirer: new options to control task iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ab5703d3f5edfbc92a1f775a48072cb85416a6e6"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-05-24 19:27:26.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-07-31 14:06:27.000000000","tz":60},"subject":"Configurable expiring_objects_task_container_per_day","message":"Configurable expiring_objects_task_container_per_day\n\nIn order to better scale the object-expirer cluster operators may\nconfigure expiring_objects_task_container_per_day to increase the number\nof task_containers created per day beyond the default value of one\nhundred to any value that better matches their cluster\u0027s topology and\nusage of the x-delete-at feature.\n\nUpgradeImpact: Two existing expirer options have been deprecated\n\n * expiring_objects_container_divisor\n * expiring_objects_account_name\n\nThe upstream maintainers are not aware of any clusters where these have\nbeen configured to non-default values.\n\nOperators are encouraged to remove their \"container_divisor\" setting and\nuse the default value of 86400 which will make more sense with\n\"task_container_per_day\".\n\nIf a cluster was deployed with a non-standard \"account_name\" operators\nshould remove the option from all configs so they are using a supported\nconfiguration going forward, but will need to deploy a stand-alone\nlegacy expirer config to clean-up old expiration tasks from the\npreviously configured account name.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I5ea9e6dc8b44c8c5f55837debe24dd76be7d6248\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c9c0e033e02dcedefdaff1b40094a07d14ab379d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c9c0e033e02dcedefdaff1b40094a07d14ab379d"}]},"branch":"refs/heads/master"},"e6d93f9effdeda2aaa46975b033bc76cdb222b1b":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2024-08-02 11:36:31.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/52/920452/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/920452/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/920452/14"}}},"commit":{"parents":[{"commit":"aa5f180203ddf7415e25bebe5c9c802477f2a5fd","subject":"DNM expirer: new options to control task iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/aa5f180203ddf7415e25bebe5c9c802477f2a5fd"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-05-24 19:27:26.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-08-02 11:20:55.000000000","tz":60},"subject":"Configurable expiring_objects_task_container_per_day","message":"Configurable expiring_objects_task_container_per_day\n\nIn order to better scale the object-expirer cluster operators may\nconfigure expiring_objects_task_container_per_day to increase the number\nof task_containers created per day beyond the default value of one\nhundred to any value that better matches their cluster\u0027s topology and\nusage of the x-delete-at feature.\n\nUpgradeImpact: Two existing expirer options have been deprecated\n\n * expiring_objects_container_divisor\n * expiring_objects_account_name\n\nThe upstream maintainers are not aware of any clusters where these have\nbeen configured to non-default values.\n\nOperators are encouraged to remove their \"container_divisor\" setting and\nuse the default value of 86400 which will make more sense with\n\"task_container_per_day\".\n\nIf a cluster was deployed with a non-standard \"account_name\" operators\nshould remove the option from all configs so they are using a supported\nconfiguration going forward, but will need to deploy a stand-alone\nlegacy expirer config to clean-up old expiration tasks from the\npreviously configured account name.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I5ea9e6dc8b44c8c5f55837debe24dd76be7d6248\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e6d93f9effdeda2aaa46975b033bc76cdb222b1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e6d93f9effdeda2aaa46975b033bc76cdb222b1b"}]},"branch":"refs/heads/master"},"93c63225b3fa485a2e8f4133ce4db5ac640db1be":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2024-08-02 13:30:58.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/52/920452/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/920452/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/920452/15"}}},"commit":{"parents":[{"commit":"e093bfc295ba204319e60039d153082bd78c5e0a","subject":"DNM expirer: new options to control task iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e093bfc295ba204319e60039d153082bd78c5e0a"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-05-24 19:27:26.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-08-02 13:23:59.000000000","tz":60},"subject":"Configurable expiring_objects_task_container_per_day","message":"Configurable expiring_objects_task_container_per_day\n\nIn order to better scale the object-expirer cluster operators may\nconfigure expiring_objects_task_container_per_day to increase the number\nof task_containers created per day beyond the default value of one\nhundred to any value that better matches their cluster\u0027s topology and\nusage of the x-delete-at feature.\n\nUpgradeImpact: Two existing expirer options have been deprecated\n\n * expiring_objects_container_divisor\n * expiring_objects_account_name\n\nThe upstream maintainers are not aware of any clusters where these have\nbeen configured to non-default values.\n\nOperators are encouraged to remove their \"container_divisor\" setting and\nuse the default value of 86400 which will make more sense with\n\"task_container_per_day\".\n\nIf a cluster was deployed with a non-standard \"account_name\" operators\nshould remove the option from all configs so they are using a supported\nconfiguration going forward, but will need to deploy a stand-alone\nlegacy expirer config to clean-up old expiration tasks from the\npreviously configured account name.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I5ea9e6dc8b44c8c5f55837debe24dd76be7d6248\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/93c63225b3fa485a2e8f4133ce4db5ac640db1be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/93c63225b3fa485a2e8f4133ce4db5ac640db1be"}]},"branch":"refs/heads/master"},"f128491303b3553f1fb1daf8c9e1ee79d6e23745":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2024-08-07 13:26:44.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/52/920452/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/920452/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/920452/16"}}},"commit":{"parents":[{"commit":"ac928f8b8b93e772d4cd9196236c8002944983e9","subject":"DNM expirer: new options to control task iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ac928f8b8b93e772d4cd9196236c8002944983e9"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-05-24 19:27:26.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-08-07 13:25:28.000000000","tz":60},"subject":"Configurable expiring_objects_task_container_per_day","message":"Configurable expiring_objects_task_container_per_day\n\nIn order to better scale the object-expirer cluster operators may\nconfigure expiring_objects_task_container_per_day to increase the number\nof task_containers created per day beyond the default value of one\nhundred to any value that better matches their cluster\u0027s topology and\nusage of the x-delete-at feature.\n\nUpgradeImpact: Two existing expirer options have been deprecated\n\n * expiring_objects_container_divisor\n * expiring_objects_account_name\n\nThe upstream maintainers are not aware of any clusters where these have\nbeen configured to non-default values.\n\nOperators are encouraged to remove their \"container_divisor\" setting and\nuse the default value of 86400 which will make more sense with\n\"task_container_per_day\".\n\nIf a cluster was deployed with a non-standard \"account_name\" operators\nshould remove the option from all configs so they are using a supported\nconfiguration going forward, but will need to deploy a stand-alone\nlegacy expirer config to clean-up old expiration tasks from the\npreviously configured account name.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I5ea9e6dc8b44c8c5f55837debe24dd76be7d6248\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f128491303b3553f1fb1daf8c9e1ee79d6e23745"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f128491303b3553f1fb1daf8c9e1ee79d6e23745"}]},"branch":"refs/heads/master"},"41fd33e45177cbe61b43075c4184bcbd605a8b76":{"kind":"REWORK","_number":17,"created":"2024-09-11 08:45:14.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/52/920452/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/920452/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/920452/17"}}},"commit":{"parents":[{"commit":"4caa65e5a3fc298d18da57a0038680bc92f0df75","subject":"DNM expirer: new options to control task iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4caa65e5a3fc298d18da57a0038680bc92f0df75"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-05-24 19:27:26.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-09-11 08:42:06.000000000","tz":60},"subject":"Configurable expiring_objects_task_container_per_day","message":"Configurable expiring_objects_task_container_per_day\n\nIn order to better scale the object-expirer cluster operators may\nconfigure expiring_objects_task_container_per_day to increase the number\nof task_containers created per day beyond the default value of one\nhundred to any value that better matches their cluster\u0027s topology and\nusage of the x-delete-at feature.\n\nUpgradeImpact: Two existing expirer options have been deprecated\n\n * expiring_objects_container_divisor\n * expiring_objects_account_name\n\nThe upstream maintainers are not aware of any clusters where these have\nbeen configured to non-default values.\n\nOperators are encouraged to remove their \"container_divisor\" setting and\nuse the default value of 86400 which will make more sense with\n\"task_container_per_day\".\n\nIf a cluster was deployed with a non-standard \"account_name\" operators\nshould remove the option from all configs so they are using a supported\nconfiguration going forward, but will need to deploy a stand-alone\nlegacy expirer config to clean-up old expiration tasks from the\npreviously configured account name.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I5ea9e6dc8b44c8c5f55837debe24dd76be7d6248\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/41fd33e45177cbe61b43075c4184bcbd605a8b76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/41fd33e45177cbe61b43075c4184bcbd605a8b76"}]},"branch":"refs/heads/master"},"2c920ec1b9b79300883aa1a3c1ffee3972163ab3":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2024-09-11 08:48:50.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/52/920452/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/920452/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/920452/18"}}},"commit":{"parents":[{"commit":"a1c4e1a467dbd796b22dae61810f0eb33e051f3e","subject":"DNM expirer: new options to control task iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a1c4e1a467dbd796b22dae61810f0eb33e051f3e"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-05-24 19:27:26.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-09-11 08:48:03.000000000","tz":60},"subject":"Configurable expiring_objects_task_container_per_day","message":"Configurable expiring_objects_task_container_per_day\n\nIn order to better scale the object-expirer cluster operators may\nconfigure expiring_objects_task_container_per_day to increase the number\nof task_containers created per day beyond the default value of one\nhundred to any value that better matches their cluster\u0027s topology and\nusage of the x-delete-at feature.\n\nUpgradeImpact: Two existing expirer options have been deprecated\n\n * expiring_objects_container_divisor\n * expiring_objects_account_name\n\nThe upstream maintainers are not aware of any clusters where these have\nbeen configured to non-default values.\n\nOperators are encouraged to remove their \"container_divisor\" setting and\nuse the default value of 86400 which will make more sense with\n\"task_container_per_day\".\n\nIf a cluster was deployed with a non-standard \"account_name\" operators\nshould remove the option from all configs so they are using a supported\nconfiguration going forward, but will need to deploy a stand-alone\nlegacy expirer config to clean-up old expiration tasks from the\npreviously configured account name.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I5ea9e6dc8b44c8c5f55837debe24dd76be7d6248\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2c920ec1b9b79300883aa1a3c1ffee3972163ab3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2c920ec1b9b79300883aa1a3c1ffee3972163ab3"}]},"branch":"refs/heads/master"},"c345a3d8ce9d71047bbff5a2f68d7d887652c488":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2024-09-11 22:00:32.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"real_uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/52/920452/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/920452/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/920452/19"}}},"commit":{"parents":[{"commit":"a92b3f39f270dc6e0ccc9a8084bdea70618e3b07","subject":"DNM expirer: new options to control task iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a92b3f39f270dc6e0ccc9a8084bdea70618e3b07"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-05-24 19:27:26.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-09-11 22:00:32.000000000","tz":0},"subject":"Configurable expiring_objects_task_container_per_day","message":"Configurable expiring_objects_task_container_per_day\n\nIn order to better scale the object-expirer cluster operators may\nconfigure expiring_objects_task_container_per_day to increase the number\nof task_containers created per day beyond the default value of one\nhundred to any value that better matches their cluster\u0027s topology and\nusage of the x-delete-at feature.\n\nUpgradeImpact: Two existing expirer options have been deprecated\n\n * expiring_objects_container_divisor\n * expiring_objects_account_name\n\nThe upstream maintainers are not aware of any clusters where these have\nbeen configured to non-default values.\n\nOperators are encouraged to remove their \"container_divisor\" setting and\nuse the default value of 86400 which will make more sense with\n\"task_container_per_day\".\n\nIf a cluster was deployed with a non-standard \"account_name\" operators\nshould remove the option from all configs so they are using a supported\nconfiguration going forward, but will need to deploy a stand-alone\nlegacy expirer config to clean-up old expiration tasks from the\npreviously configured account name.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I5ea9e6dc8b44c8c5f55837debe24dd76be7d6248\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c345a3d8ce9d71047bbff5a2f68d7d887652c488"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c345a3d8ce9d71047bbff5a2f68d7d887652c488"}]},"branch":"refs/heads/master","description":"Rebase"},"92706394d9b24c758e2c83c9c1db64c36175a950":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2024-10-02 23:08:47.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/52/920452/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/920452/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/920452/20"}}},"commit":{"parents":[{"commit":"849b0b4b6e8e14723d80c57ae6ad20b33bc83cc8","subject":"DNM expirer: new options to control task iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/849b0b4b6e8e14723d80c57ae6ad20b33bc83cc8"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-05-24 19:27:26.000000000","tz":-300},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-10-02 23:08:07.000000000","tz":-420},"subject":"Configurable expiring_objects_task_container_per_day","message":"Configurable expiring_objects_task_container_per_day\n\nIn order to better scale the object-expirer cluster operators may\nconfigure expiring_objects_task_container_per_day to increase the number\nof task_containers created per day beyond the default value of one\nhundred to any value that better matches their cluster\u0027s topology and\nusage of the x-delete-at feature.\n\nUpgradeImpact: Two existing expirer options have been deprecated\n\n * expiring_objects_container_divisor\n * expiring_objects_account_name\n\nThe upstream maintainers are not aware of any clusters where these have\nbeen configured to non-default values.\n\nOperators are encouraged to remove their \"container_divisor\" setting and\nuse the default value of 86400 which will make more sense with\n\"task_container_per_day\".\n\nIf a cluster was deployed with a non-standard \"account_name\" operators\nshould remove the option from all configs so they are using a supported\nconfiguration going forward, but will need to deploy a stand-alone\nlegacy expirer config to clean-up old expiration tasks from the\npreviously configured account name.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I5ea9e6dc8b44c8c5f55837debe24dd76be7d6248\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/92706394d9b24c758e2c83c9c1db64c36175a950"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/92706394d9b24c758e2c83c9c1db64c36175a950"}]},"branch":"refs/heads/master"},"4613ffbf9fe298789dee39f81c802454bfc9b1e4":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2024-10-04 21:01:35.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/52/920452/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/920452/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/920452/21"}}},"commit":{"parents":[{"commit":"c664d79256f649aa680d2dd68dc4368ba82b54f8","subject":"DNM expirer: new options to control task iteration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c664d79256f649aa680d2dd68dc4368ba82b54f8"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-05-24 19:27:26.000000000","tz":-300},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-10-04 21:01:10.000000000","tz":-420},"subject":"Configurable expiring_objects_task_container_per_day","message":"Configurable expiring_objects_task_container_per_day\n\nIn order to better scale the object-expirer cluster operators may\nconfigure expiring_objects_task_container_per_day to increase the number\nof task_containers created per day beyond the default value of one\nhundred to any value that better matches their cluster\u0027s topology and\nusage of the x-delete-at feature.\n\nUpgradeImpact: Two existing expirer options have been deprecated\n\n * expiring_objects_container_divisor\n * expiring_objects_account_name\n\nThe upstream maintainers are not aware of any clusters where these have\nbeen configured to non-default values.\n\nOperators are encouraged to remove their \"container_divisor\" setting and\nuse the default value of 86400 which will make more sense with\n\"task_container_per_day\".\n\nIf a cluster was deployed with a non-standard \"account_name\" operators\nshould remove the option from all configs so they are using a supported\nconfiguration going forward, but will need to deploy a stand-alone\nlegacy expirer config to clean-up old expiration tasks from the\npreviously configured account name.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I5ea9e6dc8b44c8c5f55837debe24dd76be7d6248\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4613ffbf9fe298789dee39f81c802454bfc9b1e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4613ffbf9fe298789dee39f81c802454bfc9b1e4"}]},"branch":"refs/heads/master"},"22983c09eefe45089a76cd7d343e50bf9508af8b":{"kind":"REWORK","_number":22,"created":"2024-10-24 21:11:26.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/52/920452/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/920452/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/920452/22"}}},"commit":{"parents":[{"commit":"8094e7d0b8f145a2ab2489560c7b03c6d4384ef5","subject":"add round_robin_cache_size","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8094e7d0b8f145a2ab2489560c7b03c6d4384ef5"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-10-16 14:23:12.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-10-24 18:06:45.000000000","tz":-300},"subject":"Deprecate expirer options","message":"Deprecate expirer options\n\nThe following configuration options are deprecated:\n\n * expiring_objects_container_divisor\n * expiring_objects_account_name\n\nThe upstream maintainers are not aware of any clusters where these have\nbeen configured to non-default values.\n\nUpgradeImpact:\n\nOperators are encouraged to remove their \"container_divisor\" setting and\nuse the default value of 86400.\n\nIf a cluster was deployed with a non-standard \"account_name\" operators\nshould remove the option from all configs so they are using a supported\nconfiguration going forward, but will need to deploy a stand-alone\nlegacy expirer config to clean-up old expiration tasks from the\npreviously configured account name.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I5ea9e6dc8b44c8c5f55837debe24dd76be7d6248\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/22983c09eefe45089a76cd7d343e50bf9508af8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/22983c09eefe45089a76cd7d343e50bf9508af8b"}]},"branch":"refs/heads/master"},"81fb6359f63277a2b2ab985e80622ad796e4caef":{"kind":"REWORK","_number":23,"created":"2024-10-28 22:11:18.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/52/920452/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/920452/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/920452/23"}}},"commit":{"parents":[{"commit":"798bc9759da318e7af0a9f5821af45cfc1ace198","subject":"object-expirer: add round_robin_cache_size option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/798bc9759da318e7af0a9f5821af45cfc1ace198"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-10-16 14:23:12.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-10-28 22:07:05.000000000","tz":-300},"subject":"Deprecate expirer options","message":"Deprecate expirer options\n\nThe following configuration options are deprecated:\n\n * expiring_objects_container_divisor\n * expiring_objects_account_name\n\nThe upstream maintainers are not aware of any clusters where these have\nbeen configured to non-default values.\n\nUpgradeImpact:\n\nOperators are encouraged to remove their \"container_divisor\" setting and\nuse the default value of 86400.\n\nIf a cluster was deployed with a non-standard \"account_name\" operators\nshould remove the option from all configs so they are using a supported\nconfiguration going forward, but will need to deploy a stand-alone\nlegacy expirer config to clean-up old expiration tasks from the\npreviously configured account name.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I5ea9e6dc8b44c8c5f55837debe24dd76be7d6248\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/81fb6359f63277a2b2ab985e80622ad796e4caef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/81fb6359f63277a2b2ab985e80622ad796e4caef"}]},"branch":"refs/heads/master"},"679ad81edd88a29d2d8a77ece500b87f350fb370":{"kind":"REWORK","_number":24,"created":"2024-10-31 11:15:16.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/52/920452/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/920452/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/920452/24"}}},"commit":{"parents":[{"commit":"5ba9f30919a920ed6fdd5278d544410fd694c9d5","subject":"object-expirer: add round_robin_cache_size option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5ba9f30919a920ed6fdd5278d544410fd694c9d5"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-10-16 14:23:12.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-31 11:14:59.000000000","tz":0},"subject":"Deprecate expirer options","message":"Deprecate expirer options\n\nThe following configuration options are deprecated:\n\n * expiring_objects_container_divisor\n * expiring_objects_account_name\n\nThe upstream maintainers are not aware of any clusters where these have\nbeen configured to non-default values.\n\nUpgradeImpact:\n\nOperators are encouraged to remove their \"container_divisor\" setting and\nuse the default value of 86400.\n\nIf a cluster was deployed with a non-standard \"account_name\" operators\nshould remove the option from all configs so they are using a supported\nconfiguration going forward, but will need to deploy a stand-alone\nlegacy expirer config to clean-up old expiration tasks from the\npreviously configured account name.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I5ea9e6dc8b44c8c5f55837debe24dd76be7d6248\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/679ad81edd88a29d2d8a77ece500b87f350fb370"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/679ad81edd88a29d2d8a77ece500b87f350fb370"}]},"branch":"refs/heads/master"},"4639e066e36bc3436c0d8555a536e96795218141":{"kind":"REWORK","_number":25,"created":"2024-10-31 13:31:35.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/52/920452/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/920452/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/920452/25"}}},"commit":{"parents":[{"commit":"5ba9f30919a920ed6fdd5278d544410fd694c9d5","subject":"object-expirer: add round_robin_cache_size option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5ba9f30919a920ed6fdd5278d544410fd694c9d5"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-10-16 14:23:12.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-10-31 13:30:59.000000000","tz":0},"subject":"Deprecate expirer options","message":"Deprecate expirer options\n\nThe following configuration options are deprecated:\n\n * expiring_objects_container_divisor\n * expiring_objects_account_name\n\nThe upstream maintainers are not aware of any clusters where these have\nbeen configured to non-default values.\n\nUpgradeImpact:\n\nOperators are encouraged to remove their \"container_divisor\" setting and\nuse the default value of 86400.\n\nIf a cluster was deployed with a non-standard \"account_name\" operators\nshould remove the option from all configs so they are using a supported\nconfiguration going forward, but will need to deploy a stand-alone\nlegacy expirer config to clean-up old expiration tasks from the\npreviously configured account name.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I5ea9e6dc8b44c8c5f55837debe24dd76be7d6248\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4639e066e36bc3436c0d8555a536e96795218141"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4639e066e36bc3436c0d8555a536e96795218141"}]},"branch":"refs/heads/master"},"5bd4cb7814138bbe46359a11b867031c869c0845":{"kind":"TRIVIAL_REBASE","_number":26,"created":"2024-11-01 09:55:05.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/52/920452/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/920452/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/920452/26"}}},"commit":{"parents":[{"commit":"df22032d79c74c047974c360331463a0e062c0ea","subject":"object-expirer: add round_robin_cache_size option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/df22032d79c74c047974c360331463a0e062c0ea"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-10-16 14:23:12.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-01 09:54:54.000000000","tz":0},"subject":"Deprecate expirer options","message":"Deprecate expirer options\n\nThe following configuration options are deprecated:\n\n * expiring_objects_container_divisor\n * expiring_objects_account_name\n\nThe upstream maintainers are not aware of any clusters where these have\nbeen configured to non-default values.\n\nUpgradeImpact:\n\nOperators are encouraged to remove their \"container_divisor\" setting and\nuse the default value of 86400.\n\nIf a cluster was deployed with a non-standard \"account_name\" operators\nshould remove the option from all configs so they are using a supported\nconfiguration going forward, but will need to deploy a stand-alone\nlegacy expirer config to clean-up old expiration tasks from the\npreviously configured account name.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I5ea9e6dc8b44c8c5f55837debe24dd76be7d6248\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5bd4cb7814138bbe46359a11b867031c869c0845"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5bd4cb7814138bbe46359a11b867031c869c0845"}]},"branch":"refs/heads/master"},"538c82e0fb554328d8ce72e513c583db2c20045a":{"kind":"REWORK","_number":27,"created":"2024-11-01 20:13:43.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/52/920452/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/920452/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/920452/27"}}},"commit":{"parents":[{"commit":"5ee1e72c0b03c70da8a4f07b106ff57aaf03cc10","subject":"Merge \"object-expirer: add round_robin_cache_size option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5ee1e72c0b03c70da8a4f07b106ff57aaf03cc10"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-10-16 14:23:12.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-11-01 19:08:42.000000000","tz":-300},"subject":"Deprecate expirer options","message":"Deprecate expirer options\n\nThe following configuration options are deprecated:\n\n * expiring_objects_container_divisor\n * expiring_objects_account_name\n\nThe upstream maintainers are not aware of any clusters where these have\nbeen configured to non-default values.\n\nUpgradeImpact:\n\nOperators are encouraged to remove their \"container_divisor\" setting and\nuse the default value of 86400.\n\nIf a cluster was deployed with a non-standard \"account_name\" operators\nshould remove the option from all configs so they are using a supported\nconfiguration going forward, but will need to deploy a stand-alone\nlegacy expirer config to clean-up old expiration tasks from the\npreviously configured account name.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I5ea9e6dc8b44c8c5f55837debe24dd76be7d6248\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/538c82e0fb554328d8ce72e513c583db2c20045a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/538c82e0fb554328d8ce72e513c583db2c20045a"}]},"branch":"refs/heads/master"},"63af9fd3ffa51b14c25345e00f0cb543dd732d36":{"kind":"REWORK","_number":28,"created":"2024-11-04 00:06:02.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/52/920452/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/920452/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/920452/28"}}},"commit":{"parents":[{"commit":"5ee1e72c0b03c70da8a4f07b106ff57aaf03cc10","subject":"Merge \"object-expirer: add round_robin_cache_size option\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5ee1e72c0b03c70da8a4f07b106ff57aaf03cc10"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-10-16 14:23:12.000000000","tz":-300},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-11-03 23:24:43.000000000","tz":-480},"subject":"Deprecate expirer options","message":"Deprecate expirer options\n\nThe following configuration options are deprecated:\n\n * expiring_objects_container_divisor\n * expiring_objects_account_name\n\nThe upstream maintainers are not aware of any clusters where these have\nbeen configured to non-default values.\n\nUpgradeImpact:\n\nOperators are encouraged to remove their \"container_divisor\" setting and\nuse the default value of 86400.\n\nIf a cluster was deployed with a non-standard \"account_name\" operators\nshould remove the option from all configs so they are using a supported\nconfiguration going forward, but will need to deploy a stand-alone\nlegacy expirer config to clean-up old expiration tasks from the\npreviously configured account name.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I5ea9e6dc8b44c8c5f55837debe24dd76be7d6248\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/63af9fd3ffa51b14c25345e00f0cb543dd732d36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/63af9fd3ffa51b14c25345e00f0cb543dd732d36"}]},"branch":"refs/heads/master"},"a39f35f4b20d7e66639172aa46ccb0d94bf613d9":{"kind":"REWORK","_number":29,"created":"2024-11-05 00:21:14.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/52/920452/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/920452/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/920452/29"}}},"commit":{"parents":[{"commit":"97c5deae87e4120f09df38382ffeb6d913eb065c","subject":"Merge \"slo: add more functional tests for copying\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/97c5deae87e4120f09df38382ffeb6d913eb065c"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-10-16 14:23:12.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-11-05 00:16:15.000000000","tz":-360},"subject":"Deprecate expirer options","message":"Deprecate expirer options\n\nThe following configuration options are deprecated:\n\n * expiring_objects_container_divisor\n * expiring_objects_account_name\n\nThe upstream maintainers are not aware of any clusters where these have\nbeen configured to non-default values.\n\nUpgradeImpact:\n\nOperators are encouraged to remove their \"container_divisor\" setting and\nuse the default value of 86400.\n\nIf a cluster was deployed with a non-standard \"account_name\" operators\nshould remove the option from all configs so they are using a supported\nconfiguration going forward, but will need to deploy a stand-alone\nlegacy expirer config to clean-up old expiration tasks from the\npreviously configured account name.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nChange-Id: I5ea9e6dc8b44c8c5f55837debe24dd76be7d6248\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a39f35f4b20d7e66639172aa46ccb0d94bf613d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a39f35f4b20d7e66639172aa46ccb0d94bf613d9"}]},"branch":"refs/heads/master"},"ba69dbc3b07cc933c9726e138e3becb8dfedb494":{"kind":"TRIVIAL_REBASE","_number":30,"created":"2024-11-05 22:52:54.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/52/920452/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/920452/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/920452/30"}}},"commit":{"parents":[{"commit":"7662cde704849b424ad3497b10f5fef35ae0a947","subject":"Merge \"Add oldest failed async pending tracker\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7662cde704849b424ad3497b10f5fef35ae0a947"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-10-16 14:23:12.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-11-05 22:52:25.000000000","tz":-360},"subject":"Deprecate expirer options","message":"Deprecate expirer options\n\nThe following configuration options are deprecated:\n\n * expiring_objects_container_divisor\n * expiring_objects_account_name\n\nThe upstream maintainers are not aware of any clusters where these have\nbeen configured to non-default values.\n\nUpgradeImpact:\n\nOperators are encouraged to remove their \"container_divisor\" setting and\nuse the default value of 86400.\n\nIf a cluster was deployed with a non-standard \"account_name\" operators\nshould remove the option from all configs so they are using a supported\nconfiguration going forward, but will need to deploy a stand-alone\nlegacy expirer config to clean-up old expiration tasks from the\npreviously configured account name.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nChange-Id: I5ea9e6dc8b44c8c5f55837debe24dd76be7d6248\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ba69dbc3b07cc933c9726e138e3becb8dfedb494"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ba69dbc3b07cc933c9726e138e3becb8dfedb494"}]},"branch":"refs/heads/master"},"b5cac26392139b300faf44ba4401c6377bdeca4c":{"kind":"TRIVIAL_REBASE","_number":31,"created":"2024-11-12 17:02:51.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/52/920452/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/920452/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/920452/31"}}},"commit":{"parents":[{"commit":"309b906df82fe92826dcda230d1410a534e0be3f","subject":"Merge \"Prevent installation with pyeclib 1.6.2 and 1.6.3\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/309b906df82fe92826dcda230d1410a534e0be3f"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-10-16 14:23:12.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-11-12 16:57:09.000000000","tz":-360},"subject":"Deprecate expirer options","message":"Deprecate expirer options\n\nThe following configuration options are deprecated:\n\n * expiring_objects_container_divisor\n * expiring_objects_account_name\n\nThe upstream maintainers are not aware of any clusters where these have\nbeen configured to non-default values.\n\nUpgradeImpact:\n\nOperators are encouraged to remove their \"container_divisor\" setting and\nuse the default value of 86400.\n\nIf a cluster was deployed with a non-standard \"account_name\" operators\nshould remove the option from all configs so they are using a supported\nconfiguration going forward, but will need to deploy a stand-alone\nlegacy expirer config to clean-up old expiration tasks from the\npreviously configured account name.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nChange-Id: I5ea9e6dc8b44c8c5f55837debe24dd76be7d6248\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b5cac26392139b300faf44ba4401c6377bdeca4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b5cac26392139b300faf44ba4401c6377bdeca4c"}]},"branch":"refs/heads/master"},"4d5b379862fb7ba9809bb61be67f05695cf49308":{"kind":"TRIVIAL_REBASE","_number":32,"created":"2024-11-15 22:58:23.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/52/920452/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/920452/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/920452/32"}}},"commit":{"parents":[{"commit":"89815389d5376195d9bc1b0bf65f88e9fb7ce637","subject":"Merge \"probe tests: Set default timeout for subprocesses\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/89815389d5376195d9bc1b0bf65f88e9fb7ce637"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-10-16 14:23:12.000000000","tz":-300},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2024-11-15 20:28:11.000000000","tz":-480},"subject":"Deprecate expirer options","message":"Deprecate expirer options\n\nThe following configuration options are deprecated:\n\n * expiring_objects_container_divisor\n * expiring_objects_account_name\n\nThe upstream maintainers are not aware of any clusters where these have\nbeen configured to non-default values.\n\nUpgradeImpact:\n\nOperators are encouraged to remove their \"container_divisor\" setting and\nuse the default value of 86400.\n\nIf a cluster was deployed with a non-standard \"account_name\" operators\nshould remove the option from all configs so they are using a supported\nconfiguration going forward, but will need to deploy a stand-alone\nlegacy expirer config to clean-up old expiration tasks from the\npreviously configured account name.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nChange-Id: I5ea9e6dc8b44c8c5f55837debe24dd76be7d6248\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4d5b379862fb7ba9809bb61be67f05695cf49308"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4d5b379862fb7ba9809bb61be67f05695cf49308"}]},"branch":"refs/heads/master"},"0445dd9626b5d5a6243c31f5706487da0e92e4c2":{"kind":"REWORK","_number":33,"created":"2024-11-18 19:09:46.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/52/920452/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/920452/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/920452/33"}}},"commit":{"parents":[{"commit":"ffbf17e47c0c76d4daa6e40811338e4abec1beb4","subject":"Fix duplicate prefix in exception logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ffbf17e47c0c76d4daa6e40811338e4abec1beb4"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-10-16 14:23:12.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-11-18 19:04:33.000000000","tz":-360},"subject":"Deprecate expirer options","message":"Deprecate expirer options\n\nThe following configuration options are deprecated:\n\n * expiring_objects_container_divisor\n * expiring_objects_account_name\n\nThe upstream maintainers are not aware of any clusters where these have\nbeen configured to non-default values.\n\nUpgradeImpact:\n\nOperators are encouraged to remove their \"container_divisor\" setting and\nuse the default value of 86400.\n\nIf a cluster was deployed with a non-standard \"account_name\" operators\nshould remove the option from all configs so they are using a supported\nconfiguration going forward, but will need to deploy a stand-alone\nlegacy expirer config to clean-up old expiration tasks from the\npreviously configured account name.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nChange-Id: I5ea9e6dc8b44c8c5f55837debe24dd76be7d6248\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0445dd9626b5d5a6243c31f5706487da0e92e4c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0445dd9626b5d5a6243c31f5706487da0e92e4c2"}]},"branch":"refs/heads/master"},"c7988c0a6830ea7fa2d9b14a5eb92a03f6f8e709":{"kind":"TRIVIAL_REBASE","_number":34,"created":"2024-11-29 14:47:58.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/52/920452/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/920452/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/920452/34"}}},"commit":{"parents":[{"commit":"0d9ea8d9bd00716483e812afbbd9324881c80bc9","subject":"Merge \"Add probe test for reconciler with object versioning\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0d9ea8d9bd00716483e812afbbd9324881c80bc9"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-10-16 14:23:12.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-11-29 14:38:15.000000000","tz":0},"subject":"Deprecate expirer options","message":"Deprecate expirer options\n\nThe following configuration options are deprecated:\n\n * expiring_objects_container_divisor\n * expiring_objects_account_name\n\nThe upstream maintainers are not aware of any clusters where these have\nbeen configured to non-default values.\n\nUpgradeImpact:\n\nOperators are encouraged to remove their \"container_divisor\" setting and\nuse the default value of 86400.\n\nIf a cluster was deployed with a non-standard \"account_name\" operators\nshould remove the option from all configs so they are using a supported\nconfiguration going forward, but will need to deploy a stand-alone\nlegacy expirer config to clean-up old expiration tasks from the\npreviously configured account name.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nChange-Id: I5ea9e6dc8b44c8c5f55837debe24dd76be7d6248\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c7988c0a6830ea7fa2d9b14a5eb92a03f6f8e709"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c7988c0a6830ea7fa2d9b14a5eb92a03f6f8e709"}]},"branch":"refs/heads/master"},"287fa42ab8ff101a8ab59aebd44f25028d096c1b":{"kind":"TRIVIAL_REBASE","_number":35,"created":"2025-01-16 12:23:57.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/52/920452/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/920452/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/920452/35"}}},"commit":{"parents":[{"commit":"ae6300af86c292b4ded1b4606f7e978bf7474f8a","subject":"wsgi: Reap stale workers (after a timeout) following a reload","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ae6300af86c292b4ded1b4606f7e978bf7474f8a"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-10-16 14:23:12.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-01-16 12:16:53.000000000","tz":0},"subject":"Deprecate expirer options","message":"Deprecate expirer options\n\nThe following configuration options are deprecated:\n\n * expiring_objects_container_divisor\n * expiring_objects_account_name\n\nThe upstream maintainers are not aware of any clusters where these have\nbeen configured to non-default values.\n\nUpgradeImpact:\n\nOperators are encouraged to remove their \"container_divisor\" setting and\nuse the default value of 86400.\n\nIf a cluster was deployed with a non-standard \"account_name\" operators\nshould remove the option from all configs so they are using a supported\nconfiguration going forward, but will need to deploy a stand-alone\nlegacy expirer config to clean-up old expiration tasks from the\npreviously configured account name.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nChange-Id: I5ea9e6dc8b44c8c5f55837debe24dd76be7d6248\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/287fa42ab8ff101a8ab59aebd44f25028d096c1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/287fa42ab8ff101a8ab59aebd44f25028d096c1b"}]},"branch":"refs/heads/master"},"b69a2bef45b100a2bc06b7fbb2eacd06380b117b":{"kind":"REWORK","_number":36,"created":"2025-02-07 16:33:48.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/52/920452/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/920452/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/920452/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/920452/36"}}},"commit":{"parents":[{"commit":"ae6300af86c292b4ded1b4606f7e978bf7474f8a","subject":"wsgi: Reap stale workers (after a timeout) following a reload","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ae6300af86c292b4ded1b4606f7e978bf7474f8a"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-10-16 14:23:12.000000000","tz":-300},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-02-07 16:33:34.000000000","tz":-480},"subject":"Deprecate expirer options","message":"Deprecate expirer options\n\nThe following configuration options are deprecated:\n\n * expiring_objects_container_divisor\n * expiring_objects_account_name\n\nThe upstream maintainers are not aware of any clusters where these have\nbeen configured to non-default values.\n\nUpgradeImpact:\n\nOperators are encouraged to remove their \"container_divisor\" setting and\nuse the default value of 86400.\n\nIf a cluster was deployed with a non-standard \"account_name\", operators\nshould remove the option from all configs so they are using a supported\nconfiguration going forward, but will need to deploy stand-alone expirer\nprocesses with legacy expirer config to clean-up old expiration tasks\nfrom the previously configured account name.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Jianjian Huo \u003cjhuo@nvidia.com\u003e\nChange-Id: I5ea9e6dc8b44c8c5f55837debe24dd76be7d6248\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b69a2bef45b100a2bc06b7fbb2eacd06380b117b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b69a2bef45b100a2bc06b7fbb2eacd06380b117b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
