)]}'
{"id":"openstack%2Fswift~920861","triplet_id":"openstack%2Fswift~master~I087d7f4840f651dd607cac12a4230f95082f2368","project":"openstack/swift","branch":"master","topic":"remove-multipart-orphans","attention_set":{"36487":{"account":{"_account_id":36487,"name":"Rafael Tavares Oliveira","display_name":"fulecorafa","email":"rafael.oliveira@profusion.mobi","username":"fulecorafa"},"last_update":"2024-05-30 22:24:08.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I087d7f4840f651dd607cac12a4230f95082f2368","subject":"object-storage: clean up orphan segments daemon","status":"NEW","created":"2024-05-30 20:40:24.000000000","updated":"2024-05-30 22:24:08.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"34de81bf0b11699840755374a7757905bdf1e9e9","_number":920861,"virtual_id_number":920861,"owner":{"_account_id":36487,"name":"Rafael Tavares Oliveira","display_name":"fulecorafa","email":"rafael.oliveira@profusion.mobi","username":"fulecorafa"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-05-30 22:24:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-30 20:50:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-05-30 22:24:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b02748ca1d3dba9043b30b79859ffb41ed7fe273","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36487,"name":"Rafael Tavares Oliveira","display_name":"fulecorafa","email":"rafael.oliveira@profusion.mobi","username":"fulecorafa"},"date":"2024-05-30 20:40:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3f2967e7b5271d61d3f8ca295ddc47f5f0e78291","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-30 20:50:30.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d20add2e80844709322686a6f2f7312\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/43982a2b4fc34ec2a5fb580cf495de42 : RETRY_LIMIT in 1m 14s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d0399a21d0de4d8ebbe7b43bbd2b6e0b : RETRY_LIMIT in 2m 03s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/6e11dbc30adc44579613697e24bd7392 : RETRY_LIMIT in 1m 03s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/e22c56a03eee49e3a62d81eb2f798366 : RETRY_LIMIT in 1m 05s","accounts_in_message":[],"_revision_number":1},{"id":"34de81bf0b11699840755374a7757905bdf1e9e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-30 22:24:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bf8a1cc2041043788e6178ffccb24ba9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2183fd97067e41a1b45f8c0f5ac8dd82 : SUCCESS in 8m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/5765ab8a47e544a8af5e37dcd81df493 : SUCCESS in 1h 04m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cb2b506f7a304dd38a2a99eac0cc17de : SUCCESS in 51m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a7f36009fe7e4a759792fa067995f379 : SUCCESS in 1h 06m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ae76f2dfae124368b89e49b73389a196 : SUCCESS in 18m 15s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ca2f826668e14db98aa3390065c730c9 : SUCCESS in 20m 30s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c7bbf6769c42401880ff084c7ab5386b : SUCCESS in 23m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f6ea045b3531403391e67d1d0f37389a : SUCCESS in 11m 01s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/93283ff5a6cb46edac9c2d5ed04a6fa2 : SUCCESS in 6m 44s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a4851dcfe7fc4e6aa92b8fea59c339c0 : FAILURE in 20m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/48900803d4474174951bf61141072984 : FAILURE in 18m 43s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/09fad3f9a9d343acabfaa7a4c2816b87 : FAILURE in 17m 27s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/01f1ffafef484b33ab55931667246374 : FAILURE in 16m 19s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/cfa1173f0ed24cb4b6b8373bcfa2a1d0 : FAILURE in 25m 04s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/36e35b8e966546709687eb9addd53ffe : TIMED_OUT in 1h 30m 47s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2ac81f690caf41569dfc126a0ab454b8 : TIMED_OUT in 1h 31m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5e6ece1df1c843ddbb531e63f5326e75 : TIMED_OUT in 1h 31m 42s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7f28a864b8be4e3f9417dd79821589e6 : SUCCESS in 15m 28s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/da533299384e4ce2b2ba0c4daa70983b : SUCCESS in 18m 46s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/10f9bdbddff142dfa47b18e959662a97 : SUCCESS in 15m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/351a4eeb9eb84364b32c47b9a541fa39 : SUCCESS in 6m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c02970740e7c4fbaab1783106e5f6291 : FAILURE in 17m 28s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5bc4f3cbdb4b402e90ebc5bafe7af2c6 : SUCCESS in 7m 32s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3053dbaa58af4fe18cda63a4e4e3b675 : FAILURE in 54m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8d96ea103d5b46a88c709fa8eb83376d : FAILURE in 46m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/de2594fedd6f45ecb52b0d6698554a9a : FAILURE in 40m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cb03b2a29b9a4791893a30bce0708bde : FAILURE in 19m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d331b1e227e846ab9bb155e0f6fe8877 : SUCCESS in 4m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/684fbf36df6c4506818000e0f44a4eb0 : SUCCESS in 20m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/99f9f443ee274033b6ae3d543bc287d8 : SUCCESS in 45m 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4e8bbfb6f573acb8ac1184b88c017f4e3ce07177","revisions":{"4e8bbfb6f573acb8ac1184b88c017f4e3ce07177":{"kind":"REWORK","_number":1,"created":"2024-05-30 20:40:24.000000000","uploader":{"_account_id":36487,"name":"Rafael Tavares Oliveira","display_name":"fulecorafa","email":"rafael.oliveira@profusion.mobi","username":"fulecorafa"},"ref":"refs/changes/61/920861/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/61/920861/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/920861/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/920861/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/61/920861/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/61/920861/1"}}},"commit":{"parents":[{"commit":"00398d0ca3c34ac43a40c36d2629bebd4edf66dc","subject":"object-storage: clean up orphan segment parts in MPU","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/00398d0ca3c34ac43a40c36d2629bebd4edf66dc"}]}],"author":{"name":"Rafael Oliveira","email":"rafael.oliveira@profusion.mobi","date":"2024-05-29 19:09:48.000000000","tz":-180},"committer":{"name":"Rafael Oliveira","email":"rafael.oliveira@profusion.mobi","date":"2024-05-30 20:32:59.000000000","tz":-180},"subject":"object-storage: clean up orphan segments daemon","message":"object-storage: clean up orphan segments daemon\n\nOverwriting an SLO object with a new object would result in\nthe SLO parts to not be valid anymore, but still take disk\nspace (called orphan parts). These orphans should be removed\nto clear disk space.\n\nTo actually collect the objects, a daemon was created to\nretrieve task objects registered by s3api middlewares.\nIt considers a TTL for objects so it doesn\u0027t delete parts\nof ongoing uploads.\n\nThis is very based on obj/expirer, working mostly the same\nway.\n\nOn behalf of Magalu at https://magalu.cloud/\n\nPartial-Bug: #1813202\nChange-Id: I087d7f4840f651dd607cac12a4230f95082f2368\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4e8bbfb6f573acb8ac1184b88c017f4e3ce07177"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4e8bbfb6f573acb8ac1184b88c017f4e3ce07177"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
