)]}'
{"id":"openstack%2Fswift~921318","triplet_id":"openstack%2Fswift~master~Id2662bfc5ef2f21f901f1c98e6389c4cb01818a2","project":"openstack/swift","branch":"master","topic":"p-s3-inventory","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-06-04 20:45:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id2662bfc5ef2f21f901f1c98e6389c4cb01818a2","subject":"Lazy import is not needed","status":"MERGED","created":"2024-06-04 18:11:10.000000000","updated":"2024-06-04 20:48:23.000000000","submitted":"2024-06-04 20:45:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"921318-p-s3-inventory","meta_rev_id":"7526de6b3df4f42e345a4e9a1f79d5051b3c9c95","_number":921318,"virtual_id_number":921318,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-04 20:45:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":2,"date":"2024-06-04 18:36:53.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-06-04 18:13:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":1,"date":"2024-06-04 18:36:53.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-06-04 18:13:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-04 18:13:11.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2024-06-04 18:34:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-04 18:36:53.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2024-06-04 19:17:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"53577a00f87f4fde755057b5ee9f239f19bb25b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-06-04 18:11:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b3ed4a40c2d0049d1b9063eb1c48a2b7c21e1e46","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-04 18:13:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f4e44d72be8132f577bd4dd37a8f7295052beb6e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 18:34:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09dbae5263c34099a789f421619572f9\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/dd9c0e321e0542ea8a822f34fcca3d89 : SUCCESS in 17m 52s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/55320571e8754398b638b8d74fb2b835 : SUCCESS in 21m 07s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/1c04eaf9dcfd4a3195d0db15ac20e8c2 : SUCCESS in 20m 45s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6b39ba9196e242d4a2de5e5f06c642c2 : SUCCESS in 19m 46s","accounts_in_message":[],"_revision_number":1},{"id":"cc8ebdde234f8499f93f8cda8ba8b033000808f1","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2024-06-04 18:36:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3b58e51e0b7e02a004bf02a159ef02f6b8ea3998","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 19:17:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/846732f517cb4b21a636284659d638a7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b3b9f12082c4247a964b1ab93051d08 : SUCCESS in 8m 55s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e19dc2074a1447468ef60f4a0436ee00 : SUCCESS in 1h 01m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70875f9f64164281ab1f8fd657755e06 : SUCCESS in 4m 10s","accounts_in_message":[],"_revision_number":1},{"id":"7946e545521fd12e72399d4435e0a7e62e031eb3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 19:18:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d48b3da278af2552a47514fb2af857fb70ba9db7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 20:45:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1419356ad23d48e98a1d3ff8943f632d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ed1907e037b481c80a3a28568b34a1c : SUCCESS in 7m 47s\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/92553599c7ad49c8a6da65ba02f4fdc6 : SUCCESS in 21m 40s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/af7131627fae4568b9282354df4930ed : SUCCESS in 21m 10s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/13553e26195444f6bcbc2b0b6b5d6ace : SUCCESS in 18m 28s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/46816aafcf4d489ca1ebf3c7a42c47b5 : SUCCESS in 17m 30s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/bbcd13b1435a4aafa84e6c8238c1a35a : SUCCESS in 23m 01s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3682814d8063474285eb6c55e19c0add : SUCCESS in 19m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3e5a108b3d4e444f9bfd1460b1e19d64 : SUCCESS in 18m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7b2d780663d544e99591719d1c7c629d : SUCCESS in 19m 26s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1ba0fd6c27e94a6b9d255694c9500539 : SUCCESS in 17m 25s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c507b21d08d241e8a6e3295ff8e12147 : SUCCESS in 16m 22s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0aaa979a7a7244129df6ea3c5667eb1d : SUCCESS in 18m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/85a53af770d34712816f1bc079459850 : SUCCESS in 6m 06s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5ac96dea9f7642c48817bb232da14112 : SUCCESS in 6m 12s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e2659f6218384d638fce276d2961a7bd : SUCCESS in 1h 02m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19a8240aed3e45ceb4ecbcc9f79fd4fe : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":1},{"id":"7ddc92e7d6fac8f0445a117e2291767e160f7315","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 20:45:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7526de6b3df4f42e345a4e9a1f79d5051b3c9c95","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-04 20:48:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/042b9a91b87347668aee5941e59b1161\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41ebbc46daff4654b4d1a3d5bc37f487 : SUCCESS in 45s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/767ece27a52c4aca8e718bf316948ee9 : FAILURE in 28s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/731a05d70226405f9205360c983d57b3 : FAILURE in 28s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a666010aae8aa9ba67341eeb5e69380fc20d02f5","revisions":{"a666010aae8aa9ba67341eeb5e69380fc20d02f5":{"kind":"REWORK","_number":1,"created":"2024-06-04 18:11:10.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/18/921318/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/921318/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/921318/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/921318/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/921318/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/921318/1"}}},"commit":{"parents":[{"commit":"99c629edb813c7d9fcd736f5d34b3a47d224a457","subject":"tests: realistic task_container names","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/99c629edb813c7d9fcd736f5d34b3a47d224a457"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-04 17:26:55.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-04 17:42:20.000000000","tz":-300},"subject":"Lazy import is not needed","message":"Lazy import is not needed\n\nThere was an abandoned change that made reference to a RecussionError\nwhen running a probe test that imported boto3 that had something to do\nwith eventlet, ssl and a transitive dependency on requests-mock, but the\nfix that actually got merged seemed to depend on another change to\ntox.ini that disables request-mock when we run pytest.\n\nEither way, we already import from boto3 at the top of probe tests and\nit\u0027s in test-requirements; so we require it to be installed even if you\ndon\u0027t have s3api in your pipeline.\n\nRelated-Change: I789b257635c031ac0cb6e4b5980f741e0cb5244d\nRelated-Change: I2793e335a08ad373c49cbbe6759d4e97cc420867\nRelated-Change: If14e4d2c1af2efcbc99e9b6fe10973a7eb94d589\n\nChange-Id: Id2662bfc5ef2f21f901f1c98e6389c4cb01818a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a666010aae8aa9ba67341eeb5e69380fc20d02f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a666010aae8aa9ba67341eeb5e69380fc20d02f5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
