)]}'
{"id":"openstack%2Fswift~921451","triplet_id":"openstack%2Fswift~master~I0a499dc5c50bb534a23adcbfe7c3822d8954b0e0","project":"openstack/swift","branch":"master","topic":"p-s3api-log-info-errors","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-06-25 07:21:10.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-06-25 07:21:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0a499dc5c50bb534a23adcbfe7c3822d8954b0e0","subject":"s3api: include error response reason in log_info","status":"MERGED","created":"2024-06-06 13:49:22.000000000","updated":"2024-06-25 07:22:18.000000000","submitted":"2024-06-25 07:21:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"921451-p-s3api-log-info-errors","meta_rev_id":"c72a2a38188825e454db4ef2cee71021167d1e44","_number":921451,"virtual_id_number":921451,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-25 07:21:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2024-06-24 07:15:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2024-06-25 06:10:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-06 15:14:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-18 22:20:26.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2024-06-24 07:15:39.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"f226698c4ae3f273fb57315a96a1a891d4c8439d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-06-06 13:49:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dfa2415cb74dcff15974267c0d5946b22ad106c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-06 15:14:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b141111d9e824e8a9d3871a34a37d3fd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/200f31dfa66c46f390eeb0d4292bb66e : SUCCESS in 9m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/9fb7f062f19e4f6e8deac5b3f738a9a1 : SUCCESS in 1h 16m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/08df11b88f58493b89fe0c0e7f385c67 : SUCCESS in 59m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f9de53bf7ca045b08dfdd91d93ae0c9d : SUCCESS in 1h 14m 04s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cc9185bfc30241cab85364bd62759903 : SUCCESS in 17m 52s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9ae70d36a8da416aa98011bbc2af3bf2 : SUCCESS in 19m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/49d2f3c6d6cc425fa87261ed3616c906 : SUCCESS in 21m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e4bcc77a5a9644cab4b2e74a5bcb9023 : SUCCESS in 10m 46s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a281aa5659c94ccc91dbe4f8cdde80e4 : SUCCESS in 5m 19s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/dc8258e63db74402a74ec364e0f91ba0 : FAILURE in 4m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5293d3a030e041d39259b6e2ca3f9f57 : FAILURE in 4m 28s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9be618132581457badda0c3b06064077 : SUCCESS in 19m 59s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/13e3bfe1e55a4de4b54d1d6bca09b094 : SUCCESS in 17m 13s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4b9cc26bbdfd4911860b4e8f143ca794 : SUCCESS in 24m 14s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fa52384adf7e48e180bd019a5f0458ab : SUCCESS in 19m 11s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/8ffdcce266874d8cb2f8b4fe08c4ddf2 : SUCCESS in 18m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3c4d484ea0e74b37988504d6a74694d6 : SUCCESS in 18m 57s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/579eb80c3b6d4ee3a5c409c0acd274a6 : SUCCESS in 16m 53s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a15cd5717ec84503a7d22732978f065c : SUCCESS in 15m 40s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/82ecd540ad9e4824b184959cc6a1570f : SUCCESS in 18m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f55ff03505d3480e80ee7102a2f19304 : SUCCESS in 6m 18s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e20bfb56e1ad48a08acfeae572893b52 : FAILURE in 13m 51s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/819fca541ef8496f9aad60fb0dd28a40 : SUCCESS in 7m 00s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7480e17784ab44a29969c8c290282fb2 : SUCCESS in 54m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4c234c6c68ea4519872332f53f61264b : SUCCESS in 57m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/780ab3d004d94f50a2a15ca8b91d10af : SUCCESS in 50m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/33f1b6954436463ebe38646e648aa2ac : FAILURE in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/379464aefb5a446ba42eba56c02d8b6b : SUCCESS in 4m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/490c9f16b4474be69f366c91f710cb09 : SUCCESS in 17m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cb2304650b8e4c7d8290405e7bd770a7 : SUCCESS in 55m 43s","accounts_in_message":[],"_revision_number":1},{"id":"22a03b8ac84ba8db8fffc425a887d44df877380b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-06 15:17:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e0ead286c7c4d808f27cb41e34c8fef\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/391eac84303042b0b6530de5e7662a56 : SUCCESS in 15m 17s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/35fcd5951bac4932a1cd5bd93ef117e5 : SUCCESS in 16m 28s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/3c895c123d164986938f626f1af08551 : SUCCESS in 1h 25m 55s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/8bdde888e25046b0b7b9b193c97e180b : SUCCESS in 17m 52s","accounts_in_message":[],"_revision_number":1},{"id":"54d211bcdca72067967f392395a74ab0b8c0c67b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-06-07 09:59:04.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a5a7e47904725f99c4968e41a55a9a3c10fb5cba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-07 10:27:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ced703afc6e4867bdf8e288855e378a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a88b5e5d77aa463a97ebcc905ae48b1e : SUCCESS in 19m 45s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9b09234977b54420b3f48bbee1f9277c : SUCCESS in 21m 52s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/460a764848b84a229d7419983e2a7d49 : SUCCESS in 27m 28s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/730bfed5a12249cba009a96e54d7bfa6 : SUCCESS in 19m 22s","accounts_in_message":[],"_revision_number":2},{"id":"3baeb3529dda1083067ce02f26650cc3b1f89dfe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-07 11:16:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce301798031a4f8cac024f77ce401c3c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c2b86c5ab014e1e82c8f18100993392 : SUCCESS in 8m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/8e6bd88a6c394a16ac80b2ffa4f8b19e : SUCCESS in 50m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3097491e0a2e4499bd31c8bc9c1e03c8 : SUCCESS in 53m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/97bbf67d5b914bce8000b05ea1169a77 : SUCCESS in 1h 07m 10s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/86e4bff329f54235ab9c6d5f231e53c7 : SUCCESS in 15m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/11883fc8b0674cb4a7aa0d1346fecb04 : SUCCESS in 20m 47s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c18426cbf81f4b45a51274155a80fd6b : SUCCESS in 18m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/06d980029f6d4987903bd647421de6b0 : SUCCESS in 10m 37s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1d0f6a2ebba240c3867e46559963fee7 : SUCCESS in 4m 41s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/0dc7a5f157224a5ca2038c7cc7a69419 : SUCCESS in 16m 26s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ecd9a286088542038f96a3a4ff7db960 : SUCCESS in 17m 24s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/127522a055544bedb478761f4730bf94 : SUCCESS in 19m 03s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/9116ff5f088c447890dc8981fced0b9a : SUCCESS in 17m 49s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/69fe37f9bf1048ec90d3b23d12301bee : SUCCESS in 21m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/018a795295194aa0b414eb10fc4eb036 : SUCCESS in 14m 22s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d5f6a38ae8a84eb6bee4342713bc287d : SUCCESS in 18m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4bb32054949646529533ade1490ed97d : SUCCESS in 18m 42s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d6796cf4bad24ad0b3b39bcc185e9f06 : SUCCESS in 15m 32s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/304354da38fd4b679fc1a9e6d9f86ce5 : SUCCESS in 16m 59s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6df6d3c5905845a7a94f149f83405123 : SUCCESS in 14m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/18cd8fd7b311472bb6f75b0d550b0354 : SUCCESS in 8m 55s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5ffc3e24b7904ac2801b302f8566e57e : FAILURE in 16m 13s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e3a6c77302cd4a229035fa5bba3241d1 : SUCCESS in 6m 17s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/612cba94ada045eaaa141ffd0231b46a : SUCCESS in 54m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8a070969c3f04e5d9f51e318003afd89 : SUCCESS in 38m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7e50110c78e14c78a7a5010aef3773ac : SUCCESS in 58m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ecaac646bba54f539d28c039f1803888 : SUCCESS in 19m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/106df88287124634a7af49dea6c67746 : SUCCESS in 4m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4a54ec347c124c13a2ed77b5f7539584 : SUCCESS in 18m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ec83dfcf51e14c55a2f239df53326295 : SUCCESS in 1h 04m 29s","accounts_in_message":[],"_revision_number":2},{"id":"5bfb0c88e0d8c2ec326c574a95291e0973d15d60","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-18 22:20:26.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"bbdc11c0b7b33455db865ac9b4b97973e68213a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-06-19 10:00:49.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"903439efe8b852a4d783ba6935af21362d883fea","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-06-19 10:01:01.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"8cc12324e2f8fc2eb8488d43a160f3a721a72075","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-06-19 10:06:20.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"11d80f44d263abafa9009c8ab37e38404f1e62ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-19 11:22:39.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbe06134d87447928ce6a72a5a9f2dd3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5413a89ebe484e2284891554824765b8 : SUCCESS in 8m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/a89424b242644e71a6978f199701b1d0 : SUCCESS in 1h 01m 06s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/844991ecb5b74de9ad47b06edacdc9c3 : SUCCESS in 52m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c1ebb6254d854e51a7e26e67f2bb4f8a : SUCCESS in 1h 08m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4440cd5617414fd8a42e1c974be69008 : SUCCESS in 17m 45s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/37b6d9fdbdb54f1cb5652be3cea9bc1e : SUCCESS in 19m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/26b84c93005348ad801e79da49741fd7 : SUCCESS in 19m 26s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/be822519312b4238824b583d4dbd494f : SUCCESS in 10m 37s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/272af616ff73498f94523af6707c74f4 : SUCCESS in 5m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e793f8acf3dc4f5ebd06d8448f4a1481 : SUCCESS in 18m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b88d3cf784794fff9ad25b9570bbdb7e : SUCCESS in 17m 52s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/7c44623f34424b56a113248ce60e0819 : SUCCESS in 18m 49s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/d87039f80270472292be974fc6b52eb3 : SUCCESS in 18m 06s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1df98a1c255042f399696623b10bb63c : SUCCESS in 22m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/38ba9e788fdb455d86534b7af91010b8 : SUCCESS in 16m 59s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e0649f6806934958b24d51717adb8fe4 : SUCCESS in 18m 03s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ae11393e599c4c9e9d78c00c230db1f1 : SUCCESS in 19m 00s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f5408ddf884041318001150fe4e60b53 : SUCCESS in 16m 08s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9d67d4b49e1e4936ac9550fe80c068e5 : SUCCESS in 17m 39s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c1862095ef1b401a82a1f65957b91251 : SUCCESS in 15m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/314ed8273c4f4bf2b7199191942f560f : SUCCESS in 8m 24s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/03964ca82ae64554903658415b592a55 : SUCCESS in 14m 35s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7071a5fac8c7441d9855b0d162127084 : SUCCESS in 9m 05s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3a2bce817b994fbb91246a75bf3f5455 : SUCCESS in 54m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f7e98123149a46f5936a06255a1166a9 : SUCCESS in 52m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2b7b51a580a24f7eba06d01c28c4e05d : SUCCESS in 50m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a4add37c797d48cf80eeed69cd65d90d : SUCCESS in 19m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/010ad2f55d2f4a7d94ccaf9de3946b22 : SUCCESS in 4m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/74197ce2144f49bca11f0ce6fff8c169 : SUCCESS in 20m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a4dcdbdf4e154198919a65f06a9fafd1 : SUCCESS in 1h 07m 45s","accounts_in_message":[],"_revision_number":4},{"id":"c60ab6ffc4dc601c2441d1be30906608703a0eda","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-19 11:29:48.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cdffe69595bf45059ea22e8202ee344c\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d7d364cde46442dabbbc2ce8cea31634 : SUCCESS in 14m 47s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dd82124b4801486994394333523aec7e : SUCCESS in 16m 42s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e33eba0d50f641e7870694edd7099b9c : SUCCESS in 18m 43s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/14a212e68233462187fe13179cc94d5f : SUCCESS in 1h 21m 53s","accounts_in_message":[],"_revision_number":4},{"id":"aa0bd6ae13682e924146dae37f88030d1a5761be","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-06-24 07:15:39.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"73cf0718fdc1e466db620adbc096ef5943ad8fa3","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-06-25 06:10:45.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"09167a692b80985d1bed98ef18eb2e598d3fe3a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 06:11:15.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"51b20d8bf82572c78ca79fb1d702dfb53825261c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 07:21:10.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d9c289f0cbd478e83a7cce48ea178c9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06674581a4b4400da6a27d7078e4827d : SUCCESS in 8m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/f4683bae51dc4dd3a1e7dce125d17cc7 : SUCCESS in 1h 02m 40s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6e6c547e00f746eb9af72921202b9fba : SUCCESS in 51m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/41da3c3ece8745dab4799ff870948495 : SUCCESS in 1h 04m 55s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/7736568bb95e45539355695b3a517938 : SUCCESS in 11m 12s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/323dc901399945b39f5cc0dd1fe2f607 : SUCCESS in 5m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/f875fc6303a74f2a8b998c4b0a516dce : SUCCESS in 20m 36s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/dad7b6a084734b329248ea8084ed1ef0 : SUCCESS in 19m 28s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0dad4184b1e74b3381c2b6db853b0c21 : SUCCESS in 19m 24s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/29a5876efd254851b2aacc078c3e5b51 : SUCCESS in 17m 14s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c82c8ba263654021a8809c46f382f765 : SUCCESS in 23m 44s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b3d5f73c9f264d158960d7cc2c14a0df : SUCCESS in 17m 31s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/abeaacbf290c498aa6fa4d6569d91768 : SUCCESS in 17m 52s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2ced043d3ae443458013d3f0dd115ca8 : SUCCESS in 19m 35s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4c42e034fbd24f4bb97d9f16e7593791 : SUCCESS in 15m 55s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c2e018b78c8144cc87f354b65419aa40 : SUCCESS in 14m 04s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/164940a8c94344db8981a2c9f14353fb : SUCCESS in 16m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/114a3c0ec44544fb892cd73df0d5c71c : SUCCESS in 6m 06s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/4832ce1a0fc047d7a3f3a008d9f5ac68 : SUCCESS in 7m 08s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a515f3eefb5545989a5c4fa23318214e : SUCCESS in 1h 00m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e08c79c252e347deb19ca1ddfe1ffca6 : SUCCESS in 39m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f6d6d04d94be4a738ae78cd171663c28 : SUCCESS in 39m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/25fee8ffc2b84b6fb2bd32cd48fa9296 : SUCCESS in 16m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/560bfeb5d7e140868bfbb40ed061a4af : SUCCESS in 4m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9680743d12b3488986c067d508b9537d : SUCCESS in 18m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6420b1dacc9c43be9f1aef54be26dc18 : SUCCESS in 45m 11s","accounts_in_message":[],"_revision_number":4},{"id":"6b733c19de55f52eaee7d815d5f600d0df1d419c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 07:21:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"c72a2a38188825e454db4ef2cee71021167d1e44","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 07:22:18.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d1edba10b2947138ed413b1a718883b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2a1e0b5247449a9b979a960980a0423 : SUCCESS in 47s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/e54e6641bb244056a80b5e7a4b596d1a : SUCCESS in 31s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/a73a569c08e6494b984549cfc23a7393 : SUCCESS in 29s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"5d640985f7c7104b8ce9992c63c159c58c85ebe8","revisions":{"26b2e5cafbe6295129c30b9d17aecd71f41906b2":{"kind":"REWORK","_number":1,"created":"2024-06-06 13:49:22.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/51/921451/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/51/921451/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/921451/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/921451/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/921451/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/51/921451/1"}}},"commit":{"parents":[{"commit":"1d5b1d746e027bb9509ad0e45cc0f2f50d28511f","subject":"Merge \"Lazy import is not needed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1d5b1d746e027bb9509ad0e45cc0f2f50d28511f"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-06-06 11:55:35.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-06-06 13:48:14.000000000","tz":60},"subject":"s3api: include error response reason in log_info","message":"s3api: include error response reason in log_info\n\nThere\u0027s a variety of reasons why s3api may return a 4xx or 5xx\nresponse. It\u0027s useful to get some more detail in the log message,\nparticularly when the request is a HEAD which won\u0027t get any detail in\nthe response body.\n\nChange-Id: I0a499dc5c50bb534a23adcbfe7c3822d8954b0e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/26b2e5cafbe6295129c30b9d17aecd71f41906b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/26b2e5cafbe6295129c30b9d17aecd71f41906b2"}]},"branch":"refs/heads/master"},"33411464d9a4f77a12156111be11869bcc13c279":{"kind":"REWORK","_number":2,"created":"2024-06-07 09:59:04.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/51/921451/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/51/921451/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/921451/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/921451/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/921451/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/51/921451/2"}}},"commit":{"parents":[{"commit":"1d5b1d746e027bb9509ad0e45cc0f2f50d28511f","subject":"Merge \"Lazy import is not needed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1d5b1d746e027bb9509ad0e45cc0f2f50d28511f"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-06-06 11:55:35.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-06-07 08:55:38.000000000","tz":60},"subject":"s3api: include error response reason in log_info","message":"s3api: include error response reason in log_info\n\nThere\u0027s a variety of reasons why s3api may return a 4xx or 5xx\nresponse. It\u0027s useful to get some more detail in the log message,\nparticularly when the request is a HEAD which won\u0027t get any detail in\nthe response body.\n\nChange-Id: I0a499dc5c50bb534a23adcbfe7c3822d8954b0e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/33411464d9a4f77a12156111be11869bcc13c279"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/33411464d9a4f77a12156111be11869bcc13c279"}]},"branch":"refs/heads/master"},"d957191937c59e9d85350332d5af8acbb2119e11":{"kind":"REWORK","_number":3,"created":"2024-06-19 10:00:49.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/51/921451/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/51/921451/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/921451/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/921451/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/921451/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/51/921451/3"}}},"commit":{"parents":[{"commit":"1d5b1d746e027bb9509ad0e45cc0f2f50d28511f","subject":"Merge \"Lazy import is not needed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1d5b1d746e027bb9509ad0e45cc0f2f50d28511f"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-06-06 11:55:35.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-06-19 10:00:25.000000000","tz":60},"subject":"s3api: include error response reason in log_info","message":"s3api: include error response reason in log_info\n\nThere\u0027s a variety of reasons why s3api may return a 4xx or 5xx\nresponse. It\u0027s useful to get some more detail in the log message,\nparticularly when the request is a HEAD which won\u0027t get any detail in\nthe response body.\n\nChange-Id: I0a499dc5c50bb534a23adcbfe7c3822d8954b0e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d957191937c59e9d85350332d5af8acbb2119e11"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d957191937c59e9d85350332d5af8acbb2119e11"}]},"branch":"refs/heads/master"},"5d640985f7c7104b8ce9992c63c159c58c85ebe8":{"kind":"REWORK","_number":4,"created":"2024-06-19 10:06:20.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/51/921451/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/51/921451/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/921451/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/921451/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/921451/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/51/921451/4"}}},"commit":{"parents":[{"commit":"1d5b1d746e027bb9509ad0e45cc0f2f50d28511f","subject":"Merge \"Lazy import is not needed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1d5b1d746e027bb9509ad0e45cc0f2f50d28511f"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-06-06 11:55:35.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-06-19 10:06:11.000000000","tz":60},"subject":"s3api: include error response reason in log_info","message":"s3api: include error response reason in log_info\n\nThere\u0027s a variety of reasons why s3api may return a 4xx or 5xx\nresponse. It\u0027s useful to get some more detail in the log message,\nparticularly when the request is a HEAD which won\u0027t get any detail in\nthe response body.\n\nChange-Id: I0a499dc5c50bb534a23adcbfe7c3822d8954b0e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5d640985f7c7104b8ce9992c63c159c58c85ebe8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5d640985f7c7104b8ce9992c63c159c58c85ebe8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
