)]}'
{"id":"openstack%2Fswift~921664","triplet_id":"openstack%2Fswift~master~I1d11b2974de8649a111f2462a8fef51039e35d7f","project":"openstack/swift","branch":"master","topic":"account_quota_files","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2024-07-16 03:52:53.000000000","reason":"\u003cGERRIT_ACCOUNT_7233\u003e replied on the change","reason_account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},"37027":{"account":{"_account_id":37027,"name":"Mohammed Al-Jawaheri","email":"mjawaheri02@gmail.com","username":"mjawaheri"},"last_update":"2024-07-16 04:52:06.000000000","reason":"Change was submitted"},"37085":{"account":{"_account_id":37085,"name":"Daanish Khan","display_name":"Daanish Khan","email":"daanishk@andrew.cmu.edu","username":"daanishk"},"last_update":"2024-07-16 04:52:06.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-07-16 04:52:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1d11b2974de8649a111f2462a8fef51039e35d7f","subject":"add object-count quota for accounts in middleware","status":"MERGED","created":"2024-06-10 13:11:25.000000000","updated":"2024-07-16 04:53:24.000000000","submitted":"2024-07-16 04:52:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":23,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"921664-account_quota_files","meta_rev_id":"5531faccffc218e90132815908fe0e14c3b4ebd0","_number":921664,"virtual_id_number":921664,"owner":{"_account_id":37027,"name":"Mohammed Al-Jawaheri","email":"mjawaheri02@gmail.com","username":"mjawaheri"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-16 04:52:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2024-07-16 03:52:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2024-07-16 03:52:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":37040,"name":"Md Azmain Adib Pahlowan","display_name":"Azmain Adib","email":"adib1905@gmail.com","username":"aadib"},{"_account_id":37058,"name":"Mohamed Hassaneen","email":"mohammedashoor89@gmail.com","username":"mohamed"},{"_account_id":37085,"name":"Daanish Khan","display_name":"Daanish Khan","email":"daanishk@andrew.cmu.edu","username":"daanishk"}],"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-10 14:44:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-10 14:49:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-10 21:46:23.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2024-06-13 04:00:39.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2024-06-26 19:28:29.000000000","updated_by":{"_account_id":37040,"name":"Md Azmain Adib Pahlowan","display_name":"Azmain Adib","email":"adib1905@gmail.com","username":"aadib"},"reviewer":{"_account_id":37040,"name":"Md Azmain Adib Pahlowan","display_name":"Azmain Adib","email":"adib1905@gmail.com","username":"aadib"},"state":"CC"},{"updated":"2024-06-29 17:34:41.000000000","updated_by":{"_account_id":37058,"name":"Mohamed Hassaneen","email":"mohammedashoor89@gmail.com","username":"mohamed"},"reviewer":{"_account_id":37058,"name":"Mohamed Hassaneen","email":"mohammedashoor89@gmail.com","username":"mohamed"},"state":"CC"},{"updated":"2024-07-15 17:01:38.000000000","updated_by":{"_account_id":37085,"name":"Daanish Khan","display_name":"Daanish Khan","email":"daanishk@andrew.cmu.edu","username":"daanishk"},"reviewer":{"_account_id":37085,"name":"Daanish Khan","display_name":"Daanish Khan","email":"daanishk@andrew.cmu.edu","username":"daanishk"},"state":"CC"}],"messages":[{"id":"583c79a508e6c17b1a618cac0fa244b7439fc1df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37027,"name":"Mohammed Al-Jawaheri","email":"mjawaheri02@gmail.com","username":"mjawaheri"},"date":"2024-06-10 13:11:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bc5131263a2dd9c0bd86ce95bd84b687e5839ee5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-10 14:44:14.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f141cc1069c64979be54ada034944c5d\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1f03e779c5a6469988da15e69c05fb6d : SUCCESS in 18m 52s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c01e950b9ba14154ae9b5d083f9867ca : SUCCESS in 20m 25s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/695335357640424e98d94820c12fa0b7 : TIMED_OUT in 1h 31m 00s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/9d874acff3f14764a04f58ee1b81b238 : SUCCESS in 19m 56s","accounts_in_message":[],"_revision_number":1},{"id":"cfb67367b91f5c7a9cf42210573301cfab97bddf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-10 14:49:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9a2f37e5a84475e90d29b0799aadab7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9d9773b39f24b8f9e0820166d4e4231 : SUCCESS in 8m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/9edd27f6218e43418e2d3ed2d97e52ca : SUCCESS in 1h 17m 21s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1a427d1b53e141cba9d2907f560549cd : SUCCESS in 41m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bf306d98ad6c4a48aa9bd6669102c482 : SUCCESS in 1h 00m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/da063e141efc49a387f271b4f54a984d : SUCCESS in 18m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0a159422b2f34560ac1b02b084782185 : SUCCESS in 20m 28s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/392441e945c446609e4af8f191e6e32a : SUCCESS in 19m 32s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1038f940ddef47258a49400d148672bf : SUCCESS in 10m 12s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/82a636e5efab453ebef8b16dd0eb3a03 : SUCCESS in 5m 35s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d64c3f284f96420abf911c731b2a572f : SUCCESS in 23m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a1e79822d73142faa45694e3e2a931d2 : SUCCESS in 15m 02s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b506ca9ce0584887be7b6efdae1428fc : SUCCESS in 18m 02s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/dfb1d26d0fc944fc94ab480ad13c87bb : SUCCESS in 22m 16s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/fba11b77269945099f845177d91ea51a : SUCCESS in 22m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5850dff569d341d3a6fd0ea2e74a443b : SUCCESS in 17m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a48a9910e1634e90ba836a027b0be8bf : SUCCESS in 19m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5d1a1cfe69b341d68d531bec3567be70 : SUCCESS in 17m 36s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d8efae6ff0ab4933bd0b2f7cd0aead1e : SUCCESS in 14m 28s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/9b24b116c5b14c4f8c6d80060989a08d : SUCCESS in 17m 06s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/37ea53002f1f46209b7e108fa1a4d9ef : SUCCESS in 17m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bee663d0426d44be92e0557e02e163dc : SUCCESS in 5m 59s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9be0ca4580064cd58f66f6548b22adfc : FAILURE in 18m 20s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5ce5f815e5954425868677c358b8266e : SUCCESS in 7m 27s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6f4756141d404ab490a78d056bdb5bbe : SUCCESS in 1h 00m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2bb753cf75fc4003a0e7bc29bb6ccc9d : SUCCESS in 55m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ae136a7f19c647e287df16bc5caf20fe : SUCCESS in 57m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/10c26be57e1c4d5cb27a316d0d4e909e : SUCCESS in 20m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee3c97d2204c4d36a927bb3b5c3edbd8 : SUCCESS in 4m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9931965417324a9a8534f6da629df078 : SUCCESS in 17m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7cc5b69b79454f07848938a5babca22a : SUCCESS in 45m 44s","accounts_in_message":[],"_revision_number":1},{"id":"a9710f6c16a3023ce088d44832cd6759c174851c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-10 21:46:23.000000000","message":"Patch Set 1: Code-Review+1\n\n(5 comments)","accounts_in_message":[],"_revision_number":1},{"id":"36699b4a963a0a2507961cdb894a3478ca6f2993","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37027,"name":"Mohammed Al-Jawaheri","email":"mjawaheri02@gmail.com","username":"mjawaheri"},"date":"2024-06-11 14:20:34.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"d9b6085a411e5d6ecfa946312f89129b6c7606de","author":{"_account_id":37027,"name":"Mohammed Al-Jawaheri","email":"mjawaheri02@gmail.com","username":"mjawaheri"},"date":"2024-06-11 14:42:21.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"172a1bb99dc9624c3429ac83e6889b7aa574f31e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-11 15:28:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/942d5f531fb946e48c2bdc3ba4db9e81\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0cfeb4ec87074ccd9204a884f273eced : SUCCESS in 17m 42s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c5c5a401848b4497bfd27c7d9f714590 : SUCCESS in 1h 06m 04s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/6e5c67b3bb1645068902d246b219f0e0 : SUCCESS in 20m 24s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1930c0796a1f47d4bd38c90d3ee20638 : SUCCESS in 45m 44s","accounts_in_message":[],"_revision_number":2},{"id":"5da73b3f0a17cfd054f02fbfc3b3752a49d3edd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-11 15:39:21.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/35f3d0f567f44eea9c2074f784d1b9ed\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a59ae9839637445b923d4f68a04548d1 : SUCCESS in 8m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/40418bd6a94e4c7abba2b3dd75574b13 : SUCCESS in 1h 03m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c7204f2aa3754990b6bfba76e20fd128 : SUCCESS in 59m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/17224bb86c774df09f2bc6165b125d59 : SUCCESS in 58m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a0ad227636a3487285ea77dae1ca7b8f : SUCCESS in 20m 33s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bcc4e21d494e4b46af7cedbe099826d6 : SUCCESS in 20m 13s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9e0f6adb21374bd18849ea65eb869d09 : SUCCESS in 19m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d6277fa5df234201a66de1038af9c519 : SUCCESS in 9m 03s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/ee62df1cc8914178918bef53157d3c23 : SUCCESS in 6m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/76c69c0222f148538570a13b20ced8f5 : SUCCESS in 20m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/86cc68743d014d398ff0d4b63c231a53 : SUCCESS in 21m 23s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a553320f083a4d2c83e1d62d5665bc7e : SUCCESS in 17m 56s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/7df74470a13447748157c0b83e851de9 : SUCCESS in 17m 22s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a2035022d98544e7ac994d37b7150c14 : SUCCESS in 21m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ee90208b5ebf4a38a0fb9f01ab43d31e : SUCCESS in 19m 33s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e6e7e0d3e7764b86ad9ee3b76da9c426 : SUCCESS in 19m 45s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/accc0b901c2c4cc59ece8c657a9a3981 : SUCCESS in 17m 58s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0ef53313f7f74d19a449c879334c1128 : SUCCESS in 17m 31s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/53d6e879e5cd4a7cb6b622838bb87068 : SUCCESS in 16m 43s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5ba71207a2be409b9ce47dc9aba5aeed : SUCCESS in 16m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e80d6254f1d6472594e7821ff554c49e : SUCCESS in 6m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/baebdbf688b049038675cfcdbcb768d9 : FAILURE in 15m 56s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2d65de5e6bc64d4a985b862f6c5a4875 : SUCCESS in 8m 22s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8c000f6521374d37961dd63bad781713 : FAILURE in 58m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ed3b761eeb204ebdb405f186ddb2d7cb : SUCCESS in 1h 01m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c40f1cfa1d674875bf5d22b8537804da : SUCCESS in 52m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f8f462e5ba084fde96e915dcc246a981 : SUCCESS in 19m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/642f2ff68cbe4181b8ca49980af23b6a : SUCCESS in 4m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9e2d6c0380324929bfef7cac457cd23b : SUCCESS in 17m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cf5b4a3c648f4ec08e032fe2751332bf : SUCCESS in 1h 11m 54s","accounts_in_message":[],"_revision_number":2},{"id":"d3f689ab76c52af2667d797aad45b4ed9b18acb9","author":{"_account_id":37027,"name":"Mohammed Al-Jawaheri","email":"mjawaheri02@gmail.com","username":"mjawaheri"},"date":"2024-06-12 11:56:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7758123978724c5da9965fd5b47572dedca8b982","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 13:15:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c77bc33a94bc4404bebae0af7216dfb0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a676a19c3d7946ddb62c86e75a79bf9d : SUCCESS in 10m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/859a75f9ed944dd2bcbb1dbe11501630 : SUCCESS in 1h 07m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2e0eac299cd64e74ae96ae9f7d245bfa : SUCCESS in 56m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0a18edf1a88344228c31e1f8603a8084 : SUCCESS in 1h 12m 42s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1055cb0df2534dc49bfc85e6a09ea2a8 : SUCCESS in 21m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c87d704db15941088a87f36a86dad65a : SUCCESS in 20m 13s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/589a1e7e810142c2a193d9b1f8e37468 : SUCCESS in 17m 50s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/636a41d4dec9434eaea9d44a9dd6051a : SUCCESS in 9m 57s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/cd2cbcc3b5f642dc8959d34531a3221a : SUCCESS in 6m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d1624fbe6b9d4cf18387c3fd64bd2c93 : SUCCESS in 20m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ab1782ecab214b44a5c30a474c7258e0 : SUCCESS in 26m 49s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9d6ba027e1904931af91f14407367818 : SUCCESS in 14m 39s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/cbff2dbf06c14a358954df49e8be8d9b : SUCCESS in 17m 20s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4cb3e7e2c0fc4320847693a1fbf04d1b : SUCCESS in 22m 50s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6df37234a2a740fbb3ee04eb5c80ba6d : SUCCESS in 18m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7e13fafc6b87420bb87a46fd66cda154 : SUCCESS in 17m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ab704d35a9be4810bc2be0aa51e2d872 : SUCCESS in 19m 07s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d3dfc3fc6f3e4631a24b5f579706e423 : SUCCESS in 13m 54s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/84e597465d704eeab80920742f9aa662 : SUCCESS in 15m 31s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/6822f900604c4ed3a088e857a30f7c75 : SUCCESS in 15m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ffcdf3c49cb94beba447aeaab867d3ad : SUCCESS in 7m 10s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b959adb65b044b728ac3d09eff23f8b7 : FAILURE in 15m 34s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/eb0ddebe1c854f08807a7cee1d4f45d4 : SUCCESS in 6m 42s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b5db26eeac484c3d937a888f6bb1d2a0 : SUCCESS in 56m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2923258a89a041d7922491da2f29b9ed : SUCCESS in 40m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/97b25168eea54ea6a7a7f07b2a7d439d : SUCCESS in 45m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4d0942e56e0945bbbc5801e87879f10e : SUCCESS in 19m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a25cbb03b90144308d12627234eacc3a : SUCCESS in 4m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b23fe3da3bf84f84922498b34eb8e98b : SUCCESS in 18m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f4e0440895314f08a53a3a049802c0b4 : SUCCESS in 58m 44s","accounts_in_message":[],"_revision_number":2},{"id":"a8953aaa337a447d3239b8d7483892813d377574","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-12 13:29:23.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e69cc5c2bb2b4e2384b0535c2ab63b62\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6d444fe0468141e093224d45c931ef43 : TIMED_OUT in 1h 31m 14s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dc69554349f94425a907e8c6ac0c2380 : SUCCESS in 42m 08s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/62d25e793d4a4e7c81e9358ea15df3f2 : SUCCESS in 25m 54s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/f27b5b01749e4686961c32a58f302566 : SUCCESS in 1h 07m 48s","accounts_in_message":[],"_revision_number":2},{"id":"401d66f07ed6a584d8ad4c00925ede8e3365598c","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-06-13 04:00:39.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"8d9fd7ab7057ffef963b37bf6070a9815ca00e3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37027,"name":"Mohammed Al-Jawaheri","email":"mjawaheri02@gmail.com","username":"mjawaheri"},"date":"2024-06-19 19:36:27.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"990165744d696b31fe07cfe674902eaf0551a392","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-19 20:30:13.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd9ca9048def4d85bc4b3567d3698f50\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1da2ba1c7ed2476aa5fc5e47d5322970 : SUCCESS in 15m 56s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/05084effff454f52bd04e28aacce633e : SUCCESS in 15m 45s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e03e6b302fe54dd4a7dd7f453d6b1dbf : SUCCESS in 22m 53s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a1a9f011b80744b38096a09015c5c5a4 : SUCCESS in 52m 03s","accounts_in_message":[],"_revision_number":3},{"id":"88bc489e5b5853ceb045cd2d5714350b7a22e867","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-19 21:04:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34f66d6653c947b4923654461e6883e1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/460f23fb485f4765a6508be8055154b3 : SUCCESS in 9m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/9753d6da29c04ad19d812912b4fe8453 : SUCCESS in 1h 21m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bfbd3ed6557945408e901c8c1bb16552 : SUCCESS in 42m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6bd4748712ab40ea96e11855ce0ca685 : SUCCESS in 59m 54s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b7bf4b9405304c1198ca59ed34a5a65f : SUCCESS in 16m 01s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bee5cc4a97f140d390f14b59959e3420 : SUCCESS in 17m 52s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0cc7f41cf1c34d95a254c5516a429111 : SUCCESS in 19m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3db87a8e9dff46bcac481e9d70b3c436 : SUCCESS in 9m 22s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/133915f9813e4ea58e71ef48d413da1a : SUCCESS in 4m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/36e7049cde6c42a79a3e50aa79eded20 : SUCCESS in 20m 13s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/606e9e5ce6f54de7ae8d381a6cab469c : SUCCESS in 18m 22s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/cfda204b4c9548dfa98221e094e22781 : SUCCESS in 17m 16s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/e69e4d0cecc54b98a6a9c8fde7409164 : SUCCESS in 15m 24s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/caa7aa746c164c9d9c539e7a218fcf2c : SUCCESS in 24m 27s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a7f2e38d643f4fe4991dea10e292db66 : SUCCESS in 18m 32s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6c91872bc0a54c18b09b1929fefcdd86 : SUCCESS in 18m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/afe03afda99b42dd83639aee51151273 : SUCCESS in 18m 21s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2927be81e3d0463996796f1606e19e97 : SUCCESS in 17m 27s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8af1c80cbb8248fd8380cfc7df462de8 : SUCCESS in 15m 40s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ae33487e8918450884bcd448f835a335 : SUCCESS in 20m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3d933f8a6b3f4e7fa8de5161b2797b95 : SUCCESS in 7m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a7c69cc314f6406e8643969686d8a5b1 : SUCCESS in 15m 12s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9f813253dec04363bdaf16aaedebbe4a : SUCCESS in 9m 40s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a622ffa670d94d838a5b5def30c90578 : SUCCESS in 56m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/36605163d61a4c87b4e8a9db6e82f390 : SUCCESS in 55m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e2d462ce820f4846b23f41fe6b892eaa : SUCCESS in 56m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/86b0dcb383234f27b38e61b95c5be942 : SUCCESS in 18m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ac67837829d4242a430ec66b12bf641 : SUCCESS in 4m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/51b3ad898e9a4ce4b1947484893f2ae7 : SUCCESS in 20m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ca0fda3fdc824462871a1719d4c7c230 : SUCCESS in 42m 06s","accounts_in_message":[],"_revision_number":3},{"id":"9b40ae63c8a38b984b4590a474982871c44dcb73","author":{"_account_id":37040,"name":"Md Azmain Adib Pahlowan","display_name":"Azmain Adib","email":"adib1905@gmail.com","username":"aadib"},"date":"2024-06-26 19:28:29.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6ff982c462d4f992f46277fdd69df2bfda72c286","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-28 16:34:55.000000000","message":"Patch Set 3: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"52fa1a9b74cf87e014d6e3f242740606094e19da","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-07-11 07:27:23.000000000","message":"Patch Set 3: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"d530f1eef7e5ccdc367da7e515bc343f3babfab6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37085,"name":"Daanish Khan","display_name":"Daanish Khan","email":"daanishk@andrew.cmu.edu","username":"daanishk"},"date":"2024-07-15 16:55:16.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"01f6d4a4c64a28cdd8ec7c83306d63d0013c48c4","author":{"_account_id":37085,"name":"Daanish Khan","display_name":"Daanish Khan","email":"daanishk@andrew.cmu.edu","username":"daanishk"},"date":"2024-07-15 17:01:38.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"43d314fd28401e5a341df356d18369c6e5e5b347","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-15 18:01:03.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4396494f87db44129950b40cb3dd5165\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04ed6e9593ca444196c82f487003da81 : SUCCESS in 8m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/f9e859655a7a4b0fa77a430349b67277 : SUCCESS in 49m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e86c990bd8b04465999bdd49af02dcdb : SUCCESS in 46m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dd1fd62ad1964bf99e7f1e34b1e2c7cc : SUCCESS in 1h 00m 13s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dbd205362cfe4c0b855075dd2c087dc3 : SUCCESS in 17m 35s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8527d543c316423480700f314d04eaf5 : SUCCESS in 20m 28s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/04178617801b46c9a88e33cfb59f57da : SUCCESS in 21m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/77164e8990724968a5043fd8484fd0b4 : SUCCESS in 5m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/dd46c93b2a9c496fb95cdc6867df5331 : SUCCESS in 20m 39s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e50c9691670049ad9449af277c7dd818 : SUCCESS in 15m 55s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a52bd4ce266841b79e26fb822c823b66 : SUCCESS in 20m 27s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/c47050fd3c144ccc93c5d988eb13436a : SUCCESS in 21m 25s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/198cbad30d1341a78c65430e69591d73 : SUCCESS in 22m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/a15ea0b6ccf044d99813cebaad93e80b : SUCCESS in 21m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/265099a0b9fd48dcbb323515b262d981 : SUCCESS in 19m 09s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1f5f95b7bb664b1bb8c15e42c1d28d8f : SUCCESS in 20m 37s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ec8548920c994602a8375d6f4fe67128 : SUCCESS in 17m 43s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/37d31ecf63ef41999fff8ced28f535e2 : SUCCESS in 18m 20s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f688218b763049e6bdabc48d772be45a : SUCCESS in 16m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/008019cb033d45cf84b7fb8893596384 : SUCCESS in 6m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/96ea6975d6f144d18b3cc1077e464e0d : SUCCESS in 18m 45s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/353970fc63ed4ff98d918e1166a2d1da : SUCCESS in 7m 36s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3e82d4ca9ea64c708d1b03cba92483c2 : SUCCESS in 59m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/08506e47a8584d06ae4f75630343fc12 : SUCCESS in 48m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/722e608f916e436ba87836d9b1cb77fa : SUCCESS in 51m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/01c6ed6e1f4841d0a13c1289306b4fef : SUCCESS in 17m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/16b2f242d0c647fcb9a35dc1b5a63d73 : SUCCESS in 5m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/81bebe7cc9d94cc39cb164e5acc37cce : SUCCESS in 18m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a1def2e90890429bba473a236e43a129 : SUCCESS in 58m 58s","accounts_in_message":[],"_revision_number":4},{"id":"0c29533a653e6905b29d97d32a4919ec44a50f24","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-15 18:21:01.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93e1a02719a4420286212fa19a5efc8a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/4a19e320b7a648ccb31ed0ca766af1d4 : SUCCESS in 1h 20m 29s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6e0ebe0240884a858d77c4af92ce6597 : SUCCESS in 19m 38s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a0ad7605f7de49309e45f1e21b8db66c : SUCCESS in 1h 04m 27s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/960b9b995cb440fcadb952fd6378c690 : SUCCESS in 23m 21s","accounts_in_message":[],"_revision_number":4},{"id":"fdd48abf2a427c4f8f308c2768d56c741d0388b2","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-07-16 03:52:53.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"edc7b33af35a3870aa2fbb94938343318d42bcd0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 03:53:30.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"cb681f3ac41878b1aac26b300b13ea3e41b351ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 04:52:05.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58f6215f289b4b7d8a0458075be5ac2f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e80941bc70244e65979400a85d6f68df : SUCCESS in 9m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/d8e3e13648ac41398a50267cbb7e8fce : SUCCESS in 56m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c06580c165134ab78f08c27972948970 : SUCCESS in 44m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1b32e68be00b4a6d96240dea7920905e : SUCCESS in 46m 45s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/06960aa83c7f4c15a7c18fe0c9eeecb5 : SUCCESS in 6m 11s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/803f43e36a3a44db8a5a4e2fb32a8892 : SUCCESS in 21m 08s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/382b75bf91b8479cb496f2813e9f4132 : SUCCESS in 21m 19s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/eaccf3f82957482ea03b26000768f397 : SUCCESS in 20m 51s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/5142a8549a6e43248181db6fb110ab4f : SUCCESS in 27m 05s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7e72150949e24572a3b16e494306d918 : SUCCESS in 18m 20s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/aaf1f5e8923247adb0b7f9066a1c1961 : SUCCESS in 18m 50s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/94d1862f8476468192ed08c6f0f8ba16 : SUCCESS in 14m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e23455ecdd9b4a7da79e690a20aab176 : SUCCESS in 21m 11s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/eabd4359609845b7a6b08e102248229f : SUCCESS in 15m 51s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a5c9d004a65d42c9b9c531f31f9d19ff : SUCCESS in 16m 56s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/45ac294798fe4bff9a630f9c97e503e9 : SUCCESS in 18m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9b3175830c74496ba72936e398ccc073 : SUCCESS in 6m 19s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8e9c09313a40473ca5edc419ce722a73 : SUCCESS in 6m 59s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5784e7e3340848f4a57513aa54e6a4ce : SUCCESS in 53m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/77de1e20622547afa6eb47e046da02a7 : SUCCESS in 48m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4a4bd28565d9450e901a86468909509c : SUCCESS in 49m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/abe494a30dd24a41b3b742300978a61b : SUCCESS in 19m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/062be3b1a21749bd8d669c22ab60cf12 : SUCCESS in 4m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2f6b721d791b4f8bb1d20229aa0c17df : SUCCESS in 22m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7fa46e2a541946c9a6c803728ecd5f01 : SUCCESS in 45m 14s","accounts_in_message":[],"_revision_number":4},{"id":"cb4cc282c9c2c67746f97b9d9de9f9257f08bb29","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 04:52:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"5531faccffc218e90132815908fe0e14c3b4ebd0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 04:53:24.000000000","message":"Patch Set 4:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fae0dd5cef1c4ff485963df8e750c6f5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cea9db111b744e97bf4f0c8bf1f5d199 : FAILURE in 59s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/ce844dcd9f124aa4ae9d561b59f17136 : SUCCESS in 35s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"6105b38e65b00c23d373e17ad9f89e4243395165","revisions":{"90c12f8dd0693fab99dbd8a635b0eb2fb3420a45":{"kind":"REWORK","_number":1,"created":"2024-06-10 13:11:25.000000000","uploader":{"_account_id":37027,"name":"Mohammed Al-Jawaheri","email":"mjawaheri02@gmail.com","username":"mjawaheri"},"ref":"refs/changes/64/921664/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/64/921664/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/921664/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/921664/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/921664/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/64/921664/1"}}},"commit":{"parents":[{"commit":"1d5b1d746e027bb9509ad0e45cc0f2f50d28511f","subject":"Merge \"Lazy import is not needed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1d5b1d746e027bb9509ad0e45cc0f2f50d28511f"}]}],"author":{"name":"Mohammed Al-Jawaheri","email":"mjawaheri02@gmail.com","date":"2024-06-09 15:16:05.000000000","tz":180},"committer":{"name":"Mohammed Al-Jawaheri","email":"mjawaheri02@gmail.com","date":"2024-06-10 13:10:53.000000000","tz":180},"subject":"add object-count quota for accounts in middleware","message":"add object-count quota for accounts in middleware\n\nIt is currently possible to configure \"X-Container-Meta-Quota-Bytes\"\nand \"X-Container-Meta-Quota-Count\" on containers, as well as\n\"X-Account-Meta-Quota-Bytes\" on accounts. However, there is no way\nto configure an account quota limit for the number of files or\n\"quota-count\". This limitation could potentially allow a user to\nexhaust filesystem inodes.\n\nThis change introduces the \"X-Account-Meta-Quota-Count\" header,\nallowing users with the ResellerAdmin role to add a quota-count\nlimit on accounts. The middleware will enforce this limit similarly\nto the existing quota-byte limit.\n\nCo-authored-by: Azmain Adib \u003cadib1905@gmail.com\u003e\nCo-authored-by: Daanish Khan \u003cdaanish1337@gmail.com\u003e\nCo-authored-by: Mohamed Hassaneen \u003cmohammedashoor89@gmail.com\u003e\nCo-authored-by: Nada El-Mestkawy \u003cnadamaged05@gmail.com\u003e\nCo-authored-by: Tra Bui \u003ctrabui.0517@gmail.com\u003e\n\nChange-Id: I1d11b2974de8649a111f2462a8fef51039e35d7f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/90c12f8dd0693fab99dbd8a635b0eb2fb3420a45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/90c12f8dd0693fab99dbd8a635b0eb2fb3420a45"}]},"branch":"refs/heads/master"},"96010b6ffc2f7c29c6ed415d9501444adcef359e":{"kind":"REWORK","_number":2,"created":"2024-06-11 14:20:34.000000000","uploader":{"_account_id":37027,"name":"Mohammed Al-Jawaheri","email":"mjawaheri02@gmail.com","username":"mjawaheri"},"ref":"refs/changes/64/921664/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/64/921664/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/921664/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/921664/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/921664/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/64/921664/2"}}},"commit":{"parents":[{"commit":"1d5b1d746e027bb9509ad0e45cc0f2f50d28511f","subject":"Merge \"Lazy import is not needed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1d5b1d746e027bb9509ad0e45cc0f2f50d28511f"}]}],"author":{"name":"Mohammed Al-Jawaheri","email":"mjawaheri02@gmail.com","date":"2024-06-09 15:16:05.000000000","tz":180},"committer":{"name":"Mohammed Al-Jawaheri","email":"mjawaheri02@gmail.com","date":"2024-06-11 14:20:20.000000000","tz":180},"subject":"add object-count quota for accounts in middleware","message":"add object-count quota for accounts in middleware\n\nIt is currently possible to configure \"X-Container-Meta-Quota-Bytes\"\nand \"X-Container-Meta-Quota-Count\" on containers, as well as\n\"X-Account-Meta-Quota-Bytes\" on accounts. However, there is no way\nto configure an account quota limit for the number of files or\n\"quota-count\". This limitation could potentially allow a user to\nexhaust filesystem inodes.\n\nThis change introduces the \"X-Account-Meta-Quota-Count\" header,\nallowing users with the ResellerAdmin role to add a quota-count\nlimit on accounts. The middleware will enforce this limit similarly\nto the existing quota-byte limit.\n\nCo-authored-by: Azmain Adib \u003cadib1905@gmail.com\u003e\nCo-authored-by: Daanish Khan \u003cdaanish1337@gmail.com\u003e\nCo-authored-by: Mohamed Hassaneen \u003cmohammedashoor89@gmail.com\u003e\nCo-authored-by: Nada El-Mestkawy \u003cnadamaged05@gmail.com\u003e\nCo-authored-by: Tra Bui \u003ctrabui.0517@gmail.com\u003e\n\nChange-Id: I1d11b2974de8649a111f2462a8fef51039e35d7f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/96010b6ffc2f7c29c6ed415d9501444adcef359e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/96010b6ffc2f7c29c6ed415d9501444adcef359e"}]},"branch":"refs/heads/master"},"8ea892a41e2bc5615da9ff7df6a2ad85341484b3":{"kind":"REWORK","_number":3,"created":"2024-06-19 19:36:27.000000000","uploader":{"_account_id":37027,"name":"Mohammed Al-Jawaheri","email":"mjawaheri02@gmail.com","username":"mjawaheri"},"ref":"refs/changes/64/921664/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/64/921664/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/921664/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/921664/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/921664/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/64/921664/3"}}},"commit":{"parents":[{"commit":"1d5b1d746e027bb9509ad0e45cc0f2f50d28511f","subject":"Merge \"Lazy import is not needed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1d5b1d746e027bb9509ad0e45cc0f2f50d28511f"}]}],"author":{"name":"Mohammed Al-Jawaheri","email":"mjawaheri02@gmail.com","date":"2024-06-09 15:16:05.000000000","tz":180},"committer":{"name":"Mohammed Al-Jawaheri","email":"mjawaheri02@gmail.com","date":"2024-06-19 19:36:06.000000000","tz":180},"subject":"add object-count quota for accounts in middleware","message":"add object-count quota for accounts in middleware\n\nIt is currently possible to configure \"X-Container-Meta-Quota-Bytes\"\nand \"X-Container-Meta-Quota-Count\" on containers, as well as\n\"X-Account-Meta-Quota-Bytes\" on accounts. However, there is no way\nto configure an account quota limit for the number of files or\n\"quota-count\". This limitation could potentially allow a user to\nexhaust filesystem inodes.\n\nThis change introduces the \"X-Account-Meta-Quota-Count\" header,\nallowing users with the ResellerAdmin role to add a quota-count\nlimit on accounts. The middleware will enforce this limit similarly\nto the existing quota-byte limit.\n\nCo-authored-by: Azmain Adib \u003cadib1905@gmail.com\u003e\nCo-authored-by: Daanish Khan \u003cdaanish1337@gmail.com\u003e\nCo-authored-by: Mohamed Hassaneen \u003cmohammedashoor89@gmail.com\u003e\nCo-authored-by: Nada El-Mestkawy \u003cnadamaged05@gmail.com\u003e\nCo-authored-by: Tra Bui \u003ctrabui.0517@gmail.com\u003e\n\nChange-Id: I1d11b2974de8649a111f2462a8fef51039e35d7f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8ea892a41e2bc5615da9ff7df6a2ad85341484b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8ea892a41e2bc5615da9ff7df6a2ad85341484b3"}]},"branch":"refs/heads/master"},"6105b38e65b00c23d373e17ad9f89e4243395165":{"kind":"REWORK","_number":4,"created":"2024-07-15 16:55:16.000000000","uploader":{"_account_id":37085,"name":"Daanish Khan","display_name":"Daanish Khan","email":"daanishk@andrew.cmu.edu","username":"daanishk"},"ref":"refs/changes/64/921664/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/64/921664/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/921664/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/921664/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/64/921664/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/64/921664/4"}}},"commit":{"parents":[{"commit":"1d5b1d746e027bb9509ad0e45cc0f2f50d28511f","subject":"Merge \"Lazy import is not needed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1d5b1d746e027bb9509ad0e45cc0f2f50d28511f"}]}],"author":{"name":"Mohammed Al-Jawaheri","email":"mjawaheri02@gmail.com","date":"2024-06-09 15:16:05.000000000","tz":180},"committer":{"name":"Daanish Khan","email":"daanishk@andrew.cmu.edu","date":"2024-07-15 16:54:08.000000000","tz":180},"subject":"add object-count quota for accounts in middleware","message":"add object-count quota for accounts in middleware\n\nIt is currently possible to configure \"X-Container-Meta-Quota-Bytes\"\nand \"X-Container-Meta-Quota-Count\" on containers, as well as\n\"X-Account-Meta-Quota-Bytes\" on accounts. However, there is no way\nto configure an account quota limit for the number of files or\n\"quota-count\". This limitation could potentially allow a user to\nexhaust filesystem inodes.\n\nThis change introduces the \"X-Account-Meta-Quota-Count\" header,\nallowing users with the ResellerAdmin role to add a quota-count\nlimit on accounts. The middleware will enforce this limit similarly\nto the existing quota-byte limit.\n\nCo-authored-by: Azmain Adib \u003cadib1905@gmail.com\u003e\nCo-authored-by: Daanish Khan \u003cdaanish1337@gmail.com\u003e\nCo-authored-by: Mohamed Hassaneen \u003cmohammedashoor89@gmail.com\u003e\nCo-authored-by: Nada El-Mestkawy \u003cnadamaged05@gmail.com\u003e\nCo-authored-by: Tra Bui \u003ctrabui.0517@gmail.com\u003e\n\nChange-Id: I1d11b2974de8649a111f2462a8fef51039e35d7f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6105b38e65b00c23d373e17ad9f89e4243395165"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6105b38e65b00c23d373e17ad9f89e4243395165"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
