)]}'
{"id":"openstack%2Fswift~921927","triplet_id":"openstack%2Fswift~master~Ib7f4219ef5fdef65e9cfd5701e28b5288741783e","project":"openstack/swift","branch":"master","topic":"kms-keymaster-barbican-endpoint","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-06-14 17:32:11.000000000","reason":"Change was submitted"},"6968":{"account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"last_update":"2024-06-14 17:32:11.000000000","reason":"Change was submitted"},"6476":{"account":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"last_update":"2024-06-14 17:32:11.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-06-14 16:27:13.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Ib7f4219ef5fdef65e9cfd5701e28b5288741783e","subject":"kms_keymaster: allow specifying barbican_endpoint","status":"MERGED","created":"2024-06-13 09:30:09.000000000","updated":"2024-06-14 17:33:12.000000000","submitted":"2024-06-14 17:32:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"921927-kms-keymaster-barbican-endpoint","meta_rev_id":"5a5bda941486fd6452a914cb26f3ac6c5ae0e8ae","_number":921927,"virtual_id_number":921927,"owner":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":36866,"name":"Valentin Chassignol","display_name":"Vinetos","email":"contact@vinetos.fr","username":"vinetos"},{"value":0,"_account_id":36116,"name":"Philippe SÉRAPHIN","email":"philippe.seraphin@infomaniak.com","username":"pseraf"},{"value":0,"_account_id":31945,"name":"Jacquet Axel","email":"axel.jacquet@infomaniak.com","username":"lhommerares"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-14 17:32:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"recommended":{"_account_id":31945,"name":"Jacquet Axel","email":"axel.jacquet@infomaniak.com","username":"lhommerares"},"all":[{"value":1,"date":"2024-06-14 11:53:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36866,"name":"Valentin Chassignol","display_name":"Vinetos","email":"contact@vinetos.fr","username":"vinetos"},{"value":1,"date":"2024-06-14 11:49:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36116,"name":"Philippe SÉRAPHIN","email":"philippe.seraphin@infomaniak.com","username":"pseraf"},{"value":1,"date":"2024-06-14 11:47:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31945,"name":"Jacquet Axel","email":"axel.jacquet@infomaniak.com","username":"lhommerares"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-06-14 16:27:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":36866,"name":"Valentin Chassignol","display_name":"Vinetos","email":"contact@vinetos.fr","username":"vinetos"},{"value":0,"_account_id":36116,"name":"Philippe SÉRAPHIN","email":"philippe.seraphin@infomaniak.com","username":"pseraf"},{"value":0,"_account_id":31945,"name":"Jacquet Axel","email":"axel.jacquet@infomaniak.com","username":"lhommerares"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-06-14 16:27:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}],"reviewers":{"REVIEWER":[{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31945,"name":"Jacquet Axel","email":"axel.jacquet@infomaniak.com","username":"lhommerares"},{"_account_id":36116,"name":"Philippe SÉRAPHIN","email":"philippe.seraphin@infomaniak.com","username":"pseraf"},{"_account_id":36866,"name":"Valentin Chassignol","display_name":"Vinetos","email":"contact@vinetos.fr","username":"vinetos"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-13 11:39:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-13 21:09:50.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2024-06-14 11:47:24.000000000","updated_by":{"_account_id":31945,"name":"Jacquet Axel","email":"axel.jacquet@infomaniak.com","username":"lhommerares"},"reviewer":{"_account_id":31945,"name":"Jacquet Axel","email":"axel.jacquet@infomaniak.com","username":"lhommerares"},"state":"REVIEWER"},{"updated":"2024-06-14 11:49:17.000000000","updated_by":{"_account_id":36116,"name":"Philippe SÉRAPHIN","email":"philippe.seraphin@infomaniak.com","username":"pseraf"},"reviewer":{"_account_id":36116,"name":"Philippe SÉRAPHIN","email":"philippe.seraphin@infomaniak.com","username":"pseraf"},"state":"REVIEWER"},{"updated":"2024-06-14 11:53:47.000000000","updated_by":{"_account_id":36866,"name":"Valentin Chassignol","display_name":"Vinetos","email":"contact@vinetos.fr","username":"vinetos"},"reviewer":{"_account_id":36866,"name":"Valentin Chassignol","display_name":"Vinetos","email":"contact@vinetos.fr","username":"vinetos"},"state":"REVIEWER"},{"updated":"2024-06-14 12:20:24.000000000","updated_by":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"reviewer":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"state":"REVIEWER"},{"updated":"2024-06-14 12:20:24.000000000","updated_by":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"786bfe3eac5fb5347eb86a7790ed9c0b6bf80971","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2024-06-13 09:30:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"99bc8ad67d4c2dd6f83c8c8e24ca1cdcbde46aee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2024-06-13 10:25:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b3599ece7cd4f53d9ed313a8b04c666423eb1fd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-13 11:39:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00807cf6ec944d8294fb0fada3039df4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a17d53016824e7888ea031234ba376d : SUCCESS in 7m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/4f29d31699374029b54c5cdd213efa92 : SUCCESS in 56m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/22c57beffc70459990d8ed44a3653aaf : SUCCESS in 46m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6e5f4d503e44400f91afa4117b3eb913 : SUCCESS in 1h 07m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0b43697efe0747818743aacead269bfd : SUCCESS in 18m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/42a1494555f048f8a372e989fd4c4bc4 : SUCCESS in 18m 43s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/70e03bbe0c614d00945eabf55f5774ad : SUCCESS in 20m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/acefd5ba794b4ada92ebc58c1b00291c : SUCCESS in 10m 47s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8470f78a7f3b4940af401a265433277b : SUCCESS in 6m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8d376aed0268413a9f0f498d0b642835 : SUCCESS in 19m 43s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/da2c9275bbc24de6ac042fc8bf5df59b : SUCCESS in 19m 05s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6f599306c5704dff836a5932dec5ecc7 : SUCCESS in 18m 33s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/8f3afc459d41499ca9d8fdb873b05f85 : SUCCESS in 19m 51s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/38b64b8c6f334b71a381359326e80c0c : SUCCESS in 21m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d57fb3f6d2c84d6a9873117745272174 : SUCCESS in 19m 57s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/af07d516642d49979efc0bb5cc946e4c : SUCCESS in 18m 11s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/537770d4cc834cb3a4cf18dea1a3b7d1 : SUCCESS in 20m 04s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c42b6d9831eb46fdadcbf5bcb5ad45b5 : SUCCESS in 15m 25s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/68d2793ae30c4fb69af27e80f2ae27cc : SUCCESS in 15m 31s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f7020b712b59486baf5b9a00bc746b51 : SUCCESS in 19m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/26c78229277e4838bdd8593c81364897 : SUCCESS in 9m 45s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/262ca4c2ea1941c5a4795cbf29b0b6b6 : FAILURE in 14m 20s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ce684c67cff64161af42f91967277a79 : SUCCESS in 6m 13s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c5c1ec9759ba4956a000770ba34df804 : SUCCESS in 57m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/91e06bfb073b46aba1759f2cafaa3fb8 : SUCCESS in 47m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7e2238fb2866450fbe28e215e6b9e321 : SUCCESS in 42m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2ca80164d7ac42f7bef5a6d72345db32 : SUCCESS in 18m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c4542e949d3451bb4a408041566a780 : SUCCESS in 4m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/21050c2fe415428b95d4d42ce24013e0 : SUCCESS in 17m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/47fdc31aa6604389829889881e70d289 : SUCCESS in 55m 35s","accounts_in_message":[],"_revision_number":2},{"id":"e8295312bac8dcdbb0c7aef1cab44d5eb60a5d00","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-13 11:59:05.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b0dbeeacbad46ada587dcdfd0221b38\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/196d2d49141f4f5fa3e419d7aa46d628 : SUCCESS in 18m 02s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a859f796d36f4e5a90014a15c6b43995 : SUCCESS in 16m 56s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/beb624e57fcc408da1cd7717cd441e74 : SUCCESS in 21m 51s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/23c2f194aabc4a0bbc3f5fe4ab4c3ce9 : TIMED_OUT in 1h 31m 01s","accounts_in_message":[],"_revision_number":2},{"id":"1a269d29d81fabd01dbe20846066856d35cd0845","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-13 21:09:50.000000000","message":"Patch Set 2: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"aced10811f9a8bb876645d8113416eb109815bca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2024-06-14 10:17:17.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"a53d8f5cac08ef578a06585a3d5e720df643352d","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2024-06-14 10:17:23.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"27be54d2d1caed62457e84a64be80c592d548c74","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-14 11:25:06.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a49487d47b04234b7cd7857520a4cae\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/88a15aecee3e431e87d77e3f38562503 : FAILURE in 21m 23s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0da073f3f7b24602b5d931c86d1cff5f : FAILURE in 22m 53s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b48c3c58640148ed8fd07b0cb9433823 : SUCCESS in 1h 06m 09s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/227c15d8520e4f17a883b05f3ecd7e52 : SUCCESS in 27m 33s","accounts_in_message":[],"_revision_number":3},{"id":"a556b7ecf89e337a4874e755f067065126ad912b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-14 11:28:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a69c36f4f114c969e17e7ac4015b84c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72ded51a46a34ea6b5a40637a3c71952 : SUCCESS in 8m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/b688fc0e5c1042c0965f63749fe15631 : SUCCESS in 1h 05m 16s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f138711ceb0d4e35875a07fc9bab706e : SUCCESS in 41m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2ceb1814bce843f0b25f5c100cabf18d : SUCCESS in 1h 04m 10s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7025c8daf0244ed28bbae7d58a15b9cf : SUCCESS in 22m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2ba6b1c8568e40499339c5e8a2d619ef : SUCCESS in 22m 19s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/930d6e6d0eb34b02a4ffdadea870425a : SUCCESS in 23m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3eccb26c299f45c7bc9612388bee800c : SUCCESS in 10m 45s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/45ef2a93b3b64ef88218ae93c6a9b8de : SUCCESS in 5m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c89f72d41f824e50b38e3dcf693a6642 : SUCCESS in 16m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2bffb5841f58441ba1b91ef67b512886 : SUCCESS in 18m 44s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/337497ed780c4c128b998ebf3a395f8f : SUCCESS in 18m 35s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/2d12d77f253a49309b18d66aa79b7a17 : SUCCESS in 17m 18s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/84a9ad7d72834d8c9be0c3c33697361a : SUCCESS in 22m 36s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1b47c8f4f68045998ec721e014b17de7 : SUCCESS in 20m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/16c0c4d01db44076a00fef61ad7b7011 : SUCCESS in 23m 14s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/57df4e1e19364d4ba22d2eea44a20b21 : SUCCESS in 22m 29s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c7ed900d4c9041f8a28f3b9d052eaa66 : SUCCESS in 15m 28s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8e3918d875524a4d81421dec510c01aa : SUCCESS in 15m 16s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2e4bace5608c41d9a6f496f20b6a074d : SUCCESS in 18m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/08ffa1e33a884071a675bd91961d2303 : SUCCESS in 5m 38s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/aafb58cb556f44b1a6b0e0af2b277ba1 : FAILURE in 13m 12s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a0f8c675a11f469da4259ceb130621ed : SUCCESS in 6m 08s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/df7b9b4be4f4455ea8c27209373d6bd9 : SUCCESS in 53m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2c2bd6b7add8475cb5e3a298f98c1534 : SUCCESS in 56m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9025ee645489428e922d5d74175b5d55 : SUCCESS in 48m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/08e861676cc345d7820addc1c6835657 : SUCCESS in 18m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a70e37b9acd940fd9ed8b23cd8cfbd0d : SUCCESS in 3m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/661b5759d7f9402e8185cc72a417af98 : SUCCESS in 19m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/32cec674e8bd4da5afaf4e33913c638c : SUCCESS in 54m 51s","accounts_in_message":[],"_revision_number":3},{"id":"f617b3cc492f4bb317437675d99d74e1b7f045f2","author":{"_account_id":31945,"name":"Jacquet Axel","email":"axel.jacquet@infomaniak.com","username":"lhommerares"},"date":"2024-06-14 11:47:24.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"0808f5a121e926b7285d268fe030f88077121816","author":{"_account_id":36116,"name":"Philippe SÉRAPHIN","email":"philippe.seraphin@infomaniak.com","username":"pseraf"},"date":"2024-06-14 11:49:17.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"57ebdcacf2b812b479ddc6e8ae5685ad044858ac","author":{"_account_id":36866,"name":"Valentin Chassignol","display_name":"Vinetos","email":"contact@vinetos.fr","username":"vinetos"},"date":"2024-06-14 11:53:47.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"67c92171470e758f20ba0e8d72dbc5ec059d9508","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-14 16:27:13.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b04e11ab2fbd9fb1c9966cde5e31d89ab2bb4537","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-14 16:27:45.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a83765ef524f369f687d83ea33491c787a0936a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-14 17:32:11.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44a8c695d49c493ba0dd827ba14b124e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa2554ac0aab402a87a50a2a1b8c92d7 : SUCCESS in 6m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/afb23636125c40b3863eae0cc01b381f : SUCCESS in 46m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e27aa19143394ffa8445d18b690e9017 : SUCCESS in 40m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f11b01dee0094f62b85fbfade3c045cb : SUCCESS in 56m 32s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/c311ed3761944bbcabf26f11d1e56ee5 : SUCCESS in 10m 36s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/b1fe6bcc24904cbbb4e8ee3ef4ce131c : SUCCESS in 6m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c7da42948f744458bce095bd4f365dba : SUCCESS in 19m 40s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/da790fa6a03b47d4b8edf2ec21e0e008 : SUCCESS in 22m 27s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/0d0f7b3f10af47f295a7678ec94bb46e : SUCCESS in 18m 12s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/8dcb7bcab0234a7f957545dfc6a2232f : SUCCESS in 17m 02s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6a15f8ed438044c5bc8ab4ae01942659 : SUCCESS in 23m 57s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/6c4a0e029f1541d88bf14fcbf9eb55dd : SUCCESS in 17m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4bdc9724ace24f43a976e25a1d3bd07b : SUCCESS in 15m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f42f3a773eac4eb2a4146be6bee5d60f : SUCCESS in 19m 15s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/dc56763c7b724eeaa7b011c6dbfff900 : SUCCESS in 15m 48s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/063dd103859341339c53e6dd5e2079b4 : SUCCESS in 16m 05s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e2d627fc68154d43b63c1b1124772059 : SUCCESS in 15m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/134f53ea8fd44172b08289ede308dc03 : SUCCESS in 6m 12s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/da53e738ba314c4c94c11b5136d58389 : SUCCESS in 6m 04s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d5fb74d3530748a58276dbb80fb75e1c : SUCCESS in 57m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5216924cc3d34fd5a168565dba4ec65c : SUCCESS in 47m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b7682ffabc114ba7af3c6805ef979c34 : SUCCESS in 49m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/49922a82f33b4876b11635c2baac05f3 : SUCCESS in 18m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b8d79b8ec52483191654f1f2e33ce74 : SUCCESS in 3m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b66e61d77e1a4fc1a1b6040316a5fe9b : SUCCESS in 16m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ba3bc64b16114774a94ecda1ebd0daf0 : SUCCESS in 58m 01s","accounts_in_message":[],"_revision_number":3},{"id":"42e1dafcf6377c283ddf8595bb4168dec3149a12","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-14 17:32:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"5a5bda941486fd6452a914cb26f3ac6c5ae0e8ae","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-14 17:33:12.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a33d95328d5440aa6aa9d061ff6d64e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/349b241082fa46abb67007c73a13560b : SUCCESS in 45s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/d7adb2b348024cc59d60ddbe3f88ece9 : SUCCESS in 34s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/4b52f4e9c4d54b15ae92862c6b76e91f : SUCCESS in 33s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"90da23c7d232ba443b0814fd71c39f4585d30e4c","revisions":{"2983f0205ddc52643bdb81204f17aa0db9237fec":{"kind":"REWORK","_number":1,"created":"2024-06-13 09:30:09.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/27/921927/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/27/921927/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/921927/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/921927/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/921927/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/27/921927/1"}}},"commit":{"parents":[{"commit":"1d5b1d746e027bb9509ad0e45cc0f2f50d28511f","subject":"Merge \"Lazy import is not needed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1d5b1d746e027bb9509ad0e45cc0f2f50d28511f"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2024-06-13 09:27:55.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2024-06-13 09:27:55.000000000","tz":120},"subject":"kms_keymaster: allow specifying barbican_endpoint","message":"kms_keymaster: allow specifying barbican_endpoint\n\nUnder a multi-region deployment with a single Keystone server,\nspecifying the Keystone auth credentials isn\u0027t enough. Indeed,\nCastellan succeeds when logging-in, but may use the wrong\nBarbican endpoint (if there are 2 Barbican deployed). This is\nwhat happened to us, when deploying our 2nd region.\n\nThey way to fix it would be to tell Castellan what region to use,\nunfortunately, there\u0027s no such option in Castellan. Though we may\nspecify the barbican_endpoint, which is what this patch allows.\n\nChange-Id: Ib7f4219ef5fdef65e9cfd5701e28b5288741783e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2983f0205ddc52643bdb81204f17aa0db9237fec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2983f0205ddc52643bdb81204f17aa0db9237fec"}]},"branch":"refs/heads/master"},"548676da1629031f6f498a83219e79d8994603b3":{"kind":"REWORK","_number":2,"created":"2024-06-13 10:25:57.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/27/921927/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/27/921927/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/921927/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/921927/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/921927/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/27/921927/2"}}},"commit":{"parents":[{"commit":"1d5b1d746e027bb9509ad0e45cc0f2f50d28511f","subject":"Merge \"Lazy import is not needed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1d5b1d746e027bb9509ad0e45cc0f2f50d28511f"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2024-06-13 09:27:55.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2024-06-13 10:25:45.000000000","tz":120},"subject":"kms_keymaster: allow specifying barbican_endpoint","message":"kms_keymaster: allow specifying barbican_endpoint\n\nUnder a multi-region deployment with a single Keystone server,\nspecifying the Keystone auth credentials isn\u0027t enough. Indeed,\nCastellan succeeds when logging-in, but may use the wrong\nBarbican endpoint (if there are 2 Barbican deployed). This is\nwhat happened to us, when deploying our 2nd region.\n\nThey way to fix it would be to tell Castellan what region to use,\nunfortunately, there\u0027s no such option in Castellan. Though we may\nspecify the barbican_endpoint, which is what this patch allows.\n\nChange-Id: Ib7f4219ef5fdef65e9cfd5701e28b5288741783e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/548676da1629031f6f498a83219e79d8994603b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/548676da1629031f6f498a83219e79d8994603b3"}]},"branch":"refs/heads/master"},"90da23c7d232ba443b0814fd71c39f4585d30e4c":{"kind":"REWORK","_number":3,"created":"2024-06-14 10:17:17.000000000","uploader":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"ref":"refs/changes/27/921927/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/27/921927/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/921927/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/921927/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/27/921927/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/27/921927/3"}}},"commit":{"parents":[{"commit":"1d5b1d746e027bb9509ad0e45cc0f2f50d28511f","subject":"Merge \"Lazy import is not needed\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1d5b1d746e027bb9509ad0e45cc0f2f50d28511f"}]}],"author":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2024-06-13 09:27:55.000000000","tz":120},"committer":{"name":"Thomas Goirand","email":"zigo@debian.org","date":"2024-06-14 10:17:07.000000000","tz":120},"subject":"kms_keymaster: allow specifying barbican_endpoint","message":"kms_keymaster: allow specifying barbican_endpoint\n\nUnder a multi-region deployment with a single Keystone server,\nspecifying the Keystone auth credentials isn\u0027t enough. Indeed,\nCastellan succeeds when logging-in, but may use the wrong\nBarbican endpoint (if there are 2 Barbican deployed). This is\nwhat happened to us, when deploying our 2nd region.\n\nThey way to fix it would be to tell Castellan what region to use,\nunfortunately, there\u0027s no such option in Castellan. Though we may\nspecify the barbican_endpoint, which is what this patch allows.\n\nChange-Id: Ib7f4219ef5fdef65e9cfd5701e28b5288741783e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/90da23c7d232ba443b0814fd71c39f4585d30e4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/90da23c7d232ba443b0814fd71c39f4585d30e4c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36866,"name":"Valentin Chassignol","display_name":"Vinetos","email":"contact@vinetos.fr","username":"vinetos"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
