)]}'
{"id":"openstack%2Fswift~922157","triplet_id":"openstack%2Fswift~master~I12e84745e7601d162755bc9d0f1cda7b63e92197","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I12e84745e7601d162755bc9d0f1cda7b63e92197","subject":"CI: make sure old swift is truly gone for rolling-upgrade jobs","status":"MERGED","created":"2024-06-17 21:48:48.000000000","updated":"2024-06-18 21:48:06.000000000","submitted":"2024-06-18 21:47:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"922157","meta_rev_id":"706122abf9a613a2595df95909df0f1f16eb9612","_number":922157,"virtual_id_number":922157,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-18 21:47:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-06-18 20:37:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-06-18 20:37:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-17 22:24:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-17 23:18:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ac7c766fdfcc26c1a62ec039af1a71ff1984599a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-17 21:48:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"354e464f480c21b304cbfb22623f4678b46686d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-17 22:02:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a6a2ec6efd14a29a0077d8e9f48943d8056ab5a5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-17 22:24:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/343926c43492409b87ab12754494dd4a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/a354a666689e4519ab1c4d11f81d891b : SUCCESS in 20m 54s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ac4c943b4f204f569a7983bf1e3df667 : SUCCESS in 16m 40s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4584ff1b6a0f4f80a77907d9a26faf2d : SUCCESS in 19m 18s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/518d19a8e84347f1a1174910bce6b3ce : SUCCESS in 21m 00s","accounts_in_message":[],"_revision_number":2},{"id":"897a6f9ecc98e2ae75b8894748c63814f9f0f7b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-17 23:18:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3773a5d141c4f0c94cdf7e2f6f237b0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/153b84654853459cbd18e2613d3e39d1 : SUCCESS in 9m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/5c0ac07dbe9a4f238948835a3cd3c439 : SUCCESS in 56m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8564de856fa8436890c7daee94b90870 : SUCCESS in 49m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d62837a5c89a431eba51d4671cd5d20b : SUCCESS in 1h 08m 14s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/041cc74dc6d94a29ac4c3701c15e3257 : SUCCESS in 19m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e83d2eec9302419583136d7af7d4328a : SUCCESS in 18m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a31ebf5d50f4446996ac5f7a07df25f1 : SUCCESS in 22m 04s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/15618f4b5f7b4ecc838d4b3bca7838f8 : SUCCESS in 10m 30s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9ce409f59d3b4bb8a3430e74390c00af : SUCCESS in 6m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e097ce84fe0a490aa0bf0c75b4cde5f1 : SUCCESS in 19m 32s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/62aa2fa0d15545e6bc223036bb44b960 : SUCCESS in 17m 04s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/294383c3eaa9456fa5abe00124641289 : SUCCESS in 17m 03s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/d4bf89c5d4ea43aa9d8a49b4ae80abaa : SUCCESS in 16m 46s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0571f34407d047fd8327f12c19ec710a : SUCCESS in 22m 24s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4e10f6d795f14f69a1ccd12127cab7bc : SUCCESS in 17m 37s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0eba56e2d7a5433686d5ac9ca5ef488d : SUCCESS in 19m 09s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a5797f7fdf62422aaafa101fec4925e2 : SUCCESS in 18m 33s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4af39cadf74047dca5eee767663e0fb0 : SUCCESS in 16m 23s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e8f5caa6356f4361b95959cc57cf8d50 : SUCCESS in 16m 06s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ae71710346584ec8aa1a475b519ae186 : SUCCESS in 15m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/db27c852d73742fca890ca83dcca0a64 : SUCCESS in 7m 18s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/175f63cf931743d99c0c3fc0e51816b6 : FAILURE in 14m 26s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/faa86e1ae17943268418e16d027315bc : SUCCESS in 6m 12s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c2a5e8c4c7d6455d999787fb4adf2ae7 : SUCCESS in 56m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/dbd4191dffc644d3a08585ac7a4a2185 : SUCCESS in 44m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8e09dedc054d4be489dcfa8e79bffdf7 : SUCCESS in 52m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e2d263426e78481cafee0897b7f45899 : SUCCESS in 17m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/983780f7ccb948bcbf1e981ca3cae213 : SUCCESS in 5m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8e4f393e9d3343d1a0d22f02be38234a : SUCCESS in 18m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eac679c03c47490ab2798b2ab7c13be4 : SUCCESS in 1h 10m 51s","accounts_in_message":[],"_revision_number":2},{"id":"a4ec396b6464f3076e10521784325fc135fdc99b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-18 20:37:03.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"deaf4813cd305be9c8d17217de38def3f08acb81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-18 20:37:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9ff0b6eea44d4ef925d79004d784e96787a7c4ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-18 21:47:01.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fd1851d52cc4ebcac48c88785b68710\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bf6ac007ff84fcd93408f69a1e8edf7 : SUCCESS in 8m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/0e90b7eeea82489eb2d731e8963ceeda : SUCCESS in 57m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0ab33230e51540d2b8691d71d56df94a : SUCCESS in 39m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/17350f023743432d9bff7acf3b84d68e : SUCCESS in 1h 09m 19s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/d4c2041270be45a5bbfa80122bab6a55 : SUCCESS in 10m 11s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/e3e7a485be0e4f959240951e760882ca : SUCCESS in 5m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/27fac14a8e6c4517b0e0edf82ede8b6f : SUCCESS in 19m 29s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/64cb75bb509c4f47bc4d3d7df6f86a88 : SUCCESS in 19m 09s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/136a6abc2845428981bfd063471afd4c : SUCCESS in 18m 16s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/c0f5da362ffe4ccfb97cf1fc3e95a8a0 : SUCCESS in 19m 06s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/eda8646bd3664bca8bbbe82e6b49b2e9 : SUCCESS in 23m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1ea7c36c6ed14583acbd1b6c12b997f6 : SUCCESS in 17m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/947d8712e4fc4303b60b87234c03e166 : SUCCESS in 20m 09s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e3652683251e475492903fbba78566dd : SUCCESS in 19m 04s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/826f14e49b5e4948853131d1236a8ade : SUCCESS in 15m 39s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b919dc9131374271b61c4c163e931b2d : SUCCESS in 15m 33s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/51705dad11194145add40460929ac377 : SUCCESS in 16m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f64493a804924074a5740d4114d11d51 : SUCCESS in 6m 41s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1f7ea39375744fb6988f86dee8dacb2a : SUCCESS in 6m 26s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/efdbbd85ce694274ad11cc3ad1f98ecd : SUCCESS in 1h 00m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d9d7db7a71774999badc26dcfcccc588 : SUCCESS in 51m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7046ea5725014075946be0021c948d8d : SUCCESS in 38m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8cb8e4620a6147348bbd631060daf68c : SUCCESS in 19m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad8c2b9a552b4863978a5b058715d2c0 : SUCCESS in 4m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1208dcda9dc046fe9b7c6e064435010d : SUCCESS in 17m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5a6e693440ff414fbef23dd66ebcdc8b : SUCCESS in 41m 04s","accounts_in_message":[],"_revision_number":2},{"id":"71ac2d41f2c6c0b64332d2d5c40bfa34785f4081","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-18 21:47:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"706122abf9a613a2595df95909df0f1f16eb9612","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-18 21:48:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ea96e975b934f2194fd7a676f3c1e21\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c44290c66db4f59bc7d3d52fbe2288e : SUCCESS in 45s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/d2418e81c914454d91e6c7ed2d06abea : SUCCESS in 30s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/071983275031482b92daa30fa842974c : SUCCESS in 30s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a290f445efe812ad4886ca23cca00c8bb3b48c2b","revisions":{"8174820da4eac1e2c923b2dc36143aeb2f9d1419":{"kind":"REWORK","_number":1,"created":"2024-06-17 21:48:48.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/57/922157/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/922157/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/922157/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/922157/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/922157/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/922157/1"}}},"commit":{"parents":[{"commit":"46e183df15d0a98d2595e583d2a50c98aba153e2","subject":"Merge \"add bytes of expiring objects to queue entry\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/46e183df15d0a98d2595e583d2a50c98aba153e2"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-17 21:43:10.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-17 21:48:41.000000000","tz":-420},"subject":"CI: make sure old swift is truly gone for rolling-upgrade jobs","message":"CI: make sure old swift is truly gone for rolling-upgrade jobs\n\nWhile switching how some executable scripts were configured, I saw\nsome strange rolling-upgrade failures that seemed to indicate that\nthe new invocation method was trying to be used with old code. It\nseems like it maybe has something to do with whether swift was\ninstalled to /usr/local/lib/python3.9/site-packages/ or\n/usr/local/lib64/python3.9/site-packages/ but I\u0027m not entirely\nsure.\n\nAt any rate, a proper package manager ought to uninstall the old\nversion then install the new one, so it seems reasonable to do that\nwith pip, too.\n\nChange-Id: I12e84745e7601d162755bc9d0f1cda7b63e92197\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8174820da4eac1e2c923b2dc36143aeb2f9d1419"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8174820da4eac1e2c923b2dc36143aeb2f9d1419"}]},"branch":"refs/heads/master"},"a290f445efe812ad4886ca23cca00c8bb3b48c2b":{"kind":"REWORK","_number":2,"created":"2024-06-17 22:02:05.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/57/922157/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/922157/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/922157/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/922157/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/922157/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/922157/2"}}},"commit":{"parents":[{"commit":"46e183df15d0a98d2595e583d2a50c98aba153e2","subject":"Merge \"add bytes of expiring objects to queue entry\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/46e183df15d0a98d2595e583d2a50c98aba153e2"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-17 21:43:10.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-17 22:01:35.000000000","tz":-420},"subject":"CI: make sure old swift is truly gone for rolling-upgrade jobs","message":"CI: make sure old swift is truly gone for rolling-upgrade jobs\n\nWhile switching how some executable scripts were configured, I saw\nsome strange rolling-upgrade failures that seemed to indicate that\nthe new invocation method was trying to be used with old code. It\nseems like it maybe has something to do with whether swift was\ninstalled to /usr/local/lib/python3.9/site-packages/ or\n/usr/local/lib64/python3.9/site-packages/ but I\u0027m not entirely\nsure.\n\nAt any rate, a proper package manager ought to uninstall the old\nversion then install the new one, so it seems reasonable to do that\nwith pip, too.\n\nChange-Id: I12e84745e7601d162755bc9d0f1cda7b63e92197\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a290f445efe812ad4886ca23cca00c8bb3b48c2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a290f445efe812ad4886ca23cca00c8bb3b48c2b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
