)]}'
{"id":"openstack%2Fswift~922177","triplet_id":"openstack%2Fswift~master~Iba43804fe6b5c57bd009a2d503cff300b72d8e34","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-06-18 15:45:59.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Iba43804fe6b5c57bd009a2d503cff300b72d8e34","subject":"DNM: drop boto test requirement; skip unported tests","status":"ABANDONED","created":"2024-06-18 04:06:39.000000000","updated":"2024-06-18 15:45:59.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6a24fa2ecc018bf45d67864563d6d1a9bc39552c","_number":922177,"virtual_id_number":922177,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-06-18 05:23:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-18 04:26:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-18 05:23:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8a74ff540975b07670e06b82c9fcc40b5ee8025f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-18 04:06:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1ae8b57c462c9aa41c6be669745ae40ac24a11c9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-18 04:26:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0b9b44dbf6542a6890185bed7927bd7\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/146874c1d86f40e09690092f498d7c5a : SUCCESS in 16m 11s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/01c9c7e142844c16acb796e1fb482f6b : SUCCESS in 15m 07s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/aa26a93f5fc241f5ba3610811b9ecc5d : SUCCESS in 19m 08s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/2454be41a86d4fe6b94a44ed54f80c9d : SUCCESS in 14m 11s","accounts_in_message":[],"_revision_number":1},{"id":"728b4700021a32c5e0ce9beb18113fade296bcd6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-18 05:23:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/37bbf0d2f2be4fc69985998d5776d1f4\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8c299c885c9a4577b20b9fe46ad93fbe : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3eb43cf36d0c4b6c94e19c8bfdf9cb5b : SUCCESS in 8m 37s\n- requirements-check https://zuul.opendev.org/t/openstack/build/063ff2f01bc743d7ac020175234ea2c1 : SUCCESS in 2m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/50111c8b5f364d55af96f16ce2179a37 : SUCCESS in 5m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/0687edf1f4aa4a929c9e354788ea459f : SUCCESS in 1h 11m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/598d633d6d4a4a3184963799e6779545 : SUCCESS in 51m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/55fa8dd495db45618e95682227bc9a21 : SUCCESS in 49m 11s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/29a7d18658d049af9f3efb56235b776a : SUCCESS in 17m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/127c3f7ccf6048129943b80d42e9da85 : SUCCESS in 13m 09s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/de52aa4a438949be93f6e042750e2ac0 : SUCCESS in 17m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/55197a2af57948f79e50d796312c71cc : SUCCESS in 10m 59s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a3ec928540cc4e4d93f985b02a562926 : SUCCESS in 5m 16s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/11b14d6dc121452582d98e9515de7f65 : SUCCESS in 19m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/111f31631b3441d4bf18862b9fb43100 : SUCCESS in 14m 27s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6daffd70becf487fbb50f5e647e2c738 : SUCCESS in 17m 02s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/c068bbca68fc419b96cf5a64fdcac923 : SUCCESS in 17m 09s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2ad2608e8ca948828f95ac4b30ad37df : SUCCESS in 30m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/694f6e3ba67a453a8585817c11083b79 : SUCCESS in 14m 30s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a7a86ac3246e43f0a95861265115a385 : SUCCESS in 15m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c4ae4b52c86f487cb6c7fe0e2c22682c : SUCCESS in 15m 20s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9071a64a4273436baf3d9085b25592b5 : SUCCESS in 11m 50s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/21f2a6bbf4df49ecb298d18ce2b9384b : SUCCESS in 12m 03s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3faac8b02a974b1a8c24e9dd12b1b7f6 : SUCCESS in 13m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/be80cee0b09641d4ade4106e7e3ee83f : SUCCESS in 6m 09s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/09b6d04b61324457b0aaa905bb676180 : FAILURE in 14m 29s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/bdd8fad85ae14939a34a7c02481aafb8 : SUCCESS in 6m 58s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8c499abbabad40c0b934598fbdaf6084 : FAILURE in 1h 02m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8cbaafaf1ea74458aff836ec569d1bd0 : SUCCESS in 36m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/16edf0fd361f4792b840c945b767b81c : SUCCESS in 42m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/36794cc5edfc435396abbb71ac46647e : SUCCESS in 20m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f5b40d8bb5d430bb6fff2f02bbe5734 : SUCCESS in 3m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ce4cd830d9a6446f944f3bfaae3a2018 : SUCCESS in 18m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7b3ade0737d74179b4091fa13a11b173 : SUCCESS in 46m 48s","accounts_in_message":[],"_revision_number":1},{"id":"6a24fa2ecc018bf45d67864563d6d1a9bc39552c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-18 15:45:59.000000000","message":"Abandoned\n\nMainly done to demonstrate the parent patch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7bdf514e010e14b0de9f6054a9d2ba12dd8f5718","revisions":{"7bdf514e010e14b0de9f6054a9d2ba12dd8f5718":{"kind":"REWORK","_number":1,"created":"2024-06-18 04:06:39.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/77/922177/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/77/922177/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/922177/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/922177/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/922177/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/77/922177/1"}}},"commit":{"parents":[{"commit":"dd8b7656daad1c244c94aa6d2bc8b4eb59c8cb3b","subject":"Skip boto 2.x tests if boto is not installed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dd8b7656daad1c244c94aa6d2bc8b4eb59c8cb3b"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-18 04:06:23.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-18 04:06:23.000000000","tz":-420},"subject":"DNM: drop boto test requirement; skip unported tests","message":"DNM: drop boto test requirement; skip unported tests\n\nChange-Id: Iba43804fe6b5c57bd009a2d503cff300b72d8e34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7bdf514e010e14b0de9f6054a9d2ba12dd8f5718"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7bdf514e010e14b0de9f6054a9d2ba12dd8f5718"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
