)]}'
{"id":"openstack%2Fswift~922200","triplet_id":"openstack%2Fswift~master~Ice32d7ac2aa47f0da1926b1ff1bf57cbe4a03faf","project":"openstack/swift","branch":"master","topic":"timstatsdpatch","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-06-18 22:26:21.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ice32d7ac2aa47f0da1926b1ff1bf57cbe4a03faf","subject":"statsd client: warn when legacy options are used","status":"ABANDONED","created":"2024-06-18 10:49:53.000000000","updated":"2024-06-18 22:26:21.000000000","total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"9dca4ca6a4f815d49c92a850c926ecf1d6b79bcd","_number":922200,"virtual_id_number":922200,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-06-18 12:24:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-18 11:42:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-18 12:24:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-18 20:27:08.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"}],"messages":[{"id":"c1752565920ee4895736f2234b671aea95813011","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-06-18 10:49:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"28b2d03908a1de3068bfc6114653dbedca3af765","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-18 11:42:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f003d4ef27044a72b0ccae18796e8727\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3b73f82eb866423aa743c2f537963358 : SUCCESS in 40m 42s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0867c2bf0a1e4932ba5c692320999d95 : SUCCESS in 17m 18s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f84e1bc8853d4cd6aee53269ee70117b : SUCCESS in 51m 16s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/c25c1da597de4d64889206705eb7cc45 : SUCCESS in 19m 03s","accounts_in_message":[],"_revision_number":1},{"id":"fe29e203b12ed9faa1617c262531a732156d8deb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-18 12:24:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f043c21dff5492eb8090fba5d7718dc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff83cb74d21e42b2abab0ce3b87618f4 : SUCCESS in 8m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/9cd70bc88fb0417786c9c9772f148a9a : SUCCESS in 59m 34s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/53fea77bf8d04a4cbbcee559cd2e766d : SUCCESS in 57m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f152d27f6907446393a47186cb0456f4 : SUCCESS in 1h 09m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/debd531a598945c39a5d32d8b33fd01c : SUCCESS in 17m 33s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d8bb7af97f2e46cb8a0b5820ad15b4a1 : SUCCESS in 18m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6650dceb02bf4359bbadb3a5838f28ac : SUCCESS in 19m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/511f9d6b2d7c48fdb9091869f357caa8 : SUCCESS in 11m 28s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d5725596e666451f83f491dffd54f9be : SUCCESS in 5m 14s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/76ad8d8bdeeb4288845b24482ee6d562 : SUCCESS in 19m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b3f12fe6ae3e47628c6a03416be03621 : SUCCESS in 19m 52s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/76528820057046daa191cbf5b3d09de9 : SUCCESS in 17m 38s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/ca1701f357d04164bdf4d317a7adc475 : SUCCESS in 17m 27s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f08a55096aad4d168e6f7da8bcd13e17 : SUCCESS in 23m 25s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/78a7d7b0eaba43dbb863b9d198df940f : SUCCESS in 17m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5498f7eb018a42fb90c77305e090f549 : SUCCESS in 19m 23s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3293aed1f5ad45748319d92ef1a2bcda : SUCCESS in 18m 04s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/990bc0fb64e5485a970b5257015737cc : SUCCESS in 18m 08s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/a896c1801f33429bae72a1935019ee32 : SUCCESS in 18m 53s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/c48a815bfaa44fef85a827d3d03e6cdf : SUCCESS in 16m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5897210e954c4d83871aadef095ad9b5 : SUCCESS in 7m 40s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b0d8b1e8af57480c96e74bf74d6a6b11 : FAILURE in 16m 54s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f0868c4341aa49dbbd5a766ca0bafadc : SUCCESS in 7m 28s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/410fd211b757410dab038a26ec4d9714 : SUCCESS in 58m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c304b064cc1447cbbae2cbfb1fcabae9 : SUCCESS in 58m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0890b0bce0884efe8366de573f98627a : SUCCESS in 38m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3c854037255e4d8189f36cd05e9be827 : SUCCESS in 18m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de0a99f7dd27401f8ff39cb32f1aa2cf : SUCCESS in 3m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/66aa7fe524394fbb9268be50fcedf16f : SUCCESS in 17m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/20d62aed01ac49168cd71657a632f8d0 : SUCCESS in 57m 54s","accounts_in_message":[],"_revision_number":1},{"id":"e1d027c116b42f18f3d95823cab92de3e05b8e5b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-18 20:27:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9dca4ca6a4f815d49c92a850c926ecf1d6b79bcd","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-18 22:26:21.000000000","message":"Abandoned\n\nSquashed into https://review.opendev.org/c/openstack/swift/+/919444","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"16646ce1da58ec8f991d6a7d906f6800a499feb2","revisions":{"16646ce1da58ec8f991d6a7d906f6800a499feb2":{"kind":"REWORK","_number":1,"created":"2024-06-18 10:49:53.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/00/922200/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/00/922200/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/922200/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/922200/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/00/922200/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/00/922200/1"}}},"commit":{"parents":[{"commit":"4d94b1ad04937a9a00943bb6485285da46ded443","subject":"sq? simplify legacy option parsing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4d94b1ad04937a9a00943bb6485285da46ded443"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-06-18 09:50:59.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-06-18 09:50:59.000000000","tz":60},"subject":"statsd client: warn when legacy options are used","message":"statsd client: warn when legacy options are used\n\nChange-Id: Ice32d7ac2aa47f0da1926b1ff1bf57cbe4a03faf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/16646ce1da58ec8f991d6a7d906f6800a499feb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/16646ce1da58ec8f991d6a7d906f6800a499feb2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
