)]}'
{"id":"openstack%2Fswift~922311","triplet_id":"openstack%2Fswift~feature%2Fmpu~Ic865f2fc7b32638dc55a045f91c84b0ce930f548","project":"openstack/swift","branch":"feature/mpu","topic":"merge-master","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-06-24 23:15:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic865f2fc7b32638dc55a045f91c84b0ce930f548","subject":"Merge remote-tracking branch origin/master into feature/mpu","status":"MERGED","created":"2024-06-19 12:16:49.000000000","updated":"2024-06-24 23:18:48.000000000","submitted":"2024-06-24 23:15:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"922311-merge-master","meta_rev_id":"d9a130bcf3046859478c3f6b0bc46c9c70ae7629","_number":922311,"virtual_id_number":922311,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-24 23:15:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2024-06-24 08:43:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-06-21 19:50:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2024-06-24 08:43:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-06-24 22:13:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-19 12:39:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-19 13:49:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-21 19:50:05.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"4eb32772770c77b49d71325ed30225db402a9961","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-06-19 12:16:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"93a6ae78a10cf6c63741322200cba9d63044ee9b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-19 12:39:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0518fd8423d4057a288412bdaffd411\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8155ec8746654ee38c6ca0e7266f0a21 : SUCCESS in 18m 00s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a16f1f8ed8c940d5b2e5d6edd56274dd : SUCCESS in 17m 58s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/7c36426017804db388d1984b17da5fe7 : SUCCESS in 21m 03s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a2bf356214574609b172c59ca4ad235c : SUCCESS in 19m 51s","accounts_in_message":[],"_revision_number":1},{"id":"702ba9737a643ef23ae65fc290e2cbb4e99668fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-19 13:49:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0dd652b0f4d94a37801491c9b483f41b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/198114f300114695bd9358c144497063 : SUCCESS in 7m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/93b1957232364d21a2ed25bc6346acc3 : SUCCESS in 20m 49s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c2779668c0ee4b468f9dc01205a45a60 : SUCCESS in 20m 26s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/851cee6a85004220bc2030d37b4b9687 : SUCCESS in 15m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9c99c93a6a6941a383f06a4349656606 : SUCCESS in 11m 50s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/bd457cbe02884094b45c05159e124680 : SUCCESS in 6m 32s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/10e82425520b4b6eb4ccbeb3c0ca2b90 : SUCCESS in 21m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7d1d0e0bc010415bbb4c75605c3c9737 : SUCCESS in 17m 37s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/4aec2d94acb74842abfb4e1f0f93eed6 : SUCCESS in 17m 14s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/98a6ef5d5a694116ad1224eb5dd2e589 : SUCCESS in 19m 03s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b156dd8f384b4f6e9aead3c5e75bdb18 : SUCCESS in 22m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0b636153f5184edebe727bd397633b14 : SUCCESS in 17m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7dfb289163bd46d3aa36d379fe40d758 : SUCCESS in 14m 59s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8a0dfac5047a4f0abbda5cc11541e770 : SUCCESS in 18m 13s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/14378ceda4934e62bc6720c742038db7 : SUCCESS in 16m 11s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e17434efe6d64a94aef2bb229b456e99 : SUCCESS in 16m 16s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ef609b594a4d47a48ed06e6c32720f55 : SUCCESS in 16m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/170e13a8b86248029168ab4b4ef6dec1 : SUCCESS in 6m 31s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/32761e97709240589c5f53f3e6548d07 : SUCCESS in 17m 15s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ac325fddd689461fb8844201b3470037 : SUCCESS in 6m 09s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8cf1245aa1f44b559ddbff05fca65fc5 : SUCCESS in 58m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/72feb4348f4245f6b416586c36eb1235 : SUCCESS in 19m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/164898a5a12b4e739a11dc81e52417bf : SUCCESS in 3m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3b6248d807e042839cb540c43cb440e3 : SUCCESS in 18m 40s","accounts_in_message":[],"_revision_number":1},{"id":"d3a8c5e312ebe955d7c936ce25e4353723d2c680","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-21 19:50:05.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8389c4ea08b19ea83a365e792728271a16d14b20","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-06-24 08:43:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"224fa648c778d260ce4c077da008b1182d2fa348","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-24 16:18:29.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5b060184221b1e51e7f9a9c6489dd33ea2f89907","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-24 22:12:58.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"a9c3db84fa1f66436a9bc12ca915aebaa4f5214e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-24 22:13:15.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ca19007e92a5cc2a178beab85823029dae2bc6c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-24 22:13:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"75c7619c643d8ab008ebe4ff979ac57df929da3b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-24 23:15:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa66af92625944c6a7685a6aa510d96a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdad977672aa4cc5bfa18cf0e1a9fb17 : SUCCESS in 8m 33s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/9bce9497f3e24c54aa272c6b597f6266 : SUCCESS in 9m 49s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/a35e95edc87549998a5925283e38c1c1 : SUCCESS in 4m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6b89688235384ac49aa707b93014c65e : SUCCESS in 19m 54s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9bf954c96cfc40f6a8510cc79f33040e : SUCCESS in 19m 34s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/955da1dd44584236bab1a4caeeab8733 : SUCCESS in 19m 09s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/d2b3491eb4594448be301c4c990d6881 : SUCCESS in 18m 45s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b9e3bc4795634016bb4a48575e38ff75 : SUCCESS in 29m 58s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3146170bf0c6432e83018f2fee07322d : SUCCESS in 16m 24s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/c3d1533d8f714b6dbf6a1cff55d4496a : SUCCESS in 17m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b6f3ce7a50834d86a9d989a26b46ead9 : SUCCESS in 19m 18s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/9b9358fbd47d453592b06ae5b5a77546 : SUCCESS in 17m 05s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/14acfa2dfbc144a1b175f9bab287a16c : SUCCESS in 15m 51s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d2927ab177ea461ea51376bc1d099c57 : SUCCESS in 18m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/48b208bb0eb44806aa521ccc7a78f8fe : SUCCESS in 5m 23s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/59254c588bd74dfba1eed90329e50007 : SUCCESS in 6m 49s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/72b7bfca9064400aa49c8a9d18a0a932 : SUCCESS in 1h 01m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1a2c5e3d00ac41a0a13f578312c373b5 : SUCCESS in 17m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/668c0fde99794e4a8f54167d7b3c8f15 : SUCCESS in 3m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dd46a2014e2a46c49f9817c254c0952b : SUCCESS in 18m 21s","accounts_in_message":[],"_revision_number":1},{"id":"355a1c3a3dd946e47ddb528e8ef0db769ddaad8a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-24 23:15:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d9a130bcf3046859478c3f6b0bc46c9c70ae7629","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-24 23:18:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6291867a8f304eb1a42dc31902cbdff4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2bc6fed63e24b179b48535fed110b22 : SUCCESS in 44s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/2e73e468eba2453989e98f2145b55c79 : SUCCESS in 29s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/d45c588813774f3885f9ba7943353de5 : SUCCESS in 30s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"51fd293f1f947f6972e4fa6751f94f750bbd86f2","revisions":{"51fd293f1f947f6972e4fa6751f94f750bbd86f2":{"kind":"REWORK","_number":1,"created":"2024-06-19 12:16:49.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/11/922311/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/922311/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/922311/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/922311/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/922311/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/922311/1"}}},"commit":{"parents":[{"commit":"c77fbc6546a696065757b24bbb92690ad732a89e","subject":"Merge remote-tracking branch \u0027gerrit/master\u0027 into feature/mpu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c77fbc6546a696065757b24bbb92690ad732a89e"}]},{"commit":"b33f8ff2c56767c49c669c1368be7d3f839d7a97","subject":"Merge \"CI: make sure old swift is truly gone for rolling-upgrade jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b33f8ff2c56767c49c669c1368be7d3f839d7a97"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-06-19 12:12:11.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-06-19 12:12:11.000000000","tz":60},"subject":"Merge remote-tracking branch origin/master into feature/mpu","message":"Merge remote-tracking branch origin/master into feature/mpu\n\nResolve conflicts in test/unit/common/middleware/s3api/test_obj.py\n\nChange-Id: Ic865f2fc7b32638dc55a045f91c84b0ce930f548\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/51fd293f1f947f6972e4fa6751f94f750bbd86f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/51fd293f1f947f6972e4fa6751f94f750bbd86f2"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
