)]}'
{"id":"openstack%2Fswift~922332","triplet_id":"openstack%2Fswift~master~Id88f128ecbb890272d02e8af0216508842de571b","project":"openstack/swift","branch":"master","topic":"task-container-per-day","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-07-29 19:01:50.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Id88f128ecbb890272d02e8af0216508842de571b","subject":"sq? pass the expirer_config to helpers","status":"ABANDONED","created":"2024-06-19 15:29:38.000000000","updated":"2024-07-29 19:01:50.000000000","total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"36ee98d90dd2cd1ae5fc71f08d1a30308c031462","_number":922332,"virtual_id_number":922332,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-06-19 17:53:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-19 17:43:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-19 17:53:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-24 15:10:59.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"}],"messages":[{"id":"a914c3f743d089d94f36c762433d169c251430fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-06-19 15:29:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3e105abbd555ece8f45f663a2dbc97bef25e0cbf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-06-19 16:32:12.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"b93a0efc8a35d4ea12768c51e7c80deab18655c6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-19 17:43:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bdde77aab87a47efbfd046434b9a6983\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9d03db9590084244b54233a1609b9904 : SUCCESS in 16m 56s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9a066376b4d84bed9ebccdb36577efa7 : SUCCESS in 1h 09m 31s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/2a0b3f88a34e4c90aa3e9a94e5be6c0a : SUCCESS in 20m 24s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4e8c0943ae1e4a0b8df67e74e52d0703 : SUCCESS in 49m 12s","accounts_in_message":[],"_revision_number":2},{"id":"03677569036c94945e470f17572fdf75276dff66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-19 17:53:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab06abddb67c49c588fcf5f463f76d5b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/720ac4d7ca1c4501b147298adfd33378 : SUCCESS in 7m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/3f4abc2704bb43449dfd93b192422db9 : SUCCESS in 1h 05m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d47a03a0edbf4adba74591a81c5fbf1b : SUCCESS in 47m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a1812ef2624f46ad880c0cde6c27af43 : SUCCESS in 55m 24s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/090f420dbbac4f08a13dd7b47145de86 : SUCCESS in 18m 53s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8fdb8c982fa44734891af328c76e95b0 : SUCCESS in 19m 02s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bef3a15061384cb5975b27e59fd5c7a8 : SUCCESS in 19m 56s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c0b648bd7f8d4daeada30836c87af3b6 : SUCCESS in 9m 38s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/7daffe969f724cb4bebb6280e626a80b : SUCCESS in 5m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/12e49efe04bc421f99d5aa4bd55728fd : SUCCESS in 19m 41s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4adb13c4454640d48f68dba10706dd25 : SUCCESS in 16m 06s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/433d960d03c1427cade2ceadb14200fc : SUCCESS in 17m 18s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/64f5f7a98a704fbda25a21a89a4fa903 : SUCCESS in 17m 27s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a1b0cdd605204ec8925a226ee8f52d4a : SUCCESS in 18m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/1ec83ce2eebf49e39763728f2e083008 : SUCCESS in 17m 52s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/25acc90712924bf2ad73241495826421 : SUCCESS in 18m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f1689e1d2cec44fc8acdae5f7064aa52 : SUCCESS in 19m 25s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/bef62c08cec84cbdbcadcb36028a43f0 : SUCCESS in 16m 01s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b8aab870ea404b96ad77eeeecf7e82a9 : SUCCESS in 15m 13s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/38f53e2c5a0a4545b2c24d22bbe88766 : SUCCESS in 17m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/49ff4f99ccce4db3bbaed45566edb18e : SUCCESS in 7m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/965999dd7a3f499585b8c827d841b6a6 : SUCCESS in 15m 53s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/29e4f4c7e0bd41638f20388a38c55000 : SUCCESS in 7m 01s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/00fe45181b7e4f8fbbb464930087bb46 : SUCCESS in 58m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ceaaa4bcbff24f0897bcd0f50cba981a : SUCCESS in 47m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f351c0eb2f8045b6a45421d65cc47c9b : SUCCESS in 47m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0378cff930cd43f2a3a8dce5ff2446e2 : SUCCESS in 19m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb75d5c21eac42c881b00c7d3aa5d37a : SUCCESS in 3m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4a6a31683a8b434a94b2c28baf503ada : SUCCESS in 17m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3e78f3d997614d53ba4e8926067a65bc : SUCCESS in 1h 09m 01s","accounts_in_message":[],"_revision_number":2},{"id":"ad410b06fe52e7df0db576f42b45bc1183e711ff","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-06-24 15:10:59.000000000","message":"Patch Set 2:\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"36ee98d90dd2cd1ae5fc71f08d1a30308c031462","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-07-29 19:01:50.000000000","message":"Abandoned\n\nalternative squashed into 920452: Configurable expiring_objects_task_container_per_day | https://review.opendev.org/c/openstack/swift/+/920452","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bd94e09537f79555280de827cde97ea14aebca49","revisions":{"fe60b0eacb3d8e9f2ce584f7d0a03d26301080ac":{"kind":"REWORK","_number":1,"created":"2024-06-19 15:29:38.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/32/922332/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/32/922332/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/922332/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/922332/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/922332/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/32/922332/1"}}},"commit":{"parents":[{"commit":"c503d937bdf0898e739f4febf2b5f416d3af87e9","subject":"Configurable object_expirer_task_container_per_day","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c503d937bdf0898e739f4febf2b5f416d3af87e9"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-06-19 15:27:29.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-06-19 15:27:29.000000000","tz":60},"subject":"sq? pass the expirer_config to helpers","message":"sq? pass the expirer_config to helpers\n\nChange-Id: Id88f128ecbb890272d02e8af0216508842de571b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe60b0eacb3d8e9f2ce584f7d0a03d26301080ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fe60b0eacb3d8e9f2ce584f7d0a03d26301080ac"}]},"branch":"refs/heads/master"},"bd94e09537f79555280de827cde97ea14aebca49":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-06-19 16:32:12.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/32/922332/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/32/922332/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/922332/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/922332/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/32/922332/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/32/922332/2"}}},"commit":{"parents":[{"commit":"43cc17025cd094dd9a0c866887e7b35de08a3bc4","subject":"Configurable object_expirer_task_container_per_day","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/43cc17025cd094dd9a0c866887e7b35de08a3bc4"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-06-19 15:27:29.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-06-19 16:31:02.000000000","tz":60},"subject":"sq? pass the expirer_config to helpers","message":"sq? pass the expirer_config to helpers\n\nChange-Id: Id88f128ecbb890272d02e8af0216508842de571b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bd94e09537f79555280de827cde97ea14aebca49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bd94e09537f79555280de827cde97ea14aebca49"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
