)]}'
{"id":"openstack%2Fswift~922518","triplet_id":"openstack%2Fswift~master~I49d40d10c75443def6476ed8c8d4cc4473291484","project":"openstack/swift","branch":"master","topic":"log_prefix","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-11-22 01:23:30.000000000","reason":"Change was marked work in progress"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-11-22 01:23:30.000000000","reason":"Change was marked work in progress"},"35790":{"account":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"last_update":"2024-11-22 01:23:30.000000000","reason":"Change was marked work in progress"},"36606":{"account":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"last_update":"2024-10-18 15:35:59.000000000","reason":"\u003cGERRIT_ACCOUNT_36606\u003e replied on the change","reason_account":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-11-22 01:23:30.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"I49d40d10c75443def6476ed8c8d4cc4473291484","subject":"statsd: deprecate log_ prefix for options","status":"NEW","created":"2024-06-21 20:29:27.000000000","updated":"2024-11-22 04:36:36.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":89,"unresolved_comment_count":11,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"d0ee9a0e3914ad70235866fc77dc073ba6fe0454","_number":922518,"virtual_id_number":922518,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-11-22 02:37:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2024-06-21 21:12:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-21 21:56:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-24 16:09:59.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2024-06-26 19:49:08.000000000","updated_by":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"reviewer":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"state":"CC"},{"updated":"2024-07-25 16:45:14.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2024-07-30 21:22:10.000000000","updated_by":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"reviewer":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"state":"CC"},{"updated":"2024-07-30 21:22:21.000000000","updated_by":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"reviewer":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"state":"REVIEWER"},{"updated":"2024-08-14 14:14:57.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2024-08-29 14:08:07.000000000","updated_by":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"reviewer":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"state":"REVIEWER"}],"messages":[{"id":"183a7f33e2110bd13011cb55b796abb3a0033981","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-21 20:29:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1338da44c5c8eca795956371e53b1f6fa9bfbaf8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-21 21:12:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/668f819868154407b1a313ef815a60ff\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d2649ca44779415d9e052774526a0d81 : SUCCESS in 17m 04s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/436b3943066d47069ea644a3f5f9561f : SUCCESS in 17m 12s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b242f795289e424eae79b7732e32a74d : SUCCESS in 41m 07s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7e53035ecbab43d4bd2da487ab6cfd81 : SUCCESS in 20m 13s","accounts_in_message":[],"_revision_number":1},{"id":"c60619139c555a127bd4502b40a499852c3ed634","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-21 21:56:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2c208bc3d0f4b429c38bd8fb6add06d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc44260b6487469c8c6c61daf6e3fb05 : SUCCESS in 8m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/b53327e0da1646a185c3e908fb015a52 : SUCCESS in 1h 06m 00s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/809503b52f7841af9f1ebb4eb12a61fb : SUCCESS in 44m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0ec617295c9a4d9fb73a74524cf35ceb : SUCCESS in 1h 19m 08s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/98089719553a47dca2de9c1d0169e743 : SUCCESS in 23m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fdebb9ee07b24e76abc5fe868272eb3f : SUCCESS in 15m 33s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e547e46d6b3749d59368012678540687 : SUCCESS in 20m 26s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3a517cf9d3864d1ab5b488efed67bcbb : SUCCESS in 10m 04s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1d6a98a51868475eade7bef94919034a : SUCCESS in 5m 45s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/28139ea5f5174ed5ade85bf2a99d893e : SUCCESS in 19m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1a23f57a416443ee9338a46cd12d13b1 : SUCCESS in 17m 54s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/89f50676ce8c42b49cdbcba37b30ec1f : SUCCESS in 19m 51s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/a31df86c5b514959aa138e7f69fdbc2b : SUCCESS in 17m 06s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/990b7bf39bdb4e8e822730b371fc56fd : SUCCESS in 18m 22s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/fe4f81496681487fab8ca607e494ae42 : SUCCESS in 16m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/99e513c026dc4510941e57fac7accb72 : SUCCESS in 18m 11s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ac358f620ed1487c907d0013773470c3 : SUCCESS in 19m 21s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f180080219f341d396f15313c041edf8 : SUCCESS in 14m 15s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/aad1b9840109490bbdb3dbef69399f5b : SUCCESS in 16m 19s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d986a77ec466476ca4abf3aef03b3adc : SUCCESS in 15m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b1a0a23d0a6c4290b9a77dadf012c96a : SUCCESS in 6m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5402abd62c644ad594c823d0cdf3259d : SUCCESS in 15m 59s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/0ceac0edaf1a41ef9df98009c7b6ea93 : SUCCESS in 6m 02s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0e0eeb3c71994dcbbf9bfa6418bd06c7 : SUCCESS in 56m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c40455ea827847608be5669b156bc515 : SUCCESS in 51m 01s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7adcde5956b74344a7701e76937c1e2a : SUCCESS in 50m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e676296709354193bd9409360024ed81 : SUCCESS in 19m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f257cf482da6481d90fb5046c6524227 : SUCCESS in 5m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ce50a5a1ed3349dbae78a044a9158af0 : SUCCESS in 15m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/039eb83587aa42ddbae75855d43f92f7 : SUCCESS in 52m 39s","accounts_in_message":[],"_revision_number":1},{"id":"9e3c1b90f36957fca67a0defa1fd0a8a6ecf9588","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-06-24 16:09:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6360aa502d69aff3b0f76615a8d2e54f9804f4ad","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-24 21:11:48.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"91413e4f1fffcfe1ef313434d040f731d47ce353","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-06-26 19:38:53.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"9621aa22f7067f8dbcde6d331fa5aabe2e53495f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-06-26 19:45:24.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7723da7d858e3c2f216e05df9c10d088aab952ae","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-06-26 19:49:08.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3064623cd9b9126be7ba13ae9ab6b928835b104b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 20:08:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01ba7bd90b85482cbebed52a42900fe5\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6a481ea15d254c9fbbf4c7769991293e : SUCCESS in 17m 12s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/399528e13cf74b419f806f3d784bb410 : SUCCESS in 16m 27s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/dd8e0e51bbc8480fb6e44c4cf3f746ff : SUCCESS in 20m 30s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/70425151cb59436a8748afde8e393a81 : SUCCESS in 22m 09s","accounts_in_message":[],"_revision_number":3},{"id":"e01ac529095b96579907e9ed90d525ae767a9d65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 20:54:49.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/449de837b29c4ca28718c408b0f6d905\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87c0fd3fb8a24072bfa238894fd54b7e : SUCCESS in 8m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/36801468f04447a59e65fac4061c8de7 : SUCCESS in 1h 08m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a2400953f7c44e9a8401e60ef07ae5ab : SUCCESS in 59m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cfd21917b7cf42558447732cae57751a : SUCCESS in 56m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6f87835af7e44e2a8949ebdf781b6d87 : SUCCESS in 21m 28s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/145a3f1669ac4d8694e37f9d2b7fe1e1 : SUCCESS in 21m 46s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e1c534f63ceb40dcb0cb3b4fb44b4671 : SUCCESS in 21m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4852e6daa8c84057b12be7501387342f : SUCCESS in 11m 16s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/9598f445bba94fc88ba0729453ea870a : SUCCESS in 6m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/c771013a15334adcb608e4b27430b423 : SUCCESS in 21m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a577b0edf34648a798fca8e41a3919b7 : SUCCESS in 22m 38s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/9d46187de3c44fa49330bd0ca6d2a63e : SUCCESS in 18m 14s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/3605f743f22c433893e527571273b10a : SUCCESS in 16m 56s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/cfe9fc48f6514330be138c7d1a79770c : SUCCESS in 22m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/10aec82e8d3d473bb8abd91f3095fb5f : SUCCESS in 17m 17s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/813570d88c584b7ea35697f821d26d79 : SUCCESS in 18m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3eefc5b494174fc09e3af4863c1112b8 : SUCCESS in 19m 05s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6759dc05d39943e083f99f1f1c8568f4 : SUCCESS in 17m 50s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/586947b847f84a49ad9dfdfe62bfc25c : SUCCESS in 17m 48s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/96e5afb0755d422ca59e7f820e21b6ff : SUCCESS in 17m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3315c71ea4994b07a3450eff2ac91047 : SUCCESS in 5m 48s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ef916d996a714329b4637435ae745ac1 : FAILURE in 5m 03s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1079ccf3d082458093ba781b38709b8c : SUCCESS in 5m 40s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/38c7f906918c4620b3c57d5aae089322 : SUCCESS in 55m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/90cb683e39c141d7b8da9b0daea5d716 : SUCCESS in 46m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1ee87fa6d1d74e53b585117a8df00b22 : SUCCESS in 52m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4bb9c3842e574a9ea00e7dbae992f105 : SUCCESS in 18m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/997fe89e32234de0a8928c0298be6140 : SUCCESS in 4m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7f786e4aea814334b96aecc718e61d3d : SUCCESS in 16m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/46684e90d2c541d98cdcccb7f6f6aaf8 : SUCCESS in 53m 25s","accounts_in_message":[],"_revision_number":3},{"id":"bc01f7c2a5858c1c29ac3e48a5c9ceb651080e05","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-06-27 09:16:23.000000000","message":"Patch Set 3:\n\n(8 comments)","accounts_in_message":[],"_revision_number":3},{"id":"f9263d6b3ab3a60f0208c8c2e654c7e3db70162f","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-06-27 15:56:18.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d782a5156812d1f85ec6308056e437b28f26aca1","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-06-27 16:01:03.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"8a8a7ae6e7d31d4d27db3b946054d602cc591559","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-07-09 16:31:10.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"58441ed8c22fd80fb0ca46c5c686c22bd7118c27","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-07-09 18:29:07.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"503e4fd18e1a8cc0f3a27046b134ba661585b27f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 18:54:16.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c73cf7d7dc2442a8d397281c359cf34\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cdbb020bd36840b7b161fe80ba9b64ee : SUCCESS in 24m 07s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4434f1dc6da045509029d75a3fe0ac3b : SUCCESS in 18m 20s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/658b6d6820624d22a6b452b8aebe69b2 : SUCCESS in 22m 09s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/23211e9eedae47caa3da0c51b90c7a20 : SUCCESS in 22m 28s","accounts_in_message":[],"_revision_number":4},{"id":"ceb531b64e6a0fdfb637441a9c4e53ed0e94262a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 19:29:16.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e16732ea44cf4834b9b2bbf3fd60bae5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c0317dc633a444bbb8a0015a2568ecb : SUCCESS in 8m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/66b0fa0e4e9a4a0392af6d6a0e6bdf9d : SUCCESS in 45m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1a54bff3806c491a8b95bce806616b09 : SUCCESS in 52m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f4dd5700150c441db76ff5d9319e2b7d : SUCCESS in 46m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f0b91ca75b0c4cd7937495ec8b62c5eb : SUCCESS in 20m 59s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8dec5351834e438782d8322024cc3ea8 : SUCCESS in 21m 39s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/561a6d99483047e5996fa51a044a68b4 : SUCCESS in 23m 28s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0132bb6427bd4088aea056999eb3c460 : SUCCESS in 5m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1d7b6da5234f4ab39b84c5ced7607121 : SUCCESS in 20m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0a1412416530413ba967c6561f81499b : SUCCESS in 17m 55s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8aa8fe0ebc354fb6b3094c9e8de96f0c : SUCCESS in 20m 07s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/34e0d2418bba477e93bf31085c78b748 : SUCCESS in 19m 47s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3ba1b2eda4fe4df6892b77f6da4c0e4a : SUCCESS in 27m 11s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8d1f9650b2d6419788ecd2e09202d14b : SUCCESS in 17m 23s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/155d126de72a420cb8cdcf93c63506c0 : SUCCESS in 19m 45s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/05cbfe1a322e476b9e0a92d3c0ff9cbf : SUCCESS in 17m 52s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8e1489df5f15415d8849e36f043f9fc1 : SUCCESS in 15m 02s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ff246b8582f84d2aae9c89236ba96155 : SUCCESS in 16m 02s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f984e606f569442a8bf1212006214f39 : SUCCESS in 16m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e5ddba69a8a848549e60e878efa0dfab : SUCCESS in 6m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fd782af87a8d4c69b207b3504e820756 : SUCCESS in 14m 20s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e5e7e8e4918647818a3f4b40844a0032 : SUCCESS in 6m 28s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2b735b63029a4f7ab94c4b6868fd8637 : SUCCESS in 50m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7c3068207a0e49cda048502dcbf6a17d : SUCCESS in 53m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c7d20a37a7624c8982bdd2211982f9cb : SUCCESS in 39m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f7c65ad56b4a436db644a6942accfae8 : SUCCESS in 21m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44b01ce8504642d2bca6156207a74068 : SUCCESS in 4m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dd4d9d81e33e41c6b363c150581bc1db : SUCCESS in 17m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9cd1598c2db1420181cd07b3dc717fbd : SUCCESS in 41m 25s","accounts_in_message":[],"_revision_number":4},{"id":"319d9b4c959dabec3e13a249f5ae3ef76f6601e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-07-09 23:29:58.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"634d5a8db0d57f2c67b7e0a1de4f037a30a36670","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 23:34:16.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb3f4059fbee4acf9e3d9fd7362c2894\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fcf59f09d86d46f0b69f258378c5a637 : FAILURE in 3m 03s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/217a02e02b0a4d8d988d785d5f2a8341 : FAILURE in 3m 12s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4bfadd8c393840cd8fa273a5d2134bff : FAILURE in 3m 04s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b77349a8929a426aa3245c3a2b396377 : FAILURE in 2m 32s","accounts_in_message":[],"_revision_number":5},{"id":"decd3c74bd44b635432c1024129f66e32d0e4fa9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-10 00:35:06.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd76ac9b15334308bb0b312271b3964b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a5797d7e7204ac99deb404210083092 : SUCCESS in 8m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/32f4d2d0a6de48f7b97173b9aebe9695 : SUCCESS in 1h 01m 50s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/39cdfda92e564be8871fc9fe5628370b : SUCCESS in 55m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9dca0da21a554fcfb3ed5a8f8f2074ba : SUCCESS in 59m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7ca234dd5bd34efabedef5a7dbabef95 : SUCCESS in 20m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/41e4e45404fa417fad48ff8b2a3a7871 : SUCCESS in 19m 07s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f1421edceacd48b195f59f121187d462 : SUCCESS in 21m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3b41b258cdb647a3acee35aabcb4cf86 : SUCCESS in 5m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6106a3d5159e439db87ed27169da1f5c : SUCCESS in 19m 35s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9bd38750499241f09908d71b6f2c736b : SUCCESS in 18m 33s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/23362289c06f49de8ea903920aa4547e : SUCCESS in 20m 02s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/95bcf82548a44abe8719a03ce2152c15 : SUCCESS in 15m 33s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2622b32b27f542a2bf14ff9d8d709db6 : SUCCESS in 28m 36s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/5c39f6ecee0e486ba968b710ea782579 : SUCCESS in 19m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/20422190601c4b29b0715d47356d49cc : SUCCESS in 17m 18s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ba76cc81540b4a82b55a5fd6050d472d : SUCCESS in 19m 26s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/846cfc64e3fb4c87888955bf574cc73a : SUCCESS in 19m 16s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e487d6eb36bb49af8e21932a6f3f7ba9 : SUCCESS in 16m 15s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3a04d742d64d4b47922c87f7e7cbba08 : SUCCESS in 17m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bb771fc4bb974250838b19ee20cdf8b3 : SUCCESS in 7m 11s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0862d2edc85c424492b23a70549081b5 : SUCCESS in 15m 21s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c75b618d002b4560a8284c4f48ecdf12 : SUCCESS in 6m 42s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/fa1d403563e74fbea06d4f96193c8c3e : SUCCESS in 58m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ae51abe0df4f476995adf2a93e058015 : SUCCESS in 42m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1cfb0b87917949039a3e79be4ff158a0 : SUCCESS in 31m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a89e1b3a2c9141b79a04b69bfb8e1841 : SUCCESS in 20m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39c870e148cb459aab03cfa48e528827 : SUCCESS in 4m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/00298a0186b94c70ac055f25cd1a5537 : SUCCESS in 19m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/048a5cfd476a4363a31a43cc2ea2f3b6 : SUCCESS in 39m 43s","accounts_in_message":[],"_revision_number":5},{"id":"bb941313238bce270534409c867c74b541cad303","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-07-16 11:04:42.000000000","message":"Patch Set 5:\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"66db1787e19ccb4c7c9de865e85671a42c7acf4d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-07-16 11:13:22.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"f381384985281bd62a6173db109cd28b6521e11f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 12:17:08.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2036a8d775cf45c5bf33079336ef69fe\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d078a0b34e3a4338939a307644bbdf55 : SUCCESS in 21m 57s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5e34a4b2594340b5b987a8a65e9213e8 : SUCCESS in 1h 01m 20s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/59b859b3c8c24f3ba1d8ff0bc512a893 : FAILURE in 50m 10s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/95c4bda0936347fe804375fff4933399 : SUCCESS in 1h 01m 50s","accounts_in_message":[],"_revision_number":6},{"id":"f4034e56f13cd111fdf5b23851c7b2254eaaab01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 12:26:52.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6235b8d76fa4e979ee300d0eef4da79\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1268c7d9cfc7429c843b119c19f706c5 : SUCCESS in 9m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/b6870d814133411ab45d5922e25fcd49 : SUCCESS in 1h 02m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6857e5c37b7e4d6582e247a8a8ea39cf : SUCCESS in 50m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/50290ef0d766400998eb1eeecad9306d : SUCCESS in 51m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/58df1b70017b42cb9d454eea9910536c : SUCCESS in 19m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1a130ca78043430fa02ddd7c75fbca11 : SUCCESS in 18m 40s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/01107c5447d24872a67e6b62dc497ae8 : SUCCESS in 21m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b73ee89793d449af8f8015e4f7d55bd5 : SUCCESS in 5m 03s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b74ff7977a90423b949a58888b850936 : SUCCESS in 18m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/331f01c9bfb949b78ed328f6bfaf2bc1 : SUCCESS in 18m 35s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/432b57790ef8452ba1a3037ab69cc107 : SUCCESS in 18m 29s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/ba7e4f67985a4c7ea666d3b300cdbaf2 : SUCCESS in 21m 51s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/841917c60c2948fa84ece678babc99e2 : SUCCESS in 35m 29s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/608be77c764543a087bf2443add35d98 : SUCCESS in 19m 45s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4e7aa04d33bf4f8cbeb0d7d326146fdd : SUCCESS in 17m 40s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/da7c28dbd8df4b9cb1576f8f86475eda : SUCCESS in 17m 32s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/dd06901e95ad410a8aeb3fff2445b05e : SUCCESS in 16m 51s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/7ef616b4650f4ece9a9761837d899dc1 : SUCCESS in 18m 43s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/147bbaaf2cca48ca9d630713ebe13992 : SUCCESS in 17m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/92ce65653408474e904b55fc4818e76d : SUCCESS in 5m 50s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2d129328ecd44dd78547f2aada9df52c : SUCCESS in 15m 33s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/6013616c4d49493188b2815f418ce2e9 : SUCCESS in 6m 19s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/93c34e5905404e84b357df246d0e7dbc : SUCCESS in 58m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3756ce7e2d9a402589ab900888f31f0a : SUCCESS in 51m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bd354ba257ca4d1c8ad57ba0181643dc : SUCCESS in 46m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ea74f60438be4935a0a0ce4679374636 : SUCCESS in 16m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ed5771ffd3d4647a135f684cb9e6b04 : SUCCESS in 4m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/46cb7c5190d846529f291800e31afc36 : SUCCESS in 19m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f501528546fa43c9a94b89d0f87d0377 : SUCCESS in 1h 00m 42s","accounts_in_message":[],"_revision_number":6},{"id":"c25b2f6038731c85d4cd379c3306681deb066037","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-07-16 19:21:53.000000000","message":"Patch Set 6:\n\n(6 comments)","accounts_in_message":[],"_revision_number":6},{"id":"95a228cd32216de38895026479b15355f1eec496","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-07-25 16:21:55.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"7e713b1a1978668fd847c1e010a41494111f5f9f","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-07-25 16:45:14.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"5bd8c83c9469e8f8612c3dc22fa51547885616cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 17:26:21.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ded70312af84b6291e913573841c38d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7f4af5ae8f447a4a96405c9e2a629dd : SUCCESS in 8m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/402c0d77f3de490b89fb561f91818830 : SUCCESS in 54m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/57e3009f2b0c44148cbc991a202d417f : SUCCESS in 48m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/82a20f4959684eb5bb63dbd799faff38 : SUCCESS in 58m 50s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/41e5f93cd2dd4e7ea5452797c269a2d5 : SUCCESS in 17m 53s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4d0f64a2b6784eb1843d229474328802 : SUCCESS in 20m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2fd782aa521740acba0addc1262cc980 : SUCCESS in 18m 08s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/db7e1b89c9a94e838d69d592973b8ab1 : SUCCESS in 5m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/d73e4348520a41c8abbe7831b3c0a39a : SUCCESS in 21m 51s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2e59dac32fb145709c918516f05d149f : SUCCESS in 19m 11s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f2bb3341b0bd47648fe5096c48abfab1 : SUCCESS in 22m 03s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/b420d47a728646b28320423da6045fc3 : SUCCESS in 26m 13s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/82272e4cad664fd4b8949bfcbea36af4 : SUCCESS in 27m 57s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/53844eceec224f3d81cb97b840d926c8 : SUCCESS in 19m 28s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fcac64f5e05b4a8d82eeda787d14e86f : SUCCESS in 18m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5eeea2b2be71489aa68931b1c8b890e8 : SUCCESS in 22m 03s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/124c48360aa743ed92ffecc61d7fd77b : SUCCESS in 17m 17s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/09ddc3bd1b284750b80a34ccdcc74acf : SUCCESS in 18m 28s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/e1e1f44512c34b1c8638ba8292c512a4 : SUCCESS in 18m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5fd6cf74764e43ee99ee7dabaecd8a34 : SUCCESS in 6m 12s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/90821566aeff4d089e1224ce66bebe63 : SUCCESS in 14m 56s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b34b63dbffc54b18902d46379b0528b4 : SUCCESS in 6m 36s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f675aefb12f84f78afcf984bcb10b4ba : SUCCESS in 47m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4aba4bb40d8e4065b0f8aea33d9e34ec : SUCCESS in 35m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/edc14f3428fd4f27bcdfac69ae941db1 : SUCCESS in 38m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3f4f5a4442664c6394e1e6da5ccbbc0f : SUCCESS in 18m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/019dc1955bea42aa9d9bbd1b64eb7d7a : SUCCESS in 4m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/522c8a867ce24e259ffa46e61d95ae8e : SUCCESS in 17m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b06722917cd943d393d0b7c9757500c0 : SUCCESS in 42m 35s","accounts_in_message":[],"_revision_number":7},{"id":"493030b6062f2bcc1251f3d37b35fd0d491fe240","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 17:33:13.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5de234f518c9402988518601f8b3bf14\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/cded8bba643d47d3bc0daeb4a2bfacb4 : SUCCESS in 1h 07m 20s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9ec98e0ad5be486b91e178eec0f546c9 : SUCCESS in 1h 09m 30s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/e75141ef9ba74e3d90f1dfe872bb68a3 : SUCCESS in 1h 01m 38s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ce00c63edeca4809a4782ab5d9cb8226 : SUCCESS in 21m 42s","accounts_in_message":[],"_revision_number":7},{"id":"8a8d7819a901d7fbc7d4d0504bd6b6085d2dc442","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-07-26 16:36:08.000000000","message":"Patch Set 7: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"6d3a91e6a9540d71d94d7bb16e5fabad117ba910","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-07-29 14:32:02.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":8},{"id":"f09630e267251d584281fa24c902eab439ad77e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-29 15:36:08.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad3e289315a640f4841cdef6f218bec2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4f1085083374c01bb55125018f039d0 : SUCCESS in 8m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/3e1144c28492464681a8b8595a1ab09a : SUCCESS in 1h 03m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9ee9c4bbe1244853a6730eb065a4c83f : SUCCESS in 45m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8a51cbcb1bcc432993a00b3a80a9c795 : SUCCESS in 47m 43s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bc353fdf3ab643b783f21746846c12a0 : SUCCESS in 19m 51s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2b6179a5f6fb4a1699587de966636399 : SUCCESS in 24m 07s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/298b33a9bae94ecb98c0c2e99e6c118d : SUCCESS in 20m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0ffe205a085346ef954f7a1682860422 : SUCCESS in 5m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7443623028a9475090d50dc6c228ca17 : SUCCESS in 22m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/238c69d99cfa4e8fbe3aed2182fe1111 : SUCCESS in 19m 34s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/80a871d5218f4f91a7cdf949306118a1 : SUCCESS in 29m 03s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/54450dc101e742eb878dd5f96a37d049 : SUCCESS in 16m 12s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/be435cba0f0a4b8092ff08b492950cb4 : SUCCESS in 25m 35s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/968dc8ca8f6341799fd7a660370be5d7 : SUCCESS in 17m 02s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/08e80dd73119492c959df456c8c478ec : SUCCESS in 22m 28s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6a68bbc0061a4d4d9041602355162cd4 : SUCCESS in 17m 18s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6e8532bacfad4df8abcc0dad3e3e25da : SUCCESS in 17m 50s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/561c41b6c5be49a9806ef228f8365a54 : SUCCESS in 16m 28s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/0567ed78fcc94ab6a0905cdebaca70bd : SUCCESS in 25m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2b7139059d0b47269a2cbd1085baef05 : SUCCESS in 8m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4036ba87c41040e4a982f29b8e157a87 : SUCCESS in 14m 46s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d11dcf3b4d014fdc809950e5da48fe06 : SUCCESS in 5m 51s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/68dcb93e4067427b9e24a257bdfb31f6 : SUCCESS in 45m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/09be01d207624c019b38516e0c0eedba : SUCCESS in 39m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e7b66f3c2a7c4f1db8d9ca840f025c38 : SUCCESS in 52m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/92072e1b65a44671a14c4110ce667f0b : SUCCESS in 18m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc8b174114214e01b641448c605bbf6e : SUCCESS in 4m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/241e3083a0b64ce39c4fa62cc2255494 : SUCCESS in 16m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e94d80d86c234c53989eb34ae32b1121 : SUCCESS in 41m 46s","accounts_in_message":[],"_revision_number":8},{"id":"298e515e05d1c45935eb37eb8739c10a1fea0589","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-29 17:24:57.000000000","message":"Patch Set 8:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de0d9ac7c3ff4b5b8aa4d83fe36fc140\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2f6ff6edc0624c319e31b91b87f9c087 : SUCCESS in 1h 09m 10s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f5e093bb9eae4377a716bfbf39f8e328 : SUCCESS in 52m 05s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/8c768da7e7a04c34bbb1277bf6db17d4 : SUCCESS in 1h 17m 06s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/739bcc30782b4736b87560cf065ddcee : TIMED_OUT in 1h 30m 52s","accounts_in_message":[],"_revision_number":8},{"id":"817b36d0f61f39893f532ae091ef7d6098039c33","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-07-30 21:22:10.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"9cd302a9e2b57b57af8e85431ab0db20cbf1ff17","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-07-30 21:22:21.000000000","message":"Patch Set 8: Code-Review+1","accounts_in_message":[],"_revision_number":8},{"id":"34a706412dce49d7acf1f7fdc8fb51717d69bdf5","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-07-31 09:46:04.000000000","message":"Patch Set 8:\n\n(7 comments)","accounts_in_message":[],"_revision_number":8},{"id":"d99c8f00f5d1de9aa662d6d8108e454b4ca46025","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-07-31 09:46:14.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"5ab472e0c0bb1f53e9040f184a1fe7450e5b383d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 10:51:55.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7871aa4a3ae34537afd7ec1736953da8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81deea16c39c450aa0e90031b20d94cd : SUCCESS in 9m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/ebaf8fa6cace464b8a5c0db327449357 : SUCCESS in 1h 03m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4278ce76e78d45d9869c4068586b1866 : SUCCESS in 50m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4bc10673ba04433e8fb860feac5ff146 : SUCCESS in 1h 00m 47s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/67297575ea4e48dca197551d106532ce : SUCCESS in 22m 57s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fe87f5eefca44114bc4b3d4ff61cf805 : SUCCESS in 26m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/75b66c9fd7804c6ab1e5c7c874a71b95 : SUCCESS in 16m 58s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/82cf593756604af3a95c4d39769ea749 : SUCCESS in 6m 02s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5ee11ba8eb8044c5938a07ee0d790656 : SUCCESS in 18m 28s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/49be8068ef4f4c4e9ecac88acdc1ed1d : SUCCESS in 15m 16s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/fa01f8804264407a8e59fd316f3dce86 : SUCCESS in 16m 47s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/3ba6fe6214c54cf79c65ab81ac311ef6 : SUCCESS in 28m 41s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b8eeaa64f8bf405f9576e71c35b10952 : SUCCESS in 25m 51s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d9067987d434470ab02611682d916330 : SUCCESS in 19m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7bd292249db64158adcdb67c944311c5 : SUCCESS in 18m 56s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/e4f23939071e4cf7b4165df503f6298a : SUCCESS in 21m 14s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8a32282e74ab4904b1e6c6822d9883c0 : SUCCESS in 16m 05s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5f925e040b7744ae9902d9e08014d09e : SUCCESS in 17m 03s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/89ad1b16e7d64c6b8d880c91bc98b11b : SUCCESS in 17m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a0683842460b4559bce58a01b25cea75 : SUCCESS in 5m 55s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/006cce8fcdc84c2d9b6ad49c358bf66f : SUCCESS in 14m 24s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d687172ec5e64362b05e3dd2041385c8 : SUCCESS in 6m 25s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/bf762f24ffb44e4d8ddaf029cb5296fb : SUCCESS in 56m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f1f6774993814e49bc0aba59c4c17e39 : SUCCESS in 49m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7371cfc5f3624565b7cdf771f1c960b2 : SUCCESS in 48m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/82a61889eff3411eb65c8f6eda210088 : SUCCESS in 29m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64ea0c765c994fcd9c049bfec35cc803 : SUCCESS in 5m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c435628d7a364fcea7b4fde5f5db6106 : SUCCESS in 28m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/29765cb2d0604da9a16a386dfe80c409 : SUCCESS in 53m 09s","accounts_in_message":[],"_revision_number":9},{"id":"98e9134dd7d5b2fab92124397df675042b78c652","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 11:19:11.000000000","message":"Patch Set 9:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f2b53866020f49d687d42b3a9a701cde\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ee09010f73da4704941af15c97e3d30e : TIMED_OUT in 1h 30m 54s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2e79f32b315d4ee3aa1a36b746fd2fbe : TIMED_OUT in 1h 31m 12s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/5888436468e949d6b64246d17ab34157 : SUCCESS in 1h 15m 39s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/235da22608d54116ab8d9fa747d6ff45 : TIMED_OUT in 1h 30m 52s","accounts_in_message":[],"_revision_number":9},{"id":"6807b99f9d4b2089285a1304963c2bf6b5e51d01","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-06 00:23:01.000000000","message":"Patch Set 9: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":9},{"id":"37e08c1bd68b7e597f63abe225fc8ae14a7d0980","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-08-06 11:42:54.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"1fc7beb5dae42bb090ae1e4fab7a52e10fa08e56","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-08-06 11:46:55.000000000","message":"Patch Set 10:\n\n(7 comments)","accounts_in_message":[],"_revision_number":10},{"id":"e71576e1b2b1858c6141f45480717f1a845d9155","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-06 12:49:32.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7a76accb08844550995105bd445892da\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b2b9119b3f543b1afe1cae9120c8a3a : SUCCESS in 8m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/f169b57f422941d89f5b3ba0282452d0 : SUCCESS in 55m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0e23f59307374b5a8fb29d70a82a0674 : FAILURE in 49m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/387e746b27b74088af76335ceccbff09 : SUCCESS in 56m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e272a708898640309d28489defc15e06 : SUCCESS in 19m 30s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/445989bf102d49f4964810e70eff918c : SUCCESS in 19m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/76ac9e5cab2c468ea5e4cdfbd276841d : SUCCESS in 18m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/23218ad1ced64c63a2a4fe8a4d155e1c : SUCCESS in 6m 22s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/1cb09415607743a1950c05d08a8b4c5c : SUCCESS in 20m 16s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/45bd73dbc83243dc8c0361a50de84c41 : SUCCESS in 20m 08s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c6860c96d4a3496d894b57512413371d : SUCCESS in 21m 34s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/e2a97827118e493b994e05968e7ae62a : SUCCESS in 15m 54s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/574e8c7076e14eefa95c2bf946fcd69b : SUCCESS in 24m 23s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9c27b195679c411daa1da396c615aafc : SUCCESS in 20m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3ad5808f38ad48e99cb83809fbd3e245 : SUCCESS in 16m 49s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a009c1df22544d83be7333bc04dc7a7d : SUCCESS in 21m 20s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a15fe809d2bf434f94eb6250ef2a720a : SUCCESS in 15m 25s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/2ae4595a47cc4dd5a4249ac484c68e8f : SUCCESS in 16m 51s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/124923d51dfd4486b0b053ab6e99a283 : SUCCESS in 16m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/caf497c3598f485db5cf28146a8352c1 : SUCCESS in 6m 22s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/af6805c2a46b44e2a3161e03a5d8d631 : SUCCESS in 15m 04s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7aa9545976e94ad1bc90eb7c5c7eee4a : SUCCESS in 6m 09s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/957c60d5156e46d597cfe20158d1b6c9 : SUCCESS in 48m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8e8214dc8b0c4920b277e82e1215cc7d : SUCCESS in 49m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0c55912e56c543d686f6fe2c6a5104c3 : SUCCESS in 46m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ee13f19f3e8243559c8d2b4686697cfb : SUCCESS in 18m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7200473e882440af8c23e411cc503e16 : SUCCESS in 4m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f24e7574504245ebb7754a6b3f6c3b1d : SUCCESS in 18m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d020c7c41af1439da6cc6d257256491f : SUCCESS in 59m 24s","accounts_in_message":[],"_revision_number":10},{"id":"437d126c08dc5099d4a359035860e656bc78ffd9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-06 13:15:34.000000000","message":"Patch Set 10:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d73fe7346ffd41a1a5e9d56f73c1d4a3\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b4cb35e2bff34698a4af7c4d85cb0f55 : SUCCESS in 49m 33s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/73478cbe80ae456ea03c8de88dfcd3c0 : SUCCESS in 1h 03m 19s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c29df3e15c2549eab06e32da96791693 : TIMED_OUT in 1h 30m 54s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/bee31caa0005459f9b75d9c1c5bef044 : SUCCESS in 1h 05m 08s","accounts_in_message":[],"_revision_number":10},{"id":"47895a5c90d09e6d0477b3582f35405dfe469dfa","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-08-08 14:50:32.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"7dcb3a1ba0d436a2d0972d82215e257766269d6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-08-12 11:14:36.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"1e18fb114d89a4fefaba25235e71dad849807c46","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-08-12 11:16:38.000000000","message":"Patch Set 10:\n\n(4 comments)","accounts_in_message":[],"_revision_number":10},{"id":"f326136dc2d508e3ed0fa0e615bbfca5b59e5f03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-12 12:25:19.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0dd8e928c13749238fea43488c106e14\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/433b7702a8ee432a88ea517fe44d9de6 : SUCCESS in 8m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/285db39d4a0940cb8e6622411b50e6b9 : SUCCESS in 1h 03m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cab92beda27a4c0b8c7590d714699a2d : SUCCESS in 40m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/85bd6ad941f5439ebeddfc275477555c : SUCCESS in 48m 24s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6f3c5e89e5274996bb9a4ff2aceb2892 : SUCCESS in 24m 44s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ec2675bcf6064a7e858db30d4b36bcd0 : SUCCESS in 22m 37s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/85695c9693254a09b46607acd711c14b : SUCCESS in 23m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/086c41d0697b412ba6728fb514a6f9ed : SUCCESS in 6m 34s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7eebf2b028d046dead4257eef7f1a2b0 : SUCCESS in 22m 05s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c41f321045af470986400cd6df5753a0 : SUCCESS in 17m 30s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/778650d25f2c464e8f65ce8aa280ac3d : SUCCESS in 21m 18s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/5bc4e4f6a9724e09ac1f16772ff09a84 : SUCCESS in 13m 21s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f0153255a09c46379830d6630a284ba8 : SUCCESS in 28m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/ee3c7b008d554a02beccf6d0d459a851 : SUCCESS in 16m 24s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b066bfafb9674f1b8a09c4dc10f33d0e : SUCCESS in 21m 31s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/03a292b6be5149e7ab70d9c937819922 : SUCCESS in 17m 43s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b73d4c23a720432fa5cfa2088fa5b677 : SUCCESS in 16m 41s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0b056f5d16b14f7a82d7a3ea22123f3d : SUCCESS in 19m 23s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8835f4a10edf4e67acba288250e8abed : SUCCESS in 19m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4189c3a103024123b04a0f820a391af6 : SUCCESS in 5m 39s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e76577d79a5c456fbcfbeb072d4d63f7 : SUCCESS in 14m 25s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/64a3f0918041442f8dfdb01ec2ac6fb9 : SUCCESS in 6m 09s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b5f14822427b435aa0c277855c54a7ad : SUCCESS in 43m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/171583f46a96477cbddaf7d3130f73fa : SUCCESS in 48m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a59ce49ab02b43c3959af50034cff75b : SUCCESS in 50m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/660b0d050434462e8af239a54884e40e : SUCCESS in 18m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba9a9108cb9f4cc5a17b379d7caf20f5 : SUCCESS in 3m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7bea62c9ed8a498682bf32e404dc558e : SUCCESS in 20m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fe2a3622fd0d40d289ac24aad351109f : SUCCESS in 40m 09s","accounts_in_message":[],"_revision_number":11},{"id":"4a85420c49a12d80391da157dbfeae1dc0dd2166","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-12 14:00:05.000000000","message":"Patch Set 11:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abfbc867a9b04cb58d5ad317f64ea6f3\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/81cd54428bef419fa8cacec7aea81f22 : SUCCESS in 53m 52s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/adac7ad5315e4a058e83572c7bce1977 : SUCCESS in 48m 05s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/a1367af0a0a34e7aa79171a2266798a7 : SUCCESS in 58m 18s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/ed22ab3bcb534b128b4cc1b6fd58d4f6 : TIMED_OUT in 1h 30m 53s","accounts_in_message":[],"_revision_number":11},{"id":"b3234ba42c433dbf43be8f9771de12f631e36583","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-08-14 14:14:57.000000000","message":"Patch Set 11: Code-Review+1\n\n(14 comments)","accounts_in_message":[],"_revision_number":11},{"id":"2918dcfe9abbdb4d0f7a7d3fa3c5b88cdc29b333","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-08-29 14:08:07.000000000","message":"Patch Set 11: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":11},{"id":"fd636bd82915ff49168f5572f1733bcc8724a869","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-09-02 14:32:41.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"896f345b5d8d0130de7d8225f55983936fd21d6e","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-09-02 14:32:50.000000000","message":"Patch Set 11:\n\n(10 comments)","accounts_in_message":[],"_revision_number":11},{"id":"25868e12bfac0aaf80985b7324a82c3d62aa6e02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-09-02 14:36:52.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"ebe101d22f8d20d2ef3afd7a8956c4a15667406a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-02 15:33:56.000000000","message":"Patch Set 13: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/637ae62b4f6248588966207ccc33bf42\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78e5749bd3fe47ef81aa691a336ed953 : SUCCESS in 8m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/20d4bc60cb6c4559a9b09c40ba06c4c9 : SUCCESS in 54m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4a18ade317ab4bbe9d5033142c222af1 : SUCCESS in 50m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/84dc61c7215f47b0a22551a9df25ee83 : SUCCESS in 54m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3511eac3a5d84ab0bd225a0ec7d913ab : SUCCESS in 18m 35s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7f211a08838a4138859919ef790cc14e : SUCCESS in 19m 03s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/36259f450c3840af8c086ef5c3a97970 : SUCCESS in 20m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5b346e2b244e4ddf9b16bfb3b9922c4a : SUCCESS in 5m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/36555d605dc44daf9492d5d9e249e770 : SUCCESS in 16m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/840498a104fc437f802533793ea86354 : SUCCESS in 15m 49s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f9fb746484cf48058bc30e95e033b659 : SUCCESS in 16m 52s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/1bcbea3253f34180bb57f9cfbc0cc1f9 : SUCCESS in 16m 28s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/373b3ce4400d459d9b1898f22c30de4f : SUCCESS in 21m 06s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/21b602f747614418a67d28bd3acf55f7 : SUCCESS in 17m 46s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bbf5ea62d9a141be93ccc61a723668cf : SUCCESS in 17m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/451811d3a72140f98c44231042dc7566 : SUCCESS in 20m 44s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0fc2f6eb241640cfaba8edffb3c78fbc : SUCCESS in 15m 45s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/94ba9502bb55499ba90f464e22c18812 : SUCCESS in 16m 11s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/fbb1c8fc97ad46bb8323a01d024845b8 : SUCCESS in 16m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1f01209f8ea64fac9e1d80d64ea1e340 : SUCCESS in 6m 04s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8befe7ec59364d71b74b1a03d54d3c6c : SUCCESS in 13m 30s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/13e76193e5214afc814b2f1281250070 : SUCCESS in 6m 20s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f98300da71b54b4e9c3b6980ab5130c4 : SUCCESS in 47m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/972dcf5ee51242eebd451ab95cf77dce : SUCCESS in 49m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a86bb6b1337e442cb0de9922b0afb776 : SUCCESS in 45m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6b5d0d31beed47b6a6047ceb9e0aa2cd : SUCCESS in 23m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fd43ecb64ee4110860e657734677cee : FAILURE in 3m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2c0c68e3beea4a55825fee59d94256ad : SUCCESS in 16m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0d1de642c7b74c7fb2401e585bc42c4a : SUCCESS in 44m 32s","accounts_in_message":[],"_revision_number":13},{"id":"ee8e7e403741d0af489ae74a45a43e7124f4c87a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-09-02 16:01:11.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"4c71f166f78c8d001878c0329ef69ebb66fe8520","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-02 17:11:40.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/080c62e53f9c4a0588b9512dd899fb8d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7449075e1dd64fb8941ad617fcc9e66c : SUCCESS in 9m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/3d7284f5d5ab45d98b27cd5c0eb415cb : SUCCESS in 45m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fd3678d7a068438f82a4bb977314a80e : SUCCESS in 36m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4b2c8ec98a704ebabf28ece251784876 : SUCCESS in 1h 04m 42s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1d0510d6ae654f4db26f8e19f0751701 : SUCCESS in 20m 33s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1f45636f929d4d469c2746a8b663fd6d : SUCCESS in 19m 20s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4407dc2891e44bec8a7edb4051dd2028 : SUCCESS in 21m 28s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c723f1b92ee94144993f8f71f8c56c18 : SUCCESS in 5m 49s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3f33c906ea0445b78668298c2c2a6125 : SUCCESS in 16m 58s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f6e01bdb9de245b7a710450381e7c222 : SUCCESS in 15m 14s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/3934bba7cb214e7e9be6b2ddca007038 : SUCCESS in 15m 01s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/527128f971a54009b8103b025b3e91f5 : SUCCESS in 28m 43s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/cdff20e700434a4598b6a147dd275438 : SUCCESS in 32m 22s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/395f911d182d4a90ae5ac8272328db1a : SUCCESS in 19m 43s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/4e88de59c0b84781a88453b5a1995af3 : SUCCESS in 19m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/eb6d7f39a20946738d6c7eef5306255c : SUCCESS in 19m 22s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/58f80af187fd4a31a0d8ec87082bd79f : SUCCESS in 17m 29s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/71b0e6ad0de449ad8b21a45eaa4f3808 : SUCCESS in 16m 00s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/431a95ee7c27404c9f47b114624ec14c : SUCCESS in 18m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a670e7df12a64230a2e218389f1ed5f0 : SUCCESS in 7m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/699782f1cc574621a49fc82d32df4940 : SUCCESS in 16m 08s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/76325376e0404129bb890778a3fb7d4f : SUCCESS in 6m 37s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/41a8d4ef7b4a45b2b913a66f86b858fb : SUCCESS in 53m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/223110fb874a452c86271d99be296d20 : SUCCESS in 45m 31s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a6e553864a6041bf81c5fac354ef39ca : SUCCESS in 46m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/65af3d272dc046c6a32398859362babf : SUCCESS in 19m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3df8fbd53aeb4b4581443d0ce0132ef5 : SUCCESS in 4m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/45920f085037428fb2d924cb6922a0c9 : SUCCESS in 19m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f8b7d56dfce0408ba001e555382ee03c : SUCCESS in 40m 15s","accounts_in_message":[],"_revision_number":14},{"id":"9b8f15581b5321d76ae47406068fb262a57d64a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-02 17:21:07.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ff1b4ddaf9245b090e8eee0cada9b7c\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3373f9d1f709437ea498e06685e3b0cb : SUCCESS in 43m 42s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4209c9f9dadc487f8dce686e834c6a7c : SUCCESS in 43m 10s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/afd789dae5b0474696817c99f483527c : SUCCESS in 1h 08m 13s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/896309594007410087ffeebecbc4f42e : SUCCESS in 1h 18m 15s","accounts_in_message":[],"_revision_number":14},{"id":"17b5992f79b48896f06cb6919ce31ceef807c16f","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2024-10-18 15:35:59.000000000","message":"Patch Set 14:\n\n(3 comments)","accounts_in_message":[],"_revision_number":14},{"id":"7fc4ea41d2b55f71b5af7b58a691886345a1ac88","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-11-22 01:23:30.000000000","message":"Patch Set 15: Patch Set 14 was rebased\n\nThe following files contain Git conflicts:\n* test/unit/common/test_statsd_client.py\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"a96e259ccd55f96ab153fea7fc7b0c2216e32015","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2024-11-22 01:31:57.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"1c1dc43917ec8f9695222db4771c875e2fcdb279","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-22 02:37:46.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/78154e2e0c3e4005ae1dd1348a4a125a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/818934ef79654724bafa039d5155f800 : SUCCESS in 8m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/e1d5ade275ca4d938b14c6767a527a0f : SUCCESS in 1h 00m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e9bea9c58e08402fbf741fb64f2dbb6a : SUCCESS in 49m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3f0d634d5ed042b28bd21240dd75c7e6 : SUCCESS in 33m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8ba11530f7084d32bf36c9f1581125fa : FAILURE in 34m 29s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2abbdae8bf274b8c96618e9d8b414f67 : SUCCESS in 16m 03s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d6da461d13864c71a6b2f48f861dfb29 : SUCCESS in 11m 24s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/58d8a085b1334b9b8ee665bd51db365c : SUCCESS in 15m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6d8a74c2b91e4b33aa23c622fab96b54 : SUCCESS in 4m 42s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ae5bd1a9142340479ff02ab5b691a7d6 : SUCCESS in 8m 54s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/1d9db04b0f70467098b07492e05667e7 : SUCCESS in 17m 45s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6d5d3deb1b4544a7af600c27a313e7ab : SUCCESS in 17m 38s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/24730ab6e27842d289c9808e1ee64422 : SUCCESS in 19m 16s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3d4ade70011b4419bf80bad3cd37dd8e : SUCCESS in 15m 19s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e48697c86e36423a86428765a9291ccb : SUCCESS in 17m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/115e7d9a62fa48189461a7bf00f9d605 : SUCCESS in 16m 42s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/058b0ed17e0d41c992a15508f1a6e66a : SUCCESS in 15m 45s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e936861e3c944dbab3a15cc2c1f6622e : SUCCESS in 14m 06s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/025e20fb0e5c45868548f0cdd80f8c2c : SUCCESS in 17m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c1c51524234b4c97aed0d7226e2a37c7 : FAILURE in 6m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a1ebe59b38634635b415c948eaed0782 : SUCCESS in 14m 22s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9f8410c247d949b5b1db53f7081d3450 : SUCCESS in 7m 00s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3a55430afb7c4b48ad9556f7b245fb7f : SUCCESS in 50m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6c5fba826eed4ad58929755fac610af4 : SUCCESS in 52m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a5f5bef9a4854044a10480b71e09f1e8 : SUCCESS in 48m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/790a7f37189f4f4b80f6d695dedfb3ae : SUCCESS in 19m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/043d7d9b13134513b546310d1524b423 : SUCCESS in 4m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e48b5dcb01cf4c5c98e9d93d22dd8483 : SUCCESS in 17m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b0ff4ca87a804f2aafa445a26b731bc7 : SUCCESS in 44m 19s","accounts_in_message":[],"_revision_number":16},{"id":"d0ee9a0e3914ad70235866fc77dc073ba6fe0454","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-22 04:36:36.000000000","message":"Patch Set 16:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38724302a54143bc9187c1e415f4e79a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cb2a7e000a4141579afc3ddc0dd24658 : TIMED_OUT in 3h 01m 06s","accounts_in_message":[],"_revision_number":16}],"current_revision_number":16,"current_revision":"d972f57c797de9908629ff4d97fdd94c58850c84","revisions":{"b74c760a276c7b03095283460355e8e3f203ed59":{"kind":"REWORK","_number":1,"created":"2024-06-21 20:29:27.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/18/922518/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/922518/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/922518/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/922518/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/922518/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/922518/1"}}},"commit":{"parents":[{"commit":"0c58afc7ee23fd5e5cca19318e3f055dcc79e021","subject":"Add get_statsd_client function","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0c58afc7ee23fd5e5cca19318e3f055dcc79e021"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-21 20:06:30.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-21 20:29:00.000000000","tz":-420},"subject":"statsd: deprecate log_ prefix for options","message":"statsd: deprecate log_ prefix for options\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nChange-Id: I49d40d10c75443def6476ed8c8d4cc4473291484\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b74c760a276c7b03095283460355e8e3f203ed59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b74c760a276c7b03095283460355e8e3f203ed59"}]},"branch":"refs/heads/master"},"1979b9fb382e3381a932edc0c35b10eac0d26312":{"kind":"REWORK","_number":2,"created":"2024-06-26 19:38:53.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/922518/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/922518/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/922518/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/922518/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/922518/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/922518/2"}}},"commit":{"parents":[{"commit":"37169aa6ac58082737f31c893d923427428329d1","subject":"Merge \"proxy-server: log correct path when listing parsing fails\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/37169aa6ac58082737f31c893d923427428329d1"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-21 20:06:30.000000000","tz":-420},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-06-26 19:38:02.000000000","tz":-420},"subject":"statsd: deprecate log_ prefix for options","message":"statsd: deprecate log_ prefix for options\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nChange-Id: I49d40d10c75443def6476ed8c8d4cc4473291484\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1979b9fb382e3381a932edc0c35b10eac0d26312"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1979b9fb382e3381a932edc0c35b10eac0d26312"}]},"branch":"refs/heads/master"},"98120c782d44a74011c41516ecd7c0db601f1e40":{"kind":"REWORK","_number":3,"created":"2024-06-26 19:45:24.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/922518/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/922518/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/922518/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/922518/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/922518/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/922518/3"}}},"commit":{"parents":[{"commit":"37169aa6ac58082737f31c893d923427428329d1","subject":"Merge \"proxy-server: log correct path when listing parsing fails\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/37169aa6ac58082737f31c893d923427428329d1"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-21 20:06:30.000000000","tz":-420},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-06-26 19:45:05.000000000","tz":-420},"subject":"statsd: deprecate log_ prefix for options","message":"statsd: deprecate log_ prefix for options\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nChange-Id: I49d40d10c75443def6476ed8c8d4cc4473291484\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/98120c782d44a74011c41516ecd7c0db601f1e40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/98120c782d44a74011c41516ecd7c0db601f1e40"}]},"branch":"refs/heads/master"},"4c9f75510f26306c26b806d6394d390896118dbf":{"kind":"REWORK","_number":4,"created":"2024-07-09 18:29:07.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/922518/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/922518/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/922518/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/922518/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/922518/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/922518/4"}}},"commit":{"parents":[{"commit":"37169aa6ac58082737f31c893d923427428329d1","subject":"Merge \"proxy-server: log correct path when listing parsing fails\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/37169aa6ac58082737f31c893d923427428329d1"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-21 20:06:30.000000000","tz":-420},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-07-09 18:28:49.000000000","tz":-240},"subject":"statsd: deprecate log_ prefix for options","message":"statsd: deprecate log_ prefix for options\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nChange-Id: I49d40d10c75443def6476ed8c8d4cc4473291484\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4c9f75510f26306c26b806d6394d390896118dbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4c9f75510f26306c26b806d6394d390896118dbf"}]},"branch":"refs/heads/master"},"a509bb3dd3643e612dcbdd7019941ad85d733301":{"kind":"REWORK","_number":5,"created":"2024-07-09 23:29:58.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/922518/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/922518/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/922518/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/922518/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/922518/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/922518/5"}}},"commit":{"parents":[{"commit":"37169aa6ac58082737f31c893d923427428329d1","subject":"Merge \"proxy-server: log correct path when listing parsing fails\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/37169aa6ac58082737f31c893d923427428329d1"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-21 20:06:30.000000000","tz":-420},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-07-09 23:29:40.000000000","tz":-240},"subject":"statsd: deprecate log_ prefix for options","message":"statsd: deprecate log_ prefix for options\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nChange-Id: I49d40d10c75443def6476ed8c8d4cc4473291484\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a509bb3dd3643e612dcbdd7019941ad85d733301"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a509bb3dd3643e612dcbdd7019941ad85d733301"}]},"branch":"refs/heads/master"},"6ed1b4dc1076d779dcc9f87c42f5579d6a178b6a":{"kind":"REWORK","_number":6,"created":"2024-07-16 11:13:22.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/18/922518/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/922518/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/922518/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/922518/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/922518/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/922518/6"}}},"commit":{"parents":[{"commit":"d555755423c179d23ef02ab2e0b856fd2b04fa71","subject":"proxy_logging config: unit tests and doc pointers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d555755423c179d23ef02ab2e0b856fd2b04fa71"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-21 20:06:30.000000000","tz":-420},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-07-16 11:09:05.000000000","tz":60},"subject":"statsd: deprecate log_ prefix for options","message":"statsd: deprecate log_ prefix for options\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nChange-Id: I49d40d10c75443def6476ed8c8d4cc4473291484\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6ed1b4dc1076d779dcc9f87c42f5579d6a178b6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6ed1b4dc1076d779dcc9f87c42f5579d6a178b6a"}]},"branch":"refs/heads/master"},"28f60a5897dab81744c20391fbdb9c3a56a59473":{"kind":"REWORK","_number":7,"created":"2024-07-25 16:21:55.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/18/922518/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/922518/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/922518/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/922518/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/922518/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/922518/7"}}},"commit":{"parents":[{"commit":"c581f85b7f6255895e5e58c40dbb394e331637c3","subject":"Merge \"Pass timeout as kwarg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c581f85b7f6255895e5e58c40dbb394e331637c3"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-21 20:06:30.000000000","tz":-420},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-07-25 16:21:45.000000000","tz":60},"subject":"statsd: deprecate log_ prefix for options","message":"statsd: deprecate log_ prefix for options\n\nIn ProxyLoggingMiddleware also deprecate the access_log_ prefix for\nstatsd options.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nChange-Id: I49d40d10c75443def6476ed8c8d4cc4473291484\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/28f60a5897dab81744c20391fbdb9c3a56a59473"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/28f60a5897dab81744c20391fbdb9c3a56a59473"}]},"branch":"refs/heads/master"},"2bf74940870a493d44425dcd9ed10816ff0f0bd7":{"kind":"REWORK","_number":8,"created":"2024-07-29 14:32:02.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/18/922518/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/922518/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/922518/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/922518/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/922518/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/922518/8"}}},"commit":{"parents":[{"commit":"c581f85b7f6255895e5e58c40dbb394e331637c3","subject":"Merge \"Pass timeout as kwarg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c581f85b7f6255895e5e58c40dbb394e331637c3"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-21 20:06:30.000000000","tz":-420},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-07-29 14:24:03.000000000","tz":60},"subject":"statsd: deprecate log_ prefix for options","message":"statsd: deprecate log_ prefix for options\n\nThe following statsd config options are now supported:\n\n  \u0027statsd_host\u0027\n  \u0027statsd_port\u0027\n  \u0027statsd_metric_prefix\u0027\n  \u0027statsd_default_sample_rate\u0027\n  \u0027statsd_sample_rate_factor\u0027\n\nLegacy options having the same names but with a \u0027log_\u0027 prefix are now\ndeprecated. A warning will be logged if they are used.\n\nIt is important that ops simultaneously update these statsd options in\nall proxy server config sections (e.g. [DEFAULTS] and\n[filter:proxy_logging]. A ValueError will be raised when instantiating\na StatsdClient if both a supported and deprecated option is configured\nfor the same equivalent name (e.g. configuring \u0027statsd_host\u0027 and\n\u0027log_statsd_host\u0027 will cause an error).\n\nIn ProxyLoggingMiddleware, the \u0027access_log_\u0027 prefix is also deprecated\nfor statsd options. The paste-deploy \u0027set\u0027 syntax should be used to\nvary statsd options in the proxy_logging filter section.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nChange-Id: I49d40d10c75443def6476ed8c8d4cc4473291484\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2bf74940870a493d44425dcd9ed10816ff0f0bd7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2bf74940870a493d44425dcd9ed10816ff0f0bd7"}]},"branch":"refs/heads/master"},"980eb1b603be2866d5c902cbefb88456139d29a5":{"kind":"REWORK","_number":9,"created":"2024-07-31 09:46:14.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/18/922518/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/922518/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/922518/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/922518/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/922518/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/922518/9"}}},"commit":{"parents":[{"commit":"c581f85b7f6255895e5e58c40dbb394e331637c3","subject":"Merge \"Pass timeout as kwarg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c581f85b7f6255895e5e58c40dbb394e331637c3"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-21 20:06:30.000000000","tz":-420},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-07-31 09:45:55.000000000","tz":60},"subject":"statsd: deprecate log_ prefix for options","message":"statsd: deprecate log_ prefix for options\n\nThe following statsd config options are now supported:\n\n  \u0027statsd_host\u0027\n  \u0027statsd_port\u0027\n  \u0027statsd_metric_prefix\u0027\n  \u0027statsd_default_sample_rate\u0027\n  \u0027statsd_sample_rate_factor\u0027\n\nLegacy options having the same names but with a \u0027log_\u0027 prefix are now\ndeprecated. A warning will be logged if they are used.\n\nIt is important that ops simultaneously update these statsd options in\nall proxy server config sections (e.g. [DEFAULTS] and\n[filter:proxy_logging]. A ValueError will be raised when instantiating\na StatsdClient if both a supported and deprecated option is configured\nfor the same equivalent name (e.g. configuring \u0027statsd_host\u0027 and\n\u0027log_statsd_host\u0027 will cause an error).\n\nIn ProxyLoggingMiddleware, the \u0027access_log_\u0027 prefix is also deprecated\nfor statsd options. The paste-deploy \u0027set\u0027 syntax should be used to\nvary statsd options in the proxy_logging filter section.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nChange-Id: I49d40d10c75443def6476ed8c8d4cc4473291484\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/980eb1b603be2866d5c902cbefb88456139d29a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/980eb1b603be2866d5c902cbefb88456139d29a5"}]},"branch":"refs/heads/master"},"d6f6b5ddcc38493340a873c38da447b0c79b377e":{"kind":"REWORK","_number":10,"created":"2024-08-06 11:42:54.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/18/922518/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/922518/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/922518/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/922518/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/922518/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/922518/10"}}},"commit":{"parents":[{"commit":"c581f85b7f6255895e5e58c40dbb394e331637c3","subject":"Merge \"Pass timeout as kwarg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c581f85b7f6255895e5e58c40dbb394e331637c3"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-21 20:06:30.000000000","tz":-420},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-08-06 11:42:11.000000000","tz":60},"subject":"statsd: deprecate log_ prefix for options","message":"statsd: deprecate log_ prefix for options\n\nThe following statsd config options are now supported:\n\n  \u0027statsd_host\u0027\n  \u0027statsd_port\u0027\n  \u0027statsd_metric_prefix\u0027\n  \u0027statsd_default_sample_rate\u0027\n  \u0027statsd_sample_rate_factor\u0027\n\nLegacy options having the same names but with a \u0027log_\u0027 prefix are now\ndeprecated. A warning will be logged if they are used.\n\nIt is important that ops simultaneously update these statsd options in\nall proxy server config sections (e.g. [DEFAULTS] and\n[filter:proxy_logging]. A ValueError will be raised when instantiating\na StatsdClient if both a supported and deprecated option is configured\nfor the same equivalent name (e.g. configuring \u0027statsd_host\u0027 and\n\u0027log_statsd_host\u0027 will cause an error).\n\nIn ProxyLoggingMiddleware, the \u0027access_log_\u0027 prefix is also deprecated\nfor statsd options. The paste-deploy \u0027set\u0027 syntax should be used to\nvary statsd options in the proxy_logging filter section.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nChange-Id: I49d40d10c75443def6476ed8c8d4cc4473291484\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d6f6b5ddcc38493340a873c38da447b0c79b377e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d6f6b5ddcc38493340a873c38da447b0c79b377e"}]},"branch":"refs/heads/master"},"969efd29b7ac91e5fff1388b4a5d7958467e0597":{"kind":"REWORK","_number":11,"created":"2024-08-12 11:14:36.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/18/922518/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/922518/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/922518/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/922518/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/922518/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/922518/11"}}},"commit":{"parents":[{"commit":"c581f85b7f6255895e5e58c40dbb394e331637c3","subject":"Merge \"Pass timeout as kwarg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c581f85b7f6255895e5e58c40dbb394e331637c3"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-21 20:06:30.000000000","tz":-420},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-08-12 08:38:25.000000000","tz":60},"subject":"statsd: deprecate log_ prefix for options","message":"statsd: deprecate log_ prefix for options\n\nThe following statsd config options are now supported:\n\n  \u0027statsd_host\u0027\n  \u0027statsd_port\u0027\n  \u0027statsd_metric_prefix\u0027\n  \u0027statsd_default_sample_rate\u0027\n  \u0027statsd_sample_rate_factor\u0027\n\nLegacy options having the same names but with a \u0027log_\u0027 prefix are now\ndeprecated. A warning will be logged if they are used.\n\nIt is important that ops simultaneously update these statsd options in\nall proxy server config sections (e.g. [DEFAULTS] and\n[filter:proxy_logging]. A ValueError will be raised when instantiating\na StatsdClient if both a supported and deprecated option is configured\nfor the same equivalent name (e.g. configuring \u0027statsd_host\u0027 and\n\u0027log_statsd_host\u0027 will cause an error).\n\nIn ProxyLoggingMiddleware, the \u0027access_log_\u0027 prefix is also deprecated\nfor statsd options. The paste-deploy \u0027set\u0027 syntax should be used to\nvary statsd options in the proxy_logging filter section.\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nChange-Id: I49d40d10c75443def6476ed8c8d4cc4473291484\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/969efd29b7ac91e5fff1388b4a5d7958467e0597"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/969efd29b7ac91e5fff1388b4a5d7958467e0597"}]},"branch":"refs/heads/master"},"8134a99e5eef3edc90b32220e0646cd1aa0d6aca":{"kind":"REWORK","_number":12,"created":"2024-09-02 14:32:41.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/18/922518/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/922518/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/922518/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/922518/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/922518/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/922518/12"}}},"commit":{"parents":[{"commit":"c581f85b7f6255895e5e58c40dbb394e331637c3","subject":"Merge \"Pass timeout as kwarg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c581f85b7f6255895e5e58c40dbb394e331637c3"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-21 20:06:30.000000000","tz":-420},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-09-02 14:18:04.000000000","tz":60},"subject":"statsd: deprecate log_ prefix for options","message":"statsd: deprecate log_ prefix for options\n\nThe following statsd config options are now supported:\n\n  \u0027statsd_host\u0027\n  \u0027statsd_port\u0027\n  \u0027statsd_metric_prefix\u0027\n  \u0027statsd_default_sample_rate\u0027\n  \u0027statsd_sample_rate_factor\u0027\n\nLegacy options having the same names but with a \u0027log_\u0027 prefix are now\ndeprecated. A warning will be logged if they are used.\n\nOperators should update these statsd options in all proxy server\nconfig sections (e.g. [DEFAULTS] and [filter:proxy_logging]. A warning\nwill be logged if a deprecated option is configured. The value of a\nnew style option will take precedence over the equivalent deprecated\noption if both are configured (e.g. \u0027statsd_host\u0027 will take precedence\nover \u0027log_statsd_host\u0027).\n\nIn ProxyLoggingMiddleware, the \u0027access_log_\u0027 prefix is also deprecated\nfor statsd options. The paste-deploy \u0027set\u0027 syntax should be used to\nvary statsd options in the proxy-logging filter section. The value of\na new style option in the [DEFAULTS] section will take precedence over\nthe equivalent deprecated in the [filter:proxy-logging] section option\nif both are configured (e.g. \u0027statsd_host\u0027 in [DEFAULTS] will take\nprecedence over \u0027access_log_statsd_host\u0027 in [filter:proxy-logging]).\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nChange-Id: I49d40d10c75443def6476ed8c8d4cc4473291484\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8134a99e5eef3edc90b32220e0646cd1aa0d6aca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8134a99e5eef3edc90b32220e0646cd1aa0d6aca"}]},"branch":"refs/heads/master"},"f73daf6e706b77387e1b490e99fe4525f58c8efc":{"kind":"REWORK","_number":13,"created":"2024-09-02 14:36:52.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/18/922518/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/922518/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/922518/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/922518/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/922518/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/922518/13"}}},"commit":{"parents":[{"commit":"c581f85b7f6255895e5e58c40dbb394e331637c3","subject":"Merge \"Pass timeout as kwarg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c581f85b7f6255895e5e58c40dbb394e331637c3"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-21 20:06:30.000000000","tz":-420},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-09-02 14:36:45.000000000","tz":60},"subject":"statsd: deprecate log_ prefix for options","message":"statsd: deprecate log_ prefix for options\n\nThe following statsd config options are now supported:\n\n  \u0027statsd_host\u0027\n  \u0027statsd_port\u0027\n  \u0027statsd_metric_prefix\u0027\n  \u0027statsd_default_sample_rate\u0027\n  \u0027statsd_sample_rate_factor\u0027\n\nLegacy options having the same names but with a \u0027log_\u0027 prefix are now\ndeprecated. A warning will be logged if they are used.\n\nOperators should update these statsd options in all proxy server\nconfig sections (e.g. [DEFAULTS] and [filter:proxy_logging]. A warning\nwill be logged if a deprecated option is configured. The value of a\nnew style option will take precedence over the equivalent deprecated\noption if both are configured (e.g. \u0027statsd_host\u0027 will take precedence\nover \u0027log_statsd_host\u0027).\n\nIn ProxyLoggingMiddleware, the \u0027access_log_\u0027 prefix is also deprecated\nfor statsd options. The paste-deploy \u0027set\u0027 syntax should be used to\nvary statsd options in the proxy-logging filter section. The value of\na new style option in the [DEFAULTS] section will take precedence over\nthe equivalent deprecated in the [filter:proxy-logging] section option\nif both are configured (e.g. \u0027statsd_host\u0027 in [DEFAULTS] will take\nprecedence over \u0027access_log_statsd_host\u0027 in [filter:proxy-logging]).\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nChange-Id: I49d40d10c75443def6476ed8c8d4cc4473291484\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f73daf6e706b77387e1b490e99fe4525f58c8efc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f73daf6e706b77387e1b490e99fe4525f58c8efc"}]},"branch":"refs/heads/master"},"3453d4ee6e260e78161a730dab1f052397977338":{"kind":"REWORK","_number":14,"created":"2024-09-02 16:01:11.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/18/922518/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/922518/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/922518/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/922518/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/922518/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/922518/14"}}},"commit":{"parents":[{"commit":"c581f85b7f6255895e5e58c40dbb394e331637c3","subject":"Merge \"Pass timeout as kwarg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c581f85b7f6255895e5e58c40dbb394e331637c3"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-21 20:06:30.000000000","tz":-420},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-09-02 16:00:57.000000000","tz":60},"subject":"statsd: deprecate log_ prefix for options","message":"statsd: deprecate log_ prefix for options\n\nThe following statsd config options are now supported:\n\n  \u0027statsd_host\u0027\n  \u0027statsd_port\u0027\n  \u0027statsd_metric_prefix\u0027\n  \u0027statsd_default_sample_rate\u0027\n  \u0027statsd_sample_rate_factor\u0027\n\nLegacy options having the same names but with a \u0027log_\u0027 prefix are now\ndeprecated. A warning will be logged if they are used.\n\nOperators should update these statsd options in all proxy server\nconfig sections (e.g. [DEFAULTS] and [filter:proxy_logging]. A warning\nwill be logged if a deprecated option is configured. The value of a\nnew style option will take precedence over the equivalent deprecated\noption if both are configured (e.g. \u0027statsd_host\u0027 will take precedence\nover \u0027log_statsd_host\u0027).\n\nIn ProxyLoggingMiddleware, the \u0027access_log_\u0027 prefix is also deprecated\nfor statsd options. The paste-deploy \u0027set\u0027 syntax should be used to\nvary statsd options in the proxy-logging filter section. The value of\na new style option in the [DEFAULTS] section will take precedence over\nthe equivalent deprecated in the [filter:proxy-logging] section option\nif both are configured (e.g. \u0027statsd_host\u0027 in [DEFAULTS] will take\nprecedence over \u0027access_log_statsd_host\u0027 in [filter:proxy-logging]).\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nChange-Id: I49d40d10c75443def6476ed8c8d4cc4473291484\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3453d4ee6e260e78161a730dab1f052397977338"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3453d4ee6e260e78161a730dab1f052397977338"}]},"branch":"refs/heads/master"},"08a182995c13722f041c390b5be5a2cac2d96ebb":{"kind":"REWORK","_number":15,"created":"2024-11-22 01:23:30.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/922518/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/922518/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/922518/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/922518/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/922518/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/922518/15"}}},"commit":{"parents":[{"commit":"ffbf17e47c0c76d4daa6e40811338e4abec1beb4","subject":"Fix duplicate prefix in exception logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ffbf17e47c0c76d4daa6e40811338e4abec1beb4"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-21 20:06:30.000000000","tz":-420},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-11-22 01:23:30.000000000","tz":0},"subject":"statsd: deprecate log_ prefix for options","message":"statsd: deprecate log_ prefix for options\n\nThe following statsd config options are now supported:\n\n  \u0027statsd_host\u0027\n  \u0027statsd_port\u0027\n  \u0027statsd_metric_prefix\u0027\n  \u0027statsd_default_sample_rate\u0027\n  \u0027statsd_sample_rate_factor\u0027\n\nLegacy options having the same names but with a \u0027log_\u0027 prefix are now\ndeprecated. A warning will be logged if they are used.\n\nOperators should update these statsd options in all proxy server\nconfig sections (e.g. [DEFAULTS] and [filter:proxy_logging]. A warning\nwill be logged if a deprecated option is configured. The value of a\nnew style option will take precedence over the equivalent deprecated\noption if both are configured (e.g. \u0027statsd_host\u0027 will take precedence\nover \u0027log_statsd_host\u0027).\n\nIn ProxyLoggingMiddleware, the \u0027access_log_\u0027 prefix is also deprecated\nfor statsd options. The paste-deploy \u0027set\u0027 syntax should be used to\nvary statsd options in the proxy-logging filter section. The value of\na new style option in the [DEFAULTS] section will take precedence over\nthe equivalent deprecated in the [filter:proxy-logging] section option\nif both are configured (e.g. \u0027statsd_host\u0027 in [DEFAULTS] will take\nprecedence over \u0027access_log_statsd_host\u0027 in [filter:proxy-logging]).\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nChange-Id: I49d40d10c75443def6476ed8c8d4cc4473291484\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/08a182995c13722f041c390b5be5a2cac2d96ebb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/08a182995c13722f041c390b5be5a2cac2d96ebb"}]},"branch":"refs/heads/master","description":"Rebase"},"d972f57c797de9908629ff4d97fdd94c58850c84":{"kind":"REWORK","_number":16,"created":"2024-11-22 01:31:57.000000000","uploader":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"ref":"refs/changes/18/922518/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/18/922518/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/922518/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/922518/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/18/922518/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/18/922518/16"}}},"commit":{"parents":[{"commit":"ffbf17e47c0c76d4daa6e40811338e4abec1beb4","subject":"Fix duplicate prefix in exception logging","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ffbf17e47c0c76d4daa6e40811338e4abec1beb4"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-21 20:06:30.000000000","tz":-420},"committer":{"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","date":"2024-11-22 01:31:40.000000000","tz":-300},"subject":"statsd: deprecate log_ prefix for options","message":"statsd: deprecate log_ prefix for options\n\nThe following statsd config options are now supported:\n\n  \u0027statsd_host\u0027\n  \u0027statsd_port\u0027\n  \u0027statsd_metric_prefix\u0027\n  \u0027statsd_default_sample_rate\u0027\n  \u0027statsd_sample_rate_factor\u0027\n\nLegacy options having the same names but with a \u0027log_\u0027 prefix are now\ndeprecated. A warning will be logged if they are used.\n\nOperators should update these statsd options in all proxy server\nconfig sections (e.g. [DEFAULTS] and [filter:proxy_logging]. A warning\nwill be logged if a deprecated option is configured. The value of a\nnew style option will take precedence over the equivalent deprecated\noption if both are configured (e.g. \u0027statsd_host\u0027 will take precedence\nover \u0027log_statsd_host\u0027).\n\nIn ProxyLoggingMiddleware, the \u0027access_log_\u0027 prefix is also deprecated\nfor statsd options. The paste-deploy \u0027set\u0027 syntax should be used to\nvary statsd options in the proxy-logging filter section. The value of\na new style option in the [DEFAULTS] section will take precedence over\nthe equivalent deprecated in the [filter:proxy-logging] section option\nif both are configured (e.g. \u0027statsd_host\u0027 in [DEFAULTS] will take\nprecedence over \u0027access_log_statsd_host\u0027 in [filter:proxy-logging]).\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Shreeya Deshpande \u003cshreeyad@nvidia.com\u003e\nChange-Id: I49d40d10c75443def6476ed8c8d4cc4473291484\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d972f57c797de9908629ff4d97fdd94c58850c84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d972f57c797de9908629ff4d97fdd94c58850c84"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
