)]}'
{"id":"openstack%2Fswift~922684","triplet_id":"openstack%2Fswift~feature%2Fs3api~I7c8469d7823195f50d2e873caed5944c8a7ca056","project":"openstack/swift","branch":"feature/s3api","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-06-27 16:39:32.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I7c8469d7823195f50d2e873caed5944c8a7ca056","subject":"Remove CI from defunct branch","status":"MERGED","created":"2024-06-24 22:30:08.000000000","updated":"2024-06-27 20:19:00.000000000","submitted":"2024-06-27 20:19:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"922684","meta_rev_id":"648dfd5bbe23558d8ef8f68d1077c7f404c4e963","_number":922684,"virtual_id_number":922684,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-27 20:19:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-06-27 16:39:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-06-27 16:39:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-24 22:36:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"391f4d726d3e8afc36272991171e4d8d4436611a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-24 22:30:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eda30e68e4a9a20f9bbf36a3a34014b5bcd83c19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-24 22:36:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0359b8089104ba891829264c114acaf\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3041b6ac708d4b08b8ea1956217126ac : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":1},{"id":"295a092a0cc1d25047122c54f4ea07d2894624fd","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-24 22:38:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e91ab50c3611c6c358c6d7a1fdfef3adbfffed9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-24 23:15:39.000000000","message":"Patch Set 1: Verified-2\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/swift on branch feature/losf.  The\nproblem was:\n\n  Job swift-tox-func-encryption not defined\n\nThe problem appears in the the \"opendev.org/openstack/swift\" project stanza:\n\n  project:\n      templates:\n        - publish-openstack-docs-pti\n        - periodic-stable-jobs\n        - check-requirements\n        - release-notes-jobs-python3\n        - integrated-gate-object-storage\n      check:\n        jobs:\n  ...\n\n  in \"openstack/swift/.zuul.yaml@feature/losf\", line 456","accounts_in_message":[],"_revision_number":1},{"id":"1b9821fdc017f6d6039e393b0f0abaa9aa0c4d87","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-27 02:25:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b33a29e9ea91c895714c03cb0998e7a8ea637edd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 02:26:26.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/swift on branch unmaintained/victoria.  The\nproblem was:\n\n  Job swift-tox-func-encryption not defined\n\nThe problem appears in the the \"opendev.org/openstack/swift\" project stanza:\n\n  project:\n      vars:\n        ensure_tox_version: \u0027\u003c4\u0027\n      templates:\n        - publish-openstack-docs-pti\n        - periodic-stable-jobs\n        - check-requirements\n        - release-notes-jobs-python3\n        - integrated-gate-object-storage\n  ...\n\n  in \"openstack/swift/.zuul.yaml@unmaintained/victoria\", line 377","accounts_in_message":[],"_revision_number":1},{"id":"3b7427812b14d924b1e2be57834334b1b9f76721","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-27 16:07:27.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"f92a0e56db805886f598ff9c121d6be7edaab1ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 16:38:37.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56d800fecb8d4e929d38014bcef84608\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0efe86e7d96c4c488625a24552c689dc : SUCCESS in 2m 06s","accounts_in_message":[],"_revision_number":2},{"id":"9d49e5b7b4ced93726adad56cd6f33cd542f4763","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-27 16:39:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9f99091d5898bf51fee609799bfa2774d7f5730a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 18:36:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e0a003ccc21f81d9cbf02880744082b10c435dc7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 20:19:00.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8f781a2a038458aa16af3152db86b53\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/434cddf4963b463396ef62b1b75645a2 : SUCCESS in 2m 11s","accounts_in_message":[],"_revision_number":2},{"id":"648dfd5bbe23558d8ef8f68d1077c7f404c4e963","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 20:19:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5ba6391bc6ee6262ad6bb2af36601b42ddbab684","revisions":{"1237ead7a655cf89fe7e853d5d6a1893a316291a":{"kind":"REWORK","_number":1,"created":"2024-06-24 22:30:08.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/84/922684/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/922684/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/922684/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/922684/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/922684/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/922684/1"}}},"commit":{"parents":[{"commit":"684139c60f8f9ff1e18149ebb8d37d3cc2db1d18","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/684139c60f8f9ff1e18149ebb8d37d3cc2db1d18"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-24 22:29:39.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-24 22:29:39.000000000","tz":-420},"subject":"Remove CI from defunct branch","message":"Remove CI from defunct branch\n\nChange-Id: I7c8469d7823195f50d2e873caed5944c8a7ca056\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1237ead7a655cf89fe7e853d5d6a1893a316291a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1237ead7a655cf89fe7e853d5d6a1893a316291a"}]},"branch":"refs/heads/feature/s3api"},"5ba6391bc6ee6262ad6bb2af36601b42ddbab684":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-06-27 16:07:27.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/84/922684/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/922684/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/922684/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/922684/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/922684/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/922684/2"}}},"commit":{"parents":[{"commit":"684139c60f8f9ff1e18149ebb8d37d3cc2db1d18","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/684139c60f8f9ff1e18149ebb8d37d3cc2db1d18"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-24 22:29:39.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-27 16:07:27.000000000","tz":0},"subject":"Remove CI from defunct branch","message":"Remove CI from defunct branch\n\nDepends-On: https://review.opendev.org/c/openstack/swift/+/922905\nChange-Id: I7c8469d7823195f50d2e873caed5944c8a7ca056\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5ba6391bc6ee6262ad6bb2af36601b42ddbab684"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5ba6391bc6ee6262ad6bb2af36601b42ddbab684"}]},"branch":"refs/heads/feature/s3api","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
