)]}'
{"id":"openstack%2Fswift~922691","triplet_id":"openstack%2Fswift~stable%2F2024.1~Id7f8acffe122bdd18932c523e6c1e7d042d81b03","project":"openstack/swift","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-06-25 17:26:49.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Id7f8acffe122bdd18932c523e6c1e7d042d81b03","subject":"[stable-only] CI: Remove some old rolling-upgrade jobs","status":"MERGED","created":"2024-06-24 23:59:30.000000000","updated":"2024-06-26 01:11:07.000000000","submitted":"2024-06-26 01:07:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"922691","meta_rev_id":"bf63d83bed1d95938ae86b0afb361afe172d1538","_number":922691,"virtual_id_number":922691,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-26 01:07:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-06-25 22:10:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-06-25 22:10:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-25 00:22:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-25 01:02:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d72026e879c35e59a50b4d24b7ec002b4a9271e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-24 23:59:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7610447c0169e8eac2d4fe8f34a42d6cb98e954f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-25 00:00:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9b0c5953b01fc340f4d87e61a740c21b95de9701","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 00:22:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb95e911dd4a407fb67aec863de9a323\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3c1242bdb97946a594ed1f38b37d787d : SUCCESS in 16m 07s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/85cca59f478a40f7a3e49534ee75d178 : SUCCESS in 16m 33s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/680fec7376c44c2c81f7f9fe22ffd733 : SUCCESS in 21m 46s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/637d83b9695543babfd353838f37a74f : SUCCESS in 17m 16s","accounts_in_message":[],"_revision_number":1},{"id":"e6a15e1efdae02070a0ba17c9c0a0a1e6323c85a","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 00:51:25.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec73c4c8c7404619a66b896e725a5950\n\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/181141e13f8640639a9ef380f3a3bbc2 : SUCCESS in 17m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b76cdb0f3197418bac00072d2a9e2baf : SUCCESS in 18m 14s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/c1d5bf04c3a14b4c83a258d0c97b6133 : SUCCESS in 16m 58s\n- swift-tox-py36-centos-8-stream https://zuul.opendev.org/t/openstack/build/0bd38e8c5f9e4cdf91c5c34d4df9f673 : RETRY_LIMIT in 1m 53s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/985c5cef2e9346b68b7d170202d3a467 : SUCCESS in 16m 59s\n- swift-tox-func-encryption-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/26830fb4b47a44d18c6557eaef71ce18 : SUCCESS in 19m 22s\n- swift-tox-func-ec-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/f27ef544cae145bd9bd61ae9e0f27138 : SUCCESS in 16m 24s\n- swift-multinode-rolling-upgrade-victoria https://zuul.opendev.org/t/openstack/build/9308e0477d984db6984aea5ba8aaad45 : RETRY_LIMIT in 2m 12s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/cb1037deb8d84f739eb4e45dcd3d77e9 : RETRY_LIMIT in 1m 42s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/37540a621dd54e45aa9875bcdb0bf210 : RETRY_LIMIT in 3m 01s\n- swift-multinode-rolling-upgrade-yoga https://zuul.opendev.org/t/openstack/build/fcd7e2eaab3b4af49ce320b3cbc22e41 : RETRY_LIMIT in 1m 45s\n- swift-multinode-rolling-upgrade-zed https://zuul.opendev.org/t/openstack/build/d5415cdfff504cc1ad16be4e0d1e0e0b : RETRY_LIMIT in 3m 04s\n- swift-multinode-rolling-upgrade-antelope https://zuul.opendev.org/t/openstack/build/c0b92021e861431da7e8baff33a646c6 : RETRY_LIMIT in 3m 08s\n- swift-multinode-rolling-upgrade-bobcat https://zuul.opendev.org/t/openstack/build/96c5343302724dc48805f627acf016e0 : RETRY_LIMIT in 2m 07s","accounts_in_message":[],"_revision_number":1},{"id":"0282a4c2aac36b903fed48893ea19e30e15436c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 01:02:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ecef6d2868d946cf916e9c58fa6083ae\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37f541f61b11428c8678e07d710b8ec6 : SUCCESS in 6m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/ee8bb163c17941b4b9fe5e9089b241c3 : SUCCESS in 40m 58s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/1b5b8ac13e2b4b749f305a121ccdbd16 : SUCCESS in 46m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2ecede8293194e9f832e690b27aeb764 : SUCCESS in 36m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/15ad25385e254a03a88e7c476b2e30a4 : SUCCESS in 1h 02m 28s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/098b9b02d35147ef8cbe19cc304fff58 : SUCCESS in 17m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/425a72d962174ffcbe2499cc2593f866 : SUCCESS in 19m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fdf29f0fdffc4be1beadd113d9fbc426 : SUCCESS in 19m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5a2ad609d32948c59ea8ac6ee56e067a : SUCCESS in 10m 45s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8980120934d04bb9bad6d372a6312a7f : SUCCESS in 5m 07s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8070ba5245734c7e8b3aa94fe087a8ed : SUCCESS in 20m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/16e5c75a661345d08a74efd2b063234d : SUCCESS in 19m 05s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8233441c89bc452fb9ccb0ee30d28aa6 : SUCCESS in 17m 00s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/350b612a2bbc4d21b109d8bd04272ca3 : SUCCESS in 16m 54s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/17d7bf5b3901435b9181833a4f494e90 : SUCCESS in 17m 41s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0bd36a9a48dc425eab219e1fef3ed39d : SUCCESS in 19m 02s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/88c8e0c88d0f48a8a5ab60ed5e833964 : SUCCESS in 23m 39s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2fe50a252f7e474ba80061d19e3f54eb : SUCCESS in 15m 30s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/69728d92650043748515562664b11349 : SUCCESS in 14m 57s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/324004a12d814e669bc62864d6d1092f : SUCCESS in 15m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fe92a9315beb4676aa5808635756ef12 : RETRY_LIMIT in 1m 53s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/32423fc61a2b43da9379e2184f2b3e1b : RETRY_LIMIT in 2m 42s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/d3dc706bd2fb415da4126ce665fde48d : RETRY_LIMIT in 2m 52s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/5235ec20606c4de7b97250a9b9530112 : RETRY_LIMIT in 1m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/042a0fd2053a40ed9c4371e4249f2592 : SUCCESS in 40m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/46eed3a2dd3a464c86d77cfc049b3af5 : SUCCESS in 39m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2f1f2c109e8a457695d3175c2b92613d : SUCCESS in 16m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53cf623b0de24436bed0403705622717 : SUCCESS in 3m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/809affccbbbc4b7dbdcd52aa0517e4ce : RETRY_LIMIT in 1m 47s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2933dc55a04d4466a49ab465cf65a29a : SUCCESS in 47m 39s","accounts_in_message":[],"_revision_number":1},{"id":"f17d3aeba99a8d89fad258c0bdafede8872edc12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-25 17:26:03.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d834c88f8fa6b697d7b3770669b127aa5e9c185e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-25 17:26:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dae853df6d4afafc40319763094affb61c31f624","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-25 18:00:57.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"c4932f9b29d9c250af4fe197c4a6d47b4d3b8334","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-25 18:30:50.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"356dfb4827d3b88d1753d0bb46c3037b0da9bd26","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 19:00:08.000000000","message":"Patch Set 3:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96d04f3b0cf3406fa16aa88233f3ba18\n\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6c4b1501463d4951bbcee0754fc1e56a : SUCCESS in 18m 47s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4c402a3f0b6d41069032c47d007b6e86 : SUCCESS in 20m 04s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/6727207ab6924292b28bafc1f7f65670 : SUCCESS in 18m 16s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/0eddc1acc91548ed93033d23147d5430 : SUCCESS in 18m 37s\n- swift-tox-func-encryption-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/e46a80c0fec349a9a24b55750eff0e3c : SUCCESS in 16m 23s\n- swift-tox-func-ec-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/07d56b8c142f4473b7bdf75877cb7ab9 : SUCCESS in 16m 48s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/37d17f7fa53e4d2f995f93988f62517f : SUCCESS in 20m 05s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/24ef9dd1ad424dbe9e4be3a9d196dbc3 : SUCCESS in 19m 06s\n- swift-multinode-rolling-upgrade-yoga https://zuul.opendev.org/t/openstack/build/984e62298a194a89a20e8db523eb9929 : SUCCESS in 19m 39s\n- swift-multinode-rolling-upgrade-zed https://zuul.opendev.org/t/openstack/build/20821e702a7542a2b0aa2228ff902df6 : RETRY_LIMIT in 6m 30s\n- swift-multinode-rolling-upgrade-antelope https://zuul.opendev.org/t/openstack/build/f570482c74b84fd48c995235eff601c1 : SUCCESS in 23m 07s\n- swift-multinode-rolling-upgrade-bobcat https://zuul.opendev.org/t/openstack/build/949ba32b82dd460fad89b2723ac058b1 : SUCCESS in 19m 40s","accounts_in_message":[],"_revision_number":3},{"id":"144847d5d0013c9331022221ee6cb7d2cb3df1e8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 19:01:38.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38c2de8548ba4e279ef3586f68860576\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5ee86f3a25a5426db02e635d2cdd2530 : SUCCESS in 16m 23s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/109fd279ef164c83bce5b968ef1ba25d : SUCCESS in 18m 51s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/533e0d0b059d4c30873be4edd73d48c6 : SUCCESS in 57m 46s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/af436e6ab7da4a01ae3f179890c05f81 : SUCCESS in 19m 00s","accounts_in_message":[],"_revision_number":3},{"id":"cc1afd0846c519c4b6bccdb38f115d1ec2a21831","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 19:37:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae664365edc04f4faa32e79ed0a38c76\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e067dcfe7ad440492f577dbf5577665 : SUCCESS in 9m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/0475cbddc7754e399d9d531cb51520c5 : SUCCESS in 1h 11m 30s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/817de6ce712240e99e25878404434d7b : SUCCESS in 1h 03m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b1a2d5bbe125422780b6e41ac530850e : SUCCESS in 50m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d8f5d8b45b1e45108c0a1aff42add82b : SUCCESS in 1h 05m 08s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/77d64bfbeb484149b43e2c75286da69f : SUCCESS in 19m 10s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/987e77c701cc4d30ab9832ba51f43623 : SUCCESS in 20m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f1b71873c36848279c7c5a49c8f53279 : SUCCESS in 19m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cc849bd35a7247a3b13b90869e8f5457 : SUCCESS in 10m 30s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/5571481694604d54a7bad40d4d78c799 : SUCCESS in 5m 25s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/5b13321a96ba45ad92482a4324da559e : SUCCESS in 17m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fd34e1878e3540ff86c105d05f290dde : SUCCESS in 17m 51s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/f187af29f2c442a3880894a6a1f8f0f6 : SUCCESS in 13m 04s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/679c164b8b354f8e81f8ff456b20a7ff : SUCCESS in 15m 32s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e3ddabeb9ea6458d973e388a5c49f97e : SUCCESS in 18m 03s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b73902f8c18643b1922504fc2e231a1c : SUCCESS in 17m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/92f860be435d4b429aeb3dbdb6617399 : SUCCESS in 19m 56s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/c08f22ba93e34b39b72b92fd0c531265 : SUCCESS in 15m 26s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/cd9bed7b00a94a08adb5106bc575c4b8 : SUCCESS in 15m 41s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/8281c27f970c47f3949d1d6ab702affc : SUCCESS in 15m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a170d3234b7b44b885b989a16a90440b : SUCCESS in 5m 36s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d5bb1d2166c342cfa54c8f3d5b4cb320 : FAILURE in 4m 36s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b2fc9939d7dc4379b83843ed761e5fca : SUCCESS in 6m 32s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b73f47d1e35d441184588a9736757573 : SUCCESS in 1h 03m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0785c888f3544514b578a4fa450f4e97 : SUCCESS in 58m 15s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4a07db06a2644f2996967faa2a26af96 : SUCCESS in 50m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9ba84e2b1b5f4133af3ace6c8b8ac2d1 : SUCCESS in 21m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/023ddea8d2f544739aeeac4acfd536e3 : SUCCESS in 5m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c5ad1741530d46b9a90e9362f45d9573 : SUCCESS in 24m 00s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7f97c9134eef45d2a72649fe198a52ba : SUCCESS in 1h 00m 30s","accounts_in_message":[],"_revision_number":3},{"id":"6169e343d1585bbba891bcd4e8d7ec5c58ea36f5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-25 22:10:25.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4b637654cd508377cbab5a3af92f2937c9a850a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 22:11:16.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"340f79333f39b7d6b85d7fd0013009d2ed6dc865","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 01:07:12.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/898d97e82f3b4c909d6a4657a74acd6b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b0086a9b8ae4e86be365d24508d4642 : SUCCESS in 9m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/d84ebde5b5b541d29ae00380ca8c56d2 : SUCCESS in 53m 59s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/c79f3d83fbb84d5c9c5c9263b106049a : SUCCESS in 50m 45s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f07ccd562f664c6abb1d7aa5faacad1a : SUCCESS in 41m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3628e0fd203c445ba2ca5bf0576fd359 : SUCCESS in 56m 07s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/8548363571fb4cb98bfb190646314238 : SUCCESS in 11m 08s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/3b42edc88f3d4920b8fe0f28a6a6a10b : SUCCESS in 5m 13s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3ed4a1ab14a84a7c9d6cb2405482b6e3 : SUCCESS in 16m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fecc5cee5dc340e1a3d239af4b033b04 : SUCCESS in 17m 35s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/85b636ebc72c42e982062ef7b178a982 : SUCCESS in 16m 27s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/c6f4a966a8824cb1814c704090b03ed3 : SUCCESS in 17m 33s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8641696fba884a8ebd8ee8b6b66ec6e1 : SUCCESS in 17m 14s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d5162e9d289f4e03afaa4df90e95e323 : SUCCESS in 18m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5ed97531629b4516b62aed2ee0702672 : SUCCESS in 19m 30s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/fe8ce0177ab749c183dad33bf42af57f : SUCCESS in 13m 58s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/76566b1cb8294b1d9b5c2acc61318e7c : SUCCESS in 15m 01s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/cab0d487bddf453987e1054e94f223bc : SUCCESS in 14m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7840173e997c43669f8c3eabf0712046 : SUCCESS in 5m 49s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/93409ffd579f46a895ef7c12b81e7d8c : SUCCESS in 6m 45s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6ffebf60b55748549093f3eb14e90c65 : SUCCESS in 52m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/225cf95705b74ccabf3d0f7f38c3ee60 : SUCCESS in 33m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b9a9c77832e84b33a36b75e2f4cb7913 : SUCCESS in 35m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/288090461a364dd6ae545b0709079965 : SUCCESS in 15m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03649af4df88400990039d3429315f26 : SUCCESS in 3m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0bfa9e3c6ca9405281d3817f6323c1e5 : SUCCESS in 39m 15s","accounts_in_message":[],"_revision_number":3},{"id":"f67ccdc4d4a81f7a5646d591685dce21999abf68","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 01:07:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"bf63d83bed1d95938ae86b0afb361afe172d1538","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 01:11:07.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/861af180533f40a3add51c837de7beee\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bc8d066efef4d80b85a17ee539e0061 : SUCCESS in 39s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/4929d700f36f4d5b84c741f84ac4bc71 : SUCCESS in 31s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/9f93ab28dbd344468e366a8bc9b816c3 : SUCCESS in 33s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"85d83978438a6bb07232704511b043dee3f8f769","revisions":{"1653f974aa9854f13e4b176f09d0d2b057c30429":{"kind":"REWORK","_number":1,"created":"2024-06-24 23:59:30.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/91/922691/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/922691/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/922691/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/922691/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/922691/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/922691/1"}}},"commit":{"parents":[{"commit":"46aabac092b2e721b5b81c5684498e4bd4f6759a","subject":"Update TOX_CONSTRAINTS_FILE for stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/46aabac092b2e721b5b81c5684498e4bd4f6759a"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-24 23:39:08.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-24 23:43:16.000000000","tz":-420},"subject":"[stable-only] CI: Remove some old rolling-upgrade jobs","message":"[stable-only] CI: Remove some old rolling-upgrade jobs\n\nThese are not actually defined for this branch.\n\nRelated-Change: I3516823fdacbe8fd3c2434c0de9dedd1d82980fe\nChange-Id: Id7f8acffe122bdd18932c523e6c1e7d042d81b03\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1653f974aa9854f13e4b176f09d0d2b057c30429"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1653f974aa9854f13e4b176f09d0d2b057c30429"}]},"branch":"refs/heads/stable/2024.1"},"d2f58f352b316eda41e8e5b131983ce7efc32fcc":{"kind":"REWORK","_number":2,"created":"2024-06-25 17:26:03.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/91/922691/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/922691/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/922691/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/922691/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/922691/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/922691/2"}}},"commit":{"parents":[{"commit":"3856cc05b20a2ef0ed7c812ef8813048a6e80931","subject":"CI: Move off CentOS 8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3856cc05b20a2ef0ed7c812ef8813048a6e80931"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-24 23:39:08.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-25 17:25:50.000000000","tz":-420},"subject":"[stable-only] CI: Remove some old rolling-upgrade jobs","message":"[stable-only] CI: Remove some old rolling-upgrade jobs\n\nThese are not actually defined for this branch.\n\nRelated-Change: I3516823fdacbe8fd3c2434c0de9dedd1d82980fe\nRelated-Change: I4f6b9c07af7bc768654f1a5d0c66b048e0f2c9c1\nChange-Id: Id7f8acffe122bdd18932c523e6c1e7d042d81b03\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d2f58f352b316eda41e8e5b131983ce7efc32fcc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d2f58f352b316eda41e8e5b131983ce7efc32fcc"}]},"branch":"refs/heads/stable/2024.1"},"85d83978438a6bb07232704511b043dee3f8f769":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-06-25 18:00:57.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/91/922691/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/91/922691/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/922691/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/922691/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/91/922691/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/91/922691/3"}}},"commit":{"parents":[{"commit":"4a89cafc18d5f2998f752e55e03f700dafb1a8fb","subject":"CI: Move off CentOS 8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4a89cafc18d5f2998f752e55e03f700dafb1a8fb"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-24 23:39:08.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-25 17:59:50.000000000","tz":-420},"subject":"[stable-only] CI: Remove some old rolling-upgrade jobs","message":"[stable-only] CI: Remove some old rolling-upgrade jobs\n\nThese are not actually defined for this branch.\n\nRelated-Change: I3516823fdacbe8fd3c2434c0de9dedd1d82980fe\nRelated-Change: I4f6b9c07af7bc768654f1a5d0c66b048e0f2c9c1\nChange-Id: Id7f8acffe122bdd18932c523e6c1e7d042d81b03\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/85d83978438a6bb07232704511b043dee3f8f769"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/85d83978438a6bb07232704511b043dee3f8f769"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
