)]}'
{"id":"openstack%2Fswift~922692","triplet_id":"openstack%2Fswift~stable%2F2023.2~Id7f8acffe122bdd18932c523e6c1e7d042d81b03","project":"openstack/swift","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-06-26 16:23:45.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Id7f8acffe122bdd18932c523e6c1e7d042d81b03","subject":"[stable-only] CI: Remove some old rolling-upgrade jobs","status":"MERGED","created":"2024-06-24 23:59:46.000000000","updated":"2024-06-26 20:53:40.000000000","submitted":"2024-06-26 20:50:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"922692","meta_rev_id":"47cd8835bf11712c08c1588550989d1825a00126","_number":922692,"virtual_id_number":922692,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-26 20:50:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-06-26 19:08:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-06-26 19:08:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-25 00:56:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-25 01:15:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e2c9f1a807da2b739d40259649f4241eb7b7e57a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-24 23:59:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"032ebacb1f332e9fc71e5a6391a6f2f847813016","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-25 00:01:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"48370b099d68d48dfb8d5c711cdd44c5b454a065","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 00:56:41.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2a624f675e14640b4ad1f9581d89e69\n\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/48743262b92b4c599d7338238b243a38 : SUCCESS in 15m 54s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/44464640627d48498798dce5fa34dd40 : SUCCESS in 14m 45s\n- swift-tox-py36-centos-8-stream https://zuul.opendev.org/t/openstack/build/227367faaa9e41d6808276f07688a913 : RETRY_LIMIT in 1m 38s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/43f8d31685624494bce66339be24602d : SUCCESS in 16m 13s\n- swift-tox-func-encryption-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/c54276ad271347f683b230b1ae0b27da : SUCCESS in 17m 33s\n- swift-tox-func-ec-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/a82031c1daa2403faa1c20276f532990 : SUCCESS in 16m 49s\n- swift-multinode-rolling-upgrade-victoria https://zuul.opendev.org/t/openstack/build/bae7e4635f954dda95d6369f54157380 : RETRY_LIMIT in 2m 01s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/8da8b726d79e49ea95dc79024f0e38d4 : RETRY_LIMIT in 2m 04s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/e289865147824e6abd157ecf05a0f1ef : RETRY_LIMIT in 1m 57s\n- swift-multinode-rolling-upgrade-yoga https://zuul.opendev.org/t/openstack/build/dd08d76a505542769c1f7fba8c7061a0 : RETRY_LIMIT in 1m 46s\n- swift-multinode-rolling-upgrade-zed https://zuul.opendev.org/t/openstack/build/b516873b0e8c4e7d8c5e424661555f89 : RETRY_LIMIT in 3m 09s\n- swift-multinode-rolling-upgrade-antelope https://zuul.opendev.org/t/openstack/build/8e36a573a5ae47568e5fd6febc943991 : RETRY_LIMIT in 1m 51s","accounts_in_message":[],"_revision_number":1},{"id":"0768828f086205e77e03472d8c774acfaa274ad5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 01:15:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e0bdcfff971340418585b8b5a196f322\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bc41fb3b5eb4ff59ee1d93160f53784 : SUCCESS in 8m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/8cf93eae5e5e47239379c94afd9cd5ab : SUCCESS in 54m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/903992330fc7462ea3bd5c48d12e78dc : SUCCESS in 52m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ed30e9ca48744b179c1b9ec0f7395dbb : SUCCESS in 1h 14m 43s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3e0ed4cf0e254bd68b020659aab9ecf2 : SUCCESS in 16m 18s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2b6eed5778a6413fa57b33803bb90939 : SUCCESS in 16m 46s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a3b068c844614b7490177a3aadd4e27d : SUCCESS in 20m 22s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3e0bb8f1836946bb9e43a5a6b2283d55 : SUCCESS in 10m 34s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/0045883fce9645348dcb001a9c087326 : SUCCESS in 4m 36s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/ae1bc5bca8ea4bf192afa4ad1e7189db : SUCCESS in 16m 49s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c2fb82fb6fa8435fb873ce0df66addb5 : SUCCESS in 17m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7ef34da650304393b4689359b08a8f17 : SUCCESS in 13m 16s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/e42fe3a2248e4acb872725e43ddc3e4e : SUCCESS in 15m 46s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/66c3806c808649bba4474160a8c42599 : SUCCESS in 15m 45s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3d00371f614f434ea707c99f67edd713 : SUCCESS in 15m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/821b000c110f492293bdfd1d0ceb1c03 : SUCCESS in 18m 22s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/aae8743496ac4b5e952f487b9ec63ba3 : SUCCESS in 16m 53s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/72b9e86d1d644ed1ab1ed5330d60004a : SUCCESS in 15m 28s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/63ada5073ee14a71a987a5bee723c7bd : SUCCESS in 15m 21s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/25907bb2e4db49d18a5ae5869d455b6a : SUCCESS in 14m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/91106e39e7a64d98bc213b1b950f0d2d : RETRY_LIMIT in 1m 51s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0f0bd01b56734c679fa65aeebbcff1ea : RETRY_LIMIT in 2m 33s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/834990d2c5b94b3ab30842e8b687f0a6 : RETRY_LIMIT in 1m 40s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/895d1cf69f9d4e95984e480c4005f44f : RETRY_LIMIT in 1m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/aa254f60f15744c58240fef4a51b53da : SUCCESS in 54m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f9820717da964f0e805718c66da3b679 : SUCCESS in 50m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3698b260c97341faa9c4f815e7f7cd23 : SUCCESS in 16m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee04c1a0f0154b21948a6df380c67db5 : SUCCESS in 5m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d80fe6de6e9a45cfb0d793a41646d88a : RETRY_LIMIT in 2m 15s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aea408d71cd74d07b65b3fc2f853f97c : SUCCESS in 55m 55s","accounts_in_message":[],"_revision_number":1},{"id":"57bd12c7cf0f73c1139aa35a0d00086f73791c26","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 01:33:18.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2508b97b466c4b9cb2384a362188887a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2f3e82ebb2ce41db926c4e17de0ec69d : TIMED_OUT in 1h 31m 13s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/822044035f6846cda3c131fb5a3f56be : SUCCESS in 15m 45s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/5d9684ebf0144d9b81ed034ceb742af2 : SUCCESS in 1h 23m 28s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/3a2414e9fdd54ca9af011b0cf51e675e : SUCCESS in 16m 46s","accounts_in_message":[],"_revision_number":1},{"id":"c28012b6d3f74fbbee4ca150d6e77a09892a2a22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-25 20:33:22.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"eff8f951443950c94cb0bc6fe9eb5585f2ebc0cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 21:34:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6cfd93549968407c8626b7ac56f6eac1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d69fdd0a5c2d4aedae6b4bb7f9d33b94 : SUCCESS in 8m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/92d46826e9d94aa4b8cc913812689156 : SUCCESS in 51m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/58af02d87ca64e7a90f1cdf035b4b11e : SUCCESS in 37m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c17764533e704e91892523f31d5f8189 : SUCCESS in 45m 32s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c10bf354fcef449e84b927f5c9fa5a5e : SUCCESS in 13m 45s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/98fa24a81e624ebea23d9592d9810671 : SUCCESS in 17m 22s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/37e1b7008ba24a18ac79813f7cde80f8 : SUCCESS in 16m 31s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a41b904993024f839453c6de8bb252cb : SUCCESS in 9m 06s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/48fb7d06ca024aa0af8a0c8259949e0a : SUCCESS in 5m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/3dcbf5085d7c4244829a56074014e466 : SUCCESS in 17m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/949cba99a2924b29aa195b6f700001da : SUCCESS in 15m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/693fcf92311b45ceba0a51db4f7ca006 : SUCCESS in 15m 13s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/8f19e69cd3ff44b6b82f1a781adf876e : SUCCESS in 15m 42s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/b6693b47b5334f6090799bf8eea0dcd9 : SUCCESS in 15m 38s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/00a51a09bd7742108722dca100235714 : SUCCESS in 16m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/3c5a4447d815439e90cce261d58b76ff : SUCCESS in 17m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/4fe983a7b21d4468a9d4db67c8d9b7f6 : SUCCESS in 16m 57s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/6285994da05c46019ed473a98b452916 : SUCCESS in 13m 35s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/67ba8ef09cac4c669da53df4f4725e75 : SUCCESS in 15m 59s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/805b568e90be44baa9fba0bed6ca1e29 : SUCCESS in 15m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/10d9657f6f0e4ddba3137a575a395245 : FAILURE in 6m 18s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f199ac9c65d7425884a23be34e059d51 : FAILURE in 5m 09s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e67b456e08414c07ae679c099f5eedd3 : SUCCESS in 6m 15s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/96dbe24c8dc14bf68712180cdd71370d : SUCCESS in 46m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7a98c70f17894f56a7ff1f2c98afced7 : SUCCESS in 53m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/983665cd55dd45d9a60d5e5bc07f80e1 : SUCCESS in 36m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/86d0d44995ff496d8c1eb570ca7e0098 : SUCCESS in 18m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25cc5e79c540498fa8e32b8f1e63f48d : SUCCESS in 4m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b350eaf33ab049f1adaade5a447e1a52 : SUCCESS in 18m 32s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/969d9d1db84d4b06a6913aa38caf33a8 : SUCCESS in 37m 49s","accounts_in_message":[],"_revision_number":2},{"id":"b84a441df8cc44b810c4787fd3c480f1024e2132","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 22:06:19.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/279dad22626e4bdeba841f4e9444da2a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9c048256dab5446481e0cdf302152ac3 : SUCCESS in 1h 12m 39s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d5c5780e0a394c85a464b9391f6d8f00 : TIMED_OUT in 1h 30m 54s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/39803abe25a147e899503b093c34413e : SUCCESS in 21m 31s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/5e0bd42aeaee45a993912bc0db10698d : SUCCESS in 17m 12s","accounts_in_message":[],"_revision_number":2},{"id":"3ca058a7d53897ed1114e11ee0ac99d515c147cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-25 22:26:40.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"31ad4934c6f35e566e34f2c52bb4aa996ffee751","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 23:27:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c82a8561c30b43348a6df738e646bf80\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/150f46e68265451c9c923d66253d2f38 : SUCCESS in 55m 06s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/12c29dd247a24ee29de75354788e4ae1 : SUCCESS in 58m 23s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/9783874f38f14efb81e2107afef4a035 : SUCCESS in 17m 27s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/cabd356a24624538a419ab301239a7ad : SUCCESS in 57m 57s","accounts_in_message":[],"_revision_number":3},{"id":"5667a5f0eda1874ef1ac2f0bad6bb5cf068ad9b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 00:31:49.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ae424b712ea6448ba8dd2ab101c798f1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25e350887e93452a964b93555806aea9 : SUCCESS in 9m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/661d939cfbd14e1eb16c9d1258d89d47 : SUCCESS in 1h 03m 53s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/116c463ce1dd44308d3e620225f94970 : SUCCESS in 53m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/26a576f156584aba9599fdc38525b7c8 : SUCCESS in 44m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8e79ff066a5e419d8e8b37c6924fddb3 : SUCCESS in 17m 24s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0a82ff97170d4fcab7b3eddad63e0226 : SUCCESS in 16m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/254faa989b3248feb4bd2d0a4219c728 : SUCCESS in 17m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6207c7f9f402419cb46b78f5a0851754 : SUCCESS in 9m 52s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8fcefdf623dd444da1b55bc269d3d822 : SUCCESS in 6m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/554461048d4140d7b67b1f61adafc16e : SUCCESS in 15m 54s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/46c20c2e398649489494c9ef3e172eef : SUCCESS in 15m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/aeb8417016374a69954ecf50cfe3ad94 : SUCCESS in 15m 34s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b60842516c7943f99d0d89ab95d286c4 : SUCCESS in 14m 35s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/b1a7434e26034a168b3e719a6ae8f9da : SUCCESS in 13m 39s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3610f942ea49431d9063e0c8da31a6d9 : SUCCESS in 16m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7a7c6019557b4c02b6dcac78a3886185 : SUCCESS in 17m 09s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/3cfdc24cce6f4ac78ba1ed09b76f809a : SUCCESS in 16m 13s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/fd677d7be1394a8f833a8ef15bfecc79 : SUCCESS in 13m 41s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/0370606bc523497c902d784d78a4d79a : SUCCESS in 14m 28s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/ea8cccc973be4b9897780cfdfedcb5c2 : SUCCESS in 13m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/eea528685bc94f959f6f26d4db77cfcb : SUCCESS in 7m 52s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/be7d7be76e404af39a3f808b6d78d879 : FAILURE in 5m 03s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/dbdfe9c9b7ae445c8186be759898de81 : SUCCESS in 5m 57s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ac9d9e8885d943ac9e6ea2f5e83198ce : TIMED_OUT in 2h 00m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a290e5c4bb41436488545a06f1195214 : SUCCESS in 47m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/242cf9de92f74256971feddcf007196e : SUCCESS in 49m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4020a148f9dd4ee09196f29bf7df2651 : SUCCESS in 17m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db1e3465c1ca48218dabe0babb31f93e : SUCCESS in 4m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8d8f6d9003f6472bb02c92eb46b97819 : SUCCESS in 16m 18s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c5095c8839c546fb9609f9e9fd9c571e : SUCCESS in 48m 17s","accounts_in_message":[],"_revision_number":3},{"id":"4c0571174eabb7df275fc062ba8f6947deaea784","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-26 02:55:06.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4249cb0833a697f6ed70bfebb7a064a9016bf0e3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 04:11:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0154a5bb84d14e488e91d289aa08b11f\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2600a5e2cfc049bb94d0836083fa1212 : SUCCESS in 17m 36s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4de3279aa4694b06b0264dea00ce77f5 : SUCCESS in 16m 24s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/43c5a386cbe0411fab4fd27cd7122269 : SUCCESS in 1h 13m 51s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/dbb90a23271b47dbb7e1fc545f04b8a4 : SUCCESS in 18m 25s","accounts_in_message":[],"_revision_number":3},{"id":"ded785592d64e1ed382c7d109eeb4c4565e44806","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 04:32:58.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8dba3a56fd714ccf9a233664a45d1592\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ba6adade9104804982f9a64fa0c21bb : SUCCESS in 8m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/196cf48fa6a54646a6bb51e6cd3469c0 : SUCCESS in 1h 01m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/266fd368508a4b719d5fd2075c5714d1 : SUCCESS in 46m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0cbfd1622c4d4e8e9114f4751a09be2e : SUCCESS in 1h 03m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/29b5f9556708401a96fc9f62e7e33775 : SUCCESS in 15m 57s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/edc34c57552845669d273880d4067232 : SUCCESS in 19m 38s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f85149c00a8c4847afd3f189663b2994 : SUCCESS in 17m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ccf3e3f98bf84e829e664bd297276b55 : SUCCESS in 10m 39s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/d010b6622351472799348bea73656ed0 : SUCCESS in 5m 54s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/178f080a213c4e95a1d3d269f0008f35 : SUCCESS in 18m 50s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d24dbeb382a34e1b81f296199ee480e8 : SUCCESS in 16m 22s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/638a874a78a34a82af76df22753ac217 : SUCCESS in 14m 41s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/16479f733db94d139037412a8e0fd1f1 : SUCCESS in 15m 33s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/58b180f3d93a4e469a9a6f5c68042c19 : SUCCESS in 16m 44s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/9e164f08fd674c9a95e3aafcf188a040 : SUCCESS in 17m 20s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/a290434a957a4492a35dac245b914b08 : SUCCESS in 17m 06s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/c1c47ca6b3bf4ff09554c6595307b1db : SUCCESS in 17m 25s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/67cb646dc61f450e811038d1732229bf : SUCCESS in 14m 34s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c17207f649b749c794ee179b13b3875c : SUCCESS in 14m 59s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d431ce3e5b514bfe86d9f30a5c4c4824 : SUCCESS in 14m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/59fb95424d8c48688327932618732a31 : SUCCESS in 6m 11s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/edec242d520f48e08d952781a9dc801e : FAILURE in 4m 58s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e745d0091c0c400ca564d42adcb3979d : SUCCESS in 8m 07s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6fa1ab1c98a548318a6ec06ff1630ea7 : FAILURE in 1h 12m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bd08c5c792ad4d56875fbfd71c61f1fc : SUCCESS in 47m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ffc2362b81a54c8ebc3ca3d02e837aae : SUCCESS in 50m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3aeb698c16ac48ab88a5e0c93b130021 : SUCCESS in 19m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5453f97d89484db68603bfb8634799b1 : SUCCESS in 5m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0dc6ed24a41e4a19b107a20a9c60219e : SUCCESS in 17m 22s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b5187c32a1554c6e8cafa2770a2b0a32 : SUCCESS in 56m 01s","accounts_in_message":[],"_revision_number":3},{"id":"003640c906e63bd5e0819942df8f1c0a7d018513","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-26 16:23:45.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"94f57ebe834acb971baedc27f01cb881fdcac5d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 17:24:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a00876bfabd2492699abc7b9fb883736\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a77fafe8d1dc4057acc69d4a049ff4c8 : SUCCESS in 7m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/22e3392e0f344ccba6a0673ffaa82b61 : SUCCESS in 53m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0503375eabbb48fa8dc5205d694680c4 : SUCCESS in 48m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bb1a849ca3624e0ba3a35d8ffc696f67 : SUCCESS in 54m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d2c225ef5c214a0eb1e029e551e13b84 : SUCCESS in 15m 51s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bd409761c176437fbd95119544b8db51 : SUCCESS in 16m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f819e53c19df46fbb3df2f85dfa92f5d : SUCCESS in 16m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2d6cc45225684374a3bc31e24024e8f4 : SUCCESS in 11m 31s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/319f580b4d5343e2b7c205637c2e045d : SUCCESS in 4m 52s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b1ea2e0e5b1e4c7ab9caef2c0a4aef2b : SUCCESS in 15m 37s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5552800a9928442a950758e77e555b1e : SUCCESS in 14m 33s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d1c1d522c86f4a2485d61390ba87aff0 : SUCCESS in 14m 25s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/fc43aefba87540639ea01641cf35d440 : SUCCESS in 17m 03s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/b0329423ff0c4149a3f3193b6ece4728 : SUCCESS in 13m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f309fa3d395f41f7886b1d2fc4cd32ae : SUCCESS in 18m 05s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/7ca1a7f8c6864ddfb9bbd171f59bea26 : SUCCESS in 15m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/7ab2ad2a162b43288db1fa0952cd4db5 : SUCCESS in 17m 10s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ef09f84c40f14cb1b1ef8c73ce3e2623 : SUCCESS in 16m 12s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/245409f0592447d4b0038ccbd6b4caaf : SUCCESS in 13m 31s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/565047f36576413bbcfe254ed787ef7d : SUCCESS in 15m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1b04144a42094ff188381c708871899e : SUCCESS in 7m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f6ea7077aaed4377b9aeb4c8357c16de : FAILURE in 4m 15s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/c301141c53f7472eae25056a210ee60b : SUCCESS in 8m 23s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/943d923e10bd4b40963f006e45514c5e : SUCCESS in 49m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c9fd9f42f4114a2ab799320bde842945 : SUCCESS in 43m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c865afba92be4eda963179319880cd6e : SUCCESS in 49m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/05075e76ea0841bd8f0a02252089b5ce : SUCCESS in 20m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e1ebcb9adeb4469bc2b11408cf4efd9 : SUCCESS in 4m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5df16be0532a48a4a618956b9bd1772b : SUCCESS in 23m 16s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ff97340a35ca47078bc9a6f8b77e5872 : SUCCESS in 54m 14s","accounts_in_message":[],"_revision_number":3},{"id":"0266eadf793d2b9815772910c7a1600762b8225d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 17:42:31.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09759852ece241df913824fc95cf1b58\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/582504b81c3442b0bdf5aad43403c554 : SUCCESS in 16m 09s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/484b2a0c386a41d29e74d6f98ec19e88 : SUCCESS in 15m 40s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4228b3106b8940459ec088bca03a9327 : SUCCESS in 20m 37s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/5f8c3176124440edab97080df5981726 : SUCCESS in 1h 16m 52s","accounts_in_message":[],"_revision_number":3},{"id":"683a0e509fb1379baafffbe1c32aa837750b36d1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-26 19:08:10.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"55ea7bbbeb9d08ba401b3165ed13233a0b3ba68b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 19:08:53.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"83f44c5ccdf9586b4f6df10969fdf96fab01d5aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 20:50:55.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4fcd5fd99b24116a5f0c7480d051b8f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a16ca8c73fd4f549668273f4ac122be : SUCCESS in 8m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/4f007541872e4361acbc1b8de4be404c : SUCCESS in 53m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7ede48a204ae4c97b045dcc62f932f74 : SUCCESS in 42m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e2f7dd12fd4d46a7a2f7a3f8c3d3328d : SUCCESS in 57m 10s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/8b2de26ad1ab44a3872ec8711e84ae19 : SUCCESS in 10m 24s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/de68ac1c8d35454fb8457d543fb4fb24 : SUCCESS in 5m 26s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8486b75db9974f658bdebbce79b0a526 : SUCCESS in 16m 18s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/27d7bf925a504cff88f96d70e7a87fc3 : SUCCESS in 15m 27s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c165f41ef0f6440b89693452168f0736 : SUCCESS in 16m 06s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d9de15425da14870b0a35e0d82274357 : SUCCESS in 16m 35s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/85f6a4f15c9c4716a66ea0f46539fbda : SUCCESS in 17m 55s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/56741e13bb4e4c349ae112239e574bff : SUCCESS in 18m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0e1a6a77eaac4e989bd8765139613421 : SUCCESS in 18m 19s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f7745b16ae4b46b0ba07b67b67ca933d : SUCCESS in 19m 08s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/fd337f552eda400ea0d0e677aeb8e76c : SUCCESS in 15m 14s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/164279b551564fae8d044c0cc0d7abe5 : SUCCESS in 16m 40s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/012f763512aa450fbd42e637d49d9c7e : SUCCESS in 15m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5d4f4998f31e4fad90cbdfb395008aa6 : SUCCESS in 6m 39s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3ed4258b8553440da63db07057cd0569 : SUCCESS in 6m 26s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/525453defcdf4cbcb196ab4c30336f2f : SUCCESS in 57m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b3e5f303b13d4895872c40432c1ef923 : SUCCESS in 57m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/eec7604335ae401f972f3ff38eb9684b : SUCCESS in 57m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/31adfe61769d480abeebf8ce88d1ad83 : SUCCESS in 16m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6bfb4049a2c4fa59a85e20eac32a0cf : SUCCESS in 4m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e08bac408cc7414a9d60942cdc3ed970 : SUCCESS in 57m 38s","accounts_in_message":[],"_revision_number":3},{"id":"4cce00c984033513bd5951140c485baf9a0ea2ab","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 20:50:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"47cd8835bf11712c08c1588550989d1825a00126","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 20:53:40.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/745df6e876c143ce8349c097b7e91db7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ba112fa7bc44005a271bb201c032176 : SUCCESS in 44s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/0c0df879dbe144c59efd549bbab0fbaa : SUCCESS in 38s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/f6d1ab440b43415b9ef11e644cf735e2 : SUCCESS in 31s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"0de976cf5636e6572660181ea196c3f1da5f6454","revisions":{"ee646dcfa5c8efb3036147720ed5128c311e9ad5":{"kind":"REWORK","_number":1,"created":"2024-06-24 23:59:46.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/92/922692/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/92/922692/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/922692/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/922692/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/922692/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/92/922692/1"}}},"commit":{"parents":[{"commit":"5e5fde12400fc76b9df43b3fcd13eb8a22665063","subject":"Merge \"CI: fix rolling-upgrade jobs\" into stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5e5fde12400fc76b9df43b3fcd13eb8a22665063"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-24 23:39:08.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-24 23:42:40.000000000","tz":-420},"subject":"[stable-only] CI: Remove some old rolling-upgrade jobs","message":"[stable-only] CI: Remove some old rolling-upgrade jobs\n\nThese are not actually defined for this branch.\n\nRelated-Change: Ibe46aecf0f4461be59eb206bfe9063cc1bfff706\nRelated-Change: I3516823fdacbe8fd3c2434c0de9dedd1d82980fe\nChange-Id: Id7f8acffe122bdd18932c523e6c1e7d042d81b03\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ee646dcfa5c8efb3036147720ed5128c311e9ad5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ee646dcfa5c8efb3036147720ed5128c311e9ad5"}]},"branch":"refs/heads/stable/2023.2"},"c94c1cdd463a38e95f5143ffcf19db2fc4333c64":{"kind":"REWORK","_number":2,"created":"2024-06-25 20:33:22.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/92/922692/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/92/922692/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/922692/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/922692/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/922692/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/92/922692/2"}}},"commit":{"parents":[{"commit":"506ca026b30ce338637e66ac5b12dd4bb425cd39","subject":"CI: Move off CentOS 8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/506ca026b30ce338637e66ac5b12dd4bb425cd39"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-24 23:39:08.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-25 20:32:45.000000000","tz":-420},"subject":"[stable-only] CI: Remove some old rolling-upgrade jobs","message":"[stable-only] CI: Remove some old rolling-upgrade jobs\n\nThese are not actually defined for this branch.\n\nRelated-Change: Ibe46aecf0f4461be59eb206bfe9063cc1bfff706\nRelated-Change: I3516823fdacbe8fd3c2434c0de9dedd1d82980fe\nRelated-Change: I4f6b9c07af7bc768654f1a5d0c66b048e0f2c9c1\nChange-Id: Id7f8acffe122bdd18932c523e6c1e7d042d81b03\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c94c1cdd463a38e95f5143ffcf19db2fc4333c64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c94c1cdd463a38e95f5143ffcf19db2fc4333c64"}]},"branch":"refs/heads/stable/2023.2"},"0de976cf5636e6572660181ea196c3f1da5f6454":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-06-25 22:26:40.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/92/922692/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/92/922692/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/922692/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/922692/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/92/922692/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/92/922692/3"}}},"commit":{"parents":[{"commit":"0ad40ea1a534aa9a870a516ebf67443e6c92abcd","subject":"CI: Move off CentOS 8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0ad40ea1a534aa9a870a516ebf67443e6c92abcd"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-24 23:39:08.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-25 22:26:34.000000000","tz":-420},"subject":"[stable-only] CI: Remove some old rolling-upgrade jobs","message":"[stable-only] CI: Remove some old rolling-upgrade jobs\n\nThese are not actually defined for this branch.\n\nRelated-Change: Ibe46aecf0f4461be59eb206bfe9063cc1bfff706\nRelated-Change: I3516823fdacbe8fd3c2434c0de9dedd1d82980fe\nRelated-Change: I4f6b9c07af7bc768654f1a5d0c66b048e0f2c9c1\nChange-Id: Id7f8acffe122bdd18932c523e6c1e7d042d81b03\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0de976cf5636e6572660181ea196c3f1da5f6454"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0de976cf5636e6572660181ea196c3f1da5f6454"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
