)]}'
{"id":"openstack%2Fswift~922693","triplet_id":"openstack%2Fswift~stable%2F2023.1~Id7f8acffe122bdd18932c523e6c1e7d042d81b03","project":"openstack/swift","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-06-26 22:12:02.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Id7f8acffe122bdd18932c523e6c1e7d042d81b03","subject":"[stable-only] CI: Remove some old rolling-upgrade jobs","status":"MERGED","created":"2024-06-24 23:59:53.000000000","updated":"2024-06-27 00:29:54.000000000","submitted":"2024-06-27 00:27:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"922693","meta_rev_id":"91653e9337496e2eeeb6dcb5462a31e16cffb707","_number":922693,"virtual_id_number":922693,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-27 00:27:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-06-26 22:13:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-06-26 22:13:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-25 00:52:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-25 01:09:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"923ddd6fa4f4abb469c818712f53076f95c40101","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-24 23:59:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2b2ae9f59d391c497bd1372921db3548880f98e4","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-25 00:01:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5f32359d395411c49782a84f0dc3f58578f8b8ad","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 00:52:52.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53f90a8e8b7240a8bd42b65996b1d33b\n\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6b00e7460b3543178b3c4041caeb6da7 : SUCCESS in 15m 13s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/b98713ad50944532bace4d6d95ec6d17 : SUCCESS in 12m 45s\n- swift-tox-py36-centos-8-stream https://zuul.opendev.org/t/openstack/build/6e39ed56ce134c19a0e6681ffc31dc85 : RETRY_LIMIT in 1m 50s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/ef5a02e4d8264c72a3321dbc767a556a : SUCCESS in 15m 53s\n- swift-tox-func-encryption-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/9615fe77b3364a85be406710c037a755 : SUCCESS in 15m 46s\n- swift-tox-func-ec-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/02791f84fb2249be9cc2f9f65c4fa496 : SUCCESS in 14m 08s\n- swift-multinode-rolling-upgrade-victoria https://zuul.opendev.org/t/openstack/build/0e8ded48fbd4449eb663485b1b811339 : RETRY_LIMIT in 1m 43s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/0a9e2829ad294a589e3d7daad79bf0fb : RETRY_LIMIT in 1m 58s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/d9c95125b715489080edb2537e803455 : RETRY_LIMIT in 1m 55s\n- swift-multinode-rolling-upgrade-yoga https://zuul.opendev.org/t/openstack/build/b477f5f07c2342e993eb850dbc68c3ec : RETRY_LIMIT in 3m 01s\n- swift-multinode-rolling-upgrade-zed https://zuul.opendev.org/t/openstack/build/e71f179d078c475d82e9a89b6845f71b : RETRY_LIMIT in 1m 58s","accounts_in_message":[],"_revision_number":1},{"id":"bfd0de8a66967c0837b2764f8edec7da9c3171b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 01:09:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/33e528870fe44eccb79e7c123e9c87c1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98381cda20bd494690904025c9ef45fd : SUCCESS in 9m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/9936582760b04ec89e93da8d3a77664b : SUCCESS in 1h 02m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4f18811a73994e33ad8c05d715a6e581 : SUCCESS in 44m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9a1f75a7589943d9967aa72ee80ee2db : SUCCESS in 56m 55s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7942de99a9544858929b6088cb1faaac : SUCCESS in 17m 45s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ef68c83f48454e3ba1aa0f6cbb265906 : SUCCESS in 18m 03s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c90ef7f752834c20a653ddfbe200b0f8 : SUCCESS in 19m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/40b37269abd745b6b0aa4566f1012711 : SUCCESS in 9m 22s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/a9c535c887504cff9fc7cec621b62250 : SUCCESS in 4m 43s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6f48b2b106df4755b097c003220fff5a : SUCCESS in 16m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/c39a49775a2d4054aded7d9540654b12 : SUCCESS in 16m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/30226a3c32174286aee1cb93ede79714 : SUCCESS in 15m 00s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/830e461aeb314e2ea1c72587ab91307c : SUCCESS in 18m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/50faaa9338354cbc82cdf07822ab6d4c : SUCCESS in 16m 48s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ff7c04bdc284473aa17c6af57b9644a5 : SUCCESS in 16m 54s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/16cf000794534e089deca4c03d1a2cb4 : SUCCESS in 17m 29s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f49424152972433aa92183637c6941a3 : SUCCESS in 15m 05s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/93bc351d0e2d483daed8937350e675f6 : SUCCESS in 14m 33s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/add630bf31b94eb484a2355425844829 : SUCCESS in 12m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fd7fc62f3156482b9130f0f278bff920 : RETRY_LIMIT in 1m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1a83be4d910a4d36a2e8211a4f22b2d2 : RETRY_LIMIT in 1m 53s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/170d9442647047e3ab875922bf83edff : RETRY_LIMIT in 1m 32s\n- swift-probetests-centos-8-stream https://zuul.opendev.org/t/openstack/build/cbf89b9537ad43dfa818a7137e6c8301 : RETRY_LIMIT in 1m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/70757cd16ead46468a779c99843a319a : SUCCESS in 49m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/75527a782e7d4064a440854ba9a033ae : SUCCESS in 49m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8cd74c01c30045b19ca1ac9a65066dbd : SUCCESS in 17m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98df2d33b86b474297b719219eea99d2 : SUCCESS in 4m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/611011f369a04c03a308fe2b3a2f85ed : RETRY_LIMIT in 2m 04s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b8c021ab80a84399800f5bc9fa0a3571 : SUCCESS in 39m 11s","accounts_in_message":[],"_revision_number":1},{"id":"8ddbb23a9375929ac21be62afcf00944b0f3dc92","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 01:25:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8c81023a6df43b6b06d71fe46e0a462\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/debde8273a1849d3b91b7c29f77c5e86 : SUCCESS in 1h 23m 25s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/18e580e92c744d2fac60913625098fb0 : SUCCESS in 1h 00m 13s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/74c57560f859430db5a71cd45ddb1cac : SUCCESS in 18m 17s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/373f1d513ece4ecc81ed42c30c966352 : SUCCESS in 1h 08m 04s","accounts_in_message":[],"_revision_number":1},{"id":"7bd06d08c64ffb0eedf66dcd8fa794d44ff3379f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-26 19:54:15.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ec5dcb6781b075ab384811bd399eed679c279906","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 20:56:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c8e37be526040bcbd1838fcc1bbc705\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1104ea9c6b3c44428f5ba304b21bac0c : SUCCESS in 14m 40s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1aa7c57e5bf84515815b988b8ee08932 : SUCCESS in 15m 10s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/25c065d170f74a20932157b128c9a74c : SUCCESS in 17m 28s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a448a33db3ae4b0bb8d6c6e5470e83c8 : SUCCESS in 1h 00m 25s","accounts_in_message":[],"_revision_number":2},{"id":"f25b695a2c08f0ae06de65605f873207a9fbf8cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 20:57:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0d48bba52b34ba7bf43c6ae8272866d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/125061546e3a4beeadc3d27538a8f682 : SUCCESS in 7m 14s\n- grenade https://zuul.opendev.org/t/openstack/build/b438c5a0d4744c91b804143ffb890c4b : SUCCESS in 55m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/73c2ce35663c4f238748c1272331763a : SUCCESS in 42m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3f33e56bdea34c6897c9076f6f1e3904 : SUCCESS in 1h 01m 02s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cc32d27ab5f3484698d2dd9fbd9bc6bd : SUCCESS in 17m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c3f681b66e7148399a19dc976e2e3df3 : SUCCESS in 16m 55s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/eaebf7391d7c47838d1e08be384e4328 : SUCCESS in 19m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f16375c724e047bbb410c5bc4a4c0ad1 : SUCCESS in 10m 05s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/8295f615e96e407b99703cf4b7787e0c : SUCCESS in 5m 16s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/894124017d034892baba85db21ecf418 : SUCCESS in 14m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e8564e8e2c594f1d9a11eea35076226f : SUCCESS in 14m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/cc03a8b911004a89be48359d02f1a8b2 : SUCCESS in 13m 23s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/9074bccf240e440dac0a4c4e4bca6d3b : SUCCESS in 16m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/792fe7cfdae045a9ae509cac813b9e95 : SUCCESS in 15m 18s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/215d5020eab040ca8ab99dd8f0f0908d : SUCCESS in 18m 33s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/a7f810c94a704b368bc31adff99f92b4 : SUCCESS in 16m 00s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/fe6a4ba395ed40a999a3a60535a01cbc : SUCCESS in 13m 33s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c68ab81ea9904fc28007747a45417a2b : SUCCESS in 13m 41s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4ed65902ed4b43da93aaa3bd52279a89 : SUCCESS in 15m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/575636de89ca49d4b4379c4c07c98a90 : SUCCESS in 6m 04s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c08d44ce2d1f4e879e71fec6afbd170c : FAILURE in 4m 27s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/07c1f913960844fbb2b1ee4ec5a26d14 : SUCCESS in 7m 17s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8baca902a1634fba8f4c881add264eeb : SUCCESS in 55m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/bd91fec9f63d4f1e985a82953dab9fc1 : SUCCESS in 44m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/82d7b0f8a0c9412aa02b7ddaa4f9d0fc : SUCCESS in 50m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/943386d9a316480287bd15c3fbb7696d : SUCCESS in 16m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e8bcb94753a45cd916d743dea1b8bcd : SUCCESS in 3m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e76e9d017cb948d6a22118e372440802 : SUCCESS in 24m 42s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7e7b8c6298bf4318af21910e8be83251 : SUCCESS in 46m 57s","accounts_in_message":[],"_revision_number":2},{"id":"592a043668f61ebd7bdb840f9b20581a8eb00cbc","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-26 22:12:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8b1ef701303fb1bff6e083f8d504a54a4633b812","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-26 22:13:45.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"43cea921ec816f0553327f17c04537a7f2c83064","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 22:14:35.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3455d7074fc07a30561f5d6ef1b1f6c196c73723","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 23:12:29.000000000","message":"Patch Set 2:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36d005ea1d3944a3a6bcce1cd6fa5383\n\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5ff79e416fd64138aecc3de47e8b0a94 : SUCCESS in 16m 20s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/e86d7bc1ada94383b1265f4f2999b3ba : SUCCESS in 12m 26s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/4879c56aa8804139af07db1b124b5a8d : SUCCESS in 16m 01s\n- swift-tox-func-encryption-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/ab0ea37b73564692924c95497d1b36ba : SUCCESS in 14m 03s\n- swift-tox-func-ec-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/fe9d07015a4e41b0b1cdeb50e59fa8f6 : SUCCESS in 16m 21s\n- swift-multinode-rolling-upgrade-wallaby https://zuul.opendev.org/t/openstack/build/1cbfb8342e0b4e87b870daa0c2525ea7 : RETRY_LIMIT in 4m 51s\n- swift-multinode-rolling-upgrade-xena https://zuul.opendev.org/t/openstack/build/1b6ebefdb078485fa023ef48d141551e : RETRY_LIMIT in 5m 00s\n- swift-multinode-rolling-upgrade-yoga https://zuul.opendev.org/t/openstack/build/14ace038c4954c8fb869706a74afdc1b : SUCCESS in 17m 14s\n- swift-multinode-rolling-upgrade-zed https://zuul.opendev.org/t/openstack/build/1ea8768d927444ca87a9b950ec2b762f : RETRY_LIMIT in 5m 01s","accounts_in_message":[],"_revision_number":2},{"id":"7429017097fd375fa2596c09ebd69dc85a0e77de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 00:27:47.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef714d0d8109486fa61970ea267493e8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6323f0b02c9347648686bfbdea209794 : SUCCESS in 9m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/c108b26eaf6f42b2a53e3009a06a71c3 : SUCCESS in 56m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0323fc307d3a49d4a7c0c668920a7d84 : SUCCESS in 45m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d9c03c563e4c4335a8b1434397063dad : SUCCESS in 57m 17s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/805318c2a7674317bbe4fe2bca699d06 : SUCCESS in 11m 21s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/8902da6bb7ff4bf79de82d13deb1ccd3 : SUCCESS in 5m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/bea3790308ff45999435125926d547b6 : SUCCESS in 16m 19s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/09db9a3478ed4da9a8a8603aaf583f8d : SUCCESS in 14m 36s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1fc0722c56094616a1d5fac8caf19c0e : SUCCESS in 14m 47s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/a94fbe46f4cb40ae849329b19a04ef76 : SUCCESS in 15m 00s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/14500958b86c430d839d515553624627 : SUCCESS in 15m 24s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/f6ad88ee8909452d8a9151a63e9fa8c5 : SUCCESS in 13m 52s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/90967b4dfb6e4838ae4d8470216e0650 : SUCCESS in 16m 47s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/60d57887fc7140968e933adff4d537df : SUCCESS in 14m 51s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/25986cb0c1ad443e969fc0a96ff6cba8 : SUCCESS in 14m 59s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/a7db09dbca264048bd1c24f252b1de15 : SUCCESS in 13m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9a884be102494133ab677b84039ea116 : SUCCESS in 6m 03s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/8967b6b72a664223b209160082be0c62 : SUCCESS in 8m 16s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2e8cf73061f14a74963bff7ea8020514 : SUCCESS in 49m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/743c7d032bb14985b02a4565722d43d9 : SUCCESS in 51m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a93f0af509114adea8bdfbd11a0f0905 : SUCCESS in 50m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/34b6331a09bb47c29335893450693a49 : SUCCESS in 14m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6a45d88968a439ebf53cbe9abded04a : SUCCESS in 4m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3a8052a34f3a4ab6a661f6b76765e9e3 : SUCCESS in 59m 20s","accounts_in_message":[],"_revision_number":2},{"id":"e6de2170702c35fbb3497d28e63676bbadef6cc9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 00:27:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"91653e9337496e2eeeb6dcb5462a31e16cffb707","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 00:29:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/866abf7b0b2245be98890466e2ffed4c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c33c3bbc17d544b3b9ddfe142c85afff : SUCCESS in 48s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/cbc0335851994211a7b0204b42ded068 : SUCCESS in 32s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/b0712d3ef9a94106ac28511ca2abce1e : SUCCESS in 36s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"27d66a63f3bb85df5a110e86ae6a286770d902eb","revisions":{"7b8e3c77c7c1dd587ce7248daca22c101bb8fe09":{"kind":"REWORK","_number":1,"created":"2024-06-24 23:59:53.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/93/922693/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/93/922693/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/922693/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/922693/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/922693/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/93/922693/1"}}},"commit":{"parents":[{"commit":"575d148bab7b49d8ad955d7c773f6c7e26e72b70","subject":"Merge \"CI: Remove centos-7 jobs\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/575d148bab7b49d8ad955d7c773f6c7e26e72b70"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-24 23:39:08.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-24 23:42:00.000000000","tz":-420},"subject":"[stable-only] CI: Remove some old rolling-upgrade jobs","message":"[stable-only] CI: Remove some old rolling-upgrade jobs\n\nThese are not actually defined for this branch.\n\nRelated-Change: Ibe46aecf0f4461be59eb206bfe9063cc1bfff706\nRelated-Change: I3516823fdacbe8fd3c2434c0de9dedd1d82980fe\nChange-Id: Id7f8acffe122bdd18932c523e6c1e7d042d81b03\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7b8e3c77c7c1dd587ce7248daca22c101bb8fe09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7b8e3c77c7c1dd587ce7248daca22c101bb8fe09"}]},"branch":"refs/heads/stable/2023.1"},"27d66a63f3bb85df5a110e86ae6a286770d902eb":{"kind":"REWORK","_number":2,"created":"2024-06-26 19:54:15.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/93/922693/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/93/922693/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/922693/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/922693/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/922693/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/93/922693/2"}}},"commit":{"parents":[{"commit":"2032fb214eeb41cbb30a0c71525d0db9e4899760","subject":"CI: Move off CentOS 8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2032fb214eeb41cbb30a0c71525d0db9e4899760"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-24 23:39:08.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-26 19:53:51.000000000","tz":-420},"subject":"[stable-only] CI: Remove some old rolling-upgrade jobs","message":"[stable-only] CI: Remove some old rolling-upgrade jobs\n\nThese are not actually defined for this branch.\n\nRelated-Change: Ibe46aecf0f4461be59eb206bfe9063cc1bfff706\nRelated-Change: I3516823fdacbe8fd3c2434c0de9dedd1d82980fe\nRelated-Change: I4f6b9c07af7bc768654f1a5d0c66b048e0f2c9c1\nChange-Id: Id7f8acffe122bdd18932c523e6c1e7d042d81b03\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/27d66a63f3bb85df5a110e86ae6a286770d902eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/27d66a63f3bb85df5a110e86ae6a286770d902eb"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
