)]}'
{"id":"openstack%2Fswift~922715","triplet_id":"openstack%2Fswift~master~Id904f4cc0f911f9e6e53d4b3ad7f98aacd2b335d","project":"openstack/swift","branch":"master","topic":"p-proxy-failed-get-warning","attention_set":{},"removed_from_attention_set":{"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2024-06-26 18:10:34.000000000","reason":"Change was submitted"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-06-26 18:10:34.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-06-26 16:36:46.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Id904f4cc0f911f9e6e53d4b3ad7f98aacd2b335d","subject":"proxy-server: log correct path when listing parsing fails","status":"MERGED","created":"2024-06-25 10:14:06.000000000","updated":"2024-06-26 18:11:54.000000000","submitted":"2024-06-26 18:10:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":6,"has_review_started":true,"submission_id":"922715-p-proxy-failed-get-warning","meta_rev_id":"efd694819b711fea2e01e31d24648dc65b60f734","_number":922715,"virtual_id_number":922715,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-26 18:10:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-06-26 16:36:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-06-26 16:36:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-25 10:37:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-25 11:33:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-25 17:49:08.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2024-06-25 20:15:28.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"REVIEWER"}],"messages":[{"id":"6ce4f88e3e7df990e2146b59d10ae413d0d9181c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-06-25 10:14:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9c7624d7094610ab21e35af8720bb3572fcc32d6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 10:37:29.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27d810032054471a9ce4eba692809d16\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6bd8f5f7ba7b4f31a555086d4be01d0e : SUCCESS in 15m 22s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/51f97ea49bdb48838acdd4ef47f08645 : FAILURE in 2m 50s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b03084b5b60346b0bbcf9273f2a59654 : SUCCESS in 22m 21s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/94b031cc235849fdab62b14af714560e : SUCCESS in 18m 26s","accounts_in_message":[],"_revision_number":1},{"id":"4cd6f8e951c1542db996bd830763a2a744f44bd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 11:33:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4d62371f41fb424985f6f465f4bad94d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/806efa0e1fdc482cafd2f4cd0751edeb : SUCCESS in 9m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/b9949d16effe4f6d8874f6903bf682ab : SUCCESS in 1h 01m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1103549fc7404fb18e3960391ac09b73 : SUCCESS in 57m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/72cc50c4a0fd4f729fd5c31390f554c2 : SUCCESS in 1h 06m 43s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8b07b9eaa8d1467db857c821aeb784cd : SUCCESS in 16m 52s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/df0c61662ab548c297a71a40086b54cf : SUCCESS in 18m 35s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3f77215bbf4041458e929bfd7e34f5a4 : SUCCESS in 22m 38s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b5e95b227cd242058d7e61796b9ae57f : SUCCESS in 11m 50s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/58106635a2b54bfeb6c80464ac24ee1e : SUCCESS in 5m 09s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/a5f51160799f42a690877e0819a2c278 : FAILURE in 21m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5a3a1d2b8e194a709b98a1b11d5b371e : SUCCESS in 18m 19s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/23573b71855246cebec3a2ff3d1df5e0 : SUCCESS in 16m 52s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/aa1dc6ba1f944fb693723126b022762a : SUCCESS in 18m 19s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/eee43baaeae24c94b23c5a210d169d78 : SUCCESS in 24m 19s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8ed25b34904e4f63a3863f5895e6ffdf : SUCCESS in 15m 36s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0c6e9c366a18477cbc3db4e0add1960b : SUCCESS in 14m 16s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/84c3c0887bcf4b8880167c74417d6571 : SUCCESS in 16m 46s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/25d5ea1db4d8465aae94f145606b43bb : SUCCESS in 13m 26s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/c8431faaf4a04757bf5b871e68c2b14b : SUCCESS in 13m 44s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/16187c444fa74e33ae4aa5d88c437505 : SUCCESS in 16m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6200473eacc34f17a5ee87873402cddf : SUCCESS in 6m 28s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c0b27f7999ac433199b36fd5b62257fd : FAILURE in 6m 23s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/efcc2f0e45b54764ac404cb2e62f450f : SUCCESS in 9m 51s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d36dbe673ec34338ac1cfeac401aeab8 : SUCCESS in 1h 03m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b65f54c853ff4ca48bdb85db09e10b98 : SUCCESS in 52m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2f7839d1a67245dfb7c9964306850cf5 : SUCCESS in 52m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/30234d08677a4534bc2fe682d743bb96 : FAILURE in 19m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f36054a3954a4f8480e4d4e05a97d9ad : SUCCESS in 5m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f1bdce0d80e84b5596e9c560f75c6627 : SUCCESS in 24m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c4cbeef8cec74bcb8759cff3cf2191e3 : SUCCESS in 1h 03m 15s","accounts_in_message":[],"_revision_number":1},{"id":"43280da07653db16e0c1b8b7fb701aa4c873e86f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-06-25 13:38:33.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"76ef7aa4f8666327a4e9187721013c853969589e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 14:39:50.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61894de28ecd47b9828afeac561588a1\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b705997a1fa34c1aa993a7ce6600396e : SUCCESS in 17m 59s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a18720bfa7ae4674b2e1c27c1883ec27 : SUCCESS in 59m 19s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d9724d1b3f564a61a6fbe40995d03403 : SUCCESS in 23m 29s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/d7722982ef8c42b6acb694f3f7093818 : SUCCESS in 45m 54s","accounts_in_message":[],"_revision_number":2},{"id":"993650c6e8d0458841a621ea966cbbcc8a459d43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 15:02:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/abc7c63cc2b94103a57511358fa8aa91\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe0c8b6f05a04747bfa712e5b33d903f : SUCCESS in 10m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/f90cdb328a934eae81e9017f80435fa4 : SUCCESS in 53m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/bbb6f05e859047cebbaf51d85312a1a8 : SUCCESS in 54m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/065422c667a545ec836ab8d939c7636c : FAILURE in 1h 09m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/26d0e73765064e8695e2c27ff3b0c40f : SUCCESS in 17m 33s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0b0b672b9ab14168a32217150c978315 : SUCCESS in 18m 22s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1cf1ecef93684a27ab75ce7a0133654e : SUCCESS in 19m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4d2174f0767946e5be1dbf09818e0142 : SUCCESS in 10m 09s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/94e5df10f8844603a9845f4c9a8e8354 : SUCCESS in 5m 24s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/daa494640da24aad89653c8382b0fe76 : SUCCESS in 19m 25s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bb831198f3ff49179dc23c907e64f3b7 : SUCCESS in 15m 13s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d937b78cd9654a0ca45cc2d77237d6d9 : SUCCESS in 17m 19s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/340dcd972bfb4e368c3276ed67e8bdd7 : SUCCESS in 19m 08s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f20d4f2405954550b5fe8196c1019c6a : SUCCESS in 22m 03s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e1343d90add3463fa5ae2648b9aa7677 : SUCCESS in 17m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fa9d17c5e3d54736a4fe623da8b2881d : SUCCESS in 19m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/1a4bd4a1b3e547bd8dfa6fafa85e11c9 : SUCCESS in 16m 03s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/ce42810b7bdd4e26b76987968a1a982e : SUCCESS in 16m 17s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ce524ae001c043608aa06044812eadb2 : SUCCESS in 15m 31s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5fdc4975d0f946ad8ff4cf9d955cc6d3 : SUCCESS in 16m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/003c51a5590b46e8bb90595e78a31d04 : SUCCESS in 8m 00s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9000774e66c540a5a7ffd5557d170c29 : FAILURE in 7m 20s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/25aab962f7d94ebca6b667061fdfa156 : SUCCESS in 9m 01s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0db643eab98c4f26ae12ea13d18dfc4a : SUCCESS in 1h 02m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6f883d2b2fb8418fbb176cbcf78232ab : SUCCESS in 49m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/599a007137fa4366b353239eb75e8ea2 : SUCCESS in 52m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cbd118af865043eaa0c93d807e183f71 : SUCCESS in 20m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a12ccd6dad99430b8f75d765e13f0220 : SUCCESS in 5m 34s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/40e6f0ba64404365ab719a782014dd9a : SUCCESS in 23m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6addf65857a04af8bca98a3f5012c2ff : SUCCESS in 1h 17m 18s","accounts_in_message":[],"_revision_number":2},{"id":"42f2039663a1f9b5e2bd9266408abbaab4bdbbdf","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-06-25 15:09:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"807b21cab50ec6e4514ada5565a16e980cf71e31","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 15:31:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/699bfa5e69e345ab8276e91c4812d97a\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/43151034a06440bd9b0401af0466618e : SUCCESS in 16m 54s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ad34cf9f0c0e42b68668ddb88801850b : SUCCESS in 16m 42s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/853242acd51e42a99a37ad941822801c : SUCCESS in 21m 17s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/a4df87d3fe67428394771662ebda0e54 : SUCCESS in 17m 47s","accounts_in_message":[],"_revision_number":2},{"id":"6d2e7e3a791481516d8f69ab9860f2a14ed0d077","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 16:20:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98b3f01db9fa4a8493e0e22db4af1848\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ca227745f314d549dd15af533f49065 : SUCCESS in 7m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/4eb128bb66944d629832ad7bf705bdf7 : SUCCESS in 1h 05m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6a42c711a69c42fab9f8e0ff20bdd984 : SUCCESS in 57m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6318e5d63fda49c08cbeaa7883a24cf6 : SUCCESS in 1h 04m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/63f86ddf69724b9789e83b62c4551c34 : SUCCESS in 18m 41s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ddc08b815dc2494e98a3680e04787d1d : SUCCESS in 21m 52s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b2ce444a64ba4cfcbfe5e609efe81a47 : SUCCESS in 20m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/edd26b8ecbec4a48a42f28944f550d08 : SUCCESS in 10m 00s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/155e0113195e44bda471c355c29d00a8 : SUCCESS in 5m 33s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/755169bdc7374ecf883e800782124f8e : SUCCESS in 17m 37s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/740525f5828d449ab1120ce8e49292be : SUCCESS in 14m 55s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/45d08e61c418426ebd1d50ed4721f900 : SUCCESS in 15m 32s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/51049ebbae0b42a0a09b92d24683c554 : SUCCESS in 16m 21s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1a0b7c94efd146de8719f2bde8a64e13 : SUCCESS in 22m 36s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/2dba6487e7ee4d2eaaa0543d18efdf2b : SUCCESS in 17m 21s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e1ff522f66b6448991f3031696bd7155 : SUCCESS in 18m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5a939e522dc84dc986fe62b61dd3e706 : SUCCESS in 19m 04s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/bb3d4659eba64de9a3e4a4f293bd1069 : SUCCESS in 15m 32s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5bab326ac351492f9557a3e94d58d27d : SUCCESS in 15m 17s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/3db25b890b9e46b48e6ff4cce5c5748b : SUCCESS in 17m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f9b160a999384a34b636fd40a10588a1 : SUCCESS in 7m 30s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a93501936de04b33a4013111d1866500 : FAILURE in 7m 02s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f38e0a232fb745ec939aee4b1d027bcc : SUCCESS in 6m 19s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6aaab6e9b0a54fc88a7b1d4e4a06f04a : SUCCESS in 58m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4c63a9f6498f42d1b3cc24a4b41da234 : SUCCESS in 54m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/dc91149e33954f99b1ad89e7264e65df : SUCCESS in 57m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/38635348ab024917bf2711bc3d7a4850 : SUCCESS in 18m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b122adb59d5453ca601edb4d6175a79 : SUCCESS in 5m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a639d2e0a1694e8ba36510689180e1c8 : SUCCESS in 17m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/69da44ab80c04e73a756a27e10cd1a2f : SUCCESS in 57m 52s","accounts_in_message":[],"_revision_number":2},{"id":"0a275a72269e0a12e177361ac581e570a02e3765","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-25 17:49:08.000000000","message":"Patch Set 2: Code-Review+2\n\n(7 comments)","accounts_in_message":[],"_revision_number":2},{"id":"6abb70200bbb4de25cac831a6ff5e641483d4bf0","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2024-06-25 20:15:28.000000000","message":"Patch Set 2: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"56219d333a3a0b3eb372764de74a73a05c9ab950","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-06-26 11:04:08.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"19cf31aba8cca63c7ef7f49130ec8201baca22b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 12:13:11.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2bee6da043249f4b80af580f52c614e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dc12ad41a86470fadb72e2aa6a7a315 : SUCCESS in 8m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/7967e6bbc6204be5953b88a77bc3c2ca : SUCCESS in 1h 01m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/38bde5b9558c45108907ea2464c42f68 : SUCCESS in 51m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/097bab90e6fd463baef9f67377842061 : SUCCESS in 1h 03m 18s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e425f43b25c344019f52be0e954c1d1d : SUCCESS in 18m 50s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c83af2be499b4118b6ce35e1dc02e96f : SUCCESS in 19m 45s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/08b9012825b14b8f9fd52da7c58a2ecc : SUCCESS in 20m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/cfa7a8a36d5249668ed8330a438b9026 : SUCCESS in 10m 44s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/f491991ed38240a49a1a0f90597076dd : SUCCESS in 4m 51s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4ef7f6af888b490a8cb0bb6ca8f3f49c : SUCCESS in 19m 46s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b114d01daaf44214af0e07c45fdd521b : SUCCESS in 21m 21s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/8f22c2f5dcc2488a8bdd4071f3fed58a : SUCCESS in 17m 29s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/55a905fd0e434836ad59d1ffa11cfb2d : SUCCESS in 19m 33s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/01fb583884cc42c08ae14fbaa2356eed : SUCCESS in 23m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/531ac0c347c3498c9e64b63bee933cac : SUCCESS in 17m 06s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0c8f003c78674537a2d5a4322d2bd317 : SUCCESS in 18m 29s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/ff021d7cea524fbc9f733357a29885f4 : SUCCESS in 20m 31s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/47a566fe6e4d48e08bd2cdae8059ea33 : SUCCESS in 15m 43s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8e005d3f48964f05b155d1c68b111895 : SUCCESS in 16m 55s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/798ef685368148d892ab669349a66e24 : SUCCESS in 16m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6ce0fde093904bda9c7d061c7a759e38 : SUCCESS in 6m 23s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/148290c9dd0e4801a97ae4e16b625691 : FAILURE in 4m 53s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5711a53089e044299e2f4230b3187a70 : SUCCESS in 8m 11s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/17fb3828a5e140d8b1df684ddeeb4899 : SUCCESS in 47m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2763300fd8aa451398613b98d6a7e57c : SUCCESS in 46m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a0eded02d7ab4bec94b5d6a160741b16 : SUCCESS in 38m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c3ab0950be3c4e5fba3ebdb765bf9894 : SUCCESS in 16m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/330c4cc620d141a9934d5dc2a57a83c0 : SUCCESS in 4m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4f0f748c0266456f9705d052187dcbee : SUCCESS in 17m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/284f2db4d4fe4c9ea1a104ecd4b5e25f : SUCCESS in 53m 26s","accounts_in_message":[],"_revision_number":3},{"id":"6ee5325aaf8021117a6146267948dd762554f5d6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 12:36:58.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec8601a6b8414d8599fa73a1611a1521\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c84e44e5c1da49df8b6e394020061984 : SUCCESS in 18m 28s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/099dc682ad68452b8e13a648db9b7bae : SUCCESS in 17m 44s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/f13ad63cacc1455f96d67fa8239a6d1c : SUCCESS in 24m 20s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/1847b77a771d440aa0eb8f7ea3cc4622 : TIMED_OUT in 1h 31m 07s","accounts_in_message":[],"_revision_number":3},{"id":"7271a10ca7f4cf19ff7657ee61a9819bb60181b3","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-26 16:36:46.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"faa4b179a7fbfe68fdc4b8bc9d467cdd63d71517","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 16:37:20.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5ff1e022788018a710a3826cd1f0fd4c34e386be","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 18:10:34.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f6d3e3a8a8e47cfad3d93432e274461\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccbc1a231a7b43e1baed84b061821fd9 : SUCCESS in 8m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/1727fcf3043a40209e8bcbaad6572931 : SUCCESS in 53m 34s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c724de8b8dee4b8ab3e5f3c4f7517ef6 : SUCCESS in 39m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ebaacd1adfeb4bca9b594ccade668f93 : SUCCESS in 58m 34s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/0b6bb7bf7e5d43c6926d5af36a87afe7 : SUCCESS in 10m 57s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/4f24c2ee2d30456aa0fd10029a844deb : SUCCESS in 5m 10s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b7e9d70bb9ab471284e3e52163160adc : SUCCESS in 19m 53s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5eee0f51eaa249e5a0bf6bdd3be8adea : SUCCESS in 17m 39s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/ad9569d867b843a1997ddd31485afa1b : SUCCESS in 21m 00s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/f50baec6c44c4e4cbd4ee76420a46922 : SUCCESS in 13m 51s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a271e7d6644b45239eaddfa2222d44c9 : SUCCESS in 24m 49s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d3379cc85e6440e8b4d43835f3f07395 : SUCCESS in 19m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/5ccb39725c8c42e7ab1c0f5a8763f93d : SUCCESS in 17m 40s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f0dfb2b783e24f9e86e7bc1d2d13e0c2 : SUCCESS in 18m 08s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/2409c4ec16b6441f873587e938832b47 : SUCCESS in 16m 51s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/77c88a5489af4836a56382672e745890 : SUCCESS in 16m 36s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4be43ae2918549e685ab1a980157e029 : SUCCESS in 17m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9c692cacbed14b7fbd1df9229bf1ceaf : SUCCESS in 7m 27s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/856cffb4634a4ca69cb0ac5e79cdf6a7 : SUCCESS in 6m 50s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/cb787a095a764e429eae91ae6aa3d62a : SUCCESS in 58m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fc2ddd652a88459eb2595a56a0d84bb8 : SUCCESS in 51m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d40182a70ea046e092d7b30c216e666b : SUCCESS in 54m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/39e14c5c58934cc98b7904753da49d69 : SUCCESS in 21m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/858af7ae7c2347b393edab39b98fa012 : SUCCESS in 5m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c1e4036f799e46a69dc34b7aeee78a66 : SUCCESS in 16m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1c53479c1f1f46e1a0f9695d941048aa : SUCCESS in 52m 12s","accounts_in_message":[],"_revision_number":3},{"id":"ebc4c3614eae821af354040a0f090f6100c340f4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 18:10:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"efd694819b711fea2e01e31d24648dc65b60f734","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 18:11:54.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3fedf0663a5441f88cc560e2d6157d4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a570d21d100f455fb3e56317b2f0db32 : SUCCESS in 58s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/e29b4842bcda431ba37888200815e9f6 : SUCCESS in 37s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/20a4ae4211f540faaeda766a8e8908f0 : SUCCESS in 35s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"130377bd271543708141eefc9c0d5afbf19e501d","revisions":{"078af95d2c039784cb4c79bb91f7beb4f2dc347e":{"kind":"REWORK","_number":1,"created":"2024-06-25 10:14:06.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/15/922715/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/15/922715/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/922715/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/922715/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/922715/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/15/922715/1"}}},"commit":{"parents":[{"commit":"d1a7bc6afda6c73627e3221f1be4468151a02e1c","subject":"Merge \"s3api: include error response reason in log_info\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d1a7bc6afda6c73627e3221f1be4468151a02e1c"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-06-25 09:45:22.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-06-25 10:08:15.000000000","tz":60},"subject":"proxy-server: log correct path when listing parsing fails","message":"proxy-server: log correct path when listing parsing fails\n\nThe proxy may use a sub-request to fetch a container listing response,\nfor example for updating shard ranges. If it then fails to parse the\nresponse, it should use the sub-request path when logging a warning or\nerror. Before, the proxy would use the original request path which may\nhave been to an object.\n\nChange-Id: Id904f4cc0f911f9e6e53d4b3ad7f98aacd2b335d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/078af95d2c039784cb4c79bb91f7beb4f2dc347e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/078af95d2c039784cb4c79bb91f7beb4f2dc347e"}]},"branch":"refs/heads/master"},"fff2b527353bea0d150ef20dec944ba59bb23904":{"kind":"REWORK","_number":2,"created":"2024-06-25 13:38:33.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/15/922715/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/15/922715/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/922715/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/922715/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/922715/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/15/922715/2"}}},"commit":{"parents":[{"commit":"d1a7bc6afda6c73627e3221f1be4468151a02e1c","subject":"Merge \"s3api: include error response reason in log_info\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d1a7bc6afda6c73627e3221f1be4468151a02e1c"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-06-25 09:45:22.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-06-25 13:38:24.000000000","tz":60},"subject":"proxy-server: log correct path when listing parsing fails","message":"proxy-server: log correct path when listing parsing fails\n\nThe proxy may use a sub-request to fetch a container listing response,\nfor example for updating shard ranges. If it then fails to parse the\nresponse, it should use the sub-request path when logging a warning or\nerror. Before, the proxy would use the original request path which may\nhave been to an object.\n\nChange-Id: Id904f4cc0f911f9e6e53d4b3ad7f98aacd2b335d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fff2b527353bea0d150ef20dec944ba59bb23904"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fff2b527353bea0d150ef20dec944ba59bb23904"}]},"branch":"refs/heads/master"},"130377bd271543708141eefc9c0d5afbf19e501d":{"kind":"REWORK","_number":3,"created":"2024-06-26 11:04:08.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/15/922715/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/15/922715/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/922715/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/922715/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/922715/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/15/922715/3"}}},"commit":{"parents":[{"commit":"d1a7bc6afda6c73627e3221f1be4468151a02e1c","subject":"Merge \"s3api: include error response reason in log_info\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d1a7bc6afda6c73627e3221f1be4468151a02e1c"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-06-25 09:45:22.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-06-26 10:23:18.000000000","tz":60},"subject":"proxy-server: log correct path when listing parsing fails","message":"proxy-server: log correct path when listing parsing fails\n\nThe proxy may use a sub-request to fetch a container listing response,\nfor example for updating shard ranges. If it then fails to parse the\nresponse, it should use the sub-request path when logging a warning or\nerror. Before, the proxy would use the original request path which may\nhave been to an object.\n\nChange-Id: Id904f4cc0f911f9e6e53d4b3ad7f98aacd2b335d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/130377bd271543708141eefc9c0d5afbf19e501d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/130377bd271543708141eefc9c0d5afbf19e501d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
