)]}'
{"id":"openstack%2Fswift~922747","triplet_id":"openstack%2Fswift~master~I7cccf3b8cdfe157c27fb8e0b1f432a649970a865","project":"openstack/swift","branch":"master","topic":"tmp_labeled-metrics","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2024-06-25 20:37:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7cccf3b8cdfe157c27fb8e0b1f432a649970a865","subject":"trivial: assert SendtoCalls consistently","status":"MERGED","created":"2024-06-25 15:35:12.000000000","updated":"2024-06-25 20:39:59.000000000","submitted":"2024-06-25 20:37:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"922747-tmp_labeled-metrics","meta_rev_id":"22830ad2fe45acf7f58a44c8cd1ec7c03d18ce04","_number":922747,"virtual_id_number":922747,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-25 20:37:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2024-06-25 15:43:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2024-06-25 15:43:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-25 15:43:37.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2024-06-25 15:58:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-25 16:40:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4db00a4eeae098949e77a4d75dbb1f0d990a834e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2024-06-25 15:35:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d8d19702270fd3e1379fd9aceab4654323f44cb7","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-06-25 15:43:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7a3e53a8c3f384f084e3fda84908e2d8c8e70f5f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 15:58:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f96c75faed9c4fa9bd955d889c5a9a89\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8fb1bd2d67e3430e979a65f55796a254 : SUCCESS in 20m 51s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b955b9922f0a4c40bd577a00a8241eea : SUCCESS in 16m 58s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/b92f07bc7a7244e5b0bd093b78b8de46 : SUCCESS in 21m 59s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/98cc02f71c214ee4afef1e916a708620 : SUCCESS in 18m 26s","accounts_in_message":[],"_revision_number":1},{"id":"5e0f40f4d915e5c84f5900a3807bfe5549f561a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 16:40:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba17ba9c259047ffbb2a9fe2363a8553\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b81ce5eb7c648a3bb1879094f013249 : SUCCESS in 7m 38s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e6e52e0916a2466999c8d2c3b864165c : SUCCESS in 16m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d5064e70df314407994e0aa91072fa52 : SUCCESS in 18m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/97225169e77d4665a061de3f015abcac : SUCCESS in 21m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9508b56c77164f10ba975ae13aa8dda1 : SUCCESS in 10m 04s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/12b12072a7e44f6a9b8a5b10851c154a : SUCCESS in 5m 18s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8a1ebd632498404e8545bfc2cebe63a8 : SUCCESS in 19m 38s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9dbeefec39ca4bd9a0a9d55c4b57b663 : SUCCESS in 18m 41s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/48158d2909674821abf48543f0c1e60f : SUCCESS in 15m 16s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/08dae4e54bd8487fadd622b761bd9c39 : SUCCESS in 14m 38s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/27913878d1cd44e2a3cab24a98452233 : SUCCESS in 25m 07s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8388cb59bcab4ee38e757aceeb32ef43 : SUCCESS in 16m 50s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/d2a6b071a47e4b33a210e861f2080e43 : SUCCESS in 15m 43s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/0e619cf1dd504d56be7a315f4d17d37e : SUCCESS in 16m 51s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/01a42cc615ea459587319a1d3eb92e7a : SUCCESS in 15m 33s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ec77a766bfc34e768c3b5c117abcf7f1 : SUCCESS in 14m 02s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5ca8f0253f5748c5ada7d1d8729da4b0 : SUCCESS in 14m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0b674d3a52f545a3a25f8ff0a5087a3c : SUCCESS in 51m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1eec99d336c44cb8b38e1e0b2951ad74 : SUCCESS in 45m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/989c38f879af4b8892a1a10a1712e819 : SUCCESS in 18m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83fddaf1df764cb5bc35e323b38ae545 : SUCCESS in 3m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bbb127f3162d4549835b656423eb0c88 : SUCCESS in 17m 23s","accounts_in_message":[],"_revision_number":1},{"id":"53b8a3468c93e495469dde74c0df182da768f631","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 16:40:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"84218e3bf3d94c3e7e3deb1417edc3f32e248dee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 20:37:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74b3e3f7a2d1478eb82cd0d97176fea6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6dc2ead221c41a9b9a42e9123d7e9b0 : SUCCESS in 9m 12s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/b25e00affbbb4fa9a118edb80d85c434 : SUCCESS in 11m 29s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/9c9538e7fb584caa9b9d13d21a315c47 : SUCCESS in 5m 23s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/89055bb56acb4c1f98acbbe7ac25ec2c : SUCCESS in 18m 06s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d932ef3eca8a4f2396e7030f75968f12 : SUCCESS in 21m 16s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/88dea615b77042bbbe205750a29153c9 : SUCCESS in 18m 27s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/c5da3ed51b4b48fab4780d215de762bc : SUCCESS in 14m 32s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f0686001a74f4b649f89ad41d87540c0 : SUCCESS in 23m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/3f0090fa26a94836be36bb981454967e : SUCCESS in 19m 04s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/e03f5c600f184c5887c3011d30f5a95e : SUCCESS in 19m 26s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8d375936b5ba4e2fa7d7e7a3b2fc98c2 : SUCCESS in 19m 16s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7717eaf7e5e74407ad2fc803fc99184f : SUCCESS in 15m 24s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/d306e0cef03643779790ec7055cd30c6 : SUCCESS in 16m 47s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/5ad15e0745714c568989fdd68ece1eee : SUCCESS in 16m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f6fe6bacb05c422589efe65899044d22 : SUCCESS in 6m 44s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/fa0eec861a7c41b0b4168b519ec29d08 : SUCCESS in 6m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0e1cc5565cff40028b0f8717a164d2b5 : SUCCESS in 46m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f82e132f42fc4d24a2b080e5e5cfec24 : SUCCESS in 50m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/49290b3bfc2c44fdaa8db59faa4de095 : SUCCESS in 21m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e379b7fc3194a53a926de7ba5c7541e : SUCCESS in 5m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/935abb6186064692874859da23da74cd : SUCCESS in 17m 25s","accounts_in_message":[],"_revision_number":1},{"id":"cc3678bf30f46d8dea9293be0c53464b908eea9b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 20:37:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"22830ad2fe45acf7f58a44c8cd1ec7c03d18ce04","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 20:39:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/578421a66033475cb0e80b5bc13f1d64\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/051dbcf1383949e7ac651d70ecc51f2a : SUCCESS in 55s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/debbf84a895944718d62017f686d3426 : SUCCESS in 31s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/b6c5ff78fc10425db8a5e22c378f3426 : SUCCESS in 32s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a7b69628d57370831998f4bd34df7f4ae6752e00","revisions":{"a7b69628d57370831998f4bd34df7f4ae6752e00":{"kind":"REWORK","_number":1,"created":"2024-06-25 15:35:12.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/47/922747/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/47/922747/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/922747/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/922747/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/47/922747/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/47/922747/1"}}},"commit":{"parents":[{"commit":"a21d078009e2750b8a513dd9f832723c3b51dbe0","subject":"tests: Check proxy-logging stats on the wire","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a21d078009e2750b8a513dd9f832723c3b51dbe0"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-06-25 15:29:19.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2024-06-25 15:29:19.000000000","tz":60},"subject":"trivial: assert SendtoCalls consistently","message":"trivial: assert SendtoCalls consistently\n\nUse same assertion pattern as introduced in the Related-Change and\ninclude the host and port in the assertion.\n\nChange-Id: I7cccf3b8cdfe157c27fb8e0b1f432a649970a865\nRelated-Change: I9a016de1010dd17e91bc85877413b34d2c3287c7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a7b69628d57370831998f4bd34df7f4ae6752e00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a7b69628d57370831998f4bd34df7f4ae6752e00"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
