)]}'
{"id":"openstack%2Fswift~922766","triplet_id":"openstack%2Fswift~master~I13d1867a0b1bc093fda562c82950285ea2543ab3","project":"openstack/swift","branch":"master","topic":"tmp0","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2024-07-15 21:45:08.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I13d1867a0b1bc093fda562c82950285ea2543ab3","subject":"extract build_line from send","status":"ABANDONED","created":"2024-06-25 18:33:04.000000000","updated":"2024-07-15 21:45:08.000000000","total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"74df4ac3f208eea16e9bbda189e01e88bbec5d4d","_number":922766,"virtual_id_number":922766,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-06-25 22:47:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2024-06-25 21:46:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-25 19:39:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-25 19:40:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a4c7b339a5a45dd711c0ddbaf92bd4e0a5bae9dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-06-25 18:33:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"28ce7a27540569cc5115e875d78e3ea2c53dc0dd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 19:39:45.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93a58059896444ce8106240378bb39b2\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3a87c0c1ca7e43b586876ac125260aee : FAILURE in 16m 16s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a7c93d94dee64f09b994fa43ff57f7a2 : FAILURE in 15m 56s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/81b94be160fe41b6a0d8afe44bc1d51d : SUCCESS in 22m 57s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/93453fc86b794b10ae27e8375b2dd097 : SUCCESS in 1h 05m 03s","accounts_in_message":[],"_revision_number":1},{"id":"875eee7e30024d1492fde7794d989169fb144240","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 19:40:48.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4fcf887fda644c768cfdca3267c02e3d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f2abd5aeff34bd287fd1f637cef5f5e : SUCCESS in 8m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/c07e885f8219481584ee0a3eeb94b4a4 : SUCCESS in 1h 01m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/834e125f599e497198ccb7641669a485 : SUCCESS in 48m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/245750ac849b4f9b9a4ca541c6739d94 : SUCCESS in 57m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7494a6ec4b564532b024097875478a66 : SUCCESS in 18m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9f5e35d67ae140f88b2f953ba692ef14 : SUCCESS in 18m 13s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8268795a49dd455e8a08e9d06228a73f : SUCCESS in 23m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a732f22b6ce84fc7b2a8e8e2cb2ff322 : SUCCESS in 10m 32s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/53357392c6b1437c96736ab9856eb2d0 : SUCCESS in 4m 59s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8b805df60b4d409489a41816cc853c79 : FAILURE in 20m 01s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6b6986621b944184a212806f4390b376 : FAILURE in 20m 08s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/6ee5be70b640410c9cee2ed210037444 : FAILURE in 18m 42s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/f912f0d4dafd422aa870b3565d297c19 : FAILURE in 20m 55s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d30d76a014f541238544d459cbd8ab3b : FAILURE in 21m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e432b682bfff46f2a4a58f182926c1f0 : SUCCESS in 17m 52s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0cd57e0f043f4fc1a3245975140be967 : SUCCESS in 17m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/cff5832482224e3ebf07a6251bde5c83 : SUCCESS in 19m 48s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/d9660810662d4ac3992798d582034a0f : SUCCESS in 17m 08s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/25dc9bb804824884821481a9117db89b : SUCCESS in 16m 06s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1bcd3d263f044c428508da82dd087623 : SUCCESS in 15m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e8bf1cfaf27a4633af144e2ae8204d53 : SUCCESS in 7m 47s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0c10dede446a431dbd3ab1d96104c97c : FAILURE in 4m 31s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/dc35909b3ec44d0393efd191d017cfd1 : SUCCESS in 8m 13s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ffd200cd6a014db49f4634b43b6953f4 : SUCCESS in 55m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1467102ebf64457ba3bd3f7f8e2f3e8e : SUCCESS in 38m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8d1ea5c760bc4a959c8628bd0990e1a3 : SUCCESS in 40m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dedc2a32a21149b8a1beda6170e6044e : FAILURE in 19m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/428b314973d244d689505d9aa86bcefa : FAILURE in 4m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b58a847432cb4368822ab88d9ca9da1d : SUCCESS in 17m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b16b4ce613674a5d91482dc87d39a24f : SUCCESS in 49m 02s","accounts_in_message":[],"_revision_number":1},{"id":"485a83c807dcd4e85d7da0c5b3641429342e64de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-06-25 21:37:43.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"3ab4c094236a0a3a4e206ccf7460e44a7b3ac00e","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-06-25 21:46:04.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"00d620c8effac4bc09271842c4b508bc45d74baa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 21:59:05.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b17aef80d7f940eca6ca4a7ea130a2d4\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c2907c02bcb94c3fbe6a553c0f5b0878 : FAILURE in 16m 20s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/902afedcaf1f4364bf1f4bd94c21e68f : FAILURE in 15m 12s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d6e8891499f54c7eb4dd4b43ed041b1b : SUCCESS in 20m 03s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/b85d217015f2497ab7a6db4779eb7284 : SUCCESS in 20m 15s","accounts_in_message":[],"_revision_number":2},{"id":"730ef25cef05fb24e395f5bd1414a996272dcfdd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 22:47:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a29d877139b9484388bdbc84f60c8e1a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89d3e6d9e0204528ada165d943d5a330 : SUCCESS in 8m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/01895f85fe74483f9a6a196599e3b864 : SUCCESS in 1h 04m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/996e9d48c5e94503bb00ae8534a557d5 : SUCCESS in 51m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/29948b058eb14dbcbca4b1e16d05acd8 : SUCCESS in 59m 54s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8a27d62e8eb34369ae924572373b48b7 : SUCCESS in 18m 11s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c60cfff0a5654a14a1556a0c3fe15e76 : SUCCESS in 19m 06s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7158267a501e4e1aa766699b5433e642 : SUCCESS in 18m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c03d5d294254404fa66eb47c06e5cb66 : SUCCESS in 10m 41s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/927b865e0f44412b9c32ebc7c0c25c04 : SUCCESS in 4m 57s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e7064fbf244a478194ce81a2d6f9a3c9 : FAILURE in 19m 33s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/144c49e4fc624b3f9a43469f605281da : FAILURE in 19m 40s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/19c2bef36e49435bbd35a33017645251 : FAILURE in 18m 20s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/61e0798252b647c786b417b3af237edc : FAILURE in 17m 47s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c2afea9615cd48a3b932f6378291aaa8 : FAILURE in 17m 16s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/e4a4ed042f7b494aaccba008699caaa3 : SUCCESS in 17m 50s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/0ecaf7c3cb9d488fb293b1ff1830dc76 : SUCCESS in 15m 30s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b8ff31e20bed4144a3734c39ce1878fe : SUCCESS in 18m 50s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/f2f5788471f3499893a2db88fa2b4fca : SUCCESS in 13m 45s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e26828b09480434191503804db4f682a : SUCCESS in 16m 25s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/f96b88d379cc46df80b4321dbd4c8cf9 : SUCCESS in 13m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/24762b19fe9e4a25858ac8fb213a25d3 : SUCCESS in 6m 13s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/58b55d863d2b46b8945ec154ff9bb422 : FAILURE in 4m 39s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/610c70e58fa747309a8dbdd63a870448 : SUCCESS in 7m 01s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9b99965545e3403dabff1ffee3551b2e : SUCCESS in 55m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/482ae92103824bd2aeb06e460e652c23 : SUCCESS in 48m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5049198f3c2c4a58b78dd2c5ff1c55b5 : SUCCESS in 32m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b4c1723587124742861294782573ee7b : FAILURE in 14m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68b4d982ffdd484f8aaa952e74bef563 : SUCCESS in 4m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/25832e14501e4fffad537205b37fae3d : SUCCESS in 18m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/646cc8e3509a47769e0e3adab2b8d6c3 : SUCCESS in 59m 33s","accounts_in_message":[],"_revision_number":2},{"id":"74df4ac3f208eea16e9bbda189e01e88bbec5d4d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2024-07-15 21:45:08.000000000","message":"Abandoned\n\nI don\u0027t see any comments from Yan to the effect, but it seems like this was squashed 909882: stats: API for native labeled metrics | https://review.opendev.org/c/openstack/swift/+/909882","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3fecc31285a7578793d116fac5945493ced4d3c9","revisions":{"2b7eb0b2de6d0108ec6b6dc0f7daaa1032abb837":{"kind":"REWORK","_number":1,"created":"2024-06-25 18:33:04.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/66/922766/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/922766/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/922766/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/922766/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/922766/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/922766/1"}}},"commit":{"parents":[{"commit":"e2217f43ea43179d3edd0e513ac669c62a613865","subject":"stats: API for native labeled metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e2217f43ea43179d3edd0e513ac669c62a613865"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-25 18:32:30.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-25 18:32:30.000000000","tz":-300},"subject":"extract build_line from send","message":"extract build_line from send\n\nChange-Id: I13d1867a0b1bc093fda562c82950285ea2543ab3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2b7eb0b2de6d0108ec6b6dc0f7daaa1032abb837"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2b7eb0b2de6d0108ec6b6dc0f7daaa1032abb837"}]},"branch":"refs/heads/master"},"3fecc31285a7578793d116fac5945493ced4d3c9":{"kind":"REWORK","_number":2,"created":"2024-06-25 21:37:43.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/66/922766/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/922766/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/922766/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/922766/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/922766/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/922766/2"}}},"commit":{"parents":[{"commit":"e2217f43ea43179d3edd0e513ac669c62a613865","subject":"stats: API for native labeled metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e2217f43ea43179d3edd0e513ac669c62a613865"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-25 18:32:30.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-06-25 20:35:28.000000000","tz":-300},"subject":"extract build_line from send","message":"extract build_line from send\n\nand add some tests for labels\n\nChange-Id: I13d1867a0b1bc093fda562c82950285ea2543ab3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3fecc31285a7578793d116fac5945493ced4d3c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3fecc31285a7578793d116fac5945493ced4d3c9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
