)]}'
{"id":"openstack%2Fswift~922774","triplet_id":"openstack%2Fswift~stable%2F2023.2~I4f6b9c07af7bc768654f1a5d0c66b048e0f2c9c1","project":"openstack/swift","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2024-06-26 02:57:03.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I4f6b9c07af7bc768654f1a5d0c66b048e0f2c9c1","subject":"CI: Move off CentOS 8","status":"MERGED","created":"2024-06-25 20:33:22.000000000","updated":"2024-06-26 04:01:51.000000000","submitted":"2024-06-26 03:59:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"922774","meta_rev_id":"ff1c7ef4a1568cb91534e7c6171d56c029c918d5","_number":922774,"virtual_id_number":922774,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-06-26 03:59:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-06-26 02:57:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-06-26 02:57:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-25 20:54:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-25 21:46:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"26b94bd95a7fdb0c135adeab716f9b503d9350a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-25 20:33:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dd640aca96ff3f34a61c7d041b90588c475cca75","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 20:54:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5cc675bf87e4d2bb5968fa41882aa97\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/6673e3a64a7e4ad6a2e3f62187253238 : SUCCESS in 15m 10s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/75866ae68f9b437b8e6fbfd032d9d856 : SUCCESS in 16m 47s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/6da0564d3f80471abf35b8ccaf23fb68 : SUCCESS in 17m 23s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7a5775f378914e44842b014387980b1f : SUCCESS in 19m 12s","accounts_in_message":[],"_revision_number":1},{"id":"1f84787a6ad191b2975c73e80fed51e1e0ad6aac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 21:46:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/93b047cde4df4bfa9993a599d687bc47\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e59d4c838a4f415a93b003d802847a94 : SUCCESS in 10m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/44f0af5323a94f959be33691a727df1a : SUCCESS in 53m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b1c5819650ab4fb5a8f1e11d20f46b21 : SUCCESS in 46m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fd55c9d97d0f422ba8b47ef3f6375774 : SUCCESS in 53m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0751fd3dbf58442d85fbd3b3e206f294 : SUCCESS in 18m 34s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/18e86eb0f33d42dd900cfd6520130a99 : SUCCESS in 15m 05s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6fa1a0ece94b4f5faeba6cbc67de6f16 : SUCCESS in 20m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c1a7e82c853049118002de64bc9ffafb : SUCCESS in 10m 25s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/e84f0616898c49f697a94986fa045886 : SUCCESS in 4m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/4b9c134f2e4e49ffa2f461e8bcb19a17 : SUCCESS in 15m 47s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a2f64ee8d9464433baec19f7deba256c : SUCCESS in 12m 13s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/cd3fbd87d568481992e06584859bdbd9 : SUCCESS in 15m 02s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/d87ad049cb654892a0a3b6bb0e8b86cf : SUCCESS in 16m 14s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/2c9c5cd467cc4ee0b5e4c752a6a4b685 : SUCCESS in 15m 47s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/64a3aee419304ba7aeaca05d402c1e65 : SUCCESS in 16m 15s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/fd9adce6dc0240bb90dc8b0318121343 : SUCCESS in 15m 20s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/8603b108a66b4f06826d25c060eff46f : SUCCESS in 16m 13s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/8c5d640ca60a4e1f919d77d823244442 : SUCCESS in 14m 02s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/47631eac70e340b0a3d819dade6f5e62 : SUCCESS in 14m 55s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/d08e4cfe77774cc9a6a45a293a9a9ce8 : SUCCESS in 14m 15s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7f841d16a78e47009d0a42b2cf919e0a : FAILURE in 5m 17s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/30281697325a45898a1167dee33a8963 : FAILURE in 4m 50s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f44760e61e7c49f89d46271327b3c652 : SUCCESS in 6m 25s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/222715b9d78c4db38bdfcab5ead4fee9 : SUCCESS in 58m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/874f823c98ad4125b9dd4feafe6bb0e4 : SUCCESS in 58m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/495bc926bf1d43c18bac21509d6c751b : SUCCESS in 52m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f69c0db088a64577a7b2adcb2e75f7ae : SUCCESS in 17m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cb306f4537041e3b6605176a28c6135 : SUCCESS in 5m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0faa7111269448678d691ad8dd6b7a72 : SUCCESS in 21m 42s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b80a3f08eaca4c86afaab09f304fd6b0 : SUCCESS in 38m 50s","accounts_in_message":[],"_revision_number":1},{"id":"303d3696d319ab9eaa156a2537c1c159f185b2da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-25 22:26:40.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"8975da8db81b088dd2959068cdd30a54d86f1dcf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 22:47:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d68bd15d59cb4ed69c55915fabaef186\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/809450c98cd24ecbaa665766e5ff9fb2 : SUCCESS in 16m 07s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/53a1101dc6274c4ea86e04d960e7e0cc : SUCCESS in 15m 23s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d0d8031956f047f8892403d1be84ffce : SUCCESS in 19m 28s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/238b0eaad44c42978768f2e2b5081b93 : SUCCESS in 19m 17s","accounts_in_message":[],"_revision_number":2},{"id":"aecab2ae2b779bd63b9472108faea44448ba5714","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-25 23:37:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b5d2562ae1746cdaf61c142bd6b5a15\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74ce6f1d743c433790204cdc8beb95ca : SUCCESS in 9m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/091b1f9cf3ca43dbaffdfa0ec1318211 : SUCCESS in 1h 02m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6ac7b79f3a45476ba2c1edfb3aab2f72 : SUCCESS in 36m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c62a9e59c9e5466cb7df4f24a89d79b3 : SUCCESS in 57m 58s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f1731b8633e04c8aa0f33e41df049ad5 : SUCCESS in 18m 08s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d4e5ab6f1dff4e0cb0c0e295976e8537 : SUCCESS in 17m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/762f02100bee438184b90cb61eb38c2f : SUCCESS in 18m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/08932d845ff243839a260a3b2a824516 : SUCCESS in 11m 08s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/2a6e02eb25414b9a99c4f8e145897ef1 : SUCCESS in 4m 53s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/37c2dd531a654100acba84d390b2d1f8 : SUCCESS in 17m 12s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ec2534918ea14c4786caf8049e97a39a : SUCCESS in 15m 31s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7a52f34c3b40426ab787fb9bdb01c697 : SUCCESS in 15m 34s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/adf1ce769a1546cdaea97228058e74d9 : SUCCESS in 12m 10s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/167fdab12bca4ba9951998be4aab03a5 : SUCCESS in 17m 15s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/368430ebb5c84f428a0c14a64a656890 : SUCCESS in 17m 51s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/88491eba03f14135b0b86ac57cc9892c : SUCCESS in 16m 53s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/345d13010ce446af8ffb06bb3bb7a1a6 : SUCCESS in 15m 46s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/fe48cbb2aa454586b688152f5b87331c : SUCCESS in 15m 28s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/b9472c2d9e5143f78aae74a44cc7db4c : SUCCESS in 14m 17s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/2001060890904ee3ae5bd48dd65c4ca9 : SUCCESS in 15m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1b07425efab44f5db7aac83ef30d0efc : SUCCESS in 6m 03s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8647688030bb4bb9b219a54ff1ffa786 : FAILURE in 6m 44s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f2d07510930a4bcea3f0b07b89e68a5e : SUCCESS in 6m 34s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8f1a1f20a4234e419b230ff3398d3ab4 : SUCCESS in 53m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/471ac68d74744f46b2d64af08bc7d1e5 : SUCCESS in 39m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9811d2c10af14d0ab42380907e9d65b7 : SUCCESS in 56m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/351b256afb8d4dcba203850745dc7af5 : SUCCESS in 16m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d05f41b15584ea2ae1a6f4eed229549 : SUCCESS in 5m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/49937efb9f8c4081ae4a4e64976e33b5 : SUCCESS in 16m 45s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fcfa4fa68847473cb425b0a4d12f0f52 : SUCCESS in 1h 00m 39s","accounts_in_message":[],"_revision_number":2},{"id":"e1f5681abf7313caaccc2ff13cb39702d7d27a57","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-06-26 02:57:03.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"79a3e1f0767c279c44037a1b5f9d83ee2e0ab331","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 02:57:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"aa2f5dc2ef4c537eae0e1ae4bf4c9b78c31c4435","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 03:59:39.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6fa01bd98f64e4489282b0b4fe540dd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36589cef5b104321b6aa6e5c5d0799d4 : SUCCESS in 7m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/674702947bac45319dc2efdf1c5187a5 : SUCCESS in 56m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c52708612b024cec897d2a4e2c7ed00d : SUCCESS in 46m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/92b439930eee4fb1a3f41ed168fbe640 : SUCCESS in 49m 43s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/ace5afe7289d467e93d0635f137d2729 : SUCCESS in 11m 00s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/6aca574b6b2a483c93a2e9b70f0ef91d : SUCCESS in 7m 39s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/6f0fc8e5ac514760a4e5fd7913d27d03 : SUCCESS in 16m 57s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6838190a168843d7a838dc08c33f997d : SUCCESS in 17m 59s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/082db03cf6dd495b9a20fa0bcaaf2d79 : SUCCESS in 14m 45s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/912a0fa1c5074e958e6947332a576144 : SUCCESS in 14m 22s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/13ead835d2434018abc3eb566dc67883 : SUCCESS in 16m 12s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/21245050603a41ce90392247e9e8377c : SUCCESS in 18m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/2e07724476f344869195f2a143aad8d5 : SUCCESS in 18m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/6a4ce30bd0e34bb0ad7da32551139a02 : SUCCESS in 20m 36s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a6001a7c52544d1a90625449187b527a : SUCCESS in 12m 53s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1e4f93a572934c8094877c240fae7182 : SUCCESS in 15m 05s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/cee864a844ae47a2b232c8791ec8668b : SUCCESS in 15m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/708c064a487c4943885f35c87cce8dcf : SUCCESS in 5m 48s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/241f62770d144aa280c0a28d8cac8727 : SUCCESS in 6m 24s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3fb5b8fbb2384719afb598226d616b50 : SUCCESS in 52m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/26e10d45edb14348942821268133518d : SUCCESS in 48m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bbdf064d611a41309e98780deadc237f : SUCCESS in 50m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c7d4961075824d6192283a91cc356958 : SUCCESS in 15m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0792e5a56a547f58473ffddd74cebcd : SUCCESS in 4m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dcf17aadb91e4b2c9e09daa62887f6bc : SUCCESS in 48m 34s","accounts_in_message":[],"_revision_number":2},{"id":"1d9dca95cadc2cebd1450f1df2ba02db85c7499c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 03:59:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ff1c7ef4a1568cb91534e7c6171d56c029c918d5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 04:01:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fdca131293504301a603255f0bf77798\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba6c150009464680b153f5413edf320c : SUCCESS in 44s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/8fdeb0c0fd2942b68ec2cf4b51715ce7 : SUCCESS in 30s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/73cccaad4e1e426bba226f0b5d84ce56 : SUCCESS in 30s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0ad40ea1a534aa9a870a516ebf67443e6c92abcd","revisions":{"506ca026b30ce338637e66ac5b12dd4bb425cd39":{"kind":"REWORK","_number":1,"created":"2024-06-25 20:33:22.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/74/922774/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/74/922774/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/922774/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/922774/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/922774/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/74/922774/1"}}},"commit":{"parents":[{"commit":"5e5fde12400fc76b9df43b3fcd13eb8a22665063","subject":"Merge \"CI: fix rolling-upgrade jobs\" into stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5e5fde12400fc76b9df43b3fcd13eb8a22665063"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-04 15:58:19.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-25 20:31:43.000000000","tz":-420},"subject":"CI: Move off CentOS 8","message":"CI: Move off CentOS 8\n\nThis is a combination of 2 commits.\n\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\nCI: Move probe tests to centos 9 stream\n\nPin selenium to 3.x for now, until we can run down the issues with 4.x\n\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\nCI: Move off CentOS 8\n\nRemove swift-tox-py36-centos-8-stream job entirely.\n\nMove the following jobs to CentOS 9:\n\n  - swift-tox-func-s3api-ceph-s3tests-tempauth\n  - swift-tox-func-s3api-tests-tempauth\n  - swift-multinode-rolling-upgrade, as well as the other rolling\n    upgrade jobs\n\nRemove the swift-multinode-rolling-upgrade-victoria job, as py39\nsupport (required for CentOS 9) was not added until wallaby.\n\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\nRelated-Change: I596415d17f77f48a6e8a63a61b734a8ca0865847\nChange-Id: I4f6b9c07af7bc768654f1a5d0c66b048e0f2c9c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/506ca026b30ce338637e66ac5b12dd4bb425cd39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/506ca026b30ce338637e66ac5b12dd4bb425cd39"}]},"branch":"refs/heads/stable/2023.2"},"0ad40ea1a534aa9a870a516ebf67443e6c92abcd":{"kind":"REWORK","_number":2,"created":"2024-06-25 22:26:40.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/74/922774/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/74/922774/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/922774/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/922774/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/74/922774/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/74/922774/2"}}},"commit":{"parents":[{"commit":"5e5fde12400fc76b9df43b3fcd13eb8a22665063","subject":"Merge \"CI: fix rolling-upgrade jobs\" into stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5e5fde12400fc76b9df43b3fcd13eb8a22665063"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-04 15:58:19.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2024-06-25 22:26:34.000000000","tz":-420},"subject":"CI: Move off CentOS 8","message":"CI: Move off CentOS 8\n\nThis is a combination of 2 commits.\n\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\nCI: Move probe tests to centos 9 stream\n\nPin selenium to 3.x for now, until we can run down the issues with 4.x\n\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\nCI: Move off CentOS 8\n\nRemove swift-tox-py36-centos-8-stream job entirely.\n\nMove the following jobs to CentOS 9:\n\n  - swift-tox-func-s3api-ceph-s3tests-tempauth\n  - swift-tox-func-s3api-tests-tempauth\n  - swift-multinode-rolling-upgrade, as well as the other rolling\n    upgrade jobs\n\nRemove the swift-multinode-rolling-upgrade-victoria job, as py39\nsupport (required for CentOS 9) was not added until wallaby.\n\n\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\u003d\n\nRelated-Change: I596415d17f77f48a6e8a63a61b734a8ca0865847\nChange-Id: I4f6b9c07af7bc768654f1a5d0c66b048e0f2c9c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0ad40ea1a534aa9a870a516ebf67443e6c92abcd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0ad40ea1a534aa9a870a516ebf67443e6c92abcd"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
