)]}'
{"id":"openstack%2Fswift~922833","triplet_id":"openstack%2Fswift~master~I8e3f58017883cf5f81d3d1d9835750b75aecd081","project":"openstack/swift","branch":"master","topic":"recon-reaper-dump","attention_set":{"37040":{"account":{"_account_id":37040,"name":"Md Azmain Adib Pahlowan","display_name":"Azmain Adib","email":"adib1905@gmail.com","username":"aadib"},"last_update":"2024-06-27 10:50:33.000000000","reason":"A robot voted negatively on a label"},"37085":{"account":{"_account_id":37085,"name":"Daanish Khan","display_name":"Daanish Khan","email":"daanishk@andrew.cmu.edu","username":"daanishk"},"last_update":"2024-07-04 11:30:58.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2024-07-19 06:46:41.000000000","reason":"\u003cGERRIT_ACCOUNT_7233\u003e replied on the change","reason_account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}},"hashtags":[],"change_id":"I8e3f58017883cf5f81d3d1d9835750b75aecd081","subject":"Periodic reaper recon dump showing current progress","status":"NEW","created":"2024-06-26 14:41:38.000000000","updated":"2024-07-25 00:04:52.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":18,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"4bd261bc545b16680004bf0e78741f307014b931","_number":922833,"virtual_id_number":922833,"owner":{"_account_id":37040,"name":"Md Azmain Adib Pahlowan","display_name":"Azmain Adib","email":"adib1905@gmail.com","username":"aadib"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":37040,"name":"Md Azmain Adib Pahlowan","display_name":"Azmain Adib","email":"adib1905@gmail.com","username":"aadib"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-07-23 13:43:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":37040,"name":"Md Azmain Adib Pahlowan","display_name":"Azmain Adib","email":"adib1905@gmail.com","username":"aadib"},{"value":-1,"date":"2024-07-24 07:09:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":37040,"name":"Md Azmain Adib Pahlowan","display_name":"Azmain Adib","email":"adib1905@gmail.com","username":"aadib"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37040,"name":"Md Azmain Adib Pahlowan","display_name":"Azmain Adib","email":"adib1905@gmail.com","username":"aadib"}],"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":37058,"name":"Mohamed Hassaneen","email":"mohammedashoor89@gmail.com","username":"mohamed"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-26 15:49:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-06-26 15:57:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-08 04:11:49.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2024-07-10 14:56:02.000000000","updated_by":{"_account_id":37058,"name":"Mohamed Hassaneen","email":"mohammedashoor89@gmail.com","username":"mohamed"},"reviewer":{"_account_id":37058,"name":"Mohamed Hassaneen","email":"mohammedashoor89@gmail.com","username":"mohamed"},"state":"CC"},{"updated":"2024-07-25 00:04:52.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"}],"messages":[{"id":"3121e7eb85b8e01852f8e703926a76191a6cbd68","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37040,"name":"Md Azmain Adib Pahlowan","display_name":"Azmain Adib","email":"adib1905@gmail.com","username":"aadib"},"date":"2024-06-26 14:41:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"220b7ba04555ac9555c8d0a86870e0e4dac8dc3f","author":{"_account_id":37040,"name":"Md Azmain Adib Pahlowan","display_name":"Azmain Adib","email":"adib1905@gmail.com","username":"aadib"},"date":"2024-06-26 14:54:49.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"79ad34277727afea5ac012717826e6b4c8f135a1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 15:49:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e449ec11e8f24280950823898bb37682\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/31d5cd3ac06c4f55b9c1cfda30a0fcb4 : SUCCESS in 25m 23s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/207cdd2949a049e7bc930f990da92a02 : SUCCESS in 25m 38s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/ecc75b04f51048ecbb31b67516df5f36 : SUCCESS in 30m 44s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/cf3c98a398704afd8e22e5380d701bd7 : SUCCESS in 1h 07m 23s","accounts_in_message":[],"_revision_number":1},{"id":"3e9a3879a398877e07b570d05ba196f4d4e2854d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 15:57:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/55f32f5c2add4571b3ce3fc665f473cf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87d34de5d0544226b6583a6ca4f9a81e : SUCCESS in 8m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/2295b3ce914b4886af73f0a053ab8636 : SUCCESS in 1h 02m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c8c210da2c9046aa8bf274091fcb1142 : SUCCESS in 49m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6ac61edfe1f74fdb855470530cbd0256 : FAILURE in 51m 38s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e7f5066333ed418c9a9f35b40ba70d3f : SUCCESS in 16m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fbf7b45713344eb8b8423e18253691f0 : SUCCESS in 20m 13s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/052731112534404697eea668970e4ef2 : SUCCESS in 18m 43s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1ba9c79ac22043038dd01c1cda25d0d0 : SUCCESS in 9m 10s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1a1009ef0d0a468eaee6f8a68874bc0f : SUCCESS in 5m 05s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/9a2a0c400d5c4165a038a29aa90faed3 : SUCCESS in 17m 59s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4b8efd97103f4b5192cb5191a3108735 : SUCCESS in 14m 08s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/97978a243ab74882a4d65a054e94db10 : SUCCESS in 15m 06s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/ef2dcca770354431ad6923d7b4357e49 : SUCCESS in 18m 37s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/24fff7bb48a74e2bade5379b744fd2ad : SUCCESS in 21m 30s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/f45c03c1c0e74bfeaf0d0158ced846c7 : SUCCESS in 15m 50s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/1dcb4222f004497baee3779f4f1b31b8 : SUCCESS in 15m 47s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/fe5708096a714a9a940c2621ce9b6dbf : SUCCESS in 21m 36s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/1a79819b239a4ba5abfce226e8b76e54 : SUCCESS in 19m 05s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/50db3a530c4f431c8ae182380c296431 : SUCCESS in 16m 48s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/4fc2f5e673de42fba170cf9ef9274f95 : SUCCESS in 17m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e89d2614f4ae4b67b6219c8a2e079e06 : SUCCESS in 8m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/95ac44981ed2406fb8f51f3fa912d68e : FAILURE in 5m 19s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/24cbb11873134d51939c4930597afdc3 : SUCCESS in 6m 40s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/14e649ef55f04c778bd4479f9f05248d : SUCCESS in 56m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3580bbddacdc46c8888c7276985f93f3 : SUCCESS in 46m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a3928a80808641e4ae6bc217fd684a44 : SUCCESS in 55m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1145178d30634d009ee199a6f28d85a5 : SUCCESS in 20m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f76b265e1c9445689975f1f7563271fc : SUCCESS in 4m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/51915070cbf94e56a4bbdfe032a118f1 : SUCCESS in 23m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f8cbc15504884fc19dad22006b544e62 : SUCCESS in 44m 05s","accounts_in_message":[],"_revision_number":1},{"id":"3167bc4bd2849ec163d9801133f2849851d45a4e","author":{"_account_id":37040,"name":"Md Azmain Adib Pahlowan","display_name":"Azmain Adib","email":"adib1905@gmail.com","username":"aadib"},"date":"2024-06-27 09:39:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a9e684dd95efdb8df1dbdf7b8eac345783b5e486","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 10:50:33.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/37deda0ef0814e49be68d3e82d2bd3fb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d256dcdcb6a64f0fb4f5d6f81abb44da : SUCCESS in 8m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/243137d6bc694bcc87d183573f9f9fd8 : SUCCESS in 1h 05m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3e264edf1831472a94d41d1c3a309913 : SUCCESS in 50m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/091dd834c7cf4a57b7e36b588fc6ec8b : FAILURE in 58m 28s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/baa2b97a1d504caf85f4bcf7d422f41b : SUCCESS in 20m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c7d3da9d7b584a0c9cfb72ee86131ac7 : SUCCESS in 19m 16s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a546a3ea39604596a233b9dbf02246bc : SUCCESS in 22m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b026fa5162e841f8aea12e55228b1e89 : SUCCESS in 10m 17s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/1ea808ac3c4f46e39907c24b0e4eb8a9 : SUCCESS in 6m 37s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/41cdc6c33c9e410783192546bf87a10c : SUCCESS in 31m 00s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/b9d41b8e8dbd40c6ade66efe22572adc : SUCCESS in 18m 03s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/87112d08b6614daea319634ce0d5ed96 : SUCCESS in 15m 35s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/70fc218c22d3476fa47a4ab3f247f806 : SUCCESS in 17m 15s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7bf19d03009742ad8a7aec49ce879f58 : SUCCESS in 23m 43s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/4c5062fe8b32416d98a90ae5239c63f4 : SUCCESS in 20m 55s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/97386c3c54bc48cfbd2ce7ce124eed00 : SUCCESS in 19m 00s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/f766df907e3a482ead2063aa5b41f331 : SUCCESS in 18m 22s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/97026838e7e84b9a9b13e0abd92d8acd : SUCCESS in 16m 50s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/e1572ed57877490db1dc3dcb03d784ce : SUCCESS in 15m 26s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/72aeb1539c1b43f4afe30be7f576b22c : SUCCESS in 17m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0ffaadd10a39475bb8842d7ac48889c4 : SUCCESS in 5m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e8313b72f1d04182be935aaae0a61f43 : FAILURE in 5m 39s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9e62dab46f6149ada4edf97ae07a7cc7 : SUCCESS in 7m 19s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/aad2da33ee5242289319b274fd2e6048 : SUCCESS in 59m 57s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a2d3e3d8bb2b42189a9cb6d8cd5cc08d : SUCCESS in 48m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f9995e0e9d2b4f11bb80195f6a4fd91c : SUCCESS in 39m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1868cbd96d8e47d497c993c6fe512e89 : SUCCESS in 20m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49d7a4130c3a469f98752425739440be : SUCCESS in 4m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/13306849e0984166a082df3275918d28 : SUCCESS in 16m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6bf6be44641c45feb637ff4d0463e76d : SUCCESS in 53m 00s","accounts_in_message":[],"_revision_number":1},{"id":"4865f174c56524f0b0baa1a195f67aa70cf878f1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-27 11:12:00.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4afafe634814dcb8cb71f49a2fba08b\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1acf757da14f4ea4b33eba7626677cdd : SUCCESS in 59m 43s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/af4cc43b1881459ba2a8c80525035d9e : TIMED_OUT in 1h 30m 58s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/922b8a89c7164accb9d58998c7dc0eef : SUCCESS in 22m 21s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/4a4347e03c684ccaaaa1ee3f0467b52f : SUCCESS in 1h 13m 49s","accounts_in_message":[],"_revision_number":1},{"id":"bd814eee2ad6dcc60bb584bc168852b288011644","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37040,"name":"Md Azmain Adib Pahlowan","display_name":"Azmain Adib","email":"adib1905@gmail.com","username":"aadib"},"real_author":{"_account_id":37027,"name":"Mohammed Al-Jawaheri","email":"mjawaheri02@gmail.com","username":"mjawaheri"},"date":"2024-06-30 13:52:01.000000000","message":"Patch Set 2: Patch Set 1 was rebased on behalf of \u003cGERRIT_ACCOUNT_37040\u003e\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[{"_account_id":37040,"name":"Md Azmain Adib Pahlowan","display_name":"Azmain Adib","email":"adib1905@gmail.com","username":"aadib"}],"_revision_number":2},{"id":"141b87cb76b59cb258720a9676bf7430b2f5039e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-30 14:19:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5da6499f33049c58d8744a6b5ac3dcd\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/45e06c76938542cb9374c0c39211a0d4 : SUCCESS in 19m 16s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e450b41bd84f4ee1b9eda570b7feee24 : SUCCESS in 16m 47s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/de3fd240647a4cc983f812c5b65376fd : SUCCESS in 19m 23s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/cb71babb9b1a4d6490548aafa394db66 : SUCCESS in 26m 08s","accounts_in_message":[],"_revision_number":2},{"id":"f4b1280cde78feac434017b4bbc694aed711bdbc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-30 14:59:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b123bb453613438b8722ab03ec4f40c0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b27d659e4d464244a6a76053fbc8a4f2 : SUCCESS in 8m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/60508c56a9eb4dcb84d2bcf8eb00237c : SUCCESS in 1h 04m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/32fdb657835944b4aed5013f61f837ed : SUCCESS in 49m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/379eb04dda774399b144f79815286395 : SUCCESS in 1h 00m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/122e8cf140254a95b01f0177260e30d0 : SUCCESS in 20m 50s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0f07063c4316422aa9aa0452bfc027ec : SUCCESS in 20m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bd8690e24a6d4b32ad88a25f4a185851 : SUCCESS in 19m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8f7b05465d0549748ce2fcc3344b0784 : SUCCESS in 5m 27s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/8ce6dffcc93d413fa84e1764c7243cd8 : SUCCESS in 19m 23s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/3cbe892ebed1464d9763468eb57bf57e : SUCCESS in 20m 15s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/4b9f0b2294464e45971b77461a3dde6f : SUCCESS in 18m 14s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/a41faad4cdad464fa0b07b7c8e958644 : SUCCESS in 19m 42s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7943b28ff4ad463298ffe34fc6abd47e : SUCCESS in 27m 26s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/0a6ced43e57448bfa7a17cc37cf3041a : SUCCESS in 18m 13s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/cd69fb5b2d3d4ce8bd667a4ab706c39a : SUCCESS in 17m 58s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b2633d41fe6b41dfaf9a8ca857663557 : SUCCESS in 20m 05s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/befd503810694cf1911b9bb4b9c0a883 : SUCCESS in 16m 41s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/54a02dc770d9448687b61da4822a6a24 : SUCCESS in 16m 29s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/1b526a3fd9dd49a99ff6c0f7faa671bd : SUCCESS in 17m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/15759aa45f934469b17c72334e1b8946 : SUCCESS in 5m 56s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/87d46d51349f49bf8a86fe5da25366d7 : SUCCESS in 15m 09s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/45c7adf530e84bd08b8db6918fbcd66e : SUCCESS in 6m 49s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/326b05a2240d4f46ae9384cc07f4787a : SUCCESS in 57m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c6265b715be148189c25ac1a1ea03b43 : SUCCESS in 50m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6998750b6a1047fca08e6bee3b688d63 : SUCCESS in 50m 57s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4499d0b87bbe43eeb3882a44c31843df : SUCCESS in 16m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb823ded28224175b59325794d5e1148 : SUCCESS in 4m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/31fd597c8e7b457a887492a3ca5cefd2 : SUCCESS in 16m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c5b62687f3db42e5b06a363700c1c100 : SUCCESS in 54m 44s","accounts_in_message":[],"_revision_number":2},{"id":"35a2f4193e8dd5a8602b23fca59a1369c472b23d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37085,"name":"Daanish Khan","display_name":"Daanish Khan","email":"daanishk@andrew.cmu.edu","username":"daanishk"},"date":"2024-07-04 09:24:12.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"e168428fa73d54c3a4cfd871b8256ec34a1728df","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 10:56:20.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f4416772d294ee7a5a4a91ddda84e7b\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/972be1a10efe412785b1373b38dfddca : FAILURE in 20m 08s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3e90ac04f7b54a26afa5c84168c5bf71 : FAILURE in 18m 05s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/4bfa61a5ad4f49e8b12c876d8c08ed57 : SUCCESS in 23m 22s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/6dad293625354162a66a04617577e13c : SUCCESS in 1h 29m 41s","accounts_in_message":[],"_revision_number":3},{"id":"43f3af085c12480a819577aca9c0b9689dcf44e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 11:30:58.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/aec87b00db484a3d9e1aee679de14d99\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ceb121bbf8c4bb295e4d05cd868ad09 : SUCCESS in 11m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/1be072351bcb4e3aaf3a93d1ee54101a : SUCCESS in 1h 14m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b431a5237b5f4fae9fdb7f7cb3801e7c : SUCCESS in 54m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8d3e7c90f7f749489642cf6fc1ff8503 : SUCCESS in 1h 04m 05s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dce2579b7c1d475bb88722675f9207b1 : SUCCESS in 21m 20s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d454750e3f6a4ff498a2448b4becc303 : SUCCESS in 22m 37s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/15bb377fee0840f0b0902bba51dc360a : SUCCESS in 20m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/12e09570f54a412ba26bb6bd32d94095 : SUCCESS in 5m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/7e92adce05e74df6967091c3ebcd1f7d : FAILURE in 18m 32s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fb61341b0a3c4e07af5e516b981f64d0 : FAILURE in 19m 39s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/a0649f37be924242863151444dd7e26d : FAILURE in 18m 59s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/bfae137dc281495aba6655d7a36248cf : FAILURE in 18m 27s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5e0bff26f53043c1a38c340f20df50bf : FAILURE in 24m 16s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/683322f440704c3f8ad50dc37f020778 : SUCCESS in 17m 16s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/6dd18ab34b4343228c885f77ca64c105 : SUCCESS in 26m 51s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b2b25cffcd454809a7104ea035b64bd1 : SUCCESS in 21m 29s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/bc167957a0c042679c6f5d68b2cfdd53 : SUCCESS in 16m 21s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/1abb4f5168e04a4baeb3876fdee8b92c : SUCCESS in 16m 35s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b3dfe44fc93647edb6da847fdefdc19e : SUCCESS in 18m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ce101106d6f1433dba2cc7e26d9a7d06 : SUCCESS in 8m 09s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2f0a466527c4488a9a381a75162fdbe2 : SUCCESS in 14m 50s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3f1d07043fc74c7aa3867904982b48d2 : SUCCESS in 7m 02s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/da76f7e79bec4ce8a472f5f67af25157 : TIMED_OUT in 2h 00m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0a8b9d85e3a04629a6497c5af075f7a3 : SUCCESS in 44m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/71537d4097f64e919323396acf334ef5 : SUCCESS in 48m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/eb39bff86e344216bd514fe24355426a : FAILURE in 19m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/072d985f83264fc9a3c4bcadd588d8bb : SUCCESS in 4m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/13c989537d8f46d4821fadd0be8afaff : SUCCESS in 18m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/abe59878a4dd42038d69466d414e1367 : SUCCESS in 1h 02m 15s","accounts_in_message":[],"_revision_number":3},{"id":"dd63ff41d9e09b5ade67b558d574a14b22914b55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37085,"name":"Daanish Khan","display_name":"Daanish Khan","email":"daanishk@andrew.cmu.edu","username":"daanishk"},"date":"2024-07-07 15:18:25.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"005bdf24da2ea0342b6a406b84e7d8d2187e219a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-07 15:44:18.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8955e1d365ab469588a9e4047495a694\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/566a3fdd58cc4934b8ad42131dd5c7ce : SUCCESS in 15m 34s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6059e81aa504493fa1987543bd8a657f : SUCCESS in 16m 39s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/c4c869c3abae41229744e9100ed0107d : SUCCESS in 23m 02s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/923ace2a28944523b29d0d4c7ed8aa63 : SUCCESS in 21m 41s","accounts_in_message":[],"_revision_number":4},{"id":"b4b8fdd5c05a062e21ae10b7f61902326b870b85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-07 16:22:46.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9829859650e44b8a974a0354d024d851\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1262b9c6d5be4ff6969e4d756f9e9410 : SUCCESS in 7m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/75f5cbc910a647ac802a2c5336b1f499 : SUCCESS in 39m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3228ccdc72014745a28f580540f947bf : SUCCESS in 31m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0c69248206454e3190de4d1de150f947 : SUCCESS in 1h 00m 17s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e50a97b6ec254fd49af2a1f4e1fc4497 : SUCCESS in 19m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d43dd29138f44b4d980361c69a3f9c3c : SUCCESS in 19m 15s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d94f6f1c236340e3b3ca1e75c1459a38 : SUCCESS in 20m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/25f1bff297a6442ebe155c7dac4cbc3a : SUCCESS in 4m 42s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/207a613bfa114306b7de6c7aaff7519a : FAILURE in 4m 42s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9748ebf8efad4794b233a5e7a7d78c87 : SUCCESS in 19m 41s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/1b42858c1c9d4c2d9958ea0df7c9c6e6 : SUCCESS in 18m 55s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/7839726323084200890ff8d327eee2f4 : SUCCESS in 18m 01s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6f38d05d335d4b28904132c669c64820 : SUCCESS in 23m 05s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/233f663ddeca453992c6a8d6cc558391 : SUCCESS in 17m 54s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ad2c6858ad2643ec8cc5f0cdafb152b6 : SUCCESS in 19m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/5d2775f4f29744b4a7c87b3bf1d5c29a : SUCCESS in 18m 27s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/a18d6b6ee9de42c2a2705a9996d729d2 : SUCCESS in 21m 59s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/f715c6d81e144c3e950fdb58176768d8 : SUCCESS in 16m 45s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/fb03d162b28249d789ca57e9706c4360 : SUCCESS in 17m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d03ed0d1bae547148699d7635cf2ea7f : SUCCESS in 5m 43s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1e842a54804146b5a0286eea92ccf709 : SUCCESS in 15m 05s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f119b80f078a411dbc33396fe386b3f8 : SUCCESS in 6m 29s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/57acbbe52b7348f88e18a67958475927 : SUCCESS in 59m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3d8e020df3cf44c29293c66418389983 : SUCCESS in 51m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/088b7741bdca4ec99c661d15b55bafa6 : SUCCESS in 54m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7e24756373e54b0fa3aa233d8a16eb3c : FAILURE in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3dfae9afc144ab1a3d97d07452701fc : SUCCESS in 4m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3c979778f0e44b50879a2e6ac9856f39 : SUCCESS in 17m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1c173d9451e84acc99706e41fc4d88f4 : SUCCESS in 59m 39s","accounts_in_message":[],"_revision_number":4},{"id":"91cc3ae80504aff7e957ef8a007facbe9151cd4b","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-07-08 04:11:49.000000000","message":"Patch Set 4: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":4},{"id":"1b4185a2d7b9613787ea43fa00d60cc3286d0f25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37085,"name":"Daanish Khan","display_name":"Daanish Khan","email":"daanishk@andrew.cmu.edu","username":"daanishk"},"date":"2024-07-09 14:59:47.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"38edc21be0223cbda05313bcd5aa51656424ccd5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 16:02:23.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e29ae2427ee44c6a14c673324c931e7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a81c818b32f493fa6986847e9726874 : SUCCESS in 8m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/ca4a8c9b13c049609b1fe497537f1a2e : SUCCESS in 49m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/41c8fa8c554a4d7dbd6c11ab4ddaad8c : SUCCESS in 38m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a7a522a772cf40c19f496ec9a767b62e : SUCCESS in 59m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dbf80e0000184ef7b2fec4427f6cf918 : SUCCESS in 17m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8a8dc64162374931a57f4e7cedec2b8f : SUCCESS in 21m 39s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a56c7aede654470c8eb00969814edcc2 : SUCCESS in 17m 47s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c82da57289964be086ed7b060942d29a : SUCCESS in 5m 16s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/e7f0329e7af540f68720b51cac1601d2 : SUCCESS in 21m 44s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/441458657a1044a1b93e45c5a00c9ff1 : SUCCESS in 19m 14s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/cec9feb2e0f54a3da41249eabef087d4 : SUCCESS in 22m 24s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/3da63cd0be5d41cead0cc46f6b9f5536 : SUCCESS in 21m 07s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/66097af0c2c94f5b84d80c6bc72f54bb : SUCCESS in 27m 56s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/8a776ab43a594ee996aa1a23162c52d3 : SUCCESS in 18m 01s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/9ed9faea15d24049b06bd721fa48da7c : SUCCESS in 18m 15s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/2dd91939219b4aac84e926a89e5ca02a : SUCCESS in 21m 49s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/5d5b02ddfebd47868992d9a17d66e1b7 : SUCCESS in 20m 32s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/093f0ab94a8a42f1a0922e4427bdc25c : SUCCESS in 19m 28s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/adc79db044d74d1d9d8e819fc040f1b1 : SUCCESS in 21m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b57b6ee35b7c4ca18488863b2bd9c730 : SUCCESS in 7m 50s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7e46d385c87b419aa13dc3849ce381a3 : SUCCESS in 15m 16s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f6a120a8dd124f6abd8fa256ba84ceca : SUCCESS in 9m 50s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d8a5abc1ff5a4365a456cba4ee42c02d : SUCCESS in 1h 01m 26s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/059c018734fb4816a94bf3229829014a : SUCCESS in 52m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cebc18daecc74204aa03b3bf4c1a8322 : SUCCESS in 38m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dc2882b71deb47028c4db1af6e046c8b : SUCCESS in 20m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b78db6977a8b4694aad422eaaff7a22e : SUCCESS in 4m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2c5bbc91bc704695ad573f5cdb318de6 : SUCCESS in 19m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/837fe3e9356f4ea0bb00762931275eae : SUCCESS in 42m 25s","accounts_in_message":[],"_revision_number":5},{"id":"42b386683a721c1266e197bae1756cd19d65fa1a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 16:32:50.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f9d3491fcf841aea90703c3a19835c7\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/02d78bc823a344318139e5501ec783e5 : SUCCESS in 1h 06m 38s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fc62b731a27d42e9a04be00ca45bf603 : TIMED_OUT in 1h 30m 53s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/6b8dc33b7f3e4cf882f20770e31ee63d : FAILURE in 4m 04s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/7cef3e639cbb496d847d5d58f47af883 : SUCCESS in 1h 20m 50s","accounts_in_message":[],"_revision_number":5},{"id":"0a52c69e3b733e7259b6b74285e9566df0018c6e","author":{"_account_id":37058,"name":"Mohamed Hassaneen","email":"mohammedashoor89@gmail.com","username":"mohamed"},"date":"2024-07-10 14:56:02.000000000","message":"Patch Set 5:\n\n(5 comments)","accounts_in_message":[],"_revision_number":5},{"id":"fcd4f2ebdc3626ebf9c762d89cbb9627973ab7a8","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-07-19 06:46:41.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ec8fdbc3218660a62f77cdf8ed25ec8ad99a6af4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37085,"name":"Daanish Khan","display_name":"Daanish Khan","email":"daanishk@andrew.cmu.edu","username":"daanishk"},"date":"2024-07-22 19:27:05.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"2631b900fba290e4aa2ef8e4de76b2512df89d1b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 20:27:53.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9b595557f9a4d7894e9d978734b91d6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/27a3e6b6d07c4f5dbc0e1f76036f737a : SUCCESS in 7m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/fba51e1eb2f64105976ec2a1ae9b0c86 : SUCCESS in 44m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/24a77baec44a4d9bac755ce5eaadf20d : SUCCESS in 37m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/75b0b9b1724c431084ccc9ea0547e769 : SUCCESS in 46m 54s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b0a32f39b0ff42f4aaf5d11cc6b748e6 : SUCCESS in 14m 42s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7770be68244547938921b6d0ac9364a0 : SUCCESS in 16m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6850b55c59b3418fb920489ac0fd55bf : SUCCESS in 16m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b3bf93ce13b043f299460ea9e79935f5 : SUCCESS in 5m 30s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/cff9fef63eb94987a533ece4cce06db0 : SUCCESS in 13m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/833fcb2315894d7fb535be57190e45b7 : SUCCESS in 17m 13s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/bd5d82ca54f748c9b73a02cdc431c3b9 : SUCCESS in 14m 28s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/fe2c4384b6a2417c89e25ed4eca68dab : SUCCESS in 18m 03s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/253d382e390e4ce29399ca1ecf8a9d5b : SUCCESS in 24m 46s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/047a567224ae4ba1977de5c59476cecf : SUCCESS in 17m 38s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/b6934a01cd80462db905c0819d209bde : SUCCESS in 18m 01s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/060a399f3cda4794b8b808ba387e2f60 : SUCCESS in 18m 26s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/7ebbeb87b91548679c7d5ee0ef66aabd : SUCCESS in 16m 54s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/8a9915ac89264688853ab62ea5287964 : SUCCESS in 18m 18s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/75ad43e83fe240638163081954444781 : SUCCESS in 18m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/54f2836a44614409a83a8977ca4f5ca4 : SUCCESS in 7m 32s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/944a5b402a9e4339a9b8bc11ca4494c8 : SUCCESS in 14m 41s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/114ac2f0da894c8ba17929f00a3c3cce : SUCCESS in 6m 26s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/78f7cd0696a3443daa2acedff6be98e4 : SUCCESS in 55m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a6102a8de1074065ab8a63bc810a2591 : SUCCESS in 46m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a5b3147467864bc9b6fb45ea72f4ef8c : SUCCESS in 46m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1944f6d0b94c4e2babc8fbf90a647a40 : SUCCESS in 26m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/724f2ddea09a436e934b63e75cf35f93 : SUCCESS in 4m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5e4071e6595e4c08aacb74d8924c4934 : SUCCESS in 20m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9cf8dc74ccdc449787bb8b7b9f547726 : SUCCESS in 44m 49s","accounts_in_message":[],"_revision_number":6},{"id":"dc47b3c5bc539782e2f9e401b68def1dd63ea2c6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-22 20:56:35.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53d3569af236461bb11aa251cdf75c38\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3aec605604fc42849f4caad055ff2555 : SUCCESS in 18m 02s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/71e52d07eb264711ba16364b1ef5efef : SUCCESS in 18m 56s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/03f3a04f959944ce81385783f4666430 : SUCCESS in 24m 48s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/5cd27b986a86423c8a44aff5a11770a4 : SUCCESS in 1h 27m 58s","accounts_in_message":[],"_revision_number":6},{"id":"8be39bb26cefb5434f896de722d2cfb04e335501","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37085,"name":"Daanish Khan","display_name":"Daanish Khan","email":"daanishk@andrew.cmu.edu","username":"daanishk"},"date":"2024-07-23 12:38:16.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"c45e4e07ce45bd901eee83c546275ee02363a3dd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 13:02:57.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d058ca58b394c4f8ed5d43c495a6016\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9c319baaf65645b28c56dff8752a1626 : SUCCESS in 18m 33s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6bad9dadda5c41d4a673fdeda30aea70 : SUCCESS in 19m 46s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/822ad97c1f8e41cbbcf633f7d6c5d31e : SUCCESS in 23m 34s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/62069e8497bf4ee09bffdcb29cab494d : SUCCESS in 22m 15s","accounts_in_message":[],"_revision_number":7},{"id":"930f3504c66c7768c53d212d0fbfd5643ac27b70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 13:43:35.000000000","message":"Patch Set 7: Verified-1\n\n(6 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/223d9c0b89a14f08b3173ed79e87f073\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5bf5b98a3c34f62ac31a4023878c5a2 : SUCCESS in 8m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/dea6b59f59104326a2f687e697219721 : SUCCESS in 57m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/29e06d358a574068837685a63cdc2285 : SUCCESS in 49m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4fed4324710041a3894376a13a99e4a8 : SUCCESS in 1h 04m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e38933b0506b4d7283a999813472610c : SUCCESS in 19m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e0e2736a57964c37bf98c3574afe493c : SUCCESS in 21m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8d0f1a3bc4274b9aa468366f1a007a08 : SUCCESS in 17m 49s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9c6ef3466b38419e87d5a32a54d4b6ab : SUCCESS in 6m 55s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/082b17dce67746b5a4292530f53fc95f : SUCCESS in 21m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4194c52e01dc4103a655827daf941ae3 : SUCCESS in 16m 41s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/d425ff324c9d4166b3fcdc9865bedfd7 : SUCCESS in 20m 57s\n- swift-tox-py311 https://zuul.opendev.org/t/openstack/build/1d8304d799704e9e8c69c2238517871b : SUCCESS in 26m 23s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f41e38797f94420e958be5b9ed03eb73 : SUCCESS in 28m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/26b4ef2a4d2e42c6b31d709899ceba9e : SUCCESS in 19m 09s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ba32a0fa83494e42938a3f52e9bf0fd5 : SUCCESS in 17m 07s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bd6dd0608c57454e943d4cb82b5e36fd : SUCCESS in 23m 09s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/0866db592177460ba401415cfe98fcd5 : SUCCESS in 20m 52s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/5fee046f90b14ae7b1b2a9120bb144e2 : SUCCESS in 19m 49s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b8d485519bb64b8fb85d9b759eaf6059 : SUCCESS in 22m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9be43bd38c0d48d98d96a1203b6bc8f7 : SUCCESS in 7m 18s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1b8ea6cb6fb948daaf2950e6488dad87 : SUCCESS in 18m 45s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/a9bc02c5a47b4365a803b028103aa37c : SUCCESS in 6m 40s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/275709ed04744ac18891d095305448b2 : SUCCESS in 52m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/908fb1f9ec7c42bfac318e3e756a5f25 : SUCCESS in 53m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/11ca818eb2554277ac8620228c6d21e0 : SUCCESS in 51m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/45ec4ad86c404919b397781f51f3358f : SUCCESS in 20m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ba7688dcb45400ca75c78cc4ca5fcfb : FAILURE in 4m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a1c81e849c104a72a39a3cb614263cdd : SUCCESS in 19m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f74328f79b4d418a9cce2e00c0849ba2 : SUCCESS in 50m 35s","accounts_in_message":[],"_revision_number":7},{"id":"2660599953ec9e46210dc70bdcd92343185b1f6a","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2024-07-24 07:09:39.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"4bd261bc545b16680004bf0e78741f307014b931","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2024-07-25 00:04:52.000000000","message":"Patch Set 7:\n\n(4 comments)","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"93d73ebe588b6fa624640f87c5d7ddcfc49d2f66","revisions":{"4e1fd50d355e93ceefaa118333ffae90430ab1a5":{"kind":"REWORK","_number":1,"created":"2024-06-26 14:41:38.000000000","uploader":{"_account_id":37040,"name":"Md Azmain Adib Pahlowan","display_name":"Azmain Adib","email":"adib1905@gmail.com","username":"aadib"},"ref":"refs/changes/33/922833/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/33/922833/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/922833/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/922833/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/922833/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/33/922833/1"}}},"commit":{"parents":[{"commit":"337079f21f66c4f288c5ffa736a177f446472e5e","subject":"Merge \"replicator: Ensure handoffs can clear with large handoff_delete\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/337079f21f66c4f288c5ffa736a177f446472e5e"}]}],"author":{"name":"GuardiaShield","email":"adib1905@gmail.com","date":"2024-06-26 14:32:53.000000000","tz":180},"committer":{"name":"GuardiaShield","email":"adib1905@gmail.com","date":"2024-06-26 14:39:52.000000000","tz":180},"subject":"Periodic reaper recon dump showing current progress","message":"Periodic reaper recon dump showing current progress\n\nOperators need ways to monitor reaper progress other than logs, and it seems\nlike there are no recon endpoints for reaper. This aims to add recon\nmiddleware support for reaper.\n\nThis patch is a simple refactor. Currently there are a few stats that are\nlogged by scattered variables. This patch is just a semantic refactor so that\nwe have a single unified dictionary that we can use for logging and also for\ndumping as json for recon.\n\nCo-authored-by: Mohammed Al-Jawaheri \u003cmjawaheri02@gmail.com\u003e\nCo-authored-by: Daanish Khan \u003cdaanish1337@gmail.com\u003e\nCo-authored-by: Mohamed Hassaneen \u003cmohammedashoor89@gmail.com\u003e\nCo-authored-by: Nada El-Mestkawy \u003cnadamaged05@gmail.com\u003e\nCo-authored-by: Tra Bui \u003ctrabui.0517@gmail.com\u003e\nChange-Id: I8e3f58017883cf5f81d3d1d9835750b75aecd081\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4e1fd50d355e93ceefaa118333ffae90430ab1a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4e1fd50d355e93ceefaa118333ffae90430ab1a5"}]},"branch":"refs/heads/master"},"b3f9adc9303ae7ca78334e1a83518319a0594a2d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-06-30 13:52:01.000000000","uploader":{"_account_id":37040,"name":"Md Azmain Adib Pahlowan","display_name":"Azmain Adib","email":"adib1905@gmail.com","username":"aadib"},"real_uploader":{"_account_id":37027,"name":"Mohammed Al-Jawaheri","email":"mjawaheri02@gmail.com","username":"mjawaheri"},"ref":"refs/changes/33/922833/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/33/922833/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/922833/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/922833/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/922833/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/33/922833/2"}}},"commit":{"parents":[{"commit":"0c9a524bee2adf2afab83c45d12ff7457640d97d","subject":"Provide content-type for long description in dists","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0c9a524bee2adf2afab83c45d12ff7457640d97d"}]}],"author":{"name":"GuardiaShield","email":"adib1905@gmail.com","date":"2024-06-26 14:32:53.000000000","tz":180},"committer":{"name":"Md Azmain Adib Pahlowan","email":"adib1905@gmail.com","date":"2024-06-30 13:52:01.000000000","tz":0},"subject":"Periodic reaper recon dump showing current progress","message":"Periodic reaper recon dump showing current progress\n\nOperators need ways to monitor reaper progress other than logs, and it seems\nlike there are no recon endpoints for reaper. This aims to add recon\nmiddleware support for reaper.\n\nThis patch is a simple refactor. Currently there are a few stats that are\nlogged by scattered variables. This patch is just a semantic refactor so that\nwe have a single unified dictionary that we can use for logging and also for\ndumping as json for recon.\n\nCo-authored-by: Mohammed Al-Jawaheri \u003cmjawaheri02@gmail.com\u003e\nCo-authored-by: Daanish Khan \u003cdaanish1337@gmail.com\u003e\nCo-authored-by: Mohamed Hassaneen \u003cmohammedashoor89@gmail.com\u003e\nCo-authored-by: Nada El-Mestkawy \u003cnadamaged05@gmail.com\u003e\nCo-authored-by: Tra Bui \u003ctrabui.0517@gmail.com\u003e\nChange-Id: I8e3f58017883cf5f81d3d1d9835750b75aecd081\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b3f9adc9303ae7ca78334e1a83518319a0594a2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b3f9adc9303ae7ca78334e1a83518319a0594a2d"}]},"branch":"refs/heads/master","description":"Rebase"},"bcc30e2a65f5b1e518cb2ea87b003d99bf27606c":{"kind":"REWORK","_number":3,"created":"2024-07-04 09:24:12.000000000","uploader":{"_account_id":37085,"name":"Daanish Khan","display_name":"Daanish Khan","email":"daanishk@andrew.cmu.edu","username":"daanishk"},"ref":"refs/changes/33/922833/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/33/922833/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/922833/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/922833/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/922833/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/33/922833/3"}}},"commit":{"parents":[{"commit":"0c9a524bee2adf2afab83c45d12ff7457640d97d","subject":"Provide content-type for long description in dists","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0c9a524bee2adf2afab83c45d12ff7457640d97d"}]}],"author":{"name":"GuardiaShield","email":"adib1905@gmail.com","date":"2024-06-26 14:32:53.000000000","tz":180},"committer":{"name":"Daanish Khan","email":"daanishk@andrew.cmu.edu","date":"2024-07-04 09:15:20.000000000","tz":180},"subject":"Periodic reaper recon dump showing current progress","message":"Periodic reaper recon dump showing current progress\n\nOperators need ways to monitor reaper progress other than logs, and it seems \\\nlike there are no recon endpoints for reaper. This aims to add recon \\\nmiddleware support for reaper.\n\nLogs and recon now report from same unified dictionary, since a possible \\\nuse-case is tracking progress, periodicity is based on time not on progress.\n\nCo-authored-by: Mohammed Al-Jawaheri \u003cmjawaheri02@gmail.com\u003e\nCo-authored-by: Daanish Khan \u003cdaanish1337@gmail.com\u003e\nCo-authored-by: Mohamed Hassaneen \u003cmohammedashoor89@gmail.com\u003e\nCo-authored-by: Nada El-Mestkawy \u003cnadamaged05@gmail.com\u003e\nCo-authored-by: Tra Bui \u003ctrabui.0517@gmail.com\u003e\nChange-Id: I8e3f58017883cf5f81d3d1d9835750b75aecd081\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bcc30e2a65f5b1e518cb2ea87b003d99bf27606c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bcc30e2a65f5b1e518cb2ea87b003d99bf27606c"}]},"branch":"refs/heads/master"},"0fd3ad05b6c39428b5540bb0154bbb1d55e08231":{"kind":"REWORK","_number":4,"created":"2024-07-07 15:18:25.000000000","uploader":{"_account_id":37085,"name":"Daanish Khan","display_name":"Daanish Khan","email":"daanishk@andrew.cmu.edu","username":"daanishk"},"ref":"refs/changes/33/922833/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/33/922833/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/922833/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/922833/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/922833/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/33/922833/4"}}},"commit":{"parents":[{"commit":"0c9a524bee2adf2afab83c45d12ff7457640d97d","subject":"Provide content-type for long description in dists","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0c9a524bee2adf2afab83c45d12ff7457640d97d"}]}],"author":{"name":"GuardiaShield","email":"adib1905@gmail.com","date":"2024-06-26 14:32:53.000000000","tz":180},"committer":{"name":"Daanish Khan","email":"daanishk@andrew.cmu.edu","date":"2024-07-07 15:17:59.000000000","tz":180},"subject":"Periodic reaper recon dump showing current progress","message":"Periodic reaper recon dump showing current progress\n\nOperators need ways to monitor reaper progress other than logs, and it seems like there are no recon endpoints for reaper. This aims to add recon middleware support for reaper.\n\nLogs and recon now report from same unified dictionary, since a possible use-case is tracking progress, periodicity is based on time not on progress.\n\nCo-authored-by: Mohammed Al-Jawaheri \u003cmjawaheri02@gmail.com\u003e\nCo-authored-by: Daanish Khan \u003cdaanish1337@gmail.com\u003e\nCo-authored-by: Mohamed Hassaneen \u003cmohammedashoor89@gmail.com\u003e\nCo-authored-by: Nada El-Mestkawy \u003cnadamaged05@gmail.com\u003e\nCo-authored-by: Tra Bui \u003ctrabui.0517@gmail.com\u003e\nChange-Id: I8e3f58017883cf5f81d3d1d9835750b75aecd081\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0fd3ad05b6c39428b5540bb0154bbb1d55e08231"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0fd3ad05b6c39428b5540bb0154bbb1d55e08231"}]},"branch":"refs/heads/master"},"441e6095e7d9b3154d4bb24f519e7b9e1b833ca6":{"kind":"REWORK","_number":5,"created":"2024-07-09 14:59:47.000000000","uploader":{"_account_id":37085,"name":"Daanish Khan","display_name":"Daanish Khan","email":"daanishk@andrew.cmu.edu","username":"daanishk"},"ref":"refs/changes/33/922833/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/33/922833/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/922833/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/922833/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/922833/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/33/922833/5"}}},"commit":{"parents":[{"commit":"0c9a524bee2adf2afab83c45d12ff7457640d97d","subject":"Provide content-type for long description in dists","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0c9a524bee2adf2afab83c45d12ff7457640d97d"}]}],"author":{"name":"GuardiaShield","email":"adib1905@gmail.com","date":"2024-06-26 14:32:53.000000000","tz":180},"committer":{"name":"Daanish Khan","email":"daanishk@andrew.cmu.edu","date":"2024-07-09 14:59:26.000000000","tz":180},"subject":"Periodic reaper recon dump showing current progress","message":"Periodic reaper recon dump showing current progress\n\nOperators need ways to monitor reaper progress other than logs, and it seems like there are no recon endpoints for reaper. This aims to add recon middleware support for reaper.\n\nLogs and recon now report from same unified dictionary, since a possible use-case is tracking progress, periodicity is based on time not on progress.\n\nCo-authored-by: Mohammed Al-Jawaheri \u003cmjawaheri02@gmail.com\u003e\nCo-authored-by: Daanish Khan \u003cdaanish1337@gmail.com\u003e\nCo-authored-by: Mohamed Hassaneen \u003cmohammedashoor89@gmail.com\u003e\nCo-authored-by: Nada El-Mestkawy \u003cnadamaged05@gmail.com\u003e\nCo-authored-by: Tra Bui \u003ctrabui.0517@gmail.com\u003e\nChange-Id: I8e3f58017883cf5f81d3d1d9835750b75aecd081\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/441e6095e7d9b3154d4bb24f519e7b9e1b833ca6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/441e6095e7d9b3154d4bb24f519e7b9e1b833ca6"}]},"branch":"refs/heads/master"},"a321f78e779684f0828962e9ef21e21f2061883b":{"kind":"REWORK","_number":6,"created":"2024-07-22 19:27:05.000000000","uploader":{"_account_id":37085,"name":"Daanish Khan","display_name":"Daanish Khan","email":"daanishk@andrew.cmu.edu","username":"daanishk"},"ref":"refs/changes/33/922833/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/33/922833/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/922833/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/922833/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/922833/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/33/922833/6"}}},"commit":{"parents":[{"commit":"533a601cbfa83ce1ad007b3bd83ab202602425e9","subject":"Merge \"swift-drive-audit: Work with ISO timestamps\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/533a601cbfa83ce1ad007b3bd83ab202602425e9"}]}],"author":{"name":"GuardiaShield","email":"adib1905@gmail.com","date":"2024-06-26 14:32:53.000000000","tz":180},"committer":{"name":"Daanish Khan","email":"daanishk@andrew.cmu.edu","date":"2024-07-22 19:26:29.000000000","tz":180},"subject":"Periodic reaper recon dump showing current progress","message":"Periodic reaper recon dump showing current progress\n\nOperators need ways to monitor reaper progress other than logs, and it seems like there are no recon endpoints for reaper. This aims to add recon middleware support for reaper.\n\nLogs and recon now report from same unified dictionary, since a possible use-case is tracking progress, periodicity is based on time not on progress.\n\nCo-authored-by: Mohammed Al-Jawaheri \u003cmjawaheri02@gmail.com\u003e\nCo-authored-by: Daanish Khan \u003cdaanish1337@gmail.com\u003e\nCo-authored-by: Mohamed Hassaneen \u003cmohammedashoor89@gmail.com\u003e\nCo-authored-by: Nada El-Mestkawy \u003cnadamaged05@gmail.com\u003e\nCo-authored-by: Tra Bui \u003ctrabui.0517@gmail.com\u003e\nChange-Id: I8e3f58017883cf5f81d3d1d9835750b75aecd081\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a321f78e779684f0828962e9ef21e21f2061883b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a321f78e779684f0828962e9ef21e21f2061883b"}]},"branch":"refs/heads/master"},"93d73ebe588b6fa624640f87c5d7ddcfc49d2f66":{"kind":"REWORK","_number":7,"created":"2024-07-23 12:38:16.000000000","uploader":{"_account_id":37085,"name":"Daanish Khan","display_name":"Daanish Khan","email":"daanishk@andrew.cmu.edu","username":"daanishk"},"ref":"refs/changes/33/922833/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/33/922833/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/922833/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/922833/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/922833/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/33/922833/7"}}},"commit":{"parents":[{"commit":"533a601cbfa83ce1ad007b3bd83ab202602425e9","subject":"Merge \"swift-drive-audit: Work with ISO timestamps\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/533a601cbfa83ce1ad007b3bd83ab202602425e9"}]}],"author":{"name":"GuardiaShield","email":"adib1905@gmail.com","date":"2024-06-26 14:32:53.000000000","tz":180},"committer":{"name":"Daanish Khan","email":"daanishk@andrew.cmu.edu","date":"2024-07-23 12:37:53.000000000","tz":180},"subject":"Periodic reaper recon dump showing current progress","message":"Periodic reaper recon dump showing current progress\n\nOperators need ways to monitor reaper progress other than logs, and it seems like there are no recon endpoints for reaper. This aims to add recon middleware support for reaper.\n\nLogs and recon now report from same unified dictionary, since a possible use-case is tracking progress, periodicity is based on time not on progress.\n\nCo-authored-by: Mohammed Al-Jawaheri \u003cmjawaheri02@gmail.com\u003e\nCo-authored-by: Daanish Khan \u003cdaanish1337@gmail.com\u003e\nCo-authored-by: Mohamed Hassaneen \u003cmohammedashoor89@gmail.com\u003e\nCo-authored-by: Nada El-Mestkawy \u003cnadamaged05@gmail.com\u003e\nCo-authored-by: Tra Bui \u003ctrabui.0517@gmail.com\u003e\nChange-Id: I8e3f58017883cf5f81d3d1d9835750b75aecd081\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/93d73ebe588b6fa624640f87c5d7ddcfc49d2f66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/93d73ebe588b6fa624640f87c5d7ddcfc49d2f66"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
